)]}'
{"id":"openstack%2Ftripleo-quickstart~330176","triplet_id":"openstack%2Ftripleo-quickstart~master~I40eb8560732f266a1cd4ac37187e5fbb9b620b8e","project":"openstack/tripleo-quickstart","branch":"master","topic":"bug/1569477","hashtags":[],"change_id":"I40eb8560732f266a1cd4ac37187e5fbb9b620b8e","subject":"Move default stopping point to just before overcloud deploy","status":"MERGED","created":"2016-06-15 19:57:03.000000000","updated":"2016-06-16 19:58:56.000000000","submitted":"2016-06-16 19:58:56.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"221683a37647b42b1b90e6e8a7395304200e411f","_number":330176,"virtual_id_number":330176,"owner":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-06-16 19:58:56.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"recommended":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-06-16 05:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2016-06-16 19:58:45.000000000","_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2016-06-16 19:58:45.000000000","_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-06-16 05:32:12.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2016-06-16 19:58:45.000000000","updated_by":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"reviewer":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"state":"REVIEWER"},{"updated":"2016-06-16 19:58:56.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"c51ae0fe3e124abb09f256da33696f704860e9ad","author":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"date":"2016-06-15 19:57:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"280565a4e5a9e5a97a6e3988c8a8287a60c77104","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-15 19:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop http://logs.openstack.org/76/330176/1/check/noop/9fe7441/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e61973af39d58ef9afcf502fef17a63268982530","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2016-06-15 20:38:14.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* tripleo-quickstart-gate-liberty-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-liberty-delorean-quick-basic-529/ : SUCCESS\n\n* tripleo-quickstart-gate-mitaka-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-mitaka-delorean-quick-basic-534/ : SUCCESS\n\n* tripleo-quickstart-gate-liberty-delorean-quick-ooo-usbkey https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-liberty-delorean-quick-ooo-usbkey-174/ : SUCCESS\n\n* tripleo-quickstart-gate-mitaka-delorean-quick-ooo-usbkey https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-mitaka-delorean-quick-ooo-usbkey-178/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"4a22885c5d5101c9cb6f40c49bd6f5245e9bd690","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2016-06-16 05:32:12.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks pretty useful.","accounts_in_message":[],"_revision_number":1},{"id":"f0d7f6a6d4e796477040dd1a7c281fa4c7d9c182","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2016-06-16 19:58:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8de5490e3f26b1acf610a97362b736104afac53","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-16 19:58:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"5fe9e942e765bbb7c6f008173a4dc67c6ad3f7c7","date":"2016-06-16 19:58:56.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"b41a2d5328d6d5de4ec3f2eca710017885a536eb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-06-16 19:58:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop http://logs.openstack.org/76/330176/1/gate/noop/b753fd3/ : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"844cdd856b525cd739325fed14cffa0814d864ff","revisions":{"844cdd856b525cd739325fed14cffa0814d864ff":{"kind":"REWORK","_number":1,"created":"2016-06-15 19:57:03.000000000","uploader":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"ref":"refs/changes/76/330176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/76/330176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/76/330176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/76/330176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/76/330176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/76/330176/1"}}},"commit":{"parents":[{"commit":"5e0a381ff953b3f7700fb1efa8ba1028295467dd","subject":"Gate ansible-role-tripleo-* roles","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5e0a381ff953b3f7700fb1efa8ba1028295467dd"}]}],"author":{"name":"John Trowbridge","email":"trown@redhat.com","date":"2016-06-15 19:51:08.000000000","tz":-240},"committer":{"name":"John Trowbridge","email":"trown@redhat.com","date":"2016-06-15 19:57:02.000000000","tz":-240},"subject":"Move default stopping point to just before overcloud deploy","message":"Move default stopping point to just before overcloud deploy\n\nFor both the developer and new user use cases, there is rarely a\nneed to do anything before the undercloud is installed and images\nand nodes are uploaded and registered. This patch moves up the\ndefault stopping point to just before the overcloud deploy command\nis run. This behavior is still changeable by changing the tags\npassed to quickstart.sh.\n\nAlso fixes a small UI bug, where usage instructions get print out\neven when not using quickstart\u0027s default playbook. These instructions\nare unlikely to be accurate when used with a different playbook.\n\nChange-Id: I40eb8560732f266a1cd4ac37187e5fbb9b620b8e\nCloses-Bug: 1569477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/844cdd856b525cd739325fed14cffa0814d864ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/844cdd856b525cd739325fed14cffa0814d864ff"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5e0a381ff953b3f7700fb1efa8ba1028295467dd","is_merged_in_target_branch":true,"change_id":"Ib1d48d6a79ff94d2a08ea5eef12bb11cdbacf82b","change_number":326037,"patch_set_number":25,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
