)]}'
{"id":"openstack%2Ftripleo-quickstart~433208","triplet_id":"openstack%2Ftripleo-quickstart~master~Ib0e5b361e144e676f9e836e6c24bba216ea7f1d8","project":"openstack/tripleo-quickstart","branch":"master","topic":"ara-generate-deprecation","hashtags":[],"change_id":"Ib0e5b361e144e676f9e836e6c24bba216ea7f1d8","subject":"Replace \"ara generate\" by \"ara generate html\"","status":"MERGED","created":"2017-02-13 17:30:03.000000000","updated":"2017-02-17 20:23:04.000000000","submitted":"2017-02-17 20:23:04.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdfaf80512460032315918a69145db33d8559c8f","_number":433208,"virtual_id_number":433208,"owner":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-02-17 20:23:04.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},{"value":0,"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-02-14 11:24:24.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2017-02-17 20:19:54.000000000","_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},{"value":0,"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":1,"date":"2017-02-17 20:19:54.000000000","_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},{"value":0,"date":"2017-02-17 19:17:48.000000000","_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-02-14 11:24:24.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2017-02-17 19:17:48.000000000","updated_by":{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},"reviewer":{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},"state":"REVIEWER"},{"updated":"2017-02-17 20:19:54.000000000","updated_by":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"reviewer":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"state":"REVIEWER"},{"updated":"2017-02-17 20:23:04.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5cd2af4dabe27f2021f03add7bf550b05e0024fd","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2017-02-13 17:30:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a34df167d9622d783f63872967ad13ef0c54a975","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2017-02-13 17:30:39.000000000","message":"Topic set to ara-generate-deprecation","accounts_in_message":[],"_revision_number":1},{"id":"882990a57fbc7b7967f74f87b8fac6b304d75748","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2017-02-13 18:25:52.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-master-delorean-quick-basic-1509/ : SUCCESS\n\n* tripleo-quickstart-gate-newton-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-newton-delorean-quick-basic-675/ : SUCCESS\n\n* tripleo-quickstart-gate-mitaka-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-mitaka-delorean-quick-basic-2173/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"81923c85d7f59261abcf11d6611b1673186cb3a8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-13 19:10:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-quickstart-linters-ubuntu-xenial http://logs.openstack.org/08/433208/1/check/gate-tripleo-quickstart-linters-ubuntu-xenial/9ced0fd/ : SUCCESS in 1m 47s\n- gate-tripleo-quickstart-docs-ubuntu-xenial http://docs-draft.openstack.org/08/433208/1/check/gate-tripleo-quickstart-docs-ubuntu-xenial/82fe9e9//doc/build/html/ : SUCCESS in 1m 09s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/08/433208/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/b2978c9/ : SUCCESS in 1h 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b7a63243487b77e75493d78962792f820bec2ed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-13 21:59:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq-nv http://logs.openstack.org/08/433208/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq-nv/1e48f27/ : FAILURE in 2h 50m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f128432ba6c934b0ce3f9f9782fbe1ab8d03eb7","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2017-02-14 02:58:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8e5ea3c7ab6e71025bb7044837e4bb69ee65952a","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2017-02-14 03:54:13.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-master-delorean-quick-basic-1515/ : SUCCESS\n\n* tripleo-quickstart-gate-newton-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-newton-delorean-quick-basic-681/ : SUCCESS\n\n* tripleo-quickstart-gate-mitaka-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-mitaka-delorean-quick-basic-2179/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"2e5d8ace902eec74584d2044e3718dc02517745b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-14 04:53:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-quickstart-linters-ubuntu-xenial http://logs.openstack.org/08/433208/1/check/gate-tripleo-quickstart-linters-ubuntu-xenial/042f4ee/ : SUCCESS in 54s\n- gate-tripleo-quickstart-docs-ubuntu-xenial http://docs-draft.openstack.org/08/433208/1/check/gate-tripleo-quickstart-docs-ubuntu-xenial/4acd073//doc/build/html/ : SUCCESS in 1m 27s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv http://logs.openstack.org/08/433208/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq-nv/1dacd86/ : FAILURE in 1h 52m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c657282faf3e7bd2209fef581ee21f71fe7bc13","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-14 06:07:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq-nv http://logs.openstack.org/08/433208/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq-nv/5d32c01/ : SUCCESS in 2h 36m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"998f0655a17b9cb258d3188858c67bdb67a0f4dc","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2017-02-14 11:24:24.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027d suggest to set a precise moment for the deprecation removal, or probably no one except the core developer will ever change it in time","accounts_in_message":[],"_revision_number":1},{"id":"f1f1e32cd47072143cec62f51ee4f11504609389","author":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"date":"2017-02-14 12:49:20.000000000","message":"Patch Set 1:\n\n@Gabriele, ARA does not have a set release schedule. Setting a specific date or release number would make little sense. I leave enough time for people to move on - the last deprecation was done in October 2016 and was removed this week.","accounts_in_message":[],"_revision_number":1},{"id":"218cbe2cf14e6a46628258032a707ee5b1eaf239","author":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"date":"2017-02-17 20:19:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c146860aa412ba69a67a56bb935af2842bb3a10","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-17 20:20:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"b0b8d69dc2cd460a120c6af6e534e42bc8b8cf25","date":"2017-02-17 20:23:04.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"bdfaf80512460032315918a69145db33d8559c8f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-02-17 20:23:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tripleo-quickstart-linters-ubuntu-xenial http://logs.openstack.org/08/433208/1/gate/gate-tripleo-quickstart-linters-ubuntu-xenial/6fd93e4/ : SUCCESS in 1m 41s\n- gate-tripleo-quickstart-docs-ubuntu-xenial http://docs-draft.openstack.org/08/433208/1/gate/gate-tripleo-quickstart-docs-ubuntu-xenial/4822ecb//doc/build/html/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d8b1631db098849bb7352c468d7c2d2d1544f62","revisions":{"2d8b1631db098849bb7352c468d7c2d2d1544f62":{"kind":"REWORK","_number":1,"created":"2017-02-13 17:30:03.000000000","uploader":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"ref":"refs/changes/08/433208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/433208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/433208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/433208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/433208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/433208/1"}}},"commit":{"parents":[{"commit":"ee29ca599b33d70acaec579d5631b026654688c9","subject":"Ensure gate jobs verify sphinx output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ee29ca599b33d70acaec579d5631b026654688c9"}]}],"author":{"name":"David Moreau-Simard","email":"dms@redhat.com","date":"2017-02-13 17:29:34.000000000","tz":-300},"committer":{"name":"David Moreau-Simard","email":"dms@redhat.com","date":"2017-02-13 17:29:34.000000000","tz":-300},"subject":"Replace \"ara generate\" by \"ara generate html\"","message":"Replace \"ara generate\" by \"ara generate html\"\n\nThe newest version of ARA, 0.11, deprecated \"ara generate\" in favor\nof \"ara generate html\". \"ara generate\" will be removed sometime in\nthe future so let\u0027s adjust it right away.\n\nChange-Id: Ib0e5b361e144e676f9e836e6c24bba216ea7f1d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2d8b1631db098849bb7352c468d7c2d2d1544f62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2d8b1631db098849bb7352c468d7c2d2d1544f62"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ee29ca599b33d70acaec579d5631b026654688c9","is_merged_in_target_branch":true,"change_id":"I82d6073e8f130bf6ef155e39c2620cb16ae0e1d6","change_number":425936,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
