)]}'
{"id":"openstack%2Ftripleo-quickstart~504615","triplet_id":"openstack%2Ftripleo-quickstart~master~I8a3b82831891b757f713cbe4599c4ff67d57b064","project":"openstack/tripleo-quickstart","branch":"master","topic":"bug/1717550","hashtags":[],"change_id":"I8a3b82831891b757f713cbe4599c4ff67d57b064","subject":"exclude python-dns from delorean-pike-testing","status":"ABANDONED","created":"2017-09-16 18:08:44.000000000","updated":"2018-05-09 10:50:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3b11d680f895588a715364a3eed84808c36bd81","_number":504615,"virtual_id_number":504615,"owner":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"actions":{},"labels":{"Verified":{"all":[{"date":"2017-09-16 21:09:40.000000000","_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"date":"2017-09-16 18:45:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"date":"2017-09-16 18:24:19.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},{"value":1,"date":"2017-09-19 07:50:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2017-09-19 15:53:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"date":"2017-09-18 01:15:01.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-16 18:24:19.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-16 18:45:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-16 21:09:40.000000000","updated_by":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-09-18 01:15:01.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2017-09-19 07:50:18.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2017-09-19 15:53:11.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"}],"messages":[{"id":"927cf62a7da125fefacf3cab7437896ff90ecbb0","author":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"date":"2017-09-16 18:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f51745e11ae6cb6331eb99a151fd3d0a19429627","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2017-09-16 18:24:19.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* tripleo-quickstart-gate-newton-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-newton-delorean-quick-basic-2565/ : SUCCESS\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-ocata-delorean-quick-basic-1642/ : SUCCESS\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-master-delorean-quick-basic-3401/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b11d01a7788d707f71a16970853066dd1ea6cdd5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-09-16 18:45:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- gate-tripleo-ci-centos-7-undercloud-oooq-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-undercloud-oooq-master-nv/650/ : FAILURE in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3978cf269eeb4fbfdb143554100d6129e03a16b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-16 20:53:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/15/504615/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/3118ac6/ : FAILURE in 1h 43m 04s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/15/504615/1/check/gate-tripleo-ci-centos-7-containers-multinode/2741342/ : SUCCESS in 1h 36m 21s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/15/504615/1/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/54bb6f6/ : FAILURE in 44m 24s (non-voting)\n- gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv http://logs.openstack.org/15/504615/1/check/gate-tripleo-ci-centos-7-containers-multinode-upgrades-nv/ee2f237/ : SUCCESS in 2h 39m 59s (non-voting)\n- gate-tripleo-quickstart-linters-ubuntu-xenial http://logs.openstack.org/15/504615/1/check/gate-tripleo-quickstart-linters-ubuntu-xenial/6d2de68/ : SUCCESS in 2m 34s\n- gate-tripleo-quickstart-docs-ubuntu-xenial http://docs-draft.openstack.org/15/504615/1/check/gate-tripleo-quickstart-docs-ubuntu-xenial/00ad8ba//doc/build/html/ : SUCCESS in 1m 42s\n- gate-tripleo-ci-centos-7-3nodes-multinode-nv http://logs.openstack.org/15/504615/1/check/gate-tripleo-ci-centos-7-3nodes-multinode-nv/2b2bfe8/ : FAILURE in 1h 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4cbb650e4d54d44f68f10e24016b876cb66a29ec","author":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"date":"2017-09-16 21:09:40.000000000","message":"Patch Set 1:\n\n* tq-gate-rhos-11-current-passed-ci-ci-rhos-ovb-minimal-pacemaker-public-bond https://thirdparty.logs.rdoproject.org/jenkins-tq-gate-rhos-11-current-passed-ci-ci-rhos-ovb-minimal-pacemaker-public-bond-39/ : FAILURE\n\nNo such file or directory ( https://rhos-jenkins.rhev-ci-vms.eng.rdu2.redhat.com/job/tq-gate-rhos-11-current-passed-ci-ci-rhos-ovb-minimal-pacemaker-public-bond/39/ )\n\ncommand not found ( https://rhos-jenkins.rhev-ci-vms.eng.rdu2.redhat.com/job/tq-gate-rhos-11-current-passed-ci-ci-rhos-ovb-minimal-pacemaker-public-bond/39/ )\n\nA general failure case for overcloud deployments.  ( https://rhos-jenkins.rhev-ci-vms.eng.rdu2.redhat.com/job/tq-gate-rhos-11-current-passed-ci-ci-rhos-ovb-minimal-pacemaker-public-bond/39/ )\n\nansible script exception ( https://rhos-jenkins.rhev-ci-vms.eng.rdu2.redhat.com/job/tq-gate-rhos-11-current-passed-ci-ci-rhos-ovb-minimal-pacemaker-public-bond/39/ )","accounts_in_message":[],"_revision_number":1},{"id":"1756f19ea849cec6e2a94cd00b9b82e38d8e0edf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-16 23:02:30.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/15/504615/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/4f29ed1/ : FAILURE in 2h 37m 59s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/15/504615/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/f36da84/ : SUCCESS in 2h 31m 18s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/15/504615/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/3f7ba99/ : FAILURE in 2h 30m 41s","accounts_in_message":[],"_revision_number":1},{"id":"00db9ff07f5c10d3a1aa64f5f2cdd59a5862cdd7","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2017-09-19 07:50:18.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good to me","accounts_in_message":[],"_revision_number":1},{"id":"d2b509e06981854f3ade63b1ec3e2e3b9e3573b3","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2017-09-19 15:53:11.000000000","message":"Patch Set 1: Code-Review+1\n\nshould those be noted as FIXMEs?","accounts_in_message":[],"_revision_number":1},{"id":"a41a31dc63512898a2f2a8783d43ef087e2fe841","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2017-12-18 02:17:24.000000000","message":"Removed reviewer Jenkins with the following votes:\n\n* Verified-1 by Jenkins (3)\n","accounts_in_message":[],"_revision_number":1},{"id":"23a711e63656c20a1f7bebfb163048e78d1f38bf","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2018-05-09 10:50:04.000000000","message":"Abandoned\n\nAccording to TripleO policy of patch abandonment https://github.com/openstack/tripleo-specs/blob/master/specs/policy/patch-abandomment.rst, this patch is abandoned. If you would like to continue to work on it, please ask for restoration either on #tripleo in Freenode IRC, or openstack dev mailing list - openstack-dev@lists.openstack.org with [TripleO] in subject. Thanks.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc6f8ef8fcf7ea6212dfe8af280e2758332c65d4","revisions":{"cc6f8ef8fcf7ea6212dfe8af280e2758332c65d4":{"kind":"REWORK","_number":1,"created":"2017-09-16 18:08:44.000000000","uploader":{"_account_id":8745,"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","username":"lars"},"ref":"refs/changes/15/504615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/15/504615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/15/504615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/15/504615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/15/504615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/15/504615/1"}}},"commit":{"parents":[{"commit":"f95d8833d9a50d0cc4dde6dd09eb76472b82fd94","subject":"teach repo-setup about exclude","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f95d8833d9a50d0cc4dde6dd09eb76472b82fd94"}]}],"author":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2017-09-16 16:39:27.000000000","tz":-240},"committer":{"name":"Lars Kellogg-Stedman","email":"lars@redhat.com","date":"2017-09-16 17:05:28.000000000","tz":-240},"subject":"exclude python-dns from delorean-pike-testing","message":"exclude python-dns from delorean-pike-testing\n\nThe python-dns package listed in the delorean-pike-testing repository\nprevents installation of the latest version from centos-base (due to\nrepo priorities).\n\nChange-Id: I8a3b82831891b757f713cbe4599c4ff67d57b064\nCloses-bug: #1717550\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc6f8ef8fcf7ea6212dfe8af280e2758332c65d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc6f8ef8fcf7ea6212dfe8af280e2758332c65d4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f95d8833d9a50d0cc4dde6dd09eb76472b82fd94","is_merged_in_target_branch":true,"change_id":"I85237472cd227cfc9c7758d6f61cac5c04dd310e","change_number":504614,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
