)]}'
{"id":"openstack%2Ftripleo-quickstart~565209","triplet_id":"openstack%2Ftripleo-quickstart~master~I876ee678e0d56bef68867c0aeca882e784d0e835","project":"openstack/tripleo-quickstart","branch":"master","topic":"cleaning-extras","hashtags":[],"change_id":"I876ee678e0d56bef68867c0aeca882e784d0e835","subject":"fs001: enable cleaning, disable inspection extras","status":"MERGED","created":"2018-04-30 12:35:30.000000000","updated":"2018-05-02 20:18:36.000000000","submitted":"2018-05-02 20:18:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"565209-1525292316055-9cedb2f7","meta_rev_id":"60ab20bbab374299ba25dd9c3a52ca3f4b6a9878","_number":565209,"virtual_id_number":565209,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2018-05-02 20:18:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-05-02 12:33:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"date":"2018-04-30 14:06:07.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2018-05-02 15:01:42.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2018-04-30 12:37:55.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2018-05-02 15:17:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2018-05-02 14:51:35.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"date":"2018-05-02 14:51:28.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-04-30 12:37:55.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2018-04-30 14:06:07.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-05-02 12:33:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-05-02 14:51:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2018-05-02 14:51:35.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2018-05-02 15:17:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-05-02 20:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ca8441c002f7124dcb5cbda1db077df9297b301","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-04-30 12:35:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad5bc55de805057ebf67b0b47fc688866813e845","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2018-04-30 12:37:55.000000000","message":"Patch Set 1: Code-Review+2\n\nLet\u0027s see how this works","accounts_in_message":[],"_revision_number":1},{"id":"26c0a66a1cb5da90691f2e0021e54b106bbeecad","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-04-30 13:06:39.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* tripleo-quickstart-gate-newton-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-newton-delorean-quick-basic-4782/ : SUCCESS\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-ocata-delorean-quick-basic-3862/ : SUCCESS\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-master-delorean-quick-basic-5619/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"50835c116b266a2a06c4eb8b0d2fbfe0454740f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-30 13:35:44.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d9dcbf2e7c4477b4860afe7ee502acbaa336fd03","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-04-30 13:35:45.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"ef41a21c5271a94623fc20508c5d5c91b644b021","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-04-30 13:38:10.000000000","message":"Patch Set 1:\n\nrecheck\n\nthe dependency was fixed","accounts_in_message":[],"_revision_number":1},{"id":"f7b39b64105ab4e408e565bf77abddf98fc8d583","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-04-30 14:06:07.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* tripleo-quickstart-gate-newton-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-newton-delorean-quick-basic-4785/ : SUCCESS\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-ocata-delorean-quick-basic-3865/ : SUCCESS\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/artifacts/rdo/jenkins-tripleo-quickstart-gate-master-delorean-quick-basic-5622/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c449a59dfeac335a89e424ed8324c980bfa90b56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-30 16:24:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/09/565209/1/check/build-openstack-sphinx-docs/96fac97/html/ : SUCCESS in 4m 26s\n- openstack-tox-linters http://logs.openstack.org/09/565209/1/check/openstack-tox-linters/ab246c1/ : SUCCESS in 5m 26s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/09/565209/1/check/tripleo-ci-centos-7-undercloud-oooq/a4e626f/ : SUCCESS in 1h 13m 51s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/09/565209/1/check/tripleo-ci-centos-7-undercloud-upgrades/1771ec4/ : SUCCESS in 1h 17m 18s (non-voting)\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/565209/1/check/tripleo-ci-centos-7-undercloud-containers/7ca659a/ : SUCCESS in 1h 21m 22s\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/09/565209/1/check/tripleo-ci-centos-7-3nodes-multinode/b269892/ : SUCCESS in 2h 41m 14s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/565209/1/check/tripleo-ci-centos-7-containers-multinode/6fd91e2/ : SUCCESS in 2h 35m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/09/565209/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1720dd0/ : SUCCESS in 2h 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"57f932dbf734d785a290a0c8ec9673bab93efbb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-04-30 23:57:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- gate-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch-nv/2541/ : SUCCESS in 2h 24m 13s (non-voting)\n- gate-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/241/ : SUCCESS in 1h 37m 57s\n- gate-tripleo-ci-centos-7-container-to-container-upgrades-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-container-to-container-upgrades-master-nv/2657/ : FAILURE in 58m 40s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/12556/ : SUCCESS in 2h 06m 52s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/12173/ : SUCCESS in 1h 51m 51s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/4061/ : FAILURE in 1m 44s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/4030/ : SUCCESS in 1h 28m 46s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1026/ : SUCCESS in 2h 07m 28s","accounts_in_message":[],"_revision_number":1},{"id":"461e0f2a7b80190004939d9a7fc763f32aaa971c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-05-02 09:21:51.000000000","message":"Patch Set 1:\n\ncheck-rdo\n\ncleaning works on master: https://logs.rdoproject.org/09/565209/1/openstack-check/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/Za8a81d6da500487399005bc57da3a223/undercloud/var/log/containers/ironic/ironic-conductor.log.txt.gz#_2018-04-30_22_50_46_062\n\nand on queens: https://logs.rdoproject.org/09/565209/1/openstack-check/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/Za8a81d6da500487399005bc57da3a223/undercloud/var/log/ironic/ironic-conductor.log.txt.gz#_2018-04-30_19_56_53_589\n\nextended introspection data is not returned: https://logs.rdoproject.org/09/565209/1/openstack-check/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/Za8a81d6da500487399005bc57da3a223/undercloud/var/log/ironic-inspector/ironic-inspector.log.txt.gz#_2018-04-30_19_55_17_746\n\ncleaning still disabled on old branches: https://logs.rdoproject.org/09/565209/1/openstack-check/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/Za8a81d6da500487399005bc57da3a223/undercloud/var/log/ironic/ironic-conductor.log.txt.gz#_2018-04-30_19_34_49_602","accounts_in_message":[],"_revision_number":1},{"id":"5132d63028e37b95f7cafcac169b4cf8b5b1b476","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-05-02 12:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded.\n\n- gate-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch-nv/2609/ : SUCCESS in 2h 25m 20s (non-voting)\n- gate-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/308/ : SUCCESS in 1h 41m 12s\n- gate-tripleo-ci-centos-7-container-to-container-upgrades-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-container-to-container-upgrades-master-nv/2731/ : FAILURE in 59m 17s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/12719/ : SUCCESS in 1h 58m 52s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/12358/ : SUCCESS in 2h 00m 06s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/4118/ : SUCCESS in 1h 44m 24s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/4087/ : SUCCESS in 1h 26m 59s\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1084/ : SUCCESS in 1h 59m 41s","accounts_in_message":[],"_revision_number":1},{"id":"dcf429f2b3dd892765e374dea00c0c5b93d84bb2","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-05-02 14:58:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62729b30ac0272ec5d58509035170e057642065c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2018-05-02 15:01:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74d79836a40ae5c83a216ff35f4b85f897be9431","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-05-02 15:17:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cf0290fb266d35910f78f48aed9de61c42d715f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 15:22:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"869defbf5dd4d813bece39ef492546d128b821ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 20:18:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/09/565209/1/gate/build-openstack-sphinx-docs/f829298/html/ : SUCCESS in 5m 35s\n- openstack-tox-linters http://logs.openstack.org/09/565209/1/gate/openstack-tox-linters/c0ce8be/ : SUCCESS in 4m 51s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/09/565209/1/gate/tripleo-ci-centos-7-undercloud-oooq/72a487f/ : SUCCESS in 1h 23m 13s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/09/565209/1/gate/tripleo-ci-centos-7-undercloud-containers/f4613ed/ : SUCCESS in 1h 24m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/09/565209/1/gate/tripleo-ci-centos-7-containers-multinode/6d9abe9/ : SUCCESS in 2h 29m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5ebdaa379c4c4cf518fef434b8d96972c75d70e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 20:18:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d765d4d64f2c12dc8827699dfc67cd36c80480c","revisions":{"0d765d4d64f2c12dc8827699dfc67cd36c80480c":{"kind":"REWORK","_number":1,"created":"2018-04-30 12:35:30.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/09/565209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/09/565209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/09/565209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/09/565209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/09/565209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/09/565209/1"}}},"commit":{"parents":[{"commit":"69e473985a720d06ca924ceef09dd065bfa3f086","subject":"Tag \"Add the virthost to the inventory\" as teardown-","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/69e473985a720d06ca924ceef09dd065bfa3f086"}]}],"author":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2018-04-30 12:33:26.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"divius.inside@gmail.com","date":"2018-04-30 12:34:49.000000000","tz":120},"subject":"fs001: enable cleaning, disable inspection extras","message":"fs001: enable cleaning, disable inspection extras\n\nIronic node cleaning is an important feature, required for Ceph deployments.\nThis change makes fs001 run automated cleaning before deployment.\n\nTo compensate for the increased gate run time, inspection extras are disabled.\nThey are covered by fs035 (and other featuresets with introspection).\n\nDepends-On: I0145b3731ee63b6b05f79706c9ec90a14c1b06e6\nChange-Id: I876ee678e0d56bef68867c0aeca882e784d0e835\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0d765d4d64f2c12dc8827699dfc67cd36c80480c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0d765d4d64f2c12dc8827699dfc67cd36c80480c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"69e473985a720d06ca924ceef09dd065bfa3f086","is_merged_in_target_branch":true,"change_id":"Ia0c70b49bfcabea49b673d605c1bf1775ce007e5","change_number":564712,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
