)]}'
{"id":"openstack%2Ftripleo-quickstart~580119","triplet_id":"openstack%2Ftripleo-quickstart~master~Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5","project":"openstack/tripleo-quickstart","branch":"master","topic":"gate-octavia","hashtags":[],"change_id":"Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5","subject":"FS38: switch to containerized UC and config-download","status":"MERGED","created":"2018-07-04 09:55:42.000000000","updated":"2018-09-16 01:09:12.000000000","submitted":"2018-09-16 01:09:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"580119-1537060152395-37aee634","meta_rev_id":"f5d8dfcdf295eff5723bc5c339d7cef5676546b9","_number":580119,"virtual_id_number":580119,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-09-15 16:02:11.000000000","_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2018-09-16 01:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-15 18:49:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"date":"2018-09-15 17:14:17.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"recommended":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2018-09-07 10:08:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":2,"date":"2018-09-06 16:24:25.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2018-09-14 16:08:57.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"all":[{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},{"value":0,"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"date":"2018-09-05 10:11:26.000000000","_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":1,"date":"2018-09-14 16:18:45.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-07-04 13:47:13.000000000","updated_by":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"reviewer":{"_account_id":6579,"name":"Nir Magnezi","email":"nmagnezi@redhat.com","username":"nmagnezi"},"state":"REVIEWER"},{"updated":"2018-08-15 13:50:10.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2018-08-20 13:24:55.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2018-08-20 19:05:00.000000000","updated_by":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"reviewer":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-03 23:30:59.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-09-05 10:11:26.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2018-09-06 16:24:25.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2018-09-07 10:08:52.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2018-09-14 16:18:45.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2018-09-15 17:14:17.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-15 18:49:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-09-16 01:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c1a60b7d4685ca8bf3be305e6678d283fe7d2e9b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-07-04 09:55:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34db5a5e38a79e522d24274f6d1f42366244386e","author":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"date":"2018-07-04 10:01:51.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tq-gate-rhos-12-ci-rhos-ovb-minimal-pacemaker-public-bond https://thirdparty.logs.rdoproject.org/jenkins-tq-gate-rhos-12-ci-rhos-ovb-minimal-pacemaker-public-bond-1272/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b4f38529d42b3a477cdaba15b54c23f54a7a905f","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-07-04 10:53:54.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6182/ : FAILURE\n\n* tripleo-quickstart-gate-newton-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-newton-delorean-quick-basic/5345/ : SUCCESS\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/4425/ : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"1dcb4b228575cee0140dec2be6d54580f272f4fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-04 13:30:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/1/check/build-openstack-sphinx-docs/d2d2d5a/html/ : SUCCESS in 6m 02s\n- openstack-tox-linters http://logs.openstack.org/19/580119/1/check/openstack-tox-linters/ad88d8b/ : SUCCESS in 5m 38s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-undercloud-oooq/c66b84c/ : SUCCESS in 1h 11m 48s (non-voting)\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-undercloud-upgrades/7b13ea2/ : SUCCESS in 1h 09m 50s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-undercloud-containers/3d7b05e/ : SUCCESS in 2h 40m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/eeeb428/ : SUCCESS in 1h 36m 09s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-containers-multinode/26cda80/ : SUCCESS in 2h 23m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1a66495/ : FAILURE in 1h 24m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8a0f8db/ : SUCCESS in 2h 21m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-containers-multinode-queens/2be83b5/ : SUCCESS in 2h 10m 12s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/1/check/tripleo-ci-centos-7-containers-multinode-pike/c85156f/ : SUCCESS in 2h 16m 27s","accounts_in_message":[],"_revision_number":1},{"id":"87fcb7fa56aebb0ab9da874861a364c8e37afa0f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-07-17 22:54:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"bf61ba93c83c51a09057b27a23356427c9888e35","author":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"date":"2018-07-17 22:59:28.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tq-gate-rhos-13-ci-rhos-ovb-minimal-pacemaker-public-bond https://thirdparty.logs.rdoproject.org/jenkins-tq-gate-rhos-13-ci-rhos-ovb-minimal-pacemaker-public-bond-59/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0160dcc2d4c3ec9eeca5f0496cd23eacddece54a","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-07-18 00:12:53.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-gate-newton-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-newton-delorean-quick-basic/5605/ : FAILURE\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6442/ : SUCCESS\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/4684/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"aedc4eed284cda2afd78149abc9f8997208a1473","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-07-18 01:19:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/2/check/build-openstack-sphinx-docs/e4bbfc4/html/ : SUCCESS in 5m 25s\n- openstack-tox-linters http://logs.openstack.org/19/580119/2/check/openstack-tox-linters/03b12c1/ : SUCCESS in 5m 58s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-undercloud-oooq/65f9a95/ : SUCCESS in 1h 06m 48s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-undercloud-upgrades/d61aa83/ : SUCCESS in 1h 15m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-undercloud-containers/e91048c/ : SUCCESS in 1h 30m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/df82782/ : SUCCESS in 1h 31m 13s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-containers-multinode/dec66e6/ : SUCCESS in 2h 11m 33s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/41140e9/ : SUCCESS in 2h 20m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0531921/ : SUCCESS in 2h 00m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-containers-multinode-queens/79edd1d/ : POST_FAILURE in 2h 18m 40s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/2/check/tripleo-ci-centos-7-containers-multinode-pike/a59d948/ : SUCCESS in 2h 17m 26s","accounts_in_message":[],"_revision_number":2},{"id":"520405c39a5fc3e6acbe19091ebfd562267f412e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-07-18 01:44:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/d9cb307/ : SUCCESS in 2h 24m 46s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/1e1c904/ : SUCCESS in 1h 48m 02s\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/a3cbe5d/ : FAILURE in 30m 39s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/be9ac92/ : SUCCESS in 2h 42m 21s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/8620150/ : FAILURE in 2h 46m 43s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/bfc8743/ : SUCCESS in 1h 40m 43s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/9790e8f/ : SUCCESS in 1h 24m 17s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/2/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/dd8b84c/ : SUCCESS in 1h 47m 53s","accounts_in_message":[],"_revision_number":2},{"id":"dc0068d5210234273374384546b5e579404828c1","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-08-15 10:39:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1aca68cb1a772600df793ea06fcb101f8bec816c","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-15 10:47:57.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/4904/ : FAILURE\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6662/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"eedee930fbb702f85cddc6960f1cd66596886b47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-15 13:29:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/3/check/build-openstack-sphinx-docs/7cb38b2/html/ : SUCCESS in 5m 23s\n- openstack-tox-linters http://logs.openstack.org/19/580119/3/check/openstack-tox-linters/944e31c/ : SUCCESS in 5m 14s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-undercloud-oooq/357bd67/ : SUCCESS in 1h 05m 26s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-undercloud-upgrades/abe9b40/ : SUCCESS in 1h 06m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-undercloud-containers/98011de/ : SUCCESS in 1h 39m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6c0395c/ : SUCCESS in 1h 40m 44s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-containers-multinode/6364a5b/ : SUCCESS in 2h 44m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fbbcbe5/ : SUCCESS in 2h 17m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2e003ae/ : SUCCESS in 2h 21m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-containers-multinode-rocky/d3d411f/ : FAILURE in 9m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-containers-multinode-queens/b89e3e3/ : SUCCESS in 2h 04m 10s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/3/check/tripleo-ci-centos-7-containers-multinode-pike/0a99d41/ : SUCCESS in 2h 20m 10s","accounts_in_message":[],"_revision_number":3},{"id":"f436dab96cd0dc9c281425d232c5985d1da3016f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-15 13:46:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/5d15b68/ : SUCCESS in 2h 29m 13s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/33d66ed/ : SUCCESS in 1h 54m 10s\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/6ec74a4/ : FAILURE in 3h 01m 21s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/4161ee9/ : SUCCESS in 2h 37m 13s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/43756a0/ : SUCCESS in 2h 38m 15s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/08f38ef/ : SUCCESS in 1h 59m 26s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/29fd78a/ : SUCCESS in 1h 59m 01s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/73361df/ : SUCCESS in 2h 08m 02s","accounts_in_message":[],"_revision_number":3},{"id":"668eb8d873d81bdd752dc1167e331c470d2d4e9b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2018-08-15 13:50:10.000000000","message":"Patch Set 3: Code-Review+2\n\nI\u0027d like to see this job running in CI to see if it passes with this config.","accounts_in_message":[],"_revision_number":3},{"id":"7293903bbd9dc738c6406891798bef5a6279ba3c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-08-15 15:57:08.000000000","message":"Patch Set 3:\n\nFair enough. Testing it in https://review.openstack.org/#/c/591413/","accounts_in_message":[],"_revision_number":3},{"id":"58fc2c2eacedf2151697fce4a279148c14a7d64d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2018-08-20 13:24:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"564996adbeeeea35f6db87f06ebc53de0ee98903","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-08-20 14:48:09.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8b45fb655916063e1bfc4a2d673801e90eccc43f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-20 15:22:57.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"6fad2f7cdacba396bd504818cbb1aff28196536f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-20 15:22:58.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"4f53f669486ed02a1e417cd558d44d98711da10f","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-20 16:05:27.000000000","message":"Patch Set 4:\n\nBuild Successful \n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/4945/ : SUCCESS\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6703/ : SUCCESS","accounts_in_message":[],"_revision_number":4},{"id":"51abdf67ce7f52ba41dcbd127d394fa0776cb942","author":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"date":"2018-08-20 16:53:11.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tq-gate-rhos-13-ci-rhos-ovb-featureset001 https://thirdparty.logs.rdoproject.org/jenkins-tq-gate-rhos-13-ci-rhos-ovb-featureset001-139/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"a09158482979821d6cf002b84c3bef6dd8f4045e","author":{"_account_id":24752,"name":"RedHat RDO Downstream CI","email":"rdo-ci-admins+downstream@redhat.com","username":"rdo-ci-downstream","tags":["SERVICE_USER"]},"date":"2018-08-20 19:05:00.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tq-gate-rhos-13-ci-rhos-ovb-featureset001 https://thirdparty.logs.rdoproject.org/jenkins-tq-gate-rhos-13-ci-rhos-ovb-featureset001-146/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e4c3b24aaf7a3d3529934d1aabcca0ceb235f93b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-08-30 18:29:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"054af74c389d35eca24f5cd84a95f3b7dc869a60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-30 18:29:46.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"974f625f4c57f0105db997fb969f5171f79c319b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-30 18:29:51.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"a2a2e40a61608d03e540fe4b5fd3594930533c65","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-30 18:33:05.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6815/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5058/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"356ac407de42692e27a6920510b8c159e1b3d4ff","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-08-30 18:55:23.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d3efa7616a03a0261936e732ecb7e331cfbda3a1","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-30 18:59:11.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6816/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5059/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"8424a121cc6076ff5e50b2ad80f0b4a7924282c3","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-08-30 19:27:25.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c8a6e56536049e64a7b431d0831dbc9af78d82c9","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-08-30 19:31:16.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6817/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5060/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"a350c482ae3ea4ce7afbef960c1336fb7ed63993","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-08-30 19:59:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/73777a9/ : FAILURE in 10m 09s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/2a89041/ : FAILURE in 8m 01s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/688df53/ : POST_FAILURE in 7m 47s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/15d8a19/ : FAILURE in 10m 21s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/afffa63/ : FAILURE in 5m 46s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/fc0af69/ : FAILURE in 10m 05s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/236194c/ : FAILURE in 7m 25s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/6/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/f09c8af/ : FAILURE in 7m 25s","accounts_in_message":[],"_revision_number":6},{"id":"4dc86e2c2a85962448d34ce1ad2229bffc9716ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-30 22:51:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/6/check/build-openstack-sphinx-docs/59299f0/html/ : SUCCESS in 5m 00s\n- openstack-tox-linters http://logs.openstack.org/19/580119/6/check/openstack-tox-linters/bdeccc1/ : SUCCESS in 5m 04s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-undercloud-oooq/260bb1d/ : SUCCESS in 1h 08m 19s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-undercloud-upgrades/cffd46d/ : SUCCESS in 1h 06m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-undercloud-containers/e411cb3/ : SUCCESS in 1h 36m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f83007c/ : SUCCESS in 2h 06m 36s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-containers-multinode/a7694b0/ : SUCCESS in 2h 44m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c548c9c/ : FAILURE in 2h 02m 41s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/66997ab/ : TIMED_OUT in 3h 21m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-containers-multinode-rocky/80ba5db/ : SUCCESS in 3h 00m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-containers-multinode-queens/45e8404/ : SUCCESS in 2h 09m 35s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/6/check/tripleo-ci-centos-7-containers-multinode-pike/9e6508e/ : SUCCESS in 2h 28m 51s","accounts_in_message":[],"_revision_number":6},{"id":"3b280aa5b133a29c275faae9009e39ec02a682d2","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-01 11:14:40.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"235511d92ea915933e30f1fe9333a2ae9c4917cc","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-09-01 12:03:41.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6845/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5088/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"d060e233a07e57f085a82cad320f5fd0f0dd7a11","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-01 13:55:02.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/87cd2c7/ : FAILURE in 22m 34s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/57d7fee/ : FAILURE in 1h 52m 19s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/f0d7dd4/ : FAILURE in 20m 45s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/1024795/ : SUCCESS in 2h 29m 54s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/44e9c86/ : SUCCESS in 2h 34m 08s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/59ba5b5/ : SUCCESS in 1h 44m 01s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/48dff79/ : SUCCESS in 1h 28m 49s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/7/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d85408a/ : SUCCESS in 2h 03m 31s","accounts_in_message":[],"_revision_number":7},{"id":"1107b0b4b25c238ef4f3bc6b6fc89ac70b025f27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 14:44:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/7/check/build-openstack-sphinx-docs/dc52522/html/ : SUCCESS in 4m 08s\n- openstack-tox-linters http://logs.openstack.org/19/580119/7/check/openstack-tox-linters/f5172d4/ : SUCCESS in 4m 11s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-undercloud-oooq/ac7d181/ : SUCCESS in 58m 36s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-undercloud-upgrades/1dd85e4/ : SUCCESS in 1h 13m 48s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-undercloud-containers/33ec7d3/ : SUCCESS in 1h 49m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2b435a4/ : SUCCESS in 1h 44m 11s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-containers-multinode/85cc226/ : SUCCESS in 2h 46m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cbf6919/ : FAILURE in 2h 02m 19s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/01673ac/ : TIMED_OUT in 3h 25m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-containers-multinode-rocky/75df63c/ : SUCCESS in 2h 29m 54s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-containers-multinode-queens/9e24ee8/ : SUCCESS in 2h 17m 18s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-containers-multinode-pike/2e33d46/ : SUCCESS in 2h 23m 37s\n- openstack-tox-docs http://logs.openstack.org/19/580119/7/check/openstack-tox-docs/90d4212/html/ : SUCCESS in 4m 19s\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/19/580119/7/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/c5b7fa0/ : FAILURE in 1h 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"57c816609e8d3a7bc4b68a30e0a2e669da4464cb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-03 14:16:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c57e71427e307073b59f9814b454aa8bbe9f329d","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-09-03 15:37:49.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5099/ : FAILURE\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6856/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"986458ab1efdb18fd363de89f20cb9f91d660f76","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-03 17:00:07.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/c2daf65/ : FAILURE in 20m 01s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/c1c6cb9/ : FAILURE in 1h 42m 04s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/b7d403d/ : FAILURE in 22m 12s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/f5472ba/ : SUCCESS in 2h 39m 43s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/8b98511/ : SUCCESS in 2h 30m 40s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/af298e2/ : FAILURE in 1h 08m 03s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/5630982/ : SUCCESS in 1h 29m 52s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/8/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/6cc08e1/ : SUCCESS in 2h 13m 46s","accounts_in_message":[],"_revision_number":8},{"id":"50fa77a094468a282b4828c1118f861eef523e76","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 17:38:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/8/check/build-openstack-sphinx-docs/bd46d4b/html/ : SUCCESS in 4m 42s\n- openstack-tox-linters http://logs.openstack.org/19/580119/8/check/openstack-tox-linters/54be5bd/ : SUCCESS in 4m 24s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-undercloud-oooq/0047a71/ : SUCCESS in 1h 22m 58s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-undercloud-upgrades/3b97f4b/ : SUCCESS in 1h 08m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-undercloud-containers/1d8f758/ : SUCCESS in 1h 39m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/513d847/ : SUCCESS in 1h 41m 00s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-containers-multinode/3bc15fa/ : SUCCESS in 2h 22m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/11f9567/ : FAILURE in 2h 08m 57s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/705a1c3/ : TIMED_OUT in 3h 16m 22s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-containers-multinode-rocky/e840e55/ : SUCCESS in 2h 59m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-containers-multinode-queens/e0359c3/ : SUCCESS in 2h 16m 19s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-containers-multinode-pike/5ff6343/ : SUCCESS in 2h 17m 43s\n- openstack-tox-docs http://logs.openstack.org/19/580119/8/check/openstack-tox-docs/38e851c/html/ : SUCCESS in 4m 23s\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/19/580119/8/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/4e76994/ : FAILURE in 1h 47m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f886dd295ae08c4c4e01e58743ff690cd73c950b","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-03 20:15:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b02c9a1667ab8e129fb61129288884f43636ca9d","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-09-03 21:27:12.000000000","message":"Patch Set 9:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6861/ : SUCCESS\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5104/ : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"993c412b0087346270a30a98235802f7ea9e2114","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-03 22:53:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/dd922f5/ : FAILURE in 22m 16s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/764648c/ : FAILURE in 1h 52m 02s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/eaf8278/ : FAILURE in 20m 15s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/a5943d6/ : SUCCESS in 2h 32m 39s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/d71a8ab/ : SUCCESS in 2h 31m 46s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/4f738c1/ : SUCCESS in 1h 46m 58s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/afbb93d/ : SUCCESS in 1h 35m 17s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/9/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/36cbe0c/ : SUCCESS in 2h 15m 11s","accounts_in_message":[],"_revision_number":9},{"id":"47cc8a41f90835ff33a23720910575e6bd3f6bd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-03 23:30:48.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/9/check/build-openstack-sphinx-docs/f00c1c8/html/ : SUCCESS in 4m 05s\n- openstack-tox-linters http://logs.openstack.org/19/580119/9/check/openstack-tox-linters/0a58bda/ : SUCCESS in 4m 56s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-undercloud-oooq/85263d1/ : SUCCESS in 1h 01m 06s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-undercloud-upgrades/bc65927/ : SUCCESS in 1h 12m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-undercloud-containers/c111cdc/ : SUCCESS in 1h 56m 03s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/bbf3dfa/ : SUCCESS in 1h 35m 00s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-containers-multinode/79049d4/ : SUCCESS in 2h 31m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c7fee18/ : FAILURE in 1h 58m 57s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5a7c28c/ : FAILURE in 3h 10m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-containers-multinode-rocky/1333f6c/ : SUCCESS in 2h 51m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-containers-multinode-queens/cf91369/ : SUCCESS in 2h 15m 46s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-containers-multinode-pike/f332dd9/ : FAILURE in 17m 57s\n- openstack-tox-docs http://logs.openstack.org/19/580119/9/check/openstack-tox-docs/f45864c/html/ : SUCCESS in 3m 45s\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/19/580119/9/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/863ae42/ : SUCCESS in 2h 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"4d190a946e179ba8f3cb783a73a75baa2e133f80","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-09-03 23:30:59.000000000","message":"Patch Set 9:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- tripleo-ci-centos-7-containers-multinode-pike: https://bugs.launchpad.net/bugs/1282876\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":9},{"id":"6acb1ae54fec2dbb33481d2e967d1e1e7e864d8e","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-09-04 09:08:44.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"093be5d00b1045b0d52715b9b32dea6649012334","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-04 10:07:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5ed76b0a05043d03acba1d4b32432c417e786a96","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-09-04 11:43:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6eccec7bfa151ce7be4846859b6975506a8cff90","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-04 11:58:27.000000000","message":"Patch Set 9:\n\n(1 comment)\n\nI\u0027ll remove the duplicated deployed_server in next PS.","accounts_in_message":[],"_revision_number":9},{"id":"86ef2572e92a4fe387ae0e9f5fd38cf5a590535a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-04 11:59:03.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c3f74d4e4af2c8135e8338660a5776d4c272dca9","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-09-04 13:29:50.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5112/ : FAILURE\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6869/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"c398895bed3dd34d0cb1505cf627280274436485","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-04 15:32:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/10/check/build-openstack-sphinx-docs/a9aa867/html/ : SUCCESS in 5m 04s\n- openstack-tox-linters http://logs.openstack.org/19/580119/10/check/openstack-tox-linters/adb0551/ : SUCCESS in 6m 19s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-undercloud-oooq/cf11922/ : SUCCESS in 1h 02m 13s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-undercloud-upgrades/2e14bd7/ : SUCCESS in 1h 07m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-undercloud-containers/359bb8b/ : SUCCESS in 1h 57m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5900f8f/ : SUCCESS in 1h 47m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-containers-multinode/33de461/ : SUCCESS in 2h 21m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ab1faf6/ : FAILURE in 2h 16m 37s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d42ac2f/ : TIMED_OUT in 3h 14m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-containers-multinode-rocky/81e69d4/ : SUCCESS in 2h 41m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-containers-multinode-queens/fb3ac27/ : SUCCESS in 2h 07m 13s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-containers-multinode-pike/70f080b/ : SUCCESS in 2h 20m 50s\n- openstack-tox-docs http://logs.openstack.org/19/580119/10/check/openstack-tox-docs/b513749/html/ : SUCCESS in 5m 34s\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/19/580119/10/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/e42bb39/ : TIMED_OUT in 3h 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f24687c8f4690f6fd3ac3d46ded0ddbd58d1e343","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-04 15:46:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/b7c6f81/ : FAILURE in 22m 14s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/3379bfa/ : FAILURE in 1h 41m 01s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/ee2cb7d/ : POST_FAILURE in 21m 49s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/f8aa76f/ : FAILURE in 2h 28m 09s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/9a1dd15/ : FAILURE in 1h 52m 39s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/d8d9219/ : FAILURE in 2h 07m 48s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/b175ca7/ : SUCCESS in 1h 41m 54s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/10/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b8c5808/ : FAILURE in 1h 37m 02s","accounts_in_message":[],"_revision_number":10},{"id":"4363cde861d52539d3f01010294a6066fef675bb","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-04 16:00:33.000000000","message":"Uploaded patch set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"4e1280b66137bb0726e0ad23da0d75ea99eb0ada","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-09-04 17:58:01.000000000","message":"Patch Set 11:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/6878/ : SUCCESS\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5121/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"46b0573654426e2f1d9f938150278521d16301a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-04 19:55:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/11/check/build-openstack-sphinx-docs/3ff38f6/html/ : SUCCESS in 5m 27s\n- openstack-tox-linters http://logs.openstack.org/19/580119/11/check/openstack-tox-linters/1345603/ : SUCCESS in 5m 42s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-undercloud-oooq/782159e/ : SUCCESS in 1h 07m 36s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-undercloud-upgrades/0364c28/ : SUCCESS in 1h 07m 53s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-undercloud-containers/2ec6b5c/ : SUCCESS in 1h 51m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5f36008/ : SUCCESS in 1h 42m 35s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode/84db1c5/ : SUCCESS in 2h 40m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2efffc0/ : FAILURE in 1h 45m 58s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6e1049b/ : TIMED_OUT in 3h 15m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode-rocky/f735d55/ : SUCCESS in 2h 33m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode-queens/2e775a8/ : SUCCESS in 2h 21m 15s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode-pike/625586b/ : SUCCESS in 2h 27m 03s\n- openstack-tox-docs http://logs.openstack.org/19/580119/11/check/openstack-tox-docs/71d4c2f/html/ : SUCCESS in 5m 24s\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/712830d/ : SUCCESS in 2h 13m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5441122a3768918ba7420296a23230710e347b69","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-04 20:55:22.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/6394968/ : FAILURE in 20m 30s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/ecc522e/ : FAILURE in 1h 49m 13s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/af80fa5/ : FAILURE in 18m 47s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/1275625/ : FAILURE in 2h 52m 11s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/dcfecf3/ : FAILURE in 3h 11m 08s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/50b93a5/ : SUCCESS in 1h 50m 52s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/57e3bbe/ : SUCCESS in 1h 44m 13s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/41339fe/ : SUCCESS in 2h 08m 14s","accounts_in_message":[],"_revision_number":11},{"id":"d544f142284222cb67cf1c8771d06155aa264520","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-05 10:12:35.000000000","message":"Patch Set 11:\n\nscenario010 seems to be deploying Octavia just fine. Octavia logs in controller node show no errors.\n\nhttp://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/712830d/logs/subnode-2/var/log/containers/octavia/api.log.txt.gz\n\nhttp://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/712830d/logs/subnode-2/var/log/containers/octavia/health-manager.log.txt.gz\n\nhttp://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/712830d/logs/subnode-2/var/log/containers/octavia/housekeeping.log.txt.gz\n\nhttp://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/712830d/logs/subnode-2/var/log/containers/octavia/worker.log.txt.gz","accounts_in_message":[],"_revision_number":11},{"id":"72bf56ab41a860fdb41325fcc6897e8f3c49429f","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2018-09-06 16:24:25.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"7bc56240deb6ca832e7d51503b949a9aa2720a56","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-09-07 10:04:39.000000000","message":"Patch Set 11:\n\nIs tempest disabled purposely?","accounts_in_message":[],"_revision_number":11},{"id":"19bac7aaa1a586d59818b21c2b041a913c05f9b7","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-07 10:07:34.000000000","message":"Patch Set 11:\n\nYes, I left a note there:\n\n\"# TODO(cgoncalves): enable tempest run once https://review.openstack.org/#/c/591997/ is merged\"\n\nEnablement of tempest is proposed here: https://review.openstack.org/#/c/599178/","accounts_in_message":[],"_revision_number":11},{"id":"9ef547bc85144e63a614263de7f3eb5704be8896","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2018-09-07 10:08:52.000000000","message":"Patch Set 11: Code-Review+1\n\nSorry, my bad.","accounts_in_message":[],"_revision_number":11},{"id":"30359ccc49bbe3a378a5036c15ea775818226390","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-12 04:36:02.000000000","message":"Patch Set 11:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":11},{"id":"7e3fea172d8883a93bb583bd6fac53f0658672b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-12 14:29:28.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/0a03705/ : FAILURE in 24m 27s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/3301601/ : FAILURE in 1h 59m 16s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/c81cff5/ : FAILURE in 22m 44s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/7844950/ : FAILURE in 30m 16s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/d75cd3b/ : FAILURE in 30m 16s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/43b2691/ : FAILURE in 28m 15s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/87ff217/ : FAILURE in 28m 23s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bf93366/ : FAILURE in 32m 56s","accounts_in_message":[],"_revision_number":11},{"id":"46cbf95f8612fc4f487f4d1d778a0565d5f2cdda","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2018-09-14 16:08:57.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"5902f1f1bbf5554235e168bd2d187a6db537a624","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2018-09-14 16:18:45.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"21c8b19b583522a4d15070a049516b6d0be756b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-14 16:19:07.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"e89b5e526b415a19e2f19b8cffbc9465258a2b0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 11:15:25.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"0a21c9394f7c65c24186ecacc813f14ce7e65a2a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2018-09-15 16:02:11.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"ad55024adece9ec09678c296fb64431a8eafd860","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2018-09-15 17:14:17.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/7037/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/5278/ : SUCCESS","accounts_in_message":[],"_revision_number":11},{"id":"2663b91eccb83540eabe02ed71bc4f8d28cf31bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-15 18:49:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-containers-multinode-upgrades-pike-branch/abd4801/ : FAILURE in 18m 54s (non-voting)\n- legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-multinode-1ctlr-featureset037-updates-master/9ca1473/ : SUCCESS in 1h 58m 03s (non-voting)\n- legacy-tripleo-ci-centos-7-container-to-container-upgrades-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-container-to-container-upgrades-master/853eeee/ : FAILURE in 29m 01s (non-voting)\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/94ebdeb/ : SUCCESS in 2h 26m 34s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/7231f0e/ : SUCCESS in 2h 34m 21s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/439af1c/ : SUCCESS in 1h 52m 07s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata-branch/ac5c1fd/ : SUCCESS in 1h 39m 21s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/19/580119/11/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/4402428/ : SUCCESS in 2h 14m 53s","accounts_in_message":[],"_revision_number":11},{"id":"0bb9bd3ed4f5b8c80396187505a0653086df0fa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 19:11:14.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/11/check/build-openstack-sphinx-docs/1623473/html/ : SUCCESS in 4m 54s\n- openstack-tox-linters http://logs.openstack.org/19/580119/11/check/openstack-tox-linters/bdf4dd5/ : SUCCESS in 4m 42s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-undercloud-upgrades/b2c21ea/ : SUCCESS in 1h 05m 52s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-undercloud-containers/c9a966a/ : SUCCESS in 1h 50m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/8c7df07/ : SUCCESS in 1h 45m 34s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode/eaa5d8e/ : SUCCESS in 3h 04m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/70827da/ : SUCCESS in 2h 27m 40s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3753d5f/ : FAILURE in 1h 39m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode-rocky/f3a1cc7/ : SUCCESS in 2h 52m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode-queens/2c9b665/ : SUCCESS in 2h 30m 40s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-containers-multinode-pike/f2a23d3/ : SUCCESS in 2h 14m 49s\n- openstack-tox-docs http://logs.openstack.org/19/580119/11/check/openstack-tox-docs/d6c16a4/html/ : SUCCESS in 4m 21s\n- tripleo-ci-centos-7-scenario010-multinode-oooq-container http://logs.openstack.org/19/580119/11/check/tripleo-ci-centos-7-scenario010-multinode-oooq-container/06b70ed/ : SUCCESS in 2h 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f8013d7fd674b9f308e6c3736038049396db1ea2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-15 19:11:31.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"963368756c09eff2b682d76e0fa7d2badbe053de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-16 01:09:12.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/19/580119/11/gate/build-openstack-sphinx-docs/b1a0d78/html/ : SUCCESS in 4m 25s\n- openstack-tox-linters http://logs.openstack.org/19/580119/11/gate/openstack-tox-linters/0545f61/ : SUCCESS in 4m 48s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/580119/11/gate/tripleo-ci-centos-7-undercloud-containers/919b5ee/ : SUCCESS in 2h 13m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/19/580119/11/gate/tripleo-ci-centos-7-containerized-undercloud-upgrades/eb67505/ : SUCCESS in 1h 48m 39s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/580119/11/gate/tripleo-ci-centos-7-containers-multinode/af6ca1b/ : SUCCESS in 2h 55m 25s\n- openstack-tox-docs http://logs.openstack.org/19/580119/11/gate/openstack-tox-docs/be1dfcd/html/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":11},{"id":"55efaece46b20c3f2c2c8c9c62b060330ae0948d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-16 01:09:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d64a2585daea72c248e1c07f5c76b4c8c51e2799","revisions":{"e1a6880281312407632a05e0c399c2a52f17f58c":{"kind":"REWORK","_number":1,"created":"2018-07-04 09:55:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/1"}}},"commit":{"parents":[{"commit":"3e985e45b046f3f3a685ff731512c55c23566600","subject":"Add fs038 for Octavia scenario testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3e985e45b046f3f3a685ff731512c55c23566600"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"subject":"WIP: featureset038: run Octavia tempest tests","message":"WIP: featureset038: run Octavia tempest tests\n\nThis patch also removes tempest_config (obsolete).\n\nTODO:\n  - check tempest whitelist: too many? are api tests enough or also\n  (part of) scenario tests?\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e1a6880281312407632a05e0c399c2a52f17f58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e1a6880281312407632a05e0c399c2a52f17f58c"}]},"branch":"refs/heads/master"},"5ae3f3a1580400f73f96622c4889e4828fa20921":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2018-07-17 22:54:00.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/2"}}},"commit":{"parents":[{"commit":"7d80f7e76bdafe0e0e876d50de369ad151040305","subject":"Merge \"update minimal to keep it up to date with containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7d80f7e76bdafe0e0e876d50de369ad151040305"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-17 22:54:00.000000000","tz":0},"subject":"WIP: featureset038: run Octavia tempest tests","message":"WIP: featureset038: run Octavia tempest tests\n\nThis patch also removes tempest_config (obsolete).\n\nTODO:\n  - check tempest whitelist: too many? are api tests enough or also\n  (part of) scenario tests?\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5ae3f3a1580400f73f96622c4889e4828fa20921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5ae3f3a1580400f73f96622c4889e4828fa20921"}]},"branch":"refs/heads/master"},"ddedbe536088154d470cc3d23ed1b698231f4643":{"kind":"REWORK","_number":3,"created":"2018-08-15 10:39:41.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/3"}}},"commit":{"parents":[{"commit":"aca17029237b81b0d34ef5e682dd97a52b312d48","subject":"fix if statement for standalone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/aca17029237b81b0d34ef5e682dd97a52b312d48"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-08-15 10:38:30.000000000","tz":120},"subject":"featureset038: run Octavia tempest tests","message":"featureset038: run Octavia tempest tests\n\nThis patch also removes tempest_config (obsolete).\n\nDepends-On: https://review.openstack.org/#/c/591997/\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ddedbe536088154d470cc3d23ed1b698231f4643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ddedbe536088154d470cc3d23ed1b698231f4643"}]},"branch":"refs/heads/master"},"18297a43e2de0ff793a6a00186f23f080ff5d392":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-08-20 14:48:09.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/4"}}},"commit":{"parents":[{"commit":"c231ab842a37cb9f75fd78bf7f4a7d6355329e9d","subject":"Merge \"fix full-deploy standalone logic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c231ab842a37cb9f75fd78bf7f4a7d6355329e9d"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-08-20 14:47:10.000000000","tz":120},"subject":"featureset038: run Octavia tempest tests","message":"featureset038: run Octavia tempest tests\n\nThis patch also removes tempest_config (obsolete).\n\nDepends-On: https://review.openstack.org/#/c/591997/\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/18297a43e2de0ff793a6a00186f23f080ff5d392"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/18297a43e2de0ff793a6a00186f23f080ff5d392"}]},"branch":"refs/heads/master"},"28231355359fd0c7d7abb520c74514a0a60c73e2":{"kind":"REWORK","_number":5,"created":"2018-08-30 18:29:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/5"}}},"commit":{"parents":[{"commit":"7cc62968410f5b55d7a95707c71a641c72036a08","subject":"[FS021] Switch to containerized undercloud net config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7cc62968410f5b55d7a95707c71a641c72036a08"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-08-30 18:29:22.000000000","tz":120},"subject":"featureset038: run Octavia tempest tests","message":"featureset038: run Octavia tempest tests\n\nThis patch also removes tempest_config (obsolete).\n\nDepends-On: https://review.openstack.org/#/c/591997/\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/28231355359fd0c7d7abb520c74514a0a60c73e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/28231355359fd0c7d7abb520c74514a0a60c73e2"}]},"branch":"refs/heads/master"},"cfd31ca7073af632a48855b73feebe73ff2ff24d":{"kind":"REWORK","_number":6,"created":"2018-08-30 19:27:25.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/6"}}},"commit":{"parents":[{"commit":"7cc62968410f5b55d7a95707c71a641c72036a08","subject":"[FS021] Switch to containerized undercloud net config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7cc62968410f5b55d7a95707c71a641c72036a08"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-08-30 19:26:43.000000000","tz":120},"subject":"featureset038: run Octavia tempest tests","message":"featureset038: run Octavia tempest tests\n\nThis patch also removes tempest_config (obsolete).\n\nDepends-On: https://review.openstack.org/#/c/591997/\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cfd31ca7073af632a48855b73feebe73ff2ff24d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cfd31ca7073af632a48855b73feebe73ff2ff24d"}]},"branch":"refs/heads/master"},"280e767326b0859daca342ca0ad8c25623e0fe6f":{"kind":"REWORK","_number":7,"created":"2018-09-01 11:14:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/7"}}},"commit":{"parents":[{"commit":"99ab0040ccc858abe2ec5c54f4821bcf52d4683b","subject":"Add scenario010 to the check queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/99ab0040ccc858abe2ec5c54f4821bcf52d4683b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-09-01 11:14:01.000000000","tz":120},"subject":"FS38: switch to containerized UC and config-download","message":"FS38: switch to containerized UC and config-download\n\nThis patch also switches FS038 to containerized undercloud and converts\nthe job that runs against master changes to use config-download.\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/280e767326b0859daca342ca0ad8c25623e0fe6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/280e767326b0859daca342ca0ad8c25623e0fe6f"}]},"branch":"refs/heads/master"},"cf3ac3fbda837ff044a64ab5b84e29e8f670fbb7":{"kind":"REWORK","_number":8,"created":"2018-09-03 14:16:40.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/8"}}},"commit":{"parents":[{"commit":"99ab0040ccc858abe2ec5c54f4821bcf52d4683b","subject":"Add scenario010 to the check queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/99ab0040ccc858abe2ec5c54f4821bcf52d4683b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-09-03 14:16:17.000000000","tz":120},"subject":"FS38: switch to containerized UC and config-download","message":"FS38: switch to containerized UC and config-download\n\nThis patch also switches FS038 to containerized undercloud and converts\nthe job that runs against master changes to use config-download.\nAdditionally, switch to containerized undercloud net config.\n\nRelated-Bug: #1749707\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cf3ac3fbda837ff044a64ab5b84e29e8f670fbb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cf3ac3fbda837ff044a64ab5b84e29e8f670fbb7"}]},"branch":"refs/heads/master"},"115d018083d7f3102694ffd11ca06814b6c63f14":{"kind":"REWORK","_number":9,"created":"2018-09-03 20:15:42.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/9"}}},"commit":{"parents":[{"commit":"99ab0040ccc858abe2ec5c54f4821bcf52d4683b","subject":"Add scenario010 to the check queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/99ab0040ccc858abe2ec5c54f4821bcf52d4683b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-09-03 20:14:56.000000000","tz":120},"subject":"FS38: switch to containerized UC and config-download","message":"FS38: switch to containerized UC and config-download\n\nThis patch also switches FS038 to containerized undercloud and converts\nthe job that runs against master changes to use config-download.\nAdditionally, switch to containerized undercloud net config.\n\nRelated-Bug: #1749707\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/115d018083d7f3102694ffd11ca06814b6c63f14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/115d018083d7f3102694ffd11ca06814b6c63f14"}]},"branch":"refs/heads/master"},"97d90f3d07600b1135bbea1573a4726f8b56532b":{"kind":"REWORK","_number":10,"created":"2018-09-04 11:59:03.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/10"}}},"commit":{"parents":[{"commit":"99ab0040ccc858abe2ec5c54f4821bcf52d4683b","subject":"Add scenario010 to the check queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/99ab0040ccc858abe2ec5c54f4821bcf52d4683b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-09-04 11:53:36.000000000","tz":120},"subject":"FS38: switch to containerized UC and config-download","message":"FS38: switch to containerized UC and config-download\n\nThis patch also switches FS038 to containerized undercloud and converts\nthe job that runs against master changes to use config-download.\nAdditionally, switch to containerized undercloud net config.\n\nRelated-Bug: #1749707\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/97d90f3d07600b1135bbea1573a4726f8b56532b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/97d90f3d07600b1135bbea1573a4726f8b56532b"}]},"branch":"refs/heads/master"},"d64a2585daea72c248e1c07f5c76b4c8c51e2799":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2018-09-04 16:00:33.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/19/580119/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/580119/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/580119/11"}}},"commit":{"parents":[{"commit":"99ab0040ccc858abe2ec5c54f4821bcf52d4683b","subject":"Add scenario010 to the check queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/99ab0040ccc858abe2ec5c54f4821bcf52d4683b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-07-04 09:46:37.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2018-09-04 16:00:02.000000000","tz":120},"subject":"FS38: switch to containerized UC and config-download","message":"FS38: switch to containerized UC and config-download\n\nThis patch switches FS038 to containerized undercloud and converts the\njob that runs against master changes to use config-download.\nAdditionally, switch to containerized undercloud net config.\n\nRelated-Bug: #1749707\n\nChange-Id: Iff3b8b66deb901993f9e8fe0f2a896f1da2749f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d64a2585daea72c248e1c07f5c76b4c8c51e2799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d64a2585daea72c248e1c07f5c76b4c8c51e2799"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
