)]}'
{"id":"openstack%2Ftripleo-quickstart~639324","triplet_id":"openstack%2Ftripleo-quickstart~master~I93949966d1c8255bff9c8bff18b6f584a20f77cf","project":"openstack/tripleo-quickstart","branch":"master","topic":"ipatest","hashtags":[],"change_id":"I93949966d1c8255bff9c8bff18b6f584a20f77cf","subject":"Run novajoin tempest tests in all TLS job using os_tempest","status":"ABANDONED","created":"2019-02-26 12:32:56.000000000","updated":"2020-04-30 09:27:43.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"47bc1f759485d8e727865bfb1f2ac9e5804faa19","_number":639324,"virtual_id_number":639324,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-11-15 06:54:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-15 07:52:23.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-15 11:31:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-04-17 14:47:59.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-18 09:41:09.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2019-05-17 01:46:11.000000000","updated_by":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"reviewer":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"state":"REVIEWER"},{"updated":"2019-06-12 12:02:28.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-09 10:21:28.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-09-06 14:07:43.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-09-30 17:32:38.000000000","updated_by":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"reviewer":{"_account_id":8866,"name":"Raildo Mascena de Sousa Filho","email":"rmascena@redhat.com","username":"raildo"},"state":"REVIEWER"},{"updated":"2019-11-15 06:54:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-15 07:52:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4117ff19ebadd75bbfbbbca8c0d1fca81cb47216","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-26 12:32:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2627a9e121b22fc2f4a1c7e3cec51bc496461fd","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 15:01:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8532/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6777/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"a0465186527dbda19c7cd1852485a016c04a7228","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 15:36:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-undercloud-containers/22330ea/ : SUCCESS in 1h 36m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0e19549/ : SUCCESS in 1h 52m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-containers-multinode/251c94e/ : SUCCESS in 2h 07m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/38219c9/ : SUCCESS in 2h 07m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9c4cf8b/ : SUCCESS in 2h 33m 43s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/1/check/tripleo-ci-fedora-28-standalone/2bc40a6/ : SUCCESS in 1h 33m 45s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-standalone-upgrade/d5b4331/ : SUCCESS in 1h 50m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-containers-multinode-rocky/c3089ed/ : SUCCESS in 1h 49m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-containers-multinode-queens/3e8234e/ : SUCCESS in 1h 59m 07s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/1/check/tripleo-ci-centos-7-containers-multinode-pike/b68ce5e/ : SUCCESS in 2h 49m 25s\n- openstack-tox-docs http://logs.openstack.org/24/639324/1/check/openstack-tox-docs/b240460/html/ : SUCCESS in 3m 55s\n- openstack-tox-linters http://logs.openstack.org/24/639324/1/check/openstack-tox-linters/fe6507a/ : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"29bf81d6a40930a4e29c56fcf7f6e921408455cd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-26 15:42:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cbe9230/ : FAILURE in 1h 51m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8fbe84a/ : SUCCESS in 2h 25m 51s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/3a633de/ : SUCCESS in 1h 34m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/15a5d16/ : SUCCESS in 1h 47m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2ef4c82/ : SUCCESS in 2h 16m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4cda909/ : SUCCESS in 2h 38m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/468971e/ : SUCCESS in 2h 52m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2194ff9/ : FAILURE in 2h 59m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7b92681f1485af7c54ec1c0261c82ad9920a90ed","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-26 16:20:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5f17c7a2def0867e320947475a2c72657470b80e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-26 18:47:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-undercloud-containers/31ff89d/ : SUCCESS in 1h 48m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/be4bfe1/ : SUCCESS in 2h 19m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-containers-multinode/0ee3651/ : SUCCESS in 2h 06m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9c127d8/ : SUCCESS in 2h 06m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/031c4e0/ : SUCCESS in 2h 16m 59s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/2/check/tripleo-ci-fedora-28-standalone/fab7427/ : SUCCESS in 1h 35m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-standalone-upgrade/657db23/ : SUCCESS in 1h 54m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-containers-multinode-rocky/3e66d21/ : SUCCESS in 2h 13m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-containers-multinode-queens/daac0be/ : SUCCESS in 2h 06m 11s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/2/check/tripleo-ci-centos-7-containers-multinode-pike/f338d0c/ : SUCCESS in 2h 08m 55s\n- openstack-tox-docs http://logs.openstack.org/24/639324/2/check/openstack-tox-docs/cb7a81a/html/ : SUCCESS in 3m 46s\n- openstack-tox-linters http://logs.openstack.org/24/639324/2/check/openstack-tox-linters/e33cfba/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"dc96e2c3562663cbb94b7f9e55c9622b79b1762a","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-26 19:01:36.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8533/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6778/ : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"3d116e2896e24514eb6b8a04c214d4af005ff75d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-26 19:31:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/797ddac/ : SUCCESS in 2h 30m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/99d7d0f/ : SUCCESS in 2h 31m 08s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/00592d2/ : SUCCESS in 1h 38m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/b4f5dca/ : SUCCESS in 1h 52m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/11ca584/ : FAILURE in 1h 08m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5e4ebc7/ : FAILURE in 1h 27m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c782241/ : FAILURE in 42m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ecd2dc2/ : FAILURE in 2h 49m 42s","accounts_in_message":[],"_revision_number":2},{"id":"edefbb3c44a9935b6e37ba0949b8680b2266efaa","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-02-27 03:39:52.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nThanks, just few nit picks!","accounts_in_message":[],"_revision_number":1},{"id":"e38c8f9e990dbd45385e6ffda2e41a73529720c7","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-27 12:09:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31349341be5bcfdb936f1590f5f2927ddb65c22b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-27 12:47:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f16d7759bdd86821c7c3d34b2f5c8c34d736d9b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 15:43:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-undercloud-containers/97ad2e9/ : FAILURE in 40m 05s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/761f25a/ : SUCCESS in 2h 02m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode/bf7b896/ : SUCCESS in 2h 19m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9102e73/ : SUCCESS in 2h 26m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2250c6b/ : SUCCESS in 2h 43m 24s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/3/check/tripleo-ci-fedora-28-standalone/29723e1/ : SUCCESS in 2h 06m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-standalone-upgrade/2bf0bad/ : SUCCESS in 2h 11m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-rocky/aa8d6d9/ : SUCCESS in 2h 06m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-queens/c494fe9/ : SUCCESS in 2h 08m 45s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-pike/8e3394a/ : SUCCESS in 2h 27m 37s\n- openstack-tox-docs http://logs.openstack.org/24/639324/3/check/openstack-tox-docs/d061dce/html/ : SUCCESS in 4m 43s\n- openstack-tox-linters http://logs.openstack.org/24/639324/3/check/openstack-tox-linters/a0c3c3a/ : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":3},{"id":"0fbef41fc7fa670913ec343f75360940ac180fbd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-27 15:47:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/53aedcd/ : SUCCESS in 2h 35m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/614a9ea/ : SUCCESS in 2h 27m 05s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b79ec3a/ : SUCCESS in 1h 37m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/dad9965/ : FAILURE in 1h 19m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3aaef61/ : SUCCESS in 2h 17m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/325cf14/ : SUCCESS in 2h 34m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/3ed0ba6/ : RETRY_LIMIT in 9m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/965a9ab/ : FAILURE in 2h 41m 07s","accounts_in_message":[],"_revision_number":3},{"id":"3d28c9df941cd2e51da619609358ca228dd23cb2","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 16:49:05.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8534/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6779/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"2c5d940336c0a4849184d8e04b5c228dee750498","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-27 17:24:54.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"01e7e73d45f2b033f1d991a27cd439b956b40513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 17:26:53.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c385ad4fe553543618f033080373043fd3e38a5e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-27 17:26:57.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"5d8791a8eccde3d4ecc24e1dfe35f1a63cde1d4f","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-27 17:28:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ac4a9fdf3f17032d463237573de75b91d5ca8ab0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-27 19:53:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/70e4418/ : SUCCESS in 2h 22m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f93df10/ : RETRY_LIMIT in 11m 14s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/683fda9/ : SUCCESS in 1h 35m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/9c1393e/ : RETRY_LIMIT in 9m 58s","accounts_in_message":[],"_revision_number":3},{"id":"93f7703b65491b71f20e957e5ad448ccc0bbfd10","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-27 20:04:14.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8535/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6780/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"5bfd51e81b5f81c76d158b9c77c5b4419f6dc222","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-27 21:27:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-undercloud-containers/fb5c761/ : SUCCESS in 1h 32m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/4888d88/ : SUCCESS in 1h 52m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode/16b0b71/ : SUCCESS in 1h 59m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/19ac2c0/ : SUCCESS in 2h 48m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7d8cb75/ : SUCCESS in 2h 15m 13s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/3/check/tripleo-ci-fedora-28-standalone/a03a87b/ : FAILURE in 1h 27m 01s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-standalone-upgrade/f453269/ : FAILURE in 2h 53m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-rocky/01c751b/ : SUCCESS in 1h 54m 47s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-queens/7c6b96a/ : SUCCESS in 2h 01m 47s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-pike/6b4d25c/ : SUCCESS in 2h 04m 07s\n- openstack-tox-docs http://logs.openstack.org/24/639324/3/check/openstack-tox-docs/00b0d98/html/ : SUCCESS in 6m 27s\n- openstack-tox-linters http://logs.openstack.org/24/639324/3/check/openstack-tox-linters/97b8261/ : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":3},{"id":"2e8e9ba0ffaddb0bca1ff21a628c2a0f610edf19","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-27 22:45:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"16f7c47b76dd07eecfcdf8fb385f4b669660c766","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 01:24:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-undercloud-containers/cb4f5ac/ : SUCCESS in 1h 17m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5c4347f/ : SUCCESS in 1h 54m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode/db2d12e/ : SUCCESS in 2h 02m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/606dad1/ : SUCCESS in 2h 14m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/017ffa4/ : SUCCESS in 2h 30m 05s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/3/check/tripleo-ci-fedora-28-standalone/f022800/ : FAILURE in 1h 38m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-standalone-upgrade/aa55988/ : FAILURE in 2h 03m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-rocky/51801b9/ : SUCCESS in 2h 12m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-queens/54d01e8/ : SUCCESS in 1h 46m 28s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-pike/999e6cd/ : SUCCESS in 2h 07m 57s\n- openstack-tox-docs http://logs.openstack.org/24/639324/3/check/openstack-tox-docs/8c58133/html/ : SUCCESS in 5m 49s\n- openstack-tox-linters http://logs.openstack.org/24/639324/3/check/openstack-tox-linters/6f23a0d/ : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":3},{"id":"e882721584cadc051e56f42d0655fc816b873f2c","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-28 01:29:27.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8537/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6782/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"bf355f5bd77c5c5802dd0da603c2b487f959b002","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-28 02:01:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d62bf26/ : SUCCESS in 2h 30m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f9a1abc/ : TIMED_OUT in 3h 14m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ea35392/ : SUCCESS in 1h 34m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/7bf311e/ : SUCCESS in 1h 50m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/45d6332/ : FAILURE in 2h 58m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/657f6eb/ : SUCCESS in 2h 35m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/4de9a04/ : SUCCESS in 2h 36m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e1c5520/ : FAILURE in 3h 04m 20s","accounts_in_message":[],"_revision_number":3},{"id":"ff65487a1eeaf1598bd013cb59b46688b170f46e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-02-28 05:17:27.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3ac5961c11985024d7ff14b987f3cf1a1bce0a5b","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-28 05:53:26.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8538/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6783/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"4579711e9669d1933766236b72a0a48b18d7bb9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 07:54:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-undercloud-containers/aec1209/ : SUCCESS in 1h 24m 02s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9f7edfb/ : SUCCESS in 2h 00m 22s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode/4a43a47/ : SUCCESS in 1h 57m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a71bbd0/ : SUCCESS in 1h 56m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3504757/ : SUCCESS in 2h 22m 20s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/3/check/tripleo-ci-fedora-28-standalone/bd96de9/ : FAILURE in 1h 23m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-standalone-upgrade/d51748a/ : FAILURE in 1h 53m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-rocky/234fd83/ : SUCCESS in 2h 04m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-queens/43c9039/ : SUCCESS in 1h 52m 54s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/3/check/tripleo-ci-centos-7-containers-multinode-pike/47915bd/ : SUCCESS in 2h 13m 04s\n- openstack-tox-docs http://logs.openstack.org/24/639324/3/check/openstack-tox-docs/3d8ad40/html/ : SUCCESS in 4m 24s\n- openstack-tox-linters http://logs.openstack.org/24/639324/3/check/openstack-tox-linters/28af0e2/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":3},{"id":"cf012d09abd004c4707055a1773ae4052bcb12b2","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-02-28 08:18:49.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b751e3954a2148b88e470886a0a3e38888c61e43","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-28 08:35:15.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c658ad0/ : SUCCESS in 2h 38m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3c9b30b/ : TIMED_OUT in 3h 15m 08s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/68ba8ae/ : SUCCESS in 1h 35m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/457e289/ : SUCCESS in 1h 49m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a30ec01/ : SUCCESS in 2h 08m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/19ad668/ : SUCCESS in 2h 30m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/3eba9ea/ : SUCCESS in 2h 38m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/070d6e0/ : SUCCESS in 2h 43m 46s","accounts_in_message":[],"_revision_number":3},{"id":"b42c1b502120bdbd3d6d6a06bce0d19e1ed32862","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 10:05:14.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"afc0424e4877eb08477a999e8519ae2bbddd91d7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-02-28 10:17:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c683e7356669d0391c415db534c59de5fff11c0a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-02-28 10:18:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"eacd24f66011a9ff7a44703ab49d681e2acc49e9","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-28 10:18:17.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6788/ : FAILURE\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8543/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"64d13d169a7b2295a30327045fb2e8c1c7513820","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-28 12:56:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-undercloud-containers/03ee03d/ : SUCCESS in 1h 28m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d94dcec/ : SUCCESS in 1h 59m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-containers-multinode/defb8e7/ : FAILURE in 46m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/097afc0/ : SUCCESS in 1h 53m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/53f17a1/ : SUCCESS in 2h 13m 29s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/4/check/tripleo-ci-fedora-28-standalone/c3c3089/ : FAILURE in 1h 12m 43s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-standalone-upgrade/bf95b7a/ : FAILURE in 1h 57m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-containers-multinode-rocky/43c70a3/ : SUCCESS in 2h 28m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-containers-multinode-queens/2224e37/ : SUCCESS in 2h 11m 39s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/4/check/tripleo-ci-centos-7-containers-multinode-pike/deb52fc/ : SUCCESS in 2h 10m 15s\n- openstack-tox-docs http://logs.openstack.org/24/639324/4/check/openstack-tox-docs/673574d/html/ : SUCCESS in 4m 53s\n- openstack-tox-linters http://logs.openstack.org/24/639324/4/check/openstack-tox-linters/22cb1e6/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":4},{"id":"53c30ba3ec89d52e089e0f461351dc71df12b79b","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-02-28 14:18:29.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8548/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6793/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"e6d76091d04cec1d18b7f3949939e4935323fbdb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-02-28 16:44:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/2058d2c/ : SUCCESS in 1h 22m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/50417f5/ : SUCCESS in 2h 36m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"a42b79ee934474460cd0a91bf21952a153ae5a87","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-03-01 05:13:09.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"59c0db55a44a18ebe90bb0eb017c710ddfd5857f","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-01 07:41:36.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8554/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6799/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"fb82b8e197520be6d50c5c6f89495fe28dbda0c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-01 07:48:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-undercloud-containers/dd0fd5e/ : FAILURE in 1h 07m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ba9f94a/ : SUCCESS in 1h 46m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode/784c8db/ : FAILURE in 1h 56m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cae5206/ : SUCCESS in 1h 56m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5f796c4/ : SUCCESS in 2h 12m 06s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/5/check/tripleo-ci-fedora-28-standalone/5ad87d6/ : FAILURE in 1h 14m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-standalone-upgrade/cb2f336/ : FAILURE in 1h 43m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode-rocky/0508218/ : FAILURE in 2h 14m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode-queens/619c24a/ : SUCCESS in 1h 51m 14s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode-pike/f4ffc88/ : SUCCESS in 1h 55m 09s\n- openstack-tox-docs http://logs.openstack.org/24/639324/5/check/openstack-tox-docs/3b88f1d/html/ : SUCCESS in 5m 00s\n- openstack-tox-linters http://logs.openstack.org/24/639324/5/check/openstack-tox-linters/943acd8/ : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":5},{"id":"96ee1a930c68ce0fb4c5aca34ea8eb9a9193c10e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-01 08:04:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9bcdf70/ : FAILURE in 2h 12m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b801e1e/ : FAILURE in 2h 08m 01s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ba473e4/ : FAILURE in 1h 18m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/147d98a/ : SUCCESS in 1h 50m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ceec98f/ : SUCCESS in 2h 08m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/d04087b/ : FAILURE in 2h 17m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/27924c4/ : FAILURE in 2h 19m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/03af729/ : FAILURE in 2h 48m 37s","accounts_in_message":[],"_revision_number":5},{"id":"e909a26420fa94bf3b68441c71ea681f1117a067","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-01 11:09:09.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"69ebff1d0427642122ac0e938f96a6db048dfe9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-01 11:51:21.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"e83ca223f59228fa1603a2bb014b9915d23841a5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-01 11:51:22.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"10cd48bf0aac00e7f052d1edd7a9c31fa8ff11fa","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-03-01 11:51:40.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"d369e1070f015a33d7bd57900cad64e47da0ca7b","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-01 13:48:20.000000000","message":"Patch Set 5:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8558/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6803/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"d69de34c8f0e9f0025c45691f554063b6314c576","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-01 14:14:58.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-undercloud-containers/3e9dfb3/ : SUCCESS in 1h 21m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/45b7c1a/ : SUCCESS in 2h 10m 17s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode/311c329/ : SUCCESS in 1h 45m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a41884d/ : SUCCESS in 1h 57m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/30d0fc3/ : SUCCESS in 2h 04m 39s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/5/check/tripleo-ci-fedora-28-standalone/ff70df7/ : FAILURE in 1h 24m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-standalone-upgrade/622b0bd/ : FAILURE in 1h 46m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode-rocky/e8d3631/ : SUCCESS in 2h 16m 22s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode-queens/015f2be/ : SUCCESS in 1h 46m 10s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/5/check/tripleo-ci-centos-7-containers-multinode-pike/9cc70c1/ : SUCCESS in 1h 55m 34s\n- openstack-tox-docs http://logs.openstack.org/24/639324/5/check/openstack-tox-docs/d15d0a2/html/ : SUCCESS in 3m 49s\n- openstack-tox-linters http://logs.openstack.org/24/639324/5/check/openstack-tox-linters/f9f0c16/ : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":5},{"id":"18a0661d4c42ce9cca7e3be1c418edf77b452c42","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-01 15:00:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/58aeb61/ : RETRY_LIMIT in 8m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/640eb79/ : SUCCESS in 1h 24m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2c5c8cb/ : FAILURE in 2h 55m 30s","accounts_in_message":[],"_revision_number":5},{"id":"3b1e79c9411f8073d47b3aa2472656f053b2f058","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-01 21:51:04.000000000","message":"Patch Set 5:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":5},{"id":"8e66551f77bc3049c2de24672e37c7ea366ddf54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-02 00:26:23.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3c9869b/ : SUCCESS in 2h 16m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2b6404a/ : SUCCESS in 2h 16m 46s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b7beb38/ : SUCCESS in 1h 25m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/82fe246/ : SUCCESS in 1h 48m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/47c4ada/ : SUCCESS in 2h 10m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/d40b0bd/ : SUCCESS in 2h 27m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/51b0911/ : SUCCESS in 2h 21m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/84144e9/ : FAILURE in 2h 32m 02s","accounts_in_message":[],"_revision_number":5},{"id":"35628948ac34c6e84be80158f99fabc4b3205373","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-02 09:27:40.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"cbafc070da818105c8739c1007b3688e64b91119","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-02 10:00:11.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8560/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6805/ : FAILURE","accounts_in_message":[],"_revision_number":6},{"id":"0551631a471ceb6b1faf2d1f20ab433b5b51c580","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-02 12:02:16.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6d85765/ : SUCCESS in 2h 14m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b28aa60/ : SUCCESS in 2h 10m 43s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/87643fe/ : SUCCESS in 1h 16m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/581256e/ : SUCCESS in 1h 45m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/af07581/ : SUCCESS in 2h 08m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c84eacb/ : SUCCESS in 2h 18m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/d58063d/ : SUCCESS in 2h 29m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/fe6c225/ : FAILURE in 1h 58m 02s","accounts_in_message":[],"_revision_number":6},{"id":"4fd7745b80dab2d700a4c1cc0309e9714d43185c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-02 12:06:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-undercloud-containers/4a8704f/ : SUCCESS in 1h 33m 40s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/aadc886/ : SUCCESS in 1h 53m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode/e362576/ : SUCCESS in 2h 08m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a870ecd/ : SUCCESS in 2h 09m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f113f4c/ : SUCCESS in 2h 34m 19s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/6/check/tripleo-ci-fedora-28-standalone/3a09c96/ : FAILURE in 1h 26m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-standalone-upgrade/3652838/ : FAILURE in 1h 49m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode-rocky/340e255/ : SUCCESS in 2h 04m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode-queens/b6a034d/ : SUCCESS in 1h 50m 16s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode-pike/57f391f/ : SUCCESS in 1h 54m 12s\n- openstack-tox-docs http://logs.openstack.org/24/639324/6/check/openstack-tox-docs/5773ed4/html/ : SUCCESS in 5m 05s\n- openstack-tox-linters http://logs.openstack.org/24/639324/6/check/openstack-tox-linters/0520612/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":6},{"id":"6167efb3ce24e37818ff50617c19b71518029a5d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-02 17:19:31.000000000","message":"Patch Set 6:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":6},{"id":"d2ce016d4a6815df846c32132acd8c27f288de9c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-02 19:52:38.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1368521/ : SUCCESS in 2h 18m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e642c61/ : SUCCESS in 2h 13m 01s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/bb9c7ae/ : SUCCESS in 1h 17m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/5e7314c/ : SUCCESS in 1h 50m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e444306/ : SUCCESS in 2h 03m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/400f897/ : SUCCESS in 2h 26m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/a4585ae/ : SUCCESS in 2h 26m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e8ca091/ : FAILURE in 2h 31m 14s","accounts_in_message":[],"_revision_number":6},{"id":"54e4ba71a15ed08b8dea3ff5c13b5501da9b20ce","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-02 23:43:47.000000000","message":"Patch Set 6:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":6},{"id":"1942ee37b3de275b080911d040e2782c536c2e01","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-03 02:24:10.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3213867/ : SUCCESS in 2h 35m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/cce0033/ : SUCCESS in 2h 14m 38s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/e6fe1a4/ : SUCCESS in 1h 31m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/e1f5481/ : SUCCESS in 1h 43m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/61a06a6/ : SUCCESS in 2h 05m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cbef920/ : SUCCESS in 2h 19m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/5bcddb6/ : SUCCESS in 2h 29m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2d80a0d/ : FAILURE in 2h 38m 32s","accounts_in_message":[],"_revision_number":6},{"id":"dac7913bc3e9bc4ff3d9b3d801987c2e22a4b52f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-03-03 06:29:38.000000000","message":"Patch Set 6:\n\n2019-03-03 02:01:33 |     ImportError: No module named ipaclient.remote_plugins\n\nhttps://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2d80a0d/logs/undercloud/home/zuul/tempest.log.txt.gz#_2019-03-03_02_01_33\n\nproblem in ipaclient spec file, needs to be fixed","accounts_in_message":[],"_revision_number":6},{"id":"4be6d136e43dc2986bbc463bab24e9dc3310e5c1","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-03 15:47:39.000000000","message":"Patch Set 6:\n\n@Chandan, who should I turn to for fixing it?","accounts_in_message":[],"_revision_number":6},{"id":"80eb8186db57e1ca9e95ac5649c628109255c810","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-03-04 07:47:35.000000000","message":"Patch Set 6:\n\nI am putting a patch to include ipaclient as a dependency also.","accounts_in_message":[],"_revision_number":6},{"id":"66135cd0899108352b800b7aad43d5a89856a807","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-03-04 08:01:25.000000000","message":"Patch Set 6:\n\nAdded a patch here https://review.rdoproject.org/r/#/c/19074/","accounts_in_message":[],"_revision_number":6},{"id":"6b9cf60367f969c33c586d70de9ec03a4ff1750d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-07 13:05:58.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"173caefe9e7bfa5499450ba29337663e45789580","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-07 15:46:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-undercloud-containers/6bb80bf/ : SUCCESS in 1h 29m 40s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ce152fa/ : SUCCESS in 1h 55m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode/710ad37/ : SUCCESS in 2h 03m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a7f6c79/ : SUCCESS in 2h 19m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/778257d/ : SUCCESS in 2h 34m 16s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/6/check/tripleo-ci-fedora-28-standalone/93fd43b/ : FAILURE in 1h 26m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-standalone-upgrade/6de4930/ : FAILURE in 1h 36m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode-rocky/964dfb7/ : SUCCESS in 2h 01m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode-queens/1fa3847/ : SUCCESS in 1h 52m 50s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/6/check/tripleo-ci-centos-7-containers-multinode-pike/4ce1c31/ : SUCCESS in 2h 26m 22s\n- openstack-tox-docs http://logs.openstack.org/24/639324/6/check/openstack-tox-docs/3dca6b0/html/ : SUCCESS in 5m 50s\n- openstack-tox-linters http://logs.openstack.org/24/639324/6/check/openstack-tox-linters/d57622d/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":6},{"id":"e63c45ccd6eea0847a2f7389bd9992f868792b49","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-07 15:48:39.000000000","message":"Patch Set 6:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8577/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6822/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"15dface3e7e11e7c8b617d9848968c45d42b2f3d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-07 17:39:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/359dd10/ : FAILURE in 2h 54m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3318d5b/ : FAILURE in 1h 37m 03s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/70a9373/ : SUCCESS in 1h 26m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch finger://ze01.softwarefactory-project.io:7979/69392ad416644a4c8025a467b4b97252 : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c48e835/ : SUCCESS in 2h 16m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/135adb9/ : FAILURE in 1h 20m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/22ab08d/ : FAILURE in 1h 34m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/38b8b82/ : FAILURE in 2h 18m 28s","accounts_in_message":[],"_revision_number":6},{"id":"5513a7eafcfb6d0f617aabd46bf0dff1e7a14c8c","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-11 22:45:10.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"d475dda4033c26d35bf37088e12ec1d46cce11b7","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-12 01:25:56.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8585/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6830/ : SUCCESS","accounts_in_message":[],"_revision_number":7},{"id":"9dad4691d3c9452a998666b9fa1c1d6b45e8929d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-12 01:53:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-undercloud-containers/e42eb0f/ : SUCCESS in 1h 49m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/30cc15e/ : SUCCESS in 1h 55m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode/ff7ea7f/ : FAILURE in 58m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/73c03c6/ : SUCCESS in 2h 12m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e0e3bfa/ : FAILURE in 1h 04m 26s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/7/check/tripleo-ci-fedora-28-standalone/75bb0b7/ : FAILURE in 20m 45s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-standalone-upgrade/493da0f/ : FAILURE in 1h 33m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-rocky/c167166/ : SUCCESS in 1h 58m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-queens/018101e/ : SUCCESS in 2h 07m 53s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-pike/e457fe1/ : SUCCESS in 1h 54m 31s\n- openstack-tox-docs http://logs.openstack.org/24/639324/7/check/openstack-tox-docs/2585af6/html/ : SUCCESS in 5m 42s\n- openstack-tox-linters http://logs.openstack.org/24/639324/7/check/openstack-tox-linters/c601ae3/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":7},{"id":"3cf6d7f3ad6fa5b542f11c49b7d0f521f66e37a7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-12 06:15:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a4d2e02/ : TIMED_OUT in 3h 28m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/20b4844/ : FAILURE in 2h 58m 30s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/77d8010/ : SUCCESS in 1h 35m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/7b77ddf/ : SUCCESS in 1h 50m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/cffdc86/ : SUCCESS in 2h 16m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0848638/ : SUCCESS in 2h 29m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/536cbdf/ : SUCCESS in 3h 16m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2c23012/ : TIMED_OUT in 3h 56m 02s","accounts_in_message":[],"_revision_number":7},{"id":"df0898b674627809b0b347afe1f2e2abe62c5a06","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-13 11:57:52.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"e6bafbccadc48f695ba42ce7b2da2fb42d2be928","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-13 12:01:42.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8588/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6833/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"26945b0e3b792d40daaccb174a270a1d00d8b2ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-13 14:49:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-undercloud-containers/9ac538a/ : SUCCESS in 1h 48m 10s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c6f9f48/ : SUCCESS in 2h 05m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode/27aea19/ : SUCCESS in 2h 36m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a780332/ : SUCCESS in 2h 21m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3a1c0bf/ : SUCCESS in 2h 15m 19s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/7/check/tripleo-ci-fedora-28-standalone/41510be/ : FAILURE in 1h 41m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-standalone-upgrade/9d67d13/ : FAILURE in 1h 34m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-rocky/6ac6de5/ : SUCCESS in 2h 21m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-queens/7f46437/ : SUCCESS in 1h 54m 09s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-pike/628e29d/ : SUCCESS in 2h 40m 41s\n- openstack-tox-docs http://logs.openstack.org/24/639324/7/check/openstack-tox-docs/a0fdb30/html/ : SUCCESS in 4m 51s\n- openstack-tox-linters http://logs.openstack.org/24/639324/7/check/openstack-tox-linters/d39312b/ : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":7},{"id":"6e31e09f1dfa2357c9addacdad2400b5d00b42bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-13 15:09:52.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a57b19a/ : FAILURE in 2h 00m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/483b4c9/ : FAILURE in 2h 13m 33s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ec2964b/ : SUCCESS in 1h 37m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/017a3e4/ : SUCCESS in 2h 00m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/46fa29b/ : SUCCESS in 2h 17m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/99337f9/ : SUCCESS in 2h 42m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/3ddd973/ : FAILURE in 3h 01m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/10a409c/ : FAILURE in 2h 59m 27s","accounts_in_message":[],"_revision_number":7},{"id":"12ac34309ecaf8b48232125038016a1136295812","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-18 16:29:32.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"229324aaac846e3749a608181bae2725a2c5615e","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-18 17:47:00.000000000","message":"Patch Set 7:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8604/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6849/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"00acc536420f5f1bed2f16a0a2ea0a67ac2e17d8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-18 19:55:23.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/16473b4/ : SUCCESS in 2h 43m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/03d8424/ : SUCCESS in 2h 47m 26s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a1d2507/ : SUCCESS in 1h 22m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/9566f26/ : SUCCESS in 1h 59m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b411c8f/ : SUCCESS in 2h 15m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/88bf9f7/ : SUCCESS in 2h 43m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/9b85259/ : FAILURE in 1h 55m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2d9065a/ : FAILURE in 3h 08m 48s","accounts_in_message":[],"_revision_number":7},{"id":"854b599ed02676cf5def575e4b47714269bee433","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-18 20:29:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-undercloud-containers/8469d10/ : SUCCESS in 1h 47m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d2934c6/ : SUCCESS in 1h 58m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode/76b617b/ : SUCCESS in 1h 54m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1442a2f/ : SUCCESS in 2h 04m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e0e10df/ : SUCCESS in 2h 23m 14s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/7/check/tripleo-ci-fedora-28-standalone/778ca88/ : SUCCESS in 1h 29m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-standalone-upgrade/f7b197a/ : FAILURE in 1h 42m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-rocky/bbcccca/ : SUCCESS in 2h 11m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-queens/5d55985/ : SUCCESS in 2h 17m 01s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/7/check/tripleo-ci-centos-7-containers-multinode-pike/169fc9c/ : SUCCESS in 2h 10m 13s\n- openstack-tox-docs http://logs.openstack.org/24/639324/7/check/openstack-tox-docs/80a2c4b/html/ : SUCCESS in 4m 41s\n- openstack-tox-linters http://logs.openstack.org/24/639324/7/check/openstack-tox-linters/ee2c56e/ : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":7},{"id":"e374729aa2c6e77b0c14aae5fedad0d74a45143c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-03-19 05:43:51.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"dc9246fa1bd7c21e0a43c76cf01b67da17c3557b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-19 08:02:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-undercloud-containers/cefbb8d/ : SUCCESS in 1h 23m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/929239f/ : SUCCESS in 1h 51m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-containers-multinode/42130b6/ : SUCCESS in 2h 06m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c80bb07/ : SUCCESS in 2h 03m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a5569f2/ : FAILURE in 45m 31s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/8/check/tripleo-ci-fedora-28-standalone/c404d61/ : SUCCESS in 1h 33m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-standalone-upgrade/e23dbff/ : FAILURE in 1h 42m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-containers-multinode-rocky/8e4a1dc/ : SUCCESS in 2h 02m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-containers-multinode-queens/9fa1b67/ : SUCCESS in 1h 49m 53s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/8/check/tripleo-ci-centos-7-containers-multinode-pike/6611de3/ : FAILURE in 1h 31m 07s\n- openstack-tox-docs http://logs.openstack.org/24/639324/8/check/openstack-tox-docs/4b3264e/html/ : SUCCESS in 3m 24s\n- openstack-tox-linters http://logs.openstack.org/24/639324/8/check/openstack-tox-linters/6c9a1d8/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":8},{"id":"1449fde473e00c27f84c4eccfdf921f2cc7d3b8a","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-19 08:23:14.000000000","message":"Patch Set 8:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8605/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6850/ : SUCCESS","accounts_in_message":[],"_revision_number":8},{"id":"2e90e59573237c3661bb9179b30e26251636920f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-19 09:03:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d53303b/ : SUCCESS in 2h 45m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8539478/ : SUCCESS in 2h 47m 29s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/31ff830/ : SUCCESS in 1h 21m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/027a692/ : SUCCESS in 1h 51m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/31d346a/ : SUCCESS in 2h 13m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/9009538/ : SUCCESS in 2h 36m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/8c1b3a8/ : SUCCESS in 3h 02m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/83f0cce/ : FAILURE in 3h 17m 39s","accounts_in_message":[],"_revision_number":8},{"id":"58b091996e8f8408bd1ad49f2bfbaf1eb687bec2","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-21 14:50:42.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"7ad5baec8d3f5a3dbfc0e36d424056f329a41f0d","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-21 16:16:27.000000000","message":"Patch Set 9:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8625/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6870/ : FAILURE","accounts_in_message":[],"_revision_number":9},{"id":"5fbd700a5452bbac90d8869e580348703e8256ca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-21 16:31:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze03.softwarefactory-project.io:7979/311d483b6c244123946dd9a95a6b02bc : RETRY_LIMIT in 7s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/9/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/cfb0d34/ : SUCCESS in 1h 31m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/2cfd4a1/ : FAILURE in 1h 20m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze04.softwarefactory-project.io:7979/985cb14778774f9e819097b150f609d4 : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze01.softwarefactory-project.io:7979/e4a66d4ecbc346caa6f5b135f7bf9325 : RETRY_LIMIT in 7s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/5ef9a18/ : RETRY_LIMIT in 17m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze02.softwarefactory-project.io:7979/3e41e8d8704a4b3d883e9d09a11f3b26 : RETRY_LIMIT in 6s","accounts_in_message":[],"_revision_number":9},{"id":"267a6b6f374a0376af1449589959c151114ad61d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-21 17:31:24.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-undercloud-containers/af3d14a/ : FAILURE in 1h 03m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e395638/ : SUCCESS in 1h 54m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-containers-multinode/f1ab9fc/ : SUCCESS in 2h 07m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/31de586/ : SUCCESS in 2h 13m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/bf586ec/ : SUCCESS in 2h 32m 44s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/9/check/tripleo-ci-fedora-28-standalone/1d128a8/ : SUCCESS in 1h 56m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-standalone-upgrade/ea90c08/ : FAILURE in 1h 34m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-containers-multinode-rocky/2b95e68/ : SUCCESS in 2h 24m 02s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-containers-multinode-queens/a2d4e43/ : SUCCESS in 1h 58m 06s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/9/check/tripleo-ci-centos-7-containers-multinode-pike/24141d6/ : SUCCESS in 2h 18m 55s\n- openstack-tox-docs http://logs.openstack.org/24/639324/9/check/openstack-tox-docs/fc5cde0/html/ : SUCCESS in 4m 44s\n- openstack-tox-linters http://logs.openstack.org/24/639324/9/check/openstack-tox-linters/e63a6aa/ : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":9},{"id":"7278d166ddf75b661300ccaa427ab8e06d21dff6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-28 14:08:22.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"e5b1a87231a7e4cf2d203ae42e844d13bef86aa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 14:09:03.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"2514bc6841dd26a5a5c77c5d5de10de029844988","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-28 14:09:04.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"9f683591c9bcc4ff3abd994bac354ad6d1d39a95","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-28 14:09:33.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"f7a0dbf8b1afe9dd75878d8e8d300bb59956a363","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-28 14:14:04.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8681/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6926/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"cd2d97fcc7e808f9f4ff77e80e49e50e8a701937","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 16:30:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-undercloud-containers/34cbc4c/ : SUCCESS in 1h 17m 44s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/038ef02/ : SUCCESS in 1h 44m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode/bc7c43f/ : SUCCESS in 2h 05m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/942b440/ : SUCCESS in 2h 14m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fe873b1/ : SUCCESS in 2h 06m 48s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/10/check/tripleo-ci-fedora-28-standalone/6753c5a/ : SUCCESS in 1h 38m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-standalone-upgrade/d3603f6/ : FAILURE in 1h 37m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode-rocky/7c48edd/ : SUCCESS in 2h 05m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode-queens/f8a4999/ : SUCCESS in 2h 04m 47s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode-pike/b79c750/ : SUCCESS in 2h 10m 19s\n- openstack-tox-docs http://logs.openstack.org/24/639324/10/check/openstack-tox-docs/d955c41/html/ : SUCCESS in 4m 07s\n- openstack-tox-linters http://logs.openstack.org/24/639324/10/check/openstack-tox-linters/d0ad81d/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":10},{"id":"420de0411883bd9b3b5d3505593107dfee7d0492","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-28 16:54:02.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8680/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6925/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"320e35364c7073e2b5eb50413494a7fbc1b2397b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-28 17:32:10.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5a94306/ : SUCCESS in 2h 44m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c1acf34/ : SUCCESS in 2h 46m 30s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/9be0303/ : SUCCESS in 1h 31m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/0bfa801/ : SUCCESS in 1h 55m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/89e1d82/ : SUCCESS in 2h 16m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/ca1cef0/ : SUCCESS in 2h 26m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c24dc33/ : SUCCESS in 2h 55m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/b9f3755/ : FAILURE in 3h 11m 24s","accounts_in_message":[],"_revision_number":10},{"id":"1647d5f3a471a56d08bae10290b4ee55cd554457","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-28 20:37:48.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"b953c887c5b78987ebd5a2005e448d9ddd6a21fa","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-28 23:17:50.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8686/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6931/ : SUCCESS","accounts_in_message":[],"_revision_number":10},{"id":"c130125f4b48be9d5ead5f4a61e602e1a1f2452c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-28 23:23:32.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-undercloud-containers/32b06a3/ : SUCCESS in 1h 42m 19s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/71b46c7/ : SUCCESS in 2h 05m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode/7c84d3f/ : SUCCESS in 1h 47m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2c5c37c/ : SUCCESS in 2h 15m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f86249f/ : SUCCESS in 2h 38m 26s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/10/check/tripleo-ci-fedora-28-standalone/4c190e8/ : SUCCESS in 1h 39m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-standalone-upgrade/b7124ed/ : FAILURE in 1h 34m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode-rocky/5461347/ : SUCCESS in 2h 04m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode-queens/69ef941/ : SUCCESS in 1h 53m 55s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/10/check/tripleo-ci-centos-7-containers-multinode-pike/4cfdf73/ : SUCCESS in 1h 58m 07s\n- openstack-tox-docs http://logs.openstack.org/24/639324/10/check/openstack-tox-docs/871a13f/html/ : SUCCESS in 4m 12s\n- openstack-tox-linters http://logs.openstack.org/24/639324/10/check/openstack-tox-linters/72e848c/ : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":10},{"id":"9e6234278e64d2f8ca0d629d50f5d43590087894","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-03-28 23:29:02.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d06af72/ : SUCCESS in 2h 39m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6c12dcd/ : SUCCESS in 2h 43m 14s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/7984df6/ : SUCCESS in 1h 24m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/1f71baf/ : SUCCESS in 1h 45m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7ed19a4/ : SUCCESS in 2h 09m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e58e5f2/ : SUCCESS in 2h 24m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/390515a/ : SUCCESS in 2h 49m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/10/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/681a8d1/ : FAILURE in 2h 07m 38s","accounts_in_message":[],"_revision_number":10},{"id":"a5d87772624e24e66d8ae727d9f59b4d86883b4a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-31 21:39:21.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"30965ca6fa6c41acd7ea27675e9b9f6779ebb044","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-03-31 21:39:28.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"07b7a4ed097cb7ab0906006729d49feeeb170c54","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-03-31 21:39:29.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8698/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6943/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"69997824510e9937e32969ca53f2f395e902b442","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 00:20:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-undercloud-containers/387228e/ : SUCCESS in 1h 17m 03s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2713a5b/ : SUCCESS in 1h 47m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-containers-multinode/3fed363/ : SUCCESS in 1h 54m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/447193e/ : SUCCESS in 1h 56m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f188dda/ : SUCCESS in 2h 19m 35s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/12/check/tripleo-ci-fedora-28-standalone/56e6ccb/ : SUCCESS in 1h 22m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-standalone-upgrade/3b14ef3/ : FAILURE in 1h 35m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-containers-multinode-rocky/acd66f1/ : FAILURE in 32m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-containers-multinode-queens/5deb2c0/ : SUCCESS in 1h 55m 26s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/12/check/tripleo-ci-centos-7-containers-multinode-pike/ea8f34c/ : SUCCESS in 2h 03m 53s\n- openstack-tox-docs http://logs.openstack.org/24/639324/12/check/openstack-tox-docs/c09f53f/html/ : SUCCESS in 3m 29s\n- openstack-tox-linters http://logs.openstack.org/24/639324/12/check/openstack-tox-linters/d70a176/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":12},{"id":"c10f98c5e8f92bcc17c3531f608afc5d407f09b4","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 00:22:48.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8699/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6944/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"2bc62a480cf14466c3f746a104f07c7b01614c7e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-01 00:33:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e90f6a0/ : SUCCESS in 2h 27m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ea1ebb1/ : SUCCESS in 2h 32m 45s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/edbe94c/ : SUCCESS in 1h 19m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/16aa9fa/ : SUCCESS in 2h 01m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/f0a6add/ : SUCCESS in 2h 05m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6a05619/ : SUCCESS in 2h 13m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/9b858e2/ : SUCCESS in 2h 42m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/12/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/656e761/ : FAILURE in 2h 52m 26s","accounts_in_message":[],"_revision_number":12},{"id":"f695408242cab21598e09232cf14d2871706e2e3","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-04-01 10:16:31.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"33b94517a2a82eddbf5f4f07235154ca0ea46724","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-02 07:59:08.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"39cb805bfa4bf8635f4105e05820ee7979a104dd","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 09:18:45.000000000","message":"Patch Set 13:\n\nBuild Failed \n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6947/ : FAILURE\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8702/ : FAILURE","accounts_in_message":[],"_revision_number":13},{"id":"0ef8c8ed00944b795f38a5d13fff463a1475cf00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 10:17:31.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-undercloud-containers/b94f3fd/ : SUCCESS in 1h 32m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/18152c6/ : SUCCESS in 2h 07m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-containers-multinode/d4fb7ed/ : SUCCESS in 1h 57m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9172f60/ : SUCCESS in 2h 10m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5b02ff6/ : SUCCESS in 2h 16m 58s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/13/check/tripleo-ci-fedora-28-standalone/aad136c/ : SUCCESS in 1h 29m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-standalone-upgrade/9d724b7/ : FAILURE in 1h 40m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-containers-multinode-rocky/7b6b8b4/ : SUCCESS in 2h 13m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-containers-multinode-queens/307d651/ : SUCCESS in 1h 53m 25s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/13/check/tripleo-ci-centos-7-containers-multinode-pike/53f4525/ : SUCCESS in 2h 10m 17s\n- openstack-tox-docs http://logs.openstack.org/24/639324/13/check/openstack-tox-docs/e9d8d65/html/ : SUCCESS in 4m 01s\n- openstack-tox-linters http://logs.openstack.org/24/639324/13/check/openstack-tox-linters/48f7c74/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":13},{"id":"0439661f289841ab7c1390b41ffe4213ab02b6e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-02 10:28:55.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9e62612/ : FAILURE in 1h 10m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ededb4e/ : FAILURE in 1h 09m 50s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/0d6a3ca/ : SUCCESS in 1h 28m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/731392c/ : SUCCESS in 1h 47m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/78228c0/ : SUCCESS in 2h 27m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/99b1588/ : FAILURE in 1h 33m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/53ae79c/ : FAILURE in 1h 09m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/13/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/4a51eec/ : FAILURE in 1h 33m 06s","accounts_in_message":[],"_revision_number":13},{"id":"753d76590904d04eb19714f5d04388ed9c16f6a7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-02 11:28:31.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"1f7b5338419a0fcec8ccaf9cf12526fee6376bae","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 14:08:39.000000000","message":"Patch Set 14:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8703/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6948/ : SUCCESS","accounts_in_message":[],"_revision_number":14},{"id":"b9c80ca915e76030bef84d8a2c41791024c22212","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 14:17:50.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-undercloud-containers/c7af721/ : SUCCESS in 1h 24m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/21a5cf1/ : SUCCESS in 1h 52m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-containers-multinode/ec6bc3b/ : SUCCESS in 1h 57m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a601760/ : SUCCESS in 2h 37m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3eb0614/ : FAILURE in 2h 02m 37s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/14/check/tripleo-ci-fedora-28-standalone/48bb190/ : FAILURE in 1h 28m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-standalone-upgrade/8f1768a/ : FAILURE in 1h 29m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-containers-multinode-rocky/d0de903/ : SUCCESS in 2h 03m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-containers-multinode-queens/ca8ff6e/ : SUCCESS in 2h 08m 21s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/14/check/tripleo-ci-centos-7-containers-multinode-pike/a76e630/ : SUCCESS in 2h 03m 48s\n- openstack-tox-docs http://logs.openstack.org/24/639324/14/check/openstack-tox-docs/b5c72e0/html/ : SUCCESS in 4m 40s\n- openstack-tox-linters http://logs.openstack.org/24/639324/14/check/openstack-tox-linters/260c1c2/ : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":14},{"id":"844707e8cb6f5b4a07465a779b8f5b36113d23b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-02 14:17:59.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/33cfec1/ : SUCCESS in 2h 38m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1ad1019/ : SUCCESS in 2h 42m 54s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/14/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/285960d/ : SUCCESS in 1h 24m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/14/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c6208cf/ : SUCCESS in 2h 26m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":14},{"id":"e708341f097fb3fc12bdfff9d9768e5a11440ed8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-04 08:06:32.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"2adbbdd1ef22959c30f4761f75f61ba1ffbd58e2","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-04 09:50:41.000000000","message":"Patch Set 15:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8713/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6958/ : SUCCESS","accounts_in_message":[],"_revision_number":15},{"id":"fcbbd7f85eedd2d47299f83502598f79a5edbeab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-04 10:33:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-undercloud-containers/3723363/ : SUCCESS in 1h 35m 32s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f8b13d2/ : FAILURE in 1h 27m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-containers-multinode/f69fa29/ : SUCCESS in 1h 57m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/feb4cfa/ : SUCCESS in 2h 01m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5ec0865/ : FAILURE in 2h 11m 39s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/15/check/tripleo-ci-fedora-28-standalone/9fc4092/ : FAILURE in 24m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-standalone-upgrade/11bcfce/ : FAILURE in 1h 12m 58s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-containers-multinode-stein/7490d23/ : SUCCESS in 2h 01m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-containers-multinode-rocky/c2dd175/ : SUCCESS in 2h 13m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-containers-multinode-queens/584fa33/ : SUCCESS in 1h 56m 43s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/15/check/tripleo-ci-centos-7-containers-multinode-pike/0a24acc/ : SUCCESS in 2h 09m 46s\n- openstack-tox-docs http://logs.openstack.org/24/639324/15/check/openstack-tox-docs/d956a25/html/ : SUCCESS in 5m 51s\n- openstack-tox-linters http://logs.openstack.org/24/639324/15/check/openstack-tox-linters/d31aee7/ : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":15},{"id":"46a842425f4e82c8e04ad97ade4a08216dd24573","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-04 11:22:19.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/439ef6d/ : SUCCESS in 3h 03m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9d3adba/ : SUCCESS in 2h 49m 48s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/2462c29/ : SUCCESS in 1h 21m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/f8ca29d/ : SUCCESS in 2h 05m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a714094/ : SUCCESS in 2h 21m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c14926d/ : SUCCESS in 2h 27m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/6037687/ : FAILURE in 1h 48m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/15/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/bf64834/ : FAILURE in 3h 11m 27s","accounts_in_message":[],"_revision_number":15},{"id":"f32d7e37aa490e54206e1bd39625918d93781b4a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-08 06:30:08.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"e93c2370b3357e4bdf356c300c4e81e258541864","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-08 08:11:06.000000000","message":"Patch Set 16:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8744/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6989/ : SUCCESS","accounts_in_message":[],"_revision_number":16},{"id":"e20620defa756bd74b1e5134966a3327f8354cea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 08:56:04.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-undercloud-containers/782e166/ : SUCCESS in 1h 22m 55s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6660bdb/ : SUCCESS in 1h 24m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-containers-multinode/dc7f99b/ : SUCCESS in 1h 55m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/347a02e/ : SUCCESS in 1h 58m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f1be729/ : SUCCESS in 2h 12m 31s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/16/check/tripleo-ci-fedora-28-standalone/edde169/ : FAILURE in 20m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-standalone-upgrade/e66934f/ : FAILURE in 1h 16m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-containers-multinode-stein/da2d427/ : SUCCESS in 2h 00m 18s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-containers-multinode-rocky/79548f4/ : SUCCESS in 2h 10m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-containers-multinode-queens/b703e88/ : SUCCESS in 2h 01m 23s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/16/check/tripleo-ci-centos-7-containers-multinode-pike/3a96732/ : SUCCESS in 2h 09m 41s\n- openstack-tox-docs http://logs.openstack.org/24/639324/16/check/openstack-tox-docs/8754e6e/html/ : SUCCESS in 3m 52s\n- openstack-tox-linters http://logs.openstack.org/24/639324/16/check/openstack-tox-linters/3697c61/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":16},{"id":"d2b179bc3252b04af016399802da6ec66766dfc7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-08 09:29:11.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f841789/ : SUCCESS in 2h 31m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/eb3fb2f/ : SUCCESS in 2h 45m 35s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/9d53cad/ : SUCCESS in 1h 19m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/6699ad3/ : SUCCESS in 1h 46m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/60033af/ : SUCCESS in 2h 11m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/59e774a/ : SUCCESS in 2h 30m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/5e5f2a9/ : SUCCESS in 2h 47m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/16/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/9272f8a/ : SUCCESS in 2h 56m 51s","accounts_in_message":[],"_revision_number":16},{"id":"c29ea7c69945bef8ce8784c418e1b249694be1c3","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-08 09:38:07.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c702104738f467a03d1d695294df80b62b552149","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-08 12:19:42.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-undercloud-containers/b37828c/ : SUCCESS in 1h 26m 52s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6e08ee4/ : SUCCESS in 1h 37m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-containers-multinode/56224b6/ : SUCCESS in 1h 54m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/42f4469/ : SUCCESS in 2h 15m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7ec01a4/ : SUCCESS in 2h 12m 36s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/17/check/tripleo-ci-fedora-28-standalone/54232f9/ : FAILURE in 20m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-standalone-upgrade/f6e789d/ : FAILURE in 1h 10m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-containers-multinode-stein/ca74c0a/ : SUCCESS in 1h 47m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-containers-multinode-rocky/d9c752a/ : SUCCESS in 1h 54m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-containers-multinode-queens/3bc3bc3/ : SUCCESS in 1h 50m 07s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/17/check/tripleo-ci-centos-7-containers-multinode-pike/3abe053/ : SUCCESS in 2h 17m 49s\n- openstack-tox-docs http://logs.openstack.org/24/639324/17/check/openstack-tox-docs/f1743f0/html/ : SUCCESS in 5m 07s\n- openstack-tox-linters http://logs.openstack.org/24/639324/17/check/openstack-tox-linters/efabe94/ : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":17},{"id":"44e9d9b225850b440c91684ffca9ac6053fe2084","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-08 12:20:31.000000000","message":"Patch Set 17:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8749/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/6994/ : SUCCESS","accounts_in_message":[],"_revision_number":17},{"id":"ba70d490f32720d5501d082e69e4e19f53ea2394","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-08 12:31:50.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fc305d0/ : SUCCESS in 2h 51m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4d4919b/ : SUCCESS in 2h 37m 14s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/d981368/ : SUCCESS in 1h 30m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/9073b16/ : SUCCESS in 1h 49m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3a3a664/ : SUCCESS in 2h 24m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4936bcb/ : SUCCESS in 2h 20m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/23b2de9/ : SUCCESS in 2h 40m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/17/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0c420e7/ : FAILURE in 2h 49m 11s","accounts_in_message":[],"_revision_number":17},{"id":"82a5dd7864e67b79694fd8c1fcfac7668f6ea4ab","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-12 08:43:32.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"435f13aacbf8b09b7e789adce9046316f5a48361","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-12 11:22:05.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-undercloud-containers/89a4354/ : SUCCESS in 1h 49m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f8f431e/ : SUCCESS in 1h 45m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-containers-multinode/6d29183/ : SUCCESS in 1h 54m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/576c518/ : SUCCESS in 2h 08m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e205a84/ : SUCCESS in 2h 20m 03s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/18/check/tripleo-ci-fedora-28-standalone/0baef75/ : SUCCESS in 1h 46m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-standalone-upgrade/68c898f/ : FAILURE in 1h 22m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-containers-multinode-stein/e0ee4cc/ : SUCCESS in 1h 50m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-containers-multinode-rocky/091eb25/ : SUCCESS in 2h 17m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-containers-multinode-queens/708115a/ : SUCCESS in 1h 48m 44s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/18/check/tripleo-ci-centos-7-containers-multinode-pike/1a59184/ : SUCCESS in 1h 59m 57s\n- openstack-tox-docs http://logs.openstack.org/24/639324/18/check/openstack-tox-docs/83bab89/html/ : SUCCESS in 3m 40s\n- openstack-tox-linters http://logs.openstack.org/24/639324/18/check/openstack-tox-linters/d17803c/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":18},{"id":"18d1b814cdb5c72140dacfe7fff537e8325f2916","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-12 11:26:51.000000000","message":"Patch Set 18:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8797/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/7042/ : SUCCESS","accounts_in_message":[],"_revision_number":18},{"id":"8955b926aab09611054e5a28bb59ffd5d99b245c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-12 11:40:48.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/111bc95/ : SUCCESS in 2h 49m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5069e24/ : SUCCESS in 2h 51m 05s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/5f527b6/ : SUCCESS in 1h 24m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/8ce5e9d/ : SUCCESS in 1h 52m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3f44c4b/ : SUCCESS in 2h 20m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/88c6047/ : SUCCESS in 2h 26m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/eb6f460/ : SUCCESS in 2h 54m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/18/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e0fbc3b/ : FAILURE in 1h 55m 12s","accounts_in_message":[],"_revision_number":18},{"id":"2e5ffeb3d7d54288cc1b4281e1ec8cc9deb0388a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-13 16:47:56.000000000","message":"Patch Set 19: Patch Set 18 was rebased","accounts_in_message":[],"_revision_number":19},{"id":"0fee18ade77723798e44ab7dd3441c2cdc6226a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-13 18:53:36.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-undercloud-containers/38e44e6/ : SUCCESS in 1h 18m 04s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/fa17686/ : SUCCESS in 1h 29m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode/32ed5fb/ : SUCCESS in 1h 46m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/20526bb/ : SUCCESS in 1h 57m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ad4ebe0/ : SUCCESS in 2h 03m 21s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/19/check/tripleo-ci-fedora-28-standalone/5c410de/ : SUCCESS in 1h 26m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-standalone-upgrade/8b7a121/ : FAILURE in 59m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-stein/a32f31e/ : FAILURE in 1h 26m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-rocky/ced4c14/ : SUCCESS in 1h 55m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-queens/da5962c/ : SUCCESS in 1h 46m 26s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-pike/243e7f3/ : SUCCESS in 2h 00m 05s\n- openstack-tox-docs http://logs.openstack.org/24/639324/19/check/openstack-tox-docs/d3929bc/html/ : SUCCESS in 3m 51s\n- openstack-tox-linters http://logs.openstack.org/24/639324/19/check/openstack-tox-linters/b8eb5b8/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":19},{"id":"941a9ada4f33e111a79f6c78844cc34868b3f0f0","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-13 19:25:54.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8806/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/7051/ : SUCCESS","accounts_in_message":[],"_revision_number":19},{"id":"f035302ee71374969fb370f24b2cc10c26373556","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-13 19:57:28.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/eb46727/ : SUCCESS in 2h 31m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/634d178/ : SUCCESS in 2h 42m 15s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/94e45cd/ : SUCCESS in 1h 18m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/1896c20/ : SUCCESS in 1h 46m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/45bea24/ : SUCCESS in 2h 15m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/45b41bc/ : SUCCESS in 2h 35m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/89628c4/ : SUCCESS in 3h 06m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/19/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/4542b68/ : FAILURE in 1h 41m 27s","accounts_in_message":[],"_revision_number":19},{"id":"d64c1a195635c0e32aad45aa1cd9c920dd3b83ca","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-15 05:21:34.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"9165ed7bfc259165c54b1e6757fe56970de94a48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 07:51:35.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-undercloud-containers/2e90afa/ : SUCCESS in 1h 38m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/189df9d/ : SUCCESS in 1h 56m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode/7c9853c/ : SUCCESS in 1h 53m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e776e85/ : SUCCESS in 2h 01m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fb60726/ : SUCCESS in 2h 23m 56s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/19/check/tripleo-ci-fedora-28-standalone/b8f996c/ : FAILURE in 21m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-standalone-upgrade/16eae36/ : FAILURE in 1h 26m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-stein/5456334/ : SUCCESS in 1h 58m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-rocky/3c850d2/ : SUCCESS in 2h 03m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-queens/0e8dfa1/ : FAILURE in 1h 05m 12s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/19/check/tripleo-ci-centos-7-containers-multinode-pike/4f67a56/ : SUCCESS in 2h 04m 14s\n- openstack-tox-docs http://logs.openstack.org/24/639324/19/check/openstack-tox-docs/ec1ce7b/html/ : SUCCESS in 2m 59s\n- openstack-tox-linters http://logs.openstack.org/24/639324/19/check/openstack-tox-linters/291327b/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":19},{"id":"47b281459ec3f7b1115bab2433a09230167280e9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-15 07:52:08.000000000","message":"Patch Set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"93b708b202238b69d999d2190b231fee1b7c74ee","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-15 07:58:31.000000000","message":"Patch Set 19:\n\nBuild Failed \n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/7053/ : FAILURE\n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8808/ : FAILURE","accounts_in_message":[],"_revision_number":19},{"id":"2449938303d0286bdef9bbfb359212011092acaf","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-15 09:32:06.000000000","message":"Patch Set 20:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8809/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/7054/ : SUCCESS","accounts_in_message":[],"_revision_number":20},{"id":"767b5499d18f42ce2d2cf998a7f156f6a692a585","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-15 10:15:41.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-undercloud-containers/c00cae5/ : SUCCESS in 1h 17m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9662b76/ : SUCCESS in 1h 29m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-containers-multinode/d6f0490/ : SUCCESS in 1h 47m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2392c86/ : SUCCESS in 1h 58m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d296f43/ : SUCCESS in 2h 08m 36s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/20/check/tripleo-ci-fedora-28-standalone/f7d4564/ : SUCCESS in 1h 33m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-standalone-upgrade/5ed9ac2/ : FAILURE in 58m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-containers-multinode-stein/35851b0/ : SUCCESS in 1h 42m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-containers-multinode-rocky/1eaee3d/ : SUCCESS in 1h 51m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-containers-multinode-queens/95dd636/ : SUCCESS in 1h 54m 22s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/20/check/tripleo-ci-centos-7-containers-multinode-pike/4ca28b0/ : SUCCESS in 2h 11m 13s\n- openstack-tox-docs http://logs.openstack.org/24/639324/20/check/openstack-tox-docs/33c849d/html/ : SUCCESS in 4m 24s\n- openstack-tox-linters http://logs.openstack.org/24/639324/20/check/openstack-tox-linters/c6261f6/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":20},{"id":"8d31627961ac213f767102de4c65131124874002","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-15 11:18:41.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6e64b00/ : SUCCESS in 2h 52m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9158b8a/ : SUCCESS in 2h 45m 42s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b9beea5/ : SUCCESS in 1h 32m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/57dc530/ : SUCCESS in 2h 02m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c52143f/ : SUCCESS in 2h 19m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/7b39f1b/ : SUCCESS in 2h 35m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/ca3b148/ : SUCCESS in 3h 22m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ddbf7ea/ : FAILURE in 1h 47m 37s","accounts_in_message":[],"_revision_number":20},{"id":"d6e522f21726b7c08f24ba0f7e3308d287a15d6c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-04-15 11:31:04.000000000","message":"Patch Set 20:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":20},{"id":"b5650ed9e91a011222bd9be8cc39b767134f3b17","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-15 14:34:08.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/583d674/ : SUCCESS in 2h 51m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2f05561/ : FAILURE in 2h 52m 10s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/211c498/ : SUCCESS in 1h 22m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/9a6e37f/ : SUCCESS in 1h 47m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/5175347/ : SUCCESS in 2h 15m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/22eefbe/ : FAILURE in 3h 00m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/e5490f3/ : SUCCESS in 2h 51m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c5c5fd6/ : FAILURE in 1h 35m 33s","accounts_in_message":[],"_revision_number":20},{"id":"6950fc4dc672b6d0adc1e16214c0c7bec5259ff8","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-04-16 11:19:12.000000000","message":"Patch Set 20:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":20},{"id":"f96f92012b3dcc05838fc23e625b93e29ab199ed","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-16 15:04:28.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bb0600c/ : FAILURE in 3h 00m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/105f0c8/ : SUCCESS in 2h 43m 55s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/600b1af/ : SUCCESS in 1h 26m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/1c5b338/ : SUCCESS in 1h 51m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a97304b/ : SUCCESS in 2h 24m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/94b1351/ : TIMED_OUT in 3h 43m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/950bc9b/ : SUCCESS in 3h 09m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/489537a/ : FAILURE in 2h 17m 36s","accounts_in_message":[],"_revision_number":20},{"id":"51fb14ee80285020a1dca5ea300c7118ea95130a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-04-16 16:37:29.000000000","message":"Patch Set 20:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":20},{"id":"2e292ff5a9ce1fb020aea0e0664df9ca23f09faa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-16 19:29:40.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b373573/ : SUCCESS in 2h 45m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/337f187/ : SUCCESS in 2h 43m 22s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/f2cad07/ : SUCCESS in 1h 22m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/4171bb8/ : SUCCESS in 1h 47m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/53d33f6/ : SUCCESS in 2h 12m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a6073fe/ : SUCCESS in 2h 23m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/5450d45/ : SUCCESS in 2h 49m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/79f984b/ : FAILURE in 46m 47s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"f39ff41f1f4db7e7a74c11792729f54718590500","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-04-16 23:25:05.000000000","message":"Patch Set 20:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":20},{"id":"4e738425b5fe950ef73473f6c84a9ad8cfa65a2f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-17 05:17:35.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/99fb2ed/ : SUCCESS in 2h 45m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a9e86d5/ : FAILURE in 2h 56m 26s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/58f14e3/ : SUCCESS in 1h 23m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/b482454/ : SUCCESS in 1h 49m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3718a57/ : SUCCESS in 2h 37m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/9b66903/ : FAILURE in 1h 21m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/697a57e/ : SUCCESS in 2h 52m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/20/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c8cd619/ : FAILURE in 59m 20s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"78633a89ec25b569d33d878509d671aa08e6cbe2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-17 08:09:02.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"76dd0aaafec11c94064c597170aacc5441e2cc67","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 10:52:44.000000000","message":"Patch Set 21:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8824/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/7069/ : SUCCESS","accounts_in_message":[],"_revision_number":21},{"id":"ea0739804c692455a229eb20322e0c5ccfb95885","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-17 11:18:27.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d56bf2b/ : FAILURE in 2h 37m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b60d9d7/ : SUCCESS in 2h 52m 05s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/9baa6cd/ : SUCCESS in 1h 27m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/904b866/ : SUCCESS in 1h 44m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/cd2ecc2/ : SUCCESS in 2h 14m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/298237a/ : SUCCESS in 2h 18m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/466f73d/ : SUCCESS in 3h 01m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/21/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/6a079cd/ : FAILURE in 45m 15s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"52cbd10ec3d16fb2f8583c3ae68a6f50c89dab4a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 12:12:53.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-undercloud-containers/4b02c43/ : SUCCESS in 1h 39m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1552339/ : FAILURE in 1h 31m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-containers-multinode/65ee35d/ : SUCCESS in 2h 15m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/887fb52/ : SUCCESS in 2h 10m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/aa8732c/ : SUCCESS in 2h 25m 24s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/21/check/tripleo-ci-fedora-28-standalone/af2fd3a/ : FAILURE in 1h 42m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-standalone-upgrade/de77d26/ : FAILURE in 1h 39m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-containers-multinode-stein/0e42f08/ : SUCCESS in 2h 02m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-containers-multinode-rocky/47e1ba1/ : SUCCESS in 2h 15m 44s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-containers-multinode-queens/4f3d4cd/ : SUCCESS in 2h 02m 25s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-containers-multinode-pike/dffba29/ : SUCCESS in 2h 20m 27s\n- openstack-tox-docs http://logs.openstack.org/24/639324/21/check/openstack-tox-docs/7bba956/html/ : SUCCESS in 4m 59s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/24/639324/21/check/tripleo-ci-centos-7-standalone-rocky/e6af2ff/ : SUCCESS in 1h 31m 01s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/24/639324/21/check/openstack-tox-linters/43633b2/ : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":21},{"id":"4e922e5b9b058fd666260c377aab7e8c706c9897","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-17 12:43:32.000000000","message":"Patch Set 22: Patch Set 21 was rebased","accounts_in_message":[],"_revision_number":22},{"id":"e87dae58c8bdee2a192d535fb97097c879893d24","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 14:47:59.000000000","message":"Patch Set 22:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/8826/ : FAILURE\n\n* tripleo-quickstart-gate-ocata-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-ocata-delorean-quick-basic/7071/ : SUCCESS","accounts_in_message":[],"_revision_number":22},{"id":"b955c3cdad720b7b828d0b729dde94d545c81053","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-17 16:13:34.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/50f6e77/ : SUCCESS in 2h 48m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1de7a5b/ : SUCCESS in 3h 02m 55s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/4f482c4/ : SUCCESS in 1h 33m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/138d63b/ : SUCCESS in 1h 58m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c44b73a/ : SUCCESS in 2h 26m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2ef1b90/ : SUCCESS in 2h 35m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/ca6d6c8/ : SUCCESS in 3h 26m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/317d71d/ : FAILURE in 1h 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"6349d6d8b52c28567b31dc9c6c2b650e6fc84492","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 20:35:45.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-undercloud-containers/c25bca3/ : SUCCESS in 1h 29m 05s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5d9f24c/ : FAILURE in 1h 35m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-containers-multinode/7572c5e/ : SUCCESS in 1h 55m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ebe4807/ : SUCCESS in 2h 10m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2f083f5/ : SUCCESS in 2h 18m 55s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/22/check/tripleo-ci-fedora-28-standalone/07a8111/ : SUCCESS in 1h 45m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-standalone-upgrade/c2e60db/ : FAILURE in 1h 43m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-containers-multinode-stein/818209a/ : SUCCESS in 2h 00m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-containers-multinode-rocky/2473f0c/ : SUCCESS in 2h 13m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-containers-multinode-queens/61522a5/ : SUCCESS in 2h 02m 26s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-containers-multinode-pike/81e8f4d/ : SUCCESS in 2h 12m 14s\n- openstack-tox-docs http://logs.openstack.org/24/639324/22/check/openstack-tox-docs/7d4d27e/html/ : SUCCESS in 5m 27s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/24/639324/22/check/tripleo-ci-centos-7-standalone-rocky/da36925/ : SUCCESS in 1h 28m 45s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/24/639324/22/check/openstack-tox-linters/f7871cb/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":22},{"id":"f13d4d367932c3277e8bcfcef714151a3781ee21","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-04-17 23:51:42.000000000","message":"Patch Set 22:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":22},{"id":"144ca57c2fb39515529daea0210fc84164ae269c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-18 03:06:28.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c20e8b9/ : SUCCESS in 2h 48m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/db08c60/ : FAILURE in 2h 58m 28s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b6d5475/ : SUCCESS in 1h 28m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/7ebe448/ : SUCCESS in 1h 50m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d11d938/ : SUCCESS in 2h 30m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/52b6e60/ : SUCCESS in 2h 33m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/9a18b6e/ : SUCCESS in 3h 03m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/159af02/ : FAILURE in 3h 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"7a63c8738d8f8cf3571f85b5d6df10f1e6caabc9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-04-18 04:24:42.000000000","message":"Patch Set 22:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":22},{"id":"409c904d380d63f4a25b05fa86d96772e9f76a7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-18 07:24:12.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4568f61/ : SUCCESS in 2h 30m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/81ad901/ : SUCCESS in 2h 34m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c2c8d64/ : SUCCESS in 1h 21m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/ebed253/ : SUCCESS in 1h 43m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3e8520f/ : SUCCESS in 2h 10m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e9e941a/ : SUCCESS in 2h 17m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/40ab474/ : SUCCESS in 2h 53m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/22/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/8505332/ : FAILURE in 2h 57m 03s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"b47b844ba79ded6759d4cbda0228a41e4461ebee","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-04-18 09:41:09.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"152492aba4cab9388a28850bfc7f4d405737f3b0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-04-22 22:43:29.000000000","message":"Patch Set 23: Patch Set 22 was rebased","accounts_in_message":[],"_revision_number":23},{"id":"4ad262afec9a8c01a67f96ebb7bbd7de8c61d915","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-04-23 01:56:33.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c79e12d/ : FAILURE in 2h 47m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6b604eb/ : FAILURE in 2h 51m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/e69f81e/ : SUCCESS in 1h 39m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-pike-branch/e51e9db/ : SUCCESS in 1h 46m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/54215ab/ : SUCCESS in 2h 23m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e747fb1/ : SUCCESS in 2h 38m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/3201663/ : FAILURE in 2h 49m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/23/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/8b6a0e1/ : FAILURE in 3h 10m 59s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"3fee1b91e71ab7ab43635015e0420e3f9bc1ccab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-23 04:25:08.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-undercloud-containers/6b71ace/ : SUCCESS in 1h 32m 31s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9de3a5f/ : FAILURE in 1h 27m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-containers-multinode/df74727/ : SUCCESS in 2h 00m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e0ef76e/ : SUCCESS in 2h 18m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4949eb9/ : SUCCESS in 2h 25m 31s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/23/check/tripleo-ci-fedora-28-standalone/fe67c72/ : SUCCESS in 1h 48m 04s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-standalone-upgrade/46b2f9a/ : FAILURE in 1h 48m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-containers-multinode-stein/1d3845d/ : SUCCESS in 2h 13m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-containers-multinode-rocky/a6f71e6/ : SUCCESS in 2h 17m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-containers-multinode-queens/52009ed/ : SUCCESS in 1h 59m 04s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-containers-multinode-pike/07df2c5/ : SUCCESS in 2h 36m 37s\n- openstack-tox-docs http://logs.openstack.org/24/639324/23/check/openstack-tox-docs/9055a6e/html/ : SUCCESS in 3m 53s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/24/639324/23/check/tripleo-ci-centos-7-standalone-rocky/95a8e94/ : SUCCESS in 1h 33m 21s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/24/639324/23/check/openstack-tox-linters/13b1c19/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":23},{"id":"425f9255cddaf1556c6144a0912dbede66cf2a00","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-05-16 15:16:10.000000000","message":"Patch Set 24: Patch Set 23 was rebased","accounts_in_message":[],"_revision_number":24},{"id":"10afb2f4b70477181b3e4b7c417b156618993c22","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-16 16:34:39.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/02ad791/ : FAILURE in 1h 07m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/51bab71/ : FAILURE in 1h 06m 09s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a3889a5/ : SUCCESS in 1h 16m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bdf0e1a/ : FAILURE in 1h 03m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cad1937/ : FAILURE in 1h 15m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b7e4b4b/ : FAILURE in 1h 06m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/573ff34/ : FAILURE in 1h 05m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c263dee/ : FAILURE in 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"1e4225551dd656789f4ddcae799ee7e7350396b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 20:01:50.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-undercloud-containers/1df82f6/ : SUCCESS in 1h 28m 39s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c12cdc6/ : FAILURE in 1h 52m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-containers-multinode/74076df/ : SUCCESS in 2h 16m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/32decec/ : SUCCESS in 2h 02m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/745e5f6/ : SUCCESS in 2h 39m 20s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/24/check/tripleo-ci-fedora-28-standalone/f0af4de/ : SUCCESS in 1h 35m 14s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-standalone-upgrade/e5a8aac/ : SUCCESS in 2h 09m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-standalone-upgrade-stein/2cc4101/ : SUCCESS in 2h 17m 40s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-containers-multinode-stein/ffa99d2/ : SUCCESS in 1h 49m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-containers-multinode-rocky/7df2053/ : SUCCESS in 2h 09m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-containers-multinode-queens/667f1b1/ : SUCCESS in 2h 08m 29s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-containers-multinode-pike/b0ad6da/ : SUCCESS in 2h 11m 55s\n- openstack-tox-docs http://logs.openstack.org/24/639324/24/check/openstack-tox-docs/29111fc/html/ : SUCCESS in 4m 34s\n- tripleo-ci-centos-7-standalone-rocky http://logs.openstack.org/24/639324/24/check/tripleo-ci-centos-7-standalone-rocky/b3b17dc/ : SUCCESS in 1h 35m 49s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/24/639324/24/check/openstack-tox-linters/5c0ffb4/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":24},{"id":"4668c5ab4e442f031cbcd35aabe2cfebde431ba0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-05-16 21:36:33.000000000","message":"Patch Set 24:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":24},{"id":"c42a9395bba1ef677a9b0d6b5fffb94b2b4bd80c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 00:27:56.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b8aa40b/ : FAILURE in 1h 41m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8634ced/ : SUCCESS in 2h 46m 23s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/166fce7/ : SUCCESS in 1h 29m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bde04f1/ : SUCCESS in 2h 23m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2f7b9c0/ : SUCCESS in 2h 30m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/3a67759/ : SUCCESS in 2h 49m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/d115076/ : FAILURE in 2h 35m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/f38a5f0/ : FAILURE in 2h 48m 19s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"a96978d4e9cc072d34134e1b0327e1f63bb118a9","author":{"_account_id":30402,"name":"gaoqiang","email":"gqgqgqdyx@126.com"},"date":"2019-05-17 01:46:11.000000000","message":"Patch Set 24: Code-Review+1","accounts_in_message":[],"_revision_number":24},{"id":"f649c91c39f794a0161d9b3183c725a2d23ae62e","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-05-17 21:07:07.000000000","message":"Patch Set 24:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":24},{"id":"a7662e8bb8885d33d0041abd5751bcfa92ee5632","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-17 23:38:38.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e396bac/ : SUCCESS in 2h 16m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/4732313/ : SUCCESS in 2h 24m 36s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c7fd057/ : SUCCESS in 1h 07m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ef89a78/ : SUCCESS in 2h 02m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c72b350/ : SUCCESS in 2h 05m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/597f46d/ : SUCCESS in 2h 14m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/f7fe13f/ : FAILURE in 2h 23m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/24/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/53391cc/ : FAILURE in 2h 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c7c3577c340abbb045cfa9031747ea9e7bfc748c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-12 10:45:55.000000000","message":"Patch Set 25: Commit message was updated.","accounts_in_message":[],"_revision_number":25},{"id":"0496fd5203d38f0af60a3e2f82741ec6db903c28","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-12 10:46:01.000000000","message":"Patch Set 26: Patch Set 25 was rebased","accounts_in_message":[],"_revision_number":26},{"id":"ceb0d44275d0f1c80cf75374689f254aab458ae1","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-06-12 12:02:28.000000000","message":"Patch Set 26: Code-Review+1","accounts_in_message":[],"_revision_number":26},{"id":"039eeab26147bd16be65a151d7412c64531eec3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-12 14:38:43.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7fe5bbb/ : TIMED_OUT in 3h 48m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/109d5ae/ : TIMED_OUT in 3h 46m 57s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c8a8c60/ : TIMED_OUT in 3h 11m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/042cb8f/ : SUCCESS in 2h 17m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/ffd4de9/ : SUCCESS in 2h 23m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b89761d/ : SUCCESS in 2h 25m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/a10447d/ : FAILURE in 2h 58m 13s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c6bf5ea/ : TIMED_OUT in 3h 50m 31s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"9313f774b15670dfd4d8ef8a77f2be2d31f26b09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 20:33:42.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-undercloud-containers/3242863/ : SUCCESS in 1h 47m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6abe6d4/ : FAILURE in 2h 02m 19s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6b04665/ : TIMED_OUT in 3h 27m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4b72194/ : TIMED_OUT in 3h 23m 06s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/26/check/tripleo-ci-fedora-28-standalone/ddbcf43/ : SUCCESS in 2h 03m 54s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-standalone-upgrade/0d7ea80/ : SUCCESS in 2h 20m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-standalone-upgrade-stein/90fba73/ : FAILURE in 2h 33m 26s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-stein/63bfb1e/ : SUCCESS in 1h 53m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-rocky/82c1454/ : SUCCESS in 2h 15m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-queens/4d380e7/ : SUCCESS in 1h 54m 28s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-pike/5979c86/ : SUCCESS in 2h 15m 55s\n- openstack-tox-docs http://logs.openstack.org/24/639324/26/check/openstack-tox-docs/ccdba3c/html/ : SUCCESS in 3m 49s\n- openstack-tox-linters http://logs.openstack.org/24/639324/26/check/openstack-tox-linters/3c9b054/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":26},{"id":"a239848c830479f7ab166df08f25b6fcd8cbd221","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-13 09:44:45.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"3f4fb20e7fe0033ee4210ed88fe232860561a0de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-13 14:01:19.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5e6cb0f/ : FAILURE in 1h 51m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c2ebeae/ : TIMED_OUT in 3h 45m 56s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/73ef532/ : TIMED_OUT in 3h 08m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e4171d7/ : SUCCESS in 2h 12m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/34d44bf/ : SUCCESS in 2h 13m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/9ade834/ : SUCCESS in 2h 23m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/99a5f9f/ : TIMED_OUT in 4h 14m 51s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/26/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/6ec5ac6/ : TIMED_OUT in 3h 49m 04s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"3124f7ae8244b1814ec51e5ff840ed0dfbee8d2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 14:42:24.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-undercloud-containers/b97d3ab/ : SUCCESS in 1h 55m 31s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a969e09/ : FAILURE in 1h 54m 01s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fbf5270/ : TIMED_OUT in 3h 13m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/97e786f/ : TIMED_OUT in 3h 14m 29s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/26/check/tripleo-ci-fedora-28-standalone/52bf2c8/ : FAILURE in 1h 58m 11s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-standalone-upgrade/4fdea76/ : SUCCESS in 2h 07m 14s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-standalone-upgrade-stein/bc4294d/ : SUCCESS in 2h 28m 57s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-stein/0c3526d/ : SUCCESS in 1h 58m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-rocky/73ea505/ : SUCCESS in 2h 21m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-queens/b8a37ee/ : SUCCESS in 1h 59m 03s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/26/check/tripleo-ci-centos-7-containers-multinode-pike/26f6ed3/ : SUCCESS in 2h 21m 43s\n- openstack-tox-docs http://logs.openstack.org/24/639324/26/check/openstack-tox-docs/1937b91/html/ : SUCCESS in 4m 45s\n- tox-molecule http://logs.openstack.org/24/639324/26/check/tox-molecule/de3b94a/tox/reports.html : SUCCESS in 9m 11s\n- openstack-tox-linters http://logs.openstack.org/24/639324/26/check/openstack-tox-linters/598ff4b/ : SUCCESS in 6m 19s","accounts_in_message":[],"_revision_number":26},{"id":"29a57b9e6e4f64f453c62566cfffa12c0d94d39c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-18 12:26:02.000000000","message":"Patch Set 27: Commit message was updated.","accounts_in_message":[],"_revision_number":27},{"id":"cfcb577be65b5aa16224015f88cafbd74fd01250","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-18 12:27:47.000000000","message":"Patch Set 28: Patch Set 27 was rebased","accounts_in_message":[],"_revision_number":28},{"id":"2ab97b491283dac041b3770015debd32d933a9c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 12:43:38.000000000","message":"Patch Set 28: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":28},{"id":"18369b289c65f581c4cd9845ce0815f4665d90f6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 12:43:38.000000000","message":"Patch Set 28: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":28},{"id":"a3e28d83f186d3255e6aee84e56650b604f031cd","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-18 12:46:01.000000000","message":"Patch Set 29: Commit message was updated.","accounts_in_message":[],"_revision_number":29},{"id":"b583950beb73ed27858d5d671d9e4d73f29c7cd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 15:25:11.000000000","message":"Patch Set 29: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"4e6f084b9c948cf5ccb4e3c4ebca7f6518b2b2ae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 15:25:12.000000000","message":"Patch Set 29: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":29},{"id":"b6e9a34d5438c5023b4dd7bdabae7e254416f1f2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-18 15:26:07.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"ce347f21d091a222ba5b6365af4c4cf7e3f9b803","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-18 18:11:43.000000000","message":"Patch Set 29:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5f8c68f/ : FAILURE in 2h 23m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b6e6b33/ : FAILURE in 2h 22m 16s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/fcc5655/ : FAILURE in 1h 12m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b51ac51/ : FAILURE in 2h 16m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/f30dd08/ : SUCCESS in 2h 13m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/50db7f4/ : SUCCESS in 2h 27m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/85a09ad/ : FAILURE in 2h 33m 20s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/29/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/f394093/ : FAILURE in 2h 43m 35s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"acfbec5068b5eeab78a50b4a0eae6442130a0fb8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 02:03:31.000000000","message":"Patch Set 29:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-undercloud-containers/f84b30a/ : SUCCESS in 1h 44m 29s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9344a88/ : FAILURE in 1h 29m 42s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/975bf16/ : FAILURE in 1h 48m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8fb763f/ : FAILURE in 1h 43m 47s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/29/check/tripleo-ci-fedora-28-standalone/3b59a06/ : SUCCESS in 2h 05m 06s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-standalone-upgrade/7958388/ : SUCCESS in 2h 22m 31s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-standalone-upgrade-stein/bb6498c/ : SUCCESS in 2h 02m 27s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-containers-multinode-stein/d966ac2/ : SUCCESS in 1h 56m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-containers-multinode-rocky/5078f29/ : SUCCESS in 2h 14m 22s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-containers-multinode-queens/1a85cd7/ : SUCCESS in 2h 12m 14s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/29/check/tripleo-ci-centos-7-containers-multinode-pike/a421b97/ : SUCCESS in 2h 19m 07s\n- openstack-tox-docs http://logs.openstack.org/24/639324/29/check/openstack-tox-docs/4bf8a8a/html/ : SUCCESS in 4m 04s\n- tox-molecule http://logs.openstack.org/24/639324/29/check/tox-molecule/f8657b8/tox/reports.html : SUCCESS in 9m 31s\n- openstack-tox-linters http://logs.openstack.org/24/639324/29/check/openstack-tox-linters/4d69a04/ : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":29},{"id":"fe0d08524d89d3e6b2df09a5b5041747bcfb085d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-19 05:51:16.000000000","message":"Patch Set 30: Commit message was updated.","accounts_in_message":[],"_revision_number":30},{"id":"108cc5c395d66645e5bcd3e4c0d8fbfb9f0f7b14","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-19 08:24:48.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b6dd77a/ : FAILURE in 1h 12m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/74b1f82/ : FAILURE in 1h 15m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c21cf7c/ : FAILURE in 55m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ecbcdaf/ : SUCCESS in 2h 30m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2387057/ : SUCCESS in 2h 31m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/061d713/ : FAILURE in 53m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/f2c35d4/ : FAILURE in 1h 14m 55s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/30/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/8c69214/ : FAILURE in 1h 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"44215a36aa7ae00ccf3af07980e37b6f70ad657f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-19 08:49:00.000000000","message":"Patch Set 31: Commit message was updated.","accounts_in_message":[],"_revision_number":31},{"id":"113729aac00672bd23c3aaee6c191f69aa5f558a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-19 11:05:19.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/738dcc5/ : FAILURE in 46m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9df299c/ : FAILURE in 45m 52s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/5911807/ : FAILURE in 37m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1a9b08d/ : SUCCESS in 2h 14m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a67e47c/ : SUCCESS in 2h 12m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/71fc4f4/ : FAILURE in 36m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/141a1e3/ : FAILURE in 47m 47s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/31/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/cc41cc6/ : FAILURE in 59m 00s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"de9c22d494f68f8427b71d23b0b0e68498900eb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 11:41:22.000000000","message":"Patch Set 31: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":31},{"id":"4185b955c2368e27244bc82dbf8b252ca3385272","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-24 06:55:52.000000000","message":"Patch Set 32: Commit message was updated.","accounts_in_message":[],"_revision_number":32},{"id":"5518a8ce64b6a0c3eabf5b2a5f948cdc17d7e618","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-24 06:55:59.000000000","message":"Patch Set 33: Patch Set 32 was rebased","accounts_in_message":[],"_revision_number":33},{"id":"cf37169bc4ef17beb6ea4270f0654143f360dc68","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-24 09:53:18.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f87fb3a/ : FAILURE in 1h 51m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a30fc4f/ : SUCCESS in 2h 52m 53s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/a662a33/ : SUCCESS in 1h 19m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d3e5256/ : FAILURE in 2h 20m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/79dda02/ : SUCCESS in 2h 31m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/053ad38/ : SUCCESS in 2h 40m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/6fc2a4f/ : SUCCESS in 2h 55m 22s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0d42500/ : FAILURE in 2h 51m 59s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"17efbd0c17628c2fd7aa24b92667ea11fb64da3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:16:50.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-undercloud-containers/37857e7/ : SUCCESS in 1h 44m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/edcea93/ : FAILURE in 1h 34m 08s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c870a1c/ : SUCCESS in 2h 21m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f16fa3d/ : SUCCESS in 2h 47m 19s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/33/check/tripleo-ci-fedora-28-standalone/952862a/ : SUCCESS in 1h 52m 55s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-standalone-upgrade/a88235d/ : SUCCESS in 2h 31m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-standalone-upgrade-stein/9893cd5/ : SUCCESS in 2h 08m 10s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-containers-multinode-stein/136622d/ : SUCCESS in 1h 54m 58s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-containers-multinode-rocky/7a60651/ : SUCCESS in 2h 31m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-containers-multinode-queens/fa50d48/ : SUCCESS in 2h 01m 28s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/33/check/tripleo-ci-centos-7-containers-multinode-pike/fc9d710/ : SUCCESS in 2h 17m 18s\n- openstack-tox-docs http://logs.openstack.org/24/639324/33/check/openstack-tox-docs/e0be17e/html/ : SUCCESS in 3m 51s\n- tox-molecule http://logs.openstack.org/24/639324/33/check/tox-molecule/45341ef/tox/reports.html : SUCCESS in 11m 54s\n- openstack-tox-linters http://logs.openstack.org/24/639324/33/check/openstack-tox-linters/8414323/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":33},{"id":"ab20372eb33ec8cbaa021ece98147a7c874acdc6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-24 12:52:20.000000000","message":"Patch Set 33:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":33},{"id":"6a1eecef50716c0c2f7473a6e82e81b675c8b23e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-24 16:39:03.000000000","message":"Patch Set 33:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5729ded/ : TIMED_OUT in 3h 42m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c35326b/ : SUCCESS in 2h 51m 46s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/ae22017/ : SUCCESS in 1h 22m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e8c7ae7/ : FAILURE in 2h 24m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/aeaa8ac/ : SUCCESS in 2h 09m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/0dbeb2f/ : SUCCESS in 2h 41m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c515ebc/ : SUCCESS in 3h 04m 08s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/33/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/fe3f8d1/ : FAILURE in 2h 51m 56s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"9d27042f4d04e86ad00cce24ff16589d3a049b49","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-25 07:47:39.000000000","message":"Patch Set 34: Commit message was updated.","accounts_in_message":[],"_revision_number":34},{"id":"35478f9fc7e971b285ace04a59282d5d6ad9779c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:57:40.000000000","message":"Patch Set 34: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":34},{"id":"bce6e829d89ab6847ff30eec3e6ea2b827ad958f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 10:57:40.000000000","message":"Patch Set 34: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":34},{"id":"5f0683ea7c91099164e73a2672f3a6081067d614","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-25 10:59:19.000000000","message":"Patch Set 35: Commit message was updated.","accounts_in_message":[],"_revision_number":35},{"id":"d204889234a22e742724ebab8380eca99870ebcc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 14:04:06.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7d0ba31/ : SUCCESS in 2h 36m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3d1cf6e/ : SUCCESS in 2h 44m 24s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/33c26da/ : SUCCESS in 1h 57m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/6c8213e/ : FAILURE in 2h 12m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/c886fe4/ : FAILURE in 1h 28m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1dbe435/ : FAILURE in 1h 23m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/dfbc5bf/ : FAILURE in 1h 27m 24s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/35/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/9dc8a6b/ : FAILURE in 1h 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"7dd648c5d27ceb2d34cbeff41e48873ba70a3bd2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 23:30:17.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-undercloud-containers/7065e6c/ : SUCCESS in 1h 55m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e2ea72e/ : FAILURE in 1h 36m 24s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/217c464/ : SUCCESS in 2h 16m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2ef3e5e/ : SUCCESS in 2h 29m 45s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/35/check/tripleo-ci-fedora-28-standalone/846a473/ : SUCCESS in 2h 11m 50s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-standalone-upgrade/f8e15b7/ : SUCCESS in 2h 16m 54s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-standalone-upgrade-stein/d60ad80/ : SUCCESS in 1h 57m 44s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-containers-multinode-stein/55792bf/ : SUCCESS in 1h 57m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-containers-multinode-rocky/b6b6271/ : SUCCESS in 2h 13m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-containers-multinode-queens/3c991cb/ : SUCCESS in 2h 06m 07s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/35/check/tripleo-ci-centos-7-containers-multinode-pike/b307676/ : SUCCESS in 2h 24m 50s\n- openstack-tox-docs http://logs.openstack.org/24/639324/35/check/openstack-tox-docs/39acab6/html/ : SUCCESS in 4m 08s\n- tox-molecule http://logs.openstack.org/24/639324/35/check/tox-molecule/22816b2/tox/reports.html : SUCCESS in 13m 07s\n- openstack-tox-linters http://logs.openstack.org/24/639324/35/check/openstack-tox-linters/8b9ed26/ : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":35},{"id":"daa00a1e0247d47dc97de1a5ba06c79e3c6d28a7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-26 03:30:11.000000000","message":"Patch Set 36: Patch Set 35 was rebased","accounts_in_message":[],"_revision_number":36},{"id":"ef1cc466ad04b7c28ba9737e2f1e5b1f8e1c3032","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 07:15:40.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze06.softwarefactory-project.io:7979/99e83da74341417da8a53267a5717188 : POST_FAILURE in 3h 33m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze04.softwarefactory-project.io:7979/eb38ce90aa6045609e944006ca9b9ce0 : POST_FAILURE in 3h 43m 06s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/36/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c04d6de/ : SUCCESS in 1h 23m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/36/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/84a0334/ : SUCCESS in 2h 17m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/36/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5a5ccb9/ : SUCCESS in 2h 18m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/36/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/be38429/ : SUCCESS in 2h 26m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze01.softwarefactory-project.io:7979/8c79b1cb29ab429e9cf04f872d1cc1c5 : POST_FAILURE in 3h 43m 06s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/36/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ca919ad/ : FAILURE in 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"98733494bd4fd2205f6d0ec7063292f2f832c915","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 08:24:58.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-undercloud-containers/1391167/ : SUCCESS in 1h 38m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c51fee9/ : FAILURE in 1h 27m 09s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e82feb9/ : SUCCESS in 2h 35m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/34b765a/ : SUCCESS in 3h 03m 43s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/36/check/tripleo-ci-fedora-28-standalone/cf0641b/ : SUCCESS in 2h 29m 44s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-standalone-upgrade/376bb7f/ : SUCCESS in 2h 44m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-standalone-upgrade-stein/fa48e0d/ : SUCCESS in 2h 14m 27s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-containers-multinode-stein/66ae3f2/ : SUCCESS in 1h 55m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-containers-multinode-rocky/db7d249/ : SUCCESS in 2h 13m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-containers-multinode-queens/a52b4db/ : SUCCESS in 2h 02m 50s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/36/check/tripleo-ci-centos-7-containers-multinode-pike/c0d90a4/ : SUCCESS in 2h 20m 56s\n- openstack-tox-docs http://logs.openstack.org/24/639324/36/check/openstack-tox-docs/db286d1/html/ : SUCCESS in 3m 40s\n- tox-molecule http://logs.openstack.org/24/639324/36/check/tox-molecule/be3eb64/tox/reports.html : SUCCESS in 13m 41s\n- openstack-tox-linters http://logs.openstack.org/24/639324/36/check/openstack-tox-linters/7c9a510/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":36},{"id":"2f648dde81b6e0c2381becef1c64860ba7ba5dc4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-27 08:41:06.000000000","message":"Patch Set 37: Patch Set 36 was rebased","accounts_in_message":[],"_revision_number":37},{"id":"600e5cf4b874ba65f69be2ee4818fab820d67411","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-27 11:51:33.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e4d6ab4/ : FAILURE in 58m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/39464d6/ : FAILURE in 1h 07m 12s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/074fed3/ : FAILURE in 47m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7534576/ : FAILURE in 3h 08m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/7ec0cbe/ : SUCCESS in 2h 49m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/8d1b670/ : FAILURE in 1h 30m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/d2f0a4e/ : FAILURE in 58m 16s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/37/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/93f85e6/ : FAILURE in 23m 42s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"514dcfaf4f69607549647d51196bff22ed6f722e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 15:29:03.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-undercloud-containers/ca24530/ : FAILURE in 1h 26m 54s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1272e26/ : FAILURE in 1h 39m 20s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8bafff5/ : FAILURE in 1h 10m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/48dfd03/ : FAILURE in 1h 27m 02s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/37/check/tripleo-ci-fedora-28-standalone/e61255f/ : SUCCESS in 2h 12m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-standalone-upgrade/a5c61cf/ : SUCCESS in 2h 26m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-standalone-upgrade-stein/8863da1/ : SUCCESS in 2h 03m 33s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-containers-multinode-stein/38b6e81/ : SUCCESS in 2h 03m 47s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-containers-multinode-rocky/0b027d4/ : SUCCESS in 2h 06m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-containers-multinode-queens/2f32dc9/ : SUCCESS in 1h 49m 43s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/37/check/tripleo-ci-centos-7-containers-multinode-pike/3867f10/ : SUCCESS in 2h 10m 52s\n- openstack-tox-docs http://logs.openstack.org/24/639324/37/check/openstack-tox-docs/9233868/html/ : SUCCESS in 3m 45s\n- tox-molecule http://logs.openstack.org/24/639324/37/check/tox-molecule/d1c2829/tox/reports.html : SUCCESS in 13m 51s\n- openstack-tox-linters http://logs.openstack.org/24/639324/37/check/openstack-tox-linters/45fbce6/ : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":37},{"id":"92b7fc26fc3e57da79ac0806134760018ae0addb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-01 06:55:11.000000000","message":"Patch Set 38: Patch Set 37 was rebased","accounts_in_message":[],"_revision_number":38},{"id":"7b8af2b9c0670de0c16402142b779f5a126b66fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 09:36:19.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-undercloud-containers/d0e944e/ : SUCCESS in 1h 39m 55s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5247735/ : FAILURE in 1h 25m 08s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b45c432/ : SUCCESS in 2h 13m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b5e928b/ : FAILURE in 2h 02m 21s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/38/check/tripleo-ci-fedora-28-standalone/41b9ef9/ : SUCCESS in 2h 02m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-standalone-upgrade/9b488e3/ : SUCCESS in 2h 25m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-standalone-upgrade-stein/05c20e7/ : SUCCESS in 2h 34m 49s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-stein/3142f9f/ : SUCCESS in 1h 55m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-rocky/27d9893/ : SUCCESS in 2h 08m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-queens/3904207/ : SUCCESS in 2h 23m 13s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-pike/84fc994/ : SUCCESS in 2h 16m 08s\n- openstack-tox-docs http://logs.openstack.org/24/639324/38/check/openstack-tox-docs/e92bf17/html/ : SUCCESS in 4m 12s\n- tox-molecule http://logs.openstack.org/24/639324/38/check/tox-molecule/3b43cf9/tox/reports.html : SUCCESS in 14m 01s\n- openstack-tox-linters http://logs.openstack.org/24/639324/38/check/openstack-tox-linters/f52baeb/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":38},{"id":"ea93e42a2d799ad9d0dffa7b46e259320df19fa0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 09:41:45.000000000","message":"Patch Set 38: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":38},{"id":"1389ad67b74b1051eb3857e70e31d99fac7a727b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-01 09:46:37.000000000","message":"Patch Set 38:\n\nrecheck","accounts_in_message":[],"_revision_number":38},{"id":"ba13665973ce41496eaa23c7707a8a7a262ac8ae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 13:17:48.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5b05924/ : FAILURE in 59m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ba145c5/ : FAILURE in 1h 05m 16s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/458b8dc/ : FAILURE in 49m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/f87d02b/ : SUCCESS in 2h 35m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/ae2c69f/ : SUCCESS in 2h 19m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b548040/ : SUCCESS in 3h 28m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/9c1bc72/ : FAILURE in 57m 58s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/a70b1c8/ : FAILURE in 25m 30s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"3c79ab9477ed5e9ead589c298188fa0ff1b9331f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 13:43:32.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-undercloud-containers/196abe6/ : FAILURE in 1h 24m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5b96c25/ : FAILURE in 2h 12m 03s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b97cead/ : FAILURE in 1h 28m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/998d275/ : FAILURE in 1h 29m 59s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/38/check/tripleo-ci-fedora-28-standalone/e9acc7e/ : FAILURE in 1h 20m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-standalone-upgrade/c353ec1/ : SUCCESS in 2h 35m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-standalone-upgrade-stein/1ae1112/ : SUCCESS in 2h 03m 10s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-stein/199938a/ : SUCCESS in 2h 04m 47s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-rocky/b0fb521/ : SUCCESS in 2h 09m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-queens/f12bdc6/ : SUCCESS in 2h 07m 26s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-pike/24ff4d8/ : SUCCESS in 2h 24m 02s\n- openstack-tox-docs http://logs.openstack.org/24/639324/38/check/openstack-tox-docs/b914bec/html/ : SUCCESS in 3m 40s\n- tox-molecule http://logs.openstack.org/24/639324/38/check/tox-molecule/4376070/tox/reports.html : SUCCESS in 13m 44s\n- openstack-tox-linters http://logs.openstack.org/24/639324/38/check/openstack-tox-linters/625046b/ : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":38},{"id":"ab86f0b5e663a8b10c9ce08c85b1113117727b51","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-01 15:46:48.000000000","message":"Patch Set 38:\n\nrecheck","accounts_in_message":[],"_revision_number":38},{"id":"b4f066e7b198f77b2f76d2bfe1da0efce4c06600","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 19:59:32.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/31d52dd08bb146379812b8f1baef2258 : POST_FAILURE in 3h 41m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze04.softwarefactory-project.io:7979/f52f4141a2a64ad6b6ace8d50dc642a9 : POST_FAILURE in 3h 45m 26s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 finger://ze05.softwarefactory-project.io:7979/63374f0e6b504893b7d36b934c870c37 : POST_FAILURE in 2h 25m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/aace653/ : FAILURE in 2h 58m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze03.softwarefactory-project.io:7979/67301a3598254c3291c217762618b817 : POST_FAILURE in 3h 14m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/3f9061a/ : FAILURE in 2h 16m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze04.softwarefactory-project.io:7979/b990237d8b51444b9e40261294b65618 : POST_FAILURE in 4h 07m 13s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/38/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/c5314db/ : FAILURE in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"fadc99d453eb08461b533203ba476b9d7cb82dab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 22:09:16.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-undercloud-containers/c58c060/ : SUCCESS in 1h 31m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2fe7e4d/ : FAILURE in 2h 03m 33s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a9970ff/ : SUCCESS in 2h 55m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2ac9659/ : SUCCESS in 3h 13m 56s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/38/check/tripleo-ci-fedora-28-standalone/fa9763e/ : SUCCESS in 2h 06m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-standalone-upgrade/acf140b/ : SUCCESS in 2h 08m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-standalone-upgrade-stein/08f1669/ : SUCCESS in 2h 01m 47s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-stein/fb1bce6/ : SUCCESS in 2h 12m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-rocky/6949d83/ : SUCCESS in 2h 03m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-queens/b885ac7/ : SUCCESS in 2h 04m 44s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/38/check/tripleo-ci-centos-7-containers-multinode-pike/d2c3e71/ : SUCCESS in 2h 38m 02s\n- openstack-tox-docs http://logs.openstack.org/24/639324/38/check/openstack-tox-docs/90707b1/html/ : SUCCESS in 4m 12s\n- tox-molecule http://logs.openstack.org/24/639324/38/check/tox-molecule/444e323/tox/reports.html : SUCCESS in 15m 33s\n- openstack-tox-linters http://logs.openstack.org/24/639324/38/check/openstack-tox-linters/5a59f01/ : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":38},{"id":"118a1808aab148072b1c7d0dac4eeff81cdee3ef","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-02 14:20:34.000000000","message":"Patch Set 39: Patch Set 38 was rebased","accounts_in_message":[],"_revision_number":39},{"id":"537333090390cef716b6c7ce314d2b065b9f8e00","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 16:41:40.000000000","message":"Patch Set 39: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":39},{"id":"e4138de0cc4c0d779ddb139414e9a5b095ab8c85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 16:47:22.000000000","message":"Patch Set 39: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":39},{"id":"af2e1f51a98e36622dcb0ca01e70b8796a66db37","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-03 07:14:57.000000000","message":"Patch Set 40: Patch Set 39 was rebased","accounts_in_message":[],"_revision_number":40},{"id":"5e322a63c9f046c38d23be673daec76ec1fab18d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-03 09:45:06.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3097a17/ : FAILURE in 23m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/77f1de4/ : FAILURE in 23m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/4e26fc3/ : FAILURE in 14m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8713004/ : SUCCESS in 2h 27m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/55da609/ : SUCCESS in 2h 27m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/cfdb14b/ : FAILURE in 1h 44m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/f1937ac/ : FAILURE in 23m 14s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/3e92298/ : FAILURE in 25m 33s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"713c1b1d9dc952e9847fdba08e971801dd535011","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 10:36:21.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-undercloud-containers/7928eca/ : FAILURE in 28m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d317377/ : FAILURE in 1h 21m 37s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6d7d1c1/ : FAILURE in 32m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b94a5f2/ : FAILURE in 22m 01s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/40/check/tripleo-ci-fedora-28-standalone/7a65796/ : FAILURE in 43m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-standalone-upgrade/22de334/ : FAILURE in 2h 12m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-standalone-upgrade-stein/957f07c/ : FAILURE in 1h 38m 58s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-stein/05939f9/ : FAILURE in 2h 29m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-rocky/008d4f9/ : FAILURE in 2h 27m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-queens/70de0c4/ : FAILURE in 1h 48m 38s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-pike/3f9a21d/ : FAILURE in 2h 20m 39s\n- openstack-tox-docs http://logs.openstack.org/24/639324/40/check/openstack-tox-docs/4d6d8b8/html/ : SUCCESS in 3m 58s\n- tox-molecule http://logs.openstack.org/24/639324/40/check/tox-molecule/6ed2dac/tox/reports.html : SUCCESS in 12m 56s\n- openstack-tox-linters http://logs.openstack.org/24/639324/40/check/openstack-tox-linters/eeacf95/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":40},{"id":"8f973c79c91e9daec429566ca93bd85fe39ca0b5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-05 07:30:12.000000000","message":"Patch Set 40:\n\nrecheck","accounts_in_message":[],"_revision_number":40},{"id":"a08bf133775932506aaa70fe81526b76d285e566","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 10:33:11.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-undercloud-containers/e172215/ : SUCCESS in 1h 41m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/b6ca375/ : FAILURE in 1h 55m 13s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8aa28fa/ : SUCCESS in 2h 47m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b9b26cf/ : SUCCESS in 2h 45m 39s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/40/check/tripleo-ci-fedora-28-standalone/a2415bc/ : FAILURE in 51m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-standalone-upgrade/bb80463/ : SUCCESS in 2h 12m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-standalone-upgrade-stein/cd16fd2/ : SUCCESS in 2h 25m 43s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-stein/4e5f381/ : SUCCESS in 2h 06m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-rocky/27494c5/ : SUCCESS in 2h 15m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-queens/6858476/ : SUCCESS in 2h 08m 50s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/40/check/tripleo-ci-centos-7-containers-multinode-pike/b2761d8/ : SUCCESS in 2h 23m 31s\n- openstack-tox-docs http://logs.openstack.org/24/639324/40/check/openstack-tox-docs/ec15232/html/ : SUCCESS in 5m 01s\n- tox-molecule http://logs.openstack.org/24/639324/40/check/tox-molecule/5757cda/tox/reports.html : SUCCESS in 17m 09s\n- openstack-tox-linters http://logs.openstack.org/24/639324/40/check/openstack-tox-linters/b182dea/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":40},{"id":"6c81619f52c6b758fc8f8bcdd5abe15870b3e69a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 10:40:31.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/569786b/ : SUCCESS in 2h 49m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/67a844e/ : SUCCESS in 3h 08m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/1aab702/ : SUCCESS in 1h 30m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7f853ba/ : SUCCESS in 2h 18m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/75fa000/ : SUCCESS in 2h 15m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/3d894a9/ : SUCCESS in 2h 44m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/06e93cb/ : SUCCESS in 3h 06m 16s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/40/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/af80b17/ : FAILURE in 2h 40m 51s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"46f8c0038f6dba3f3a2beaa8a93452414541d77f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-08 06:43:58.000000000","message":"Patch Set 41: Patch Set 40 was rebased","accounts_in_message":[],"_revision_number":41},{"id":"430d57edf3c0d6611e3a662d07ffa58c8ce256c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 09:44:46.000000000","message":"Patch Set 41: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":41},{"id":"b8df18c21994097b06972c79089b6a7a876aee74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 09:45:08.000000000","message":"Patch Set 41: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":41},{"id":"1ab575c77edf584e17d13de0cf8de6ee1d39b870","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-08 10:02:44.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"caf8760054003051a9c00529993e7d441dafb486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 11:21:55.000000000","message":"Patch Set 41:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":41},{"id":"30e5ec73f234b475ece336d6e278fbc6a82cfe61","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 11:21:55.000000000","message":"Patch Set 41:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":41},{"id":"09aa31d75bcd8ef781e76a69d6f3ae27186cdf26","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-08 11:22:52.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"b9cc2156bc332dbfdde72b497550044595da94bb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-08 14:30:31.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b19aaef/ : SUCCESS in 2h 41m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c3f848c/ : SUCCESS in 2h 27m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b4dbcf8/ : SUCCESS in 2h 24m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/a229f2a/ : RETRY_LIMIT in 5m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/487449a/ : RETRY_LIMIT in 4m 44s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze01.softwarefactory-project.io:7979/a064cb6edfab4782860e00426518a64d : RETRY_LIMIT in 2h 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"7a810c9e8f963da74c5ee63a561711bcea7ce721","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 20:31:12.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-undercloud-containers/824a8c3/ : FAILURE in 1h 06m 22s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/8b0b8e5/ : FAILURE in 2h 31m 58s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1d66002/ : SUCCESS in 2h 35m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6fbafd2/ : FAILURE in 1h 16m 18s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/41/check/tripleo-ci-fedora-28-standalone/fc19b63/ : FAILURE in 2h 20m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade/cddef50/ : SUCCESS in 2h 49m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade-stein/61cef08/ : SUCCESS in 2h 58m 34s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-stein/c8e562e/ : SUCCESS in 3h 08m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-rocky/22236ff/ : SUCCESS in 2h 52m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-queens/e46c3a8/ : SUCCESS in 2h 11m 46s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-pike/2b5c7fa/ : SUCCESS in 2h 25m 50s\n- openstack-tox-docs http://logs.openstack.org/24/639324/41/check/openstack-tox-docs/2e0af48/html/ : SUCCESS in 4m 25s\n- tox-molecule http://logs.openstack.org/24/639324/41/check/tox-molecule/4f891ca/tox/reports.html : SUCCESS in 15m 08s\n- openstack-tox-linters http://logs.openstack.org/24/639324/41/check/openstack-tox-linters/0b6761a/ : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":41},{"id":"600535d848c5b78ac02c7b37b902a0aa798512d6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-09 06:40:59.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"5191e3148dd7ee864d4a1fcbb33be136f5ec9f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 10:09:57.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-undercloud-containers/4eb804b/ : SUCCESS in 1h 38m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/6f27fe5/ : FAILURE in 2h 02m 30s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bb576bc/ : SUCCESS in 2h 15m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4070502/ : SUCCESS in 3h 09m 37s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/41/check/tripleo-ci-fedora-28-standalone/4391d1d/ : SUCCESS in 1h 59m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade/f7bcfe8/ : SUCCESS in 2h 40m 10s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade-stein/798f045/ : SUCCESS in 2h 05m 53s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-stein/95f4c4c/ : SUCCESS in 2h 30m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-rocky/e8ea836/ : SUCCESS in 2h 05m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-queens/205a320/ : SUCCESS in 2h 02m 45s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-pike/6ebb408/ : SUCCESS in 2h 19m 28s\n- openstack-tox-docs http://logs.openstack.org/24/639324/41/check/openstack-tox-docs/84f6889/html/ : SUCCESS in 6m 16s\n- tox-molecule http://logs.openstack.org/24/639324/41/check/tox-molecule/85e9f07/tox/reports.html : SUCCESS in 13m 55s\n- openstack-tox-linters http://logs.openstack.org/24/639324/41/check/openstack-tox-linters/b31d2b2/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":41},{"id":"a0dda1b690d35706177555d850d502f51eb8190e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 10:53:21.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/37000a2492e24cde93033c3c59564edf : POST_FAILURE in 3h 50m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze07.softwarefactory-project.io:7979/e2b029ed32a24f27815c922116608c31 : POST_FAILURE in 3h 52m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze04.softwarefactory-project.io:7979/e4dc6f2fb5044152ba093adf4cda4bba : POST_FAILURE in 3h 34m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0c2607f/ : SUCCESS in 2h 32m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze02.softwarefactory-project.io:7979/680c66c320af45d781e50ac20b61a390 : POST_FAILURE in 3h 53m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze05.softwarefactory-project.io:7979/cbbaff4b7de742e0b1a14757202bb9a5 : POST_FAILURE in 4h 05m 49s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze06.softwarefactory-project.io:7979/b3fb89b8edaa4c24999012021a7d0527 : POST_FAILURE in 4h 09m 19s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"2cc1d965213fbb17ce8b55cfba0ea6d4e41037bc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-15 09:42:48.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"a314882c3bfda09ac7b055fa9ac0c044c3bf2629","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 12:53:59.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4a5dea4/ : SUCCESS in 2h 57m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0aa5cf8/ : SUCCESS in 2h 42m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/79612d6/ : FAILURE in 2h 21m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6eae0ac/ : SUCCESS in 2h 24m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5a39aff/ : SUCCESS in 2h 37m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/877ef69/ : SUCCESS in 3h 09m 17s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/232fbce/ : FAILURE in 3h 01m 51s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"64c67ab6ef61023c09430da501f8e92e7880d4c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 13:33:01.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-undercloud-containers/94d8afa/ : SUCCESS in 1h 52m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2ab3eb3/ : TIMED_OUT in 3h 48m 26s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3672c95/ : SUCCESS in 2h 42m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/85d88c2/ : FAILURE in 2h 41m 30s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/41/check/tripleo-ci-fedora-28-standalone/0ffc124/ : SUCCESS in 2h 10m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade/0378f6f/ : FAILURE in 2h 08m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-stein/5146a95/ : SUCCESS in 2h 18m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-rocky/be1fd2f/ : SUCCESS in 2h 26m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-queens/c7d0ef9/ : SUCCESS in 2h 05m 53s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-pike/db2b235/ : SUCCESS in 2h 30m 12s\n- openstack-tox-docs http://logs.openstack.org/24/639324/41/check/openstack-tox-docs/41659b6/html/ : SUCCESS in 4m 31s\n- openstack-tox-molecule http://logs.openstack.org/24/639324/41/check/openstack-tox-molecule/0d513e8/tox/reports.html : SUCCESS in 14m 07s\n- openstack-tox-linters http://logs.openstack.org/24/639324/41/check/openstack-tox-linters/0ef144e/ : SUCCESS in 8m 58s","accounts_in_message":[],"_revision_number":41},{"id":"bdb9bf68b13e70f100c7c6071d60bf1e42367068","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 06:24:30.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"fb96cbe773b97749819db62479d3058a2a55cfdd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 09:33:49.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/63eddf7/ : SUCCESS in 2h 47m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c49ca9d/ : SUCCESS in 2h 57m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1619b70/ : SUCCESS in 2h 02m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/088b9db/ : SUCCESS in 2h 29m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/48745f3/ : SUCCESS in 2h 57m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/74a1b46/ : FAILURE in 2h 09m 07s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/b754789/ : FAILURE in 3h 07m 58s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"f7be41373f8fa94c8bb27ff03973d9bdb3c71c18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 10:05:44.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-undercloud-containers/a673bce/ : SUCCESS in 1h 55m 01s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c180e2a/ : FAILURE in 2h 17m 07s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8a1ca16/ : SUCCESS in 2h 38m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f1a2471/ : FAILURE in 3h 09m 22s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/24/639324/41/check/tripleo-ci-fedora-28-standalone/f451cda/ : SUCCESS in 1h 40m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade/5b5fd66/ : FAILURE in 2h 04m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-stein/6a74e16/ : SUCCESS in 2h 24m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-rocky/9aaa46d/ : SUCCESS in 2h 22m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-queens/3d25aa5/ : SUCCESS in 2h 09m 56s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-pike/b3d2f2b/ : SUCCESS in 2h 34m 21s\n- openstack-tox-docs http://logs.openstack.org/24/639324/41/check/openstack-tox-docs/14e9132/html/ : SUCCESS in 4m 18s\n- openstack-tox-molecule http://logs.openstack.org/24/639324/41/check/openstack-tox-molecule/47f2d80/tox/reports.html : SUCCESS in 19m 47s\n- openstack-tox-linters http://logs.openstack.org/24/639324/41/check/openstack-tox-linters/6a3e489/ : SUCCESS in 10m 04s","accounts_in_message":[],"_revision_number":41},{"id":"756003def824da4858b974a69010fc239df00bfe","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-08 13:01:28.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"3dfd30ffcd49d2030e61c276962d07bfe8d2a5e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 14:11:07.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/626a0ad/ : FAILURE in 1h 06m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze02.softwarefactory-project.io:7979/03ff2d8efdf94ad2ba09749f3a13144e : RETRY_LIMIT in 8s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/66f6883/ : RETRY_LIMIT in 7m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/7b6054e/ : RETRY_LIMIT in 5m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze02.softwarefactory-project.io:7979/7a3c60890ffc4bcc8f0f35cd59db783b : RETRY_LIMIT in 7s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze01.softwarefactory-project.io:7979/5537dda5948f4b12bbba78c6480508e7 : RETRY_LIMIT in 7s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/855df81/ : RETRY_LIMIT in 6m 34s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-rhel-8-standalone-rdo/d01f398/ : FAILURE in 14m 54s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"f8af285815285830d9caf5d42d777d7ee5855b34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 20:18:35.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-undercloud-containers/9ce3a59/ : SUCCESS in 1h 33m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/f874f85/ : FAILURE in 1h 55m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode/8c1e06d/ : SUCCESS in 2h 33m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e359239/ : SUCCESS in 2h 34m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/24c7ab3/ : SUCCESS in 2h 34m 04s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/24/639324/41/check/tripleo-ci-fedora-28-standalone/b9c4108/ : FAILURE in 1h 27m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade/cc1993d/ : SUCCESS in 2h 12m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-stein/240de0c/ : SUCCESS in 2h 03m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-rocky/5619191/ : SUCCESS in 2h 14m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-queens/0367f73/ : SUCCESS in 2h 02m 52s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-pike/5af6030/ : TIMED_OUT in 3h 13m 40s\n- openstack-tox-docs https://logs.opendev.org/24/639324/41/check/openstack-tox-docs/0d5163d/html/ : SUCCESS in 4m 01s\n- openstack-tox-molecule https://logs.opendev.org/24/639324/41/check/openstack-tox-molecule/a2c9c90/tox/reports.html : SUCCESS in 13m 17s\n- openstack-tox-linters https://logs.opendev.org/24/639324/41/check/openstack-tox-linters/dfe4160/ : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":41},{"id":"3087ff00bb5e3558cce54c8fd82cab1b31b4e000","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-09 10:21:28.000000000","message":"Patch Set 41:\n\nrecheck","accounts_in_message":[],"_revision_number":41},{"id":"db3b5d1ed26ce290119ce9e403beccde60ff6bb4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 13:14:20.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/eab598e/ : SUCCESS in 2h 32m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bbdd0ca/ : SUCCESS in 2h 46m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/bf6d186/ : FAILURE in 2h 32m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/346e453/ : SUCCESS in 2h 19m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/ced58b9/ : SUCCESS in 2h 42m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/81a90c3/ : SUCCESS in 2h 45m 47s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/df0955c/ : FAILURE in 2h 49m 46s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-rhel-8-standalone-rdo/a74e8bc/ : FAILURE in 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"99dbaf1ca3eda6ddc4c51859e084c1cca12baabe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 13:55:31.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-undercloud-containers/dca340e/ : SUCCESS in 1h 42m 06s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a3b4f93/ : FAILURE in 1h 54m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode/fb130b4/ : SUCCESS in 2h 41m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/226cd8c/ : SUCCESS in 2h 05m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d0569ed/ : SUCCESS in 2h 29m 22s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/24/639324/41/check/tripleo-ci-fedora-28-standalone/7a92eec/ : SUCCESS in 1h 36m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-standalone-upgrade/52c0212/ : SUCCESS in 2h 50m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-stein/84fe5db/ : SUCCESS in 2h 10m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-rocky/09e771f/ : SUCCESS in 2h 31m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-queens/249ce33/ : SUCCESS in 2h 50m 19s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/24/639324/41/check/tripleo-ci-centos-7-containers-multinode-pike/2a77d02/ : SUCCESS in 2h 21m 40s\n- openstack-tox-docs https://logs.opendev.org/24/639324/41/check/openstack-tox-docs/d6b32bf/html/ : SUCCESS in 4m 28s\n- openstack-tox-molecule https://logs.opendev.org/24/639324/41/check/openstack-tox-molecule/df3c8f9/tox/reports.html : SUCCESS in 14m 28s\n- openstack-tox-linters https://logs.opendev.org/24/639324/41/check/openstack-tox-linters/a1e63a2/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":41},{"id":"0383c7b1285241b54313298132585b52069ed6eb","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-25 10:12:59.000000000","message":"Patch Set 41:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":41},{"id":"7f7bdffaa9320142fd2f1923b1144489d3bb7547","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-25 13:09:36.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8fe8698/ : FAILURE in 2h 34m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3f90104/ : SUCCESS in 2h 40m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/12b5a59/ : SUCCESS in 2h 08m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/982cde5/ : SUCCESS in 2h 19m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/51a5d96/ : SUCCESS in 2h 55m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/41/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/16faabc/ : FAILURE in 2h 45m 14s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"6fcca1daa57e8a8d80b508b87fa381ee80ff3cb0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-25 13:20:40.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"fcee1ec69f195d5d91b3ced6d475b6c9ef03e677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-25 16:05:36.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/564d85cbb14149eaa831717021b04fe6 : SUCCESS in 1h 21m 26s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/313ae89774e647cfbfae85c305569093 : FAILURE in 1h 44m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/475ffc9976b74cb5b6c2a7b30d17643a : SUCCESS in 2h 12m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/916bd3575adc48a8aa28dac774068c15 : SUCCESS in 1h 59m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/cc2d0d83a08849a2b0133f86450faf76 : SUCCESS in 1h 59m 38s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/24c485990e1b4c08a6606e4418a74b27 : SUCCESS in 1h 53m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/926c4066fd1747a18dd705f400b904a5 : SUCCESS in 2h 33m 37s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/594627b0df5b4429bc5ccc4921110c76 : SUCCESS in 2h 25m 29s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/86d4fc10102a4f48ab7e7228a4d592b8 : SUCCESS in 2h 00m 08s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/10ac51f9b12240aeadd4b6460c25cd6d : SUCCESS in 2h 01m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393aa4273c5643f48ba8411efe7539f7 : SUCCESS in 4m 37s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c62fb85ef9ce4853a3953a6f6ccd5976 : SUCCESS in 13m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1a8c8b112738441bb9bca6deea14069b : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":42},{"id":"7850e3b6b72015b7431b8ef906ec6886888592df","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-25 16:18:38.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/42/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b2ddff5/ : SUCCESS in 2h 33m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/42/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/24b2689/ : SUCCESS in 2h 45m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/42/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0072107/ : SUCCESS in 2h 15m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/42/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5d968c8/ : SUCCESS in 2h 08m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/42/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/edf81d1/ : SUCCESS in 2h 47m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/42/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/0fd9818/ : FAILURE in 2h 56m 15s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"5edb15d168b1e2b9887f6b9b7e398578ccb32cd9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-26 05:34:43.000000000","message":"Patch Set 43: Commit message was updated.","accounts_in_message":[],"_revision_number":43},{"id":"d9903f91fc167e9ffa9c4015443edb775b3511e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 08:13:16.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/81ca4af01b5846e1ad8b3f7b3130522d : SUCCESS in 1h 27m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3a2c95dd668240fa9185b0c9621a3f01 : FAILURE in 1h 40m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5c7b4bd58f8046eeb5a2c5b6bf9988ff : SUCCESS in 2h 33m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cd6b6689b18a4184a07449e5f18aa0e4 : SUCCESS in 2h 24m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e060eb031d8344429b795915b10fac7a : SUCCESS in 2h 15m 31s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/3a7508f8ffa5495b8e3dd12c4a560d2b : SUCCESS in 1h 47m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/4343d9910e374410aecc4acd57c13ded : SUCCESS in 2h 29m 02s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/84df53143ad740129b1ebca58a61e45e : SUCCESS in 2h 15m 52s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/cf51b7d3dc784bd09f507089d64962eb : SUCCESS in 2h 19m 16s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/927ee22311004aee881285a922c42a34 : SUCCESS in 2h 14m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39d0adfa9b9a44d59a97d62545091d52 : SUCCESS in 4m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0cc8e5ed940d4a44a5cec6b7e21ce34c : SUCCESS in 15m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/efbf164d359242f9af73af40dd38bded : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":43},{"id":"ad1f1fb71a289874281704f1e9822150e4008d07","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 09:17:34.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/43/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6aa8505/ : SUCCESS in 2h 36m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/43/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5fa12a0/ : SUCCESS in 2h 44m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/43/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e58678d/ : SUCCESS in 2h 22m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/43/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e6a3fe7/ : FAILURE in 2h 55m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/43/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/586fe27/ : TIMED_OUT in 3h 41m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/43/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/69d3757/ : FAILURE in 2h 58m 10s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"deabf6dc256e87ccc8862488e6acfd0fc3527c71","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-26 10:18:17.000000000","message":"Uploaded patch set 44.","accounts_in_message":[],"_revision_number":44},{"id":"2ee150a76193fc1e7c77e3bd3d0c352414f0584a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-26 10:19:59.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"c993e0ea236db8968cfdaa6c7e97158c547c8b9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 13:11:05.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a591ff2e83f140c7adc664d7685fd8c5 : SUCCESS in 1h 28m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/140ba7ac8de848de897d8fde9c0a0252 : FAILURE in 1h 41m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/70a44a10c3a44fe8b445a7498ffb1e68 : SUCCESS in 2h 37m 24s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/91f4079d2bca43489024b5ca3b335679 : SUCCESS in 2h 06m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/050ca0c13ea94efc9fa3b1ce0c1cd42a : SUCCESS in 2h 18m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a2dd24ff9c404f849310103848438f77 : SUCCESS in 1h 49m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/99a04791ea9543898d5b8ecff9eebbba : SUCCESS in 2h 44m 53s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/e9cd6ce463b848fbb73ab87a3c6fc1bb : SUCCESS in 2h 31m 16s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/4d6dd87b10164cb882d39148b10a0045 : SUCCESS in 2h 21m 53s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/4f398edfd50c4880850dee88b8781f76 : SUCCESS in 2h 12m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a0f6189aa7e4f269c3c9d898b39d69c : SUCCESS in 4m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4ab052c07c3e41eba79c6fd40ae8eb54 : SUCCESS in 18m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c46126602f87428eaf663f9ea85e42dd : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":45},{"id":"ff44cef7e39e9f9a8363c012bc29b199965c2573","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 13:25:41.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/45/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/777435f/ : SUCCESS in 2h 29m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/45/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/83b3388/ : SUCCESS in 2h 41m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/45/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7cb9b0c/ : FAILURE in 2h 17m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/45/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b3ad80b/ : SUCCESS in 2h 22m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/45/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/8a32c6b/ : FAILURE in 3h 04m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/45/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/af1a76d/ : FAILURE in 2h 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"18b03b4e75ef8215299601e1dbaf8e79afab3720","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-26 13:35:52.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"3baa6de06a4fcab94024ac5aef4a22caad085b8b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-26 13:56:38.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"ce9a7094df64e2dd7c5e6a03aa1bda19d86fff65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 17:37:17.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ebdf6a336367447a8878f2d80e26ce0c : SUCCESS in 1h 49m 52s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/026610239458441a95b66cff5c713b6a : FAILURE in 2h 12m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d01d84eb1a994477811da666e87d27b2 : SUCCESS in 2h 39m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2100c13b451148d0bcc2525df7f94a7b : SUCCESS in 2h 31m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a492041ea0a6400f99313377d5d7213b : SUCCESS in 2h 58m 56s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/88bd1d9550d34d5b869e872e09945ca3 : SUCCESS in 2h 08m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/19ab964ac88a48198789038e6077ad98 : TIMED_OUT in 3h 31m 46s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/3146bc9c30954764aebf2275b3c20a23 : SUCCESS in 2h 24m 57s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/facb9af605034b359552e6e266a33828 : SUCCESS in 2h 21m 10s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/5892d0aec86a47fd85bc2190033ed2bc : SUCCESS in 2h 34m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6913fc638da64f6d9dc5ecb2f6b00629 : SUCCESS in 5m 00s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2fa689697b124e86995b4a63f1d18de2 : SUCCESS in 15m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/67926f3a3c8941dba98c3970f2e999b4 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":47},{"id":"dc0fa9a45f9ec53267c4308e690538cdc4561908","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 17:48:54.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9e11f98/ : FAILURE in 3h 26m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze04.softwarefactory-project.io:7979/51e2036305f14bfb9518e8576f1c270b : POST_FAILURE in 3h 48m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3638a0b/ : SUCCESS in 2h 17m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/bfc0e15/ : SUCCESS in 2h 16m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze03.softwarefactory-project.io:7979/cc9298931b6b42c08ce15e5ddb1e2f71 : POST_FAILURE in 3h 46m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e1b65c3/ : FAILURE in 28m 43s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"a66f1007b38a1f6dc6f4428cbff6044fcced9fbc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-26 17:54:31.000000000","message":"Patch Set 47:\n\nrecheck","accounts_in_message":[],"_revision_number":47},{"id":"67e337b6d91959d791b5f581c793b5f64379275a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 20:55:23.000000000","message":"Patch Set 47:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b91592a/ : SUCCESS in 2h 58m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b3c95da/ : SUCCESS in 2h 58m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b0d431d/ : SUCCESS in 2h 12m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/04bd60d/ : FAILURE in 2h 48m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/828e9dd/ : SUCCESS in 2h 49m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/47/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/b18c135/ : FAILURE in 1h 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"0d1f9d35a12b74265e5d8fdb994f0a79d78bad41","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 20:56:55.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8b78446093d44e88a5564e9d0ce4058e : SUCCESS in 1h 52m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2d4cbe13ae854283ad33e902ed535916 : FAILURE in 2h 14m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/92e3a42be198452fb4d53b677522297d : SUCCESS in 2h 56m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/05d8f0814d2e4357a66e996f2cb1a9be : SUCCESS in 2h 22m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2b1184124617421d98c31158bb3af80b : SUCCESS in 2h 47m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/41f64ceb98f64306a07f5a6d84abb1a8 : FAILURE in 1h 28m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/ab04b8df4d134adb929a4f92494e04d1 : SUCCESS in 2h 47m 15s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/3441886f3e0f4f39bc797eccb613f867 : SUCCESS in 2h 42m 32s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/53e8725cb95c4620a68408dcaa1c5dc9 : SUCCESS in 2h 14m 59s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/da79d256b73b4e99a41cbcf872605b64 : SUCCESS in 2h 36m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c5e705f2ca45e68a8ea95e191fa650 : SUCCESS in 6m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/236506f9153a4ba79b7c052360678ac1 : SUCCESS in 14m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e96241bfe14d4388a0ac256fc973491a : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":47},{"id":"9f0d7317a9f2fef6ac3ffa2e09cc9c3ca44e2fdf","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-27 03:56:31.000000000","message":"Patch Set 48: Patch Set 47 was rebased","accounts_in_message":[],"_revision_number":48},{"id":"5d24b754664a3fdc31eadc7893373b7cec5a28d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 06:51:18.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6015bc1693da41b494db86eeb778282c : SUCCESS in 1h 39m 22s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cd48f95884c944f3bcdab06878f4e33a : FAILURE in 2h 01m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/5662f56283844e2f95df9b2f3962fd24 : SUCCESS in 2h 52m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/92ffbc8ccb074fc694d08c4b52a61f98 : SUCCESS in 2h 18m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b714df2e18974d0892f373515dc8749c : SUCCESS in 2h 35m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e324a55ac0e3426f85737bed8e2bdc40 : SUCCESS in 2h 21m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/ef0b103ba2474fc8a9226c0f0717ecb5 : SUCCESS in 2h 27m 50s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/cdeccde240ff4c20ab8aa290b38da7c7 : SUCCESS in 2h 31m 43s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/7be775ec6d2e4025a55de96c6b4faed8 : SUCCESS in 2h 06m 01s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/06d3b00e8dc9407fbb706ddc5c9d3452 : SUCCESS in 2h 20m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc33addb0b3a43a99d85250989df3dfa : SUCCESS in 4m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b0121ac739d24232b35eb42bafadc03b : SUCCESS in 15m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c5cdb0b70c0249bcac3f217f36a6a196 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":48},{"id":"8c2b11acee9f31ffda29599f997f2f7ccd2e6190","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 07:15:48.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/48/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/41f6a07/ : SUCCESS in 3h 15m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/48/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/5a72f4c/ : SUCCESS in 3h 13m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/48/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3df9f2e/ : FAILURE in 2h 10m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/48/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0b37d93/ : SUCCESS in 2h 11m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/48/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/d6fcc20/ : SUCCESS in 3h 17m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/48/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/26e9f65/ : FAILURE in 3h 17m 45s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"a4bad9ba937360d6f424ea869c8d9c14891aa219","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-27 11:23:04.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"e57a0ed61c143ca878b6738438ba207a6a1cd78a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 14:25:44.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/49/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6d5fc27/ : SUCCESS in 2h 51m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/49/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/45d654b/ : SUCCESS in 2h 54m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/49/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/f65f6b5/ : FAILURE in 2h 07m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/49/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/27c2515/ : SUCCESS in 2h 24m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/49/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/cc00146/ : SUCCESS in 2h 48m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/49/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/531a8a9/ : FAILURE in 3h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"bfd9f5763cbd0529ff6bd627b57428d8dc22764e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 14:31:27.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a18a7268ae7c4ee3928dcd3bf368bbdc : SUCCESS in 1h 41m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/e0327e81e1344e9fac7230808bafb22f : FAILURE in 1h 59m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/0227360ded774786b3b9db0b17732672 : SUCCESS in 2h 41m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7aa1291ab02948d492b7fc3d470566bd : SUCCESS in 2h 27m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5dda1253b4ab4750bd1a6ae633cccfa1 : SUCCESS in 2h 50m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b33d2fc7aefd41f489298d60a8e0be0c : SUCCESS in 2h 29m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/56f9fef6c769425ba9ee1b56a05a1773 : SUCCESS in 2h 50m 54s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/d6657fa8431148e09e1470dce425010a : SUCCESS in 2h 29m 36s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3683e786652940db8d48482f1578fa70 : SUCCESS in 1h 59m 41s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/92fa529c10dd42a693ff00e0db315a94 : SUCCESS in 2h 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b30a5de7bfc4758b5a0707f18e58d73 : SUCCESS in 7m 56s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/917e8e06aa1d48ad8830aae8325d5463 : SUCCESS in 15m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/874327eddbfd4980863d386f23fef9e9 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":49},{"id":"03678c09e7b69365c1c1fb282456e8328955d156","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-28 05:07:37.000000000","message":"Patch Set 50: Published edit on patch set 49.","accounts_in_message":[],"_revision_number":50},{"id":"e98e1326c3c57938151ac81e858ed29c4121d5bb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-28 05:07:45.000000000","message":"Patch Set 51: Patch Set 50 was rebased","accounts_in_message":[],"_revision_number":51},{"id":"0075219cfb7cbeb1e160ac1d031ecc35bb713c62","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 08:06:13.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c5494d18ae5c4f059140f8df4170beb9 : SUCCESS in 1h 51m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cb320118c20847dd8b466130eaca1f9a : FAILURE in 2h 45m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/210cf7fd52354e688fd853a3ae906986 : SUCCESS in 2h 53m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fbe6e3a910174a4dbbbd20af920920ff : SUCCESS in 2h 23m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1fca70fe3cb44d18ac17145c718dcf39 : SUCCESS in 2h 39m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/de090832dfcb4d93bc17e6e66c03384c : SUCCESS in 2h 11m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/ba002edbb34143b38a97fa34edaf2501 : FAILURE in 1h 07m 43s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/e8b8227e7aae4dc0b5310dd66288630f : SUCCESS in 2h 40m 02s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/2c34d855ee9f4fbd8b4a570c0806055e : SUCCESS in 2h 32m 23s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/b9360343a4ab450f8bde972d51fe37fc : SUCCESS in 2h 20m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a343d3ca699400983d1c8dcef95f708 : SUCCESS in 6m 11s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a2adcfad45be4045b11d23add9967159 : SUCCESS in 17m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/739cf83634674139b647cad22274721b : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":51},{"id":"3f92844422a2a29b8192ee9bd7345e0fb1fe16e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-28 08:17:54.000000000","message":"Patch Set 51: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/51/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0af5377/ : SUCCESS in 3h 02m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/24/639324/51/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2fdd19d/ : SUCCESS in 2h 44m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/51/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/4431dce/ : FAILURE in 2h 22m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/51/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/4dc43a6/ : SUCCESS in 2h 26m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/51/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c1f15bc/ : FAILURE in 3h 08m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/51/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/a98973b/ : FAILURE in 3h 08m 46s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"ef68f5ee88a64c0554fa3dcdf38f4eb017f8b4e5","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-28 10:03:02.000000000","message":"Patch Set 51:\n\nrecheck","accounts_in_message":[],"_revision_number":51},{"id":"f2d2d7b83fc2c08efc416f8bd53cb913f440f630","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-28 10:09:40.000000000","message":"Patch Set 52: Published edit on patch set 51.","accounts_in_message":[],"_revision_number":52},{"id":"f81fae914c54620e289b895bfd9304d350a31ba3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 13:30:36.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2c572bdc3b1e4af9b474371fff5ef7b5 : SUCCESS in 1h 51m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/eb34599444ec418f9dd205fe8efe32c4 : FAILURE in 2h 01m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/543d6a94d3644bae8517d40aa3d46c04 : SUCCESS in 2h 53m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e33afcf764df4747b7b1562f72e1c4cd : SUCCESS in 3h 08m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f352504d3514493d8a6f871420f12b6a : SUCCESS in 2h 50m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a804923900764c65a5f170103e3d4f64 : SUCCESS in 2h 23m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/9611fca4699b4383aa42118fbc810ca0 : SUCCESS in 2h 25m 05s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/4422d9bf1e384008801e0a286d55c6a2 : SUCCESS in 2h 21m 21s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/24eae6fc21dd4e80bac6a5362355c762 : SUCCESS in 2h 24m 48s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/b08b6f8c003846e99d66c44f34354f1d : SUCCESS in 2h 28m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734168d612f348e6b6d088ac6d6f4385 : SUCCESS in 5m 11s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1e59a956c1d1463384f0754c7f135894 : SUCCESS in 14m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a8499a90b3244fb2b36ebcefa4311626 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":52},{"id":"da66458dbeaf9605099cea5b37f0bb8bddf53745","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-28 14:09:25.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/52/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1e67689/ : SUCCESS in 2h 53m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze04.softwarefactory-project.io:7979/891f1ca545c44cdeac5e396a358040b2 : POST_FAILURE in 3h 51m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/52/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e101f3b/ : SUCCESS in 2h 22m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/52/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e72669f/ : SUCCESS in 2h 21m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze03.softwarefactory-project.io:7979/8e42bfc3547d418b86e4f17fa4c65d4b : POST_FAILURE in 3h 56m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze05.softwarefactory-project.io:7979/32a2bc4f601e43fb8282d1d459e8a927 : POST_FAILURE in 3h 57m 50s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"90d62a483d0417d6dad2cbbd770667cf26eefc17","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-29 09:08:05.000000000","message":"Patch Set 53: Published edit on patch set 52.","accounts_in_message":[],"_revision_number":53},{"id":"eece88dd0ea6b478ec02a2872d9cf4d9756c2232","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 12:01:21.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/394caa3/ : FAILURE in 1h 41m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8c75549/ : SUCCESS in 2h 08m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cce064c/ : FAILURE in 2h 51m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/905d917/ : FAILURE in 2h 48m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/f8f200a/ : FAILURE in 2h 00m 21s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"7d73d2bcd9f195772e8f9e0ed043422e6c992835","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 12:32:43.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/fa04317e0cfd481cb0507e397c93cac4 : SUCCESS in 1h 42m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/36fb049812474856a6c6838e93ff6898 : FAILURE in 1h 57m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cc67cad909834886a6d821ba99291346 : SUCCESS in 2h 41m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f10a48adebbf45c0b7641a026df15072 : SUCCESS in 2h 34m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ea839d16febf4c08b12c7864b54edbd1 : SUCCESS in 2h 54m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b3968708ba5545d6b57bb784eae5556d : SUCCESS in 2h 02m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/a961e0339b7e467f835cd5e9f1e09fbe : SUCCESS in 2h 42m 12s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/3a0f115f702e4b2fadfee8462d4689b9 : SUCCESS in 2h 19m 26s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/0fb7611938814da093b174bc84b02f14 : SUCCESS in 2h 12m 52s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/559982f23ac0449888d00f8a3cf126c2 : SUCCESS in 2h 26m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/061ae8f1a21a45e9be021f14d5cc9d25 : SUCCESS in 5m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/53fc278928d943f59d31000c1bd923e7 : SUCCESS in 16m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/44ccb851b328448283daf1904d53ca1b : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":53},{"id":"d0fd86b01d7e05306e0db42f15631be6f291919c","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-01 08:57:22.000000000","message":"Patch Set 53:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":53},{"id":"1fa2f1363a3cd37b37a6f8e9c1f331dda56b22e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-01 10:34:18.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ab5bdd7/ : FAILURE in 17m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/4f5d13f/ : FAILURE in 1h 29m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3876ee6/ : FAILURE in 1h 31m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b6b6a52/ : FAILURE in 19m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/432c3a3/ : FAILURE in 20m 03s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"b7bba9d06e63688e77f7c77fdc6156eb7ab41d85","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-05 11:41:34.000000000","message":"Patch Set 53:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":53},{"id":"5a143b8dea9455317abc772ca9542fd27ef23620","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-05 14:46:41.000000000","message":"Patch Set 53:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6dc74ac/ : FAILURE in 2h 50m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d0bd8ba/ : SUCCESS in 2h 11m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cb74dcd/ : SUCCESS in 2h 10m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/d0e2d39/ : SUCCESS in 2h 50m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/53/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/68987b5/ : FAILURE in 3h 04m 21s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"87c132ed4575ceec225dc814a1d77db634ed9204","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-06 07:32:11.000000000","message":"Patch Set 54: Published edit on patch set 53.","accounts_in_message":[],"_revision_number":54},{"id":"8a0972702e91dab1acf66ca8d824800b7596ca54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-06 10:23:16.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/54/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2dca89f/ : SUCCESS in 2h 49m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/54/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/625b8d4/ : SUCCESS in 2h 14m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/54/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5d54807/ : SUCCESS in 2h 18m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/54/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1107557/ : SUCCESS in 2h 48m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/54/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/eafb5d4/ : FAILURE in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"095d5cc9e1be27aca1a033c5eff0cf95e019857c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-06 10:25:07.000000000","message":"Patch Set 55: Published edit on patch set 54.","accounts_in_message":[],"_revision_number":55},{"id":"cb9054fc76a54dc1fb694751e99435b775cdb71b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-06 13:21:49.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/de7f356/ : FAILURE in 1h 49m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7693a5f/ : SUCCESS in 2h 08m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/df8cf8c/ : SUCCESS in 2h 15m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/ff149f8/ : SUCCESS in 2h 55m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/16c0967/ : FAILURE in 2h 05m 30s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"b6c829255e3d88b0ee5a6c33ccad14c96a5c557a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-06 14:07:43.000000000","message":"Patch Set 55:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":55},{"id":"18430e791b54fe56c50283db28af0edd39999360","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 14:28:44.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e8a65d0ea30f4b55b27e177c1585ef7c : SUCCESS in 1h 46m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/c37baf59542f4fcb9102c4acd587f401 : FAILURE in 2h 06m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/dec58618a9c74bba82f2725c4e7e8354 : SUCCESS in 2h 57m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/522b06768db44537aa51494d546fbbdf : SUCCESS in 2h 11m 51s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5e92aa1f359544e08af88cbee97bfa44 : SUCCESS in 2h 19m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/247ef654ce51428eb61b2ec1a4d923eb : SUCCESS in 1h 59m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/f2969bafece2453cae60c9e239bd18a6 : SUCCESS in 2h 12m 40s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/2840132130ee45c994352e3aca00bacd : SUCCESS in 2h 29m 28s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/8dd52ac0d47a447c8a4376e4ba091999 : SUCCESS in 2h 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab58db4dde2b4d44919a4facf0162298 : SUCCESS in 4m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/bc2117683f0242d7ab0d4303e3b89d08 : SUCCESS in 11m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/eeab52a6e03e4d968b5faba5daa3c057 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":55},{"id":"d245c22f6cc646fbc071430b7bcbd9eefaec0181","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-06 17:22:28.000000000","message":"Patch Set 55:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/293ea75/ : SUCCESS in 2h 55m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/4fae95c/ : FAILURE in 2h 07m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b2425d2/ : SUCCESS in 2h 13m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/6bbb57c/ : FAILURE in 2h 55m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/55/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/36a8fc5/ : FAILURE in 3h 13m 10s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"011b8661f7f722eab867dcdd043d6b99f4bf4aa6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-09-09 10:53:40.000000000","message":"Patch Set 56: Published edit on patch set 55.","accounts_in_message":[],"_revision_number":56},{"id":"fba977552272dec673852cab9e6bbd9a2a7f3d59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 13:45:09.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/c502146573a64ea8a61fb59cd97865da : SUCCESS in 1h 44m 25s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/bfe7525c11334dffa6f2fd43453ee0f7 : FAILURE in 1h 56m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7262fb6b16aa479287af3436b4d645c1 : SUCCESS in 2h 41m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b770d7377cc64e7489ef5ac4f05b088e : SUCCESS in 2h 21m 18s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/e0d91e77ee33426490800874d2193054 : SUCCESS in 2h 36m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c5ae6a591da54ff7b6d053930c4eb85a : SUCCESS in 2h 18m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/418762ad55214130b26318bd9151ebb1 : SUCCESS in 2h 34m 36s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/fc8356b17af44159a2a1c79d929861cf : SUCCESS in 2h 22m 17s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/b52004b85eee49e08535c0c882e988c0 : SUCCESS in 2h 16m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/172a022818f04d87af24e524c0224155 : SUCCESS in 4m 10s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/406a82e5c58a42ed8b6646ca15c4c8c2 : SUCCESS in 11m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/83cbb105fdf042ca9b35b2be51b7b84e : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":56},{"id":"885b7310f7cfbfead6d34478b988449a494da5cb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-09 14:24:46.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/56/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/217b178/ : SUCCESS in 3h 30m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/56/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/87affc9/ : FAILURE in 2h 59m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/56/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/99743c6/ : SUCCESS in 3h 17m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/56/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/4df8594/ : FAILURE in 3h 29m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/56/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2415189/ : FAILURE in 2h 52m 25s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"352c84b5b67993da0644ea71a079e4fe1c003fb8","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-11-13 18:38:55.000000000","message":"Patch Set 57: Patch Set 56 was rebased","accounts_in_message":[],"_revision_number":57},{"id":"f93e900201b177c4c8cf79a0a17bbdf20e46a2a4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-13 22:08:31.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/57/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8d9c0ea/ : SUCCESS in 2h 47m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/57/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/47c94ea/ : SUCCESS in 3h 27m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/57/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/9154e01/ : SUCCESS in 2h 20m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/57/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/beea088/ : SUCCESS in 2h 59m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/57/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/43c4f25/ : FAILURE in 3h 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"e5572c62d8887390ff8c0310ce6d80278a2771df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 22:24:31.000000000","message":"Patch Set 57: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6e14f904e51c4229b3959354fce8b8f1 : SUCCESS in 1h 45m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3b1b640b3d7c469ebda5599e7ffe44ce : FAILURE in 1h 32m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/acd8291d313541b59e600f19ecfb4799 : SUCCESS in 2h 26m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/34bc8a1ef5b24d72b16297c1f718b2a5 : SUCCESS in 2h 01m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/199121949b06401885353b4cc49dd278 : SUCCESS in 2h 23m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/967f1d7a221a4760b5e12015dfc7ba0d : SUCCESS in 3h 01m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/67388974a0334418a0a7bbada251a79b : SUCCESS in 2h 44m 48s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/1c8a37c13fd74fad9261aff80c1cbc69 : TIMED_OUT in 3h 36m 58s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/710cbe02eefa47c6925c157a002e790f : SUCCESS in 2h 48m 23s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/97736726a99841f3a29c244443daaa67 : FAILURE in 40m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d6ee067f9c14cd0bd0ad2053f12fbcd : SUCCESS in 5m 03s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e2130e10c3a1462ba8e44fedbf00bb89 : SUCCESS in 18m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/996e3cd881f84b15874a2d03eac18e09 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":57},{"id":"bc45304289c8a38c970f923d4eb0cfede330abec","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-14 02:55:17.000000000","message":"Uploaded patch set 58.","accounts_in_message":[],"_revision_number":58},{"id":"d38274c2ec4a5548d2a812077960e32f2ffb7fd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 05:56:50.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/dfa441780a854ffc83fad24530b9ab5d : SUCCESS in 1h 52m 46s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cad5c871f1594a4594397a30f6678498 : FAILURE in 1h 22m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d3233b08075f4f9bb61056ed603c72ce : SUCCESS in 2h 47m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0e643e474abc47a8922c1198d5279644 : SUCCESS in 2h 00m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/883d38681dcb4ac085df9727fdcabe3a : SUCCESS in 2h 22m 04s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7a5daaf9386d40bb9e9ac4dfb540b562 : SUCCESS in 2h 07m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3d96e1816bb44e6ba45cecf54e55c186 : FAILURE in 36m 18s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/c26379b1e8b24165b43c2fad7b71a2be : SUCCESS in 2h 06m 00s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/7c692f6fc41e486eaf32aece38fbbea4 : SUCCESS in 2h 16m 10s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/fee1d10bfd68437dbea453a049e07439 : SUCCESS in 2h 02m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eada313d56744b7adf882f76206682d : SUCCESS in 3m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6f0dff19aa6d4f08a78ce979ca49ccd5 : SUCCESS in 12m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/85f3e50720604d99afb0bdd4052b0df4 : FAILURE in 5m 58s","accounts_in_message":[],"_revision_number":58},{"id":"004227925e9ad381aa38aca489e41d7ba766885b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 06:11:50.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/58/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3d37091/ : SUCCESS in 2h 43m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/58/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b25a761/ : SUCCESS in 3h 15m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/58/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/e93b640/ : SUCCESS in 2h 18m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/58/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e6de9f1/ : SUCCESS in 2h 45m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/58/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/d51f2ef/ : FAILURE in 3h 02m 16s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"4a1d1c40911bd30add48ee72c866bc59a1cecf37","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-14 06:41:54.000000000","message":"Uploaded patch set 59.","accounts_in_message":[],"_revision_number":59},{"id":"0cc6ca12477934865f86e7731eb1e77e84b54966","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 09:24:06.000000000","message":"Patch Set 59: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/e8170a5fe0544ff48fabdc264ef7036e : SUCCESS in 1h 27m 24s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8a54ba3bea814ed7be78d283755512d6 : FAILURE in 1h 21m 17s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/92c673d929db4948b1ae73c2d344cd4c : SUCCESS in 2h 18m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4e1bc54ee1c945ebae6ba6584a4aaaf3 : SUCCESS in 2h 06m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/d1703445cffe40ebb77d7717b128594e : SUCCESS in 2h 04m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/61910e03868a4871919a7614fcdcbd09 : SUCCESS in 1h 54m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/47e2dba2988d4d14a7bb0f56bb73544a : SUCCESS in 2h 15m 12s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/26647434cddd4f149b64ead0dbfc9922 : SUCCESS in 2h 06m 47s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/d384232245d94bea8818fd4606c1d8ef : FAILURE in 1h 55m 48s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/aba75bbafcd345c9bec52d244e82b9c2 : SUCCESS in 1h 58m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9667249483a84ac4a88b43e52ba3cae5 : SUCCESS in 5m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fac5bb6faf57433b952f9a543bcbb363 : SUCCESS in 14m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/71043b8fbb92456c8e96fd5f5ceddce2 : FAILURE in 8m 45s","accounts_in_message":[],"_revision_number":59},{"id":"dfa407dcb2913fd9a5a3a8879d0adde59a86df4d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 10:34:12.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/59/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e900e40/ : SUCCESS in 3h 41m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/59/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/997c334/ : SUCCESS in 2h 09m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/59/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/019d50e/ : SUCCESS in 2h 25m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/59/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/6d83ec0/ : SUCCESS in 3h 04m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/59/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/e017676/ : FAILURE in 3h 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"3a1e173865c067875c1fb9d0377dcffa1b4298f9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-14 10:47:30.000000000","message":"Uploaded patch set 60.","accounts_in_message":[],"_revision_number":60},{"id":"7b3a0ca42024ae1f8ac6ea5b7d0a46f0883c7317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 13:31:07.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/3ace314ead9d4d26b72a12b461a98ae5 : POST_FAILURE in 2h 27m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/2cd07a5774d44d769982d5474317b9ad : FAILURE in 1h 25m 17s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4257e7f116844fd88cab84db92771cb6 : SUCCESS in 2h 42m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/28ea8afde92f4cdc818101b50afae22c : SUCCESS in 2h 20m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/59e2cc54f3e142fc88d2db6448026cc4 : SUCCESS in 2h 09m 04s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/1926f416d2124a05a4e95873d540b128 : SUCCESS in 1h 55m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f753884f1b0a4ea396dc8eacf542dda1 : SUCCESS in 2h 29m 08s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/11285a4ac2624e21b1f90fe8afbd01e6 : SUCCESS in 2h 14m 14s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/314c76e305594e0ab0c5dce183067a2c : SUCCESS in 2h 20m 54s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/2a3d917cdca84f4fa45d6fbb380acf16 : SUCCESS in 2h 15m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/878b2e05a22f4ae3acd61e11c1813a78 : SUCCESS in 5m 08s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1ec7f145bafe48579a7ba0798f91af7c : SUCCESS in 13m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/926484d91f65478f817d1306ca4dcd99 : FAILURE in 7m 15s","accounts_in_message":[],"_revision_number":60},{"id":"e0f610784e15c5c50618bb63acdc1800eafa8207","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 14:02:39.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/60/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2a2bd81/ : SUCCESS in 2h 42m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/60/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0330041/ : SUCCESS in 2h 01m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/60/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/20aaa47/ : SUCCESS in 2h 15m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/60/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/eadae95/ : SUCCESS in 3h 06m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/60/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/d51f5fe/ : FAILURE in 3h 01m 49s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"0c90eb475477453fbfe4d1e572e960d8c423b57b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-14 14:35:32.000000000","message":"Uploaded patch set 61.","accounts_in_message":[],"_revision_number":61},{"id":"d60206019035e86874ac67c8b6ec32f8cba29bff","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-14 14:36:01.000000000","message":"Uploaded patch set 62.","accounts_in_message":[],"_revision_number":62},{"id":"953a93d32fe483eaf33582dc7ee1e4c1e66f3d51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 18:27:15.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/94dfda3b10fa49d587a5665794fff919 : SUCCESS in 2h 08m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/db45135c0bea467e87a1c80e0854f4d4 : FAILURE in 1h 44m 13s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ad9d4ed3e7e14fbb8670161b96c129dd : SUCCESS in 2h 48m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a239f9a8b968446988d14e68b812ff40 : SUCCESS in 2h 21m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6353d7c1e8bb4ac4a8a0ef09ba740348 : SUCCESS in 2h 43m 14s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a4a590cacbe44f769d9b40b736a98e42 : SUCCESS in 2h 22m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/b90368a5f49b42ebb1e6578171ee1d88 : TIMED_OUT in 3h 21m 47s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/2536e23aa0f640dfad1b0df6640f9d17 : SUCCESS in 2h 30m 31s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/2690800241c241f19878dcae40c9726b : SUCCESS in 3h 09m 32s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/699489c04bbe42a6b04eca0878597173 : SUCCESS in 2h 08m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6043ad926b33494cad7febf2771c016a : SUCCESS in 6m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a46465b5965c4682b664832c8520d6bf : SUCCESS in 17m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5d779152587a4d23959638bad9846afa : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":62},{"id":"4f13083b6438bb7f065b47775ef59bcaa96350f1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-14 19:41:21.000000000","message":"Patch Set 62: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/62/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a6812f0/ : TIMED_OUT in 5h 01m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/62/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/dd64819/ : FAILURE in 3h 28m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/62/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/fa50905/ : SUCCESS in 2h 30m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/62/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/a8711f5/ : SUCCESS in 2h 53m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/62/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/2100110/ : FAILURE in 3h 19m 15s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"fbf4a6a4b7e8f05bd6cff5cf833caf296776f763","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-15 03:50:05.000000000","message":"Patch Set 63: Patch Set 62 was rebased","accounts_in_message":[],"_revision_number":63},{"id":"c733670e2dc93a6442bd6f6b8a72156f35c587f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 06:54:45.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/acca7e0f445b4411bd8e86f9a4f69955 : FAILURE in 1h 35m 15s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/b9d2463564e242e18f88af2504888814 : FAILURE in 1h 20m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/329860920830441d859784a8d93aa038 : SUCCESS in 2h 59m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4153c6dcdc6c446ca1687793ab38d43c : SUCCESS in 2h 23m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/09c8e148772244d99868f22c093059a7 : SUCCESS in 2h 35m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/cbd65749bc7047aa8e1ff34e8f54a9a1 : SUCCESS in 2h 05m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/71a8c56cc559409bbdd2e8e98d02f18a : SUCCESS in 2h 39m 08s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/7aa116ce4e9f450499c907b3ae7bd69d : SUCCESS in 2h 36m 45s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/a163099f9a01471a9587c515abc76f47 : SUCCESS in 2h 40m 36s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/eac3be02708a4424a086ee3a06cfdf5b : SUCCESS in 2h 13m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19512f18e5444b8ba10147e60944d295 : SUCCESS in 4m 43s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5386983b9f414494b48ebe4b560df318 : SUCCESS in 13m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/86a197e0c26f49bc9442e81621155e62 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":63},{"id":"f3f985c1f4949d360ff35f1fb87aefaf1e8ecef9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-15 07:52:23.000000000","message":"Patch Set 63: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/24/639324/63/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/7fa0079/ : SUCCESS in 2h 42m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/24/639324/63/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/925cedd/ : SUCCESS in 2h 13m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/24/639324/63/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cfaf4ed/ : SUCCESS in 4h 00m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/24/639324/63/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/35ad043/ : SUCCESS in 2h 53m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/24/639324/63/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/ea9c7ba/ : FAILURE in 3h 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"c75bb689861967e6e2e7b846fcdbfc174e0c89b2","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-04-30 09:27:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":63}],"current_revision_number":63,"current_revision":"d600b59808c76d74e94b67eb1b4858f2e9cbe3fa","revisions":{"219734d73be755c54fc22794a0bbf1ca903d61a8":{"kind":"REWORK","_number":1,"created":"2019-02-26 12:32:56.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/1"}}},"commit":{"parents":[{"commit":"275361c63beb5fa210305ebc974c41b24ef0696d","subject":"Fix freeipa role for quickstart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/275361c63beb5fa210305ebc974c41b24ef0696d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:32:24.000000000","tz":120},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nDepends-On: https://review.openstack.org/#/c/639323/\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/219734d73be755c54fc22794a0bbf1ca903d61a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/219734d73be755c54fc22794a0bbf1ca903d61a8"}]},"branch":"refs/heads/master"},"d29ec1bbcf7a399a730c876ff5612c30d1350da7":{"kind":"REWORK","_number":2,"created":"2019-02-26 16:20:09.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/2"}}},"commit":{"parents":[{"commit":"275361c63beb5fa210305ebc974c41b24ef0696d","subject":"Fix freeipa role for quickstart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/275361c63beb5fa210305ebc974c41b24ef0696d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 16:19:52.000000000","tz":120},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nDepends-On: https://review.openstack.org/#/c/639323/\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d29ec1bbcf7a399a730c876ff5612c30d1350da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d29ec1bbcf7a399a730c876ff5612c30d1350da7"}]},"branch":"refs/heads/master"},"7830ba33e62675029617cc049f09f2309125dab6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-02-27 12:47:37.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/3"}}},"commit":{"parents":[{"commit":"275361c63beb5fa210305ebc974c41b24ef0696d","subject":"Fix freeipa role for quickstart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/275361c63beb5fa210305ebc974c41b24ef0696d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-27 12:47:34.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/639651\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7830ba33e62675029617cc049f09f2309125dab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7830ba33e62675029617cc049f09f2309125dab6"}]},"branch":"refs/heads/master"},"9bfb57c6394836faa99725dcd21197834a2d4601":{"kind":"REWORK","_number":4,"created":"2019-02-28 10:18:15.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/4"}}},"commit":{"parents":[{"commit":"275361c63beb5fa210305ebc974c41b24ef0696d","subject":"Fix freeipa role for quickstart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/275361c63beb5fa210305ebc974c41b24ef0696d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-02-28 10:18:10.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/639651\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9bfb57c6394836faa99725dcd21197834a2d4601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9bfb57c6394836faa99725dcd21197834a2d4601"}]},"branch":"refs/heads/master"},"0604709ee59cb231ae12686cd6235ef4ec68d7ff":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-03-01 05:13:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/5"}}},"commit":{"parents":[{"commit":"08bc574294a9b3494025569d5a5fab3791db47d3","subject":"Ensure cirros-0.3.6 is referenced consistently","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/08bc574294a9b3494025569d5a5fab3791db47d3"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-03-01 05:13:09.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/639651\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0604709ee59cb231ae12686cd6235ef4ec68d7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0604709ee59cb231ae12686cd6235ef4ec68d7ff"}]},"branch":"refs/heads/master"},"925057b1c62517b2341ef85c297f29b84d7dfbc3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-03-02 09:27:40.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/6"}}},"commit":{"parents":[{"commit":"a0507051d5a51037a308b11686a55565f2adecdf","subject":"Merge \"Ignore novajoin tests in job that don\u0027t support it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a0507051d5a51037a308b11686a55565f2adecdf"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-03-02 09:27:40.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/639651\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/925057b1c62517b2341ef85c297f29b84d7dfbc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/925057b1c62517b2341ef85c297f29b84d7dfbc3"}]},"branch":"refs/heads/master"},"f67a18a0673e871f097cf126a2711026ddd0b085":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-03-11 22:45:10.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/7"}}},"commit":{"parents":[{"commit":"003f78cdfb12b1cb6ea7c907b624312a27871412","subject":"fix the f28 container namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/003f78cdfb12b1cb6ea7c907b624312a27871412"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-03-11 22:45:10.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/639651\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f67a18a0673e871f097cf126a2711026ddd0b085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f67a18a0673e871f097cf126a2711026ddd0b085"}]},"branch":"refs/heads/master"},"e9d2ad97bac75a1d9e1948f282273a02cc0e796b":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2019-03-19 05:43:51.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/8"}}},"commit":{"parents":[{"commit":"003f78cdfb12b1cb6ea7c907b624312a27871412","subject":"fix the f28 container namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/003f78cdfb12b1cb6ea7c907b624312a27871412"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-03-19 05:43:40.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/644482/\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e9d2ad97bac75a1d9e1948f282273a02cc0e796b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e9d2ad97bac75a1d9e1948f282273a02cc0e796b"}]},"branch":"refs/heads/master"},"ea41b5e7c4417c0c317778b1cf542adf7472b969":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-03-21 14:50:42.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/9"}}},"commit":{"parents":[{"commit":"bf4da290a4c787c633f06bed402a3fbcf39e7003","subject":"Merge \"Add libvirt env for qemu-guest task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bf4da290a4c787c633f06bed402a3fbcf39e7003"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-03-21 14:50:42.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/644482/\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ea41b5e7c4417c0c317778b1cf542adf7472b969"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ea41b5e7c4417c0c317778b1cf542adf7472b969"}]},"branch":"refs/heads/master"},"eb387d9fb948a3a93ec782288c3e572114a35e30":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-03-28 14:08:22.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/10"}}},"commit":{"parents":[{"commit":"f94a1a7988e56ab1f5dc13395cb543ea8ba9797e","subject":"Improve master release config reusability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f94a1a7988e56ab1f5dc13395cb543ea8ba9797e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-03-28 14:08:22.000000000","tz":0},"subject":"WIP: run novajoin tempest tests","message":"WIP: run novajoin tempest tests\n\nprevious: https://review.openstack.org/#/c/639323/\n\nDepends-On: https://review.openstack.org/#/c/644482/\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb387d9fb948a3a93ec782288c3e572114a35e30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb387d9fb948a3a93ec782288c3e572114a35e30"}]},"branch":"refs/heads/master"},"55c6eb8b3afee8b32e896b1a1cb75a7b86f6c7cd":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-03-31 21:39:21.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/11"}}},"commit":{"parents":[{"commit":"f94a1a7988e56ab1f5dc13395cb543ea8ba9797e","subject":"Improve master release config reusability","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f94a1a7988e56ab1f5dc13395cb543ea8ba9797e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-03-31 21:39:13.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/55c6eb8b3afee8b32e896b1a1cb75a7b86f6c7cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/55c6eb8b3afee8b32e896b1a1cb75a7b86f6c7cd"}]},"branch":"refs/heads/master"},"c135eccb2d3c70c3f7e181d86a69917d42f5f0c6":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-03-31 21:39:28.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/12"}}},"commit":{"parents":[{"commit":"e321fd3f0c18d7b106889822c7765a268ed3c046","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e321fd3f0c18d7b106889822c7765a268ed3c046"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-03-31 21:39:28.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c135eccb2d3c70c3f7e181d86a69917d42f5f0c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c135eccb2d3c70c3f7e181d86a69917d42f5f0c6"}]},"branch":"refs/heads/master"},"51167212ac4cf6003bafbd41e7d043b8a0292844":{"kind":"REWORK","_number":13,"created":"2019-04-02 07:59:08.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/13"}}},"commit":{"parents":[{"commit":"e321fd3f0c18d7b106889822c7765a268ed3c046","subject":"Merge \"Replace openstack.org git:// URLs with https://\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e321fd3f0c18d7b106889822c7765a268ed3c046"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-02 07:58:57.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/51167212ac4cf6003bafbd41e7d043b8a0292844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/51167212ac4cf6003bafbd41e7d043b8a0292844"}]},"branch":"refs/heads/master"},"663820c0858c55fb94146396df51363023887b87":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-04-02 11:28:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/14"}}},"commit":{"parents":[{"commit":"f163c03aa800785e022889e036be0e61f9ab7a8e","subject":"Merge \"Improve virt-resize UX\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f163c03aa800785e022889e036be0e61f9ab7a8e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-04-02 11:28:31.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/663820c0858c55fb94146396df51363023887b87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/663820c0858c55fb94146396df51363023887b87"}]},"branch":"refs/heads/master"},"a920b4bb730a40f2c81e4cb1cab2367c93d10222":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-04-04 08:06:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/15"}}},"commit":{"parents":[{"commit":"1da70f5141ca1fd54203a1f4fc2e35e1208add1b","subject":"Add promotion-testing-hash-stein.yml for Fedora-28","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1da70f5141ca1fd54203a1f4fc2e35e1208add1b"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-04-04 08:06:32.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a920b4bb730a40f2c81e4cb1cab2367c93d10222"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a920b4bb730a40f2c81e4cb1cab2367c93d10222"}]},"branch":"refs/heads/master"},"6f0f757573c908df0ae95c2977ee14e1c2b11d59":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-04-08 06:30:08.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/16"}}},"commit":{"parents":[{"commit":"58b8136a7bf109b75c52c7fb20f6a67389ee7867","subject":"align fs020/21 settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/58b8136a7bf109b75c52c7fb20f6a67389ee7867"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-04-08 06:30:08.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6f0f757573c908df0ae95c2977ee14e1c2b11d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6f0f757573c908df0ae95c2977ee14e1c2b11d59"}]},"branch":"refs/heads/master"},"8851d47a5c87a1ef181192aa9d3484e98974fc13":{"kind":"REWORK","_number":17,"created":"2019-04-08 09:38:07.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/17"}}},"commit":{"parents":[{"commit":"58b8136a7bf109b75c52c7fb20f6a67389ee7867","subject":"align fs020/21 settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/58b8136a7bf109b75c52c7fb20f6a67389ee7867"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-08 09:38:03.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8851d47a5c87a1ef181192aa9d3484e98974fc13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8851d47a5c87a1ef181192aa9d3484e98974fc13"}]},"branch":"refs/heads/master"},"a7cff628eb02506d851752a8fa45447fcbc89610":{"kind":"REWORK","_number":18,"created":"2019-04-12 08:43:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/18"}}},"commit":{"parents":[{"commit":"58b8136a7bf109b75c52c7fb20f6a67389ee7867","subject":"align fs020/21 settings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/58b8136a7bf109b75c52c7fb20f6a67389ee7867"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-12 08:43:00.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\nFixed the novajoin.keytab path\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a7cff628eb02506d851752a8fa45447fcbc89610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a7cff628eb02506d851752a8fa45447fcbc89610"}]},"branch":"refs/heads/master"},"894f8dad4d0821e253584f6a8e35a3f0e079a8d3":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-04-13 16:47:56.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/19"}}},"commit":{"parents":[{"commit":"badaaf7b17a7a5b33c2ce2d2906ddd3bda45aac2","subject":"Merge \"Create libvirt nodes dynamically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/badaaf7b17a7a5b33c2ce2d2906ddd3bda45aac2"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-04-13 16:47:56.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\nFixed the novajoin.keytab path\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/894f8dad4d0821e253584f6a8e35a3f0e079a8d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/894f8dad4d0821e253584f6a8e35a3f0e079a8d3"}]},"branch":"refs/heads/master"},"06a16f16d04e38135ac4b2d1761c9c9aeb201356":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2019-04-15 07:52:08.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/20"}}},"commit":{"parents":[{"commit":"badaaf7b17a7a5b33c2ce2d2906ddd3bda45aac2","subject":"Merge \"Create libvirt nodes dynamically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/badaaf7b17a7a5b33c2ce2d2906ddd3bda45aac2"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-04-15 07:52:03.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job","message":"Run novajoin tempest tests in all TLS job\n\nLet\u0027s use the tempest_format to packages\nFixed the novajoin.keytab path\n\nDepends-On: https://review.openstack.org/#/c/652499\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/06a16f16d04e38135ac4b2d1761c9c9aeb201356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/06a16f16d04e38135ac4b2d1761c9c9aeb201356"}]},"branch":"refs/heads/master"},"9e56db04a6ce248036db5984b31839c3a2b2d0b8":{"kind":"REWORK","_number":21,"created":"2019-04-17 08:09:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/21"}}},"commit":{"parents":[{"commit":"badaaf7b17a7a5b33c2ce2d2906ddd3bda45aac2","subject":"Merge \"Create libvirt nodes dynamically\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/badaaf7b17a7a5b33c2ce2d2906ddd3bda45aac2"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar","email":"chkumar@redhat.com","date":"2019-04-17 08:08:00.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.openstack.org/#/c/652986/\nDepends-On: https://review.openstack.org/#/c/652983/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9e56db04a6ce248036db5984b31839c3a2b2d0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9e56db04a6ce248036db5984b31839c3a2b2d0b8"}]},"branch":"refs/heads/master"},"7db5e9e73d6849626792fe32707b82a8deef8f49":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2019-04-17 12:43:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/22"}}},"commit":{"parents":[{"commit":"62724ea59185dc387716a0d84561c65d32bcadd1","subject":"Merge \"Remove \"validate_services\" param from featureset\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/62724ea59185dc387716a0d84561c65d32bcadd1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-04-17 12:43:32.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.openstack.org/#/c/652986/\nDepends-On: https://review.openstack.org/#/c/652983/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7db5e9e73d6849626792fe32707b82a8deef8f49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7db5e9e73d6849626792fe32707b82a8deef8f49"}]},"branch":"refs/heads/master"},"48dcbffaa5034397ae2860bc6baca103eb9673f2":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-04-22 22:43:29.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/23"}}},"commit":{"parents":[{"commit":"bde79b5a4116e73a45dd00cc54288ec8067f2a76","subject":"OpenDev Migration Patch 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bde79b5a4116e73a45dd00cc54288ec8067f2a76"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-04-22 22:43:29.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.openstack.org/#/c/652986/\nDepends-On: https://review.openstack.org/#/c/652983/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/48dcbffaa5034397ae2860bc6baca103eb9673f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/48dcbffaa5034397ae2860bc6baca103eb9673f2"}]},"branch":"refs/heads/master"},"5e6197830ae6b2e3d75a8497bb252550d6515068":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-05-16 15:16:10.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/24/639324/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/24"}}},"commit":{"parents":[{"commit":"f55b0bc3fe42a4cb1e9cff4dc987ca2cf0314aa9","subject":"Revert \"pin Wekzeug module to 0.15.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f55b0bc3fe42a4cb1e9cff4dc987ca2cf0314aa9"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-05-16 15:16:10.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.openstack.org/#/c/652986/\nDepends-On: https://review.openstack.org/#/c/652983/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5e6197830ae6b2e3d75a8497bb252550d6515068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5e6197830ae6b2e3d75a8497bb252550d6515068"}]},"branch":"refs/heads/master"},"0e3ec1c54f070ece7b93eb8bc318785e1c0a02db":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2019-06-12 10:45:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/25"}}},"commit":{"parents":[{"commit":"f55b0bc3fe42a4cb1e9cff4dc987ca2cf0314aa9","subject":"Revert \"pin Wekzeug module to 0.15.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f55b0bc3fe42a4cb1e9cff4dc987ca2cf0314aa9"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-12 10:45:50.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/664568/\nDepends-On: https://review.opendev.org/#/c/664586/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0e3ec1c54f070ece7b93eb8bc318785e1c0a02db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0e3ec1c54f070ece7b93eb8bc318785e1c0a02db"}]},"branch":"refs/heads/master"},"6e8db851b707dbb64882d1520f20ad5fe6a9bdba":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2019-06-12 10:46:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/26"}}},"commit":{"parents":[{"commit":"69adeda0e4e2dc4231f755a87a0eed638458dba1","subject":"Merge \"add the missing o to tripleo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/69adeda0e4e2dc4231f755a87a0eed638458dba1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-12 10:46:01.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/664568/\nDepends-On: https://review.opendev.org/#/c/664586/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6e8db851b707dbb64882d1520f20ad5fe6a9bdba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6e8db851b707dbb64882d1520f20ad5fe6a9bdba"}]},"branch":"refs/heads/master"},"a1e67b3587d36a01d8bf98033475a26a482f7f8d":{"kind":"NO_CODE_CHANGE","_number":27,"created":"2019-06-18 12:26:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/27"}}},"commit":{"parents":[{"commit":"69adeda0e4e2dc4231f755a87a0eed638458dba1","subject":"Merge \"add the missing o to tripleo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/69adeda0e4e2dc4231f755a87a0eed638458dba1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-18 12:25:57.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/664568/\nDepends-On: https://review.opendev.org/#/c/664586/\nDepends-On: https://review.opendev.org/665604\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a1e67b3587d36a01d8bf98033475a26a482f7f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a1e67b3587d36a01d8bf98033475a26a482f7f8d"}]},"branch":"refs/heads/master"},"0cafb96e434c13755470dd96e12eccf2c8c262c2":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2019-06-18 12:27:47.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/28"}}},"commit":{"parents":[{"commit":"091d0b982ec77f8803b7ffe72c01e8b107033c25","subject":"Merge \"Fix typos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/091d0b982ec77f8803b7ffe72c01e8b107033c25"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-18 12:27:47.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/664568/\nDepends-On: https://review.opendev.org/#/c/664586/\nDepends-On: https://review.opendev.org/665604\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0cafb96e434c13755470dd96e12eccf2c8c262c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0cafb96e434c13755470dd96e12eccf2c8c262c2"}]},"branch":"refs/heads/master"},"9bc2bdb70a6266de5c50d2f7c3e12a0cdf66d776":{"kind":"NO_CODE_CHANGE","_number":29,"created":"2019-06-18 12:46:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/29"}}},"commit":{"parents":[{"commit":"091d0b982ec77f8803b7ffe72c01e8b107033c25","subject":"Merge \"Fix typos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/091d0b982ec77f8803b7ffe72c01e8b107033c25"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-18 12:45:57.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\nDepends-On: https://review.opendev.org/#/c/664586/\nDepends-On: https://review.opendev.org/665604\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9bc2bdb70a6266de5c50d2f7c3e12a0cdf66d776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9bc2bdb70a6266de5c50d2f7c3e12a0cdf66d776"}]},"branch":"refs/heads/master"},"8942336492e7e60a0e9f78e6dca3e7fcd38ab18f":{"kind":"NO_CODE_CHANGE","_number":30,"created":"2019-06-19 05:51:16.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/30"}}},"commit":{"parents":[{"commit":"091d0b982ec77f8803b7ffe72c01e8b107033c25","subject":"Merge \"Fix typos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/091d0b982ec77f8803b7ffe72c01e8b107033c25"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-19 05:51:10.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/665604\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8942336492e7e60a0e9f78e6dca3e7fcd38ab18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8942336492e7e60a0e9f78e6dca3e7fcd38ab18f"}]},"branch":"refs/heads/master"},"cc0f4c82f38a92e1ee49e646651aca8b95a718ed":{"kind":"NO_CODE_CHANGE","_number":31,"created":"2019-06-19 08:49:00.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/31"}}},"commit":{"parents":[{"commit":"091d0b982ec77f8803b7ffe72c01e8b107033c25","subject":"Merge \"Fix typos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/091d0b982ec77f8803b7ffe72c01e8b107033c25"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-19 08:48:50.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/665604\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc0f4c82f38a92e1ee49e646651aca8b95a718ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc0f4c82f38a92e1ee49e646651aca8b95a718ed"}]},"branch":"refs/heads/master"},"a781498f4f9c5216396f787c68f94e6c390a351f":{"kind":"NO_CODE_CHANGE","_number":32,"created":"2019-06-24 06:55:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/32"}}},"commit":{"parents":[{"commit":"091d0b982ec77f8803b7ffe72c01e8b107033c25","subject":"Merge \"Fix typos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/091d0b982ec77f8803b7ffe72c01e8b107033c25"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-24 06:55:48.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a781498f4f9c5216396f787c68f94e6c390a351f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a781498f4f9c5216396f787c68f94e6c390a351f"}]},"branch":"refs/heads/master"},"d9efff45e8f057364958062ef50bf1cf36e1ddd4":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2019-06-24 06:55:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/33"}}},"commit":{"parents":[{"commit":"78131ae2004e705c238526becba158d3a66e23f0","subject":"Add undercloud_cloud_domain in featureset049","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78131ae2004e705c238526becba158d3a66e23f0"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-24 06:55:59.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d9efff45e8f057364958062ef50bf1cf36e1ddd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d9efff45e8f057364958062ef50bf1cf36e1ddd4"}]},"branch":"refs/heads/master"},"de5d60099471dce2629877e28fb4c2ecdfc4b428":{"kind":"NO_CODE_CHANGE","_number":34,"created":"2019-06-25 07:47:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/34"}}},"commit":{"parents":[{"commit":"78131ae2004e705c238526becba158d3a66e23f0","subject":"Add undercloud_cloud_domain in featureset049","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78131ae2004e705c238526becba158d3a66e23f0"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-25 07:47:34.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/de5d60099471dce2629877e28fb4c2ecdfc4b428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/de5d60099471dce2629877e28fb4c2ecdfc4b428"}]},"branch":"refs/heads/master"},"61f0f8f87ddefb7010259a7f66f2fe20b52a2c2c":{"kind":"NO_CODE_CHANGE","_number":35,"created":"2019-06-25 10:59:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/35"}}},"commit":{"parents":[{"commit":"78131ae2004e705c238526becba158d3a66e23f0","subject":"Add undercloud_cloud_domain in featureset049","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78131ae2004e705c238526becba158d3a66e23f0"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-25 10:59:14.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nDepends-On: https://review.opendev.org/#/c/667323/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/61f0f8f87ddefb7010259a7f66f2fe20b52a2c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/61f0f8f87ddefb7010259a7f66f2fe20b52a2c2c"}]},"branch":"refs/heads/master"},"8ad71394e404cd0098f3787f560edd551b020435":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2019-06-26 03:30:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/36"}}},"commit":{"parents":[{"commit":"6d8420ff3ca6898b314981dfbdd450a0936bf9de","subject":"Merge \"Bump Ansible to 2.7.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6d8420ff3ca6898b314981dfbdd450a0936bf9de"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-26 03:30:11.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nDepends-On: https://review.opendev.org/#/c/667323/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8ad71394e404cd0098f3787f560edd551b020435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8ad71394e404cd0098f3787f560edd551b020435"}]},"branch":"refs/heads/master"},"ec2275682d369a8be1c02ec6a11b866b63bd177a":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2019-06-27 08:41:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/37"}}},"commit":{"parents":[{"commit":"7279f085c6cb963043ba757402d42f36018c88a4","subject":"Merge \"master promotion use the same repos as master\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7279f085c6cb963043ba757402d42f36018c88a4"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-06-27 08:41:06.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nDepends-On: https://review.opendev.org/#/c/667323/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ec2275682d369a8be1c02ec6a11b866b63bd177a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ec2275682d369a8be1c02ec6a11b866b63bd177a"}]},"branch":"refs/heads/master"},"1ff0f6a7354ca4d4ef8dbeeadc59561e778455e6":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2019-07-01 06:55:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/38"}}},"commit":{"parents":[{"commit":"2a81c126b6b5927c6308a0392143facc1f8d444a","subject":"Merge \"Remove featureset057\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2a81c126b6b5927c6308a0392143facc1f8d444a"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-01 06:55:11.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nDepends-On: https://review.opendev.org/#/c/667323/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1ff0f6a7354ca4d4ef8dbeeadc59561e778455e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1ff0f6a7354ca4d4ef8dbeeadc59561e778455e6"}]},"branch":"refs/heads/master"},"cc2ee96774119d3a460da04200c701c7a51b3bd9":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2019-07-02 14:20:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/39"}}},"commit":{"parents":[{"commit":"9df739dd2a569a92545072c86cf9c4a8efdce796","subject":"Fix failure in featureset039","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9df739dd2a569a92545072c86cf9c4a8efdce796"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-02 14:20:34.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nDepends-On: https://review.opendev.org/#/c/667323/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc2ee96774119d3a460da04200c701c7a51b3bd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc2ee96774119d3a460da04200c701c7a51b3bd9"}]},"branch":"refs/heads/master"},"204466f936310ade0d114ad11290d93aded2ef40":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2019-07-03 07:14:57.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/40"}}},"commit":{"parents":[{"commit":"d2659094dcaf6c400058d47d62a250b4178ba983","subject":"Merge \"Turn off ansible lint after deployment.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d2659094dcaf6c400058d47d62a250b4178ba983"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-03 07:14:57.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nDepends-On: https://review.opendev.org/#/c/667323/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/204466f936310ade0d114ad11290d93aded2ef40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/204466f936310ade0d114ad11290d93aded2ef40"}]},"branch":"refs/heads/master"},"a1e96a453c4f39d56897b5dd78010f43869b7b2b":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2019-07-08 06:43:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/41"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-08 06:43:58.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\n\nDepends-On: https://review.opendev.org/#/c/664568/\n\nDepends-On: https://review.opendev.org/#/c/664586/\n\nDepends-On: https://review.opendev.org/#/c/665604/\n\nDepends-On: https://review.opendev.org/#/c/667323/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a1e96a453c4f39d56897b5dd78010f43869b7b2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a1e96a453c4f39d56897b5dd78010f43869b7b2b"}]},"branch":"refs/heads/master"},"9ac924cc9664f9fe1bc59df3684cf714f5aba9f7":{"kind":"REWORK","_number":42,"created":"2019-08-25 13:20:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/42"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-25 13:20:01.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/678418/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9ac924cc9664f9fe1bc59df3684cf714f5aba9f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9ac924cc9664f9fe1bc59df3684cf714f5aba9f7"}]},"branch":"refs/heads/master"},"ba529410bb8ab5b59072de74749c0873c240935d":{"kind":"NO_CODE_CHANGE","_number":43,"created":"2019-08-26 05:34:43.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/43"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-26 05:34:38.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/678418/\nDepends-On: https://review.opendev.org/#/c/677925/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ba529410bb8ab5b59072de74749c0873c240935d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ba529410bb8ab5b59072de74749c0873c240935d"}]},"branch":"refs/heads/master"},"cac9940e5fb060964dd1c34b20bfb5b9183febe2":{"kind":"REWORK","_number":44,"created":"2019-08-26 10:18:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/44"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-26 10:17:38.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/678418/\nDepends-On: https://review.opendev.org/677925\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cac9940e5fb060964dd1c34b20bfb5b9183febe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cac9940e5fb060964dd1c34b20bfb5b9183febe2"}]},"branch":"refs/heads/master"},"ca14a90a43b7a98ced2e031c11e1ffb134448664":{"kind":"REWORK","_number":45,"created":"2019-08-26 10:19:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/45"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-26 10:19:46.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/678418/\nDepends-On: https://review.opendev.org/677925\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ca14a90a43b7a98ced2e031c11e1ffb134448664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ca14a90a43b7a98ced2e031c11e1ffb134448664"}]},"branch":"refs/heads/master"},"f37f7178721ea6c501ee31e392a52923f26b1b67":{"kind":"REWORK","_number":46,"created":"2019-08-26 13:35:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/46"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-26 13:35:40.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/678418/\nDepends-On: https://review.opendev.org/677925\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f37f7178721ea6c501ee31e392a52923f26b1b67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f37f7178721ea6c501ee31e392a52923f26b1b67"}]},"branch":"refs/heads/master"},"888e16dd9e8fd1b97043db85accaba64a6329238":{"kind":"REWORK","_number":47,"created":"2019-08-26 13:56:38.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/47"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-26 13:56:26.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/678418/\nDepends-On: https://review.opendev.org/677925\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/888e16dd9e8fd1b97043db85accaba64a6329238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/888e16dd9e8fd1b97043db85accaba64a6329238"}]},"branch":"refs/heads/master"},"c40bfccad595f5b885eaa7d74fe75a8bf8b9023a":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2019-08-27 03:56:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/48"}}},"commit":{"parents":[{"commit":"3e76b9d54dbd54f44517b0da534f0e179cda6887","subject":"Merge \"Add settings from fs001 missing in fs035\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3e76b9d54dbd54f44517b0da534f0e179cda6887"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-27 03:56:31.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/678418/\nDepends-On: https://review.opendev.org/677925\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c40bfccad595f5b885eaa7d74fe75a8bf8b9023a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c40bfccad595f5b885eaa7d74fe75a8bf8b9023a"}]},"branch":"refs/heads/master"},"1996efb16a27ede94496bdeada43a4e83f1ed8ef":{"kind":"REWORK","_number":49,"created":"2019-08-27 11:23:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/49"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-27 11:22:47.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1996efb16a27ede94496bdeada43a4e83f1ed8ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1996efb16a27ede94496bdeada43a4e83f1ed8ef"}]},"branch":"refs/heads/master"},"3818a386188b0b61f468109ed878f663a60cc696":{"kind":"REWORK","_number":50,"created":"2019-08-28 05:07:37.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/50"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-28 05:07:29.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3818a386188b0b61f468109ed878f663a60cc696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3818a386188b0b61f468109ed878f663a60cc696"}]},"branch":"refs/heads/master"},"41141e1f3d54385806134137950e848bc4f04620":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2019-08-28 05:07:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/51"}}},"commit":{"parents":[{"commit":"932acbe66eb477a45d9532b13a8173a97be1c590","subject":"Merge \"Install selinux shim from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/932acbe66eb477a45d9532b13a8173a97be1c590"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-28 05:07:45.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/41141e1f3d54385806134137950e848bc4f04620"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/41141e1f3d54385806134137950e848bc4f04620"}]},"branch":"refs/heads/master"},"badcc14f82d44d4d7fdc9e95b74251ba6491a28a":{"kind":"REWORK","_number":52,"created":"2019-08-28 10:09:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/52"}}},"commit":{"parents":[{"commit":"932acbe66eb477a45d9532b13a8173a97be1c590","subject":"Merge \"Install selinux shim from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/932acbe66eb477a45d9532b13a8173a97be1c590"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-28 10:09:34.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/badcc14f82d44d4d7fdc9e95b74251ba6491a28a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/badcc14f82d44d4d7fdc9e95b74251ba6491a28a"}]},"branch":"refs/heads/master"},"6f1e76f5411b91bdd1c818b7f05097b384eb9837":{"kind":"REWORK","_number":53,"created":"2019-08-29 09:08:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/53"}}},"commit":{"parents":[{"commit":"932acbe66eb477a45d9532b13a8173a97be1c590","subject":"Merge \"Install selinux shim from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/932acbe66eb477a45d9532b13a8173a97be1c590"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-29 09:07:42.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6f1e76f5411b91bdd1c818b7f05097b384eb9837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6f1e76f5411b91bdd1c818b7f05097b384eb9837"}]},"branch":"refs/heads/master"},"67c7f2dfe3eb237c8be763b4193967bbdc76b5d9":{"kind":"REWORK","_number":54,"created":"2019-09-06 07:32:11.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/54"}}},"commit":{"parents":[{"commit":"932acbe66eb477a45d9532b13a8173a97be1c590","subject":"Merge \"Install selinux shim from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/932acbe66eb477a45d9532b13a8173a97be1c590"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-09-06 07:32:05.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/67c7f2dfe3eb237c8be763b4193967bbdc76b5d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/67c7f2dfe3eb237c8be763b4193967bbdc76b5d9"}]},"branch":"refs/heads/master"},"fd3cffd6d5407ba493daa609b32de61c1c30dd78":{"kind":"REWORK","_number":55,"created":"2019-09-06 10:25:07.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/55"}}},"commit":{"parents":[{"commit":"932acbe66eb477a45d9532b13a8173a97be1c590","subject":"Merge \"Install selinux shim from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/932acbe66eb477a45d9532b13a8173a97be1c590"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-09-06 10:25:02.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fd3cffd6d5407ba493daa609b32de61c1c30dd78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fd3cffd6d5407ba493daa609b32de61c1c30dd78"}]},"branch":"refs/heads/master"},"e0d42c6d12c93da7add7a29b311d7eff364c0d48":{"kind":"REWORK","_number":56,"created":"2019-09-09 10:53:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/56"}}},"commit":{"parents":[{"commit":"932acbe66eb477a45d9532b13a8173a97be1c590","subject":"Merge \"Install selinux shim from pypi\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/932acbe66eb477a45d9532b13a8173a97be1c590"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-09-09 10:53:33.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e0d42c6d12c93da7add7a29b311d7eff364c0d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e0d42c6d12c93da7add7a29b311d7eff364c0d48"}]},"branch":"refs/heads/master"},"26051964ff62efa16940a63e3c4e1ed6ce0702a4":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2019-11-13 18:38:55.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/24/639324/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/57"}}},"commit":{"parents":[{"commit":"5f9754c829900754783d8ed2f7aa3c5c30fa9d9e","subject":"Merge \"Add CloudSIG stein/train release files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5f9754c829900754783d8ed2f7aa3c5c30fa9d9e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2019-11-13 18:38:55.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/26051964ff62efa16940a63e3c4e1ed6ce0702a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/26051964ff62efa16940a63e3c4e1ed6ce0702a4"}]},"branch":"refs/heads/master"},"0510de3128c29154c7379d3c5ef57467fa4966d9":{"kind":"REWORK","_number":58,"created":"2019-11-14 02:55:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/58"}}},"commit":{"parents":[{"commit":"5f9754c829900754783d8ed2f7aa3c5c30fa9d9e","subject":"Merge \"Add CloudSIG stein/train release files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5f9754c829900754783d8ed2f7aa3c5c30fa9d9e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-14 02:54:50.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0510de3128c29154c7379d3c5ef57467fa4966d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0510de3128c29154c7379d3c5ef57467fa4966d9"}]},"branch":"refs/heads/master"},"cbc1d1813d67fd7713d4edfa9eb8fa097f47f676":{"kind":"REWORK","_number":59,"created":"2019-11-14 06:41:54.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/59"}}},"commit":{"parents":[{"commit":"5f9754c829900754783d8ed2f7aa3c5c30fa9d9e","subject":"Merge \"Add CloudSIG stein/train release files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5f9754c829900754783d8ed2f7aa3c5c30fa9d9e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-14 06:41:37.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cbc1d1813d67fd7713d4edfa9eb8fa097f47f676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cbc1d1813d67fd7713d4edfa9eb8fa097f47f676"}]},"branch":"refs/heads/master"},"f9435506c726fedc2a61beaca9c2af57d89a65d4":{"kind":"REWORK","_number":60,"created":"2019-11-14 10:47:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/60"}}},"commit":{"parents":[{"commit":"5f9754c829900754783d8ed2f7aa3c5c30fa9d9e","subject":"Merge \"Add CloudSIG stein/train release files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5f9754c829900754783d8ed2f7aa3c5c30fa9d9e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-14 10:47:17.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f9435506c726fedc2a61beaca9c2af57d89a65d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f9435506c726fedc2a61beaca9c2af57d89a65d4"}]},"branch":"refs/heads/master"},"73c0cfa5f31c182255170461b64fb6d48ec8740d":{"kind":"REWORK","_number":61,"created":"2019-11-14 14:35:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/61"}}},"commit":{"parents":[{"commit":"5f9754c829900754783d8ed2f7aa3c5c30fa9d9e","subject":"Merge \"Add CloudSIG stein/train release files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5f9754c829900754783d8ed2f7aa3c5c30fa9d9e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-14 14:35:05.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/694311/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/73c0cfa5f31c182255170461b64fb6d48ec8740d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/73c0cfa5f31c182255170461b64fb6d48ec8740d"}]},"branch":"refs/heads/master"},"3db11e049b2f839b893dfa3d419ce9e19ab13ded":{"kind":"REWORK","_number":62,"created":"2019-11-14 14:36:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/62"}}},"commit":{"parents":[{"commit":"5f9754c829900754783d8ed2f7aa3c5c30fa9d9e","subject":"Merge \"Add CloudSIG stein/train release files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5f9754c829900754783d8ed2f7aa3c5c30fa9d9e"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-14 14:35:49.000000000","tz":330},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/694311/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3db11e049b2f839b893dfa3d419ce9e19ab13ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3db11e049b2f839b893dfa3d419ce9e19ab13ded"}]},"branch":"refs/heads/master"},"d600b59808c76d74e94b67eb1b4858f2e9cbe3fa":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2019-11-15 03:50:05.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/24/639324/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/639324/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/639324/63"}}},"commit":{"parents":[{"commit":"1e2976d687a1c070fd68601c3361c3b6385197df","subject":"Disable Paunch on the undercloud job (fs027)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1e2976d687a1c070fd68601c3361c3b6385197df"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-02-26 12:02:04.000000000","tz":120},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-15 03:50:05.000000000","tz":0},"subject":"Run novajoin tempest tests in all TLS job using os_tempest","message":"Run novajoin tempest tests in all TLS job using os_tempest\n\nDepends-On: https://review.opendev.org/#/c/694311/\n\nChange-Id: I93949966d1c8255bff9c8bff18b6f584a20f77cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d600b59808c76d74e94b67eb1b4858f2e9cbe3fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d600b59808c76d74e94b67eb1b4858f2e9cbe3fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
