)]}'
{"id":"openstack%2Ftripleo-quickstart~662055","triplet_id":"openstack%2Ftripleo-quickstart~master~If737085d7a27afca9de3357be29f7ef303452dfa","project":"openstack/tripleo-quickstart","branch":"master","topic":"bug/1830956","hashtags":[],"change_id":"If737085d7a27afca9de3357be29f7ef303452dfa","subject":"tripleo-quickstart fails due to python-pip","status":"ABANDONED","created":"2019-05-29 17:47:50.000000000","updated":"2020-04-30 08:53:04.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c4489d72377bdbd8ffd210c84a5fab2623398976","_number":662055,"virtual_id_number":662055,"owner":{"_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"date":"2019-06-26 16:25:00.000000000","_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"date":"2019-06-26 16:12:35.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":1,"date":"2019-05-30 00:01:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-05-29 19:23:39.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},{"value":1,"date":"2019-05-30 09:18:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2019-08-25 09:49:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-06-03 21:49:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-18 11:41:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-08-25 09:50:11.000000000","_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"date":"2019-08-25 09:50:05.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"date":"2019-08-25 09:50:00.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-29 19:23:39.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 00:01:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-30 09:18:25.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-06-03 21:49:42.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-18 11:41:40.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-06-26 16:12:35.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-08-25 09:49:49.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-08-25 09:50:00.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-25 09:50:05.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-25 09:50:11.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"}],"messages":[{"id":"049d72f04a4d72371e83e09f7dc7614c50219e73","author":{"_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},"date":"2019-05-29 17:47:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f69e2cc75a73ce32fae00b9d07c53c0384b0ed0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-29 19:23:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/51254fd/ : FAILURE in 1h 31m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/c75cf27/ : SUCCESS in 1h 09m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/fb5b6ed/ : RETRY_LIMIT in 3m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5e1d625/ : RETRY_LIMIT in 2m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bc004cdc217a9f293ca8fd946a8fd89ecf7dba0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 00:01:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-undercloud-containers/d4df68a/ : SUCCESS in 1h 35m 28s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/07a3a17/ : FAILURE in 1h 48m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode/7c9e894/ : SUCCESS in 1h 52m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4172405/ : SUCCESS in 2h 12m 21s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d48f841/ : SUCCESS in 2h 25m 12s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/55/662055/1/check/tripleo-ci-fedora-28-standalone/9777b48/ : SUCCESS in 1h 45m 43s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-standalone-upgrade/a79bb2e/ : FAILURE in 1h 35m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/e62b9cc/ : SUCCESS in 2h 03m 14s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-stein/cff6fa0/ : SUCCESS in 1h 57m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-rocky/4fb1622/ : SUCCESS in 2h 11m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-queens/b892724/ : SUCCESS in 2h 10m 50s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-pike/b994b77/ : SUCCESS in 2h 18m 52s\n- openstack-tox-docs http://logs.openstack.org/55/662055/1/check/openstack-tox-docs/ec9bab7/html/ : SUCCESS in 3m 16s\n- build-openstack-releasenotes http://logs.openstack.org/55/662055/1/check/build-openstack-releasenotes/fde676a/html/ : SUCCESS in 3m 39s\n- openstack-tox-linters http://logs.openstack.org/55/662055/1/check/openstack-tox-linters/5f24791/ : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"44688bade0c05830a3524e6e34bee7312300c93b","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-30 09:18:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2222ea746cc3d31ea7381ce7860dc503cea09c5b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-03 21:49:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe90789013759f354a938b04af84c974025bc759","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-06-18 11:41:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e68a670f450abbe917b1bc2094f1401e8bb37a60","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-06-26 16:12:35.000000000","message":"Patch Set 1:\n\nrecheck last run very old","accounts_in_message":[],"_revision_number":1},{"id":"b1137cff316ef18c5eabb30213b26fc25a4298ce","author":{"_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},"date":"2019-06-26 16:25:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f77a91f9dfa4b90f9fbbf77b858bc1a2e96dee7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-26 18:55:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0170898/ : FAILURE in 1h 04m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1666117/ : SUCCESS in 2h 29m 44s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/1ae7f99/ : SUCCESS in 1h 17m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/aef1803/ : SUCCESS in 2h 05m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/57013ca/ : SUCCESS in 2h 06m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/8e606bb/ : SUCCESS in 2h 15m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/55/662055/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/25b7dc0/ : SUCCESS in 2h 40m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13fc352b5a8bcd6c97f6ab09bb4127ab6909e2fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 02:12:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-undercloud-containers/b0e580c/ : SUCCESS in 1h 34m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/9e32840/ : FAILURE in 1h 19m 32s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c162edc/ : SUCCESS in 2h 20m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7a35f30/ : SUCCESS in 2h 16m 29s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/55/662055/1/check/tripleo-ci-fedora-28-standalone/34f308c/ : SUCCESS in 1h 36m 35s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-standalone-upgrade/fb5e87e/ : SUCCESS in 2h 06m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/34d3aa7/ : SUCCESS in 2h 12m 56s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-stein/80f6f9e/ : SUCCESS in 1h 58m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-rocky/415d101/ : SUCCESS in 2h 07m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-queens/d6cb37a/ : SUCCESS in 1h 59m 21s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/55/662055/1/check/tripleo-ci-centos-7-containers-multinode-pike/a448925/ : SUCCESS in 2h 15m 07s\n- openstack-tox-docs http://logs.openstack.org/55/662055/1/check/openstack-tox-docs/3e6dd2f/html/ : SUCCESS in 3m 54s\n- build-openstack-releasenotes http://logs.openstack.org/55/662055/1/check/build-openstack-releasenotes/8269e1c/html/ : SUCCESS in 3m 08s\n- tox-molecule http://logs.openstack.org/55/662055/1/check/tox-molecule/0560303/tox/reports.html : SUCCESS in 13m 08s\n- openstack-tox-linters http://logs.openstack.org/55/662055/1/check/openstack-tox-linters/54b220d/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1cf635aed4bb32c0409897ed29f24e7bef929b55","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-25 09:49:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3d3bcde21f12debbe2a487d46a8b07cf6d3a4d4","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-04-30 08:53:04.000000000","message":"Abandoned\n\nWith nobody tracking this is better to abandon it but I think that EPEL not being supported explained it better.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98d1ae27b4ee8e38be47a8a27b01a1aa49138ceb","revisions":{"98d1ae27b4ee8e38be47a8a27b01a1aa49138ceb":{"kind":"REWORK","_number":1,"created":"2019-05-29 17:47:50.000000000","uploader":{"_account_id":25235,"name":"Treva WIlliams","display_name":"trilliams","email":"tribecca@tribecc.us","username":"trilliams"},"ref":"refs/changes/55/662055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/55/662055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/55/662055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/55/662055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/55/662055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/55/662055/1"}}},"commit":{"parents":[{"commit":"1bdabc9f5a832a55eff6e9276f605967be544cca","subject":"Merge \"fs027: test SSL with FQDN based undercloud_public_host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1bdabc9f5a832a55eff6e9276f605967be544cca"}]}],"author":{"name":"trilliams","email":"tribecca@tribecc.us","date":"2019-05-29 17:39:05.000000000","tz":-300},"committer":{"name":"trilliams","email":"tribecca@tribecc.us","date":"2019-05-29 17:39:05.000000000","tz":-300},"subject":"tripleo-quickstart fails due to python-pip","message":"tripleo-quickstart fails due to python-pip\n\nThe python-pip package was renamed in EPEL to reflect python version,\nwhich causes the quickstart installer to fail early on CentOS 7 due to\ninability to find the correct package. Editing the name to python2-pip\nin bindep.txt seems to fix the issue per local tests.\n\nChange-Id: If737085d7a27afca9de3357be29f7ef303452dfa\nCloses-bug: #1830956\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/98d1ae27b4ee8e38be47a8a27b01a1aa49138ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/98d1ae27b4ee8e38be47a8a27b01a1aa49138ceb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
