)]}'
{"id":"openstack%2Ftripleo-quickstart~667219","triplet_id":"openstack%2Ftripleo-quickstart~master~Idebd929b40097442feee3efbdee3084cb38c9838","project":"openstack/tripleo-quickstart","branch":"master","topic":"ansible2.8-oooq","hashtags":[],"change_id":"Idebd929b40097442feee3efbdee3084cb38c9838","subject":"Bump tripleo quickstart ansible to 2.8","status":"MERGED","created":"2019-06-24 20:36:47.000000000","updated":"2019-08-09 12:13:52.000000000","submitted":"2019-08-09 12:13:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"667219-1565352832903-25fc65c6","meta_rev_id":"55f95eb829fb201ebae3ebd9e81addef63b32460","_number":667219,"virtual_id_number":667219,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-08-09 12:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-08-09 09:26:42.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":2,"date":"2019-08-09 00:08:18.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2019-08-09 00:02:47.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-08-08 15:06:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2019-08-09 07:21:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-09 10:15:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2019-08-09 09:26:42.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":27379,"name":"jrosser","inactive":true},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":27379,"name":"jrosser","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-26 14:34:13.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-07-26 17:08:52.000000000","updated_by":{"_account_id":27379,"name":"jrosser","inactive":true},"reviewer":{"_account_id":27379,"name":"jrosser","inactive":true},"state":"REVIEWER"},{"updated":"2019-08-08 15:06:17.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-08-09 00:08:18.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-09 07:21:39.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-08-09 09:26:42.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-09 10:15:37.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-08-09 12:13:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3e493aa8a0749fbcf5f7121f062565ef9f753a7f","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-24 20:36:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2d76f879d8cbde139e9358c6dadfdd716ab113a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-24 20:40:43.000000000","message":"Topic changed from ooo2.8 to ansible2.8-oooq","accounts_in_message":[],"_revision_number":1},{"id":"acf0103c73f988193920782034f7e493cb23207c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 22:26:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/1/check/tripleo-ci-centos-7-standalone/0111704/ : FAILURE in 12m 13s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/1/check/tripleo-ci-fedora-28-standalone/605f9ae/ : SUCCESS in 1h 44m 30s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/1/check/tripleo-ci-centos-7-standalone-upgrade/7a7ed8d/ : FAILURE in 12m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/19/667219/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/618b91d/ : FAILURE in 11m 11s\n- openstack-tox-docs http://logs.openstack.org/19/667219/1/check/openstack-tox-docs/c248599/html/ : SUCCESS in 3m 38s\n- tox-molecule http://logs.openstack.org/19/667219/1/check/tox-molecule/63f289b/tox/reports.html : FAILURE in 10m 34s\n- openstack-tox-linters http://logs.openstack.org/19/667219/1/check/openstack-tox-linters/fe81c62/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"09248c17aaa23ed62425977fc2d02c8fe62712a0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-25 08:57:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7432ce57ea6d1775e17873d2fdcefca694166428","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-25 08:58:57.000000000","message":"Topic changed from ooo28 to ansible2.8-oooq","accounts_in_message":[],"_revision_number":2},{"id":"8ce3c42f4a1926494081a46b91b1ac9b62b07a13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 13:13:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone finger://ze02.openstack.org/973474225bb248848a2dd1a946defd44 : RETRY_LIMIT in 48m 19s\n- tripleo-ci-fedora-28-standalone finger://ze01.openstack.org/0e30080d7ddb4edba44cd11ef5c652c9 : RETRY_LIMIT in 50m 21s\n- tripleo-ci-centos-7-standalone-upgrade finger://ze02.openstack.org/6e86fa7ac88540b8876c6d458b0c735d : RETRY_LIMIT in 54m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein finger://ze11.openstack.org/1a412f67ea4c4073bcf4685f76595d2d : RETRY_LIMIT in 48m 04s\n- openstack-tox-docs http://logs.openstack.org/19/667219/2/check/openstack-tox-docs/b604159/html/ : SUCCESS in 5m 21s\n- tox-molecule http://logs.openstack.org/19/667219/2/check/tox-molecule/b1f4aa4/tox/reports.html : SUCCESS in 13m 31s\n- openstack-tox-linters http://logs.openstack.org/19/667219/2/check/openstack-tox-linters/28ad361/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"da6c18f38b5a6095395bffdc7784d852e56a98ea","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-25 13:43:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3137bd17251c2e4f306af275180dbe38eaf905e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 01:00:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/2/check/tripleo-ci-centos-7-standalone/cc5b00b/ : FAILURE in 1h 55m 38s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/2/check/tripleo-ci-fedora-28-standalone/3ff9c3c/ : SUCCESS in 1h 38m 08s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/2/check/tripleo-ci-centos-7-standalone-upgrade/0b5d6ef/ : SUCCESS in 2h 01m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/19/667219/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/3bfa6a3/ : SUCCESS in 2h 21m 23s\n- openstack-tox-docs http://logs.openstack.org/19/667219/2/check/openstack-tox-docs/66a122a/html/ : SUCCESS in 4m 47s\n- tox-molecule http://logs.openstack.org/19/667219/2/check/tox-molecule/d86a53a/tox/reports.html : SUCCESS in 16m 44s\n- openstack-tox-linters http://logs.openstack.org/19/667219/2/check/openstack-tox-linters/257f293/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"92cc7848b1e32a5520e8fb662c7a6305a62c4f3f","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-06-26 14:34:13.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b29f93d55ed245e0cf715abf0e34312bb8f58ffd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 00:30:34.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/3/check/tripleo-ci-centos-7-standalone/54e8648/ : FAILURE in 1h 50m 12s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/3/check/tripleo-ci-fedora-28-standalone/53e07ae/ : SUCCESS in 1h 44m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/3/check/tripleo-ci-centos-7-standalone-upgrade/4bd59b7/ : SUCCESS in 2h 31m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/19/667219/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/b9ed1fc/ : SUCCESS in 2h 20m 20s\n- openstack-tox-docs http://logs.openstack.org/19/667219/3/check/openstack-tox-docs/a7d7936/html/ : SUCCESS in 3m 47s\n- tox-molecule http://logs.openstack.org/19/667219/3/check/tox-molecule/9c2d20c/tox/reports.html : SUCCESS in 14m 33s\n- openstack-tox-linters http://logs.openstack.org/19/667219/3/check/openstack-tox-linters/b57fac3/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"e4f0e37eb21b94a528bfe7fe7bfaba900db09dee","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-01 16:21:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"4688a6d93f0e3aaf81720c6c670c9fcfb9622dcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 01:04:10.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/4/check/tripleo-ci-centos-7-standalone/a81eb57/ : FAILURE in 1h 39m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/4/check/tripleo-ci-fedora-28-standalone/e8fe81e/ : SUCCESS in 1h 50m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/4/check/tripleo-ci-centos-7-standalone-upgrade/a75128c/ : SUCCESS in 2h 20m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/19/667219/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/463f876/ : SUCCESS in 2h 16m 01s\n- openstack-tox-docs http://logs.openstack.org/19/667219/4/check/openstack-tox-docs/5f00f85/html/ : SUCCESS in 6m 11s\n- tox-molecule http://logs.openstack.org/19/667219/4/check/tox-molecule/3ce2c73/tox/reports.html : SUCCESS in 17m 38s\n- openstack-tox-linters http://logs.openstack.org/19/667219/4/check/openstack-tox-linters/c1acb5a/ : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":4},{"id":"06196a1bd5556dc6d9179c1d0f07bd0f6675303a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-09 07:10:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"86fd1f867db34d99edec46a7878c7d326af1d9b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 10:32:42.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/4/check/tripleo-ci-centos-7-standalone/4dc1d7d/ : FAILURE in 1h 48m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/4/check/tripleo-ci-fedora-28-standalone/e8abf5b/ : SUCCESS in 1h 40m 54s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/4/check/tripleo-ci-centos-7-standalone-upgrade/5c20cdc/ : SUCCESS in 2h 21m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/19/667219/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/1b8ea98/ : SUCCESS in 2h 38m 05s\n- openstack-tox-docs http://logs.openstack.org/19/667219/4/check/openstack-tox-docs/b053823/html/ : SUCCESS in 5m 57s\n- tox-molecule http://logs.openstack.org/19/667219/4/check/tox-molecule/1d6701d/tox/reports.html : SUCCESS in 17m 09s\n- openstack-tox-linters http://logs.openstack.org/19/667219/4/check/openstack-tox-linters/3056240/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":4},{"id":"52f8920e6aea0ec835c267b78ab03885b601b1e6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-22 08:53:37.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"6c8459707d756cfec62bac3ff48487c2c674ade8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 10:48:54.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/5/check/tripleo-ci-centos-7-standalone/a34fc53/ : FAILURE in 1h 43m 39s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/5/check/tripleo-ci-fedora-28-standalone/2d4ef7c/ : SUCCESS in 1h 36m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/5/check/tripleo-ci-centos-7-standalone-upgrade/02a9286/ : FAILURE in 1h 48m 50s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/5/check/openstack-tox-docs/2a810c2/html/ : SUCCESS in 4m 39s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/5/check/openstack-tox-molecule/626266c/tox/reports.html : SUCCESS in 23m 00s\n- openstack-tox-linters http://logs.openstack.org/19/667219/5/check/openstack-tox-linters/890d908/ : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":5},{"id":"e8d42a5c638598ffc03606b1a3350c3242378a7a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-23 09:06:03.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"1a9bb0ace39031d356d36b74f506e907fba89da2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 11:20:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/6/check/tripleo-ci-centos-7-standalone/43b9d3d/ : SUCCESS in 1h 49m 29s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/6/check/tripleo-ci-fedora-28-standalone/803cbbc/ : SUCCESS in 1h 35m 46s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/6/check/tripleo-ci-centos-7-standalone-upgrade/99dd4eb/ : FAILURE in 1h 53m 06s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/6/check/openstack-tox-docs/a47b07f/html/ : SUCCESS in 4m 58s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/6/check/openstack-tox-molecule/3aec4bc/tox/reports.html : SUCCESS in 15m 05s\n- openstack-tox-linters http://logs.openstack.org/19/667219/6/check/openstack-tox-linters/d77a835/ : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":6},{"id":"4a14445595579ff9705564a3739e89621f7e15b4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-24 17:38:07.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"46e76fab9734ce0b647d8e2c12180b20733929a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 19:45:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/7/check/tripleo-ci-centos-7-standalone/34869cc/ : FAILURE in 2h 00m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/7/check/tripleo-ci-fedora-28-standalone/456460e/ : SUCCESS in 1h 31m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/7/check/tripleo-ci-centos-7-standalone-upgrade/51ab03a/ : FAILURE in 1h 45m 30s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/7/check/openstack-tox-docs/8388345/html/ : SUCCESS in 4m 45s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/7/check/openstack-tox-molecule/f4ee687/tox/reports.html : SUCCESS in 19m 54s\n- openstack-tox-linters http://logs.openstack.org/19/667219/7/check/openstack-tox-linters/0f5b2c0/ : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":7},{"id":"bdd2461dffca0fd634e483d32280f8a47adb1c64","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 04:30:52.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"b361d25ee7cd6fe4148c47999ab64f5330ffd95d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 08:24:24.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"ead027670c6fc7d3ef60bb011554934a386fa82b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 08:24:44.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"10aaf9f3ad455d1a36ee4718360a8828d9b720be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 10:12:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/8/check/tripleo-ci-centos-7-standalone/72c1811/ : SUCCESS in 1h 30m 33s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/8/check/tripleo-ci-fedora-28-standalone/8ab09cb/ : SUCCESS in 1h 28m 01s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/8/check/tripleo-ci-centos-7-standalone-upgrade/3a6acf0/ : FAILURE in 1h 43m 03s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/8/check/openstack-tox-docs/d1e4b49/html/ : SUCCESS in 4m 26s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/8/check/openstack-tox-molecule/9072197/tox/reports.html : SUCCESS in 13m 29s\n- openstack-tox-linters http://logs.openstack.org/19/667219/8/check/openstack-tox-linters/ff84975/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":8},{"id":"6570349f911b554c67c6a41123b0987c19ddf676","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-26 12:30:35.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"f994a4d03ddd708d5e45779e6a2b0187d4f0731f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 14:44:04.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/9/check/tripleo-ci-centos-7-standalone/a100492/ : SUCCESS in 1h 45m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/9/check/tripleo-ci-fedora-28-standalone/cc74d09/ : SUCCESS in 1h 32m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/9/check/tripleo-ci-centos-7-standalone-upgrade/4eb6a89/ : FAILURE in 1h 49m 06s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/9/check/openstack-tox-docs/1090528/html/ : SUCCESS in 4m 50s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/9/check/openstack-tox-molecule/d30143f/tox/reports.html : SUCCESS in 15m 00s\n- openstack-tox-linters http://logs.openstack.org/19/667219/9/check/openstack-tox-linters/cabdf7d/ : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":9},{"id":"5fdb98e21638161a7414563d731091d6600b0dd8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-26 16:19:17.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"aac75480e1cc8c14bbc16902ce4cbdb9aa29c376","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 16:40:21.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"1c557398509e840ee34898274dd04cea14c6a411","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-26 16:40:41.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"3f7a00393d88538500afac581cdc9bd8ccff24ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 16:43:10.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"294e3a6a232251d7470c037308dc923fb103d801","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-26 17:08:27.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"5a8a810fc00d5bb7305c9446c3ffdd9d46ff91b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 19:34:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/12/check/tripleo-ci-centos-7-standalone/43393a9/ : SUCCESS in 1h 38m 54s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/12/check/tripleo-ci-fedora-28-standalone/09b4197/ : SUCCESS in 1h 48m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/12/check/tripleo-ci-centos-7-standalone-upgrade/dc2d4ec/ : SUCCESS in 2h 00m 08s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/12/check/openstack-tox-docs/d67a374/html/ : SUCCESS in 4m 50s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/12/check/openstack-tox-molecule/f53ca87/tox/reports.html : SUCCESS in 15m 42s\n- openstack-tox-linters http://logs.openstack.org/19/667219/12/check/openstack-tox-linters/e8a934a/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":12},{"id":"427a59e6092984115f097c9b0d03e9f6ff4d5650","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-29 06:25:27.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"80af16d6b7958f9e557e0817b598d3072a095a16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 09:34:50.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/12/check/tripleo-ci-centos-7-standalone/e160545/ : SUCCESS in 1h 51m 17s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/12/check/tripleo-ci-fedora-28-standalone/228d115/ : SUCCESS in 1h 37m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/12/check/tripleo-ci-centos-7-standalone-upgrade/1c00562/ : SUCCESS in 1h 59m 46s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/12/check/openstack-tox-docs/5981523/html/ : SUCCESS in 4m 00s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/12/check/openstack-tox-molecule/68cfdfe/tox/reports.html : SUCCESS in 14m 59s\n- openstack-tox-linters http://logs.openstack.org/19/667219/12/check/openstack-tox-linters/3b9717a/ : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":12},{"id":"bf87da9d642c6872916f77f18266d244603e8df8","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-30 14:28:04.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"be8638208f27afd2ec4bc739b14e2e9510c6b742","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:51:46.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/667219/13/check/tripleo-ci-centos-7-standalone/5e9b0cb/ : SUCCESS in 1h 39m 11s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/667219/13/check/tripleo-ci-fedora-28-standalone/ffd2bc5/ : SUCCESS in 1h 31m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/19/667219/13/check/tripleo-ci-centos-7-standalone-upgrade/b6afbac/ : SUCCESS in 2h 23m 33s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/19/667219/13/check/openstack-tox-docs/32a0479/html/ : SUCCESS in 4m 24s\n- openstack-tox-molecule http://logs.openstack.org/19/667219/13/check/openstack-tox-molecule/2da6ace/tox/reports.html : SUCCESS in 14m 44s\n- openstack-tox-linters http://logs.openstack.org/19/667219/13/check/openstack-tox-linters/e03de20/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":13},{"id":"a58396d6b81b8cfdaf728cd7ba6f78def1dfe0c5","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-04 19:20:22.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"0d1b1b76fb2a200de867a6853ddfdbef14bde08c","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-04 19:20:28.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"9a8d4275e946261aa8a0c7fe9054db88fc0d1b67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 20:41:15.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/19/667219/15/check/tripleo-ci-centos-7-standalone/04e8eeb/ : FAILURE in 37m 04s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/19/667219/15/check/tripleo-ci-fedora-28-standalone/965e4ef/ : FAILURE in 45m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/19/667219/15/check/tripleo-ci-centos-7-standalone-upgrade/ffa4673/ : FAILURE in 42m 30s (non-voting)\n- openstack-tox-docs https://logs.opendev.org/19/667219/15/check/openstack-tox-docs/7058d13/html/ : SUCCESS in 3m 31s\n- openstack-tox-molecule https://logs.opendev.org/19/667219/15/check/openstack-tox-molecule/6de41e6/tox/reports.html : SUCCESS in 14m 32s\n- openstack-tox-linters https://logs.opendev.org/19/667219/15/check/openstack-tox-linters/2efe42c/ : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":15},{"id":"0479b500a0b2034ae2d6fb4be96d9c3820c3b71a","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-04 21:29:55.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"1896bee85cbdd93bc3cfceb4ad0b6a28f9861702","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 23:00:37.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/19/667219/15/check/tripleo-ci-centos-7-standalone/1739cd7/ : SUCCESS in 1h 29m 37s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/19/667219/15/check/tripleo-ci-fedora-28-standalone/83f4699/ : FAILURE in 37m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/19/667219/15/check/tripleo-ci-centos-7-standalone-upgrade/385acb1/ : FAILURE in 42m 01s (non-voting)\n- openstack-tox-docs https://logs.opendev.org/19/667219/15/check/openstack-tox-docs/acd3918/html/ : SUCCESS in 3m 01s\n- openstack-tox-molecule https://logs.opendev.org/19/667219/15/check/openstack-tox-molecule/b0614c2/tox/reports.html : SUCCESS in 13m 34s\n- openstack-tox-linters https://logs.opendev.org/19/667219/15/check/openstack-tox-linters/a149d8e/ : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":15},{"id":"d3edd5d922653fe1e0cddee6ce8dc2a7a9e12fd5","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-07 22:43:46.000000000","message":"Patch Set 16: Commit message was updated.","accounts_in_message":[],"_revision_number":16},{"id":"e4527c2cf0b0e770a644b62d2bac5311d8c83dbc","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-07 22:43:58.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"c073b2ff77f3b37c8876a9d742abf15dcb58fa01","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 00:48:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/19/667219/17/check/tripleo-ci-centos-7-standalone/c299907/ : SUCCESS in 1h 40m 02s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/19/667219/17/check/tripleo-ci-fedora-28-standalone/52c525b/ : SUCCESS in 1h 33m 37s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/19/667219/17/check/tripleo-ci-centos-7-standalone-upgrade/060c01a/ : SUCCESS in 1h 55m 25s (non-voting)\n- openstack-tox-docs https://logs.opendev.org/19/667219/17/check/openstack-tox-docs/f679423/html/ : SUCCESS in 5m 05s\n- openstack-tox-molecule https://logs.opendev.org/19/667219/17/check/openstack-tox-molecule/06eed75/tox/reports.html : SUCCESS in 17m 42s\n- openstack-tox-linters https://logs.opendev.org/19/667219/17/check/openstack-tox-linters/680ca7f/ : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":17},{"id":"f74b052399935187370cce66eb97e3fb3eab5c29","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-08 03:52:18.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"d414f84130f5d9bda090b8861b0fa8ed638de0b0","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-08 09:15:21.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"b87f0b069db2d13a6d52b9a494aca13a096f87a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 14:26:35.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/19/667219/17/check/tripleo-ci-centos-7-standalone/b65cd83/ : SUCCESS in 1h 41m 12s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/19/667219/17/check/tripleo-ci-fedora-28-standalone/6d70b4b/ : SUCCESS in 1h 30m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/19/667219/17/check/tripleo-ci-centos-7-standalone-upgrade/ab7523f/ : SUCCESS in 3h 23m 47s (non-voting)\n- openstack-tox-docs https://logs.opendev.org/19/667219/17/check/openstack-tox-docs/f164a3c/html/ : SUCCESS in 5m 26s\n- openstack-tox-molecule https://logs.opendev.org/19/667219/17/check/openstack-tox-molecule/65d233f/tox/reports.html : SUCCESS in 15m 36s\n- openstack-tox-linters https://logs.opendev.org/19/667219/17/check/openstack-tox-linters/758c3d0/ : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":17},{"id":"69b037fab317ad9a8145ec9ebe0ba29f6252c66c","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-08 15:03:46.000000000","message":"Patch Set 18: Commit message was updated.","accounts_in_message":[],"_revision_number":18},{"id":"4b39f9f4406a91dcd1d60811597f247c4e0610c0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-08 15:06:17.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"e95f83d76a60e5ca2cab76f1edec8504b67777f8","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-08 15:06:41.000000000","message":"Patch Set 18: Code-Review+1\n\nwaiting on zuul","accounts_in_message":[],"_revision_number":18},{"id":"1ec570cb431fbc15c73102148096c7af1dda42ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 16:31:40.000000000","message":"Patch Set 18: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":18},{"id":"6a72cd531306ade33c24dde3c57cb6456b8ffce2","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-08 16:32:08.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"853842034d7734fa1d88ce5a5b9b7bf4d5b031bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 23:22:52.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/19/667219/18/check/tripleo-ci-centos-7-standalone/9e52129/ : SUCCESS in 1h 53m 32s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/19/667219/18/check/tripleo-ci-fedora-28-standalone/9f0890e/ : SUCCESS in 1h 32m 35s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/19/667219/18/check/tripleo-ci-centos-7-standalone-upgrade/a1972fd/ : SUCCESS in 2h 09m 56s (non-voting)\n- openstack-tox-docs https://logs.opendev.org/19/667219/18/check/openstack-tox-docs/f69d22c/html/ : SUCCESS in 4m 23s\n- openstack-tox-molecule https://logs.opendev.org/19/667219/18/check/openstack-tox-molecule/041692f/tox/reports.html : SUCCESS in 14m 39s\n- openstack-tox-linters https://logs.opendev.org/19/667219/18/check/openstack-tox-linters/dc14418/ : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":18},{"id":"2a14ce5686269296dcb0414ffb3a167950cda07e","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-09 00:02:47.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"dcc240e89ac27564c1537b23d2cc1e06710177f8","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-09 00:08:18.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"b5728ace0d98569cb2592b046cd4d6695e73085a","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-08-09 07:21:39.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"ae396481c21c25826165a0b5d1849c000cacf295","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-09 09:26:42.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"df2096e813170a1e841aaafcce72200267032d57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 09:26:57.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"9fa4239952585dfbe990d429ef0bafa167605260","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-09 10:15:37.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"23e63bcbcdc6cd53ca5eacd50e454c91cdceb176","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 12:13:52.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/19/667219/18/gate/openstack-tox-docs/0ec8983/html/ : SUCCESS in 4m 44s\n- openstack-tox-molecule https://logs.opendev.org/19/667219/18/gate/openstack-tox-molecule/28ecd61/tox/reports.html : SUCCESS in 15m 49s\n- openstack-tox-linters https://logs.opendev.org/19/667219/18/gate/openstack-tox-linters/df62ed4/ : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":18},{"id":"55f95eb829fb201ebae3ebd9e81addef63b32460","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 12:13:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"f95ed60441647fd54e32eeff65dd7f10da0f7a53","revisions":{"0e5941aa4296202e28ecf5ac81808c2b1d5d36a2":{"kind":"REWORK","_number":1,"created":"2019-06-24 20:36:47.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/1"}}},"commit":{"parents":[{"commit":"fa2df6e286e2e87dcef9b36932f62825748f612d","subject":"Merge \"Set featureset021 to run only skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fa2df6e286e2e87dcef9b36932f62825748f612d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0e5941aa4296202e28ecf5ac81808c2b1d5d36a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0e5941aa4296202e28ecf5ac81808c2b1d5d36a2"}]},"branch":"refs/heads/master"},"b0a8440204b8281ed99c274a437c2a3381ea03da":{"kind":"REWORK","_number":2,"created":"2019-06-25 08:57:05.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/2"}}},"commit":{"parents":[{"commit":"fa2df6e286e2e87dcef9b36932f62825748f612d","subject":"Merge \"Set featureset021 to run only skip list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fa2df6e286e2e87dcef9b36932f62825748f612d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-25 08:56:55.000000000","tz":180},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b0a8440204b8281ed99c274a437c2a3381ea03da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b0a8440204b8281ed99c274a437c2a3381ea03da"}]},"branch":"refs/heads/master"},"414cb98f37b2ed04ee4054e5ef0812d6beb79293":{"kind":"REWORK","_number":3,"created":"2019-06-26 14:34:13.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/19/667219/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/3"}}},"commit":{"parents":[{"commit":"6d8420ff3ca6898b314981dfbdd450a0936bf9de","subject":"Merge \"Bump Ansible to 2.7.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6d8420ff3ca6898b314981dfbdd450a0936bf9de"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-06-26 14:33:47.000000000","tz":60},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/414cb98f37b2ed04ee4054e5ef0812d6beb79293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/414cb98f37b2ed04ee4054e5ef0812d6beb79293"}]},"branch":"refs/heads/master"},"2ce5d33ec3b714b17b9fb4f65ad158812e62ca35":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-01 16:21:25.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/4"}}},"commit":{"parents":[{"commit":"2a81c126b6b5927c6308a0392143facc1f8d444a","subject":"Merge \"Remove featureset057\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2a81c126b6b5927c6308a0392143facc1f8d444a"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-07-01 16:21:25.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2ce5d33ec3b714b17b9fb4f65ad158812e62ca35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2ce5d33ec3b714b17b9fb4f65ad158812e62ca35"}]},"branch":"refs/heads/master"},"9b891ecae2781ab2ce18fdf440f17163d402686c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-22 08:53:37.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/5"}}},"commit":{"parents":[{"commit":"2a81c126b6b5927c6308a0392143facc1f8d444a","subject":"Merge \"Remove featureset057\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2a81c126b6b5927c6308a0392143facc1f8d444a"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-07-22 08:53:35.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9b891ecae2781ab2ce18fdf440f17163d402686c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9b891ecae2781ab2ce18fdf440f17163d402686c"}]},"branch":"refs/heads/master"},"a115be76077f85d3491d8c01f0147864e16ccf03":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2019-07-23 09:06:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/19/667219/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/6"}}},"commit":{"parents":[{"commit":"2a81c126b6b5927c6308a0392143facc1f8d444a","subject":"Merge \"Remove featureset057\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2a81c126b6b5927c6308a0392143facc1f8d444a"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-23 09:05:55.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\nDepends-On: https://review.opendev.org/#/c/672231/\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a115be76077f85d3491d8c01f0147864e16ccf03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a115be76077f85d3491d8c01f0147864e16ccf03"}]},"branch":"refs/heads/master"},"a430d5f9be21a021363681ce25fd90eb05a9843b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-24 17:38:07.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/19/667219/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/7"}}},"commit":{"parents":[{"commit":"4acd58cedf2816140c2fde81d65d13cbe8e6ab6d","subject":"Temporary exclude ceph-ansible rc11 from nautilus repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4acd58cedf2816140c2fde81d65d13cbe8e6ab6d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-24 17:38:07.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\nDepends-On: https://review.opendev.org/#/c/672231/\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a430d5f9be21a021363681ce25fd90eb05a9843b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a430d5f9be21a021363681ce25fd90eb05a9843b"}]},"branch":"refs/heads/master"},"d67ab67fe2cc39f4b4e270a507962aa45afa57cc":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-07-25 04:30:52.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/19/667219/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/8"}}},"commit":{"parents":[{"commit":"c9a04b18694a1d10d9480524049eee10c23ec4fa","subject":"Include ansible-config_template in delorean-current repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c9a04b18694a1d10d9480524049eee10c23ec4fa"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-25 04:30:52.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\nDepends-On: https://review.opendev.org/#/c/672231/\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d67ab67fe2cc39f4b4e270a507962aa45afa57cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d67ab67fe2cc39f4b4e270a507962aa45afa57cc"}]},"branch":"refs/heads/master"},"863a466498fb8a6289f693b1a8db4a2e7797a688":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-07-26 12:30:35.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/19/667219/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/9"}}},"commit":{"parents":[{"commit":"6133bf14f3330d923759e8ae4fdfc12820f90a00","subject":"Merge \"Revert \"Temporary exclude ceph-ansible rc11 from nautilus repo\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6133bf14f3330d923759e8ae4fdfc12820f90a00"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-26 12:30:35.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\nDepends-On: https://review.opendev.org/#/c/672231/\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/863a466498fb8a6289f693b1a8db4a2e7797a688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/863a466498fb8a6289f693b1a8db4a2e7797a688"}]},"branch":"refs/heads/master"},"1cac351805369749d1af80770ccb4e2efb6c369e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-07-26 16:19:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/19/667219/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/10"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-26 16:19:17.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\nDepends-On: https://review.opendev.org/#/c/672231/\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1cac351805369749d1af80770ccb4e2efb6c369e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1cac351805369749d1af80770ccb4e2efb6c369e"}]},"branch":"refs/heads/master"},"a6e1ad72a4ee449e99dc9cf7d957ff30a08f1c43":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2019-07-26 16:40:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/19/667219/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/11"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-26 16:40:34.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\n\nDepends-On: https://review.opendev.org/#/c/672231/\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a6e1ad72a4ee449e99dc9cf7d957ff30a08f1c43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a6e1ad72a4ee449e99dc9cf7d957ff30a08f1c43"}]},"branch":"refs/heads/master"},"4fe6c30d607c4e3fddf80b817a04efc1a8405fad":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2019-07-26 17:08:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/19/667219/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/12"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-26 17:08:20.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/#/c/671989/\n\nDepends-On: https://review.opendev.org/673021\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4fe6c30d607c4e3fddf80b817a04efc1a8405fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4fe6c30d607c4e3fddf80b817a04efc1a8405fad"}]},"branch":"refs/heads/master"},"e4d15caeae945366febe655bf5f772969b3f7ec3":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2019-07-30 14:28:04.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/13"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-07-30 14:28:01.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nDepends-On: https://review.opendev.org/673021\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e4d15caeae945366febe655bf5f772969b3f7ec3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e4d15caeae945366febe655bf5f772969b3f7ec3"}]},"branch":"refs/heads/master"},"86bf2a5ee3080fabb9d035f56f973fe749c1f86f":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2019-08-04 19:20:22.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/14"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-08-04 19:20:18.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/86bf2a5ee3080fabb9d035f56f973fe749c1f86f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/86bf2a5ee3080fabb9d035f56f973fe749c1f86f"}]},"branch":"refs/heads/master"},"8497e15e725795d2eeb54b6ba8eaf92727261077":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-08-04 19:20:28.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/15"}}},"commit":{"parents":[{"commit":"464df34ed4a3e8d0638654a3b9675529e885f3e5","subject":"Merge \"[RHEL8] Fixed rdo_dlrn in release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/464df34ed4a3e8d0638654a3b9675529e885f3e5"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-08-04 19:20:28.000000000","tz":0},"subject":"WIP: bump ansible to 2.8","message":"WIP: bump ansible to 2.8\n\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8497e15e725795d2eeb54b6ba8eaf92727261077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8497e15e725795d2eeb54b6ba8eaf92727261077"}]},"branch":"refs/heads/master"},"1b624ceda6f831e52c6dd10ec40dfc339cd45f21":{"kind":"NO_CODE_CHANGE","_number":16,"created":"2019-08-07 22:43:46.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/16"}}},"commit":{"parents":[{"commit":"464df34ed4a3e8d0638654a3b9675529e885f3e5","subject":"Merge \"[RHEL8] Fixed rdo_dlrn in release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/464df34ed4a3e8d0638654a3b9675529e885f3e5"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-08-07 22:43:43.000000000","tz":0},"subject":"Bump tripleo quickstart ansible to 2.8","message":"Bump tripleo quickstart ansible to 2.8\n\nDepends-On: https://review.opendev.org/675223\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1b624ceda6f831e52c6dd10ec40dfc339cd45f21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1b624ceda6f831e52c6dd10ec40dfc339cd45f21"}]},"branch":"refs/heads/master"},"18d9f243e95fc398cce0121273178f21bd65b73d":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-08-07 22:43:58.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/17"}}},"commit":{"parents":[{"commit":"bb7a19d5b8bc21604292dc62e7a1870f3d74a3b0","subject":"Add py3 support for get-overcloud-nodes script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bb7a19d5b8bc21604292dc62e7a1870f3d74a3b0"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-08-07 22:43:58.000000000","tz":0},"subject":"Bump tripleo quickstart ansible to 2.8","message":"Bump tripleo quickstart ansible to 2.8\n\nDepends-On: https://review.opendev.org/675223\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/18d9f243e95fc398cce0121273178f21bd65b73d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/18d9f243e95fc398cce0121273178f21bd65b73d"}]},"branch":"refs/heads/master"},"f95ed60441647fd54e32eeff65dd7f10da0f7a53":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2019-08-08 15:03:46.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/19/667219/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/667219/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/667219/18"}}},"commit":{"parents":[{"commit":"bb7a19d5b8bc21604292dc62e7a1870f3d74a3b0","subject":"Add py3 support for get-overcloud-nodes script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bb7a19d5b8bc21604292dc62e7a1870f3d74a3b0"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:36:36.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-08-08 15:03:38.000000000","tz":0},"subject":"Bump tripleo quickstart ansible to 2.8","message":"Bump tripleo quickstart ansible to 2.8\n\nDepends-On: https://review.opendev.org/675344\nChange-Id: Idebd929b40097442feee3efbdee3084cb38c9838\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f95ed60441647fd54e32eeff65dd7f10da0f7a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f95ed60441647fd54e32eeff65dd7f10da0f7a53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
