)]}'
{"id":"openstack%2Ftripleo-quickstart~667222","triplet_id":"openstack%2Ftripleo-quickstart~master~I7957f097bd55573eb0afed2964a26269a2b7b7ab","project":"openstack/tripleo-quickstart","branch":"master","topic":"ansible2.8-oooq","hashtags":[],"change_id":"I7957f097bd55573eb0afed2964a26269a2b7b7ab","subject":"DNM: test ansible2.8","status":"ABANDONED","created":"2019-06-24 20:42:28.000000000","updated":"2019-08-10 10:23:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c2de66c1ded51ac3a6b6cb859061384a5645e6b","_number":667222,"virtual_id_number":667222,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-08-08 01:35:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-08 02:58:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 01:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-08 02:58:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"61b92d4c0d0b0b00a19ce684d994633904ee37af","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-24 20:42:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08d79182adac6d92add45baf573f2990a7093b6e","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-24 20:42:46.000000000","message":"Topic changed from ooo2.8 to ansible2.8-oooq","accounts_in_message":[],"_revision_number":1},{"id":"d7e0871818a5e88c51cba647cdec4f643e8a3a87","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-24 21:00:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/22/667222/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4fed917/ : FAILURE in 14m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/22/667222/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f95cd8a/ : FAILURE in 14m 27s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/22/667222/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/fc3cda7/ : FAILURE in 4m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/22/667222/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c03fbad/ : FAILURE in 15m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/22/667222/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/71c01ba/ : FAILURE in 14m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/22/667222/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/714dd50/ : FAILURE in 14m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/22/667222/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c51dd51/ : FAILURE in 16m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd6b4275b03411661c7ad401e954200048c19b0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 23:02:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-undercloud-containers/de4d7d7/ : FAILURE in 12m 16s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/cbd84de/ : FAILURE in 11m 51s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ef9cca0/ : FAILURE in 12m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0026bda/ : FAILURE in 12m 35s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/22/667222/1/check/tripleo-ci-fedora-28-standalone/011ee5e/ : SUCCESS in 1h 50m 29s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-standalone-upgrade/541ba13/ : FAILURE in 11m 42s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/817d74a/ : FAILURE in 12m 42s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-containers-multinode-stein/46014eb/ : FAILURE in 14m 59s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-containers-multinode-rocky/3f260d1/ : FAILURE in 14m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-containers-multinode-queens/3bd57b4/ : FAILURE in 13m 43s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/22/667222/1/check/tripleo-ci-centos-7-containers-multinode-pike/c782d3d/ : FAILURE in 14m 17s\n- openstack-tox-docs http://logs.openstack.org/22/667222/1/check/openstack-tox-docs/402214d/html/ : SUCCESS in 4m 53s\n- tox-molecule http://logs.openstack.org/22/667222/1/check/tox-molecule/b2472d2/tox/reports.html : FAILURE in 8m 53s\n- openstack-tox-linters http://logs.openstack.org/22/667222/1/check/openstack-tox-linters/0b3fa01/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"98a9f8001b8d2bbcd797554c744564ab160ea174","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-06-25 08:58:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"8745b328292d47d151857f3317e0c444d4ad4224","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-25 12:53:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze01.softwarefactory-project.io:7979/ce99af6b4fa14a9599ce7af55d06d338 : POST_FAILURE in 3h 44m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze02.softwarefactory-project.io:7979/34d5f077f8fd43779a6ad9b4c0419739 : POST_FAILURE in 3h 43m 29s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/22/667222/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/fb8e7cf/ : SUCCESS in 1h 20m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/22/667222/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/45a6f8d/ : SUCCESS in 3h 12m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze03.softwarefactory-project.io:7979/bc45adcc08304d9ba2f560c588af6a62 : POST_FAILURE in 3h 16m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze07.softwarefactory-project.io:7979/01cab84908bb4afe8bf8fb808313b7a5 : POST_FAILURE in 3h 21m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze05.softwarefactory-project.io:7979/7493fa3580954e02b6ba62f57281c6a6 : POST_FAILURE in 3h 53m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1ad01ecb2e6bb704e1cd707f050e13ce4977a4c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 14:34:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers finger://ze04.openstack.org/af4326cea97647a5baa49aa3efab55b4 : RETRY_LIMIT in 13m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades finger://ze03.openstack.org/dc838a87a21d412097d2a530835db796 : RETRY_LIMIT in 14m 10s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates finger://ze01.openstack.org/aec56870c5574228b53d8bf6f811d027 : RETRY_LIMIT in 52m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/22/667222/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7810970/ : RETRY_LIMIT in 23m 16s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/22/667222/2/check/tripleo-ci-fedora-28-standalone/8c9e590/ : FAILURE in 31m 13s\n- tripleo-ci-centos-7-standalone-upgrade finger://ze10.openstack.org/3450df9ab11f49cf8dc8e4d9f990f0bc : RETRY_LIMIT in 58m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein finger://ze09.openstack.org/69ef40618403452ea7f9813e5fc5c092 : RETRY_LIMIT in 13m 36s\n- tripleo-ci-centos-7-containers-multinode-stein finger://ze05.openstack.org/d32958f3ddc048d7a86bd2dfade433ad : RETRY_LIMIT in 49m 45s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky finger://ze08.openstack.org/5a96902263c74aeda38eb2294641d491 : RETRY_LIMIT in 57m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens finger://ze07.openstack.org/cf8b85539ba544d68a0df245893c7f76 : RETRY_LIMIT in 44m 24s\n- tripleo-ci-centos-7-containers-multinode-pike finger://ze01.openstack.org/0f1308bf12cd4353b42de798719110c5 : RETRY_LIMIT in 23m 53s\n- openstack-tox-docs http://logs.openstack.org/22/667222/2/check/openstack-tox-docs/3ff6dbc/html/ : SUCCESS in 5m 02s\n- tox-molecule http://logs.openstack.org/22/667222/2/check/tox-molecule/8b2c850/tox/reports.html : FAILURE in 6m 36s\n- openstack-tox-linters http://logs.openstack.org/22/667222/2/check/openstack-tox-linters/4306cf2/ : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":2},{"id":"599a49ba737dbee5a1705a3b9454a6f9a497df3b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-01 16:23:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"30d0cdd410b5be4ef7efa24e9221c3b0e7e46f02","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 21:10:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze05.softwarefactory-project.io:7979/b96f742d068b47cbb6fbedd35e9d8a23 : POST_FAILURE in 3h 31m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze01.softwarefactory-project.io:7979/d74415efa7da453489138bbdd1e493b3 : POST_FAILURE in 3h 25m 08s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 finger://ze03.softwarefactory-project.io:7979/c6192b6b8dad4a829da258b542762e49 : POST_FAILURE in 2h 21m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/22/667222/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/01cc17a/ : FAILURE in 2h 55m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze05.softwarefactory-project.io:7979/d9c955911ff345c79d382aff3a7392a8 : POST_FAILURE in 3h 03m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze02.softwarefactory-project.io:7979/d839e66206454cb0a1d219a9a5b88887 : POST_FAILURE in 3h 13m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze05.softwarefactory-project.io:7979/6f80b8e107394b27b6fe74bd0ec106ce : POST_FAILURE in 3h 49m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee992574e854e1a58050a551678bf8b48dfe669d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 01:28:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-undercloud-containers/b9b8274/ : SUCCESS in 1h 24m 17s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/444df3f/ : FAILURE in 2h 08m 48s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/931f455/ : SUCCESS in 2h 13m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c6d0b44/ : SUCCESS in 2h 46m 37s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/22/667222/3/check/tripleo-ci-fedora-28-standalone/da51493/ : SUCCESS in 1h 57m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-standalone-upgrade/29f498c/ : SUCCESS in 2h 42m 31s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/3424a44/ : SUCCESS in 2h 05m 32s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-stein/30f2d13/ : SUCCESS in 2h 01m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-rocky/624d495/ : SUCCESS in 2h 15m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-queens/2c28c43/ : SUCCESS in 2h 13m 00s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-pike/6fbc9b1/ : SUCCESS in 2h 14m 29s\n- openstack-tox-docs http://logs.openstack.org/22/667222/3/check/openstack-tox-docs/c84985f/html/ : SUCCESS in 6m 15s\n- tox-molecule http://logs.openstack.org/22/667222/3/check/tox-molecule/5730b42/tox/reports.html : SUCCESS in 16m 07s\n- openstack-tox-linters http://logs.openstack.org/22/667222/3/check/openstack-tox-linters/c79d679/ : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":3},{"id":"e8841fa10652cebc31c9ca21f0beda5b262f6a94","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-09 07:16:53.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"549c7f22a4101e3fd2d44429ce745c324872818a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 11:17:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze01.softwarefactory-project.io:7979/9623097b4d9243b2bb8da424c237deee : POST_FAILURE in 3h 58m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze04.softwarefactory-project.io:7979/1c8a085855a04d1dad5eac5fd4c8b857 : POST_FAILURE in 3h 52m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze01.softwarefactory-project.io:7979/010386efc006489c8cba154f5129762b : POST_FAILURE in 3h 14m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze04.softwarefactory-project.io:7979/3c84e34eff9d4baf8d20be0f02516796 : POST_FAILURE in 3h 05m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze07.softwarefactory-project.io:7979/75e640eaee4c45138e9367deee7e9c3f : POST_FAILURE in 3h 28m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze03.softwarefactory-project.io:7979/a19d3dd081b24ec6aa4d7015b90cc5a1 : POST_FAILURE in 3h 52m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ba56aad761901cdb089c789ec48d2338e0a77406","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 11:56:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-undercloud-containers/6f1c335/ : SUCCESS in 1h 44m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d62d04e/ : FAILURE in 2h 12m 13s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c6dc3e2/ : SUCCESS in 2h 05m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/acfd200/ : SUCCESS in 2h 16m 25s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/22/667222/3/check/tripleo-ci-fedora-28-standalone/f539131/ : SUCCESS in 1h 53m 53s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-standalone-upgrade/1b2e625/ : SUCCESS in 2h 17m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/79eb678/ : SUCCESS in 2h 13m 52s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-stein/673ffd3/ : SUCCESS in 2h 21m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-rocky/c140c7f/ : SUCCESS in 2h 09m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-queens/925b910/ : SUCCESS in 2h 17m 30s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/22/667222/3/check/tripleo-ci-centos-7-containers-multinode-pike/f15bebb/ : SUCCESS in 2h 19m 18s\n- openstack-tox-docs http://logs.openstack.org/22/667222/3/check/openstack-tox-docs/2908fd7/html/ : SUCCESS in 3m 58s\n- tox-molecule http://logs.openstack.org/22/667222/3/check/tox-molecule/6809d0e/tox/reports.html : SUCCESS in 14m 58s\n- openstack-tox-linters http://logs.openstack.org/22/667222/3/check/openstack-tox-linters/b2f2cd3/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6b30ac5c8b81699f2f2100ed183481788953a596","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-22 08:54:27.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c7a47392134d3a24238c338501482f14570cda2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 11:35:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-undercloud-containers/c7d5ad3/ : SUCCESS in 1h 32m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/1807f08/ : FAILURE in 1h 58m 44s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode/55f4208/ : FAILURE in 2h 24m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a85741c/ : SUCCESS in 2h 20m 17s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ccc65c5/ : FAILURE in 2h 34m 14s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/22/667222/4/check/tripleo-ci-fedora-28-standalone/87f691c/ : SUCCESS in 1h 33m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-standalone-upgrade/31b2509/ : FAILURE in 1h 54m 22s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-stein/ba31571/ : SUCCESS in 2h 12m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-rocky/03a3d10/ : SUCCESS in 2h 35m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-queens/a7c21e2/ : SUCCESS in 2h 20m 13s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-pike/80d8084/ : SUCCESS in 2h 12m 21s\n- openstack-tox-docs http://logs.openstack.org/22/667222/4/check/openstack-tox-docs/0265ddc/html/ : SUCCESS in 4m 24s\n- openstack-tox-molecule http://logs.openstack.org/22/667222/4/check/openstack-tox-molecule/23792af/tox/reports.html : SUCCESS in 13m 33s\n- openstack-tox-linters http://logs.openstack.org/22/667222/4/check/openstack-tox-linters/c90ee9a/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":4},{"id":"81ef4c2af6c6fb8011eafc0f27d14f28ceda6368","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 11:35:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/155a0a0/ : RETRY_LIMIT in 13m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0fc3293/ : RETRY_LIMIT in 11m 02s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d852d49/ : RETRY_LIMIT in 12m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/17084ef/ : RETRY_LIMIT in 11m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1e108c6/ : RETRY_LIMIT in 13m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/f33757c/ : FAILURE in 1h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"67dfa4ba32624f5455f21d724155d7abb15cd706","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-30 14:29:00.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"04438b83ec2cf23b8ba35ec0ac01ef09261d9ad6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 17:40:21.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/550bbfc/ : FAILURE in 1h 06m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze05.softwarefactory-project.io:7979/c0177a7f1049450ca63fd59013bbaeab : RETRY_LIMIT in 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze07.softwarefactory-project.io:7979/96168219866b41528b1fb1ca0e5809a6 : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/36cdb9a/ : SUCCESS in 2h 17m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/22/667222/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/a8b72c7/ : FAILURE in 1h 15m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze06.softwarefactory-project.io:7979/39e0483fd77d48faafdfe6d792d5d80b : RETRY_LIMIT in 6s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"09c737177184db5504c19ba41fdb94a1a5cfa554","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:43:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-undercloud-containers/72d9378/ : SUCCESS in 1h 34m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/df39401/ : FAILURE in 1h 46m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode/9866f5b/ : FAILURE in 2h 19m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a9f07b4/ : SUCCESS in 2h 31m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/872dd11/ : SUCCESS in 2h 24m 23s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/22/667222/4/check/tripleo-ci-fedora-28-standalone/b94ce9f/ : SUCCESS in 1h 33m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-standalone-upgrade/3cae27a/ : SUCCESS in 2h 05m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-stein/2641456/ : FAILURE in 2h 42m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-rocky/c2c7ca4/ : FAILURE in 2h 21m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-queens/1f1f97b/ : FAILURE in 2h 14m 02s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/22/667222/4/check/tripleo-ci-centos-7-containers-multinode-pike/b556497/ : SUCCESS in 2h 28m 58s\n- openstack-tox-docs http://logs.openstack.org/22/667222/4/check/openstack-tox-docs/da98662/html/ : SUCCESS in 3m 52s\n- openstack-tox-molecule http://logs.openstack.org/22/667222/4/check/openstack-tox-molecule/7c79325/tox/reports.html : SUCCESS in 13m 47s\n- openstack-tox-linters http://logs.openstack.org/22/667222/4/check/openstack-tox-linters/7ac0b43/ : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c7194086c82e02c0db8ad7fc5024e56fcde33588","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-04 19:20:37.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"25e7ce2f32aa0d48607776caf4dcbb6b738dd3ba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 22:07:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-undercloud-containers/a1443fa/ : SUCCESS in 1h 29m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2057e95/ : FAILURE in 56m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-containers-multinode/2970bfe/ : SUCCESS in 2h 22m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/881311a/ : SUCCESS in 1h 57m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/54542f7/ : SUCCESS in 2h 33m 49s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/22/667222/5/check/tripleo-ci-fedora-28-standalone/7ef7057/ : FAILURE in 41m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-standalone-upgrade/e99f31e/ : FAILURE in 36m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-containers-multinode-stein/11e1f26/ : FAILURE in 1h 16m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-containers-multinode-rocky/d360e2b/ : SUCCESS in 1h 58m 18s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-containers-multinode-queens/0dda400/ : SUCCESS in 2h 06m 09s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/22/667222/5/check/tripleo-ci-centos-7-containers-multinode-pike/7fa7ccc/ : SUCCESS in 1h 59m 15s\n- openstack-tox-docs https://logs.opendev.org/22/667222/5/check/openstack-tox-docs/4dcf415/html/ : SUCCESS in 5m 20s\n- openstack-tox-molecule https://logs.opendev.org/22/667222/5/check/openstack-tox-molecule/8eb47e4/tox/reports.html : SUCCESS in 13m 24s\n- openstack-tox-linters https://logs.opendev.org/22/667222/5/check/openstack-tox-linters/44fb608/ : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":5},{"id":"b2b55e868c12434c4eee31a32873fa635918e436","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-04 22:45:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/22/667222/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3d57465/ : SUCCESS in 3h 07m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/22/667222/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/1f9d0a3/ : SUCCESS in 3h 22m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/22/667222/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/6e25b59/ : SUCCESS in 2h 22m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/22/667222/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b65139a/ : SUCCESS in 2h 21m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/22/667222/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/cfeb1e7/ : FAILURE in 1h 09m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/22/667222/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/62dfed9/ : FAILURE in 1h 53m 06s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/22/667222/5/openstack-check/tripleo-ci-rhel-8-standalone-rdo/d9f1dfc/ : FAILURE in 14m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"09d891fddbbd7cf303bba8ad0f862e000a8d5c79","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-07 22:44:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"b3bfa71aa58e01308d0c1b60d4e9b91f7bcbb330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 01:35:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-undercloud-containers/8148db3/ : SUCCESS in 1h 26m 58s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3486a48/ : FAILURE in 1h 53m 21s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-containers-multinode/22a8622/ : SUCCESS in 2h 17m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/daa9e66/ : SUCCESS in 2h 24m 51s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/089fc9c/ : SUCCESS in 2h 34m 05s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/22/667222/6/check/tripleo-ci-fedora-28-standalone/01af7e7/ : SUCCESS in 1h 43m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-standalone-upgrade/38ce192/ : SUCCESS in 2h 03m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-containers-multinode-stein/8bc4ade/ : SUCCESS in 2h 10m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-containers-multinode-rocky/5c6dc86/ : SUCCESS in 2h 28m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-containers-multinode-queens/8484744/ : SUCCESS in 2h 18m 55s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/22/667222/6/check/tripleo-ci-centos-7-containers-multinode-pike/a586a89/ : SUCCESS in 2h 30m 10s\n- openstack-tox-docs https://logs.opendev.org/22/667222/6/check/openstack-tox-docs/3016b8a/html/ : SUCCESS in 3m 51s\n- openstack-tox-molecule https://logs.opendev.org/22/667222/6/check/openstack-tox-molecule/d437f0f/tox/reports.html : SUCCESS in 14m 58s\n- openstack-tox-linters https://logs.opendev.org/22/667222/6/check/openstack-tox-linters/cedb323/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":6},{"id":"2898b7c175cce70925ce344ffe5b99f28991a37b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-08 02:58:19.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/22/667222/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/ca08720/ : SUCCESS in 2h 33m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/22/667222/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/f3146f8/ : SUCCESS in 2h 35m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/22/667222/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b46c771/ : SUCCESS in 2h 25m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/22/667222/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b13e3c9/ : FAILURE in 2h 53m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/22/667222/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/de79ec2/ : SUCCESS in 2h 38m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/22/667222/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/fd1c6c4/ : SUCCESS in 3h 25m 50s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/22/667222/6/openstack-check/tripleo-ci-rhel-8-standalone-rdo/ab47eee/ : FAILURE in 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6c2de66c1ded51ac3a6b6cb859061384a5645e6b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-10 10:23:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9398867dbaba264849bccd134d4dcfceba456e1e","revisions":{"5642cd3762e3e01154a4de31610d10dc83979213":{"kind":"REWORK","_number":1,"created":"2019-06-24 20:42:28.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/22/667222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/22/667222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/1"}}},"commit":{"parents":[{"commit":"0e5941aa4296202e28ecf5ac81808c2b1d5d36a2","subject":"WIP: bump ansible to 2.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0e5941aa4296202e28ecf5ac81808c2b1d5d36a2"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:42:14.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:42:14.000000000","tz":180},"subject":"DNM: test ansible2.8","message":"DNM: test ansible2.8\n\nChange-Id: I7957f097bd55573eb0afed2964a26269a2b7b7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5642cd3762e3e01154a4de31610d10dc83979213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5642cd3762e3e01154a4de31610d10dc83979213"}]},"branch":"refs/heads/master"},"28764bfb726b551ab24427ba83d835ab016cb589":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-25 08:58:31.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/22/667222/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/22/667222/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/2"}}},"commit":{"parents":[{"commit":"b0a8440204b8281ed99c274a437c2a3381ea03da","subject":"WIP: bump ansible to 2.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b0a8440204b8281ed99c274a437c2a3381ea03da"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:42:14.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-25 08:58:15.000000000","tz":180},"subject":"DNM: test ansible2.8","message":"DNM: test ansible2.8\n\nChange-Id: I7957f097bd55573eb0afed2964a26269a2b7b7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/28764bfb726b551ab24427ba83d835ab016cb589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/28764bfb726b551ab24427ba83d835ab016cb589"}]},"branch":"refs/heads/master"},"15b989c2281432d798095559ce5d427a8b8a7416":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-01 16:23:25.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/22/667222/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/22/667222/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/3"}}},"commit":{"parents":[{"commit":"2ce5d33ec3b714b17b9fb4f65ad158812e62ca35","subject":"WIP: bump ansible to 2.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2ce5d33ec3b714b17b9fb4f65ad158812e62ca35"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:42:14.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-07-01 16:23:25.000000000","tz":0},"subject":"DNM: test ansible2.8","message":"DNM: test ansible2.8\n\nChange-Id: I7957f097bd55573eb0afed2964a26269a2b7b7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/15b989c2281432d798095559ce5d427a8b8a7416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/15b989c2281432d798095559ce5d427a8b8a7416"}]},"branch":"refs/heads/master"},"ffa466e41273964349ea6c53bb48557fcb453d1b":{"kind":"NO_CHANGE","_number":4,"created":"2019-07-22 08:54:27.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/22/667222/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/22/667222/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/4"}}},"commit":{"parents":[{"commit":"9b891ecae2781ab2ce18fdf440f17163d402686c","subject":"WIP: bump ansible to 2.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9b891ecae2781ab2ce18fdf440f17163d402686c"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:42:14.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-07-22 08:54:27.000000000","tz":0},"subject":"DNM: test ansible2.8","message":"DNM: test ansible2.8\n\nChange-Id: I7957f097bd55573eb0afed2964a26269a2b7b7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ffa466e41273964349ea6c53bb48557fcb453d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ffa466e41273964349ea6c53bb48557fcb453d1b"}]},"branch":"refs/heads/master"},"1633248315972a7555f72ac8a21277c753dee79b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-04 19:20:37.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/22/667222/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/22/667222/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/5"}}},"commit":{"parents":[{"commit":"8497e15e725795d2eeb54b6ba8eaf92727261077","subject":"WIP: bump ansible to 2.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8497e15e725795d2eeb54b6ba8eaf92727261077"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:42:14.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-08-04 19:20:37.000000000","tz":0},"subject":"DNM: test ansible2.8","message":"DNM: test ansible2.8\n\nChange-Id: I7957f097bd55573eb0afed2964a26269a2b7b7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1633248315972a7555f72ac8a21277c753dee79b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1633248315972a7555f72ac8a21277c753dee79b"}]},"branch":"refs/heads/master"},"9398867dbaba264849bccd134d4dcfceba456e1e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-07 22:44:04.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/22/667222/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/22/667222/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/22/667222/6"}}},"commit":{"parents":[{"commit":"18d9f243e95fc398cce0121273178f21bd65b73d","subject":"Bump tripleo quickstart ansible to 2.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/18d9f243e95fc398cce0121273178f21bd65b73d"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-06-24 20:42:14.000000000","tz":180},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2019-08-07 22:44:04.000000000","tz":0},"subject":"DNM: test ansible2.8","message":"DNM: test ansible2.8\n\nChange-Id: I7957f097bd55573eb0afed2964a26269a2b7b7ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9398867dbaba264849bccd134d4dcfceba456e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9398867dbaba264849bccd134d4dcfceba456e1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
