)]}'
{"id":"openstack%2Ftripleo-quickstart~668402","triplet_id":"openstack%2Ftripleo-quickstart~master~I5b827d8e2011fb39e1df7786f012e5551cc4e0d9","project":"openstack/tripleo-quickstart","branch":"master","topic":"bug/1834655","hashtags":[],"change_id":"I5b827d8e2011fb39e1df7786f012e5551cc4e0d9","subject":"Fix failure in featureset039","status":"MERGED","created":"2019-07-01 09:48:48.000000000","updated":"2019-07-01 21:47:35.000000000","submitted":"2019-07-01 21:47:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668402-1562017655662-a4a7ac61","meta_rev_id":"1b5040e5eb6748d6493c73fd9f8a6929a7795196","_number":668402,"virtual_id_number":668402,"owner":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-07-01 21:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-01 13:51:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"date":"2019-07-01 14:24:01.000000000","_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-07-01 12:29:48.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2019-07-01 09:50:48.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"date":"2019-07-01 09:52:37.000000000","_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"date":"2019-07-01 09:52:39.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2019-07-01 18:57:09.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-01 09:52:46.000000000","_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 09:52:37.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-07-01 09:52:39.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-07-01 09:52:46.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-07-01 12:29:48.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-07-01 13:51:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-01 18:57:09.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-01 21:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3614f2974d2c53d85c526cb23ecf8a76bc6296b","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-01 09:48:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc23ac785e6691ab4661af219368afb1b959ef6","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-01 09:50:48.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"93301979dc3b678c88dd3ed616935ed026f90bd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 12:19:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-undercloud-containers/672cea2/ : SUCCESS in 1h 35m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/70ed36f/ : FAILURE in 1h 59m 17s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ae088d4/ : SUCCESS in 2h 24m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8bee7d1/ : FAILURE in 1h 13m 08s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/02/668402/1/check/tripleo-ci-fedora-28-standalone/a64fe92/ : SUCCESS in 1h 36m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-standalone-upgrade/9ec2a97/ : SUCCESS in 2h 18m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/b7ddb74/ : SUCCESS in 2h 11m 30s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-containers-multinode-stein/fc9b0e8/ : SUCCESS in 2h 00m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-containers-multinode-rocky/d54ff0a/ : SUCCESS in 2h 03m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-containers-multinode-queens/bb3cc0a/ : SUCCESS in 1h 58m 22s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/02/668402/1/check/tripleo-ci-centos-7-containers-multinode-pike/9fc7f3b/ : SUCCESS in 2h 14m 58s\n- openstack-tox-docs http://logs.openstack.org/02/668402/1/check/openstack-tox-docs/57c9c04/html/ : SUCCESS in 3m 26s\n- tox-molecule http://logs.openstack.org/02/668402/1/check/tox-molecule/c1252fb/tox/reports.html : SUCCESS in 13m 36s\n- openstack-tox-linters http://logs.openstack.org/02/668402/1/check/openstack-tox-linters/dd7f69f/ : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"3f58944fa6d1e773330258238457a26857a8702e","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-01 12:29:41.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"8a775aa92f1129672605ae6a15e84d498d8c7f5b","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-07-01 12:29:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f52cc497d76641c9b7bada7952045f5a7c502073","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 13:51:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/32a57f0/ : SUCCESS in 3h 23m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/61a335d/ : SUCCESS in 3h 27m 31s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/688a64d/ : SUCCESS in 1h 22m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0813afa/ : SUCCESS in 2h 13m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/f8bfde2/ : SUCCESS in 2h 04m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/9fadd1d/ : FAILURE in 1h 31m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 finger://ze06.softwarefactory-project.io:7979/fbb003ac88f747a2a13ae937665cef0d : POST_FAILURE in 4h 01m 08s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 finger://ze04.softwarefactory-project.io:7979/a91cc005d5bb475bb96bcf77c986fa91 : POST_FAILURE in 3h 45m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32ebefb9e9fe3cd3ad8d2d75f4838843001c34e1","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-01 14:23:53.000000000","message":"Patch Set 1:\n\nAlthouhg we had a POST_FAILURE on featureset039 the job pass the point of the issue and tempest finishes successfully http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/a91cc00/logs/tempest.html.gz so we are good to merge","accounts_in_message":[],"_revision_number":1},{"id":"b8f7b8f8a87e36019573cebb11deb5dabc3e560f","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-01 14:24:01.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"6e869566af5f883e71bcac6ab9334fd051e772c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 18:09:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze07.softwarefactory-project.io:7979/3070d603f274435baa66e88db61fcf42 : POST_FAILURE in 3h 25m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze03.softwarefactory-project.io:7979/f9616be1bd8a45e48333534654e4884d : POST_FAILURE in 3h 28m 27s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/0996954/ : SUCCESS in 2h 14m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a0a27e0/ : FAILURE in 2h 47m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch finger://ze03.softwarefactory-project.io:7979/58d9118ecb9b49e1bdd1d65d64b4f4fe : POST_FAILURE in 3h 00m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze07.softwarefactory-project.io:7979/6520857ad22f4530b7f4b18cbc57e347 : POST_FAILURE in 3h 16m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/99ed3ba/ : FAILURE in 1h 54m 41s (non-voting)\n- tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039 http://logs.rdoproject.org/02/668402/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp_1supp-featureset039/feb0a89/ : FAILURE in 3h 43m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d083d3f11345823c6864f3d6724b33cbdd06ccba","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-01 18:57:09.000000000","message":"Patch Set 1: Workflow+1\n\n039 passed the failing step, so ok to merge","accounts_in_message":[],"_revision_number":1},{"id":"d116cef34f29bc35faa821bbe39eddf8a7ff3013","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 18:57:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef72be7dee349e39c880ffd637fabf1372be14d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:47:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4ec8b7a23b1c2e96b0522b0d4bdc0b5acafcf7a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 21:47:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/02/668402/1/gate/tripleo-ci-centos-7-undercloud-containers/7e9b6ba/ : SUCCESS in 1h 44m 00s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/02/668402/1/gate/tripleo-ci-centos-7-containers-multinode/e4fee3f/ : SUCCESS in 2h 14m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/02/668402/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cef2ca6/ : SUCCESS in 2h 42m 37s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/02/668402/1/gate/tripleo-ci-centos-7-standalone-upgrade-stein/17310e3/ : SUCCESS in 2h 15m 15s\n- openstack-tox-docs http://logs.openstack.org/02/668402/1/gate/openstack-tox-docs/8d7f137/html/ : SUCCESS in 3m 35s\n- tox-molecule http://logs.openstack.org/02/668402/1/gate/tox-molecule/1be2664/tox/reports.html : SUCCESS in 13m 21s\n- openstack-tox-linters http://logs.openstack.org/02/668402/1/gate/openstack-tox-linters/244dd7e/ : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9df739dd2a569a92545072c86cf9c4a8efdce796","revisions":{"9df739dd2a569a92545072c86cf9c4a8efdce796":{"kind":"REWORK","_number":1,"created":"2019-07-01 09:48:48.000000000","uploader":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"ref":"refs/changes/02/668402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/02/668402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/668402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/668402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/668402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/668402/1"}}},"commit":{"parents":[{"commit":"2a81c126b6b5927c6308a0392143facc1f8d444a","subject":"Merge \"Remove featureset057\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2a81c126b6b5927c6308a0392143facc1f8d444a"}]}],"author":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-07-01 09:48:31.000000000","tz":120},"committer":{"name":"Arx Cruz","email":"arxcruz@gmail.com","date":"2019-07-01 09:48:31.000000000","tz":120},"subject":"Fix failure in featureset039","message":"Fix failure in featureset039\n\nWith the upgrade to ansible 2.7 this role started to fail on\nfeatureset039 due the miss of python3.\nThis patch fix it removing the args from shell module and updating the\npython script to use /usr/bin/env.\n\nChange-Id: I5b827d8e2011fb39e1df7786f012e5551cc4e0d9\nCloses-Bug: #1834655\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9df739dd2a569a92545072c86cf9c4a8efdce796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9df739dd2a569a92545072c86cf9c4a8efdce796"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"2a81c126b6b5927c6308a0392143facc1f8d444a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
