)]}'
{"id":"openstack%2Ftripleo-quickstart~669223","triplet_id":"openstack%2Ftripleo-quickstart~master~Iad1734ad789ad91915c4e0374cb2afcc1f289b23","project":"openstack/tripleo-quickstart","branch":"master","topic":"molecule","hashtags":[],"change_id":"Iad1734ad789ad91915c4e0374cb2afcc1f289b23","subject":"Adds centos-8 testing to install-deps","status":"MERGED","created":"2019-07-04 21:30:09.000000000","updated":"2020-03-30 15:24:30.000000000","submitted":"2020-03-30 15:18:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669223-1585581487626-6c540cf1","meta_rev_id":"3c4a37a8103ff8ec4fdabf2f77ccb90f3a51c988","_number":669223,"virtual_id_number":669223,"owner":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-30 15:18:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-03-25 17:23:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-25 11:59:27.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2020-03-30 07:22:19.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-30 08:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2020-03-30 09:48:35.000000000","_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 21:31:18.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-07-04 21:31:29.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-07-04 21:31:52.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-09-11 16:51:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-01-30 16:48:04.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2020-01-30 18:40:34.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-03-25 11:59:27.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-25 17:23:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-30 07:22:19.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-03-30 08:25:14.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-03-30 15:18:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9dc9703bec3fb9e799310bac2086647f393f2cfd","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-07-04 21:30:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"712131a03f9416303307c7b298b9c2e180698ee6","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-07-04 22:00:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6850e8c724cb68ea651a9bca9fec09d9b667d232","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 00:18:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-undercloud-containers/7fffc60/ : SUCCESS in 1h 23m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/062f0e5/ : FAILURE in 1h 44m 49s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b5e4273/ : SUCCESS in 2h 03m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8f12c0f/ : SUCCESS in 2h 14m 27s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/23/669223/2/check/tripleo-ci-fedora-28-standalone/cd37a88/ : FAILURE in 25m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-standalone-upgrade/0ad722d/ : SUCCESS in 2h 02m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-standalone-upgrade-stein/f5bf449/ : SUCCESS in 2h 07m 49s\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-containers-multinode-stein/25a7e45/ : SUCCESS in 2h 05m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-containers-multinode-rocky/d9a2432/ : SUCCESS in 2h 00m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-containers-multinode-queens/d05e891/ : SUCCESS in 1h 48m 07s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/23/669223/2/check/tripleo-ci-centos-7-containers-multinode-pike/e54421c/ : SUCCESS in 2h 04m 31s\n- openstack-tox-docs http://logs.openstack.org/23/669223/2/check/openstack-tox-docs/c18fa97/html/ : SUCCESS in 3m 32s\n- tox-molecule http://logs.openstack.org/23/669223/2/check/tox-molecule/a693165/tox/reports.html : FAILURE in 11m 07s\n- openstack-tox-linters http://logs.openstack.org/23/669223/2/check/openstack-tox-linters/531ca62/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"166ea6b5e4c261a52db1781f506fbdd22a1cec9e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 01:00:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/23/669223/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b97ad45/ : FAILURE in 1h 03m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/23/669223/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6e19950/ : FAILURE in 1h 01m 42s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/23/669223/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b11df0f/ : SUCCESS in 1h 32m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/23/669223/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c7afd6a/ : SUCCESS in 2h 03m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/23/669223/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3504cbc/ : SUCCESS in 2h 10m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/23/669223/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/2319d3b/ : SUCCESS in 2h 23m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/23/669223/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/95cb6f2/ : SUCCESS in 2h 57m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d96732b307b87c4285ff5a1cef828f90025cc668","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-07-05 11:41:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e71d17ec937dba23d067907b7eba932497123765","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-07-05 11:41:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"506b5a0c41554492a76fd25f818ada07b18adae3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 11:59:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tox-molecule http://logs.openstack.org/23/669223/4/check/tox-molecule/d26ee59/tox/reports.html : FAILURE in 16m 05s\n- openstack-tox-linters http://logs.openstack.org/23/669223/4/check/openstack-tox-linters/83ae14e/ : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":4},{"id":"9d0fa7d420371f42a43b2e477fbfa0616728e995","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-05 14:27:43.000000000","message":"Patch Set 4:\n\nLooks like the tox-molcule test is failing run Ansible on RHEL 8 because it can\u0027t find the Python interpreter. We either need to update Ansible to 2.8 as part of the CI job or somehow set the host variable ansible_python_interpreter\u003d/usr/libexec/platform-python for the RHEL 8 instance.","accounts_in_message":[],"_revision_number":4},{"id":"3c341c7e5d41feb27ebc688005483f9183adbd38","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 14:28:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/23/669223/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/74e15e2/ : SUCCESS in 2h 30m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/23/669223/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2bdafc2/ : SUCCESS in 2h 45m 15s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/23/669223/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/2664012/ : SUCCESS in 1h 17m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/23/669223/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/9eca93e/ : FAILURE in 2h 12m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/23/669223/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2cc9e76/ : SUCCESS in 2h 05m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/23/669223/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c9e5a53/ : SUCCESS in 2h 33m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/23/669223/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/0eb8a6d/ : FAILURE in 1h 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"330baf26f931e97999c9949a48e7ca4917103b55","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-07-05 14:42:59.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"53a5cc6619fab26ebea62f84e64fbe4b4fcb0920","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 14:51:17.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"b07429914fd763d3e0167daf23dcf4f4b2ee4c8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 14:51:18.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":5},{"id":"9fa8a1bd1a529d7b0ef0d6941e3a88d9afe20951","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-09-05 16:11:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"21fab7228658e7f16025795d6c4c47773adb826b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-05 19:09:57.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/23/669223/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b7ce884/ : SUCCESS in 2h 41m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/23/669223/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/8f66118/ : FAILURE in 1h 02m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/23/669223/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6d6ecfd/ : SUCCESS in 2h 14m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/23/669223/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/d4424bd/ : SUCCESS in 2h 56m 55s","accounts_in_message":[],"_revision_number":6},{"id":"48e72e8b7e833266f970822dd58545be1fefc4a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-05 23:50:10.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ea1a5998f3184e0b8f9acdd4d05dbc1d : SUCCESS in 2m 10s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ff6ad717cc20473d959aa831b6be3437 : SUCCESS in 12m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/05290f3a48c949b9a48bcaa486ed3844 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":6},{"id":"9be6c6e4194134d9971f8f84b9d8b34e37602c15","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-09-10 13:14:43.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"7c69a8244012404f274f8390f7d2bc0207a9050d","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-09-10 13:15:02.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"77b594121f304cc7894bfaeeb9a30c4427ff250b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 13:15:48.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch : NODE_FAILURE in 0s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"868f4639f3fc3412d6c7a44ba0da3f55f643dc1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 13:49:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6c1f08fa4573464bae81bd26aa441d01 : SUCCESS in 13m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99a7d84bc30b44c4a7dd9b0a4a7c8abe : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":8},{"id":"9ef3a0e806b6e2136faa76a1c3309def49055503","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-09-11 09:03:40.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"82e0f4474d8707847f7a3754e5b4fe01ba0649b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 09:22:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/01369bc03b6849f79fcb60d52ada3365 : SUCCESS in 13m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9571642bbd6e4dd6949119fdff1efab0 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":8},{"id":"be507d3bcdc0ba9a4c8daa3f30fc768d767e504a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 12:25:16.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/23/669223/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e8757df/ : SUCCESS in 2h 55m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/23/669223/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/0ca1a2e/ : SUCCESS in 3h 20m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/23/669223/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/ca2eaca/ : SUCCESS in 2h 10m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/23/669223/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/343f212/ : SUCCESS in 2h 46m 38s","accounts_in_message":[],"_revision_number":8},{"id":"65ad145554a9391ec30e49ed5723a6218ddfc27b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-09-11 16:51:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6373b0cf5f79e056630a5343bfb285e81edf4a0c","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-09-25 15:13:15.000000000","message":"Patch Set 8:\n\nThis patch uses the new RHEL Universal Base Image (UBI) which is provided with no strings attached and is free to use. [1] Although, I think it would be easier for us to now use CentOS 8 since it is released. [2] Should we look into updating this patch (when CentOS 8 images land in Docker Hub) or should we work towards a new patch?\n1. https://www.redhat.com/en/blog/introducing-red-hat-universal-base-image\n2. https://blog.centos.org/2019/09/centos-8-and-centos-stream-released/","accounts_in_message":[],"_revision_number":8},{"id":"4dc677ec1e1e50d2d70663ed20aa79476b3bc887","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-01-29 10:39:05.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"87dc9f3b5960b9f0f1bdd14c80a50ef86db5fe48","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-01-29 10:40:03.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"db0c8f032179ff6f09144a4f4cee5a52dea6c8e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 10:58:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/dff4657e17dc41e49fa25cc69415d086 : SUCCESS in 16m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49c0420a0c534160b3203a677485a7bb : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":9},{"id":"7ec757d83e456296fdbb9f4d36b45fc7cf17282c","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-01-29 11:08:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"427c53359e74d53675ed173773d2359064e8b98c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-29 13:58:32.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/5f6234dd058244c99b1cd90a7176df99 : SUCCESS in 1h 52m 09s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/5c33016435c9450ab57e981aed35df3f : SUCCESS in 1h 16m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ce9b833f20f0449cb925c1dd43b96d5b : FAILURE in 1h 30m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c95c3cbd8d144e66b873a3856ea2d6bb : SUCCESS in 2h 44m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f15848d1e61a459c972c20702de3b44d : SUCCESS in 2h 00m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4ba4cc151073416e8492229654496adf : SUCCESS in 2h 11m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/18bea18bb9f9402886bf532953a45c1e : SUCCESS in 1h 59m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/b26634caa4ae424aac7001dfe6ad22d7 : SUCCESS in 2h 45m 31s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/8505fb96c39646c7a4ade9be95fac49a : SUCCESS in 2h 13m 54s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/8318d34e27cd4e75baa65bc42ff6c050 : SUCCESS in 2h 31m 15s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/4697b6b579d94d5bab5f6504408644bb : SUCCESS in 2h 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5e110ab4e954550983a3376f505fcfb : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecfc4dd995f54b23b69ed205272115eb : SUCCESS in 3m 46s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c7879e35ab9b424e9471a7466a7a7396 : SUCCESS in 13m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/be5afffec2d8481c8adfa5e823902296 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":10},{"id":"59784df67cc5364ae889d8b1a869e02f9b6f1679","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-29 15:00:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/61b8fa2504f4414c95ad6cbecb16a174 : SUCCESS in 2h 44m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/de2c5a2986cb4c699be19f04fbeec1dd : SUCCESS in 2h 03m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/f98a418e430c4e40a702da781f47c1c9 : SUCCESS in 2h 24m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/1963b9f7d4f14a3d9e10ed889b4a3602 : SUCCESS in 2h 46m 58s","accounts_in_message":[],"_revision_number":10},{"id":"c9c9ea8db711e5af434bd90c75ce6c3f9d369cc1","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2020-01-30 16:48:04.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"f022adde379dc17e88dbdc7dc60cb3e330907904","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-01-30 18:40:34.000000000","message":"Patch Set 10: Code-Review+2\n\nseems ok - molecule tests pass","accounts_in_message":[],"_revision_number":10},{"id":"74f460aca1f2970938b94c6d290be4230e5919bf","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-02-10 09:37:40.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4ed01dd5dbb7efcefa8e5842732bed943e126761","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-02-10 09:42:13.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bcbd510c0b5179fed0682d3e3938cc108712662c","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-02-10 09:47:05.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"94da9719ca5012d9b5ccdd1dbf9b4e1ed8cc3e99","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-02-10 09:51:26.000000000","message":"Patch Set 10:\n\nSorin Sbarnea wrote:\n\n\u003e (1 comment)\nAck, thanks","accounts_in_message":[],"_revision_number":10},{"id":"b0af20a553c2d875fe4a0c5a39c07c941fb8795f","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-02-10 09:51:30.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"9f95fb6c11ba362e76f9ef600cf201c7966cfe43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 12:33:03.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/d62ade5dd9744d6eabe3ccf6444b0905 : SUCCESS in 1h 36m 29s\n- tripleo-ci-centos-7-containers-undercloud-minion https://zuul.opendev.org/t/openstack/build/cfeec96570254407b62b988f49e2fc2b : SUCCESS in 1h 34m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8b79c6dbd3e64b2fa80146ffb6bf6818 : SUCCESS in 2h 19m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/e90660ec47374fb2bccac0e4d8f8dcde : SUCCESS in 2h 37m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3c558e63dd5147dca8301ac5683469a1 : SUCCESS in 2h 24m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1671a665910640fd86d39da27e206126 : SUCCESS in 2h 26m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a37e447b7bac4d6f8f2b686f0b325d4d : SUCCESS in 2h 37m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a72474897527468dacfb9d59fc8c26c2 : SUCCESS in 2h 35m 45s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/9378cb372e85416e9429d4652f76226f : SUCCESS in 2h 27m 42s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/2a721cb243c3462dbfbb2517e37616c9 : SUCCESS in 2h 39m 55s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/b343c5898a0e4ba6b416eb5fb7db3861 : SUCCESS in 2h 31m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be37adc36c34e12abec73a634e35499 : SUCCESS in 5m 19s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/90ebc2d1b4f849e19048f5138e865197 : SUCCESS in 14m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/66189cd70fc64cd18507e59b0838081a : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":11},{"id":"a18a136886a32e289a822e9b717222dbc3f79270","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-02-10 12:45:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b805ab834d494a3f94096193741d8a06 : SUCCESS in 2h 57m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/014ad63f33c04f109ea3f1b34a0fb98a : SUCCESS in 2h 08m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/f6f82b9410ad43ffbab4ce60bcc9d752 : SUCCESS in 2h 29m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/0fc6ad81284e4625b13e231629e3f8d2 : SUCCESS in 2h 52m 46s","accounts_in_message":[],"_revision_number":11},{"id":"26f22b8d042243351a8a8b113aa259ae47ff50d8","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-25 11:51:34.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b0a77b1ab111ef285b0bd73b1109b4d9637a2fa5","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-03-25 11:59:27.000000000","message":"Patch Set 12:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-quick-basic https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-quick-basic/178/ : FAILURE","accounts_in_message":[],"_revision_number":12},{"id":"788aef1f8fdd6155f9c269a45c34bdedc5b84331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 15:14:48.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e0aca97f01a64a3d98fd7a8a02527db4 : SUCCESS in 1h 40m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9841c49f71fd4984a071dbf83b49de3b : SUCCESS in 2h 46m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6371d7d0b05e46b490a7ee0ec1829ad9 : SUCCESS in 2h 41m 31s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6f3dd4dbf4eb40398605a4c4094e7e1a : SUCCESS in 1h 51m 12s\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/551e87eb36ff4e5c8c8e1771a6e1880d : SUCCESS in 2h 50m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/4f2fd65917c1482c98f47f7bae293d5b : SUCCESS in 3h 18m 29s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/151c6817dcc348f09fd501a200a00549 : SUCCESS in 3h 00m 09s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/dbc2733037f74981a0b5ce11aa281116 : SUCCESS in 3h 14m 29s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3348fc6238d14c83b5bf51d0b0bd0c28 : SUCCESS in 3h 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e9b05499734bc6ae59bb757f92e2a8 : SUCCESS in 4m 59s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/47c1d271d0ca49078ed8743f31d80d77 : SUCCESS in 14m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6964574bbad04a08b9707faf0420554d : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":12},{"id":"563e780904828db116878bb2e256fceca60756d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-03-25 17:23:14.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a292febf102d4818b4d77ba22d6c1029 : SUCCESS in 3h 36m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/3282198ecdf544ae9e280d0aab242f2e : SUCCESS in 2h 12m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/9f5dccca1d1140e4a3e19da3a3bc1387 : FAILURE in 2h 18m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/7cbcd7d5438f4e56a953355ec7b3cedd : SUCCESS in 3h 05m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/11673d42b3674c6f9f9cebe1b2a52536 : RETRY_LIMIT in 5m 28s","accounts_in_message":[],"_revision_number":12},{"id":"bb9d1ebca552b76bcc1a171c33b31b687964ceed","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-03-26 08:07:15.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"eb5cee8b4ce04ad5f821b1c3825fe47ca9341982","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-27 11:46:52.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7f2f18731a5ab6d6755307a86794492f09ca1203","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-03-30 07:22:19.000000000","message":"Patch Set 12: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":12},{"id":"642faea35e537a70a1e6935675bd461f426cbd86","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2020-03-30 08:25:14.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"c3f5021874424d239fd14c9685509b25eaab7fef","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2020-03-30 09:48:35.000000000","message":"Patch Set 12: Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"21f05289aa2fd0abafbd6975c885c7a636635955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 09:48:53.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"41ffc366b6a3927864ebbd53a15e331999dfa070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 15:18:07.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ba01676c94b546a79a7fc16d8fffb28e : SUCCESS in 1h 54m 38s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f1b339bd6dce4e9d93dd0c3bd6a6bb35 : SUCCESS in 2h 50m 54s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/bfd625b68f7e44bd8fe1590f9e81959f : SUCCESS in 1h 57m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c6c28f53193c46ab848da579904c2bbb : SUCCESS in 2h 39m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d36f190c60241b9bac8230f3e29a602 : SUCCESS in 4m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/cccc4546484845c7826e230673032923 : SUCCESS in 12m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/16558b2ab589443ca68db6dc169b17f9 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":12},{"id":"b7edbb12563b137f6d3d51ae4f47dd4a89035c5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 15:18:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"e0ae8602f28aa30b6154b515249a6f6e94d47b67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 15:24:30.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5ef4f5f1ad40a6b2dc7200381573b6 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"49c5640704c17a6a11aed4273a949240327183ae","revisions":{"d0ca5438773578630efb373f3a1a2bca1829aaf5":{"kind":"REWORK","_number":1,"created":"2019-07-04 21:30:09.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/1"}}},"commit":{"parents":[{"commit":"d2659094dcaf6c400058d47d62a250b4178ba983","subject":"Merge \"Turn off ansible lint after deployment.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d2659094dcaf6c400058d47d62a250b4178ba983"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:29:17.000000000","tz":60},"subject":"Adds RHEL8 testing to molecule","message":"Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d0ca5438773578630efb373f3a1a2bca1829aaf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d0ca5438773578630efb373f3a1a2bca1829aaf5"}]},"branch":"refs/heads/master"},"6c417f637926f7aebc247e886fefff6bf192c4c7":{"kind":"REWORK","_number":2,"created":"2019-07-04 22:00:15.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/2"}}},"commit":{"parents":[{"commit":"d2659094dcaf6c400058d47d62a250b4178ba983","subject":"Merge \"Turn off ansible lint after deployment.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d2659094dcaf6c400058d47d62a250b4178ba983"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:59:53.000000000","tz":60},"subject":"Adds RHEL8 testing to molecule","message":"Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nRemoves harcoded override which prevented the detectio of the platform\nfrom working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6c417f637926f7aebc247e886fefff6bf192c4c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6c417f637926f7aebc247e886fefff6bf192c4c7"}]},"branch":"refs/heads/master"},"e533d9e0c55e3e0e80f8d209586c569fbdf4f72c":{"kind":"REWORK","_number":3,"created":"2019-07-05 11:41:39.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/3"}}},"commit":{"parents":[{"commit":"d2659094dcaf6c400058d47d62a250b4178ba983","subject":"Merge \"Turn off ansible lint after deployment.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d2659094dcaf6c400058d47d62a250b4178ba983"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-05 11:40:26.000000000","tz":60},"subject":"WIP: Adds RHEL8 testing to molecule","message":"WIP: Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nRemoves harcoded override which prevented the detectio of the platform\nfrom working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e533d9e0c55e3e0e80f8d209586c569fbdf4f72c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e533d9e0c55e3e0e80f8d209586c569fbdf4f72c"}]},"branch":"refs/heads/master"},"2cb1bdad0b0c2c0334c391cddb0a107015a75682":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-05 11:41:48.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/4"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-05 11:41:48.000000000","tz":0},"subject":"WIP: Adds RHEL8 testing to molecule","message":"WIP: Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nRemoves harcoded override which prevented the detectio of the platform\nfrom working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2cb1bdad0b0c2c0334c391cddb0a107015a75682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2cb1bdad0b0c2c0334c391cddb0a107015a75682"}]},"branch":"refs/heads/master"},"8dcece5f50b551b33bf39191c8863f6b8c4e5cc6":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-05 14:42:59.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/5"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-05 14:42:34.000000000","tz":60},"subject":"WIP: Adds RHEL8 testing to molecule","message":"WIP: Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nRemoves harcoded override which prevented the detectio of the platform\nfrom working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\nDepends-On: https://review.opendev.org/669366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8dcece5f50b551b33bf39191c8863f6b8c4e5cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8dcece5f50b551b33bf39191c8863f6b8c4e5cc6"}]},"branch":"refs/heads/master"},"0dde23ac9dfd9188cc23c2060616800f0631a489":{"kind":"REWORK","_number":6,"created":"2019-09-05 16:11:38.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/6"}}},"commit":{"parents":[{"commit":"7299586340f3a8e6b2cf6f6ccfaf40308b5f2263","subject":"Merge \"install ceph-ansible on rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7299586340f3a8e6b2cf6f6ccfaf40308b5f2263"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-09-05 15:50:22.000000000","tz":60},"subject":"WIP: Adds RHEL8 testing to molecule","message":"WIP: Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nRemoves harcoded override which prevented the detectio of the platform\nfrom working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\nDepends-On: https://review.opendev.org/669366\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0dde23ac9dfd9188cc23c2060616800f0631a489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0dde23ac9dfd9188cc23c2060616800f0631a489"}]},"branch":"refs/heads/master"},"78b2a7c3ef85f87241da50a62f9372f57918e884":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2019-09-10 13:14:43.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/7"}}},"commit":{"parents":[{"commit":"7299586340f3a8e6b2cf6f6ccfaf40308b5f2263","subject":"Merge \"install ceph-ansible on rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7299586340f3a8e6b2cf6f6ccfaf40308b5f2263"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-09-10 13:13:55.000000000","tz":0},"subject":"WIP: Adds RHEL8 testing to molecule","message":"WIP: Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nRemoves harcoded override which prevented the detectio of the platform\nfrom working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\nDepends-On: https://review.opendev.org/669366\nDepends-On: https://review.opendev.org/681251 reports.html\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78b2a7c3ef85f87241da50a62f9372f57918e884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78b2a7c3ef85f87241da50a62f9372f57918e884"}]},"branch":"refs/heads/master"},"c7f9006347c0b4ba9369f07fa2b0dbfaea0019c8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-10 13:15:02.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/8"}}},"commit":{"parents":[{"commit":"c0b5fdcdde5b736f0baf005bf1f5ffc83722597d","subject":"Merge \"Add ceph-ansible to promotion jobs on RHEL8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c0b5fdcdde5b736f0baf005bf1f5ffc83722597d"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-09-10 13:15:02.000000000","tz":0},"subject":"WIP: Adds RHEL8 testing to molecule","message":"WIP: Adds RHEL8 testing to molecule\n\nEffectively starts running a RHEL8 container (ubi8:python-36) in\nparallel with the other platforms.\n\nRemoves harcoded override which prevented the detectio of the platform\nfrom working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\nDepends-On: https://review.opendev.org/669366\nDepends-On: https://review.opendev.org/681251 reports.html\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c7f9006347c0b4ba9369f07fa2b0dbfaea0019c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c7f9006347c0b4ba9369f07fa2b0dbfaea0019c8"}]},"branch":"refs/heads/master"},"5ed57907c1455401c9bda3fea6ad3bb76418fe4f":{"kind":"REWORK","_number":9,"created":"2020-01-29 10:40:03.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/9"}}},"commit":{"parents":[{"commit":"9d6ab5408f4724d7eb3f401e5ee6b27779ad54fa","subject":"Merge \"Run multinode and undercloud jobs nova-less\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9d6ab5408f4724d7eb3f401e5ee6b27779ad54fa"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-01-29 10:29:44.000000000","tz":0},"subject":"Adds centos-8 testing to install-deps","message":"Adds centos-8 testing to install-deps\n\nRemoves harcoded override which prevented the detection of the\nplatform from working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5ed57907c1455401c9bda3fea6ad3bb76418fe4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5ed57907c1455401c9bda3fea6ad3bb76418fe4f"}]},"branch":"refs/heads/master"},"c7e63f400459ab8f2e6aba92f520a80db6d581e8":{"kind":"REWORK","_number":10,"created":"2020-01-29 11:08:11.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/10"}}},"commit":{"parents":[{"commit":"9d6ab5408f4724d7eb3f401e5ee6b27779ad54fa","subject":"Merge \"Run multinode and undercloud jobs nova-less\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9d6ab5408f4724d7eb3f401e5ee6b27779ad54fa"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-01-29 11:07:51.000000000","tz":0},"subject":"Adds centos-8 testing to install-deps","message":"Adds centos-8 testing to install-deps\n\nRemoves harcoded override which prevented the detection of the\nplatform from working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c7e63f400459ab8f2e6aba92f520a80db6d581e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c7e63f400459ab8f2e6aba92f520a80db6d581e8"}]},"branch":"refs/heads/master"},"d221effce03641ae4745094e47469a8e57c712c7":{"kind":"REWORK","_number":11,"created":"2020-02-10 09:47:05.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/11"}}},"commit":{"parents":[{"commit":"f81189367e1a2ec1ce15e764d10bef54ac07f9d1","subject":"Revert \"Remove hardcoding of python interpreter for rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f81189367e1a2ec1ce15e764d10bef54ac07f9d1"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-02-10 09:45:46.000000000","tz":0},"subject":"Adds centos-8 testing to install-deps","message":"Adds centos-8 testing to install-deps\n\nRemoves harcoded override which prevented the detection of the\nplatform from working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d221effce03641ae4745094e47469a8e57c712c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d221effce03641ae4745094e47469a8e57c712c7"}]},"branch":"refs/heads/master"},"49c5640704c17a6a11aed4273a949240327183ae":{"kind":"REWORK","_number":12,"created":"2020-03-25 11:51:34.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/23/669223/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/669223/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/669223/12"}}},"commit":{"parents":[{"commit":"16beb60b8f6f6dc7233dd3a8e2e839ec48ee8a64","subject":"Run py2 in case of CentOS7 on OVB","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/16beb60b8f6f6dc7233dd3a8e2e839ec48ee8a64"}]}],"author":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-07-04 21:13:26.000000000","tz":60},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2020-03-25 11:47:33.000000000","tz":0},"subject":"Adds centos-8 testing to install-deps","message":"Adds centos-8 testing to install-deps\n\nRemoves harcoded override which prevented the detection of the\nplatform from working.\n\nChange-Id: Iad1734ad789ad91915c4e0374cb2afcc1f289b23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/49c5640704c17a6a11aed4273a949240327183ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/49c5640704c17a6a11aed4273a949240327183ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
