)]}'
{"id":"openstack%2Ftripleo-quickstart~671017","triplet_id":"openstack%2Ftripleo-quickstart~master~Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40","project":"openstack/tripleo-quickstart","branch":"master","topic":"rhel8_periodic","hashtags":[],"change_id":"Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40","subject":"[RHEL8] promotion-testing-hash-master release file","status":"MERGED","created":"2019-07-16 11:56:25.000000000","updated":"2019-07-26 13:06:01.000000000","submitted":"2019-07-26 13:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671017-1564146361539-31ef7618","meta_rev_id":"ca6e0afa2961e5c5b4c993002bb865b39dfdd6b7","_number":671017,"virtual_id_number":671017,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-07-26 13:06:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-25 19:15:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-07-26 10:31:13.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-07-26 10:06:46.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2019-07-26 10:33:18.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-16 11:57:11.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-07-16 11:57:14.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-07-16 11:57:32.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-16 11:57:42.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-07-16 17:46:46.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-07-18 01:57:02.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-19 15:43:37.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-07-23 16:39:46.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-07-25 19:15:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-26 10:06:46.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-07-26 10:33:18.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-07-26 13:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6510074622aeff554e3aef70af16115f2c0b55f8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-16 11:56:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9323097a03dcee982572d1a644be7eda787d638","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-07-16 12:11:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nI think we can merge it even with the errors. Everything here will be adjusted during the test iterations","accounts_in_message":[],"_revision_number":1},{"id":"72317e4917a2d45a01d840876ed7baf36961ef5d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-16 12:55:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b911fa5a8c09429d003cccb250f6dfea5ec6a9e","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-07-16 15:40:15.000000000","message":"Patch Set 2: Code-Review+2\n\nLet\u0027s merge it non working","accounts_in_message":[],"_revision_number":2},{"id":"ed6e233fa740c28250d25b32ac19a02c6da0b2c1","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-07-16 17:46:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3031c7d9e7dbd89f11e8e998f2ee5973c921e017","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:49:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-undercloud-containers/db34469/ : SUCCESS in 1h 29m 11s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2a8fc7a/ : FAILURE in 2h 08m 43s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/47674c0/ : SUCCESS in 2h 43m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f724fff/ : FAILURE in 2h 47m 32s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/2/check/tripleo-ci-fedora-28-standalone/695065f/ : SUCCESS in 1h 46m 03s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-standalone-upgrade/3c1ab67/ : FAILURE in 2h 06m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-stein/75a7731/ : SUCCESS in 2h 15m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-rocky/bfefe45/ : SUCCESS in 2h 45m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-queens/ac39e59/ : SUCCESS in 2h 08m 49s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-pike/2c1250e/ : TIMED_OUT in 3h 12m 22s\n- openstack-tox-docs http://logs.openstack.org/17/671017/2/check/openstack-tox-docs/f6dfeeb/html/ : SUCCESS in 5m 32s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/2/check/openstack-tox-molecule/5e694dc/tox/reports.html : FAILURE in 14m 06s\n- openstack-tox-linters http://logs.openstack.org/17/671017/2/check/openstack-tox-linters/da969bc/ : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":2},{"id":"72c8df3ad7b21c76ca28dc12c259ff0caec3ddde","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 10:54:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3bf7f53/ : FAILURE in 1h 10m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b5b1ead/ : SUCCESS in 2h 44m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7441c4c/ : FAILURE in 2h 44m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/63d78ab/ : SUCCESS in 2h 26m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/defd3e0/ : SUCCESS in 2h 50m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/92cb937/ : FAILURE in 1h 14m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47e91be0f9a27bd3ba47b166a489da8cf3f38527","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-18 01:57:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c311be093ed47f1f75bba7414146f2b479aa4860","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 04:51:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-undercloud-containers/7b06f6b/ : SUCCESS in 1h 50m 16s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7774eea/ : FAILURE in 1h 58m 15s (non-voting)\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a46af19/ : SUCCESS in 2h 38m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3422f97/ : FAILURE in 1h 10m 07s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/2/check/tripleo-ci-fedora-28-standalone/a939201/ : FAILURE in 1h 42m 05s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-standalone-upgrade/587f6da/ : FAILURE in 2h 01m 07s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-stein/aa19590/ : SUCCESS in 2h 49m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-rocky/35368c8/ : SUCCESS in 2h 47m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-queens/0cdc939/ : SUCCESS in 2h 12m 33s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/2/check/tripleo-ci-centos-7-containers-multinode-pike/fee28a7/ : SUCCESS in 2h 29m 36s\n- openstack-tox-docs http://logs.openstack.org/17/671017/2/check/openstack-tox-docs/51aa1a4/html/ : SUCCESS in 6m 49s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/2/check/openstack-tox-molecule/cfb71b9/tox/reports.html : FAILURE in 12m 25s\n- openstack-tox-linters http://logs.openstack.org/17/671017/2/check/openstack-tox-linters/4b12068/ : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f53af80e2cfaae970204cc7b8ff49b7d34d7b03c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 07:36:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/25dd45e/ : RETRY_LIMIT in 3m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze01.softwarefactory-project.io:7979/36dac1ca2c0b4377b00e326e1747a5fc : POST_FAILURE in 3h 44m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch finger://ze03.softwarefactory-project.io:7979/aff7a1d1345440a48ca3548ecdb1977d : RETRY_LIMIT in 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cc48958/ : RETRY_LIMIT in 5m 40s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5e0cc1e/ : RETRY_LIMIT in 5m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/e6e61c9/ : RETRY_LIMIT in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ced012f065179ce0cee75bae7e5959c482654b43","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-19 15:42:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"36b5130ebe649776084ee2e6159b114a6bffb3dd","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-07-19 15:43:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ef20d5c9abab1c656c35b4325274dba543a37bf2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 18:20:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d94b7a8/ : FAILURE in 2h 11m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/671017/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/039e4fc/ : SUCCESS in 2h 36m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/671017/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/16c5177/ : SUCCESS in 2h 09m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/038d0ef/ : SUCCESS in 2h 23m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/fd46cb4/ : SUCCESS in 2h 34m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/78ef32e/ : FAILURE in 1h 48m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc75f7ed1a6ea8989c4924a7c6767515a1c7bc22","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 20:37:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-undercloud-containers/4996d70/ : SUCCESS in 1h 47m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ce36161/ : FAILURE in 2h 10m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-containers-multinode/fa00157/ : SUCCESS in 2h 31m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8635885/ : SUCCESS in 2h 10m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3ed4e9c/ : FAILURE in 2h 27m 22s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/3/check/tripleo-ci-fedora-28-standalone/6f8889f/ : FAILURE in 21m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-standalone-upgrade/17bdc81/ : FAILURE in 1h 53m 32s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-containers-multinode-stein/983522d/ : SUCCESS in 2h 12m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-containers-multinode-rocky/fc68636/ : POST_FAILURE in 3h 12m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-containers-multinode-queens/541a366/ : SUCCESS in 2h 01m 21s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/3/check/tripleo-ci-centos-7-containers-multinode-pike/cad06be/ : SUCCESS in 2h 12m 11s\n- openstack-tox-docs http://logs.openstack.org/17/671017/3/check/openstack-tox-docs/0b03899/html/ : SUCCESS in 5m 25s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/3/check/openstack-tox-molecule/0c04e11/tox/reports.html : FAILURE in 19m 33s\n- openstack-tox-linters http://logs.openstack.org/17/671017/3/check/openstack-tox-linters/ddd65d4/ : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":3},{"id":"2f65ac6a434bfdb6ec2082115aea0b3f78cf1472","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-21 15:10:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"ce2f32ebb4ee7de6fab651fefe76ec0b6790b650","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-21 18:00:13.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-undercloud-containers/46e3736/ : SUCCESS in 2h 01m 50s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ea0dce6/ : FAILURE in 1h 47m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-containers-multinode/7495aec/ : SUCCESS in 2h 12m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b8a1a30/ : SUCCESS in 2h 18m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ff846be/ : FAILURE in 2h 27m 58s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/4/check/tripleo-ci-fedora-28-standalone/f6b8484/ : FAILURE in 21m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-standalone-upgrade/6dcb622/ : FAILURE in 1h 50m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-containers-multinode-stein/7dd7aca/ : SUCCESS in 2h 29m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-containers-multinode-rocky/80b2f08/ : SUCCESS in 2h 13m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-containers-multinode-queens/838b07a/ : SUCCESS in 2h 00m 31s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/4/check/tripleo-ci-centos-7-containers-multinode-pike/810f663/ : SUCCESS in 2h 20m 45s\n- openstack-tox-docs http://logs.openstack.org/17/671017/4/check/openstack-tox-docs/6a51323/html/ : SUCCESS in 5m 17s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/4/check/openstack-tox-molecule/b936a06/tox/reports.html : FAILURE in 11m 26s\n- openstack-tox-linters http://logs.openstack.org/17/671017/4/check/openstack-tox-linters/6f6b586/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"2c25cdb630471f14d33c8b285c2e858c300daf16","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-21 18:03:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d4ec7a1/ : SUCCESS in 2h 23m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/671017/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/df0cc62/ : SUCCESS in 2h 44m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/671017/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a341359/ : SUCCESS in 2h 02m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2b4529b/ : SUCCESS in 2h 38m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/20d70cb/ : SUCCESS in 2h 25m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/00f1b98/ : SUCCESS in 2h 51m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0744172f20df03736f45b4a65861611227879a77","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-22 08:28:32.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"dd5849853c63565426f668df7d8175cb8bc71416","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 11:06:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-undercloud-containers/2720d29/ : FAILURE in 48m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/059a062/ : FAILURE in 1h 49m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode/133d0b4/ : FAILURE in 52m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/7439dfa/ : FAILURE in 1h 06m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/653110e/ : FAILURE in 1h 08m 47s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/5/check/tripleo-ci-fedora-28-standalone/e4fc62c/ : FAILURE in 51m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-standalone-upgrade/d17a324/ : FAILURE in 1h 44m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-stein/08384a6/ : SUCCESS in 2h 18m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-rocky/cfac192/ : SUCCESS in 2h 29m 58s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-queens/e15ea16/ : SUCCESS in 2h 18m 13s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-pike/0026471/ : SUCCESS in 2h 18m 24s\n- openstack-tox-docs http://logs.openstack.org/17/671017/5/check/openstack-tox-docs/5202d5e/html/ : SUCCESS in 4m 25s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/5/check/openstack-tox-molecule/f5d303f/tox/reports.html : SUCCESS in 16m 18s\n- openstack-tox-linters http://logs.openstack.org/17/671017/5/check/openstack-tox-linters/0353203/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":5},{"id":"64df27f57096eb8141eb13c0fb4f7634e6ed4cea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 13:57:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/3fc1f54/ : RETRY_LIMIT in 12m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ba01d30/ : RETRY_LIMIT in 12m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ab5881e/ : RETRY_LIMIT in 14m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/65b04fc/ : SUCCESS in 3h 37m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/3a75c6a/ : RETRY_LIMIT in 14m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/2f05840/ : FAILURE in 3h 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4537011b47353d422e4c3fc4bcdff22f5df58017","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-23 06:01:29.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"cbd5ab6c9491b6929d40df7921a92846feb78220","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-23 06:12:30.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c82053d0183d477d6405a377b4f2acf2e017d57","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 08:45:25.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/77be09e/ : SUCCESS in 2h 42m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/981fe82/ : SUCCESS in 2h 42m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/356376c/ : SUCCESS in 2h 10m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/6ca80a5/ : SUCCESS in 2h 11m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/162c238/ : SUCCESS in 2h 30m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/ee17eb2/ : SUCCESS in 2h 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c5ace417885ddf7a43256ced4fb713ff2e8738c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 10:35:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-undercloud-containers/db535d1/ : SUCCESS in 1h 36m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/eb0b514/ : FAILURE in 1h 44m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode/d6beece/ : FAILURE in 2h 24m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/df72456/ : SUCCESS in 2h 10m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0102cc1/ : SUCCESS in 2h 15m 05s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/5/check/tripleo-ci-fedora-28-standalone/5f8861b/ : SUCCESS in 1h 59m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-standalone-upgrade/8b06a71/ : FAILURE in 1h 39m 44s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-stein/8595928/ : SUCCESS in 1h 56m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-rocky/4a3ff37/ : SUCCESS in 2h 35m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-queens/ebbf27a/ : SUCCESS in 2h 15m 05s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/5/check/tripleo-ci-centos-7-containers-multinode-pike/b939c16/ : TIMED_OUT in 3h 13m 26s\n- openstack-tox-docs http://logs.openstack.org/17/671017/5/check/openstack-tox-docs/fc9b829/html/ : SUCCESS in 5m 19s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/5/check/openstack-tox-molecule/0f11786/tox/reports.html : SUCCESS in 15m 51s\n- openstack-tox-linters http://logs.openstack.org/17/671017/5/check/openstack-tox-linters/f8bd0b7/ : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":5},{"id":"5b9bd6c38376df854b6d933df79e56a007538b51","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-23 16:39:46.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a98e0af66eb1dc9b52ac9ad05eaf888002177a81","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-24 07:26:55.000000000","message":"Patch Set 5: Code-Review+1\n\nok revoting for the discussion then","accounts_in_message":[],"_revision_number":5},{"id":"13dbb67c343c39842b2ee23c894a6189d618c5b1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 07:14:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5cc1ca78936c98aff650a6cdb1858e619b191a93","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 07:14:51.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a0dcc00ccf646a7d375c7892f92446cf4bb50c9c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 07:15:06.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"a80ba91e06512b576c6bd83f3ae94ac795f3f78d","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-25 07:53:33.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"cee9cddc2189289ce46a1b75074f9706970995df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 10:03:43.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-undercloud-containers/1db277e/ : SUCCESS in 1h 23m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/09c0c36/ : FAILURE in 2h 15m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode/1670140/ : SUCCESS in 2h 08m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bb99dba/ : SUCCESS in 2h 16m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/397dea4/ : SUCCESS in 2h 12m 56s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/7/check/tripleo-ci-fedora-28-standalone/21c1d31/ : SUCCESS in 1h 38m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-standalone-upgrade/11fde2d/ : FAILURE in 1h 39m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-stein/5a6bf89/ : SUCCESS in 2h 12m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-rocky/8b3a4d3/ : SUCCESS in 2h 39m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-queens/50d9f5c/ : SUCCESS in 1h 55m 33s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-pike/88d0c1b/ : SUCCESS in 2h 15m 13s\n- openstack-tox-docs http://logs.openstack.org/17/671017/7/check/openstack-tox-docs/49472c9/html/ : SUCCESS in 5m 16s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/7/check/openstack-tox-molecule/eeb18d0/tox/reports.html : FAILURE in 8m 42s\n- openstack-tox-linters http://logs.openstack.org/17/671017/7/check/openstack-tox-linters/3abbe12/ : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":7},{"id":"bd37748a604e9a471db7ccf87ebaf3d0cc4ee259","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 10:08:41.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"40ec9d6f90243853a724dee573449fce7919fa5e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 12:21:12.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b5c4551/ : SUCCESS in 2h 30m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/671017/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/3e2c449/ : SUCCESS in 2h 34m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/671017/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/957572b/ : SUCCESS in 2h 11m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5f51827/ : FAILURE in 3h 14m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/95b1a6a/ : SUCCESS in 2h 31m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/f60597f/ : TIMED_OUT in 5h 02m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fe7e6bb0a7f220e59cb3aa09da854229338274e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 13:14:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-undercloud-containers/e6d0a13/ : SUCCESS in 1h 28m 20s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7278c4c/ : FAILURE in 2h 02m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode/11ac2ca/ : SUCCESS in 2h 18m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8572f78/ : SUCCESS in 2h 15m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b59dd21/ : SUCCESS in 2h 18m 18s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/7/check/tripleo-ci-fedora-28-standalone/1dd9248/ : SUCCESS in 1h 46m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-standalone-upgrade/b3a128b/ : FAILURE in 2h 00m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-stein/f2fe54c/ : SUCCESS in 2h 21m 52s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-rocky/9697e44/ : SUCCESS in 2h 27m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-queens/ef3cb47/ : SUCCESS in 2h 22m 41s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/7/check/tripleo-ci-centos-7-containers-multinode-pike/192ed33/ : SUCCESS in 2h 34m 10s\n- openstack-tox-docs http://logs.openstack.org/17/671017/7/check/openstack-tox-docs/2d640dd/html/ : SUCCESS in 4m 59s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/7/check/openstack-tox-molecule/bda128f/tox/reports.html : SUCCESS in 14m 06s\n- openstack-tox-linters http://logs.openstack.org/17/671017/7/check/openstack-tox-linters/7868c5f/ : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":7},{"id":"23e1b14997beed7622ab116a6f6cd55be0ea888f","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-07-25 15:10:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"96baaade51629a897b6e2e7420c628ab835fe412","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 15:55:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ee26290ff58cb45e9fb0585644e2f87878351ca6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-25 15:55:59.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9a90eb2be22b98bc739302b7aae66adadb5fcd8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 19:15:01.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/671017/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8dade34/ : FAILURE in 3h 13m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/671017/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d873eba/ : FAILURE in 3h 16m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/671017/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/47616fb/ : SUCCESS in 2h 09m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/671017/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/487d1bc/ : SUCCESS in 2h 23m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/671017/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/52928aa/ : SUCCESS in 2h 27m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/671017/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/dec0b27/ : SUCCESS in 2h 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4dee76c5278f4e9a63ce0d7a5efd3f15e316a89b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 19:42:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-undercloud-containers/ea899e3/ : SUCCESS in 1h 31m 23s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/60ae72c/ : FAILURE in 2h 05m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-containers-multinode/5616c18/ : SUCCESS in 2h 07m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/38dd03d/ : SUCCESS in 2h 08m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/abcbf4b/ : SUCCESS in 2h 36m 10s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/671017/8/check/tripleo-ci-fedora-28-standalone/e0c711c/ : SUCCESS in 1h 25m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-standalone-upgrade/e74e641/ : FAILURE in 1h 50m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-containers-multinode-stein/8d6e50c/ : SUCCESS in 2h 48m 17s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-containers-multinode-rocky/e071e52/ : SUCCESS in 2h 26m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-containers-multinode-queens/bb93ee8/ : SUCCESS in 2h 09m 55s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/671017/8/check/tripleo-ci-centos-7-containers-multinode-pike/9d91756/ : SUCCESS in 2h 15m 57s\n- openstack-tox-docs http://logs.openstack.org/17/671017/8/check/openstack-tox-docs/aef3cd7/html/ : SUCCESS in 4m 14s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/8/check/openstack-tox-molecule/17a30ae/tox/reports.html : SUCCESS in 15m 14s\n- openstack-tox-linters http://logs.openstack.org/17/671017/8/check/openstack-tox-linters/f1282df/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":8},{"id":"913e0eb06153a03245a9eb9898b3f21d8d6dfe96","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-07-26 10:06:46.000000000","message":"Patch Set 8: Code-Review+2\n\nOk to merge a iterate after. Remember there\u0027s no expectation for the jobs to work right now","accounts_in_message":[],"_revision_number":8},{"id":"a85d0718ba9ad7982b2fbc6043ffa6ca16f4e853","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-26 10:31:13.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"bdda17f825dd0e0cfbd1fee3505052cfcda9b2e3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-26 10:33:18.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"106316028ad7242d1601f4b4dd1f4625e7b5cebc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 10:33:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f6de10cada964d7e65d5c96710c0e6589cb74240","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 13:06:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"d8a1565ac18d38a31ed0bddb8e48d248434c73e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 13:06:01.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/671017/8/gate/tripleo-ci-centos-7-undercloud-containers/3cb43cd/ : SUCCESS in 1h 28m 37s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/671017/8/gate/tripleo-ci-centos-7-containers-multinode/cfcbf7e/ : SUCCESS in 2h 27m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/671017/8/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a6dc3b5/ : SUCCESS in 2h 20m 01s\n- openstack-tox-docs http://logs.openstack.org/17/671017/8/gate/openstack-tox-docs/50dfad6/html/ : SUCCESS in 4m 03s\n- openstack-tox-molecule http://logs.openstack.org/17/671017/8/gate/openstack-tox-molecule/3b471b3/tox/reports.html : SUCCESS in 17m 30s\n- openstack-tox-linters http://logs.openstack.org/17/671017/8/gate/openstack-tox-linters/5af4664/ : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"086e83d8d8784f3992bb577c478a46ce18347454","revisions":{"ed306be289323457cf8372a35288741e20249395":{"kind":"REWORK","_number":1,"created":"2019-07-16 11:56:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/1"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ed306be289323457cf8372a35288741e20249395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ed306be289323457cf8372a35288741e20249395"}]},"branch":"refs/heads/master"},"13ecee73db877ea6d58a2e8217612e8e28cca61a":{"kind":"REWORK","_number":2,"created":"2019-07-16 12:55:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/2"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 12:55:03.000000000","tz":330},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/13ecee73db877ea6d58a2e8217612e8e28cca61a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/13ecee73db877ea6d58a2e8217612e8e28cca61a"}]},"branch":"refs/heads/master"},"0fdc981207517467127efed2e68c30a4d397d8d3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-19 15:42:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/3"}}},"commit":{"parents":[{"commit":"b8b85b348da761906a28e74995a8c4ccb92785ae","subject":"Added RedHat-8 release file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b8b85b348da761906a28e74995a8c4ccb92785ae"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-19 15:42:27.000000000","tz":0},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0fdc981207517467127efed2e68c30a4d397d8d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0fdc981207517467127efed2e68c30a4d397d8d3"}]},"branch":"refs/heads/master"},"70441303a5df0f3cc8986d9d746d02eb04c62948":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-21 15:10:00.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/4"}}},"commit":{"parents":[{"commit":"326c8bde59ba13e45ca66a00ad576ab0f4262ed3","subject":"Merge \"Added RedHat-8 release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/326c8bde59ba13e45ca66a00ad576ab0f4262ed3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-21 15:10:00.000000000","tz":0},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/70441303a5df0f3cc8986d9d746d02eb04c62948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/70441303a5df0f3cc8986d9d746d02eb04c62948"}]},"branch":"refs/heads/master"},"6b9a70eb826557f74a9f4b2709a5412198dd1d18":{"kind":"REWORK","_number":5,"created":"2019-07-22 08:28:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/5"}}},"commit":{"parents":[{"commit":"326c8bde59ba13e45ca66a00ad576ab0f4262ed3","subject":"Merge \"Added RedHat-8 release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/326c8bde59ba13e45ca66a00ad576ab0f4262ed3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-22 07:46:25.000000000","tz":0},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6b9a70eb826557f74a9f4b2709a5412198dd1d18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6b9a70eb826557f74a9f4b2709a5412198dd1d18"}]},"branch":"refs/heads/master"},"a15c2d0931888ab5122986059dda2d0b221b2813":{"kind":"REWORK","_number":6,"created":"2019-07-25 07:14:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/6"}}},"commit":{"parents":[{"commit":"326c8bde59ba13e45ca66a00ad576ab0f4262ed3","subject":"Merge \"Added RedHat-8 release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/326c8bde59ba13e45ca66a00ad576ab0f4262ed3"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-25 07:14:28.000000000","tz":330},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a15c2d0931888ab5122986059dda2d0b221b2813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a15c2d0931888ab5122986059dda2d0b221b2813"}]},"branch":"refs/heads/master"},"130a494a116b1352166c6131fc79cf6b7b83cf50":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-25 07:15:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/7"}}},"commit":{"parents":[{"commit":"c9a04b18694a1d10d9480524049eee10c23ec4fa","subject":"Include ansible-config_template in delorean-current repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c9a04b18694a1d10d9480524049eee10c23ec4fa"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-25 07:15:06.000000000","tz":0},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/130a494a116b1352166c6131fc79cf6b7b83cf50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/130a494a116b1352166c6131fc79cf6b7b83cf50"}]},"branch":"refs/heads/master"},"086e83d8d8784f3992bb577c478a46ce18347454":{"kind":"REWORK","_number":8,"created":"2019-07-25 15:55:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/671017/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/671017/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/671017/8"}}},"commit":{"parents":[{"commit":"c9a04b18694a1d10d9480524049eee10c23ec4fa","subject":"Include ansible-config_template in delorean-current repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c9a04b18694a1d10d9480524049eee10c23ec4fa"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-16 11:53:30.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-25 15:55:54.000000000","tz":330},"subject":"[RHEL8] promotion-testing-hash-master release file","message":"[RHEL8] promotion-testing-hash-master release file\n\nIn order to test periodic fs01 jobs, we need\npromotion-testing-hash-master release file for RHEL8 so\nthat tripleo-current gets promoted.\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1146\n\nChange-Id: Iafc0a3ce3d882b2e60fe2c484dc188a8ff73ba40\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/086e83d8d8784f3992bb577c478a46ce18347454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/086e83d8d8784f3992bb577c478a46ce18347454"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
