)]}'
{"id":"openstack%2Ftripleo-quickstart~671447","triplet_id":"openstack%2Ftripleo-quickstart~master~I96925574858223348faf35521fb2afdf7710e0db","project":"openstack/tripleo-quickstart","branch":"master","topic":"os_tempest","hashtags":[],"change_id":"I96925574858223348faf35521fb2afdf7710e0db","subject":"Enable os_tempest for master multinode jobs","status":"MERGED","created":"2019-07-18 10:05:27.000000000","updated":"2019-07-19 22:55:08.000000000","submitted":"2019-07-19 22:55:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671447-1563576908717-5870c8eb","meta_rev_id":"8237ae2310dcd92e6a4fd4e3819b336d6fb397e0","_number":671447,"virtual_id_number":671447,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-07-19 13:48:56.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-07-19 22:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-07-18 15:37:57.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-07-19 09:22:10.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2019-07-18 11:58:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2019-07-18 16:47:40.000000000","_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2019-07-18 14:45:14.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2019-07-19 09:22:10.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 10:06:18.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-07-18 10:06:39.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-07-18 10:10:10.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-18 11:28:29.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-18 11:58:45.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-07-18 14:45:14.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-18 15:37:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-18 16:47:40.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-07-19 09:22:10.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-07-19 22:55:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"407d4aa9c7b027deda33696ff90f9d0b6da21254","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-18 10:05:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f84a09cd9f1f73537013c68911940c8fb763e07e","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-18 10:10:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e19bf8d82a640c50278f59a771687712515cf765","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-18 10:27:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"7732c40a8869ff1d76e9234b74059c8732655da4","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-18 10:42:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c7d356b6bd28371e41a3d141479a2aacf249cfc6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-18 10:44:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"270cb907b2f23dcf6d2fd9e1cb413e087a67eb63","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-18 10:46:55.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nlets see it green before merge though obviously","accounts_in_message":[],"_revision_number":2},{"id":"cc17581d51ddb4d9d3f1f9099d4ab8ea7480d0d1","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-18 11:28:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6a0c5f7f881db3ce6d7cbb4f25c4d87b04737bd3","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-07-18 11:49:36.000000000","message":"Patch Set 2:\n\n(2 comments)\n\na query and a suggestion inline","accounts_in_message":[],"_revision_number":2},{"id":"e19d50ded69d1fd2d9859aaff076269e17ecd0e7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-18 11:54:39.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9a5e01a055402d481dba89f3301a72899b7dafb7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-18 11:54:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"efbe3ba3be193d881ff7cc734204cab7317cd0e9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-07-18 11:58:40.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c8b06c040c3c7911e9503391eca86a8d5ba43ef5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-07-18 11:58:45.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bbe36add808c69fbf0a3f51de634eade66031c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:38:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-undercloud-containers/1d00b60/ : SUCCESS in 1h 21m 19s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/05fdd99/ : FAILURE in 2h 12m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode/815e689/ : SUCCESS in 2h 16m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cde0761/ : SUCCESS in 2h 20m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b6f4d7c/ : FAILURE in 2h 22m 39s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/47/671447/4/check/tripleo-ci-fedora-28-standalone/51b24ea/ : SUCCESS in 1h 34m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-standalone-upgrade/d33922f/ : FAILURE in 1h 42m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-stein/f3bb34a/ : SUCCESS in 2h 02m 27s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-rocky/429dffc/ : SUCCESS in 2h 29m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-queens/7ad9c31/ : SUCCESS in 2h 10m 44s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-pike/1708382/ : SUCCESS in 2h 23m 22s\n- openstack-tox-docs http://logs.openstack.org/47/671447/4/check/openstack-tox-docs/4b99f02/html/ : SUCCESS in 4m 00s\n- openstack-tox-molecule http://logs.openstack.org/47/671447/4/check/openstack-tox-molecule/e491925/tox/reports.html : SUCCESS in 13m 21s\n- openstack-tox-linters http://logs.openstack.org/47/671447/4/check/openstack-tox-linters/21674f9/ : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"d83336897cd7c11fdcd3dce12ecee67c014e4ab3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-18 14:45:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7192029f1581401779252fb28915c2c54d33f8ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 15:37:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/fb6cad2/ : SUCCESS in 2h 53m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/41a88b8/ : SUCCESS in 2h 35m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d14b9ca/ : TIMED_OUT in 3h 37m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cfac9a6/ : SUCCESS in 2h 49m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/93971eb/ : SUCCESS in 2h 31m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/35e217f/ : SUCCESS in 2h 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5bc0de620b819ce404d3db095ce03401705bce69","author":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"date":"2019-07-18 16:47:40.000000000","message":"Patch Set 4: Code-Review+2\n\nfeatureset010 is tested in tripleo-ci-centos-7-containers-multinode, and it works with os-tempest. would be nice to check if previous branches are still working with validate tempest.\nTHis is an example of how we would really just need a molecule test here to check that the we are getting the correct value from the featureset, and not really needing to run a whole set of jobs end to end.\nWe could save resources here by just checking that the resulting configuration is what we expect.\nChandan asked me to wait for that test before merging.","accounts_in_message":[],"_revision_number":4},{"id":"80a20dcb21bec1e7f4a3cec901f16fa1a3011ec6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-19 06:42:04.000000000","message":"Patch Set 4:\n\n\u003e featureset010 is tested in tripleo-ci-centos-7-containers-multinode,\n \u003e and it works with os-tempest. would be nice to check if previous\n \u003e branches are still working with validate tempest.\n \u003e THis is an example of how we would really just need a molecule test\n \u003e here to check that the we are getting the correct value from the\n \u003e featureset, and not really needing to run a whole set of jobs end\n \u003e to end.\n \u003e We could save resources here by just checking that the resulting\n \u003e configuration is what we expect.\n \u003e Chandan asked me to wait for that test before merging.\n\nTesting here https://review.opendev.org/671660","accounts_in_message":[],"_revision_number":4},{"id":"19f45d6572e6639412c8730ef0e649c4e4c2c569","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-19 09:17:15.000000000","message":"Patch Set 4:\n\nThe job passed here: http://logs.openstack.org/60/671660/1/check/tripleo-ci-centos-7-containers-multinode/150b249/logs/tempest.html.gz and is triggered using validate-tempest role","accounts_in_message":[],"_revision_number":4},{"id":"02d7a86cef64854f9eaf44199479bd3f1e3559bb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-19 09:22:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c84181f0361f069da5d8464eee9f92189bc80ac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 09:22:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1f6b4e2ba7f6b9e651ceda502bb7f2759941950e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:11:22.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/47/671447/4/gate/tripleo-ci-centos-7-undercloud-containers/b2f1b28/ : FAILURE in 1h 41m 07s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/47/671447/4/gate/tripleo-ci-centos-7-containers-multinode/5d7de73/ : SUCCESS in 2h 27m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/47/671447/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/48f31d7/ : SUCCESS in 2h 30m 26s\n- openstack-tox-docs http://logs.openstack.org/47/671447/4/gate/openstack-tox-docs/3a82cb8/html/ : SUCCESS in 5m 28s\n- openstack-tox-molecule http://logs.openstack.org/47/671447/4/gate/openstack-tox-molecule/a2c6285/tox/reports.html : SUCCESS in 16m 00s\n- openstack-tox-linters http://logs.openstack.org/47/671447/4/gate/openstack-tox-linters/3cfb9c2/ : SUCCESS in 7m 26s","accounts_in_message":[],"_revision_number":4},{"id":"fc0166b604532bcf5ec88f321d3d41e438c3a112","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-19 13:48:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fdcd928d35b0a6d093f4214f17eccdedfb550a17","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 16:47:33.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0c89979/ : FAILURE in 2h 08m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/e9b7885/ : SUCCESS in 2h 30m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/eedcc63/ : SUCCESS in 2h 07m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/41615c5/ : SUCCESS in 2h 20m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1afea30/ : SUCCESS in 2h 56m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/47/671447/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c4661e1/ : SUCCESS in 2h 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dc3ad6ff51e1b3f36d8a3f77a8ba1108f6fe950a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:48:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-undercloud-containers/953498d/ : SUCCESS in 1h 48m 13s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3bc02af/ : FAILURE in 2h 07m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode/fee9ca7/ : SUCCESS in 2h 36m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f1b3c50/ : SUCCESS in 2h 22m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fa98b1c/ : FAILURE in 2h 05m 44s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/47/671447/4/check/tripleo-ci-fedora-28-standalone/1f427fe/ : FAILURE in 21m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-standalone-upgrade/7bc5097/ : FAILURE in 1h 48m 45s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-stein/ca3eafc/ : SUCCESS in 2h 16m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-rocky/8bea59e/ : SUCCESS in 2h 21m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-queens/bf6674e/ : SUCCESS in 2h 18m 05s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/47/671447/4/check/tripleo-ci-centos-7-containers-multinode-pike/8f0d8a5/ : SUCCESS in 2h 21m 05s\n- openstack-tox-docs http://logs.openstack.org/47/671447/4/check/openstack-tox-docs/60a6856/html/ : SUCCESS in 4m 03s\n- openstack-tox-molecule http://logs.openstack.org/47/671447/4/check/openstack-tox-molecule/b4c62d7/tox/reports.html : SUCCESS in 16m 52s\n- openstack-tox-linters http://logs.openstack.org/47/671447/4/check/openstack-tox-linters/e8d58df/ : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":4},{"id":"01bd5091499bde0381239ba4bbdef248b5438122","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:48:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d0a85e8effe9bc6a8fb87d310ae482f9f7634dec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 22:55:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/47/671447/4/gate/tripleo-ci-centos-7-undercloud-containers/e357ff9/ : SUCCESS in 1h 39m 20s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/47/671447/4/gate/tripleo-ci-centos-7-containers-multinode/07e06a0/ : SUCCESS in 2h 13m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/47/671447/4/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a6189b9/ : SUCCESS in 2h 27m 32s\n- openstack-tox-docs http://logs.openstack.org/47/671447/4/gate/openstack-tox-docs/60e2679/html/ : SUCCESS in 5m 25s\n- openstack-tox-molecule http://logs.openstack.org/47/671447/4/gate/openstack-tox-molecule/4cc7032/tox/reports.html : SUCCESS in 14m 11s\n- openstack-tox-linters http://logs.openstack.org/47/671447/4/gate/openstack-tox-linters/f20a355/ : SUCCESS in 5m 52s","accounts_in_message":[],"_revision_number":4},{"id":"8ecd1d2a652660cabf8339ba8184e9af4fbf6fdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 22:55:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"53ad72adf4b76063765a13f7a5cba714111fbd3f","revisions":{"23cd9c07035a0fabbf4039696387600a30f2359e":{"kind":"REWORK","_number":1,"created":"2019-07-18 10:05:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/47/671447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/47/671447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/1"}}},"commit":{"parents":[{"commit":"d3ecde97750eb0dd8c231702fc44a1038e4d5a15","subject":"Avoids use of ignore_errors while loading OS specific variables","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d3ecde97750eb0dd8c231702fc44a1038e4d5a15"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-18 10:03:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-18 10:03:05.000000000","tz":330},"subject":"Enable os_tempest for master multinode jobs","message":"Enable os_tempest for master multinode jobs\n\nAs overcloud clouds.yaml support landed in Tripleo for train\nrelease and os_tempest ansible role is used for running\ntempest. Multinode container job runs for other releases also\nso it updates the fs010 to run os_tempest for master only.\n\nCloses-Bug: #1837019\n\nChange-Id: I96925574858223348faf35521fb2afdf7710e0db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/23cd9c07035a0fabbf4039696387600a30f2359e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/23cd9c07035a0fabbf4039696387600a30f2359e"}]},"branch":"refs/heads/master"},"fdc369e04f201958f86a7fb88b49e4acbda18cdd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-18 10:27:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/47/671447/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/47/671447/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/2"}}},"commit":{"parents":[{"commit":"3dbb6e794d1d54325d840c239386c624af2c9238","subject":"Run multinode jobs on all patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3dbb6e794d1d54325d840c239386c624af2c9238"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-18 10:03:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-18 10:27:01.000000000","tz":0},"subject":"Enable os_tempest for master multinode jobs","message":"Enable os_tempest for master multinode jobs\n\nAs overcloud clouds.yaml support landed in Tripleo for train\nrelease and os_tempest ansible role is used for running\ntempest. Multinode container job runs for other releases also\nso it updates the fs010 to run os_tempest for master only.\n\nCloses-Bug: #1837019\n\nChange-Id: I96925574858223348faf35521fb2afdf7710e0db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fdc369e04f201958f86a7fb88b49e4acbda18cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fdc369e04f201958f86a7fb88b49e4acbda18cdd"}]},"branch":"refs/heads/master"},"40b77403743432b7914b9c3d8dabfa917f028287":{"kind":"REWORK","_number":3,"created":"2019-07-18 11:54:53.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/47/671447/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/47/671447/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/3"}}},"commit":{"parents":[{"commit":"3dbb6e794d1d54325d840c239386c624af2c9238","subject":"Run multinode jobs on all patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3dbb6e794d1d54325d840c239386c624af2c9238"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-18 10:03:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-18 11:54:05.000000000","tz":330},"subject":"Enable os_tempest for master multinode jobs","message":"Enable os_tempest for master multinode jobs\n\nAs overcloud clouds.yaml support landed in Tripleo for train\nrelease and os_tempest ansible role is used for running\ntempest. Multinode container job runs for other releases also\nso it updates the fs010 to run os_tempest for master only.\n\nCloses-Bug: #1837019\n\nChange-Id: I96925574858223348faf35521fb2afdf7710e0db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/40b77403743432b7914b9c3d8dabfa917f028287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/40b77403743432b7914b9c3d8dabfa917f028287"}]},"branch":"refs/heads/master"},"53ad72adf4b76063765a13f7a5cba714111fbd3f":{"kind":"REWORK","_number":4,"created":"2019-07-18 11:58:40.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/47/671447/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/47/671447/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/671447/4"}}},"commit":{"parents":[{"commit":"3dbb6e794d1d54325d840c239386c624af2c9238","subject":"Run multinode jobs on all patches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3dbb6e794d1d54325d840c239386c624af2c9238"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-18 10:03:05.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2019-07-18 11:58:31.000000000","tz":0},"subject":"Enable os_tempest for master multinode jobs","message":"Enable os_tempest for master multinode jobs\n\nAs overcloud clouds.yaml support landed in Tripleo for train\nrelease and os_tempest ansible role is used for running\ntempest. Multinode container job runs for other releases also\nso it updates the fs010 to run os_tempest for master only.\n\nCloses-Bug: #1837019\n\nChange-Id: I96925574858223348faf35521fb2afdf7710e0db\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/53ad72adf4b76063765a13f7a5cba714111fbd3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/53ad72adf4b76063765a13f7a5cba714111fbd3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
