)]}'
{"id":"openstack%2Ftripleo-quickstart~672075","triplet_id":"openstack%2Ftripleo-quickstart~master~I5de0fc46c6b89592b42e230b73bc9a4048530b69","project":"openstack/tripleo-quickstart","branch":"master","hashtags":[],"change_id":"I5de0fc46c6b89592b42e230b73bc9a4048530b69","subject":"Fail when $VIRTHOST is localhost","status":"MERGED","created":"2019-07-22 15:40:56.000000000","updated":"2019-09-19 19:10:54.000000000","submitted":"2019-09-19 19:09:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672075-1568920174789-a9f074f5","meta_rev_id":"37fbf221b6ff87519079626773c4d87595c3ef81","_number":672075,"virtual_id_number":672075,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-09-19 19:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-09-16 15:55:00.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-09-17 07:00:49.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2019-09-16 12:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2019-09-16 12:55:18.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2019-09-16 12:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-16 12:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":1,"date":"2019-09-16 12:55:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2019-09-17 08:30:21.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 15:57:51.000000000","updated_by":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"reviewer":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"state":"REVIEWER"},{"updated":"2019-08-09 10:22:24.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-09-13 06:01:21.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-09-15 07:40:18.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-09-16 11:33:12.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-09-16 15:55:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-17 08:30:21.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-09-19 19:09:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"005121cb67639abb8914db385ec1e4c41f9d250c","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-07-22 15:40:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cca6205b4a81bae3990b0797adeaea45bd590158","author":{"_account_id":25877,"name":"Luke Short","email":"ekultails@gmail.com","username":"ekultails"},"date":"2019-07-22 15:57:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d88cc6af26ce82d332ac934c69101a6258d6b8d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 18:41:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/8df1990/ : FAILURE in 1h 47m 45s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7a38f6c/ : SUCCESS in 2h 34m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/afc513b/ : FAILURE in 2h 14m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/94f417c/ : FAILURE in 2h 59m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/6fbf1d0/ : SUCCESS in 2h 37m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/310273d/ : SUCCESS in 2h 45m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a374a7b4693ad2de1e924dd825aeb1ce573357a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 19:19:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-undercloud-containers/27c98b3/ : SUCCESS in 1h 29m 21s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/2b4a74f/ : FAILURE in 2h 01m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode/63c6ac9/ : SUCCESS in 2h 33m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e24d0bd/ : SUCCESS in 2h 19m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/031e6bd/ : SUCCESS in 2h 23m 37s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/75/672075/1/check/tripleo-ci-fedora-28-standalone/223460c/ : SUCCESS in 1h 32m 52s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-standalone-upgrade/21c1c78/ : FAILURE in 1h 51m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-stein/64f99b8/ : SUCCESS in 2h 16m 12s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-rocky/8f7cd7f/ : SUCCESS in 2h 08m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-queens/7084001/ : SUCCESS in 2h 21m 34s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-pike/da93018/ : SUCCESS in 2h 24m 43s\n- openstack-tox-docs http://logs.openstack.org/75/672075/1/check/openstack-tox-docs/8379d48/html/ : SUCCESS in 4m 19s\n- openstack-tox-molecule http://logs.openstack.org/75/672075/1/check/openstack-tox-molecule/7200001/tox/reports.html : SUCCESS in 18m 51s\n- openstack-tox-linters http://logs.openstack.org/75/672075/1/check/openstack-tox-linters/7b8e21f/ : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0b85a199de0709d35c572286d1aed4e649ac7c15","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-07-23 14:36:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3c4b6d6234a4ac81e3c073c89f45ff40d2d0cfae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 17:09:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/dce9e61/ : SUCCESS in 2h 28m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/6d8eb19/ : FAILURE in 2h 14m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ca99847/ : FAILURE in 2h 18m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/bc68d6f/ : SUCCESS in 2h 27m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/0a638f5/ : SUCCESS in 2h 31m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/75/672075/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/643e679/ : FAILURE in 1h 35m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1125cef6bb4764ae957d8b29bd935fd0772e0c5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:41:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-undercloud-containers/8706ec8/ : SUCCESS in 1h 36m 27s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/3e08431/ : FAILURE in 2h 12m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode/04f9d08/ : SUCCESS in 2h 19m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/49ab6b8/ : SUCCESS in 2h 13m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9b85783/ : SUCCESS in 2h 18m 38s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/75/672075/1/check/tripleo-ci-fedora-28-standalone/aa3f312/ : SUCCESS in 1h 35m 39s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-standalone-upgrade/644cb32/ : FAILURE in 1h 50m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-stein/2b40112/ : SUCCESS in 2h 25m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-rocky/494e96e/ : SUCCESS in 2h 24m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-queens/eec4bfd/ : SUCCESS in 2h 06m 23s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/75/672075/1/check/tripleo-ci-centos-7-containers-multinode-pike/3e74040/ : SUCCESS in 3h 02m 22s\n- openstack-tox-docs http://logs.openstack.org/75/672075/1/check/openstack-tox-docs/3bf6a1d/html/ : SUCCESS in 5m 41s\n- openstack-tox-molecule http://logs.openstack.org/75/672075/1/check/openstack-tox-molecule/edff50c/tox/reports.html : SUCCESS in 17m 27s\n- openstack-tox-linters http://logs.openstack.org/75/672075/1/check/openstack-tox-linters/a9b4bd5/ : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"981dc40a57e58ed4bc78097571719a16a06917fa","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-07-30 14:47:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"359f6ba9a92c37a69dd8a1f1717eb179f92a0c3c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 18:04:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/91ace3e/ : FAILURE in 1h 43m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze03.softwarefactory-project.io:7979/af70ded99f9a48589b70505a25ac2a00 : RETRY_LIMIT in 7s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/cbf9697/ : SUCCESS in 2h 16m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/da294ce/ : SUCCESS in 2h 16m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/8cc6b3f/ : FAILURE in 1h 12m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/75/672075/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c830357/ : SUCCESS in 3h 09m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88a68eaa25d60c655e60206658e89a402b835de3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 21:50:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-undercloud-containers/c1e2b08/ : SUCCESS in 1h 32m 36s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/e4005b1/ : FAILURE in 2h 00m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-containers-multinode/56dfe16/ : SUCCESS in 2h 17m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4dcbef0/ : SUCCESS in 2h 14m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/848e251/ : FAILURE in 2h 07m 31s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/75/672075/2/check/tripleo-ci-fedora-28-standalone/8236e32/ : SUCCESS in 1h 42m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-standalone-upgrade/121e1c8/ : SUCCESS in 2h 20m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-containers-multinode-stein/d0071fe/ : SUCCESS in 2h 10m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-containers-multinode-rocky/81366bd/ : SUCCESS in 2h 18m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-containers-multinode-queens/f810401/ : FAILURE in 1h 57m 23s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/75/672075/2/check/tripleo-ci-centos-7-containers-multinode-pike/4779169/ : SUCCESS in 2h 22m 56s\n- openstack-tox-docs http://logs.openstack.org/75/672075/2/check/openstack-tox-docs/47b2a55/html/ : SUCCESS in 5m 19s\n- openstack-tox-molecule http://logs.openstack.org/75/672075/2/check/openstack-tox-molecule/0c1454f/tox/reports.html : SUCCESS in 16m 05s\n- openstack-tox-linters http://logs.openstack.org/75/672075/2/check/openstack-tox-linters/d6e86d6/ : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d2397497e985feeab03119918f24eb541ea321a2","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-08-05 12:30:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"41be4dfb5160105a00d01d504bd97f67bea5cc4c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:08:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-undercloud-containers/e648acf/ : SUCCESS in 1h 45m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a6bdf2b/ : FAILURE in 1h 06m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-containers-multinode/265afa3/ : SUCCESS in 2h 15m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/22bfcf2/ : SUCCESS in 2h 17m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d8e499f/ : SUCCESS in 2h 32m 09s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/75/672075/3/check/tripleo-ci-fedora-28-standalone/4ef163c/ : SUCCESS in 1h 47m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-standalone-upgrade/adff737/ : FAILURE in 42m 35s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-containers-multinode-stein/4becac4/ : FAILURE in 1h 11m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-containers-multinode-rocky/7c41aca/ : SUCCESS in 2h 20m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-containers-multinode-queens/24c7c44/ : SUCCESS in 2h 04m 37s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/75/672075/3/check/tripleo-ci-centos-7-containers-multinode-pike/24767ea/ : SUCCESS in 2h 14m 39s\n- openstack-tox-docs https://logs.opendev.org/75/672075/3/check/openstack-tox-docs/2679115/html/ : SUCCESS in 6m 50s\n- openstack-tox-molecule https://logs.opendev.org/75/672075/3/check/openstack-tox-molecule/164fa34/tox/reports.html : SUCCESS in 13m 33s\n- openstack-tox-linters https://logs.opendev.org/75/672075/3/check/openstack-tox-linters/b055ace/ : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":3},{"id":"0c7b6daadeda38573f7bb3a7a4773c947701d2cc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 15:43:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/79f1d77/ : SUCCESS in 2h 34m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fc357cd/ : SUCCESS in 2h 48m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e65d135/ : FAILURE in 2h 31m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b7a5eeb/ : SUCCESS in 3h 09m 53s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/97c9878/ : FAILURE in 1h 01m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/75/672075/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/e4bcfa0/ : FAILURE in 2h 57m 15s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/75/672075/3/openstack-check/tripleo-ci-rhel-8-standalone-rdo/9e2651f/ : FAILURE in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d722ed0d7b40bb28cf8c7e2e0aba46a63b482626","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-08-07 15:27:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"19bed081454b55535817050051321291910356d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 18:30:23.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/af92432/ : FAILURE in 2h 18m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8fa9beb/ : SUCCESS in 2h 41m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/fcffc24/ : FAILURE in 2h 32m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/01e024f/ : SUCCESS in 2h 21m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/2eac389/ : SUCCESS in 2h 35m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/a3ec201/ : FAILURE in 2h 59m 24s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-rhel-8-standalone-rdo/9bf493c/ : FAILURE in 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"495becae68a8e08439e34621faf06cff6f0a07c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 21:19:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-undercloud-containers/44f40f9/ : SUCCESS in 1h 54m 53s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0657652/ : FAILURE in 2h 09m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode/7b07218/ : FAILURE in 2h 29m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1e60df2/ : SUCCESS in 2h 09m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b60b200/ : SUCCESS in 2h 23m 46s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/75/672075/4/check/tripleo-ci-fedora-28-standalone/c58f3ab/ : SUCCESS in 2h 10m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-standalone-upgrade/e7f1a16/ : SUCCESS in 2h 18m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-stein/cb9b64a/ : FAILURE in 7m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-rocky/29ca60f/ : SUCCESS in 2h 24m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-queens/91333be/ : SUCCESS in 2h 01m 34s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-pike/43bb69b/ : SUCCESS in 2h 24m 12s\n- openstack-tox-docs https://logs.opendev.org/75/672075/4/check/openstack-tox-docs/4709329/html/ : SUCCESS in 4m 06s\n- openstack-tox-molecule https://logs.opendev.org/75/672075/4/check/openstack-tox-molecule/b885282/tox/reports.html : SUCCESS in 17m 09s\n- openstack-tox-linters https://logs.opendev.org/75/672075/4/check/openstack-tox-linters/53defca/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":4},{"id":"41ef2c6fefc22451b762c395002e3fcebf1061ec","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-09 10:22:24.000000000","message":"Patch Set 4: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"ed729bf263a072d158cc1eba8c9e08d18a39b75f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-09 13:07:03.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/074db13/ : SUCCESS in 2h 42m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/a918c5c/ : SUCCESS in 2h 40m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/cbe42de/ : SUCCESS in 2h 15m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2f4944d/ : SUCCESS in 2h 26m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/0984650/ : FAILURE in 1h 49m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/fc06fcb/ : RETRY_LIMIT in 6m 59s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/75/672075/4/openstack-check/tripleo-ci-rhel-8-standalone-rdo/69c9260/ : FAILURE in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9d188528694e0e3779681430cd68d54d93f513c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 13:51:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-undercloud-containers/5add9f3/ : SUCCESS in 2h 11m 03s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/a6830e0/ : FAILURE in 2h 00m 53s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode/c01f14d/ : SUCCESS in 2h 15m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c3a151c/ : SUCCESS in 2h 02m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/fb05004/ : SUCCESS in 2h 35m 11s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/75/672075/4/check/tripleo-ci-fedora-28-standalone/1f9b232/ : SUCCESS in 1h 25m 57s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-standalone-upgrade/a4c5122/ : SUCCESS in 2h 13m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-stein/010e364/ : SUCCESS in 2h 24m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-rocky/7ff016c/ : SUCCESS in 2h 26m 04s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-queens/607a5a7/ : SUCCESS in 2h 16m 22s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/75/672075/4/check/tripleo-ci-centos-7-containers-multinode-pike/e1f2e89/ : SUCCESS in 2h 17m 25s\n- openstack-tox-docs https://logs.opendev.org/75/672075/4/check/openstack-tox-docs/315db3a/html/ : SUCCESS in 5m 22s\n- openstack-tox-molecule https://logs.opendev.org/75/672075/4/check/openstack-tox-molecule/f7320a3/tox/reports.html : SUCCESS in 15m 54s\n- openstack-tox-linters https://logs.opendev.org/75/672075/4/check/openstack-tox-linters/cf90f4e/ : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":4},{"id":"1e707023cb162ab84329639992c2d9f98e3c4d43","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-08-12 15:55:08.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"73b33d1b5230530f62ab177891a7eac07ccdf854","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-12 20:03:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/426d31a/ : SUCCESS in 3h 27m 41s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/99bfadb/ : SUCCESS in 3h 32m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ea6e133/ : SUCCESS in 2h 50m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/678742d/ : SUCCESS in 3h 06m 28s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch finger://ze07.softwarefactory-project.io:7979/658591f02da441369fc8b842f644659e : POST_FAILURE in 3h 27m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/75/672075/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/773de5a/ : SUCCESS in 3h 27m 42s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/75/672075/5/openstack-check/tripleo-ci-rhel-8-standalone-rdo/d11188f/ : FAILURE in 25m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2aeed6445decdca8832310b1e3f98c6d76e64ee1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 22:34:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/66ebbaa67f4646d189ba020fb3cabdf6 : SUCCESS in 1h 44m 08s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/444e1cf350b74b2db37ea538764a441b : FAILURE in 1h 57m 57s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f01b3b1ac6de4c0ab10e0ad862af7573 : SUCCESS in 2h 13m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/41daf7b409904dedb7506a16c4b9f0a9 : SUCCESS in 2h 14m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4e789e5775584101b8a2df44488c39e4 : SUCCESS in 2h 29m 18s (non-voting)\n- tripleo-ci-fedora-28-standalone https://zuul.opendev.org/t/openstack/build/50a60265b2ef4e47a9ad0e7d1467481c : SUCCESS in 1h 35m 22s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/c69fdb636e4345888338667869b10a1b : SUCCESS in 2h 31m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/de1a61fb1d5e499b9753147d02afa901 : SUCCESS in 2h 07m 35s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/104c5d79bbc34619aa8df6b66dbe17f7 : SUCCESS in 2h 29m 50s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/87a525db3fe54f6fa1ba467eacd2f7a0 : SUCCESS in 2h 09m 15s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/e68da05d1d0341409be0ede6e532e2fe : SUCCESS in 2h 17m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98015125018f41cca54c26aa0f079a13 : SUCCESS in 4m 22s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/39faa840dc5e46eea4aa4103664254e8 : SUCCESS in 15m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5b86a112ce8a45a7b5df4068481e856b : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":5},{"id":"74439a1924d5a9aad6b886c7e67d031bc9ccfc13","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-08-16 15:49:10.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"cab830bdc7e836d7848cf0816790519f11f492e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 18:19:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/a94579a0d26f4e38b9cbe3545111a2d6 : SUCCESS in 1h 43m 54s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/43e2b5d1664b4597bc3f8327d0c6eeab : POST_FAILURE in 1h 53m 11s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c2de0ef67e7048d2b769d667cb1bf3d0 : SUCCESS in 2h 12m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7a8c48312bbe41a68e5049cd4fdc98f8 : SUCCESS in 2h 19m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a793bff6616242e1a3d46d9f08e29e3d : SUCCESS in 2h 12m 17s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/573fd202e8e842c2acd2db22c9b7273a : SUCCESS in 2h 11m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/a7d7b41b1fad4cdf932d13b42de57adf : SUCCESS in 2h 01m 58s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/22f584736fe84e91996f4fb2609a0179 : SUCCESS in 2h 23m 55s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/fcd30ebc101147b2b0f489b976392393 : SUCCESS in 2h 07m 22s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/45d3060b46944917adbc71ec076eeb3b : SUCCESS in 2h 06m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2a462105334f41b3dc9db7378364ac : SUCCESS in 6m 14s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3453513997464f7faa3b7c241fa09af9 : SUCCESS in 16m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39906863bf4f4be492d798fb33d1ad62 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":6},{"id":"8a9c270b7e46462c64126ff2f0eb85bcf9dbdf6a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 18:32:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6d7db2c/ : SUCCESS in 2h 34m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d441d17/ : SUCCESS in 2h 41m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ef062c4/ : SUCCESS in 2h 37m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2a171af/ : SUCCESS in 2h 13m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/2fafc1f/ : FAILURE in 2h 04m 30s","accounts_in_message":[],"_revision_number":6},{"id":"4eb1a757570d9bcf883adfa81e5c7bbdcc16c579","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-08-25 09:40:34.000000000","message":"Patch Set 6:\n\nI\u0027m not sure it won\u0027t work if you specify localhost or 127.0.0.1, I think I deployed and it was OK. Maybe better just to add a warning, but not to break the installation.","accounts_in_message":[],"_revision_number":6},{"id":"631bd586b1e520332e61f5267def9ccf6edc1963","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-26 09:14:22.000000000","message":"Patch Set 6:\n\n\u003e I\u0027m not sure it won\u0027t work if you specify localhost or 127.0.0.1, I\n \u003e think I deployed and it was OK. Maybe better just to add a warning,\n \u003e but not to break the installation.\n\nThe documentation explicitly says to not use 127.0.0.1 because it\u0027s reserved, if that changes, maybe we should update the documentation..","accounts_in_message":[],"_revision_number":6},{"id":"e09240e3f83299926c32297d7a07e84502cd023d","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-08-26 13:08:39.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"eb22ef1a4843e3c18fa4a6043b6e8494e25b62ed","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-08-26 13:10:21.000000000","message":"Patch Set 7:\n\n@Sagi, thanks for the review.  As @Arx pointed out, the documentation explicitly says not to use localhost or 127.0.0.1.  The installation fails in unexpected ways when you do.\n\nI added this patch because I lost hours trying to figure out why the installation failed.  I think it is better to fail fast when we know the inputs are wrong and that the installation will fail regardless.","accounts_in_message":[],"_revision_number":7},{"id":"711244ed60eddd9a127f9ec04005725a0a1a53e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-26 16:05:22.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4c0fb13/ : SUCCESS in 2h 49m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/75/672075/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/0ba1773/ : SUCCESS in 2h 55m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b0929ec/ : FAILURE in 2h 09m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/abfe0dd/ : SUCCESS in 2h 17m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/cab0f0d/ : FAILURE in 1h 43m 39s","accounts_in_message":[],"_revision_number":7},{"id":"c15e14906eefa30769ae89d6975bb11296ad8bf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 16:07:07.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/6a3eb7d905304ba3b0b410e9cf21ea84 : SUCCESS in 1h 49m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/8a71fbee8f284f138881d123d4ae3997 : FAILURE in 2h 24m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/cd31a66f82a141b7b7722824dc0d5c63 : SUCCESS in 2h 49m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/11a5f382307140b3ae30e388f94e8a46 : SUCCESS in 2h 44m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/19ce115e4e3842528686a1e260a46c85 : SUCCESS in 2h 35m 18s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/2bc2160813424737b083e3174f1ec4af : SUCCESS in 2h 27m 34s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/aa933d7d9f8f46a5bf7ad3a5eb9d2f9c : SUCCESS in 2h 25m 37s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c5a7177725434ffda7dae1c8fd9403f0 : FAILURE in 1h 36m 42s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/08635381e6774c65b963f12085c72c79 : SUCCESS in 2h 23m 39s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/e7fcb31bb70043be8a37bb996b6e2f93 : SUCCESS in 2h 13m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88905c2837b04c9599b9af475cd60b55 : SUCCESS in 4m 53s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/32fdb65b26aa4213bea647a3efda22cd : SUCCESS in 16m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b0ed641c164f46b5837ebdda5eebeec4 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":7},{"id":"e28f6cd181677fdc31ad91b5c0628d3cc94910c5","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-09-04 14:29:14.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"443d4955d8220887c4895497b14a61dbed835d65","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 18:48:31.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/1db2218/ : SUCCESS in 2h 53m 12s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/babbba9/ : FAILURE in 2h 15m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/494490f/ : SUCCESS in 2h 23m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/313dc81/ : TIMED_OUT in 3h 41m 50s","accounts_in_message":[],"_revision_number":8},{"id":"77bd934af7fcccc1ea686a0400f8646d2bac5e1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 21:37:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/4c4e3220527a421ba2523fdfd1cf5ed6 : SUCCESS in 1h 56m 56s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/dcc98d6a27804010905357773f1c4fe9 : FAILURE in 1h 56m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/614432b1786b438082eaa9c1ad7ced63 : SUCCESS in 2h 31m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/aa38fed00cf64817aa046dcd7ffe6b15 : SUCCESS in 2h 26m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/5fa8af9b055d4807ae61dad5986ab5fd : SUCCESS in 2h 50m 44s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/26bde6d21b284a3d9b22d7178bd33c3a : SUCCESS in 2h 14m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/d86a3ef02a954716bd8e437aebf50d5b : SUCCESS in 2h 42m 21s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/a6e4feeea786448ca40af2e77f4bb81f : SUCCESS in 2h 38m 27s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/98f29cd08c9449beae078ce8e49cd5a4 : SUCCESS in 2h 42m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b8f7d6cac649c18b1b222db4b9480d : SUCCESS in 4m 45s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a1653dc4a52a46f28796c042726f877c : SUCCESS in 15m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f207f4795cf646caa2a8c6dc8ea46e75 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":8},{"id":"be56795e3acb275c283dbddcbaa7b306bfecb0d0","author":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"date":"2019-09-13 06:01:21.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"1bf87b5007d9a9497e84aee9227b76a1824280e2","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-09-15 07:40:18.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"599b5d787f61cc2e2dd41ba098b916a30144c2d1","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-09-16 11:33:12.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0e7bf62ecd0863d3a0b3f35095ad5c1105d5dca1","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2019-09-16 12:55:18.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"e889c71ade32f5035b5b0d957bb679edd3f9db43","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 15:55:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/6b222c5/ : FAILURE in 2h 59m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d136711/ : SUCCESS in 2h 10m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/2674fb1/ : SUCCESS in 2h 14m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/92a5e58/ : SUCCESS in 2h 50m 44s","accounts_in_message":[],"_revision_number":9},{"id":"a54c05e3f08973406d9bdda42bdfbbdd5a95a385","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 17:46:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/92e69fe9661d432c99bab4576abddeaa : SUCCESS in 1h 49m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/f6241e38a2a44e3faab4e3cfcb880eb0 : FAILURE in 2h 30m 29s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/878ff09e58d54e7cae6a399ab18b5e1c : SUCCESS in 3h 19m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9315ffca79b54b168e1fb5033df0a803 : SUCCESS in 2h 51m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6a56fc9e93404e218273c984e08f169b : SUCCESS in 2h 24m 34s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0f7fc3da4baf485391b3ed2eba360c0d : SUCCESS in 2h 35m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/dfada896f2804b3284bb842997655187 : SUCCESS in 2h 49m 09s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/78a44b3ecaca4c02bdb64505d94ffe06 : SUCCESS in 2h 37m 15s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/c6ec359e8e114cbb8c68efa72f566cfc : SUCCESS in 2h 13m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65506201a8b54d17994013ad237cc3e8 : SUCCESS in 5m 37s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/894bb423b5594d8eb17ce02525938f89 : SUCCESS in 13m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77d39c049ffc4ef486314cd725ef10e7 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":9},{"id":"d348370f2463abac37b5d79fb4841c99679d703f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-09-17 07:00:49.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c95c2b1d47788c2e54113b5f0ad8b526dbf34c38","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-09-17 07:01:35.000000000","message":"Patch Set 9:\n\n(I\u0027ll +A in a bit unless someone else beats me to it or there is any other update)","accounts_in_message":[],"_revision_number":9},{"id":"b1efbf772d5117b3e41729cfc823084a9d2c2f65","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-09-17 08:30:21.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"bd7d831750967ce4730ad8b23224e47e5760ff4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 08:30:43.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"7833c659f73cfa0f55a350423af0ccc062ca7fa1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-17 20:13:08.000000000","message":"Abandoned\n\nWe are facing gate issue: https://bugs.launchpad.net/tripleo/+bug/1844446\n\nTo clear the gate we need to abandon this patch and I will restore once the gate is ready again to land patches in TripleO. Please don\u0027t touch this patch, and ask on #tripleo Wes or Emilien for any question. Thanks for your patience.","accounts_in_message":[],"_revision_number":9},{"id":"c5f03efcae1663d8a74dba57e50d42cd6088c247","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-19 02:26:38.000000000","message":"Restored\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"0ce6081a360e3987c781e765602884d409bbb154","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-19 06:06:09.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/95486c0/ : SUCCESS in 3h 31m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/63c91d2/ : FAILURE in 1h 54m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/fd68cb3/ : SUCCESS in 3h 04m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/75/672075/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/9b9f611/ : FAILURE in 2h 28m 14s","accounts_in_message":[],"_revision_number":9},{"id":"9b9369934accdb3893e239c610eb33701f6472a9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 06:29:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/09c015237e90438fb18c2e016667bb9c : SUCCESS in 2h 09m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/262b82e3a8934d58bad3bd4f356f77d0 : FAILURE in 2h 01m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/18616794953641e9a1092f571145a2db : SUCCESS in 3h 08m 26s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/713663176674417fa250778790f0ef68 : SUCCESS in 2h 39m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4983550a87cc4cbd9bb5938e6db0b0cb : SUCCESS in 2h 48m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/d70b53852fc04b64868fb1238d1f60e8 : SUCCESS in 2h 45m 17s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/c8a8c9c606e14f6aa5e5989957c6b21b : SUCCESS in 2h 54m 30s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/f1418fb4139f4c9382bbb6d5c29734e5 : SUCCESS in 2h 43m 49s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a76a634d6fbf48c6a709f9530fe3ba16 : SUCCESS in 2h 24m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723d4916883a466895406450fcd656ac : SUCCESS in 4m 40s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3a0591c3f1074f0088aa395be543239d : SUCCESS in 12m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a041516a6f434b4297332878c4de57f9 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":9},{"id":"bff830647c9c37e42809a18953a0a800ae139d87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 06:30:14.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"27ca971cb4a5dfa2c032e7850e0eace4216be9ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:09:34.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/65be8c11290d4aee9c9d810514cdd285 : SUCCESS in 2h 17m 17s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/4450a5aeff444b769c93e34df5249563 : SUCCESS in 3h 26m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4534bae1508047bea50a8adb009c471b : SUCCESS in 3h 18m 17s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/adb958c1ca494695b193f2fd5cf5a337 : SUCCESS in 2h 53m 56s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/be8f994778b343ddae9dd82080c93b72 : SUCCESS in 2h 42m 44s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/ab632145ddd8492cb01c04d0acfbfdd6 : SUCCESS in 2h 27m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cd57c18e834570ad0918a35a526162 : SUCCESS in 5m 00s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6d39f7dd423745aa819e5f514adbdc36 : SUCCESS in 14m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4d25edac89b4455dac8a50e61297eeac : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":9},{"id":"3dfc45e331e21e202f171dc29b6f67a69df27951","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:09:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"37fbf221b6ff87519079626773c4d87595c3ef81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:10:54.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7b0e2afa9f140b59baf1f8d0530144f : SUCCESS in 52s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"7467149907c6ad523e139a77680901247d1a3baf","revisions":{"2161adf5f10dd9d14fe7fb4f8f57577c0ed92009":{"kind":"REWORK","_number":1,"created":"2019-07-22 15:40:56.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/1"}}},"commit":{"parents":[{"commit":"326c8bde59ba13e45ca66a00ad576ab0f4262ed3","subject":"Merge \"Added RedHat-8 release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/326c8bde59ba13e45ca66a00ad576ab0f4262ed3"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:40:55.000000000","tz":-300},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2161adf5f10dd9d14fe7fb4f8f57577c0ed92009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2161adf5f10dd9d14fe7fb4f8f57577c0ed92009"}]},"branch":"refs/heads/master"},"e2adf210fcb23a74cfb35ede3229387737aec54a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-07-30 14:47:15.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/2"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-30 14:47:15.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e2adf210fcb23a74cfb35ede3229387737aec54a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e2adf210fcb23a74cfb35ede3229387737aec54a"}]},"branch":"refs/heads/master"},"3371b7d7dd846987e2303a57a0a4c8d28172b001":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-05 12:30:56.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/3"}}},"commit":{"parents":[{"commit":"464df34ed4a3e8d0638654a3b9675529e885f3e5","subject":"Merge \"[RHEL8] Fixed rdo_dlrn in release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/464df34ed4a3e8d0638654a3b9675529e885f3e5"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-08-05 12:30:56.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3371b7d7dd846987e2303a57a0a4c8d28172b001"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3371b7d7dd846987e2303a57a0a4c8d28172b001"}]},"branch":"refs/heads/master"},"9533aeacd6514ec0b15e9141db3dcdabd05ed462":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-07 15:27:11.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/4"}}},"commit":{"parents":[{"commit":"0316539d81ba6f3c8950e665e317af3b4af8ea21","subject":"Merge \"[RHEL8]Fix release file for DLRN package builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0316539d81ba6f3c8950e665e317af3b4af8ea21"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-08-07 15:27:11.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9533aeacd6514ec0b15e9141db3dcdabd05ed462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9533aeacd6514ec0b15e9141db3dcdabd05ed462"}]},"branch":"refs/heads/master"},"2eb9ae46224b7665fc57b182934b11bcfd36bf3f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-12 15:55:08.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/5"}}},"commit":{"parents":[{"commit":"236c35deb1bd87608e7bee01a8f4ebb1195532d7","subject":"Remove \u0027change_when\u0027 when using include_role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/236c35deb1bd87608e7bee01a8f4ebb1195532d7"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-08-12 15:55:08.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2eb9ae46224b7665fc57b182934b11bcfd36bf3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2eb9ae46224b7665fc57b182934b11bcfd36bf3f"}]},"branch":"refs/heads/master"},"ebf68c342df60277ef127b4531bd5e189c7b1400":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-08-16 15:49:10.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/6"}}},"commit":{"parents":[{"commit":"ec61b56821b867b167dbabf0554df37bb9437258","subject":"Use py3 firstly in running get-overcloud-nodes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ec61b56821b867b167dbabf0554df37bb9437258"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-08-16 15:49:10.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ebf68c342df60277ef127b4531bd5e189c7b1400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ebf68c342df60277ef127b4531bd5e189c7b1400"}]},"branch":"refs/heads/master"},"c6c33715f7f529a8e3f843e0acacde9039c8ac8a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-26 13:08:39.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/7"}}},"commit":{"parents":[{"commit":"8c74d174a5a2c10cffa1567c83a0a27eabc935a4","subject":"Merge \"[Doc] Fix links to featureset files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8c74d174a5a2c10cffa1567c83a0a27eabc935a4"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-08-26 13:08:39.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c6c33715f7f529a8e3f843e0acacde9039c8ac8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c6c33715f7f529a8e3f843e0acacde9039c8ac8a"}]},"branch":"refs/heads/master"},"66baf177119eb14fb6eafe69fc5369f247cb9908":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-09-04 14:29:14.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/8"}}},"commit":{"parents":[{"commit":"8f490ae8252a507e0f0f0a91972431d55a204ed0","subject":"Merge \"Move  *_container_cli settings to extras-common\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8f490ae8252a507e0f0f0a91972431d55a204ed0"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-09-04 14:29:14.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/66baf177119eb14fb6eafe69fc5369f247cb9908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/66baf177119eb14fb6eafe69fc5369f247cb9908"}]},"branch":"refs/heads/master"},"7467149907c6ad523e139a77680901247d1a3baf":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-16 12:55:18.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/75/672075/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/75/672075/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/75/672075/9"}}},"commit":{"parents":[{"commit":"c0b5fdcdde5b736f0baf005bf1f5ffc83722597d","subject":"Merge \"Add ceph-ansible to promotion jobs on RHEL8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c0b5fdcdde5b736f0baf005bf1f5ffc83722597d"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-07-22 15:38:07.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2019-09-16 12:55:18.000000000","tz":0},"subject":"Fail when $VIRTHOST is localhost","message":"Fail when $VIRTHOST is localhost\n\nThere are known issues when using tripleo-quickstart\nwith localhost or 127.0.0.1 as the target host. See [1].\n\nThis patch adds a check to make sure that neither one of\nthose addresses are being used as $VIRTHOST and fails with\na message to use 127.0.0.2 instead.\n\n[1] https://docs.openstack.org/tripleo-quickstart/2.1.0/readme.html#deploying-on-localhost\n\nChange-Id: I5de0fc46c6b89592b42e230b73bc9a4048530b69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7467149907c6ad523e139a77680901247d1a3baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7467149907c6ad523e139a77680901247d1a3baf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
