)]}'
{"id":"openstack%2Ftripleo-quickstart~673005","triplet_id":"openstack%2Ftripleo-quickstart~master~Iee4eaaad37d1ac1d595e2430c1cbea09349522b8","project":"openstack/tripleo-quickstart","branch":"master","topic":"ruck_sprint13","hashtags":[],"change_id":"Iee4eaaad37d1ac1d595e2430c1cbea09349522b8","subject":"Generate fs matrix from source yaml files","status":"MERGED","created":"2019-07-26 15:08:58.000000000","updated":"2019-08-08 19:47:48.000000000","submitted":"2019-08-08 19:47:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673005-1565293668598-00db80d6","meta_rev_id":"bbe1173948dab84d0ec88c6ccba41966a6b7c838","_number":673005,"virtual_id_number":673005,"owner":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-08 19:47:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-06 22:02:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-08-08 07:23:17.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2019-08-08 16:57:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-08-08 16:59:55.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2019-08-08 16:53:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-08 16:57:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":1,"date":"2019-08-08 16:59:57.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-29 02:11:04.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-07-29 02:11:09.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-29 02:11:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-30 12:02:29.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-06 22:02:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-08 07:23:17.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-08 16:53:51.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-08-08 16:57:21.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-08-08 16:57:39.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-08-08 16:59:57.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-08 19:47:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7c9c3aec8f77a2fb5e8de971d9c18ad909f021a","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-26 15:08:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e59c9d8bf87debe011d01fcb595d51118a102cf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 16:27:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/05/673005/1/check/openstack-tox-docs/56e6c91/html/ : SUCCESS in 4m 38s\n- openstack-tox-molecule http://logs.openstack.org/05/673005/1/check/openstack-tox-molecule/806cbdb/tox/reports.html : SUCCESS in 17m 13s\n- openstack-tox-linters http://logs.openstack.org/05/673005/1/check/openstack-tox-linters/771c636/ : FAILURE in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f13dd62e6fe5b0f73232ed316a8f2f9c815f2f08","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-26 19:26:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"900c2a6773e658d65ca2a79c64c36a9ea0e0271c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-26 22:29:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/673005/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e73ce0e/ : SUCCESS in 2h 19m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/673005/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ce197fd/ : SUCCESS in 2h 35m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/05/673005/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c5d836d/ : SUCCESS in 2h 04m 56s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/05/673005/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/5f82cba/ : SUCCESS in 2h 17m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/05/673005/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/5f9262f/ : SUCCESS in 2h 33m 06s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/05/673005/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/42c9145/ : SUCCESS in 3h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"36f64e6c05b9d2b6635b26c0628cd150c51af6fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 23:04:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-undercloud-containers/3e51e32/ : SUCCESS in 1h 35m 18s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/04f853e/ : FAILURE in 1h 49m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-containers-multinode/f57a2e3/ : SUCCESS in 3h 04m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d5b4cdf/ : SUCCESS in 2h 24m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/48c1bed/ : SUCCESS in 2h 44m 50s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/673005/2/check/tripleo-ci-fedora-28-standalone/233e0bb/ : SUCCESS in 1h 37m 40s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-standalone-upgrade/2cedff2/ : SUCCESS in 1h 59m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-containers-multinode-stein/69a2d92/ : SUCCESS in 2h 29m 40s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-containers-multinode-rocky/3ce73cb/ : SUCCESS in 2h 31m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-containers-multinode-queens/ca943b3/ : SUCCESS in 2h 09m 29s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/05/673005/2/check/tripleo-ci-centos-7-containers-multinode-pike/494b071/ : SUCCESS in 2h 20m 48s\n- openstack-tox-docs http://logs.openstack.org/05/673005/2/check/openstack-tox-docs/799e78b/html/ : SUCCESS in 3m 59s\n- openstack-tox-molecule http://logs.openstack.org/05/673005/2/check/openstack-tox-molecule/a4787b3/tox/reports.html : SUCCESS in 13m 31s\n- openstack-tox-linters http://logs.openstack.org/05/673005/2/check/openstack-tox-linters/4471cdb/ : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"695c2983dc09f4f739bd636f3f051bc07f06fc47","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-29 02:10:32.000000000","message":"Patch Set 2:\n\ngenerated matrix: http://logs.openstack.org/05/673005/2/check/openstack-tox-docs/799e78b/html/feature-configuration.html","accounts_in_message":[],"_revision_number":2},{"id":"8a11e02259eac252953df8dcb9a1df745e486521","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-29 07:24:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\ncool story bro!","accounts_in_message":[],"_revision_number":2},{"id":"68762ec2e3147ccc550c4d2e153c41f8f021e866","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-29 07:33:24.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nThanks for working on this, it is a nice addition, few comments and can we include this script under tox.ini so that it will be called when tox -e linters is called.","accounts_in_message":[],"_revision_number":2},{"id":"cadc0fd9aea5400c365cb9a71cb3eac51a96311c","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-29 18:53:15.000000000","message":"Patch Set 2:\n\n(6 comments)\n\n\u003e (5 comments)\n \u003e \n \u003e Thanks for working on this, it is a nice addition, few comments and\n \u003e can we include this script under tox.ini so that it will be called\n \u003e when tox -e linters is called.\n\nlinters include the script currently. What we can do for a more automatic update is to integrate the script with tox/sphinx... in a follow-up change maybe...","accounts_in_message":[],"_revision_number":2},{"id":"808a5195a2613c1a71a267eaa39135c9d86cea8a","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-29 18:53:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"23aa4727527891e4f42dae65c5bfcfdfe3d89ef1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 21:32:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/673005/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/55e4b3e/ : SUCCESS in 2h 27m 22s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/673005/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/8d1af68/ : SUCCESS in 2h 29m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/05/673005/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/5a73ccb/ : FAILURE in 2h 37m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/05/673005/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/33dbb36/ : SUCCESS in 2h 11m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/05/673005/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/757078e/ : SUCCESS in 2h 23m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/05/673005/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/ddfbdbf/ : SUCCESS in 2h 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2c7ba6c3f68ebc5047addb89de5da1184c5d4eb6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:58:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-undercloud-containers/f27c5c2/ : SUCCESS in 1h 35m 37s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/42fba94/ : FAILURE in 1h 59m 26s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-containers-multinode/1e73e0d/ : SUCCESS in 2h 24m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/453fb1b/ : SUCCESS in 2h 24m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7c570e8/ : SUCCESS in 2h 23m 29s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-standalone/8c2894f/ : SUCCESS in 1h 44m 08s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/673005/3/check/tripleo-ci-fedora-28-standalone/c905895/ : FAILURE in 48m 11s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-standalone-upgrade/09a4bd3/ : SUCCESS in 2h 09m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-containers-multinode-stein/2933851/ : SUCCESS in 2h 15m 47s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-containers-multinode-rocky/5776c72/ : SUCCESS in 2h 30m 42s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-containers-multinode-queens/3c24a6b/ : SUCCESS in 1h 57m 18s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-containers-multinode-pike/a74c873/ : FAILURE in 2h 16m 03s\n- openstack-tox-docs http://logs.openstack.org/05/673005/3/check/openstack-tox-docs/f6af665/html/ : SUCCESS in 9m 09s\n- openstack-tox-molecule http://logs.openstack.org/05/673005/3/check/openstack-tox-molecule/25929fb/tox/reports.html : SUCCESS in 16m 56s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/05/673005/3/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/aadeb94/ : FAILURE in 1h 20m 59s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/05/673005/3/check/openstack-tox-linters/a11567f/ : FAILURE in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"950f467622c72afba4fa191e23d76df543265dc1","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-30 07:00:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c879193156ac2e509d239b94adc964cf409726d3","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-07-30 12:02:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nthis is fucking awesome, thank you :)\nPlease update it so that it will only run w/ python3 or at least doc that fact.","accounts_in_message":[],"_revision_number":3},{"id":"34a230831f9548066e60bf5165291d8a4166feaa","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-07-30 12:10:07.000000000","message":"Patch Set 3: Code-Review+1\n\n+1 before the changes required for python3 just to show how happy and excited I am with this!","accounts_in_message":[],"_revision_number":3},{"id":"b3892309ecc190a3bc050782c438250254120c55","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-30 12:14:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"452178add40ac914d0f95dc6d3047a72ab2271bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 15:50:14.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/51be3e0/ : SUCCESS in 3h 33m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d551bc0/ : SUCCESS in 3h 30m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c0fbdea/ : FAILURE in 2h 26m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8cf324d/ : SUCCESS in 3h 10m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/cb057bc/ : SUCCESS in 3h 29m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/dbb6bd5/ : SUCCESS in 3h 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"32ba92b64fce0df3fd092dd5ba44a5168346c0fb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 18:17:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-undercloud-containers/32f80f3/ : SUCCESS in 1h 30m 41s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/d56cb7a/ : FAILURE in 2h 16m 37s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode/03454ac/ : SUCCESS in 2h 11m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3b18dcc/ : SUCCESS in 2h 29m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/63f651e/ : SUCCESS in 2h 19m 00s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-standalone/8aa0cd9/ : FAILURE in 58m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/673005/4/check/tripleo-ci-fedora-28-standalone/e4fd107/ : FAILURE in 1h 04m 19s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-standalone-upgrade/cd03c57/ : SUCCESS in 2h 00m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-stein/5de0b6f/ : SUCCESS in 2h 13m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-rocky/0ac8965/ : SUCCESS in 2h 27m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-queens/b3e0e70/ : SUCCESS in 2h 34m 38s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-pike/4256ea5/ : SUCCESS in 2h 21m 28s\n- openstack-tox-docs http://logs.openstack.org/05/673005/4/check/openstack-tox-docs/15db641/html/ : SUCCESS in 5m 40s\n- openstack-tox-molecule http://logs.openstack.org/05/673005/4/check/openstack-tox-molecule/865d2c1/tox/reports.html : SUCCESS in 13m 23s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/907cc1c/ : FAILURE in 1h 07m 19s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/05/673005/4/check/openstack-tox-linters/13fa3f8/ : FAILURE in 6m 13s","accounts_in_message":[],"_revision_number":4},{"id":"df4872936b021591c15bb1db4a3e8a8f188a900f","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-31 07:01:13.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nrecheck\n\n    https://bugs.launchpad.net/tripleo/+bug/1837388 tripleo-ci jobs failing intermittently with \"Image prepare failed: 401 Client Error: Unauthorized\"\n\nhttp://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-standalone/8aa0cd9/logs/undercloud/var/log/tripleo-container-image-prepare.log.txt.gz#_2019-07-30_16_30_25_438","accounts_in_message":[],"_revision_number":4},{"id":"535eb7b23e0b1727ddaa9eb7cfd671d22d7a32e2","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-31 17:23:21.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"241416f15d1fb1e286c1dd44ebff2611185bb53c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 20:48:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9e4c5af/ : SUCCESS in 2h 26m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/7e7ffdb/ : SUCCESS in 2h 57m 04s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b332389/ : SUCCESS in 2h 36m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/07fdaec/ : SUCCESS in 2h 14m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/f1791cc/ : SUCCESS in 2h 31m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/b39ffac/ : SUCCESS in 3h 12m 05s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/05/673005/4/openstack-check/tripleo-ci-rhel-8-standalone-rdo/14a8af6/ : FAILURE in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3525607005ba360877ebbc01e1b156fb8120d481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 23:24:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-undercloud-containers/1a5380b/ : SUCCESS in 1h 29m 49s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/7ff51d1/ : FAILURE in 1h 47m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode/71ddfb9/ : SUCCESS in 2h 25m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e6fecba/ : SUCCESS in 2h 07m 32s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3277f59/ : SUCCESS in 2h 22m 32s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-standalone/164fb83/ : SUCCESS in 1h 43m 03s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/05/673005/4/check/tripleo-ci-fedora-28-standalone/02e244f/ : FAILURE in 42m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-standalone-upgrade/9634e4c/ : FAILURE in 1h 38m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-stein/3ff4c90/ : SUCCESS in 2h 03m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-rocky/68b9cc3/ : SUCCESS in 2h 17m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-queens/f1bf542/ : SUCCESS in 2h 31m 12s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-containers-multinode-pike/ef890a9/ : SUCCESS in 2h 19m 18s\n- openstack-tox-docs http://logs.openstack.org/05/673005/4/check/openstack-tox-docs/92d0ecf/html/ : SUCCESS in 3m 47s\n- openstack-tox-molecule http://logs.openstack.org/05/673005/4/check/openstack-tox-molecule/ae13c02/tox/reports.html : SUCCESS in 14m 40s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container http://logs.openstack.org/05/673005/4/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/eaff392/ : SUCCESS in 2h 07m 48s (non-voting)\n- openstack-tox-linters http://logs.openstack.org/05/673005/4/check/openstack-tox-linters/97dbf02/ : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":4},{"id":"97bed3ec347df04d2ba0b1cff8c4be93099b4051","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-08-01 13:25:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ba6e256d7e02e97cd0c289faa4ec207c48d29ab5","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-08-01 13:26:29.000000000","message":"Patch Set 5:\n\nfixed linters in previous ps","accounts_in_message":[],"_revision_number":5},{"id":"56dbca1a423ff5ca4ebcbd69efaaef260049472b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 18:28:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/673005/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0c5e3c4/ : SUCCESS in 2h 47m 55s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/673005/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/49a7145/ : SUCCESS in 2h 40m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/05/673005/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/a301a84/ : SUCCESS in 2h 10m 10s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/05/673005/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/06fb4ab/ : SUCCESS in 2h 14m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/05/673005/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/0bb954f/ : TIMED_OUT in 3h 57m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/05/673005/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/1e7b0b5/ : TIMED_OUT in 5h 00m 24s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/05/673005/5/openstack-check/tripleo-ci-rhel-8-standalone-rdo/946b8f1/ : FAILURE in 13m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b015ecbda1180dd143efac110d88698c2643eec6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 00:09:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-undercloud-containers/939478c/ : SUCCESS in 1h 34m 47s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/05aa6c0/ : TIMED_OUT in 3h 12m 05s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-containers-multinode/4fffff7/ : SUCCESS in 2h 28m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d8af7e2/ : SUCCESS in 2h 42m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7791b3c/ : SUCCESS in 2h 25m 08s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-standalone/7a24cd2/ : SUCCESS in 2h 51m 34s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/05/673005/5/check/tripleo-ci-fedora-28-standalone/416f392/ : FAILURE in 45m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-standalone-upgrade/47fb9a2/ : FAILURE in 1h 45m 55s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-containers-multinode-stein/f98fbb9/ : SUCCESS in 2h 12m 58s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-containers-multinode-rocky/b302248/ : SUCCESS in 2h 14m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-containers-multinode-queens/a1df321/ : SUCCESS in 1h 59m 24s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-containers-multinode-pike/7137a49/ : SUCCESS in 2h 56m 08s\n- openstack-tox-docs https://logs.opendev.org/05/673005/5/check/openstack-tox-docs/81cd93c/html/ : SUCCESS in 5m 26s\n- openstack-tox-molecule https://logs.opendev.org/05/673005/5/check/openstack-tox-molecule/d3b013b/tox/reports.html : SUCCESS in 13m 00s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/05/673005/5/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/692003c/ : FAILURE in 2h 20m 57s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/05/673005/5/check/openstack-tox-linters/cb6c658/ : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":5},{"id":"b69cb551fb702cc25d6bcd37f6296b228f9e6e13","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-05 07:42:21.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fd6d21c3597b8d8eb86e9ccfed01861a471b8611","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-08-05 10:46:28.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dcfc6c47e9376b2fb5e9148636208651ed83c599","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-08-05 15:28:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"60229eac2be5a75a4b8b87ce933cdd68dc26567c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:10:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-undercloud-containers/d0b1db8/ : SUCCESS in 1h 29m 34s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/0571509/ : FAILURE in 1h 09m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-containers-multinode/7d00781/ : SUCCESS in 2h 14m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1ba16fb/ : SUCCESS in 2h 19m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/65e5e00/ : SUCCESS in 2h 23m 44s (non-voting)\n- tripleo-ci-centos-7-standalone https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-standalone/7a8e959/ : SUCCESS in 1h 46m 31s\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/05/673005/6/check/tripleo-ci-fedora-28-standalone/7577674/ : SUCCESS in 1h 31m 09s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-standalone-upgrade/c51f77b/ : FAILURE in 42m 49s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-containers-multinode-stein/48cb243/ : FAILURE in 1h 14m 15s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-containers-multinode-rocky/ade74c4/ : SUCCESS in 2h 29m 46s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-containers-multinode-queens/664d826/ : SUCCESS in 2h 26m 17s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-containers-multinode-pike/1d73a7d/ : SUCCESS in 2h 22m 11s\n- openstack-tox-docs https://logs.opendev.org/05/673005/6/check/openstack-tox-docs/1c80035/html/ : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://logs.opendev.org/05/673005/6/check/build-openstack-releasenotes/a68286c/html/ : SUCCESS in 2m 57s\n- openstack-tox-molecule https://logs.opendev.org/05/673005/6/check/openstack-tox-molecule/94e7d7b/tox/reports.html : SUCCESS in 13m 29s\n- tripleo-ci-centos-7-scenario009-multinode-oooq-container https://logs.opendev.org/05/673005/6/check/tripleo-ci-centos-7-scenario009-multinode-oooq-container/706948b/ : FAILURE in 1h 55m 02s (non-voting)\n- openstack-tox-linters https://logs.opendev.org/05/673005/6/check/openstack-tox-linters/2f366a2/ : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":6},{"id":"6c382800f346178f911d65419bec8b6f488c87b4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 19:26:15.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/673005/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/b4fc9b4/ : SUCCESS in 2h 58m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/673005/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/65589aa/ : SUCCESS in 2h 50m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/05/673005/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d18975d/ : TIMED_OUT in 3h 55m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/05/673005/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/cd5786e/ : SUCCESS in 2h 30m 32s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/05/673005/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b43f314/ : FAILURE in 56m 33s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/05/673005/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/bc6b7ef/ : SUCCESS in 2h 48m 49s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/05/673005/6/openstack-check/tripleo-ci-rhel-8-standalone-rdo/8e1b7e3/ : FAILURE in 16m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"74d88fd2f042e0714ea4a66713a873d44d068493","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-05 21:09:01.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"1d81ac859f7fb4208f547147655087620ca4e20b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-06 07:01:12.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b32a8be6b8266a5ae7bf0c77609a82d1c4c48b48","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-08-06 13:15:50.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f39376886c256145ff8f1ff63898a35b8808e675","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-08-06 13:16:54.000000000","message":"Patch Set 6:\n\nAlso, please move the fs-matrix.py from root to somewhere inside doc/ so we avoid poluting the root folder.","accounts_in_message":[],"_revision_number":6},{"id":"df6f5768f59a83dc378e853458789fd7f8bd4137","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-08-06 13:18:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nsee comments (easy bits to fix)","accounts_in_message":[],"_revision_number":6},{"id":"f05b50cda7013cbd4e9dbbf83bd753a72ef52985","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-06 13:21:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"03d4569d86d235629b9c97fdbaa5aad44004e450","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-08-06 16:38:18.000000000","message":"Patch Set 6:\n\n(5 comments)\n\na new patchset is coming...","accounts_in_message":[],"_revision_number":6},{"id":"3ee5173c642bbb8dc7af68fc952088d36e5f6854","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-08-06 19:10:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c0ad069254bbfd73d85130d74701c5d450419ab1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-06 22:02:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/05/673005/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/0f062cc/ : FAILURE in 2h 49m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/05/673005/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9ffba9b/ : FAILURE in 1h 04m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/05/673005/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/aaea086/ : FAILURE in 2h 19m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/05/673005/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/391a72e/ : SUCCESS in 2h 22m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/05/673005/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/c4f4119/ : SUCCESS in 2h 37m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/05/673005/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/c2e688e/ : SUCCESS in 2h 33m 46s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/05/673005/7/openstack-check/tripleo-ci-rhel-8-standalone-rdo/044df36/ : FAILURE in 16m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0600031890d7a44bc8ffb771f80ed5f5133ab23f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-06 23:58:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-undercloud-containers/bb9dca4/ : SUCCESS in 1h 30m 57s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/cfbef06/ : FAILURE in 2h 00m 00s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-containers-multinode/d75232b/ : SUCCESS in 2h 23m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/35ec81a/ : SUCCESS in 2h 12m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/06a01cc/ : SUCCESS in 2h 21m 16s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/05/673005/7/check/tripleo-ci-fedora-28-standalone/b29a415/ : SUCCESS in 2h 02m 14s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-standalone-upgrade/386053c/ : SUCCESS in 2h 23m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-containers-multinode-stein/43f7d80/ : SUCCESS in 2h 21m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-containers-multinode-rocky/e9bb48a/ : SUCCESS in 2h 15m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-containers-multinode-queens/f090f56/ : SUCCESS in 2h 12m 17s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/05/673005/7/check/tripleo-ci-centos-7-containers-multinode-pike/3f2e319/ : SUCCESS in 2h 10m 39s\n- openstack-tox-docs https://logs.opendev.org/05/673005/7/check/openstack-tox-docs/959dcf4/html/ : SUCCESS in 3m 34s\n- build-openstack-releasenotes https://logs.opendev.org/05/673005/7/check/build-openstack-releasenotes/6d094d6/html/ : SUCCESS in 3m 19s\n- openstack-tox-molecule https://logs.opendev.org/05/673005/7/check/openstack-tox-molecule/2cbd5da/tox/reports.html : SUCCESS in 15m 47s\n- openstack-tox-linters https://logs.opendev.org/05/673005/7/check/openstack-tox-linters/8989143/ : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":7},{"id":"72d014d232b1d0b50a279496c5b2cc0ba7773cc5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-08-08 07:23:17.000000000","message":"Patch Set 7: Code-Review+2\n\nthanks didn\u0027t get warnings from tox -e docs this time","accounts_in_message":[],"_revision_number":7},{"id":"6b1066c5a283214bb69cc4a0febbc6316a7247e6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-08 16:53:51.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"cccb8399b17a4ddf17c70623677d79b46e3d6ce0","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-08 16:54:42.000000000","message":"Patch Set 7:\n\nHere is the beautiful output https://logs.opendev.org/05/673005/7/check/openstack-tox-docs/959dcf4/html/feature-configuration.html thanks rfolco :-)","accounts_in_message":[],"_revision_number":7},{"id":"a91c389ffb8255379b2c53dfeb0ad771bdcc8a9f","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-08-08 16:57:21.000000000","message":"Patch Set 7: Code-Review+1\n\nGTG!","accounts_in_message":[],"_revision_number":7},{"id":"2aa434534a3edb98dd4984958ed8b458a04513af","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-08-08 16:57:39.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b826e464256f6b217255c63e76cfad9d1b7b54e2","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-08 16:59:55.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a9fd30f8d49bc95d147c57bab9a06db724d539a2","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-08 16:59:57.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"6919ae3ace9ea649411a9ef894e13adcec28ff73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 17:00:17.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"870a1e4ac9a389761d0da1ce55fe2b829ecf661c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 19:47:48.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/05/673005/7/gate/tripleo-ci-centos-7-undercloud-containers/adaaf39/ : SUCCESS in 1h 46m 49s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/05/673005/7/gate/tripleo-ci-centos-7-containers-multinode/0c230f0/ : SUCCESS in 2h 31m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/05/673005/7/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c10ad1d/ : SUCCESS in 2h 32m 11s\n- openstack-tox-docs https://logs.opendev.org/05/673005/7/gate/openstack-tox-docs/5b705aa/html/ : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://logs.opendev.org/05/673005/7/gate/build-openstack-releasenotes/f115f4f/html/ : SUCCESS in 3m 26s\n- openstack-tox-molecule https://logs.opendev.org/05/673005/7/gate/openstack-tox-molecule/b0a3eae/tox/reports.html : SUCCESS in 15m 34s\n- openstack-tox-linters https://logs.opendev.org/05/673005/7/gate/openstack-tox-linters/e13d93c/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":7},{"id":"75c67ccb066403690f6d626740edd10c546105f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 19:47:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"edaa1be4bcb53db87ea61a3bfd26e94a012cb5bc","revisions":{"dcb98ddb27e6d35e48d9ce0689be63fd5184fada":{"kind":"REWORK","_number":1,"created":"2019-07-26 15:08:58.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/05/673005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/05/673005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/1"}}},"commit":{"parents":[{"commit":"6133bf14f3330d923759e8ae4fdfc12820f90a00","subject":"Merge \"Revert \"Temporary exclude ceph-ansible rc11 from nautilus repo\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6133bf14f3330d923759e8ae4fdfc12820f90a00"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 12:16:18.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 15:07:47.000000000","tz":-180},"subject":"Generate fs matrix from source yaml files","message":"Generate fs matrix from source yaml files\n\nUse featureset yaml files as the unique source of truth.\n\nWhen updating fs yaml files, one should run a python code to parse\nfeatureset yaml files and generate the matrix in\ndoc/source/featureset-configuration.rst. The output of the python\nscript should be copied into rst file.\n\nChange includes:\n\n  - Normalized matrix headers w/ 5 chars\n  - Legend w/ acronyms and symbols automatically generated\n  - Features marked as enabled, disabled, not found, or enabled by\n    release.\n  - Removed default settings enabled after release XYZ that are\n    not controlled by the featureset file: e.g. cnf-dl.\n  - Removed settings not present in any fs: ssl uc\n    (SSL Undercloud), NI Bond (Network Isolation Bond), nonha\n    (non-HA), PM (Pacemaker), IPSec (IPSec tunnels).\n  - Merged SSL overcloud INT and EXT into a single SSL OC.\n\nChange-Id: Iee4eaaad37d1ac1d595e2430c1cbea09349522b8\nTask: https://tree.taiga.io/project/tripleo-ci-board/task/1202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dcb98ddb27e6d35e48d9ce0689be63fd5184fada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dcb98ddb27e6d35e48d9ce0689be63fd5184fada"}]},"branch":"refs/heads/master"},"70275e3b0b04cff91a4a1f54efaf9931b8af34d5":{"kind":"REWORK","_number":2,"created":"2019-07-26 19:26:55.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/05/673005/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/05/673005/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/2"}}},"commit":{"parents":[{"commit":"6133bf14f3330d923759e8ae4fdfc12820f90a00","subject":"Merge \"Revert \"Temporary exclude ceph-ansible rc11 from nautilus repo\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6133bf14f3330d923759e8ae4fdfc12820f90a00"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 12:16:18.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 19:25:39.000000000","tz":-180},"subject":"Generate fs matrix from source yaml files","message":"Generate fs matrix from source yaml files\n\nUse featureset yaml files as the unique source of truth.\n\nWhen updating fs yaml files, one should run a python code to parse\nfeatureset yaml files and generate the matrix in\ndoc/source/featureset-configuration.rst. The output of the python\nscript should be copied into rst file.\n\nChange includes:\n\n  - Normalized matrix headers w/ 5 chars\n  - Legend w/ acronyms and symbols automatically generated\n  - Features marked as enabled, disabled, not found, or enabled by\n    release.\n  - Removed default settings enabled after release XYZ that are\n    not controlled by the featureset file: e.g. cnf-dl.\n  - Removed settings not present in any fs: ssl uc\n    (SSL Undercloud), NI Bond (Network Isolation Bond), nonha\n    (non-HA), PM (Pacemaker), IPSec (IPSec tunnels).\n  - Merged SSL overcloud INT and EXT into a single SSL OC.\n\nChange-Id: Iee4eaaad37d1ac1d595e2430c1cbea09349522b8\nTask: https://tree.taiga.io/project/tripleo-ci-board/task/1202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/70275e3b0b04cff91a4a1f54efaf9931b8af34d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/70275e3b0b04cff91a4a1f54efaf9931b8af34d5"}]},"branch":"refs/heads/master"},"06a9d4a183d34243c47376ce106a6e77c9e78e8c":{"kind":"REWORK","_number":3,"created":"2019-07-29 18:53:34.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/05/673005/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/05/673005/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/3"}}},"commit":{"parents":[{"commit":"6133bf14f3330d923759e8ae4fdfc12820f90a00","subject":"Merge \"Revert \"Temporary exclude ceph-ansible rc11 from nautilus repo\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6133bf14f3330d923759e8ae4fdfc12820f90a00"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 12:16:18.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-29 18:50:45.000000000","tz":-180},"subject":"Generate fs matrix from source yaml files","message":"Generate fs matrix from source yaml files\n\nUse featureset yaml files as the unique source of truth.\n\nWhen updating fs yaml files, one should run a python code to parse\nfeatureset yaml files and generate the matrix in\ndoc/source/featureset-configuration.rst. The output of the python\nscript should be copied into rst file.\n\nChange includes:\n\n  - Normalized matrix headers w/ 5 chars\n  - Legend w/ acronyms and symbols automatically generated\n  - Features marked as enabled, disabled, not found, or enabled by\n    release.\n  - Removed default settings enabled after release XYZ that are\n    not controlled by the featureset file: e.g. cnf-dl.\n  - Removed settings not present in any fs: ssl uc\n    (SSL Undercloud), NI Bond (Network Isolation Bond), nonha\n    (non-HA), PM (Pacemaker), IPSec (IPSec tunnels).\n  - Merged SSL overcloud INT and EXT into a single SSL OC.\n\nChange-Id: Iee4eaaad37d1ac1d595e2430c1cbea09349522b8\nTask: https://tree.taiga.io/project/tripleo-ci-board/task/1202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/06a9d4a183d34243c47376ce106a6e77c9e78e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/06a9d4a183d34243c47376ce106a6e77c9e78e8c"}]},"branch":"refs/heads/master"},"09b8b1481ff0a973977fb3cc00d1a4ff2928a50a":{"kind":"REWORK","_number":4,"created":"2019-07-30 12:14:56.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/05/673005/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/05/673005/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/4"}}},"commit":{"parents":[{"commit":"6133bf14f3330d923759e8ae4fdfc12820f90a00","subject":"Merge \"Revert \"Temporary exclude ceph-ansible rc11 from nautilus repo\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6133bf14f3330d923759e8ae4fdfc12820f90a00"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 12:16:18.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-30 12:14:26.000000000","tz":-180},"subject":"Generate fs matrix from source yaml files","message":"Generate fs matrix from source yaml files\n\nUse featureset yaml files as the unique source of truth.\n\nWhen updating fs yaml files, one should run a python code to parse\nfeatureset yaml files and generate the matrix in\ndoc/source/featureset-configuration.rst. The output of the python\nscript should be copied into rst file.\n\nChange includes:\n\n  - Normalized matrix headers w/ 5 chars\n  - Legend w/ acronyms and symbols automatically generated\n  - Features marked as enabled, disabled, not found, or enabled by\n    release.\n  - Removed default settings enabled after release XYZ that are\n    not controlled by the featureset file: e.g. cnf-dl.\n  - Removed settings not present in any fs: ssl uc\n    (SSL Undercloud), NI Bond (Network Isolation Bond), nonha\n    (non-HA), PM (Pacemaker), IPSec (IPSec tunnels).\n  - Merged SSL overcloud INT and EXT into a single SSL OC.\n\nChange-Id: Iee4eaaad37d1ac1d595e2430c1cbea09349522b8\nTask: https://tree.taiga.io/project/tripleo-ci-board/task/1202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/09b8b1481ff0a973977fb3cc00d1a4ff2928a50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/09b8b1481ff0a973977fb3cc00d1a4ff2928a50a"}]},"branch":"refs/heads/master"},"a88857c5938670df7f9777831462849fbb0f09fa":{"kind":"REWORK","_number":5,"created":"2019-08-01 13:25:13.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/05/673005/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/05/673005/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/5"}}},"commit":{"parents":[{"commit":"6133bf14f3330d923759e8ae4fdfc12820f90a00","subject":"Merge \"Revert \"Temporary exclude ceph-ansible rc11 from nautilus repo\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6133bf14f3330d923759e8ae4fdfc12820f90a00"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 12:16:18.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-08-01 13:24:34.000000000","tz":-180},"subject":"Generate fs matrix from source yaml files","message":"Generate fs matrix from source yaml files\n\nUse featureset yaml files as the unique source of truth.\n\nWhen updating fs yaml files, one should run a python code to parse\nfeatureset yaml files and generate the matrix in\ndoc/source/featureset-configuration.rst. The output of the python\nscript should be copied into rst file.\n\nChange includes:\n\n  - Normalized matrix headers w/ 5 chars\n  - Legend w/ acronyms and symbols automatically generated\n  - Features marked as enabled, disabled, not found, or enabled by\n    release.\n  - Removed default settings enabled after release XYZ that are\n    not controlled by the featureset file: e.g. cnf-dl.\n  - Removed settings not present in any fs: ssl uc\n    (SSL Undercloud), NI Bond (Network Isolation Bond), nonha\n    (non-HA), PM (Pacemaker), IPSec (IPSec tunnels).\n  - Merged SSL overcloud INT and EXT into a single SSL OC.\n\nChange-Id: Iee4eaaad37d1ac1d595e2430c1cbea09349522b8\nTask: https://tree.taiga.io/project/tripleo-ci-board/task/1202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a88857c5938670df7f9777831462849fbb0f09fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a88857c5938670df7f9777831462849fbb0f09fa"}]},"branch":"refs/heads/master"},"f188620f125c959858107569bb235b329cc2a2e9":{"kind":"REWORK","_number":6,"created":"2019-08-05 15:28:11.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/05/673005/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/05/673005/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/6"}}},"commit":{"parents":[{"commit":"464df34ed4a3e8d0638654a3b9675529e885f3e5","subject":"Merge \"[RHEL8] Fixed rdo_dlrn in release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/464df34ed4a3e8d0638654a3b9675529e885f3e5"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 12:16:18.000000000","tz":-180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-08-05 15:27:51.000000000","tz":60},"subject":"Generate fs matrix from source yaml files","message":"Generate fs matrix from source yaml files\n\nUse featureset yaml files as the unique source of truth.\n\nWhen updating fs yaml files, one should run a python code to parse\nfeatureset yaml files and generate the matrix in\ndoc/source/featureset-configuration.rst. The output of the python\nscript should be copied into rst file.\n\nChange includes:\n\n  - Normalized matrix headers w/ 5 chars\n  - Legend w/ acronyms and symbols automatically generated\n  - Features marked as enabled, disabled, not found, or enabled by\n    release.\n  - Removed default settings enabled after release XYZ that are\n    not controlled by the featureset file: e.g. cnf-dl.\n  - Removed settings not present in any fs: ssl uc\n    (SSL Undercloud), NI Bond (Network Isolation Bond), nonha\n    (non-HA), PM (Pacemaker), IPSec (IPSec tunnels).\n  - Merged SSL overcloud INT and EXT into a single SSL OC.\n\nChange-Id: Iee4eaaad37d1ac1d595e2430c1cbea09349522b8\nTask: https://tree.taiga.io/project/tripleo-ci-board/task/1202\nCo-Authored-By: Sorin Sbarnea \u003cssbarnea@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f188620f125c959858107569bb235b329cc2a2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f188620f125c959858107569bb235b329cc2a2e9"}]},"branch":"refs/heads/master"},"edaa1be4bcb53db87ea61a3bfd26e94a012cb5bc":{"kind":"REWORK","_number":7,"created":"2019-08-06 19:10:29.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/05/673005/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/05/673005/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/05/673005/7"}}},"commit":{"parents":[{"commit":"0316539d81ba6f3c8950e665e317af3b4af8ea21","subject":"Merge \"[RHEL8]Fix release file for DLRN package builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0316539d81ba6f3c8950e665e317af3b4af8ea21"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-26 12:16:18.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-08-06 19:07:30.000000000","tz":-180},"subject":"Generate fs matrix from source yaml files","message":"Generate fs matrix from source yaml files\n\nUse featureset yaml files as the unique source of truth.\n\nWhen updating fs yaml files, one should run \u0027tox -e docs\u0027\nto update the featureset matrix. Tox calls a python code\nto parse featureset yaml files and generate the matrix in\ndoc/source/featureset-configuration-generated.rst, which\nin turn, is imported into the featureset configuration page.\n\nChange includes:\n\n  - Normalized matrix headers w/ 5 chars\n  - Legend w/ acronyms and symbols automatically generated\n  - Features marked as enabled, disabled, not found, or enabled by\n    release.\n  - Removed default settings enabled after release XYZ that are\n    not controlled by the featureset file: e.g. cnf-dl.\n  - Removed settings not present in any fs: ssl uc\n    (SSL Undercloud), NI Bond (Network Isolation Bond), nonha\n    (non-HA), PM (Pacemaker), IPSec (IPSec tunnels).\n  - Merged SSL overcloud INT and EXT into a single SSL OC.\n\nChange-Id: Iee4eaaad37d1ac1d595e2430c1cbea09349522b8\nTask: https://tree.taiga.io/project/tripleo-ci-board/task/1202\nCo-Authored-By: Sorin Sbarnea \u003cssbarnea@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/edaa1be4bcb53db87ea61a3bfd26e94a012cb5bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/edaa1be4bcb53db87ea61a3bfd26e94a012cb5bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
