)]}'
{"id":"openstack%2Ftripleo-quickstart~673517","triplet_id":"openstack%2Ftripleo-quickstart~master~I6f12c263899bc96a5daa1d108d4b8fade77baaf6","project":"openstack/tripleo-quickstart","branch":"master","hashtags":[],"change_id":"I6f12c263899bc96a5daa1d108d4b8fade77baaf6","subject":"[RHEL8]Fix release file for DLRN package builds","status":"MERGED","created":"2019-07-30 13:22:44.000000000","updated":"2019-08-05 23:57:08.000000000","submitted":"2019-08-05 23:57:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673517-1565049428889-f6c17697","meta_rev_id":"badcec8fe03e04fec415a480672386e8d83cd21c","_number":673517,"virtual_id_number":673517,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2019-08-05 23:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-05 12:59:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":2,"date":"2019-08-05 14:09:58.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2019-08-05 20:37:26.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"value":1,"date":"2019-08-05 20:39:53.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 13:23:09.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-07-30 13:23:25.000000000","updated_by":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"reviewer":{"_account_id":10022,"name":"Gabriele Cerami","email":"gcerami@redhat.com","username":"panda"},"state":"REVIEWER"},{"updated":"2019-07-30 13:23:41.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-30 13:23:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-07-30 13:23:56.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-07-30 14:18:55.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2019-07-31 09:03:04.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-08-05 12:59:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 20:37:26.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-05 20:39:53.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-05 23:57:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e97de268244a0375e22742b3433a35c6c1e0a95d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-30 13:22:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb5e75c246f9cca5d1d6b82ab42cef14ce9447b5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-30 13:55:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"effcf3df687fe70923b6dc17b558d3511a733a58","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-30 14:18:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"780403302bafa9705d3a518657539a96fa9bd765","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-30 14:25:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"720f1bc3655964f0de5063a84783c4277f7065ef","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-30 14:28:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c74940b6d73421ac559ebe538ca46aee1ef3df66","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-30 18:18:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"51f0fd1b18c082e366cad95a16266328c0dab3e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 22:03:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/673517/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c03ea85/ : SUCCESS in 2h 34m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/673517/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b8121fd/ : SUCCESS in 2h 29m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/673517/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/867deaa/ : SUCCESS in 3h 43m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/673517/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/8336459/ : SUCCESS in 2h 12m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/673517/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/8629462/ : SUCCESS in 2h 24m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/673517/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/9024024/ : SUCCESS in 2h 49m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9a5e0738d724000c64059c375c33c9524e5f709e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 00:37:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-undercloud-containers/bf428e9/ : SUCCESS in 1h 27m 48s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/c5b8114/ : FAILURE in 2h 14m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-containers-multinode/1d2cb83/ : SUCCESS in 2h 10m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b72a49a/ : SUCCESS in 2h 12m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/ab59d05/ : SUCCESS in 2h 25m 45s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/673517/4/check/tripleo-ci-fedora-28-standalone/3639ce1/ : SUCCESS in 1h 41m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-standalone-upgrade/35c9336/ : SUCCESS in 2h 02m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-containers-multinode-stein/232e64a/ : SUCCESS in 2h 02m 30s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-containers-multinode-rocky/dc9105e/ : SUCCESS in 2h 38m 03s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-containers-multinode-queens/6529961/ : SUCCESS in 2h 15m 01s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/673517/4/check/tripleo-ci-centos-7-containers-multinode-pike/7860cdb/ : SUCCESS in 2h 24m 23s\n- openstack-tox-docs http://logs.openstack.org/17/673517/4/check/openstack-tox-docs/0112206/html/ : SUCCESS in 3m 46s\n- openstack-tox-molecule http://logs.openstack.org/17/673517/4/check/openstack-tox-molecule/6c8cd0a/tox/reports.html : SUCCESS in 14m 41s\n- openstack-tox-linters http://logs.openstack.org/17/673517/4/check/openstack-tox-linters/b935790/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":4},{"id":"957e8d21f2215086e11ccd4c65b2cc7d18c8adfd","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-31 06:09:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"43c5241dfc9b9ccf1e9a95d90b6d85cc64378ce3","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-31 06:46:41.000000000","message":"Patch Set 5: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"e8c3815ee3cb1481bf32b0fe65790e85971c0b06","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-31 09:02:43.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7732042be3b6c8a65c9bc158ec5866876fdbb210","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-31 09:03:04.000000000","message":"Patch Set 5: Code-Review-1\n\ncos i just worfklowed https://review.opendev.org/#/c/673709/3","accounts_in_message":[],"_revision_number":5},{"id":"e68ff09bdf9dc9403fe548ba050c333f8c765292","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 09:03:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/673517/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/e979f95/ : SUCCESS in 2h 35m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/673517/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/aee2eee/ : FAILURE in 1h 24m 44s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/673517/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3dcfe8b/ : FAILURE in 2h 07m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/673517/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/0547655/ : SUCCESS in 2h 18m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/673517/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/640b8e4/ : SUCCESS in 2h 52m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/673517/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/d733164/ : SUCCESS in 2h 51m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8590067d43eb4457713db13c9482da616fb0c860","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-31 09:04:22.000000000","message":"Patch Set 5: Workflow-1\n\nThis is a wip and it is related to dlrn packages.","accounts_in_message":[],"_revision_number":5},{"id":"5467c5ae77fabc9d20abe17407eb0a0bca183e40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 10:02:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-undercloud-containers/22032c0/ : SUCCESS in 1h 30m 35s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/5e69502/ : FAILURE in 1h 40m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-containers-multinode/d22eb14/ : SUCCESS in 2h 26m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/23e2f64/ : SUCCESS in 2h 13m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a05daad/ : SUCCESS in 2h 12m 02s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/673517/5/check/tripleo-ci-fedora-28-standalone/eb6a996/ : SUCCESS in 1h 38m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-standalone-upgrade/e13262e/ : SUCCESS in 2h 05m 24s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-containers-multinode-stein/a78d777/ : SUCCESS in 2h 00m 01s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-containers-multinode-rocky/8cd0ce6/ : SUCCESS in 2h 41m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-containers-multinode-queens/5ef7f00/ : SUCCESS in 2h 16m 38s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/673517/5/check/tripleo-ci-centos-7-containers-multinode-pike/2fb24c0/ : SUCCESS in 2h 15m 18s\n- openstack-tox-docs http://logs.openstack.org/17/673517/5/check/openstack-tox-docs/b6f1aec/html/ : SUCCESS in 5m 26s\n- openstack-tox-molecule http://logs.openstack.org/17/673517/5/check/openstack-tox-molecule/23f0019/tox/reports.html : SUCCESS in 22m 54s\n- openstack-tox-linters http://logs.openstack.org/17/673517/5/check/openstack-tox-linters/8975fcc/ : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c28f2eca83f45852460274641fb507d9b879c458","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-31 10:59:27.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"189a691b975fb25294a37290a182e85383614652","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2019-07-31 11:00:00.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"08515d396682ae87963831ead1bdf554b550e8d7","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-31 12:42:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f7f438945bd953ee079e8636ac98975129cdf8da","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-31 16:48:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/673517/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2178017/ : SUCCESS in 2h 26m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/673517/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/c0303b0/ : FAILURE in 2h 13m 08s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/673517/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/3cce822/ : TIMED_OUT in 4h 04m 47s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/673517/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b9acbac/ : FAILURE in 2h 53m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/673517/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/ecdcba7/ : TIMED_OUT in 4h 00m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/673517/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/957273f/ : SUCCESS in 2h 45m 59s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/17/673517/8/openstack-check/tripleo-ci-rhel-8-standalone-rdo/b685edb/ : FAILURE in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4aeb8325e33d60090a161317f007c03ab527084b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 18:58:38.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-undercloud-containers/5c6630b/ : SUCCESS in 1h 21m 30s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/afab687/ : FAILURE in 2h 12m 02s (non-voting)\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-containers-multinode/ef4b449/ : SUCCESS in 2h 18m 56s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b8d30d6/ : SUCCESS in 2h 18m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/03c0d7b/ : SUCCESS in 2h 24m 47s (non-voting)\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/17/673517/8/check/tripleo-ci-fedora-28-standalone/677d614/ : SUCCESS in 1h 50m 55s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-standalone-upgrade/a89a292/ : SUCCESS in 1h 52m 43s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-containers-multinode-stein/23b0d31/ : SUCCESS in 2h 12m 36s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-containers-multinode-rocky/dc179b2/ : SUCCESS in 2h 24m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-containers-multinode-queens/14d8d2f/ : SUCCESS in 2h 21m 02s\n- tripleo-ci-centos-7-containers-multinode-pike http://logs.openstack.org/17/673517/8/check/tripleo-ci-centos-7-containers-multinode-pike/bdf5a1c/ : SUCCESS in 2h 13m 35s\n- openstack-tox-docs http://logs.openstack.org/17/673517/8/check/openstack-tox-docs/b9ab510/html/ : SUCCESS in 3m 33s\n- openstack-tox-molecule http://logs.openstack.org/17/673517/8/check/openstack-tox-molecule/7fd9fe4/tox/reports.html : SUCCESS in 14m 34s\n- openstack-tox-linters http://logs.openstack.org/17/673517/8/check/openstack-tox-linters/27a8afe/ : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":8},{"id":"ac8d38dcc1e81e8370089bbd66e7885a32d79e4b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-05 09:54:40.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c7d4833429f9176630ee254564946a12b050815a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 12:32:14.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-undercloud-containers/2d38702/ : SUCCESS in 1h 25m 18s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-containerized-undercloud-upgrades/ba4344b/ : FAILURE in 1h 01m 51s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-containers-multinode/2ad4c92/ : SUCCESS in 2h 16m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8f1dfc0/ : SUCCESS in 2h 35m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0796b29/ : SUCCESS in 2h 27m 26s (non-voting)\n- tripleo-ci-fedora-28-standalone https://logs.opendev.org/17/673517/9/check/tripleo-ci-fedora-28-standalone/2455e3d/ : FAILURE in 1h 03m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-standalone-upgrade/0f0eb55/ : FAILURE in 43m 08s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-containers-multinode-stein/59cd3c9/ : FAILURE in 1h 09m 14s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-rocky https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-containers-multinode-rocky/97c1af6/ : SUCCESS in 2h 18m 56s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-containers-multinode-queens/fd2b708/ : SUCCESS in 2h 16m 49s\n- tripleo-ci-centos-7-containers-multinode-pike https://logs.opendev.org/17/673517/9/check/tripleo-ci-centos-7-containers-multinode-pike/bac5423/ : SUCCESS in 2h 14m 52s\n- openstack-tox-docs https://logs.opendev.org/17/673517/9/check/openstack-tox-docs/457138d/html/ : SUCCESS in 4m 34s\n- openstack-tox-molecule https://logs.opendev.org/17/673517/9/check/openstack-tox-molecule/713d269/tox/reports.html : SUCCESS in 14m 23s\n- openstack-tox-linters https://logs.opendev.org/17/673517/9/check/openstack-tox-linters/57fe7b6/ : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":9},{"id":"571a7f719228c4bd26ef3a76ea6846ef4c7ad022","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 12:59:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/17/673517/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a47fa0d/ : SUCCESS in 2h 49m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/17/673517/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/659ebc2/ : SUCCESS in 3h 00m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/17/673517/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/9b6cc8c/ : SUCCESS in 2h 21m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/17/673517/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/b35691d/ : SUCCESS in 2h 48m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/17/673517/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/97b73f8/ : FAILURE in 1h 02m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053 http://logs.rdoproject.org/17/673517/9/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset053/e04eb2c/ : SUCCESS in 3h 03m 05s (non-voting)\n- tripleo-ci-rhel-8-standalone-rdo http://logs.rdoproject.org/17/673517/9/openstack-check/tripleo-ci-rhel-8-standalone-rdo/9910d56/ : FAILURE in 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b0e99c545a1f55317ffa75d004f38ee58f3ae8a7","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-05 14:09:58.000000000","message":"Patch Set 9: Code-Review+2\n\nNoting that the standalone job still fails\n\u003cchandankumar\u003e rlandy: as the container with that hash does not exists","accounts_in_message":[],"_revision_number":9},{"id":"321c2057bf0f01efd730002d848e68d1ca0e9775","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-05 20:37:26.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"e43fabc255513498af0510d0dfdcf8161ba54113","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-05 20:39:53.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"4b36e4b5cb639e4cfdaa976f801e93f1ac055f45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 20:40:10.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"90653694c959d094f87006d66dc4eec5d77b34b7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 23:57:08.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://logs.opendev.org/17/673517/9/gate/tripleo-ci-centos-7-undercloud-containers/42c31e9/ : SUCCESS in 1h 50m 30s\n- tripleo-ci-centos-7-containers-multinode https://logs.opendev.org/17/673517/9/gate/tripleo-ci-centos-7-containers-multinode/f6ec9fe/ : SUCCESS in 2h 36m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://logs.opendev.org/17/673517/9/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3cb91b2/ : SUCCESS in 2h 27m 27s\n- openstack-tox-docs https://logs.opendev.org/17/673517/9/gate/openstack-tox-docs/b48f4ed/html/ : SUCCESS in 4m 22s\n- openstack-tox-molecule https://logs.opendev.org/17/673517/9/gate/openstack-tox-molecule/fa1333d/tox/reports.html : SUCCESS in 14m 14s\n- openstack-tox-linters https://logs.opendev.org/17/673517/9/gate/openstack-tox-linters/6647882/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":9},{"id":"badcec8fe03e04fec415a480672386e8d83cd21c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 23:57:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"12e4abeaac7060d309ab405f65346c465f9547e0","revisions":{"1ca87974b6c3438ca2becbfee975fa8fed91351c":{"kind":"REWORK","_number":1,"created":"2019-07-30 13:22:44.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/673517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/1"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1ca87974b6c3438ca2becbfee975fa8fed91351c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1ca87974b6c3438ca2becbfee975fa8fed91351c"}]},"branch":"refs/heads/master"},"0c65c0ce9f30317b82477647effd44c923944a30":{"kind":"REWORK","_number":2,"created":"2019-07-30 13:55:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/673517/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/2"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:55:01.000000000","tz":330},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0c65c0ce9f30317b82477647effd44c923944a30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0c65c0ce9f30317b82477647effd44c923944a30"}]},"branch":"refs/heads/master"},"e74ef0d10132e1a16071baf4893d7dc1ff20e73e":{"kind":"REWORK","_number":3,"created":"2019-07-30 14:25:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/673517/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/3"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 14:25:43.000000000","tz":330},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e74ef0d10132e1a16071baf4893d7dc1ff20e73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e74ef0d10132e1a16071baf4893d7dc1ff20e73e"}]},"branch":"refs/heads/master"},"747930e3b89997b03adb7723087af6de47e3e044":{"kind":"REWORK","_number":4,"created":"2019-07-30 18:18:15.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/673517/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/4"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 18:17:15.000000000","tz":330},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n* Added workaround for building packages using DLRN for rhel8\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/747930e3b89997b03adb7723087af6de47e3e044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/747930e3b89997b03adb7723087af6de47e3e044"}]},"branch":"refs/heads/master"},"6532ade7fcdd716b46e1e065a97e345887799a8b":{"kind":"REWORK","_number":5,"created":"2019-07-31 06:09:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/673517/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/5"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-31 06:08:18.000000000","tz":330},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n* Added workaround for building packages using DLRN for rhel8\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6532ade7fcdd716b46e1e065a97e345887799a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6532ade7fcdd716b46e1e065a97e345887799a8b"}]},"branch":"refs/heads/master"},"0982f61d7749c623b3f8324b97db5ca2b969b85d":{"kind":"REWORK","_number":6,"created":"2019-07-31 10:59:27.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/17/673517/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/6"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-07-31 10:58:53.000000000","tz":0},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n* Added workaround for building packages using DLRN for rhel8\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0982f61d7749c623b3f8324b97db5ca2b969b85d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0982f61d7749c623b3f8324b97db5ca2b969b85d"}]},"branch":"refs/heads/master"},"61f9a91788da67fff9a09a29a0f03ccfb67c3fb9":{"kind":"REWORK","_number":7,"created":"2019-07-31 11:00:00.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/17/673517/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/7"}}},"commit":{"parents":[{"commit":"4516faa7ac75611c0697d9f77e8db696b3aa85a1","subject":"Merge \"[RHEL8] promotion-testing-hash-master release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4516faa7ac75611c0697d9f77e8db696b3aa85a1"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Javier Peña","email":"jpena@redhat.com","date":"2019-07-31 10:59:57.000000000","tz":0},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n* Added workaround for building packages using DLRN for rhel8\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/61f9a91788da67fff9a09a29a0f03ccfb67c3fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/61f9a91788da67fff9a09a29a0f03ccfb67c3fb9"}]},"branch":"refs/heads/master"},"97e4bb30df274d8089ef518769a8ed6e0924bc36":{"kind":"REWORK","_number":8,"created":"2019-07-31 12:42:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/673517/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/8"}}},"commit":{"parents":[{"commit":"d268d31b6a6909ed024ba636a7489367298c936a","subject":"Add dlrn_hash_tag_newest to RHEL-8 repo file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d268d31b6a6909ed024ba636a7489367298c936a"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-31 12:42:14.000000000","tz":330},"subject":"[RHEL8]Fix release file for dlrn","message":"[RHEL8]Fix release file for dlrn\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock is needed by dlrn for building packages, now avaailble\n  through rhui, included in release file.\n* Added workaround for building packages using DLRN for rhel8\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/97e4bb30df274d8089ef518769a8ed6e0924bc36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/97e4bb30df274d8089ef518769a8ed6e0924bc36"}]},"branch":"refs/heads/master"},"12e4abeaac7060d309ab405f65346c465f9547e0":{"kind":"REWORK","_number":9,"created":"2019-08-05 09:54:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/17/673517/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/17/673517/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/17/673517/9"}}},"commit":{"parents":[{"commit":"464df34ed4a3e8d0638654a3b9675529e885f3e5","subject":"Merge \"[RHEL8] Fixed rdo_dlrn in release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/464df34ed4a3e8d0638654a3b9675529e885f3e5"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-30 13:20:48.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-05 09:54:14.000000000","tz":330},"subject":"[RHEL8]Fix release file for DLRN package builds","message":"[RHEL8]Fix release file for DLRN package builds\n\n* https://softwarefactory-project.io/r/#/c/15950/ sets the\n  dlrn_target to redhat and same is updated in release file.\n* Mock and systemd-container is needed by dlrn for building packages,\n  now availble through rhui, included in release file.\n* Added /usr/share/rhn/RHN-ORG-TRUSTED-SSL-CERT file needed for\n  RHEL-8 mock cfg.\n\nChange-Id: I6f12c263899bc96a5daa1d108d4b8fade77baaf6\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/12e4abeaac7060d309ab405f65346c465f9547e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/12e4abeaac7060d309ab405f65346c465f9547e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
