)]}'
{"id":"openstack%2Ftripleo-quickstart~678242","triplet_id":"openstack%2Ftripleo-quickstart~master~I564f7fc7c30f292e7f81f935032d12e7ee2b0e0c","project":"openstack/tripleo-quickstart","branch":"master","topic":"temp-increase-timeout","hashtags":[],"change_id":"I564f7fc7c30f292e7f81f935032d12e7ee2b0e0c","subject":"DNM  Test review to allow vexxhost qualification to  complete","status":"ABANDONED","created":"2019-08-23 14:28:48.000000000","updated":"2020-03-17 20:12:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"820b1dfc403eb8fd2340dea3b911aceb7d8f9abc","_number":678242,"virtual_id_number":678242,"owner":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2019-08-23 18:12:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-08-23 18:20:11.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":-1,"date":"2019-08-23 14:29:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"date":"2019-08-23 14:29:16.000000000","_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 14:29:16.000000000","updated_by":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"reviewer":{"_account_id":9061,"name":"David Moreau Simard","email":"moi@dmsimard.com","username":"dmsimard"},"state":"REVIEWER"},{"updated":"2019-08-23 18:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 18:20:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ba651b7545a70ad3e4c0ff2b3728a3ffc825bc5","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-23 14:28:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8973d6e782cc45aa2d827eb4f26ec82d2de2858a","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-23 14:29:23.000000000","message":"Patch Set 1: Code-Review-1\n\njust for test","accounts_in_message":[],"_revision_number":1},{"id":"8b858b9ccb005815b7334c0a5223ac84a6e3f658","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 18:12:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/2a2f1304fea24f8480d3ab632521710d : SUCCESS in 1h 54m 31s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4d7ac99683d840ceab0ddd71a40a7ad6 : FAILURE in 2h 02m 10s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/eedcf0111f36473dbd7eca626d844f87 : SUCCESS in 3h 04m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b50aa08c1e9540bf843cf3573a7bed72 : SUCCESS in 2h 38m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/7f5b99f2fc6e4f8499247630a35cfc8f : SUCCESS in 2h 37m 08s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f405ff69ee684402a14379ae948b98c3 : SUCCESS in 2h 02m 41s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/476d6d0758734436be4f1f80733c53dd : SUCCESS in 2h 07m 18s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/988343398bef469d8b66c6c9e8eb2dc6 : SUCCESS in 2h 22m 49s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/d8213b86ece04ab28fa17081c4ba467f : SUCCESS in 2h 11m 36s\n- tripleo-ci-centos-7-containers-multinode-pike https://zuul.opendev.org/t/openstack/build/7fa4c4f9e1b8473bb939c6aa23263240 : SUCCESS in 2h 47m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba64f1ddbb0347f0b9c3f446d1c14f5e : SUCCESS in 4m 22s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/cd0d939a044a449d9bb22df0ff042ee1 : SUCCESS in 15m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c5a0d5c646b6401d90c4c208515284e3 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"fa6c56793864f323310838715f04476aa27d5d7d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 18:20:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/42/678242/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/2862d18/ : FAILURE in 3h 38m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/42/678242/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/2feda9a/ : FAILURE in 2h 52m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/42/678242/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/37bc878/ : FAILURE in 2h 17m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/42/678242/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/3b9e3c2/ : FAILURE in 1h 18m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/42/678242/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/fb1eb8d/ : FAILURE in 1h 44m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0ce1b21912ec0b0c900123e685d3fbfa75622476","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-03-17 20:12:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aed157a02563f8c72c969733090d0a54123939c5","revisions":{"aed157a02563f8c72c969733090d0a54123939c5":{"kind":"REWORK","_number":1,"created":"2019-08-23 14:28:48.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/42/678242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/42/678242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/42/678242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/42/678242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/42/678242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/42/678242/1"}}},"commit":{"parents":[{"commit":"53608c26bb74b2d8dfe8f9220175bf7fb1eac133","subject":"Merge \"Use config_drive for network basicops tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/53608c26bb74b2d8dfe8f9220175bf7fb1eac133"}]}],"author":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2019-08-23 14:27:44.000000000","tz":-240},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2019-08-23 14:27:44.000000000","tz":-240},"subject":"DNM  Test review to allow vexxhost qualification to  complete","message":"DNM  Test review to allow vexxhost qualification to  complete\n\nChange-Id: I564f7fc7c30f292e7f81f935032d12e7ee2b0e0c\nRelated-Bug: #1840616\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/aed157a02563f8c72c969733090d0a54123939c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/aed157a02563f8c72c969733090d0a54123939c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
