)]}'
{"id":"openstack%2Ftripleo-quickstart~687460","triplet_id":"openstack%2Ftripleo-quickstart~master~Id56735b698fa70626847806db793d51458c36b4b","project":"openstack/tripleo-quickstart","branch":"master","topic":"os_tempest","hashtags":[],"change_id":"Id56735b698fa70626847806db793d51458c36b4b","subject":"Move fs020 to os_tempest from train onwards","status":"MERGED","created":"2019-10-09 06:39:25.000000000","updated":"2019-10-30 16:32:37.000000000","submitted":"2019-10-30 16:31:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687460-1572453089854-b8c1e99e","meta_rev_id":"35e48a30e16a2b443f2c351a59d5f26292eed20e","_number":687460,"virtual_id_number":687460,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-30 16:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-10-30 10:36:04.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2019-10-30 05:31:13.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2019-10-30 09:14:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-10-30 13:48:12.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-30 05:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":1,"date":"2019-10-30 13:48:54.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 06:40:34.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2019-10-21 07:19:28.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2019-10-29 09:04:07.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2019-10-30 09:14:53.000000000","updated_by":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"reviewer":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"state":"REVIEWER"},{"updated":"2019-10-30 10:36:04.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-30 13:48:54.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-10-30 16:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"695c88e20ed524accef9e467a6844cf39366ce93","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-09 06:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f0928ea34f3e83c635ad549e385094bfbdf06ad","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-09 06:45:56.000000000","message":"Patch Set 1:\n\nTesting here: https://review.rdoproject.org/r/#/c/23034/","accounts_in_message":[],"_revision_number":1},{"id":"3414cffe8667227fe3389b20b0ffd08682602fd1","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-09 08:55:51.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nWould be nice to fix this nit, but i\u0027m not blocking...","accounts_in_message":[],"_revision_number":1},{"id":"a1ead0029b23b2b38b99db5db6643b7d406b2230","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-09 09:40:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4ec532f/ : SUCCESS in 2h 47m 49s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/e947931/ : FAILURE in 1h 42m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/a355259/ : SUCCESS in 2h 15m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/860c26b/ : SUCCESS in 2h 59m 24s","accounts_in_message":[],"_revision_number":1},{"id":"764b6ed06cffcc2517b5fe14442551e3caf4a424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 10:14:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ebf37e10982d43f4be9863f693d7273b : SUCCESS in 2h 39m 38s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/3028f541069947c9b8810e6cf2a8d390 : FAILURE in 2h 28m 25s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ff7fb50246014ea782e2f7c7c0185bca : TIMED_OUT in 3h 30m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c198b23e27524070aabde11d2a1d326f : SUCCESS in 2h 46m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/98d4c20ce79344319618e6acef96e527 : SUCCESS in 2h 58m 49s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6f9c3d7109574666bcb1ed718a5d30db : FAILURE in 2h 21m 16s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/e1346f0d1556419c830f61c625050d44 : TIMED_OUT in 3h 29m 50s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/94c8eddd250e43d58ee384b5cc1cd00f : SUCCESS in 2h 36m 17s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/ddf6e02eb46047fab87210f8d6c36718 : SUCCESS in 2h 47m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41657d8d8d1a465cb02633dba4d0c209 : SUCCESS in 27m 05s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5f43867bde144ae194e4fc73c1d1c941 : SUCCESS in 11m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/606bcccc9ef142c1bb6647fbcf0940cc : SUCCESS in 11m 40s","accounts_in_message":[],"_revision_number":1},{"id":"bf5d787b5bbed4be4fc2cdccb92b5a6b675d4b8b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-09 10:42:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a377a2b76cee37ca47e148538e62d6239f60e059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 13:35:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/826853278a7e4bed95bfdda82abb1aaf : SUCCESS in 1h 56m 19s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/057aea4e07dd4ac48ab2ed6111265f6f : FAILURE in 2h 04m 38s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f292f5663e664d849780b3b430275e4d : SUCCESS in 2h 44m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c17fd0733fdb4418968fa6723143051c : SUCCESS in 2h 27m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/c39e5795a5ac4234b545baecb58523cc : SUCCESS in 2h 24m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/e7ff53c8bb7740c2be13e761e817d147 : FAILURE in 1h 46m 50s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/8842d45485214441b7c005fa6b1cda23 : SUCCESS in 2h 25m 03s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/591e3f3b916d4a03a7ef05c4836ea55f : SUCCESS in 2h 47m 45s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/e3f7766a6b4f4ed4b45553339aa141d4 : SUCCESS in 2h 27m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad52c99319d348478a7bb9d28d9121dc : SUCCESS in 4m 30s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fc74522cfd694dd6a243356d959e6ce8 : SUCCESS in 12m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae45e087a2c242e7bd9087f152d79240 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"6a7b1e1bf1714f562c3a37bf0e674cff50f26592","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-09 14:28:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/a024edc/ : SUCCESS in 2h 54m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/ba2f84b/ : FAILURE in 1h 38m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/83a1eb0/ : FAILURE in 2h 46m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b72bff3/ : FAILURE in 3h 44m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a2d35ae01e3b3d0537cb8891b4f80bc578d76fc2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-10 05:18:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8e7c5b13ae1e933a1dda3c19e90c596b415ccfc6","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-10 07:10:55.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\ntimeout at https://review.rdoproject.org/r/#/c/23034/ looks like is there something else needed","accounts_in_message":[],"_revision_number":3},{"id":"1cf648488a87b0690fdf488da3f684df4f8f0b35","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-10 08:26:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/cb05ba6/ : FAILURE in 1h 51m 30s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/d3b30af/ : FAILURE in 1h 43m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/f847d49/ : SUCCESS in 2h 19m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/6b2acac/ : SUCCESS in 3h 06m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3840780d041725aa7b11dc424a15b192cfe0c119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 08:38:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/8fe7835d21af47549b1cf845500df4dc : SUCCESS in 2h 04m 43s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d50dcbae99654688aa74596efeb97287 : FAILURE in 2h 07m 09s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/f842b88204f14a059055a46b40e91d9b : SUCCESS in 3h 14m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6b1d0e42d2ca4886a761a159c160419f : SUCCESS in 2h 26m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/bc76c60049374f729d152397bd815411 : SUCCESS in 2h 46m 36s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f2074fc0fe854020a8b17aed8812c7e6 : FAILURE in 2h 11m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/d41c1e2403af47d3adcf4a5448b80b87 : SUCCESS in 2h 42m 00s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/917e32fa857d44ad92c4fb314f16eeda : SUCCESS in 2h 50m 31s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/5ae29ad88528416a8a87934253cec4fa : SUCCESS in 2h 00m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e390beda5264ed4823b26e96010a155 : SUCCESS in 4m 09s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e0b7a2ed6b8641f5b65119246df1734e : SUCCESS in 11m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/98d79ddbb31449caae760ccbf872852b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":3},{"id":"1fe3239a076e0244bac0efbd5b40acdc13621e1c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-16 06:34:31.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"67b1e17ad890113d9d4c266b3e522504aae5c739","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-16 09:19:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/5739858/ : SUCCESS in 2h 42m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/2a92a9e/ : FAILURE in 1h 58m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/052ccfd/ : SUCCESS in 2h 43m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/b1b9a09/ : SUCCESS in 2h 43m 42s","accounts_in_message":[],"_revision_number":4},{"id":"3c3c4fe2ab61a02420806cdc02686eafaed01075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:18:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/ac1484ca5afb4cbcab1729b0c24d5e96 : SUCCESS in 2h 07m 59s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/71485f56c20d4f31b2f4763ad15ee4e8 : FAILURE in 2h 14m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/7c6d95f28b2949d4a6fad61f10e9bbc9 : SUCCESS in 3h 11m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a8a3735b142d4953af81bd3f879500d4 : SUCCESS in 2h 50m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6dd047b9917b474db34e0ada07e5ed61 : SUCCESS in 2h 24m 16s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/a539c4a4b4c844e3b1e66f13e2078aa9 : FAILURE in 1h 52m 06s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/7fab75cdde364bea8621ad64af6506ca : SUCCESS in 2h 53m 33s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/bc413a997eba4385924eff69b7afa40e : SUCCESS in 2h 31m 37s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a20c2c1e9a684d32989a5a95b51106b8 : SUCCESS in 2h 57m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7d8ad057c446d6a6fb8f4e2f81243d : SUCCESS in 8m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/cc1d0e6eefdd48469fe133fe08a48edf : SUCCESS in 18m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00942c94d25948529f2fe3a8b14451bf : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":4},{"id":"c36878cb9d2dcec16e1aa8a775d7c894df20417d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-17 09:17:23.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"5ff4b428f1bd1031ccc3ea2ed6f1fb2b5d8f4bc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 12:53:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/af2c6a1d4a5f4eabaff1c74613dab6bb : SUCCESS in 2h 21m 26s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/7a633bd26e6f469a86c8d240cb5eb902 : FAILURE in 2h 08m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/30428db3bc9549df9ed81dcae1d88d77 : FAILURE in 3h 11m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/75dbef40c0c34506b3363e101f65f0a6 : SUCCESS in 3h 20m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/13313513c1a74d79b162d048b29b6c25 : TIMED_OUT in 3h 28m 51s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/cc55551d510b4adcbbab17caac86fae1 : FAILURE in 2h 17m 31s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/2c12effacfeb42bab030d9aae90364b5 : FAILURE in 2h 57m 40s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/cc7541d86eb142b19520d79448baed70 : TIMED_OUT in 3h 34m 26s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/c51dc93adb8248f59c25a6c50f0d0326 : FAILURE in 2h 43m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def4579492784504a5c6ec92c32b6a39 : SUCCESS in 4m 22s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5d17849561ab4931a052a5d41a7c6d53 : SUCCESS in 11m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/395c7cc2fe6c441ea34f061ee72d6cf8 : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":5},{"id":"c205b783bf602939bbe1cf2c4146a1e19f669a6c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-18 01:43:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/17abe3b/ : FAILURE in 2h 31m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/b83d765/ : FAILURE in 1h 50m 25s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/7cbf696/ : SUCCESS in 2h 34m 23s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1dda4dc/ : SUCCESS in 2h 49m 38s","accounts_in_message":[],"_revision_number":5},{"id":"f8e951193b4e12a180d8f1de1444cbb2be2d6c2c","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-10-21 07:19:28.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck\n\n\nnow looks green there http://logs.rdoproject.org/34/23034/5/check/periodic-tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-master/2516d0c/","accounts_in_message":[],"_revision_number":5},{"id":"11e0bb474fd264c74b6f16dfaebdae0d195e2e0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 09:59:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f1362ee93680476b93962a9d88eb2806 : SUCCESS in 1h 24m 14s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/fa51bf36502a43ad9b92f944bad673e7 : SUCCESS in 1h 31m 23s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/33c7c1dff828413ebc6cedc08446676e : FAILURE in 2h 25m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/66784adaf5b04df0b824ea07a60619cb : SUCCESS in 2h 27m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/4d5f1dc6763f4bc699852c4b069e912e : SUCCESS in 2h 22m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/796c08c47b8a47d58dd79f74415f4aa0 : FAILURE in 1h 44m 33s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/29ece48371fa4d2eb5649bc35eaccaff : SUCCESS in 2h 36m 59s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/d2b777969a06425dbdb054894118ee5e : SUCCESS in 2h 28m 05s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/bbb8141fd0d848059f9057af35da6900 : SUCCESS in 2h 17m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430c2240dd9243ed99969e642ad1a6c8 : SUCCESS in 4m 54s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2baed94c85be4b97907cd63b97a164db : SUCCESS in 11m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3c8a74e295e14c4787a308990520acb8 : FAILURE in 4m 26s","accounts_in_message":[],"_revision_number":5},{"id":"750599edfbdcc06bdfb23e546a57a9f5b185ab3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-21 10:47:30.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/bd08829/ : FAILURE in 3h 26m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/1207b39/ : FAILURE in 1h 36m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/d3a2acd/ : SUCCESS in 2h 15m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/62fa838/ : SUCCESS in 3h 20m 37s","accounts_in_message":[],"_revision_number":5},{"id":"57b87fdb1e172af74fb50a9300400fe2440e92a8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-24 09:24:04.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"5fbf62a11aef7f26aec00ccbc06b8c6da1762ff9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-24 12:52:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/643c7ec/ : FAILURE in 2h 54m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/c2e0c30/ : FAILURE in 1h 45m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/47d20c3/ : SUCCESS in 2h 19m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/e1147b3/ : SUCCESS in 3h 26m 02s","accounts_in_message":[],"_revision_number":6},{"id":"fdd6f24008861f400ca5c26f4b5512424af2b567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:25:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/b3eea5be4dcc46d5b1c54d10e2c2eaea : SUCCESS in 2h 08m 42s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/12bb11f4b6f546318fd2b2333c06067a : FAILURE in 2h 04m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/20ad5731fa2d439cb7a0169c53f36e1a : FAILURE in 2h 43m 15s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/da548f8991c143bd92b1c9946f25627e : SUCCESS in 2h 32m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a4e2e08fab4e4e258d9c2b7e5cc7966d : SUCCESS in 2h 32m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b1ef119769314b44b6a5f0ca7194fc84 : FAILURE in 1h 38m 39s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/258cb2017c6f46e3927411ec44f996c7 : SUCCESS in 2h 58m 52s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/bfcae587f63e4f1eafdb895153412e25 : SUCCESS in 3h 05m 37s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/baab2bcf4b724b7b83f7143033991636 : SUCCESS in 2h 44m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a5b4271be8427b955a3456bf4d5cc6 : SUCCESS in 6m 18s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6ff214496f15426489a197fe57da1626 : SUCCESS in 12m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/19ca8d50b87649c49eee62dd4be8d26b : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":6},{"id":"6ff667d0491156709a5645050a8d0ecd0a77e992","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2019-10-29 09:04:07.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"f3f14cc3fe010e81f062a86ed010f31f755adebc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-10-30 05:31:13.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"b14f72a45de0c9da9f68a2c460feb3e15dba7fe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 08:44:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/93b53c7a8dc842a28554c16f0ee78dde : SUCCESS in 1h 30m 47s\n- tripleo-ci-centos-7-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/4711b8ffd9ee45c5a449c150453ec1a6 : FAILURE in 2h 02m 28s (non-voting)\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/49274c3523634376ab969fe85447ea64 : SUCCESS in 3h 12m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/908c706f13184f1d8931efc1f5df918a : SUCCESS in 2h 14m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6cc4425619ad41aca5be38a646676dc6 : SUCCESS in 2h 24m 13s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade https://zuul.opendev.org/t/openstack/build/fa8899d1118f40ef9ff53fd9fb5d37ca : FAILURE in 1h 36m 48s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/45bb11533436434990159437c48284d4 : SUCCESS in 2h 29m 42s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/e1f2995151e640c3ac908ad398de65f3 : SUCCESS in 2h 13m 12s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/201bfe3685e74337b8961a474cb7b22d : SUCCESS in 2h 12m 41s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/57dbe0290dee4f4c86a63f7a63ff5e17 : SUCCESS in 2h 08m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3876187277de41849430aec508ca834c : SUCCESS in 4m 42s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d31ba400d1594396b7020af3071df0e5 : SUCCESS in 17m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0fa6371722384db0909c491b85f6c3a4 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":7},{"id":"3cf2e08bff3b59b045f765cafbcb4024a79c2c9b","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-10-30 09:14:53.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"620768d23325f9e2e2c68c9fdb960d5e3c063f97","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-30 10:36:04.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/60/687460/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4cc6d5b/ : TIMED_OUT in 5h 01m 39s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch http://logs.rdoproject.org/60/687460/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch/7a50fd0/ : FAILURE in 3h 13m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch http://logs.rdoproject.org/60/687460/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch/9634f1a/ : SUCCESS in 2h 26m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch http://logs.rdoproject.org/60/687460/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch/1b1f9db/ : SUCCESS in 2h 46m 55s","accounts_in_message":[],"_revision_number":7},{"id":"c8556fedaef33d634e03d969c44a1b4e3d0931f8","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-10-30 13:48:12.000000000","message":"Patch Set 7: Code-Review+2\n\nhttp://logs.rdoproject.org/34/23034/5/check/periodic-tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-master/2516d0c/job-output.txt shows fs020 using os_tempest","accounts_in_message":[],"_revision_number":7},{"id":"d100686510d6e672d9dd7d9b58cf514c497cd3ab","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-10-30 13:48:54.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"eae7b15c2a932f1959059b9c1b8219af2d4d70f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 13:49:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"7f65434412ca1cfda31598cc59554d57f993ce02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 16:31:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/36f8848092ba4dea8a5c05b5ae7cedba : SUCCESS in 1h 21m 21s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/78459b40616e45ce82f7c58fdcdb9649 : SUCCESS in 2h 32m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f4b915832c124b598ade2486faad5388 : SUCCESS in 2h 02m 50s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/b1e72b4548f14c40ac14bca1b90b71e5 : SUCCESS in 2h 16m 25s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/0eeaffa62e5445ecbfa17dc1049b8e82 : SUCCESS in 2h 31m 02s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/0d24880671044702a33a98284791e977 : SUCCESS in 2h 35m 17s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/3212aab758b24c6cb2b9390039a32763 : SUCCESS in 2h 24m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9d6d929dc5412aac90048c30f70cd8 : SUCCESS in 3m 58s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0e1f93f3ec4e47d8884ebd4395e4843c : SUCCESS in 11m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2f3facd09b764dbbadfbcd320566e736 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":7},{"id":"77e8df32e5a852d7519b1e06a2d73be7906c2a5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 16:31:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"a7be01ce1e4940047346c05951048e93e0841855","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-30 16:32:37.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969f55fda50b4078a9b00e64bdf13bdf : SUCCESS in 49s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"7c7713c3d789b800d65d804647d4b972db0332a9","revisions":{"0a2a67bb646b5c146b2db327f2954c71c540ecdb":{"kind":"REWORK","_number":1,"created":"2019-10-09 06:39:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/687460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/60/687460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/1"}}},"commit":{"parents":[{"commit":"a40813e388bdfb2a84742a7693a87ba2989e1e4e","subject":"Adapt os_tempest in FS001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a40813e388bdfb2a84742a7693a87ba2989e1e4e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:38:47.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:39:22.000000000","tz":330},"subject":"Move fs020 to os_tempest from train onwards","message":"Move fs020 to os_tempest from train onwards\n\nChange-Id: Id56735b698fa70626847806db793d51458c36b4b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0a2a67bb646b5c146b2db327f2954c71c540ecdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0a2a67bb646b5c146b2db327f2954c71c540ecdb"}]},"branch":"refs/heads/master"},"544f678b46ab161d814a0c14c83903fa4d79a27c":{"kind":"REWORK","_number":2,"created":"2019-10-09 10:42:47.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/687460/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/60/687460/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/2"}}},"commit":{"parents":[{"commit":"a40813e388bdfb2a84742a7693a87ba2989e1e4e","subject":"Adapt os_tempest in FS001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a40813e388bdfb2a84742a7693a87ba2989e1e4e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:38:47.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 10:42:34.000000000","tz":330},"subject":"Move fs020 to os_tempest from train onwards","message":"Move fs020 to os_tempest from train onwards\n\nChange-Id: Id56735b698fa70626847806db793d51458c36b4b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/544f678b46ab161d814a0c14c83903fa4d79a27c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/544f678b46ab161d814a0c14c83903fa4d79a27c"}]},"branch":"refs/heads/master"},"56ac0e762ea7c602bf18488cdea6ebc8b8d3681c":{"kind":"REWORK","_number":3,"created":"2019-10-10 05:18:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/687460/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/60/687460/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/3"}}},"commit":{"parents":[{"commit":"a40813e388bdfb2a84742a7693a87ba2989e1e4e","subject":"Adapt os_tempest in FS001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a40813e388bdfb2a84742a7693a87ba2989e1e4e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:38:47.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-10 05:18:01.000000000","tz":330},"subject":"Move fs020 to os_tempest from train onwards","message":"Move fs020 to os_tempest from train onwards\n\nChange-Id: Id56735b698fa70626847806db793d51458c36b4b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/56ac0e762ea7c602bf18488cdea6ebc8b8d3681c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/56ac0e762ea7c602bf18488cdea6ebc8b8d3681c"}]},"branch":"refs/heads/master"},"e1172bd5c8bc4920b67a0c3653f1ea45fef65d12":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-10-16 06:34:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/687460/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/60/687460/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/4"}}},"commit":{"parents":[{"commit":"a40813e388bdfb2a84742a7693a87ba2989e1e4e","subject":"Adapt os_tempest in FS001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a40813e388bdfb2a84742a7693a87ba2989e1e4e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:38:47.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-16 06:32:59.000000000","tz":330},"subject":"Move fs020 to os_tempest from train onwards","message":"Move fs020 to os_tempest from train onwards\n\nIt is getting tested here:\nhttps://review.rdoproject.org/r/#/c/23034/\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1330\n\nChange-Id: Id56735b698fa70626847806db793d51458c36b4b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e1172bd5c8bc4920b67a0c3653f1ea45fef65d12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e1172bd5c8bc4920b67a0c3653f1ea45fef65d12"}]},"branch":"refs/heads/master"},"671ed8b6208e57164b1d83b270ae51d37f0a5306":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-10-17 09:17:23.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/687460/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/60/687460/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/5"}}},"commit":{"parents":[{"commit":"a40813e388bdfb2a84742a7693a87ba2989e1e4e","subject":"Adapt os_tempest in FS001","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a40813e388bdfb2a84742a7693a87ba2989e1e4e"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:38:47.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-17 09:17:19.000000000","tz":0},"subject":"Move fs020 to os_tempest from train onwards","message":"Move fs020 to os_tempest from train onwards\n\nIt is getting tested here:\nhttps://review.rdoproject.org/r/#/c/23034/\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1330\n\nDepends-On: https://review.opendev.org/#/c/689106/\n\nChange-Id: Id56735b698fa70626847806db793d51458c36b4b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/671ed8b6208e57164b1d83b270ae51d37f0a5306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/671ed8b6208e57164b1d83b270ae51d37f0a5306"}]},"branch":"refs/heads/master"},"ba2590d35438dfd8a694cacb0f7777b53ba0d68c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-10-24 09:24:04.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/687460/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/60/687460/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/6"}}},"commit":{"parents":[{"commit":"f87828ce9cfbc2d5b342c72833ca77df218e0971","subject":"increase deploy timeout in ci.centos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f87828ce9cfbc2d5b342c72833ca77df218e0971"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:38:47.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-24 09:24:04.000000000","tz":0},"subject":"Move fs020 to os_tempest from train onwards","message":"Move fs020 to os_tempest from train onwards\n\nIt is getting tested here:\nhttps://review.rdoproject.org/r/#/c/23034/\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1330\n\nDepends-On: https://review.opendev.org/#/c/689106/\n\nChange-Id: Id56735b698fa70626847806db793d51458c36b4b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ba2590d35438dfd8a694cacb0f7777b53ba0d68c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/ba2590d35438dfd8a694cacb0f7777b53ba0d68c"}]},"branch":"refs/heads/master"},"7c7713c3d789b800d65d804647d4b972db0332a9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-10-30 05:31:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/60/687460/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/60/687460/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/60/687460/7"}}},"commit":{"parents":[{"commit":"c673a295f9a84efe91b3137b807a8229cd8370b8","subject":"Set the amphora image to stein branch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c673a295f9a84efe91b3137b807a8229cd8370b8"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-09 06:38:47.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-10-30 05:31:13.000000000","tz":0},"subject":"Move fs020 to os_tempest from train onwards","message":"Move fs020 to os_tempest from train onwards\n\nIt is getting tested here:\nhttps://review.rdoproject.org/r/#/c/23034/\n\nhttps://tree.taiga.io/project/tripleo-ci-board/task/1330\n\nDepends-On: https://review.opendev.org/#/c/689106/\n\nChange-Id: Id56735b698fa70626847806db793d51458c36b4b\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7c7713c3d789b800d65d804647d4b972db0332a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7c7713c3d789b800d65d804647d4b972db0332a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
