)]}'
{"id":"openstack%2Ftripleo-quickstart~726752","triplet_id":"openstack%2Ftripleo-quickstart~master~I6ae79a276d3cd2af929646455d17ce46fede770a","project":"openstack/tripleo-quickstart","branch":"master","topic":"octavia_ostempest","hashtags":[],"change_id":"I6ae79a276d3cd2af929646455d17ce46fede770a","subject":"Move fs062 to os_tempest","status":"MERGED","created":"2020-05-11 08:26:41.000000000","updated":"2020-05-17 19:27:38.000000000","submitted":"2020-05-17 19:26:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726752-1589743586907-e2a209be","meta_rev_id":"c309bfdcd3d3b11a8e0e6ee89bb41035346eaf02","_number":726752,"virtual_id_number":726752,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"date":"2020-05-17 13:54:33.000000000","_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-17 19:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-16 13:05:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"recommended":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2020-05-16 10:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":1,"date":"2020-05-16 10:14:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2020-05-16 10:14:28.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"value":0,"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-05-16 10:14:33.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 08:27:29.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2020-05-11 08:27:42.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2020-05-11 08:28:06.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-05-11 08:31:24.000000000","updated_by":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"reviewer":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"state":"REVIEWER"},{"updated":"2020-05-14 21:53:18.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-05-15 01:43:09.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-05-15 05:41:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-05-16 13:05:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-17 13:54:33.000000000","updated_by":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"reviewer":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"state":"REVIEWER"},{"updated":"2020-05-17 19:26:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2f8586586a8f9ecf4fc1e754f233b20eeaeec02","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-11 08:26:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4635c31760799a8bd34db27db8b22f4a743c1019","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-11 08:27:48.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"f1ed276847fa3b037d1b871140079edd15ce4260","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-05-11 08:31:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6db36dff4c2af3d5934ca46001da6fc5dacbc75","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 11:11:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1099e6e8f3434c5d94af3f42861b1600 : SUCCESS in 2h 34m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/47eb453c88324ff48c8eab1cded2eca8 : SUCCESS in 2h 04m 17s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/3073687c13c840708e175d1010862741 : SUCCESS in 2h 19m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/095b57ded9434c248cd29148a372e272 : SUCCESS in 2h 37m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/dce46a9ab79b418f8a4363d658317af4 : SUCCESS in 2h 42m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1770367283827d2bad16063cb15e71d6cfdcb33c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 11:21:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1513b362cdfa4f4eaf0588ca9878f32a : SUCCESS in 1h 30m 48s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b0460b6aceeb4733ad5ac99a4a7c4920 : SUCCESS in 2h 24m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/eff84be58c92440d92a9768fddadba9c : SUCCESS in 2h 15m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4b0ef8e852f743af8a41426d286ef3f1 : SUCCESS in 2h 10m 22s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/599b9fc51183448a90041cf1c515b5fd : SUCCESS in 2h 38m 45s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/15934811509a4e84a165c10ce9c03729 : SUCCESS in 2h 10m 47s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c5b4d238556248d6a18de92892f1f3b3 : SUCCESS in 2h 39m 11s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a1dd768944a54c239770169e78fcd65b : SUCCESS in 2h 17m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a7d9a7f2b1746b4b739fe902eea91c8 : SUCCESS in 5m 56s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a537d0de9454449b80d89423f1696fad : FAILURE in 1h 51m 39s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/6d7fc3c5b5bd420da0afec26aaffdb40 : FAILURE in 1h 28m 50s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e4c024b1d1b64626a88dce3faf4b8b1c : SUCCESS in 13m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6b2e9a84e754fd587b7096b1bed7802 : SUCCESS in 9m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9e77a4a34832ca1cf1f6bff688e3a76e2d538bf1","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-11 11:27:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73f6273572987370daf6e167dc6338cfec9de10d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-11 14:46:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b33c82945d154d5a9032e249b3bfc7aa : SUCCESS in 2h 41m 19s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/805cbc112feb4ef4834bfecf0e3c7cab : SUCCESS in 1h 59m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/7630f151c7904ca2a11401ab540acbf4 : SUCCESS in 2h 23m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/6a23c3eb296e4f27922c4025bf7d99c6 : SUCCESS in 2h 35m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/2f3bd07166e8477b8486090ea2f6c197 : SUCCESS in 2h 53m 37s","accounts_in_message":[],"_revision_number":2},{"id":"199676ad4e77116fec06d600527b13abd565ea42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 19:08:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ab5d80a2f2364e5c873cab1bfed9f505 : SUCCESS in 1h 30m 13s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/720255f249ff412bbe74fcfbe70df731 : SUCCESS in 2h 18m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/eed7633d08d544ffa27e5073e567c9c4 : SUCCESS in 2h 00m 41s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a296e89d2a1f4b2db3649b1f44c1a574 : SUCCESS in 1h 46m 41s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/71f374b57e1f43b5bf12f4b50c8ba33c : SUCCESS in 2h 58m 44s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/153887fb1b7841f2bed0609d06ee8217 : SUCCESS in 2h 28m 43s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/937269f2b56d4a8390c1a363bc9d3f99 : SUCCESS in 3h 08m 34s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/147b9d17491e424183579bf915dbecf5 : SUCCESS in 2h 10m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9507c9550c244628b098ad286632276 : SUCCESS in 5m 47s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/9bd4552622ab4315ba0c4c1862622d46 : FAILURE in 1h 24m 45s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/4d728389f41944b4a2b55d8082694030 : FAILURE in 1h 22m 54s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/287849b490c14f7abdddeb1c1a203361 : SUCCESS in 13m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fe44c892481045eba1e85ccbd61202f3 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"47ba1c6a95e2d358b70f668e641ff03bec68be37","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-12 08:05:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5efe48a82c76b547ba0a535ad75517cb3b4f3045","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-12 11:35:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b703d8f175c84cb68d8bf1803990fdf2 : SUCCESS in 2h 41m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/11fff597902347ce881087b846159684 : SUCCESS in 2h 03m 42s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/3276a710e3314673ae5492c912efa1af : SUCCESS in 2h 16m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/5138387a088e4d6f8d66fa4deb0df753 : SUCCESS in 2h 37m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/98cf3e8e7d154d85a65a4ddcd7aee7a5 : SUCCESS in 2h 37m 41s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/534a5841a6384014af2aec2b36ccab8f : SUCCESS in 3h 01m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c01608a07a2e41504cd7c677b979979448a4a479","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 14:26:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/509238009fd74ad79841638414d009ce : SUCCESS in 1h 27m 35s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b41c57cd952e4438a083376cb2c3776d : SUCCESS in 2h 12m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/52f8596cf84044ce80231f961532fa62 : SUCCESS in 2h 00m 31s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/89ed2b3ead50485d9f5fbb039e9f4000 : SUCCESS in 2h 01m 08s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/8b6a242a413f4293a3a54db1b34cb373 : SUCCESS in 2h 56m 00s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/2ad397a8b3cc4946bf5d58d47233ae87 : SUCCESS in 2h 18m 05s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/9575ac2fb75c42db833eae34c7dd1e4a : SUCCESS in 2h 38m 56s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/dd3307265e8a432881d0c942bbf75f31 : SUCCESS in 1h 53m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12602589050a4755a52b98c53753ac01 : SUCCESS in 4m 39s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a539bb41ac324facb6719c57b383ac27 : FAILURE in 1h 27m 50s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b1491881cd7948da8f874109af5cfd51 : FAILURE in 1h 15m 03s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/055ed92defd345ff9bc797ea96901790 : SUCCESS in 12m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/226f7c489566454d84fb9748e903f164 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b0c86e5ca4df9916ce105b1231f100a84e5b3850","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-13 08:13:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"01a6540d4de9f371e63efa9f29ff1fc8ba9be9dc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-13 12:08:16.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e0b0b4c05b3e30fcefd643d46e61340537dfbf1c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-13 16:05:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1ccd0b6df5e643129b9925c8593f5c23 : SUCCESS in 2h 44m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/542df133a02a4d9496cec77bb4a972fc : SUCCESS in 2h 02m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/dae3b86ffbec48e38618a5a40c8d8cb7 : SUCCESS in 2h 30m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/1509d32f855d439b8baecaa8b336d66b : SUCCESS in 2h 42m 20s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/99d20b2cbcdc476abc8f0ef3d9878a1c : SUCCESS in 2h 38m 58s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c5366b55ed444eacb746a8ae37815f20 : FAILURE in 1h 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"62288e02ffb85c62fd425425a86de4c03aeffc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:07:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/efc8fe7fdd16434d854f0d78bd4269b2 : SUCCESS in 1h 29m 01s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ad1379bb01c2456d8a6a34f40489af3e : SUCCESS in 2h 12m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f145c0d1cfb4474d9c606b2a1642004b : SUCCESS in 1h 57m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/075eca9e09064e90a85aff8bbffcd054 : SUCCESS in 1h 51m 05s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3c88499e39c2407db450229bc898d19f : SUCCESS in 2h 58m 22s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/7bcfe63f7b514aa4840953bf70e00ef0 : SUCCESS in 2h 41m 47s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/2d712c10542f44e3b5a5cee0f9030235 : SUCCESS in 2h 34m 24s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/b49349ee05c44b02b3bc198adda6e9dc : SUCCESS in 2h 00m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e2d7ad633274a479c0b5cc5413852d7 : SUCCESS in 5m 18s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/a8c5fcf97cdd46d4b22cf2875f9e1de3 : SUCCESS in 1h 40m 27s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/4783f1d655d34b9589ba1c81da1dcd33 : SUCCESS in 1h 28m 58s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1ac46bbe7cae4387a31c811f9aff5dbe : SUCCESS in 12m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6cfa1772b3348f2abbc6eeb7d221fb4 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":5},{"id":"7666c87fe8f4fb97f8f1ec3bdfbf1da4cff0edd3","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-13 18:36:20.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"679171cc93f7dc72063d191b77f202dec670d70a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-13 22:24:45.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ee9beb0c01394590946c29853152d3b5 : SUCCESS in 2h 42m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/41739973a8e94de1a9081d0613b12e67 : SUCCESS in 2h 03m 34s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/9c089d0cb0f440c9ad92b05e06f1e712 : SUCCESS in 2h 31m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/0ea0061f830146c2bbac3d446691ae75 : SUCCESS in 2h 51m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/5bbe4c496ab743279842afeebcddbc27 : SUCCESS in 2h 46m 30s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/f7ea5ba947614b5a9aad850ca4a1ab83 : FAILURE in 1h 20m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"532c862b0cbbcfb17da545a2717d3f3889d2bd1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 22:33:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d9e7f183b5cc46f9ac85bd97872b7ce8 : SUCCESS in 1h 12m 30s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f6dbea13690246e685121e760989da00 : SUCCESS in 2h 08m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b8417b0bfe04b58b28e9c8dbfb06fe6 : SUCCESS in 2h 08m 27s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2146b343e95a48d5ba14d8605c93d15e : SUCCESS in 1h 35m 32s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/8c84960ee5a14afea98d511407d6d467 : SUCCESS in 2h 42m 22s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/b26ef81f0f764fcd8ca49dcda9952b58 : SUCCESS in 3h 02m 07s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/99452efd842c471893d20bdb9428ce3b : SUCCESS in 2h 29m 57s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/6ea0d1b995bc4233be5a1da6cf7f6473 : SUCCESS in 2h 06m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f1b592cf8a541af96918023fd421150 : SUCCESS in 5m 12s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/e8d4150f75834094b1a0bc51917f778f : FAILURE in 2h 22m 06s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e5d9bee1805c4de3933283afbd432aeb : SUCCESS in 1h 48m 49s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/01d25a6993a64105937511576d2fa75e : SUCCESS in 13m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f77e736e271b47a28ce5319b1a45e067 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":6},{"id":"7e9f84fddae97221db61225b08341b8dd68f9fd9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-14 07:29:27.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"30b399710355d26769a22d4b624a7f95ecf09c1c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-14 10:20:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6286f3a4553441e1b45fb1b4f5fa2001 : SUCCESS in 2h 37m 11s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/a7af6997239e4755b541c1d39a8e0fcc : SUCCESS in 2h 03m 16s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/9752dd36083e4bd0b52cc82cdd07d60d : SUCCESS in 2h 14m 51s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/433029340f1042cda4116dcd7c553c4d : SUCCESS in 2h 37m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/6a5866cd9b3a442689fc6b8957ee8719 : SUCCESS in 2h 42m 35s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/7c92f64d43fa485eb62cbc156892c473 : SUCCESS in 2h 50m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"70ec1e4240ccc2baead057d476c01b661e65211a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 10:32:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/dd29cc9bab514eddb826e3c6aaf4a7fc : SUCCESS in 2h 09m 11s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/446c3444f9e74260bd939b2aae8890d3 : SUCCESS in 2h 47m 28s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1631147222714793b0dfa0409d6d4f3b : SUCCESS in 2h 24m 21s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1d7992c2188d4801a13ce0ef0d6bcac1 : SUCCESS in 1h 53m 56s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ba5f7069cb444f20a175108a710596cc : SUCCESS in 3h 01m 03s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/4b43f2a0159040ae81fc264f0debb771 : SUCCESS in 2h 09m 03s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/ece2003d8f1b4e53a57febb0d2e6e09e : SUCCESS in 2h 41m 41s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/a1c125c9b66945ba8205289c839d5a93 : SUCCESS in 2h 08m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc08195f353f487da52fbc21bf08ad96 : SUCCESS in 4m 41s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/439fa652176040a6a5c6b526efbaa73c : SUCCESS in 1h 24m 04s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e69fd3a4f98847d98a801bda42915a7e : SUCCESS in 1h 32m 50s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c7ccd52f195940c38970993cdc16d610 : SUCCESS in 16m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00b089931f5c42ebb0b94b5965154627 : SUCCESS in 9m 14s","accounts_in_message":[],"_revision_number":7},{"id":"968e991a2af4a0b523f06a25fb0fc935349bb2af","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-14 10:35:06.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"59f925e8f1bc3123b5a15f99768eaf6f04cc66a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 13:08:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d6c7e4d36e53421d8b3468ca950ce2a1 : SUCCESS in 1h 28m 00s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fce5ee6a547c44fca9836c32f09f4dac : SUCCESS in 2h 18m 55s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ef66d4ef104d4ae9868a099d33a1bc2a : SUCCESS in 2h 06m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/efc37ce6e1f64c5cbf9c9c04d5dd1717 : SUCCESS in 1h 52m 02s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/58b04246deaa40f998b8c68963a9c364 : SUCCESS in 2h 27m 02s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/81a790535c774f37ae573dd15088d03e : SUCCESS in 2h 11m 44s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/c4b937c1e02d4e7d85051825f3163349 : SUCCESS in 2h 28m 46s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/16eb7c4458b34517851e6a20901d5797 : SUCCESS in 2h 11m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81ecc95bfda04a229261d8554974c974 : SUCCESS in 4m 43s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7081ef6dc88c459e9dc33fa94226ece8 : FAILURE in 1h 28m 05s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/b947e231bad54387bb6d9b06e2bea08b : FAILURE in 1h 33m 31s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/176b2c645adf4538ac207583c691fe2c : SUCCESS in 12m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ca169ed4ced8458b9c3ad5aeaff22b2a : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":8},{"id":"25a3790635b840185dc36ba2d1576d6a774455fd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-14 13:48:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0b97fee95fbe401a9b523d905a6b7254 : SUCCESS in 2h 41m 15s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/ab636bd036cd469b943ce21730737944 : SUCCESS in 2h 13m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/4b1402cbc3a34d8db91c80b9217d38be : SUCCESS in 2h 20m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/91f032eb8c834a209acc0af52ec83159 : SUCCESS in 2h 36m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/bcb13fefd2c64be5b1494d7b33990754 : SUCCESS in 2h 39m 27s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/3f196a6ee48c4e2d97b84c29ead32ad7 : SUCCESS in 2h 55m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"92ff33d2b1e93b5e1cdf30e2739877afcad418d6","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-14 13:58:22.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"18f5a9d76c62e9b277ab3cdda57feaa4d6d2d892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 17:41:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4a32272d4ced4d90aa21e8d195766fa2 : SUCCESS in 1h 26m 13s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/40c88b41ba8b4c5b8004aa41f6466a72 : SUCCESS in 2h 44m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8553fcb0ae9f43ad8eca41b039219e5e : SUCCESS in 2h 11m 04s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cffd74b4780f4945b8b330b8308ab54a : SUCCESS in 1h 36m 55s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a209aed406c9448fb63a70ff252ca137 : SUCCESS in 2h 21m 45s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/988be0f3ba4947139ad9c028d9fca76d : SUCCESS in 2h 17m 16s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/2552b27c532d493c84da1d8d83749a2e : SUCCESS in 2h 07m 37s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/62467ede8fac4044b2128fedd0185bff : SUCCESS in 2h 28m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/513b76e4ea184cd18b4457c278c4bfaa : SUCCESS in 5m 13s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/f94c5b3e791c4c34a344f0fcf098bde4 : SUCCESS in 1h 36m 27s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/e41433c865e64df2820b999ca7aa09ae : SUCCESS in 1h 44m 46s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f855ff6589764aa88441cb98ff67cc2f : SUCCESS in 13m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c85a3ccc612a4952952a82cda038e0f8 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":9},{"id":"44fc591303b73346d72ba98822fbd0443e8945e8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-14 18:32:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/004ccf0afa804dec8f23f015736bb962 : SUCCESS in 2h 39m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/4cb88c2969f641f19c0851c94d3da68d : SUCCESS in 2h 00m 00s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/4f6b4d0315274ea29cd596664a79f17e : SUCCESS in 2h 19m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/b9d1ee3b42834fd685e8f5e2f194acfb : SUCCESS in 2h 36m 38s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/4d9d29e5c8f54682999a8b18335b191e : SUCCESS in 2h 43m 42s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c1245d5de9b14a84949796eee5d121e9 : SUCCESS in 2h 55m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"78c571162d8e1d9a2691c3b9290265d5a452b2f5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-14 18:37:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1bbc79d22e69267ddecfe99ca9f5d14fa4e1b9fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 21:37:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7202c7b1c65242f8ac1d53d2c57f4625 : POST_FAILURE in 54m 11s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1284c757e4e348a8b0cdac06f80d4800 : SUCCESS in 2h 16m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ff2ca401177741129a9a3e71bb7b6142 : SUCCESS in 2h 13m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9b5ce2a3071d4133b617f1c4827d5cd6 : SUCCESS in 1h 43m 41s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/cee01016d65541fdae0a9ca611eb09a8 : SUCCESS in 2h 27m 20s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/7b9a67e000114178808f4929192ff984 : SUCCESS in 2h 47m 41s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/af930522743543e9bd64be89dec5b2ab : SUCCESS in 2h 29m 19s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/99583bec73ae453298143a246e4ae0ef : SUCCESS in 2h 39m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6750c651dbce4728ba645dfd2e739b02 : SUCCESS in 5m 25s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7db6ef63ecd9454684c040bc5c490cd8 : SUCCESS in 1h 37m 57s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/d0afe98d629f43bba79817d9f8525d13 : SUCCESS in 1h 34m 54s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/324850ab3611471f8213434bde9e457e : SUCCESS in 13m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a86ee7700fd44ab0b4aa63c1d3ae7f6f : SUCCESS in 8m 35s","accounts_in_message":[],"_revision_number":10},{"id":"b6a51072554d6b780f9cb5e94b0276d74e6005ae","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-14 21:48:01.000000000","message":"Patch Set 10: -Code-Review\n\nhttps://af0459d07c145bbcdd66-07d6f138b8d357dfe3a40986bb499ebe.ssl.cf5.rackcdn.com/726752/10/check/tripleo-ci-centos-8-scenario010-standalone/7db6ef6/logs/undercloud/var/log/tempest/tempest_run.log\n\nhttps://e6fa2e4699d062fc6c72-9355d865b880bd099576064727df95b2.ssl.cf1.rackcdn.com/726752/10/check/tripleo-ci-centos-8-scenario010-ovn-provider-standalone/d0afe98/logs/undercloud/var/log/tempest/tempest_run.log\n Both are passing","accounts_in_message":[],"_revision_number":10},{"id":"c245eb27f2bd3d0088491839c5402b10724a28ec","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-05-14 21:53:18.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c6a09b3d026eb07cd6a27881da4bc99c4178c094","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-15 00:21:09.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7e4f71a8578a4ac39950278cb1057f7c : SUCCESS in 2h 38m 21s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/01aa5d9cb4834205a62ed0383d1b38da : SUCCESS in 2h 00m 57s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/f65402b7e7de4c50b1e56a20c7e75d04 : SUCCESS in 2h 18m 46s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/8ac40574f45842868387e3b6b227b201 : SUCCESS in 2h 37m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/3b9fca67f86648e080770b4777c0a60b : SUCCESS in 2h 46m 05s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/102731743e1744839cb406f1cf85d085 : SUCCESS in 2h 36m 58s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"beb91dabd0ac22c04ebba72206c78a5c6146e537","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-05-15 01:43:09.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"e6cf6c110fefa9acf5805cd53db04ac85ee11ca5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 01:43:22.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"536521299f2b824574e3c0b90bd3b0d1a1c77f0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 05:02:05.000000000","message":"Patch Set 10: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"42f7f88bad4b18f5a68ebbab94a38461d72a53f8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-15 05:41:14.000000000","message":"Patch Set 10: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"6eb918e80774f995cd4c3d45a6f361dd6fee995a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 08:39:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f56e573f9d04ccf8c959db1e1f61955 : SUCCESS in 1h 25m 10s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c35ed4998e3a483e9ddd02ae250ddab7 : SUCCESS in 2h 20m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d2d4fb2667ad481b9640481b15d3b225 : SUCCESS in 2h 07m 38s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8dcb3b17b6c1474bbc35b4be2025652b : SUCCESS in 1h 48m 32s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a3da9f580a2742b9bdbe882697841779 : SUCCESS in 2h 53m 47s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/d7eb60af54aa4b6e9ff0951bf6712612 : SUCCESS in 2h 32m 39s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/7e33300263b445bd83104e0940ab9cf1 : SUCCESS in 2h 45m 38s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/f429c8ca3c304e71bbbc928876d4d439 : SUCCESS in 1h 50m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffb18539abc4735987e272eca33e0f1 : SUCCESS in 4m 42s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/7a0bf08502d34bb1b62a7c81763fef07 : SUCCESS in 1h 39m 27s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/07e0c2344da7409f8d3b804abbcfc2d6 : FAILURE in 2h 15m 31s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/29656ce0ba7943e5b09673a8d7266839 : SUCCESS in 10m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1a0bd37b8f3d46eea1c6988b97989c33 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":10},{"id":"49d55abf43e771fe1f725e6f1859921a2a36e70c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-15 09:16:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/af624a6cabbb4392b49bf7e2be06fbf1 : SUCCESS in 2h 39m 18s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/26368016dd1543aca2f0e44fda05c59b : SUCCESS in 2h 02m 43s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/a27016f2713a4306870fb024d57bbb52 : SUCCESS in 2h 19m 07s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/bf243be3da2b4243b0e4a23d19030d57 : SUCCESS in 2h 39m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/f1a4dd58374b4822b11f5e07e4f472e4 : SUCCESS in 2h 42m 19s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/28c8af704b1c43318b15be08a38ed15b : SUCCESS in 3h 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3fbd631796e5e9e1f2b6f5127fe02f639e537003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 12:19:41.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2c9dec32e1b35fa806c61f10d39e5d831d7c5a21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 17:32:58.000000000","message":"Patch Set 10:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"28465512df9311008011cc0f44c41b24d83d5429","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 20:45:55.000000000","message":"Patch Set 10: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"2843a5e1d1203b60c6c73a5c59465c9c51b4a0ff","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-16 10:14:28.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"d1582fa58561d59b369c4d8b8bc9f948baf1bfda","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-16 10:14:33.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"44b955ee2c48a0fd6216e13bdb8db9e72a64b0cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 12:58:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6d423274d58649f1916d760f2d051e91 : SUCCESS in 1h 17m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c2bc0b354fa64bf79f15cfea0bfaea7f : SUCCESS in 2h 20m 11s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/efdc8b22657246228bbff8f6d54dd185 : SUCCESS in 1h 54m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b61d51baa7eb492d848fe0328bdd861a : SUCCESS in 1h 44m 33s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/c0d0d4d524834289bc8009d52f9885aa : SUCCESS in 2h 41m 57s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/76636834033f4748adbebe1486399148 : SUCCESS in 2h 19m 53s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/f04f92a7476d4157ab839b8f1607c92a : SUCCESS in 2h 32m 28s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/108f928c76354a1494be6304f431c411 : SUCCESS in 2h 14m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ea8ca6ddff45da940509152dce5fca : SUCCESS in 6m 27s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/1a3eda9ceab14e698f3e2a70e7253fac : SUCCESS in 1h 26m 13s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/21d1da1bf3dc46c5bebb2ff3c0f1b870 : FAILURE in 1h 17m 39s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/0c662a18974c49e8b35cba341e343fde : SUCCESS in 13m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55501c9f37744bc794166204527e8c8f : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":11},{"id":"d3d8ee302dd9dbf075eb14fac972ec7c9342ae52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 13:01:28.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"b6cccd8e18236a421f3ec7e8a339418ae35b18b7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-16 13:05:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/737be52b564f4266992cd40732ad916c : SUCCESS in 2h 34m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/5762858e6c4848cc97608f675ca27a50 : SUCCESS in 2h 07m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/92f7ba3a9dee4155bb36c95c3162720d : SUCCESS in 2h 24m 31s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/7bcb68b1d95a4726b2a45ec32791645d : SUCCESS in 2h 43m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/1ff7adba9b804d3b9341c110f40e3780 : SUCCESS in 2h 42m 37s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/989fe0c299364555a4e7a255788f1aa9 : SUCCESS in 2h 49m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e0575759ac8a6948487b434a5083bc65b4580229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 15:51:04.000000000","message":"Patch Set 11: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"6d8c10f77ec4953c85030d9dd272cc29c295516c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-16 23:10:30.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"064d3f10a8188e5aa7c0789f529eb32ac1a27060","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 01:44:18.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/23a9a8031b57487788e17f1789fa785e : SUCCESS in 1h 14m 42s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e8af928f5b4b4bdb920cf683837fd1fc : SUCCESS in 1h 52m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/99242e44406142e28905242df8ba220d : SUCCESS in 1h 43m 58s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/aaab93e4a79c433da4f79b1a97e1250a : FAILURE in 1h 55m 24s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/2f605fc03b0e4bb9a4a59488af4a70b6 : SUCCESS in 2h 32m 24s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/50f0a0d4bbd14757915f9687e3a0ccc5 : SUCCESS in 2h 15m 29s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/2b81dbb2084f4ce1813d9227c75219e5 : SUCCESS in 2h 10m 41s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/5254f64b32b949bc8254942241b1b762 : SUCCESS in 1h 49m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3399c8b37cc04d5593d3bbb231b5e8d3 : SUCCESS in 4m 03s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/269feebc280042748acf45bc662eb8e7 : FAILURE in 17m 05s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/dc08adf2bbe34a689d0b2cc6aa4ee74b : SUCCESS in 1h 27m 35s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/51b61a4cd81249c581f6e20d5a4a9577 : SUCCESS in 11m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ecdcf7fe432c402986732dc562170d13 : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":11},{"id":"7d9fb43fc23f9d6d57c745a897aa543d054c402a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-17 01:57:50.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ddbb3009d9784f51aa4b6bef572e0255 : SUCCESS in 2h 35m 59s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/da3b428f05ad4b8584981863326b911d : SUCCESS in 2h 07m 03s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/0938c6dbd19f4ac286d66e9a1846c794 : SUCCESS in 2h 16m 36s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/41d3f82d328244babda6e10567cef385 : SUCCESS in 2h 45m 48s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/14b9538e1ab340bc98864fc0e0e56cff : SUCCESS in 2h 44m 20s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/96837e4157a94df3bd1252e5160ae48a : SUCCESS in 2h 44m 16s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"de2e3018acc7130c0384109c0d8ef463cac62c08","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-05-17 11:02:48.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"6516ed3b490c5b6bf7517a3df76ed12c80a1983e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 13:25:34.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5bf9eebb1e5348afac4e20fec7c53f23 : SUCCESS in 1h 30m 47s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/eefa5c3ab0204a4688f4c64fcad3cbe7 : SUCCESS in 2h 18m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e2573167c9654165a5b9853b709414a6 : SUCCESS in 2h 21m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d564ad83157a43a99d732dd9c7e1ceb5 : SUCCESS in 1h 43m 30s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/11734647bfd04523b94d67d232526045 : FAILURE in 36m 39s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/3b82dc63fec04f639c0577ce93e48705 : SUCCESS in 2h 08m 03s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/518acb69973d4ed19dda94b4d0e06b49 : SUCCESS in 2h 11m 32s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/be01fe33c7144c5ba05f9e9832809296 : SUCCESS in 2h 12m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa4780e265b4e84805d4f50ad1421db : SUCCESS in 4m 21s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/3b2c66d694ce4c97bff91e0ea4e3ebf3 : FAILURE in 2h 13m 55s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/c905abb50e124a6e974727c615fe6adb : SUCCESS in 2h 07m 54s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/bd29181c9e9c4e79b92893872a815db2 : SUCCESS in 10m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/91d92cde08b64b87888304645e2232a0 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":11},{"id":"01c80d1809a7e76474aa0d4f019f92cfcb03c136","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-17 13:48:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d0e01a897f8347db87f0114329565188 : SUCCESS in 2h 34m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/e804629850584e0eb61c85308a520173 : SUCCESS in 1h 57m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/aa65bbac85aa41f6ad43ccd8e18265a4 : SUCCESS in 2h 16m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/add11b4c739044d2b20e994ef760cea4 : SUCCESS in 2h 33m 58s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/591b4699ca95429095f5fbc23786289e : SUCCESS in 2h 44m 35s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/236ba25323734d9a974e31664a534487 : SUCCESS in 2h 43m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e2fe4b464901fd1d2d26d8ca405b6384470bb3b8","author":{"_account_id":20182,"name":"Bhagyashri Shewale","email":"bshewale@redhat.com","username":"bhagyashris"},"date":"2020-05-17 13:54:33.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"de14e42be6539691f7adacbbcaf4eee95efa0e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:35:02.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4dee6df47bbe475cbe40b599afec1ce9 : SUCCESS in 1h 27m 58s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/62d90df287664783817c65e2c15687e2 : SUCCESS in 2h 07m 40s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cdc6ffbbf81b4393a3c8043fee189309 : SUCCESS in 1h 50m 44s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/987de60ad05949768dfc8f372f9ca93e : SUCCESS in 1h 48m 27s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/fafa720de9a84836984950a25a014ee2 : SUCCESS in 2h 38m 34s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/f35708957e224a50a4c876906f3f4330 : SUCCESS in 2h 24m 48s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/b878cdba6f13416489876e60d8d8551d : SUCCESS in 2h 10m 00s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/eaa5c7545d4b4450bd07d7a78442059d : SUCCESS in 2h 09m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfe9ad378b0464e862e59eaf1bacf26 : SUCCESS in 5m 34s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/aa3e0f9517a24025979f756424a9bfc0 : SUCCESS in 1h 42m 16s\n- tripleo-ci-centos-8-scenario010-ovn-provider-standalone https://zuul.opendev.org/t/openstack/build/47bfe749f6784b289c3bc1c1129ce7a7 : SUCCESS in 1h 44m 00s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1c92b26a710243d5bacd4e88d69c05fe : SUCCESS in 12m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/25541162a17345aba6808bdc8a055126 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":11},{"id":"da63357a3b0ab451a4544da1418381649a66f040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 16:35:13.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"426a13bd9f665a76e6792663a350582b87236e3a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-17 16:40:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/622ff3ee5b7e4b509690f4f63189e4bf : SUCCESS in 2h 32m 14s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-queens-branch https://review.rdoproject.org/zuul/build/adfdf8c8eaae484faba71736335f5027 : SUCCESS in 2h 09m 27s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-rocky-branch https://review.rdoproject.org/zuul/build/fe62c9de55634ba3b4553aa1de8d33ee : SUCCESS in 2h 24m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-stein-branch https://review.rdoproject.org/zuul/build/59de2752126146e88aacf257142cfad3 : SUCCESS in 2h 41m 01s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-train-branch https://review.rdoproject.org/zuul/build/9a1bae7ca80e47bda8a2dbc688c122aa : SUCCESS in 2h 44m 15s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/173a619b59fd4abb84aecfb2284b8ac4 : SUCCESS in 2h 42m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"763d9a26dcbe7df7dfacdb90ee5d92b6d38d08fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 19:26:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"be191e213d2546c4b3eb17a1d2d43f1ce0d56de6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 19:26:26.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e501fe11a1ef46378e92c6e39c4b91bb : SUCCESS in 1h 29m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/20720226a23847e19261cd5d129da1a7 : SUCCESS in 2h 14m 14s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c366a99b54e64faba3a3e10f6c5c37a8 : SUCCESS in 1h 41m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c1ee0d691c58488696067fd95bd2501f : SUCCESS in 1h 49m 04s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/0207a7242a7a4c09b2d04222c961ddb5 : SUCCESS in 2h 46m 41s\n- tripleo-ci-centos-7-containers-multinode-stein https://zuul.opendev.org/t/openstack/build/fe491d9ffe0d44c2822186545b9498dd : SUCCESS in 2h 15m 11s\n- tripleo-ci-centos-7-containers-multinode-rocky https://zuul.opendev.org/t/openstack/build/dff45e029caa4e5bad9e3a7aced018cd : SUCCESS in 2h 28m 01s\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/d6bf534f7af543a9b5b3fac450a33fc0 : SUCCESS in 2h 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9c2281db4b47709ff38ff84117f088 : SUCCESS in 4m 27s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/54c09391b382404f9fce4bb0706b5a82 : SUCCESS in 1h 47m 11s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8b14454127414107a012f3440bcd3c42 : SUCCESS in 11m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/abfc3e3510c7448598db8481093b4a9d : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":11},{"id":"0cce3f3644b496977e9e83b8f8b6f03e1f011890","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 19:27:38.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0565b31cff4cc092d66dd8be2f71c8 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"5fe440af2b3714ba4b6ce0e66046394fd2794dfa","revisions":{"cc72d2f3925d3c943fb252bb139c8d483cbe48e1":{"kind":"REWORK","_number":1,"created":"2020-05-11 08:26:41.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/1"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc72d2f3925d3c943fb252bb139c8d483cbe48e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc72d2f3925d3c943fb252bb139c8d483cbe48e1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a5f238d6c2a8d257fc4436552eabf32aa83fe273":{"kind":"REWORK","_number":2,"created":"2020-05-11 11:27:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/2"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 11:27:01.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a5f238d6c2a8d257fc4436552eabf32aa83fe273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a5f238d6c2a8d257fc4436552eabf32aa83fe273"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a3db83dcf005b35bf655fe610d30b939bcab5e4f":{"kind":"REWORK","_number":3,"created":"2020-05-12 08:05:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/3"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-12 08:05:15.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a3db83dcf005b35bf655fe610d30b939bcab5e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/a3db83dcf005b35bf655fe610d30b939bcab5e4f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"851319c2cf96f37c4ef1b6cb07e4fbb40ab81c73":{"kind":"REWORK","_number":4,"created":"2020-05-13 08:13:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/4"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-13 08:12:39.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851319c2cf96f37c4ef1b6cb07e4fbb40ab81c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851319c2cf96f37c4ef1b6cb07e4fbb40ab81c73"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"63b90eb5697bde26fee009d86ea117f4d94e7198":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-05-13 12:08:16.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/5"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-13 12:07:54.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nDepends-On: https://review.opendev.org/727739\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/63b90eb5697bde26fee009d86ea117f4d94e7198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/63b90eb5697bde26fee009d86ea117f4d94e7198"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"4eb2b6c4eb6cfb3854252d0aa5a92d1b7338e751":{"kind":"REWORK","_number":6,"created":"2020-05-13 18:36:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/6"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-13 18:35:41.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nDepends-On: https://review.opendev.org/727739\nDepends-On: https://review.opendev.org/#/c/727854/\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4eb2b6c4eb6cfb3854252d0aa5a92d1b7338e751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4eb2b6c4eb6cfb3854252d0aa5a92d1b7338e751"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"aa735d0fbb0618f9b32d19ee31d4072fed76b520":{"kind":"REWORK","_number":7,"created":"2020-05-14 07:29:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/7"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-14 07:29:00.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nDepends-On: https://review.opendev.org/727739\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/aa735d0fbb0618f9b32d19ee31d4072fed76b520"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/aa735d0fbb0618f9b32d19ee31d4072fed76b520"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"e57143dc7c82f24158a2c1e08aa00ec2c1a244a9":{"kind":"REWORK","_number":8,"created":"2020-05-14 10:35:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/8"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-14 10:35:03.000000000","tz":0},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nDepends-On: https://review.opendev.org/727739\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e57143dc7c82f24158a2c1e08aa00ec2c1a244a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e57143dc7c82f24158a2c1e08aa00ec2c1a244a9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"968da68a323e007611a2a79a7f9123928e6b7470":{"kind":"REWORK","_number":9,"created":"2020-05-14 13:58:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/9"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-14 13:58:10.000000000","tz":330},"subject":"[WIP] move fs062 to os_tempest","message":"[WIP] move fs062 to os_tempest\n\nDepends-On: https://review.opendev.org/727739\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/968da68a323e007611a2a79a7f9123928e6b7470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/968da68a323e007611a2a79a7f9123928e6b7470"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"8d87e76757ce43a9a5235df52e923e341253af02":{"kind":"REWORK","_number":10,"created":"2020-05-14 18:37:42.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/10"}}},"commit":{"parents":[{"commit":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","subject":"Merge \"Update docs building\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cc028fa584f6d53bc5e4c428e0e0d199ed3ae194"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-14 18:36:54.000000000","tz":330},"subject":"Move fs062 to os_tempest","message":"Move fs062 to os_tempest\n\nIn order to create public network, neutron tempest services needs\nto be added.\n\nDepends-On: https://review.opendev.org/727739\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8d87e76757ce43a9a5235df52e923e341253af02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8d87e76757ce43a9a5235df52e923e341253af02"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cc028fa584f6d53bc5e4c428e0e0d199ed3ae194","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5fe440af2b3714ba4b6ce0e66046394fd2794dfa":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2020-05-16 10:14:28.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/52/726752/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/52/726752/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/52/726752/11"}}},"commit":{"parents":[{"commit":"42ff8812883cd39df40018c234ab3ade646246a8","subject":"Merge \"Switch minimal featureset to os_tempest for train+\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/42ff8812883cd39df40018c234ab3ade646246a8"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-11 08:26:05.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-05-16 10:14:28.000000000","tz":0},"subject":"Move fs062 to os_tempest","message":"Move fs062 to os_tempest\n\nIn order to create public network, neutron tempest services needs\nto be added.\n\nDepends-On: https://review.opendev.org/727739\n\nChange-Id: I6ae79a276d3cd2af929646455d17ce46fede770a\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5fe440af2b3714ba4b6ce0e66046394fd2794dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5fe440af2b3714ba4b6ce0e66046394fd2794dfa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"42ff8812883cd39df40018c234ab3ade646246a8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
