)]}'
{"id":"openstack%2Ftripleo-quickstart~743919","triplet_id":"openstack%2Ftripleo-quickstart~master~Ic17bf6e69b1620bdcca339f436e1399b29c0437e","project":"openstack/tripleo-quickstart","branch":"master","topic":"bug/1889219","hashtags":[],"change_id":"Ic17bf6e69b1620bdcca339f436e1399b29c0437e","subject":"Switch Standalone environment to centos8","status":"MERGED","created":"2020-07-30 07:07:50.000000000","updated":"2020-08-10 11:17:30.000000000","submitted":"2020-08-10 11:16:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743919-1597058181009-60d190a1","meta_rev_id":"b547ef7ded65598a8653c8ea98af1da5be29d271","_number":743919,"virtual_id_number":743919,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 11:16:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-10 06:41:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-10 06:54:00.000000000","_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2020-08-10 06:51:22.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2020-08-10 08:14:00.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2020-08-10 08:14:00.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 21:07:54.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-10 06:41:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-10 06:51:22.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2020-08-10 06:54:00.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-10 08:14:00.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-08-10 11:16:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8bd5430ae6457e951ee51cce578392f67a63dd9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 07:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dc49133240fc87344d034cf6c1e8ed757194b84","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 07:10:47.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/1/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7efb10510ddac6ae724051e23ccd8a923d6a648a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 07:28:53.000000000","message":"Patch Set 1:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":1},{"id":"2d54f3e0a25977f7307f449054c8099cf2a9abb7","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 07:38:07.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/2/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"88a039023645d17b33944974e4bc0b1d10cde8e9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 07:56:35.000000000","message":"Patch Set 1:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":1},{"id":"97b47bc782e4e895094506cf8e973aebf396fb10","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 08:05:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/3/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"034237c4539cce3deb90251e0a7e4976f49cd923","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 08:08:45.000000000","message":"Patch Set 1:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":1},{"id":"1186795ef42c5d8fd3cd548b377137d52fbc4f33","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 08:18:06.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/4/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"257063ad41232081f75d667bbacf44a00aa19fdd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 09:11:01.000000000","message":"Patch Set 1:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":1},{"id":"21c732da0411dc180233c0d8b10384c9c423a2bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-30 09:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0346fc6bf4574eac9139be46a4f8cb25 : SUCCESS in 2h 08m 28s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/ef7f83d49ab545e8baab160ae0eee7f3 : SUCCESS in 2h 08m 54s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/b6586c9174ce4178bfce1dd46de237d2 : SUCCESS in 2h 21m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32e295ccb55627953294499ee518592b68f0fa0a","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 10:43:37.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/6/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"71ff7636a538f25d7f7644716eee8b70afc9c551","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 11:06:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"135db792b4dbfece3e97509600a5cce688d25443","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 11:15:20.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/7/ : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"0221790165be2e574a9db838e2d53977b26e5fc8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 11:24:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"364f9b2d5c8f498506ae67045a7a105636765cdc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-30 13:52:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e5c241265c974366845870b53c551a78 : SUCCESS in 2h 04m 53s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/81f8dc64b65c47b4a41ef174b5e04721 : FAILURE in 1h 17m 50s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c4f1ad5ae28744978d1937ab89fd4064 : SUCCESS in 2h 27m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d91153d9cc89749c3d2e71fc75341201ea91bd1","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 14:37:13.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/8/ : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"6bdbf91fdf3efa5da13e3bd59cef21766899d1e7","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-07-30 14:38:16.000000000","message":"Patch Set 3:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":3},{"id":"f847921151c08052071ccf171f4cf26296b979ab","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-07-30 17:26:56.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/9/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"713cb1943db3e3a6754357c9307dec02dbf17880","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-07-31 15:44:17.000000000","message":"Patch Set 3: Code-Review+1\n\n/me needs to play w/ this","accounts_in_message":[],"_revision_number":3},{"id":"4298839719e6c16b4bb33b9b6d5f8abf547d4110","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-03 09:23:02.000000000","message":"Patch Set 3:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":3},{"id":"9afeb3df03a1e9a3b7547b8c9bae3a2af144b443","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-03 09:27:07.000000000","message":"Patch Set 3:\n\n\u003e /me needs to play w/ this\n\nWith this patch and others tqe patches at https://review.opendev.org/#/q/topic:bug/1889219+(status:open+OR+status:merged) following should work:-\n\nbash quickstart.sh -R master-tripleo-ci --no-clone --tags all -p quickstart-extras-standalone.yml -e @./config/environments/standalone_centos_libvirt.yml -c config/general_config/featureset052.yml  $VIRTHOST","accounts_in_message":[],"_revision_number":3},{"id":"3246a6f7f79421d0e4f0696292c5255527236bda","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-03 11:46:06.000000000","message":"Patch Set 3:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/10/ : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"9f4a05330d46eb176420b7cd0968c14b99fa22c8","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-03 18:03:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bad4bb58f2853e635893850963bb31cc7d645099","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-03 18:42:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a73391e3b79dbd7804a371bc3ade6f6b876fdc7","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-03 21:37:53.000000000","message":"Patch Set 3: Code-Review-1\n\ntempest fails to setup the network.. but got passed the deploy","accounts_in_message":[],"_revision_number":3},{"id":"05a34a560932e3c249e32cd5bb718181e851d6b8","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-03 21:39:29.000000000","message":"Patch Set 3:\n\nTASK [os_tempest : Ping router ip address] ****************************************************\ntask path: /var/tmp/standalone/usr/local/share/ansible/roles/os_tempest/tasks/tempest_resources.yml:283\nMonday 03 August 2020  14:38:11 -0600 (0:00:00.040)       1:54:25.601 ********* \nFAILED - RETRYING: Ping router ip address (5 retries left).\nFAILED - RETRYING: Ping router ip address (4 retries left).\nFAILED - RETRYING: Ping router ip address (3 retries left).\nFAILED - RETRYING: Ping router ip address (2 retries left).\nFAILED - RETRYING: Ping router ip address (1 retries left).\nfatal: [undercloud]: FAILED! \u003d\u003e {\n    \"attempts\": 5,\n    \"changed\": true,\n    \"cmd\": \"set -e\\nping -c2 \\\"192.168.24.102\\\"\\n\",\n    \"delta\": \"0:00:03.084743\",\n    \"end\": \"2020-08-03 20:39:24.412120\",\n    \"rc\": 1,\n    \"start\": \"2020-08-03 20:39:21.327377\"\n}\n\nSTDOUT:\n\nPING 192.168.24.102 (192.168.24.102) 56(84) bytes of data.\nFrom 192.168.24.2 icmp_seq\u003d1 Destination Host Unreachable\nFrom 192.168.24.2 icmp_seq\u003d2 Destination Host Unreachable\n\n--- 192.168.24.102 ping statistics ---\n2 packets transmitted, 0 received, +2 errors, 100% packet loss, time 30ms","accounts_in_message":[],"_revision_number":3},{"id":"b4d022fef8684c974da7fb90d0ba5e752f9e8c51","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-04 05:34:21.000000000","message":"Patch Set 3:\n\n\u003e tempest fails to setup the network.. but got passed the deploy\n\n@weshay can u cross check if u have https://review.opendev.org/#/c/743913/ patch included, with this it should have passed.","accounts_in_message":[],"_revision_number":3},{"id":"f843024968ce5508393f649ea5474a5676fb55db","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-04 05:35:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a977c341a53142f35c8df0e4971563cddfc827f2","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-04 07:02:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"055576a7d1937378b3cde8f0a60ed4cd5f590b8f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-04 09:05:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a95759924c474dceb6b1ceeb24f5d0ef : SUCCESS in 2h 01m 46s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/08534038ad464fc0837590364b710369 : SUCCESS in 2h 00m 46s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/3afbc03b91e24104821245833b8798c3 : FAILURE in 1h 57m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b3212f4672e8a318206c5e81b4dbb70824d95c36","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-04 10:26:11.000000000","message":"Patch Set 4:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/15/ : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"b2cc0d351e0c9062809012c2ed7f2542bc600f8c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-04 11:40:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"023a82597323472a6b23951e11892ba301c5ddff","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-04 14:01:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7e2a0cbc40c14526a48d7b28f2410830 : SUCCESS in 2h 17m 44s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/5635d8c86d034746b631d11874b78c0d : SUCCESS in 1h 59m 26s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/10638474403d44258431a70beb459ca7 : SUCCESS in 2h 19m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"731a8d5beafa6abf251078c9bfeb61e11cfffb8b","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-04 15:19:49.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/16/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"35502d1a2e3ae9691104306d2ede3f186d31ad7a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-04 15:44:31.000000000","message":"Patch Set 5:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":5},{"id":"670311d37e3e70b80ef961ecdbec72b594843a31","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-04 19:15:33.000000000","message":"Patch Set 5:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/17/ : SUCCESS","accounts_in_message":[],"_revision_number":5},{"id":"4913babdbb5806d5719140e6d160c30690f18f8a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-04 21:07:48.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nreally nice work Yatin, have some questions.. please work w/ Chandan and Marios on the standalone config part.  IMHO using the centos qcow2 in images.rdoproject.org would also be a ++","accounts_in_message":[],"_revision_number":5},{"id":"f0142077c31ddf0d040a9664fdfdc4da202c6767","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-04 21:07:54.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"9ab70e69d124aa5f7b0a7d1692c13e005d6b8279","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-05 05:54:29.000000000","message":"Patch Set 5:\n\n(3 comments)\n\n\u003e (3 comments)\n \u003e \n \u003e really nice work Yatin, have some questions.. please work w/\n \u003e Chandan and Marios on the standalone config part.  IMHO using the\n \u003e centos qcow2 in images.rdoproject.org would also be a ++\n\nSure will check with both chandan and marios to move forward and will update to images.rdoproject.org.","accounts_in_message":[],"_revision_number":5},{"id":"d7bb49217392175e89fe87f14446e42f13ab435c","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-05 06:18:02.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ad19bf55a54ef55b953a84f499e3d2d58388e82e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-05 09:03:44.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"081ff9b643920487f258ad0c3da14cb3326eaec6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-05 09:34:50.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1ca535034014ac6c4ed1652c5bfc702a26081621","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-05 11:14:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/99fb877db9b04cbfbd3fe9ad8e6e7852 : FAILURE in 1h 13m 27s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/237a2f7f23744b498725e117e448f9dd : FAILURE in 1h 12m 50s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/e9e25eb2c1354251bd7e97b56bb0f8a9 : FAILURE in 1h 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d6f572736b5345ef0d2da3c1a6df4f2f6ca86940","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-05 12:17:16.000000000","message":"Patch Set 6:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/18/ : SUCCESS","accounts_in_message":[],"_revision_number":6},{"id":"b13e82776bbe3aa57aa2710802b879562628d2b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 12:59:43.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fde7b3fe011d4f21a615ddd0d105fb09 : SUCCESS in 1h 30m 15s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1d6dd3c7d39a4462b4eab37edf367485 : SUCCESS in 2h 04m 19s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e985618294c441a680ce0e75379afc41 : FAILURE in 2h 46m 10s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/83947510f33e434d924dfd80dc4e58a6 : SUCCESS in 1h 49m 47s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/655c4ae8fd24403280a4b66a424885c7 : FAILURE in 1h 09m 13s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/35bf00641f6a44e384bab529cf07cc22 : FAILURE in 2h 10m 59s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/79e470f93374400caadfec542bcc64d1 : TIMED_OUT in 3h 06m 36s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/928f1c8f28bf44179ff9e1d2757ef31e : SUCCESS in 2h 34m 48s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/59f8e0cb42ef4cb7b77842235819daf4 : FAILURE in 1h 50m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cb172da58b495dbb9e9510798843f4 : SUCCESS in 7m 00s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/003f12cb706649119f3232bdf8110363 : FAILURE in 1h 19m 18s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/72e1004e4a92415bbb1ffcbe53c257ae : SUCCESS in 12m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/174047dab69c436582f34e1b0c279b7a : FAILURE in 7m 11s","accounts_in_message":[],"_revision_number":6},{"id":"58f73590f74b6297f616b7093bcdb5bed14557d4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-06 05:21:15.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"fcb5e5f1e840cf07387dbc550c18c86754927836","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-06 05:26:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4ef7ac1aab7fb4b658a2fb5db0f13469a052187f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-06 05:26:49.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"22b5b7ed27c43af37f991970c26466dae8dac24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 08:05:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/86b8438edd19437f860009fcc5a11ca1 : SUCCESS in 1h 20m 52s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/1805e07c19bb4fba87d6a00b6c3a5606 : FAILURE in 1h 45m 27s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1aefa68696874e558389d0a458ab223b : SUCCESS in 1h 58m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/99836b48b751488b8fccc7acdf0b0cd7 : SUCCESS in 1h 27m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/b523e083eb2545cd9b427b7dd4f42f6e : SUCCESS in 1h 59m 34s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f93804c3c68e4f72882c77e926b9943a : SUCCESS in 1h 36m 13s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/58c90f2678b5419db92c8eaae7365646 : SUCCESS in 2h 12m 06s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/61c324bd84df4bbf9c8e753a0365917f : SUCCESS in 2h 17m 13s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/7b6e19bcf5c3436d957c6749e0a60049 : SUCCESS in 2h 16m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f55187910274815a5253df1d0e8efc6 : SUCCESS in 4m 49s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/f3cb6073825a4fc1b6498e3c801df90d : FAILURE in 1h 26m 54s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fcc504252a6b44ac92c1e4e430ce1373 : SUCCESS in 11m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d69b569e589a4d2589c3a0f2d847bf1a : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":9},{"id":"4dccbadde16848d08957e04fce546f971b447256","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-06 10:37:04.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a8018aac22db49b581623cd54068e0d7 : RETRY_LIMIT in 1h 34m 56s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/d0f3e78352654846b9ee9b9353cb845f : FAILURE in 1h 36m 11s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/e62409d9066a48dc85fcb8ebf13ee221 : RETRY_LIMIT in 49m 53s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"78a69b773a1880393f89b1a1d8cbed82b844af4a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-07 08:00:11.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)\n\ncan\u0027t spot any nits but I need to look at it at least once more","accounts_in_message":[],"_revision_number":9},{"id":"b146fae2d15834b8640429e9f4c3d3430684f29f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-07 12:50:04.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"52941c2199708da61e608e624ac4df40ff9a9520","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-07 12:56:32.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d6cda5d60192a75d6180b8efc53291d21652dbe8","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-07 13:20:12.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"df67d4404a2b7c77a9b93cb4781a703d4a13f8e1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-07 13:53:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2c7343e9afd219d6b942316d5eb1b01094e864fd","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-07 14:00:21.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0cb6c3ed3de3ea613c9096e71d872bef9bd34ae1","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-07 14:03:31.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/21/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"4aafe9af0d4a5143fd99933bf7a4a6d9e6673ec4","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-07 14:03:45.000000000","message":"Patch Set 10:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/19/ : FAILURE","accounts_in_message":[],"_revision_number":10},{"id":"bb26e932d22acf6626abe6c194d98a18237c8b44","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-07 17:32:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/752d5358b3e04412933a5892e9ce8601 : SUCCESS in 2h 00m 43s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/729e36b30cca465d93cc6ab2e3f42849 : SUCCESS in 2h 00m 21s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/c4235c9082f64bd2bf7252da6f76aa57 : POST_FAILURE in 3h 27m 24s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"34b338ad4066fa05bd25bf03c975c12a96fcc94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 18:30:39.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f16b9d98570b47339490e882f4716b66 : SUCCESS in 1h 18m 33s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/d38fe513925f4ce2950c219c42028600 : SUCCESS in 1h 46m 48s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/dbab1ba8506943c891d51ecaa01fed37 : SUCCESS in 2h 56m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/df697732bce64f559f0d00a824450c7b : SUCCESS in 1h 29m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/6a0f8687a3d74280b993bd1fe773f731 : SUCCESS in 1h 59m 09s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1debae5b81bc4f138fa79ffb406806ce : SUCCESS in 1h 30m 53s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/44b17ad788434d4e9e5ad08d4870f296 : SUCCESS in 2h 05m 06s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ff6719cd7d6541519748b58c247b11bb : SUCCESS in 3h 05m 40s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/cae01beaa43f424298f8aaf4d33fe7f6 : SUCCESS in 2h 48m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18bd6d2398f346b884953bb1fd1a8ae3 : SUCCESS in 5m 40s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/6b8ef17870a1485a8a962a15a8b84ea7 : FAILURE in 1h 35m 16s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b5a70387367a4b7692219e8affd5bbe8 : SUCCESS in 11m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/798fadc377c242afb6fc292dbfa68a4e : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":11},{"id":"aec8b6c61fd12b92b0c4c9aae0474aea6236401a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-07 18:39:08.000000000","message":"Patch Set 11:\n\nrerun ykarel","accounts_in_message":[],"_revision_number":11},{"id":"fa7f1b6a73b3f4afdee31b2a36203891bcd74258","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-07 20:00:07.000000000","message":"Patch Set 11:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/22/ : FAILURE","accounts_in_message":[],"_revision_number":11},{"id":"cc3d9287ecf4479fb160504622f74e4756fa4086","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-10 04:06:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"98418f8d7ee304ab4bdc5f00d0a03c3f3e0fb531","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-08-10 04:12:00.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f8b786ae61094fa7fb1cde8dc76c46adee1b474e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-08-10 06:41:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c07826bdf7ee488f88669de5ab905ed7 : SUCCESS in 2h 05m 05s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/438c5cebc0a0435e885dbb08eeeedcdd : SUCCESS in 2h 27m 26s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/d2cefa65b8714fe0817db3125d999203 : SUCCESS in 2h 16m 29s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2d9393c93e7c786e56bd317c5db512ddaf95be61","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2020-08-10 06:51:22.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"c7df745c7349f4cad98c576669d18e04119bfd66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 06:52:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/91274bbf08534ed0a0f1352e1792438a : SUCCESS in 1h 12m 24s\n- tripleo-ci-centos-8-containerized-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/cc64afb1542644ecae216689d49f464f : SUCCESS in 1h 43m 41s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c156228a1da044ceac5fa29dad75649d : SUCCESS in 2h 33m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/451eee6e2791413485256e47773d8683 : SUCCESS in 1h 58m 35s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/ad760c997ddb4e46b960eff8cbe8178c : SUCCESS in 2h 00m 01s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d08fc96491714778aec7c682f7bf65c9 : SUCCESS in 1h 54m 29s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/909f7a8a00704cb0997851a95a84d4ea : SUCCESS in 2h 31m 39s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/4eb0c083a0dd4538b4b867e6f84b70a3 : SUCCESS in 2h 27m 16s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/4db348fa2de741deb853c359d02c6af2 : SUCCESS in 2h 20m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52072ccca96a454f97d6dd703c22622c : SUCCESS in 4m 08s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0425e7c5f47d44898780850cd2601f7b : FAILURE in 1h 29m 15s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5fa3a2b53b624857b8c64c00a5cf6ff7 : SUCCESS in 10m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a47b6e1a3b014f7e8bab35ffe6ff74bd : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":12},{"id":"73b20903920462936a7db0635591107d8dbcdd72","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2020-08-10 06:54:00.000000000","message":"Patch Set 12:\n\nBuild Successful \n\n* tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052 https://ci.centos.org/job/tripleo-quickstart-gate-master-delorean-full-ykarel-featureset052/24/ : SUCCESS","accounts_in_message":[],"_revision_number":12},{"id":"87fa22e2e85d01fcb6feaeda291c460b0739b304","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-08-10 08:14:00.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"8986af15f10056c584f01d35a125f2a82f4a1cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 08:14:07.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"dfe810c5e6e61e82f1f62fe6ca232e4d16e2fd65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 11:16:20.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca2fd635f2984b6bae7a1e7d39e7741e : SUCCESS in 1h 28m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/706cbc9a297d480ab142215020d4c023 : SUCCESS in 2h 02m 45s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1516d87115ed468c8f0b7c6214206dc9 : SUCCESS in 1h 52m 30s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b32e38c0eae24e7fa350f3961aa6e4bb : SUCCESS in 2h 00m 57s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/860818567e3642ce99f4ffc5c0362b01 : SUCCESS in 2h 19m 34s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/9462887082614aa09636cf9e43be475a : SUCCESS in 2h 19m 46s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/acdd358ed5f54bac80524ef67a7c06a9 : SUCCESS in 2h 55m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c613321707814d888cc614ac97e568ba : SUCCESS in 5m 07s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fd6ef1efdbbd406cb440acaca0240ad7 : SUCCESS in 11m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c172fbde45b40bb9332bb288e4df290 : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":12},{"id":"93ff84a4cee476e5fa3731a9036b77ae317584ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 11:16:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"8554aa93cc2196d4c860b584f7a1ade44fb65710","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 11:17:30.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87faf8e0d4d14ddab551ab229d3c517a : SUCCESS in 49s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"814b080a60f08dc6d292138cb5a1603c6885672c","revisions":{"d03370bf5aeb81bbc3c675dcb15c59195c8d8dbb":{"kind":"REWORK","_number":1,"created":"2020-07-30 07:07:50.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/1"}}},"commit":{"parents":[{"commit":"031aa65b26263af24b3226c46d496953b11589fe","subject":"Pin pip to last good version - 20.1.1 for c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/031aa65b26263af24b3226c46d496953b11589fe"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 07:07:04.000000000","tz":330},"subject":"[WIP] Switch Standalone environment to centos8","message":"[WIP] Switch Standalone environment to centos8\n\nAlso add some os tempest options to fs052.\n\nCloses-Bug: #1889219\nDepends-On: https://review.opendev.org/#/c/743916/\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d03370bf5aeb81bbc3c675dcb15c59195c8d8dbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d03370bf5aeb81bbc3c675dcb15c59195c8d8dbb"}]},"branch":"refs/heads/master"},"2074353a84e2b9cb05db693e8c6411b738f88b15":{"kind":"REWORK","_number":2,"created":"2020-07-30 11:06:20.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/2"}}},"commit":{"parents":[{"commit":"031aa65b26263af24b3226c46d496953b11589fe","subject":"Pin pip to last good version - 20.1.1 for c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/031aa65b26263af24b3226c46d496953b11589fe"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 11:01:01.000000000","tz":330},"subject":"[WIP] Switch Standalone environment to centos8","message":"[WIP] Switch Standalone environment to centos8\n\nAlso add some os tempest options to fs052 and update\nrelease file for container builds.\n\nCloses-Bug: #1889219\nDepends-On: https://review.opendev.org/#/c/743916/\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2074353a84e2b9cb05db693e8c6411b738f88b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2074353a84e2b9cb05db693e8c6411b738f88b15"}]},"branch":"refs/heads/master"},"c1cafcf68de1adf6eef1378ed1aeac7d705bfc02":{"kind":"REWORK","_number":3,"created":"2020-07-30 11:24:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/3"}}},"commit":{"parents":[{"commit":"031aa65b26263af24b3226c46d496953b11589fe","subject":"Pin pip to last good version - 20.1.1 for c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/031aa65b26263af24b3226c46d496953b11589fe"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 11:23:08.000000000","tz":330},"subject":"[WIP] Switch Standalone environment to centos8","message":"[WIP] Switch Standalone environment to centos8\n\nAlso add some os tempest options to fs052 and update\nrelease file for container builds.\n\nCloses-Bug: #1889219\nDepends-On: https://review.opendev.org/#/c/743916/\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c1cafcf68de1adf6eef1378ed1aeac7d705bfc02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c1cafcf68de1adf6eef1378ed1aeac7d705bfc02"}]},"branch":"refs/heads/master"},"5d5d850f9ebc4f384d519ece795ba09dce6eea0e":{"kind":"REWORK","_number":4,"created":"2020-08-04 07:02:18.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/4"}}},"commit":{"parents":[{"commit":"031aa65b26263af24b3226c46d496953b11589fe","subject":"Pin pip to last good version - 20.1.1 for c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/031aa65b26263af24b3226c46d496953b11589fe"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-04 06:57:39.000000000","tz":330},"subject":"[WIP] Switch Standalone environment to centos8","message":"[WIP] Switch Standalone environment to centos8\n\nAlso add some os tempest options to fs052 and update\nrelease file for container builds.\n\nCloses-Bug: #1889219\nDepends-On: I67d37134372f9d71e4a5fc82740051b98cfdf3e1\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5d5d850f9ebc4f384d519ece795ba09dce6eea0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5d5d850f9ebc4f384d519ece795ba09dce6eea0e"}]},"branch":"refs/heads/master"},"e23fd7689676df721f95abe787433bf8ee81b0c7":{"kind":"REWORK","_number":5,"created":"2020-08-04 11:40:07.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/5"}}},"commit":{"parents":[{"commit":"031aa65b26263af24b3226c46d496953b11589fe","subject":"Pin pip to last good version - 20.1.1 for c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/031aa65b26263af24b3226c46d496953b11589fe"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-04 11:32:51.000000000","tz":330},"subject":"[WIP] Switch Standalone environment to centos8","message":"[WIP] Switch Standalone environment to centos8\n\nAlso add some os tempest options to fs052 and update\nrelease file for container builds.\n\nCloses-Bug: #1889219\nDepends-On: I67d37134372f9d71e4a5fc82740051b98cfdf3e1\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e23fd7689676df721f95abe787433bf8ee81b0c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e23fd7689676df721f95abe787433bf8ee81b0c7"}]},"branch":"refs/heads/master"},"0e1c421322d5e8b7b2d50d8ce3d85898abcd0b9b":{"kind":"REWORK","_number":6,"created":"2020-08-05 09:34:50.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/6"}}},"commit":{"parents":[{"commit":"031aa65b26263af24b3226c46d496953b11589fe","subject":"Pin pip to last good version - 20.1.1 for c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/031aa65b26263af24b3226c46d496953b11589fe"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-05 09:33:20.000000000","tz":330},"subject":"Switch Standalone environment to centos8","message":"Switch Standalone environment to centos8\n\nFollowing changes are done with this patch:-\n- Ussuri+ releases are CentOS8 only, switch the standalone\nenvironment settings to reflect the same.\n- Add some os tempest options in standalone environment\nfile to get tempest run with it.\n- Enable repo setup on virthost it will be needed to get\nrepos needed by DLRN run.\n- Add settings to not prepare overcloud nodes as standalone\ndeploy doesn\u0027t need those.\n- Also add some vars required for standalone deployment.\n\nThis environment file will be used to test standalone\ndeployments on ci.centos.org.\n\nCloses-Bug: #1889219\nDepends-On: I87a0214a344a9b23dd267e412a37b2abe0eb0571\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0e1c421322d5e8b7b2d50d8ce3d85898abcd0b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0e1c421322d5e8b7b2d50d8ce3d85898abcd0b9b"}]},"branch":"refs/heads/master"},"dd803ffab8058ab463cebf571a6cbab0f1444c88":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-08-06 05:21:15.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/7"}}},"commit":{"parents":[{"commit":"bde0d0084ce37a4e66f74e1c2164eaea06cb3b51","subject":"Add native variable to locally build containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bde0d0084ce37a4e66f74e1c2164eaea06cb3b51"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-08-06 05:21:15.000000000","tz":0},"subject":"Switch Standalone environment to centos8","message":"Switch Standalone environment to centos8\n\nFollowing changes are done with this patch:-\n- Ussuri+ releases are CentOS8 only, switch the standalone\nenvironment settings to reflect the same.\n- Add some os tempest options in standalone environment\nfile to get tempest run with it.\n- Enable repo setup on virthost it will be needed to get\nrepos needed by DLRN run.\n- Add settings to not prepare overcloud nodes as standalone\ndeploy doesn\u0027t need those.\n- Also add some vars required for standalone deployment.\n\nThis environment file will be used to test standalone\ndeployments on ci.centos.org.\n\nCloses-Bug: #1889219\nDepends-On: I87a0214a344a9b23dd267e412a37b2abe0eb0571\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dd803ffab8058ab463cebf571a6cbab0f1444c88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dd803ffab8058ab463cebf571a6cbab0f1444c88"}]},"branch":"refs/heads/master"},"2ac5748c5c71aedcb35ff0e9d286fc48bd4ee9f4":{"kind":"REWORK","_number":8,"created":"2020-08-06 05:26:32.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/8"}}},"commit":{"parents":[{"commit":"031aa65b26263af24b3226c46d496953b11589fe","subject":"Pin pip to last good version - 20.1.1 for c7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/031aa65b26263af24b3226c46d496953b11589fe"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-06 05:19:41.000000000","tz":330},"subject":"Switch Standalone environment to centos8","message":"Switch Standalone environment to centos8\n\nFollowing changes are done with this patch:-\n- Ussuri+ releases are CentOS8 only, switch the standalone\nenvironment settings to reflect the same.\n- Add some os tempest options in standalone environment\nfile to get tempest run with it.\n- Enable repo setup on virthost it will be needed to get\nrepos needed by DLRN run.\n- Add settings to not prepare overcloud nodes as standalone\ndeploy doesn\u0027t need those.\n- Also add some vars required for standalone deployment.\n\nThis environment file will be used to test standalone\ndeployments on ci.centos.org.\n\nCloses-Bug: #1889219\nDepends-On: I87a0214a344a9b23dd267e412a37b2abe0eb0571\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2ac5748c5c71aedcb35ff0e9d286fc48bd4ee9f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2ac5748c5c71aedcb35ff0e9d286fc48bd4ee9f4"}]},"branch":"refs/heads/master"},"fd010b945fc5335fd02b4795c8903fde3414eeab":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-08-06 05:26:49.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/9"}}},"commit":{"parents":[{"commit":"bde0d0084ce37a4e66f74e1c2164eaea06cb3b51","subject":"Add native variable to locally build containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bde0d0084ce37a4e66f74e1c2164eaea06cb3b51"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-08-06 05:26:49.000000000","tz":0},"subject":"Switch Standalone environment to centos8","message":"Switch Standalone environment to centos8\n\nFollowing changes are done with this patch:-\n- Ussuri+ releases are CentOS8 only, switch the standalone\nenvironment settings to reflect the same.\n- Add some os tempest options in standalone environment\nfile to get tempest run with it.\n- Enable repo setup on virthost it will be needed to get\nrepos needed by DLRN run.\n- Add settings to not prepare overcloud nodes as standalone\ndeploy doesn\u0027t need those.\n- Also add some vars required for standalone deployment.\n\nThis environment file will be used to test standalone\ndeployments on ci.centos.org.\n\nCloses-Bug: #1889219\nDepends-On: I87a0214a344a9b23dd267e412a37b2abe0eb0571\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fd010b945fc5335fd02b4795c8903fde3414eeab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fd010b945fc5335fd02b4795c8903fde3414eeab"}]},"branch":"refs/heads/master"},"82f94a5dce5ac0f2565c6e58ec98767a885731ad":{"kind":"REWORK","_number":10,"created":"2020-08-07 12:56:32.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/10"}}},"commit":{"parents":[{"commit":"bde0d0084ce37a4e66f74e1c2164eaea06cb3b51","subject":"Add native variable to locally build containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bde0d0084ce37a4e66f74e1c2164eaea06cb3b51"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-07 12:51:54.000000000","tz":330},"subject":"Switch Standalone environment to centos8","message":"Switch Standalone environment to centos8\n\nFollowing changes are done with this patch:-\n- Ussuri+ releases are CentOS8 only, switch the standalone\nenvironment settings to reflect the same.\n- Add some os tempest options in standalone environment\nfile to get tempest run with it.\n- Enable repo setup on virthost it will be needed to get\nrepos needed by DLRN run.\n- Add settings to not prepare overcloud nodes as standalone\ndeploy doesn\u0027t need those.\n- Also add some vars required for standalone deployment.\n\nThis environment file will be used to test standalone\ndeployments on ci.centos.org.\n\nCloses-Bug: #1889219\nDepends-On: I87a0214a344a9b23dd267e412a37b2abe0eb0571\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/82f94a5dce5ac0f2565c6e58ec98767a885731ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/82f94a5dce5ac0f2565c6e58ec98767a885731ad"}]},"branch":"refs/heads/master"},"949e393cb274e53fc05f082c62fed6b2bad39351":{"kind":"REWORK","_number":11,"created":"2020-08-07 14:00:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/11"}}},"commit":{"parents":[{"commit":"bde0d0084ce37a4e66f74e1c2164eaea06cb3b51","subject":"Add native variable to locally build containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bde0d0084ce37a4e66f74e1c2164eaea06cb3b51"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-07 13:55:44.000000000","tz":330},"subject":"Switch Standalone environment to centos8","message":"Switch Standalone environment to centos8\n\nFollowing changes are done with this patch:-\n- Ussuri+ releases are CentOS8 only, switch the standalone\nenvironment settings to reflect the same.\n- Add some os tempest options in standalone environment\nfile to get tempest run with it.\n- Enable repo setup on virthost it will be needed to get\nrepos needed by DLRN run.\n- Add settings to not prepare overcloud nodes as standalone\ndeploy doesn\u0027t need those.\n- Also add some vars required for standalone deployment.\n- Use git+file for tripleo-quickstart-extras path to\navoid quickstart-extras installation from pip cache and instead\ninstall it from specified path.\n\nThis environment file will be used to test standalone\ndeployments on ci.centos.org.\n\nCloses-Bug: #1889219\nDepends-On: I87a0214a344a9b23dd267e412a37b2abe0eb0571\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/949e393cb274e53fc05f082c62fed6b2bad39351"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/949e393cb274e53fc05f082c62fed6b2bad39351"}]},"branch":"refs/heads/master"},"814b080a60f08dc6d292138cb5a1603c6885672c":{"kind":"REWORK","_number":12,"created":"2020-08-10 04:12:00.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/19/743919/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/19/743919/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/19/743919/12"}}},"commit":{"parents":[{"commit":"bde0d0084ce37a4e66f74e1c2164eaea06cb3b51","subject":"Add native variable to locally build containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bde0d0084ce37a4e66f74e1c2164eaea06cb3b51"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-07-30 06:52:34.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-08-10 04:09:23.000000000","tz":330},"subject":"Switch Standalone environment to centos8","message":"Switch Standalone environment to centos8\n\nFollowing changes are done with this patch:-\n- Ussuri+ releases are CentOS8 only, switch the standalone\nenvironment settings to reflect the same.\n- Add some os tempest options in standalone environment\nfile to get tempest run with it.\n- Enable repo setup on virthost it will be needed to get\nrepos needed by DLRN run.\n- Add settings to not prepare overcloud nodes as standalone\ndeploy doesn\u0027t need those.\n- Also add some vars required for standalone deployment.\n- Do not use egg_info for tripleo-quickstart-extras path to\navoid quickstart-extras installation from pip cache and instead\ninstall it from specified path.\n\nThis environment file will be used to test standalone\ndeployments on ci.centos.org.\n\nCloses-Bug: #1889219\nDepends-On: I87a0214a344a9b23dd267e412a37b2abe0eb0571\nChange-Id: Ic17bf6e69b1620bdcca339f436e1399b29c0437e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/814b080a60f08dc6d292138cb5a1603c6885672c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/814b080a60f08dc6d292138cb5a1603c6885672c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
