)]}'
{"id":"openstack%2Ftripleo-quickstart~763747","triplet_id":"openstack%2Ftripleo-quickstart~master~I569732617bd74302529918a7f303f76ed0c2b23b","project":"openstack/tripleo-quickstart","branch":"master","topic":"bug/1903033","hashtags":[],"change_id":"I569732617bd74302529918a7f303f76ed0c2b23b","subject":"Enable gating repo if it\u0027s set from Zuul","status":"MERGED","created":"2020-11-23 08:23:27.000000000","updated":"2020-11-26 21:26:19.000000000","submitted":"2020-11-26 21:25:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"763747-bug/1903033","meta_rev_id":"f778a19a016892530b4a303b10ff8ca3477b189d","_number":763747,"virtual_id_number":763747,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-26 21:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-11-23 12:43:17.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":2,"date":"2020-11-23 08:58:21.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2020-11-23 09:03:17.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2020-11-26 14:10:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":1,"date":"2020-11-23 09:03:17.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2020-11-24 18:52:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-23 08:58:21.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2020-11-23 09:03:17.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2020-11-23 12:43:17.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-23 16:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-24 18:52:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-26 14:10:29.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"}],"messages":[{"id":"37403d1e724f209e88f439b49135abe5be3f9a77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-23 08:23:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e191040f60fac391dfc6e50373f893fc2cfdfd82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-23 08:43:01.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dedae9516ecd8338c2ec3ee11586c80bbb94808b","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2020-11-23 08:58:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"be2e6582498c58426585ec1844698753a1b54627","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-11-23 09:03:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e82b40868a0526f17abb46a96c5badbf3674d2a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-23 12:43:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2a7de406aed04aae9af9945af1b345d4 : SUCCESS in 2h 45m 01s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/d7e13495ea224f3babc9919e4a9e47a3 : SUCCESS in 2h 28m 59s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/8cada120f3684816a77a2c6a9be28c02 : FAILURE in 1h 29m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8df02e5b7f3be883bbc3923a0ba9ddd0226232dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-23 16:52:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1a59cc1ccf6e4684a2324adc4be5c52e : SUCCESS in 3h 57m 09s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/6fc0047a51e0479eaa60d82708ca35b4 : SUCCESS in 4h 09m 16s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/db34b40cce174d85acdefe10001e32a0 : SUCCESS in 3h 22m 01s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/181b77d186494b048a3081eb1a3ea29c : SUCCESS in 2h 40m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/56256f39236648708ba456a110b576bf : SUCCESS in 1h 32m 14s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/899832ed63bf4a288248b845f98a5523 : SUCCESS in 1h 34m 16s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1bad6f27588b439cadf2779a682be2c7 : SUCCESS in 2h 13m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5dc05ac39f9e467893a625735ab6caf6 : SUCCESS in 1h 57m 37s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/2bbe78cb9a1d49f98c4e42352efbffb5 : SUCCESS in 1h 55m 42s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8b8406588970459db5c701cf7913dc79 : SUCCESS in 1h 24m 02s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/d3e7656d9a774a77a0f2b25318c356b2 : SUCCESS in 3h 26m 28s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/6cd0eaa906cc45b8b3e3b40ede885b45 : SUCCESS in 3h 10m 38s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ab789b0c7eb14d28abf7dd5d72ce5778 : SUCCESS in 2h 24m 33s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/d6d1f73ec6ab441c80d2c0eba61a1ef5 : SUCCESS in 1h 52m 00s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3717253c9d6041e6bfee7c690233c66b : SUCCESS in 2h 09m 20s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/2bacca65e24d4cd08eafb761a0d5f9a6 : FAILURE in 1h 44m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ab5c2082f54c25b01ebe243adb43cc : SUCCESS in 4m 11s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/73394adf3ee34582befef811bae59279 : SUCCESS in 2h 48m 22s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4a33ed67e99342fa8c8421ca3dcdb19a : SUCCESS in 9m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ad53c52e1d34957a96d3ae4f7d3cc8b : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":2},{"id":"329fb8acbef10a5b6e63e4bd12f8619d01de6b55","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-24 18:52:16.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5d368234986ea15a90c26c0d4f3d3a5faa0d35a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 00:10:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"519fabaa5fb0fcd55cbe42a575adaa247788ad05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 18:50:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/368e19e4fbb444848b637866debec09d : SUCCESS in 2h 56m 11s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7de5a48990754b07909492ed276bcdef : FAILURE in 1h 07m 40s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/88c01705b23045d195348d711944014c : FAILURE in 1h 07m 29s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/00ab394a472640d38f4500708c2b64ff : FAILURE in 34m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e42925cceab24af59abee86deb51ca9c : SUCCESS in 1h 21m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/50a7694caa14431aa0a211d29f71da24 : SUCCESS in 1h 57m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4a1ac92022e441d19a8c08a9a919a3d8 : SUCCESS in 1h 21m 33s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/348f82a8b8794d5a9b77d7d615407173 : SUCCESS in 1h 46m 51s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b2f14daa87438aa46a62f66e84a7cf : SUCCESS in 5m 11s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ee65acb37b3f4ada94ab58e368d42a7f : SUCCESS in 9m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ce463924488543f99ab3c0fe1dad80a9 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":2},{"id":"ebc196e6123073ab723aeca8c0f498309409366a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-11-26 05:28:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"919184568284abe0090b09450305a95b7ed1e917","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-11-26 08:21:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3a6c1a4413c2492c9fa8f8380cfd6c2e : SUCCESS in 2h 33m 02s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-ussuri-branch https://review.rdoproject.org/zuul/build/484713319de949ee93b06dfbb1e91e90 : SUCCESS in 2h 50m 42s (non-voting)\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-vexxhost https://review.rdoproject.org/zuul/build/78929cf81804451cb6c46ed9eef5e8b9 : SUCCESS in 2h 44m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bc7c2abb8dbeb44a7ac7b2f97ad680ffc85413a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 11:31:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b0c7cc2735ee4521b8d1adada877e067 : SUCCESS in 5h 49m 43s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/2aa0a2c56f904461932568d341e54fef : SUCCESS in 5h 56m 54s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/dac32533fd224ec2bcc84f7d1a2191ac : SUCCESS in 2h 56m 14s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e08126862a464afb868fc9eb656a6613 : SUCCESS in 3h 30m 34s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5a2e68e00c5e490ab84696f4aa06f947 : SUCCESS in 1h 15m 16s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e61f3d3dbc7f411183447482a7511331 : SUCCESS in 1h 34m 53s (non-voting)\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fa65ca7f5a29448c83c52ce46ef07957 : SUCCESS in 1h 59m 37s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/618f0e1f3db14d86ae437a7cdfbb442d : SUCCESS in 1h 54m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/f5eb41b2c8f74315b942a7610c7fbdc9 : SUCCESS in 1h 56m 32s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c0d520bfaa39408d9c750cae07c2bf28 : SUCCESS in 1h 26m 41s\n- tripleo-ci-centos-7-content-provider-train https://zuul.opendev.org/t/openstack/build/9e4879f2acdc4f008a9c533dfc93b556 : SUCCESS in 3h 25m 24s\n- tripleo-ci-centos-7-content-provider-queens https://zuul.opendev.org/t/openstack/build/5f133ca16a204730b17222f7ced01993 : SUCCESS in 2h 48m 49s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/b9607c9e248b4457bf344039c1940c38 : SUCCESS in 1h 56m 34s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/afefff66d33c43ccbdf2afb40f964f2b : SUCCESS in 2h 40m 29s\n- tripleo-ci-centos-7-containers-multinode-train https://zuul.opendev.org/t/openstack/build/365f83241465476b8162a8c41798ec65 : SUCCESS in 2h 07m 19s (non-voting)\n- tripleo-ci-centos-7-containers-multinode-queens https://zuul.opendev.org/t/openstack/build/7ce034e5f536436ba9501d4b3c22916d : FAILURE in 1h 39m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa09d29f36cd4523a09c5f3079216094 : SUCCESS in 4m 18s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/b297946650a84dee9eec95d52e243ff6 : SUCCESS in 3h 21m 41s (non-voting)\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/832fce8e0b2c49df911aafbf435c5630 : SUCCESS in 9m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4c1c3e057a63408c86b5fb96151acbcb : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":2},{"id":"2b8a3770756068d09ee5f6459fd7b44361d88c15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 11:32:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"294aac9215b18c0d66495013ec0b3aac1a04fb53","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2020-11-26 14:10:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"97ddcf9b01b00683f8660d97bb445631050b3585","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 21:25:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/51a1892d5d1943ea99f388bdc8bdfc53 : SUCCESS in 2h 47m 32s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/20dbf6b9a9f743c497f7e9a33fdbaf97 : SUCCESS in 57m 57s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/38de9445fb524da18dd25c77b5bb3c48 : SUCCESS in 2h 48m 30s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/35cfdf0ecf814cf1a6ad39204d09c93d : SUCCESS in 2h 52m 36s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9af185b87b2b4420b034818fd4be1cdd : SUCCESS in 1h 44m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8bb95c33c8aa49009996d0d79a551d6c : SUCCESS in 1h 44m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9de6a43ac59d4f2792db6b9d145b43ca : SUCCESS in 1h 27m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/71ce010c777a4bc4ba6299041e228107 : SUCCESS in 1h 33m 47s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/045315d9878748d49facace98d9cadc4 : SUCCESS in 1h 47m 26s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/11380d7ec8e24380bb56a029dd0f10bc : SUCCESS in 2h 00m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a49477cf8e47948acdfede5c546efe : SUCCESS in 4m 19s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ca2449f68ca643b6bf7062fb8d999b1b : SUCCESS in 9m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1a6ec1c95be84fee84513f6518bce042 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"15219e9a2338fa3c052a1bbc49b3369165685457","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 21:25:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f778a19a016892530b4a303b10ff8ca3477b189d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 21:26:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a1553fc6ed14d189d63cbbd40afa234 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"457b109640eca4d7d11b77f774f053f730fe69d8","revisions":{"d36432de3a4e0f07fc4878d8779d8916db15981b":{"kind":"REWORK","_number":1,"created":"2020-11-23 08:23:27.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/47/763747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/47/763747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/1"}}},"commit":{"parents":[{"commit":"0af58227dda0f77ca5dcdfa31e6e8303cb5ff34b","subject":"Merge \"Fix nit in victoria release file using the wrong docker registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0af58227dda0f77ca5dcdfa31e6e8303cb5ff34b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-11-23 08:17:03.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-11-23 08:23:07.000000000","tz":330},"subject":"Enable gating repo if it\u0027s set from Zuul","message":"Enable gating repo if it\u0027s set from Zuul\n\nFor master/victoria it was enabled with [1], others have\nit missing and thus result in patches not being tested\nin provider/consumer jobs of releases before victoria.\n\n[1] https://review.opendev.org/c/openstack/tripleo-quickstart/+/746973\n\nRelated-Bug: #1903033\nChange-Id: I569732617bd74302529918a7f303f76ed0c2b23b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d36432de3a4e0f07fc4878d8779d8916db15981b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d36432de3a4e0f07fc4878d8779d8916db15981b"}]},"branch":"refs/heads/master"},"457b109640eca4d7d11b77f774f053f730fe69d8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-11-23 08:43:01.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/47/763747/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/47/763747/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/47/763747/2"}}},"commit":{"parents":[{"commit":"0af58227dda0f77ca5dcdfa31e6e8303cb5ff34b","subject":"Merge \"Fix nit in victoria release file using the wrong docker registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0af58227dda0f77ca5dcdfa31e6e8303cb5ff34b"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-11-23 08:17:03.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2020-11-23 08:42:54.000000000","tz":0},"subject":"Enable gating repo if it\u0027s set from Zuul","message":"Enable gating repo if it\u0027s set from Zuul\n\nFor master/victoria it was enabled with [1], others have\nit missing and thus result in patches not being tested\nin provider/consumer jobs of releases before victoria.\n\n[1] https://review.opendev.org/c/openstack/tripleo-quickstart/+/746973\n\nRelated-Bug: #1903033\nChange-Id: I569732617bd74302529918a7f303f76ed0c2b23b\nDepends-On: https://review.opendev.org/#/c/763711\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/457b109640eca4d7d11b77f774f053f730fe69d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/457b109640eca4d7d11b77f774f053f730fe69d8"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
