)]}'
{"id":"openstack%2Ftripleo-quickstart~777108","triplet_id":"openstack%2Ftripleo-quickstart~master~Ide514b7f2ea94a1a82bd414840a9adb0a79ade98","project":"openstack/tripleo-quickstart","branch":"master","topic":"ephemeral-heat","hashtags":[],"change_id":"Ide514b7f2ea94a1a82bd414840a9adb0a79ade98","subject":"Support using ephemeral Heat","status":"MERGED","created":"2021-02-23 13:46:28.000000000","updated":"2021-05-19 21:33:42.000000000","submitted":"2021-05-19 21:30:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"777108-ephemeral-heat","meta_rev_id":"f2c7c180892777fdcca216ebfb2a1f755be2dc09","_number":777108,"virtual_id_number":777108,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-19 21:30:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-05-19 15:14:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-05-19 15:05:17.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-05-19 18:40:03.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-19 17:25:33.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-05-19 17:25:33.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-23 16:54:51.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 19:07:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 19:26:24.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"CC"},{"updated":"2021-05-12 10:16:33.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-05-13 18:18:46.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-05-13 18:19:56.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-13 18:19:56.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-05-13 18:19:56.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-05-14 05:45:31.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"},{"updated":"2021-05-14 05:45:31.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-05-14 06:58:03.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"}],"messages":[{"id":"071caedf70db05635bb3c62373c7d6fc64dd47aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-23 13:46:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3654bd8abb4b2164b29bb62fa194c3c2ab8e1bb2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-23 13:47:35.000000000","message":"Topic set to ephemeral-heat","accounts_in_message":[],"_revision_number":1},{"id":"93217fd13b12a5281325d822e3327475bf80c12a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-23 16:54:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/67ab5754acab4717b24cfdb5578065e7 : SUCCESS in 3h 04m 26s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/f8f8d7b8cef344a6aa2583c9261ea89e : SUCCESS in 2h 43m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7859eb7f5e964d15aeb50f60a316fdae5b2c08e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:07:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c4a3d6b97d8746f791af76d77a960787 : SUCCESS in 3h 34m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fab58323a8ef44b9be9a1ed84ebf274a : SUCCESS in 1h 22m 45s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/abcc34566af2446dab8a99df0fa15b90 : FAILURE in 59m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/db8b138bc3984f798e86fb64609ed86d : SUCCESS in 1h 43m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9f8b290d47224e10919ef1c2b1081126 : SUCCESS in 8m 36s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/961c78270c53463faf42d51ffa145976 : SUCCESS in 3h 54m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a80749f7483f4fb48d097c11c6fc0d4b : SUCCESS in 3h 58m 11s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0151ad0dbe014414aa6bc00ec6cf7353 : SUCCESS in 3h 26m 11s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/5d308057adc146c3add0d6531e8f781e : SUCCESS in 2h 14m 47s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ecc4ff1fcee74526b2eb4f42f21a059e : SUCCESS in 2h 11m 01s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/281122e5d1fa451fbd985ae42b61e458 : SUCCESS in 2h 03m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e581149716c748dc9161145e0cc9e9e1 : SUCCESS in 5m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6c38e148f1a8418f97b75dc2739e0b74 : SUCCESS in 9m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d9d39c3928a94cddacdea57c80f4abc7 : SUCCESS in 1h 53m 22s","accounts_in_message":[],"_revision_number":1},{"id":"090d5bb3bbc0c52c97467826c97cf037e35f3d7e","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-23 21:52:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"acd63d6778a69f0e0a6e1f112dc86415b67d2fea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-24 00:52:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ef2e5287396c44c28e93f3ed67371d7e : SUCCESS in 2h 57m 40s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/c5d33f08c006415a86f9fa146eaed200 : FAILURE in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79b50514d685d7f78b7b314b9f7ca9a5dd2a0164","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 01:02:55.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"e005e81aedc14882cb1c104d9bc5ff11cf6afd9b","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-24 01:08:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a3f7584bf2423fbff3fc9ccf3b3390ec07767e9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-24 03:59:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9aa57e01060b49d89a5f45057eaf18cd : SUCCESS in 2h 48m 36s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/ba554c6fb4814f97be966fba51d55ec9 : SUCCESS in 2h 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4df798b0817de192c8af64941a4c3e0e9a4c98b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:50:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0f54ec3422444828b2432758b0168cdd : SUCCESS in 3h 22m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/601ab5ca3917434eaabd7ef800d191a4 : SUCCESS in 1h 27m 25s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5411d7d57e01422fb6c2199a274e5e3e : FAILURE in 1h 12m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/029ccade521947409fbc14f85ad8e8e2 : SUCCESS in 1h 29m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/175a145eb24d48e38843e9905fda18b5 : SUCCESS in 8m 33s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1d41cdeef38543e4864f255926a0d08d : SUCCESS in 2h 28m 33s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2ab9bde5ac1b443981b754b6af88bd66 : SUCCESS in 2h 39m 25s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9e0fd9bd958a40ff8a69c9830c1df627 : SUCCESS in 2h 04m 12s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/1b3d6b803d9e4a80b310c9aa9000edb1 : FAILURE in 53m 39s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/3277271ead424c8caac2c632e6c526e5 : FAILURE in 1h 10m 12s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/6ac5afcf224c46a8949da264d8a87e38 : FAILURE in 1h 14m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be7774a7f2849f6833ea084648cf9a5 : SUCCESS in 5m 18s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b45a62ef1b824660a23d6966d6d3ff8d : SUCCESS in 10m 33s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/53e98295a2724964a5e301fd2c89919b : FAILURE in 1h 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"25b0efa801ba2751a3090c33780289aeabf9d4b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-24 13:20:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"49fa643bb0de460435ca9d43dd2c5a24dfc60a33","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-24 16:01:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2a18490f38474a51a0330e2a10a1f713 : SUCCESS in 2h 38m 12s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/1dca3b0ea907496dacc8f421952e973e : SUCCESS in 2h 37m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89e0f0d10347a1b5333b56a42882d92f014fc854","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 18:33:22.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"f169b1b705e53625184552005d73030f20ec2f43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-24 18:33:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7668d65982e75fc515a277c7d14e14c96d257b1f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-24 21:17:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3e3ea4087dd34919af12c981872a9974 : SUCCESS in 2h 41m 19s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/de6217c5d106427e8e33b95a4ce11c0b : SUCCESS in 2h 24m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40dd0a976f42fa02d050cdbab8f7f5b8ca4a2f46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 02:16:03.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"4e1cab881525fc97f1f1c7a7475c8d0fceb7000a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-25 02:17:03.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"40c0b925e9ef6add496b92dec2532ac5ee919bc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-25 04:31:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ad7d360b9b744daeadb84f6a021b1b8a : FAILURE in 46m 20s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/1ada0594d8ba43cbb30d51c4cfb9e6f6 : SUCCESS in 2h 11m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a1c1118f696628b81ee8d6cb79fe44a2c76a6487","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 07:17:09.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3924bb6905344bd5b8a7e842a45af914 : SUCCESS in 2h 14m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7c906c457b52499ab4a7cf1d19c67ad2 : FAILURE in 29m 38s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/136cb4aced704dc383ba1bba6d2dc747 : FAILURE in 34m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/86bbbbe0a849497094bbd9b780e2b984 : FAILURE in 39m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af91d629254b490590d8003e798a5760 : SUCCESS in 9m 02s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b70d79b2658b48b6bb6f630064956156 : SUCCESS in 3h 23m 26s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/aeea608e2f4842a9b377ea3efeffd3cd : SUCCESS in 3h 38m 14s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/764695677daa45e1838caa4a3775e7fe : SUCCESS in 3h 17m 16s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/983cbafb1566499eb6f9317831ed2e36 : SUCCESS in 2h 02m 21s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/714f34b692164df9af1648888e5646b4 : SUCCESS in 2h 10m 29s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/e83fd6403bb24b2fa24c8bc5132e7a4b : SUCCESS in 2h 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4983a1a733147d49de428d3b54e5fe2 : SUCCESS in 5m 17s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2e49e3da810f447d9ae1b56c9b7f24d4 : SUCCESS in 8m 58s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0b0089d312954b2a908aac46b14c656c : FAILURE in 33m 46s","accounts_in_message":[],"_revision_number":4},{"id":"43a6c5c02205f7347883cdf7056f93d2f9b8b6d2","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-25 20:06:41.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"6aaa8fb26532e73dcc37946899322c55f6f9b428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 20:28:11.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"4ca4a1fdf484fc97a920b62864e30dd349da3367","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-25 20:28:42.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"e38cf51b6877f71fe8ba25ee823b1063d2b0bc5e","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-25 20:28:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c191c488d4ec7dc752bb14ca622b89b7bce295c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-25 22:42:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f6303166849243c0b4b19e557aeab9f8 : FAILURE in 48m 55s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/fbb840e0c8d94ed1b4d855260b59b72d : SUCCESS in 2h 09m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"35d9eb6afd85f9c31e424a723ab8de5977890290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:28:03.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a859c3373fe34d6590120d2f6629a194 : SUCCESS in 2h 07m 15s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2a1ba6f3cc434cb5828ce79a15b2af4d : FAILURE in 32m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/fb6d1e2923be40cc8810b4edbbaaee0f : FAILURE in 38m 06s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f4bc5c4f6baa42fda14c80d4ccacf661 : FAILURE in 39m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/185634d51ec349948a9a2a413a6a1b31 : SUCCESS in 7m 36s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/fc02e4c3d07840bca86aa2de21f65f14 : SUCCESS in 3h 08m 58s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0cca4f96435a49cba7e1f5ef8b216826 : SUCCESS in 3h 02m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/34ab4cb79e6a4ee487d12ee5c00db1b2 : SUCCESS in 3h 16m 42s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/fa57a27966fa47ca96bada374656d4d8 : SUCCESS in 1h 56m 03s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/a0a09230a9d94738aac0975e864ebfd8 : SUCCESS in 1h 53m 31s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f3efeeb53f864e13ab1130e71f3684ab : SUCCESS in 2h 06m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd28dd1b633548d7b4e42afd189d8aa5 : SUCCESS in 5m 33s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/21db35c3bc7b4c62af58fdf1fe1f611b : SUCCESS in 10m 33s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/41b14acb56ad4c68ad10a46eff413e9c : FAILURE in 38m 16s","accounts_in_message":[],"_revision_number":4},{"id":"aaff5097dc8dde121d9824cc6fd69a85f06912fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-26 00:30:59.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f4cc98fd29b2cbaa6860e3f9c19522fc915727ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-26 02:17:09.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"5a042d65e5e16b0e24ad97e7fa7078b43a69a4b1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-26 04:43:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/df6f1f6c86bc4fd7965d09e47ba44213 : FAILURE in 1h 29m 15s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/1fc48e49d37c4353b333976e6f7ccd94 : SUCCESS in 2h 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"af2f73ecbfe1c55758a7980f34c2dc68363b2005","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 06:26:00.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cea4344ef1e74966819e241038832500 : SUCCESS in 2h 53m 39s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d3cca007d14f41978d2d35f469790b4f : SUCCESS in 59m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/81897bfc11454fdfa4c47731f620a58d : FAILURE in 51m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3f64bd85659a4616a51cef05a08ff342 : SUCCESS in 1h 21m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/af54e4204b2640fabcc16f32b814886f : SUCCESS in 7m 20s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8a77bbc35a0b44d28930c179e08d59d2 : SUCCESS in 3h 55m 19s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7a37523931f74a30a524e8e358025e7c : SUCCESS in 3h 22m 12s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/41c11a43abc34449ba5e629fd0ab24c9 : SUCCESS in 3h 17m 28s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/0dbf06e764c8473cabcbc2e6697e623c : SUCCESS in 1h 35m 31s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/2dadd02e302a44ba9a061305b1f810ef : SUCCESS in 1h 55m 44s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/1199f698026f4e5f8bf48816e1cc05ad : SUCCESS in 1h 44m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed86251616b241688bc656510e96c59c : SUCCESS in 6m 05s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ac07299c164643a19dd0c8cecd5bad03 : SUCCESS in 9m 32s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d7bb32e7e37c45f0bafa72ff8f581898 : FAILURE in 55m 50s","accounts_in_message":[],"_revision_number":6},{"id":"82342e421d86840fc0ddc4fda1529d545f544c8c","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-26 13:52:12.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"8e5f78f28b7000aa09de899604929a93ed8dd9e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-26 14:10:08.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d728867fbd94e80efabc9d2f3ec22b9182614c8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:06:24.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d7d0ae452102498fb606a4472eb89292 : SUCCESS in 2h 42m 33s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/37dc98354da94424a07da71d6efa2493 : SUCCESS in 1h 08m 09s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/38bc0b1246ce4942b8254aa0aad240d1 : FAILURE in 57m 52s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/008e1777c85347f28a59b79803af310c : SUCCESS in 1h 27m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5d7365c2def4881a2d21f524817b7f7 : SUCCESS in 5m 58s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7e76662978ba4900a2b0ea8418418930 : SUCCESS in 3h 03m 54s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/49f195a7eb494ac4a279041066fe7911 : SUCCESS in 2h 39m 34s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/0e912c6100d649ecbe2e6d1a024fa307 : SUCCESS in 3h 00m 50s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/174e24bbc13a47459d177289b9c17d9b : SUCCESS in 1h 58m 14s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/3c35b056eb9c453b97fc833c82432ab2 : FAILURE in 1h 38m 06s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/473169af28ba4d8cabec419b5f74a7ec : SUCCESS in 2h 06m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74887012ee654037ad94085d5ff5fb8e : SUCCESS in 4m 13s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5feb5e06dc0a436db7a5fbc6ae8f9db9 : SUCCESS in 10m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d1869907d5d24e7399e775e1551b4e2b : FAILURE in 58m 43s","accounts_in_message":[],"_revision_number":6},{"id":"0756ab74326ba6a1a2d4b8de26d5468ca632f42d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-26 18:46:56.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d050892b060c01d39ad19b438cd02fe99e45a426","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-26 19:03:45.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f969402b40cc92cc19952b273671a872c2e6f947","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:28:55.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"310ec99145e50c68698fe87e1be1d93c73d24a2c","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-26 22:29:11.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"64bdcd7a97387157b6b7973d13f5b84b1bdba18f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-26 23:56:04.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8f8ee03b6b7b4b16b1ea85019400345b : FAILURE in 1h 24m 13s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/150f3777f4b44495a1dc0e69c6a63208 : FAILURE in 1h 15m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7067c21a6431a44bc15e58ca89d6214652997022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 01:52:11.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/be54ee9b962e42dca033dea6ba326e5a : SUCCESS in 3h 03m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b59be7d798ac47c7aced216e267fbb1e : SUCCESS in 1h 19m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f83a1ca1226e4e0eb6bd2cd1f7971265 : FAILURE in 1h 05m 08s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c755344ec50441059f7edb59f4b30a49 : SUCCESS in 1h 34m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c6d73935e35b4d71a1cda79b5ee3c1b1 : SUCCESS in 6m 41s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d5f3927e34764dba80c10c245a8e607d : SUCCESS in 3h 12m 03s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3f687721ebfa4a3f86b7d5077496c38b : SUCCESS in 2h 23m 06s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e3e4ba9e1c00423582606b8883854433 : SUCCESS in 1h 59m 35s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/965ac26f7b76415598387198b9ae1060 : FAILURE in 59m 54s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/d237df2ffd7c4c83aae4cd69889a92fe : FAILURE in 1h 09m 36s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a7a4962859474de89f4712b02767865f : FAILURE in 1h 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a46b5786e84f168491cd139151ed34 : SUCCESS in 6m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5d6189a15eb943548cdffca42b42efbf : SUCCESS in 9m 45s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/613e9baac5e847b9a5e358f6577af102 : FAILURE in 1h 04m 10s","accounts_in_message":[],"_revision_number":6},{"id":"7039e5099761abaa4d753771478b2b60b193066f","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-01 00:53:11.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"695f3c0598f03a2dc46a1d36ea45a90c1ab37ff4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-01 01:11:42.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"317e9037019303834f2bdc869517a61b18be63d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 03:49:17.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2ae86fcfd4614bacb07ae6a4c27a85c1 : SUCCESS in 2h 41m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bd3d1ed3d6644ea28b73e75abb0c5f26 : SUCCESS in 1h 05m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/75e92d96d5364d3abe7cc0a1bf41dd6a : FAILURE in 1h 39m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/54efff5b3f2548c69e0b245ae3c6cba3 : SUCCESS in 7m 13s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a7ec903364ab4e4d9efb13761b6bfce9 : SUCCESS in 2h 17m 12s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5d18e75aad6740f6a6bd084f2f0de243 : SUCCESS in 2h 10m 03s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4cfb07c3ec1f4946b39a2a23db7be2bb : SUCCESS in 2h 26m 29s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/e3e226f898d2410d805f578c0ea26230 : FAILURE in 1h 01m 53s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/650ec714056e4197b705e4beaf5cf992 : FAILURE in 59m 56s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a2e31f962371419e914aeb737f794e64 : FAILURE in 1h 21m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69385e63783c47f795a7528fc12d72a5 : SUCCESS in 4m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/948f2daa41d94b28b98025ec9823b50b : SUCCESS in 1h 13m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5a0ea86025e54ce1ae501acacde2e3db : SUCCESS in 9m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e62a89a761964918bff94465bd2a76bb : FAILURE in 1h 00m 19s","accounts_in_message":[],"_revision_number":6},{"id":"bd9bd2dcdf843b191be21f346188b6a91436ae59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-01 12:47:05.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8b1db680b8e8a49e3a795ef39bd4a1364c5322c3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-01 13:30:56.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4ea921fd3b034e8a9a62398b2f3926c7 : RETRY_LIMIT in 12m 25s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/774c6e85360943798b7aac69464394d3 : FAILURE in 30m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5d3a95e51648a935a2b072d070397faba608263c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:18:16.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b9b2b85855084a4cbac325a43fc494f1 : SUCCESS in 2h 54m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/65fe2295a5784ea3b4ee5e6499d76fcb : FAILURE in 19m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f033a830f0e74c8591b3fb182fbabc47 : FAILURE in 1h 11m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/32e2280337f74bfdbcb4ba40e4d17a05 : SUCCESS in 7m 33s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/699784167de94c0fa2665a47c8677367 : SUCCESS in 3h 14m 34s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ed90269bb34a4e2dbd99b150707a98ee : SUCCESS in 2h 46m 38s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/59cdd932c4d6475297196864b7d9f27a : SUCCESS in 3h 00m 06s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/9325aefe216b448f8f2d78d3314b6972 : SUCCESS in 1h 58m 30s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6d4e3278e4c44595a1953ed39a6a749c : SUCCESS in 1h 39m 26s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/1df1879f83d44e4f8958bd6e5c2d9dba : SUCCESS in 2h 05m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406177229cf749288036e9c750362fab : SUCCESS in 4m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/daa767cedbf74ea8a0e99ca58d66317b : SUCCESS in 1h 21m 05s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/04f2a12874464e07ae2e4caaa5897d8b : SUCCESS in 9m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a3570359a53f4ead861da4bf876be742 : FAILURE in 38m 09s","accounts_in_message":[],"_revision_number":7},{"id":"32f0c89c436cd16c17b6150898eaf4439a376026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-01 16:27:40.000000000","message":"Uploaded patch set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"2fa006a3b125c89ad4010f7e426f0b985f5d98fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-01 17:19:02.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/62857c463db343bc9d6a13d17c3a433a : FAILURE in 48m 32s\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-victoria-branch https://review.rdoproject.org/zuul/build/ca6406ea6189445bb54c54ff5b158b03 : FAILURE in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6b36c0be15a67a752af13fa4fc48163709966437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:01:44.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d1bcdce5356b49d59cc0e89020c29166 : SUCCESS in 2h 51m 43s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/74cae24ff2b1473fb3d52623ccbaae5e : FAILURE in 24m 01s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/062ff8d8587b49b9b43705387a8d9be0 : FAILURE in 30m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2927def3da3a486e8d6184812d4b0984 : SUCCESS in 15m 04s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/98eee1d363c94efe8e03886d63b6e51e : SUCCESS in 3h 05m 05s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4ce079a04eec47b79813d83cab667096 : SUCCESS in 3h 30m 14s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a9e082518a3444528fe0cd0bda738fb5 : SUCCESS in 3h 48m 12s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/93329f2549b34846b256a0beda0e5b27 : SUCCESS in 1h 51m 58s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/d220da126c104a89b9e2ee339f8677b7 : SUCCESS in 2h 20m 02s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/56d401e27c7b45b394615d7b1699f4fd : SUCCESS in 2h 30m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336ba4f643ab475e9008feab60abd398 : SUCCESS in 5m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/67c42c24e7504bbca9ac650a66c944f0 : SUCCESS in 1h 26m 50s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7cef0c739caf4ec69a0a34154d6bb814 : SUCCESS in 9m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dc1ce1524ea04623954341d62d2397eb : FAILURE in 28m 59s","accounts_in_message":[],"_revision_number":8},{"id":"8311b2e67caa580fb1d40f1798ce7ff5b0675a9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-02 12:48:06.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"0a84bdeb69f193929fb14ad6b93e3be45509e4e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-02 13:37:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5646b1fb47f84f06b89c6ebed859c609 : FAILURE in 45m 39s","accounts_in_message":[],"_revision_number":9},{"id":"581c7a0d1c3ed79f994df847b7f4fa4bd2e874ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:06:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5e124d41c6bf444a9dbe89d74fe0571c : SUCCESS in 2h 46m 49s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f041860c893d4e8f96744aaa9f7d14b5 : FAILURE in 19m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6783397224bf4054b6fd9f4e3bd2f992 : FAILURE in 29m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8882f484be464456947633b0c1d88faa : SUCCESS in 7m 40s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1c98d7ac2ca9402ba2d9d05d7e992a29 : SUCCESS in 3h 01m 01s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c8396c49cf9c4bf2af51ddc5dcc532e8 : SUCCESS in 3h 06m 16s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f107ad062b5c41fa916e627f077462ec : SUCCESS in 3h 05m 22s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/ba6dde8d116d4bb3bb92c2d947180350 : SUCCESS in 1h 51m 29s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/a2d0bba474f2422e93d9123c257a824a : SUCCESS in 2h 01m 50s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/4c364da331174fcba484d96487eb6d12 : SUCCESS in 2h 07m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16fdcae7c8c94fa9b9ddbbee61027bd6 : SUCCESS in 5m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/76a60923da9f4681bf944acede050d19 : SUCCESS in 1h 30m 01s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/b52cff95e6864c38af22a324331ef000 : SUCCESS in 9m 20s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1196b09455fa459ebce81d43d8fd9bb1 : FAILURE in 23m 08s","accounts_in_message":[],"_revision_number":9},{"id":"22061eaadfb7e15276f6ca1456f1405b5c1f4e32","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-02 22:57:59.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"d2d40c20bd583deaf0865038d9b26549e81ab896","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-02 23:50:14.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5499ed7cc0e14001a00319ed36e0b487 : FAILURE in 48m 43s","accounts_in_message":[],"_revision_number":9},{"id":"2d3e016a51c77b53ce5a7c6f01d6065164b561d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 01:38:42.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6879ef143795425bbdaa58d502f302a7 : SUCCESS in 2h 13m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f77504dbb4f54b9fb1c5179b4da13a85 : FAILURE in 18m 49s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/92ef2b02d0eb4853bf78e53e61ef7971 : FAILURE in 42m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00d5fda7835e441bbeff413bce8f8495 : SUCCESS in 7m 57s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/194d3089c0bb439da8aab48be4d1d5c1 : SUCCESS in 2h 29m 27s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/928f795ed9ab41eb87c339e81ada4ae1 : SUCCESS in 2h 24m 24s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/5d2b805f18c04af09b5bd17eb18b65b7 : SUCCESS in 2h 14m 55s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/a8ece41ff5b0404f9cf0cde55baf4c5b : SUCCESS in 1h 27m 40s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ea3aef660614412b89a1ad6b6da0e3e9 : SUCCESS in 1h 34m 40s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/e1c749edb2094c42894509a789f347ba : SUCCESS in 1h 33m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f24702e05fb4f69977e25ee54e74580 : SUCCESS in 4m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/05f8fdb6fe5e4508bc27f3a6158ce685 : SUCCESS in 1h 08m 21s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/baaec2a1d7324910981cf4082a904056 : SUCCESS in 10m 21s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1497135d45904348b2df889a92fd63ca : FAILURE in 21m 56s","accounts_in_message":[],"_revision_number":9},{"id":"aaacb7ca730b526fffb0b3fdbe0294076a40525a","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-03 01:50:58.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"3c4f89e440dc2c24aa1bc5becf5f2d4f09bf3cba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-03 02:42:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/07c8e6e0f38f4825a89f34df76d52cb8 : FAILURE in 49m 09s","accounts_in_message":[],"_revision_number":9},{"id":"883c3ede4e21b638a3aab0234e7130176b362944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 04:54:23.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4f5ddde38786446a9f5ec51b50ae4a2b : SUCCESS in 2h 37m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/111a512797ae4d56bc7c9864c17c7bad : FAILURE in 19m 25s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4632e051f2394783ad711335d090208b : FAILURE in 48m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8f697b76e0d84a9abb6c06bc98f74f09 : SUCCESS in 7m 39s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8e1119eb1ec844c18ca9e32c68153245 : SUCCESS in 2h 49m 19s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/7a27b3b68c0444d38fa835461f759e3c : SUCCESS in 2h 52m 23s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/fd1cdf403bc0494aa8187f942c92d62a : SUCCESS in 2h 43m 51s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8d9f19ce964f449fa0108920eaa454f2 : SUCCESS in 1h 50m 29s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/729960328191400dbe5fea2c9cf06646 : SUCCESS in 1h 57m 44s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3219a5f66e9441d086f780a0b28d6e2f : SUCCESS in 1h 51m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca36b10d47334d23bc914c093261f73c : SUCCESS in 5m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d3de97d7a5414aeba4dff331ac6dd470 : SUCCESS in 1h 23m 30s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/650af5d6c461410998c782b065048413 : SUCCESS in 9m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/40e8a7aa23904e6a8d46f0234e83b19e : FAILURE in 24m 09s","accounts_in_message":[],"_revision_number":9},{"id":"65d293aa3c1deb425c618f8021fc7610e34cc758","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-03 13:56:18.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"63dec93c1a8194144ecde2d332dde618169c2609","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-03 15:19:15.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"e0c9b91c9a729f22215114a200b964153ccc9724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 15:19:25.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"9d13c87d1431d2e67e0cdef2e12ccb481fa7259f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-03 15:20:15.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"e543d21f7e66304ecc8600accce8844e4184d04d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-03 16:59:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ace47926245e4145af318c0e6850fd73 : FAILURE in 58m 58s","accounts_in_message":[],"_revision_number":9},{"id":"e2e3f45bdabbcd7fc5f292cc284cb079024662cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 22:55:17.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/57a77e1a6fd74f7dbfcf68df47290052 : SUCCESS in 3h 32m 05s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/85e6af3a38cc469c97c116ebbc286343 : RETRY_LIMIT in 9m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/03d2370a343748f8b1700eac3b233074 : RETRY_LIMIT in 11m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ac5b83d7253b45e0bb420f6d545e9cbb : SUCCESS in 11m 23s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/29972da3880b4f1b9f1705652edb7612 : SUCCESS in 3h 16m 11s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5f683741d97d46838de64f9f921428e3 : SUCCESS in 3h 17m 14s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/8c374448fce846ce8605fd89f1441627 : SUCCESS in 3h 11m 55s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/2e5f61b434374170af5c70cc8bf7c326 : RETRY_LIMIT in 9m 34s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/75d731fb77c040df8058a40e262b709d : SUCCESS in 2h 09m 44s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/abb86e658e8e40c7904bc70b0bdbcc99 : SUCCESS in 2h 09m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6403d10a3694424996f4755686a88a12 : SUCCESS in 5m 50s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/561764988e484e2e936b4b24b52a9525 : RETRY_LIMIT in 9m 09s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/9940625882d34ff6bd658710df529eb9 : SUCCESS in 11m 04s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c2442da696f04d76b64d00753979225d : RETRY_LIMIT in 9m 46s","accounts_in_message":[],"_revision_number":9},{"id":"f8ee8eaa81db197da7dc353e4dc374e930354560","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-03 23:06:33.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"933a40ba7ec576ccf50f688e2f1ba6fbe0abe897","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-03 23:59:26.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2ff3ca16089f4d7b831765eb666dd3c6 : FAILURE in 49m 22s","accounts_in_message":[],"_revision_number":9},{"id":"7c8170eea165c34b2d5e8edab40f151445be9778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-04 01:58:06.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b8cd3649de57adb60ed44e14b7c17b3c995304e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-04 04:32:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/80feb36359d44da4913312d6f7c38c20 : SUCCESS in 2h 31m 56s","accounts_in_message":[],"_revision_number":10},{"id":"9045300f4c8b42473a73e99b14f4af1427652e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 05:58:26.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5aa0bf33033841d19c2ff99961bd9859 : SUCCESS in 3h 06m 30s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/11acba6089764ed8b4f819cd33e7bdab : SUCCESS in 1h 04m 25s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8705e04818fc427ab2259035dd43e153 : FAILURE in 53m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2e248012e8b24009a5e6f822b9cef472 : SUCCESS in 8m 05s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7845675f18bc432fbd52e37fd8075ac7 : SUCCESS in 3h 14m 05s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b0ade8910c004028867528447ff21289 : SUCCESS in 2h 51m 38s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1689756a7a0a4058930476e8390280de : SUCCESS in 3h 46m 42s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/1414375f105349458418405257705a6d : SUCCESS in 1h 58m 51s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/eb45e1b15b6e4885bc397e31fd149595 : SUCCESS in 1h 51m 37s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f56ac1a50fac45858b9771bf296aad64 : SUCCESS in 2h 20m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a51600a86e456e9d12ef5ff9b4a08a : SUCCESS in 5m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/19da1726ad5f4885a60bebbd6a0ad71f : SUCCESS in 1h 25m 01s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/000bbdbadf8248e396207a6b795293b6 : SUCCESS in 9m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fe58d35787884435bd77eaf216e33552 : SUCCESS in 1h 49m 20s","accounts_in_message":[],"_revision_number":10},{"id":"b877d2243746a468eff93f00b7685c9fee2bf43f","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-04 22:34:31.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"db2c6163234b6f99f949d8c74ed59708c8429af7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-05 01:34:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8b037fbb6eb843c89d8e55e8552d0e34 : SUCCESS in 2h 57m 42s","accounts_in_message":[],"_revision_number":10},{"id":"5eaa173e4f36ae5d17896275678eda8010307a30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 02:19:33.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b18f9724f4b240e0ada06b60b3bdbb94 : SUCCESS in 3h 35m 23s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fcaef741eee5442c9d40f91227dae024 : SUCCESS in 1h 11m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cb516b9dac14452e8b3ee243ea22715a : FAILURE in 1h 03m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b2a153de6aec4a75ba9f44f66fed6aa4 : SUCCESS in 7m 13s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/407d80014f364a7284be15ab6bdfe931 : SUCCESS in 3h 22m 25s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3c6c2caefab448e0af22b024095fcdd0 : SUCCESS in 3h 20m 35s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2162d82705eb471e8454c653937b0ec9 : SUCCESS in 3h 07m 05s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/f92f7e2c67674801a7d01a7599b6e5ad : SUCCESS in 2h 09m 21s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/e12a226bfa804ad1987d000943c40890 : SUCCESS in 2h 07m 26s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/e25b39e4f5f94983997db42c136ff85a : SUCCESS in 2h 03m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc6c820dc5e54c1595e2353ca4d84600 : SUCCESS in 5m 19s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5cc985b3b9c747d0b82c2c7cb6a7dc28 : SUCCESS in 1h 33m 52s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1444ccdbbc7146e6922d96bfd030438c : SUCCESS in 9m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/41dca3973b744a5d9e88787d9a0f8814 : SUCCESS in 1h 57m 28s","accounts_in_message":[],"_revision_number":10},{"id":"f46b2998198b7fc5a843b79e917dfbecf59288c9","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-05 03:46:19.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"24a81963d17d5782e9224f2d3d164ac065aa141c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-05 06:36:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/96e4276e42ef4a948bd2696222aebb24 : SUCCESS in 2h 47m 33s","accounts_in_message":[],"_revision_number":10},{"id":"5b2cf157d85cb8e2d781814122a12be2fdc5aca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 07:31:46.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1893a605bc77447fa088a051052304f0 : SUCCESS in 3h 16m 20s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/070d6586728545c6aea6bb0ff8d40982 : SUCCESS in 1h 20m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/07fa519faa694eb08531c5c5a453f3ff : SUCCESS in 1h 51m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f52fd0c815f043f98fa04799de5f854c : SUCCESS in 6m 40s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/e14e53ca3d5c4e97b4cf091982733181 : SUCCESS in 3h 15m 05s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3b4bb804165c4f27a47fe9e4731a45c4 : SUCCESS in 3h 22m 55s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a8bce12f522f45719cec322cc17aa765 : SUCCESS in 3h 30m 29s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/b91f42d03a214041bc230954ad151917 : SUCCESS in 1h 58m 23s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/a7f4ea64166d47aba9925f0f6efeb8bc : SUCCESS in 2h 05m 59s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/95988a8ecab440d9990dd9ce989fac92 : SUCCESS in 2h 19m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2f3ec717d54b2da06062242b042137 : SUCCESS in 4m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/de6c119be0c74e9f8504952176b93e68 : SUCCESS in 1h 31m 25s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/461bfee5e6cb4b2eb84f04beb3919cdd : SUCCESS in 8m 58s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c6c5b2ab80e7470a915a41eb6ab1e4a5 : SUCCESS in 1h 51m 04s","accounts_in_message":[],"_revision_number":10},{"id":"bff9dba41ccfd10e771a23d0d99cdfb0838f2af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-14 15:19:42.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8cd0641d5c31d475fed2692391efb0294d132188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 15:20:34.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"735619c2888145c6be4d4d5ea8ea372603b75f89","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-14 15:20:48.000000000","message":"Patch Set 11: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":11},{"id":"db54ce7e39b42b9f4ec0d75bfa54808dfcb4ba60","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-14 20:28:57.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"09a3e96645ecb6cba1c88ef7ea9aa61d0155e59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 23:41:58.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f02b8679960645f7bd7609da46587a51 : SUCCESS in 3h 01m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/90ad9d14fe5e4b0bb0353f3db8037afa : SUCCESS in 1h 12m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a0f499ec0ef943f9b0bf44f93d30cc59 : FAILURE in 56m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ed71cd834b7e4b988e404de4c9027059 : SUCCESS in 7m 21s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/47786adb1e114c40aa732280961b7d47 : SUCCESS in 2h 01m 22s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1898e14bcff74b2ba811563481888a1a : SUCCESS in 3h 01m 19s (non-voting)\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6cd82b77c674484e9f9a4f2e5f64fa09 : SUCCESS in 2h 56m 24s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/bec73ab7fe1a4b8d9ac52532989fbbc0 : SUCCESS in 1h 39m 49s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9536de79eeb947dda4b86a50c83ae8e3 : FAILURE in 54m 53s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/2f2e0a8da9b1402d914c31c76e231ed4 : SUCCESS in 1h 50m 23s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/b3f32de8fc1b4f7295bbcbdacc7aac41 : SUCCESS in 1h 48m 00s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/15be41d0faf848658bd8b58f4dae59f4 : RETRY_LIMIT in 10m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8302936a627541ba905f2fd50f3e410b : SUCCESS in 5m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c304f33df6c247bcb6c097971c8a8b88 : SUCCESS in 1h 26m 09s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/deccfae32a604f888c176a64de690b0d : SUCCESS in 10m 28s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/41397d7f379944278a1677537dd2df3b : SUCCESS in 1h 47m 23s","accounts_in_message":[],"_revision_number":11},{"id":"4bb9227189ebb2138911cc30a014e442ae6dfae1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-15 00:17:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2365a21ec7094dc59342708f8e80d9f3 : SUCCESS in 3h 23m 38s","accounts_in_message":[],"_revision_number":11},{"id":"a12d1af56b1b78a7957b79a7ac4c8c5a0b8c19fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-20 15:51:03.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"54983c0dab06f4d3adf98c4cec8803827067bd4c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-20 17:40:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9b01b77993c54ebe923f4ca2d66f233b : FAILURE in 1h 47m 08s","accounts_in_message":[],"_revision_number":12},{"id":"dd1e5173b9acecacf3afd833307c804ff3b0a752","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:54:47.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"77cf58c1a58356d2438c712b2a3e8b624734b42c","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-20 19:58:27.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"ad53a71b02e5ef4eed6c13cb8970956b1d789116","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-20 22:28:16.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/516dc6f1d4ef43c993763e1d0f8421c5 : FAILURE in 2h 00m 51s","accounts_in_message":[],"_revision_number":12},{"id":"9993cfefb664ab8b70536c65304df71fc89e1932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 23:57:59.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/df5cc1cbbfbf4e93823a899d49861900 : SUCCESS in 3h 11m 45s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1b10151deede457fbb6e34a3052f896f : SUCCESS in 1h 17m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2c75236fb4664420bb2244d2a9db4b92 : FAILURE in 1h 03m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/03c4209165fa4e45a5122bccedfc398e : SUCCESS in 7m 44s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e5fcb77ed2804d1ab8c3c040e8092201 : SUCCESS in 2h 27m 04s (non-voting)\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/73049596827f4bc98a4348e59072f31d : SUCCESS in 3h 12m 11s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/cf5ab7638a3c4a2a9941b090a70a79e9 : SUCCESS in 3h 41m 31s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e456dea6f65a4a16bfdfac1386f36ebf : SUCCESS in 3h 16m 37s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/7e6564965e6b4050a03104d349e39c90 : FAILURE in 1h 00m 07s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/019988d66e1c45d0af7543b63bbf420c : SUCCESS in 1h 50m 49s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/3c88c9f3e4744bbcac0c3b8c1bc2c557 : SUCCESS in 2h 15m 08s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/4156d402b1104082a087b87550ab4599 : FAILURE in 2h 04m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f619b016514533b39f4bb794d2e81a : SUCCESS in 4m 52s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a698dc95410e4c3ca18b5daa27a9836e : SUCCESS in 1h 32m 50s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c15c60f7eb744e258f094d580f71f50e : SUCCESS in 10m 18s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7fd68f823e6444619f6043cd6db4ad74 : FAILURE in 1h 07m 09s","accounts_in_message":[],"_revision_number":12},{"id":"d64afcc06638f729603d3bd099bcd767c34fc37f","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-21 02:25:32.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"1ef2f220ccf1ec4cdd1398b067e48024bd96b327","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 02:32:04.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"138a28c950b9580d98ac715b06f09cbc145c6d61","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-21 02:32:21.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"09f53cc5293890ba9b8fdcb790c4b7697910df8c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-21 02:33:12.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":12},{"id":"a56a597e8e6bfe37d37a90ee1ad2cdf43dd9a165","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-21 04:38:27.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d1e6d05267c94eba9a00bf8767509281 : FAILURE in 2h 01m 55s","accounts_in_message":[],"_revision_number":12},{"id":"34ede6595bef054c16a613ea9c5e7e60dc2c0b65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 06:39:26.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2810089af66e4bc19f328740597e7fd6 : SUCCESS in 3h 19m 15s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1c5cfd22ab004636bafbf77c79eab462 : SUCCESS in 1h 31m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1309ad28cb684384860568e89974b052 : FAILURE in 1h 08m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5418259b6dde4bf0a29b09c986ee3982 : SUCCESS in 6m 55s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/37aad612c33346a7b8ef61b53a420b86 : SUCCESS in 2h 30m 09s (non-voting)\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8856d7c659a34934af80fcff51d26f97 : SUCCESS in 3h 49m 13s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d02a8a547f754933a7a588a06f1cb382 : SUCCESS in 3h 49m 50s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/c08cc5cec1e345be91acaae4812e5204 : SUCCESS in 3h 27m 11s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/04d8e5d0776943f2908d36b99b395bf5 : FAILURE in 1h 02m 21s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/05ed0ea5b71b454b85dee511ca3a7d29 : SUCCESS in 2h 12m 37s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f084d9fe18e947db9cb65d38341c9639 : SUCCESS in 2h 16m 19s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/82467922dc0147229253e8870835f4c5 : SUCCESS in 2h 05m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4febae4e8a2142e58b6d9acd30d6adc2 : SUCCESS in 4m 55s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a445ad997b1c45e48adf2c7b52f84db1 : SUCCESS in 1h 35m 59s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ffa98b4d302841bba3501ac57844d686 : SUCCESS in 10m 17s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/543a6454a994482580e4b396cf52acb5 : FAILURE in 1h 06m 23s","accounts_in_message":[],"_revision_number":12},{"id":"323b152821bfdb5b0f304c3e8d906b8a70548905","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-21 12:04:21.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"d00c8c8bf0a58687d39c2697bc381656bcac9a99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-21 15:32:45.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"124605c69744bca1163245f11d1ef0781dfb14c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 15:55:15.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e1933f73c86c460fac3780bcb1d03600 : FAILURE in 8m 42s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ac38744e1b493d960d8a05c058404b : FAILURE in 4m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f8608080c10f4774a62796c35054e78c : SUCCESS in 11m 14s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":13},{"id":"673bc2da1116010552a8012befe1089941ee02c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-21 16:07:08.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"43bfc845e14cdecb2982217701879c643f7bc594","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-21 19:10:19.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ded6fe1ca7a04b03b568e06b78819c0a : SUCCESS in 2h 52m 30s","accounts_in_message":[],"_revision_number":14},{"id":"4b2f70ba432d849f24597bcb8fc644bbc4b72ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 19:48:31.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ed446b7193034a38a77da2822a3b3c2e : SUCCESS in 3h 27m 03s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/54157660df2f4c40b7e2f8a101eab1b1 : SUCCESS in 1h 16m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7c8e1d9330a2409f86e98f937d3bfac9 : SUCCESS in 1h 54m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/62b62a10fd3e4ffb93b2cb8f78312681 : SUCCESS in 7m 32s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c59f1cb7bf974f4fb4645c8b6d89281d : SUCCESS in 3h 06m 09s (non-voting)\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/2fba7575af704581ab22cbe416df9991 : SUCCESS in 2h 44m 53s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5e62039623aa4524abe263d81e4a9b84 : SUCCESS in 3h 01m 27s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b9c6148e3357439e8c19e287652c1ec8 : SUCCESS in 3h 14m 04s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/254332c3922e476cb24783e5cbe4bde9 : SUCCESS in 1h 54m 13s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/1e1ac3327a8849dbb44c7b63e9d956a7 : FAILURE in 1h 24m 30s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/14f388d6a5694c8ca17a212aadd8ff1e : FAILURE in 1h 38m 21s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/586532063d8b4b00b4b50b70d87b9a69 : FAILURE in 1h 57m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96deb2a545984586949fa27e78a9da4c : SUCCESS in 6m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c5d8b10df97644a2b2c409289a6bd9c7 : SUCCESS in 1h 29m 43s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/d8bd88d161a2496081192f0c689348c9 : SUCCESS in 11m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f68b9718286041ea9fb99d504fe4e604 : SUCCESS in 1h 44m 17s","accounts_in_message":[],"_revision_number":14},{"id":"f0fe6efce8f037eadb06f7dafa0c70428ec997d4","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-21 21:30:42.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"b4e9f1f4029f055327dadd12fee7424ed5e5ea3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-21 23:22:10.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6f34907962a0447bb23dc2d3a03474fe : FAILURE in 1h 49m 27s","accounts_in_message":[],"_revision_number":14},{"id":"7af111b5a6296b3e579be7cfe4b044c3644fbdfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 01:23:53.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f503e34b17974735939631012848e349 : SUCCESS in 3h 40m 36s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1e972ae51a8845d7b7f20602177db134 : SUCCESS in 1h 25m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/96164157d52045e79be95e1d6488fa34 : SUCCESS in 2h 06m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/61726dc0202744d89941eaa5e70afbfc : SUCCESS in 6m 44s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/523ab813e23b4da78e7c1d5c0fdd5db5 : SUCCESS in 2h 59m 52s (non-voting)\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/05b6f9289f17477c93644a544abe03fd : SUCCESS in 3h 17m 51s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f33865d820124932a4308a5abef71b79 : SUCCESS in 3h 03m 58s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a38a7380fa544ebabeca96a38bb56fec : SUCCESS in 3h 10m 15s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/34b184d499364a8dbe2b1902452df22b : SUCCESS in 1h 48m 42s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/474069b80daf40158877fd0490a580e3 : SUCCESS in 1h 58m 29s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/42ad6f680e5b4b3bab77f522f71be3e2 : SUCCESS in 1h 53m 11s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ddfc866c9a92473c96fab033671f9efc : SUCCESS in 1h 58m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f1e55525594f32aa0813ee1c04983e : SUCCESS in 4m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f475928e5530498a9b4a3953e579f632 : SUCCESS in 1h 31m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4f8cad6f3af54b2f924921c84d121541 : SUCCESS in 9m 41s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6b5da9bce84d41b6acafebaceeeffa37 : SUCCESS in 1h 58m 29s","accounts_in_message":[],"_revision_number":14},{"id":"3d39461e4c5ee7585f22efe82cd50e6e49fcfb17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-23 20:48:30.000000000","message":"Uploaded patch set 15: Commit message was updated.","accounts_in_message":[],"_revision_number":15},{"id":"2c40c3d2ccd78187131c2526d74d4f4ef9443b10","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-23 21:28:06.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/22566ec0a30b4b75ad786860705ae37a : FAILURE in 36m 56s","accounts_in_message":[],"_revision_number":15},{"id":"cb785bee73f8adb5f59f0fcfbe721bbe54d25317","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-24 01:14:49.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fe14af44604c412a96f3c5f97c3df7f3 : FAILURE in 33m 23s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa03b94680f64b109651b70aa9e02421 : SUCCESS in 8m 21s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/b85340904b7f4db682dabf3854cd7126 : SUCCESS in 2h 56m 50s (non-voting)\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ccd6784571fd44f18a44a369ca43ddbb : SUCCESS in 2h 35m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/61f6f6c363974254a8c737937d03ccd4 : SUCCESS in 2h 39m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/639dad4a699b48e3991774ff5db8678f : SUCCESS in 3h 43m 02s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/efae4bc2f946432a8f71d9ef63d623d4 : SUCCESS in 1h 51m 34s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/e69b04dd42bc4a03a5ee96089b61f525 : SUCCESS in 1h 35m 51s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/4c73d96b80f8480b9a5426f38ca120de : SUCCESS in 1h 41m 44s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/dd74d2a915a34d9097fc50aacc5e6b61 : SUCCESS in 2h 20m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf749a392a74e069f4390327fae644b : SUCCESS in 4m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/81380c8079a644c38ace24c1f66a7db2 : SUCCESS in 10m 37s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":15},{"id":"251a5b901c227f79a6a4fb28ada365df80feb614","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-24 19:01:24.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"aae50e72e609dfd908ceb07e1e160225868dc0ea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-24 19:53:35.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c99dc037340843ce9ba1cea38b6fac67 : FAILURE in 47m 24s","accounts_in_message":[],"_revision_number":15},{"id":"fa6cfbee22e54ed787766495e4e0ef4b52945a33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-24 22:43:18.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/903b349632c44a658284769874c800e3 : FAILURE in 31m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b74d5c8ed4744e9ba9a753c9026c992e : SUCCESS in 7m 22s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/0ad31bfed57b4a55969cfdd904bb8bdc : SUCCESS in 3h 20m 26s (non-voting)\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/223bef5c7d2d4537902954755d4b7662 : SUCCESS in 3h 09m 28s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b1f7951c8c1844739b9d9a7138c9149f : SUCCESS in 2h 56m 51s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a0c565bfa09c4bdeb53d913a882a5674 : SUCCESS in 3h 19m 55s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/162e86bf311747bfbb687050da1b2f8c : SUCCESS in 2h 17m 21s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/e0c8ffba0f5d4a0a9670275c03e4f6cb : SUCCESS in 2h 09m 09s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/a051f6a735134e3793de499234c73344 : SUCCESS in 1h 49m 02s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/d4cc0c7372ad42fe9b72ca0a9022c366 : SUCCESS in 2h 19m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc8898256ba4223998cfab5b06d2587 : SUCCESS in 4m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a6b314619b9e4dba8d711cc326945935 : SUCCESS in 8m 59s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":15},{"id":"084f929e9f3a8f815abd9435fe0042b6981276d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-04-29 22:29:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c5e610d605a7328aa8f93fcc9b5a9686a8965913","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-30 00:06:20.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ff9fb8fc5fcf403ea79a6fc408cb0c92 : FAILURE in 1h 34m 20s","accounts_in_message":[],"_revision_number":16},{"id":"8016a5ad8368c9b1722e4dbd7edd9607233591f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 01:43:26.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/94fad9fa1dbd44dab2fb4071cf6c3107 : SUCCESS in 2h 58m 40s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bc3cd267f049491c93fb126cbc887c5a : SUCCESS in 1h 14m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/28cebd5384ce4f1bbd32ef2cb5bf3891 : FAILURE in 59m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/493adffc7f3f46269c7173ab9443a794 : SUCCESS in 7m 27s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6e91ab570e0146f8927a837f122f8135 : SUCCESS in 2h 50m 11s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/2a11416eceac49f198aa863e5d2267ab : SUCCESS in 2h 56m 29s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/b2d6fd0c37944250916bb8c1175b9a0b : SUCCESS in 2h 54m 57s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/225751889b184cc38e246c42323ae87c : SUCCESS in 2h 54m 44s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/6ee4774d245a4d1a80fc88572598617b : SUCCESS in 1h 46m 25s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/cc484ecc94284533a0d104e3d6f37996 : SUCCESS in 1h 52m 20s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/2d372ca5d400438a9b5512a88b0fa48d : SUCCESS in 1h 52m 11s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/74af6e4c464f454e930dc3ae4bec00b2 : SUCCESS in 1h 58m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144558bcda7f4ad2aef3cce3b7104a11 : SUCCESS in 8m 50s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a454dd13244c44e5845e715024736d60 : SUCCESS in 1h 39m 13s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/268b7f37f28b4a048e99c4b00de53685 : SUCCESS in 10m 25s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/33f5d15b7b4147bda82ce5d48dc65da3 : FAILURE in 1h 03m 27s","accounts_in_message":[],"_revision_number":16},{"id":"f9d0eeb44ba8fe68efc90f79bd27a8471e337cc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-03 19:07:08.000000000","message":"Uploaded patch set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"ecb4805f5594c8d94b9c959520eee54adfba18fc","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-05-03 19:26:24.000000000","message":"Patch Set 17:\n\nyou might get faster feedback if you update the zuul layout temporarily to only run the content-provider and containers-multinode.","accounts_in_message":[],"_revision_number":17},{"id":"ee849e5063fb5c0f068965d12c4752bc2c6c91e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-03 22:10:02.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6f3d2526b36e4ff8b85a32925de7f33b : SUCCESS in 2h 41m 13s","accounts_in_message":[],"_revision_number":17},{"id":"3d92dc2cdd8b4befdf21c07fd8ad6c301bd7a247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 23:00:43.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f86db17289604266b8477f9f07307130 : SUCCESS in 2h 40m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a0a9b00fa54946f6880e1d6c8c8476e3 : SUCCESS in 1h 00m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/49e4acab71f04a2a8a9973c3a8eecc60 : SUCCESS in 1h 28m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/618d0426a59541c195d1f1e06a4e1d1e : SUCCESS in 7m 41s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f981e64a82ed4620b17bff3cc95bfd0c : SUCCESS in 3h 35m 03s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/86f733fa12ad4572abf868f009902a2c : SUCCESS in 3h 08m 01s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/1beffd749018464dac8474231fd331eb : SUCCESS in 2h 50m 47s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9472ce18d2264c8ebf5c8b0891805513 : SUCCESS in 3h 16m 48s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/8d727b131bb3489a9a895d68bdbd9419 : FAILURE in 2h 06m 35s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/214b12c5b6a24b01af14d2d4277cb0fd : FAILURE in 1h 38m 20s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/bd9228e24865432098dc825729d56047 : FAILURE in 1h 41m 02s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/8cb62a1751ad450ab5cb38b276195339 : FAILURE in 1h 59m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b7bfa863074d62a11d0a0e5dc67777 : SUCCESS in 4m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/df27c9f8dc0d40f2a9a5485ee14f8dc3 : SUCCESS in 1h 17m 21s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/1597f0410309465aaf7206f1937a3fc2 : SUCCESS in 9m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b865eb3679d04464b761b04d0c614a25 : SUCCESS in 1h 38m 05s","accounts_in_message":[],"_revision_number":17},{"id":"e9947603c6c4d0963b9f00222a659c92cfd8b2b9","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-04 00:09:07.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"fa2a73b653cc9fa22ad744a40e5992cd0bba245a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-04 03:29:29.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8fd4df3b42294e9fb4c48d189e1ca433 : SUCCESS in 2h 46m 03s","accounts_in_message":[],"_revision_number":17},{"id":"b0550501d1282f558d09383f3dfa5bffa4a41c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 05:41:06.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2e37db6c225846bd86ca96f33b442bd2 : SUCCESS in 4h 37m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4ce4103fbd5e474980d3a984653a64bf : SUCCESS in 53m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d6965a710a7f449da6a93c535c3bf18b : SUCCESS in 1h 30m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e352a92cd09b42528de5390601c6eb40 : SUCCESS in 5m 32s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/32cf25cefd9a4a56bf845dcf2301f0a5 : SUCCESS in 3h 01m 59s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/8da0fd48be92493eb583e1c8812b87d9 : SUCCESS in 5h 04m 02s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/cd8e3d944f714df0988d49ee2e924aed : SUCCESS in 2h 55m 37s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/3f0a273155744749a7417bcb39825174 : SUCCESS in 3h 06m 43s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/13eef675d5d44bcea16216bde8cd1bba : FAILURE in 1h 57m 30s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/95f8aa2fc286403c83a9a03d0f76a3fe : SUCCESS in 1h 35m 14s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/e8604773b50c42ebad10b61d2b8ba5fa : SUCCESS in 1h 53m 26s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/215517fcdccf421398df182bbbf4effa : SUCCESS in 1h 40m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1822cb75f8dd4a83af59c6f9ac84fd35 : SUCCESS in 4m 06s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c1448033aa7940cd8dd9f798ae36988a : SUCCESS in 1h 11m 26s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6d29759802354c9d8f122c3d2f1a9cc3 : SUCCESS in 9m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c0ee359460a748f4974449bd4e325d70 : SUCCESS in 1h 28m 42s","accounts_in_message":[],"_revision_number":17},{"id":"cc962a0a7b573e54cffc08c689fc682020a3b4d2","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-04 11:24:33.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"f14700bbc79cfea7262d97fce0a133e8427d1601","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-04 14:37:04.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/12512352898e439793967026694a3063 : SUCCESS in 2h 48m 03s","accounts_in_message":[],"_revision_number":17},{"id":"94cefa7eb884a85a2b3e9bda6909b19c513144aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 15:36:21.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c5774b20c6c84fe082f16c978a7733e9 : SUCCESS in 3h 11m 05s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bc9f73dad5dd40978889ebddfc0a0c3a : SUCCESS in 1h 10m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0c4a9fdec97d40e29b1450be58e102df : SUCCESS in 1h 41m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4fefdcce1f2b4d34b241fe5a0ee396ce : SUCCESS in 6m 51s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/79c187771c9b408cac032e0a84f6ec77 : SUCCESS in 3h 36m 43s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/2c5b1bb410b0408bb9533d39979aa385 : SUCCESS in 3h 48m 35s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/1c62cd02f1af464a83e4b4a489bf1308 : SUCCESS in 3h 30m 22s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/78c4e9ce353443c58376cff3ede01db2 : SUCCESS in 2h 48m 05s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/83989f03a0ab4cdba7c93702c9d0595f : FAILURE in 2h 06m 13s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/b1252438ec6b41abaccfcbc4997971d1 : SUCCESS in 2h 18m 33s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/db9ad01582d8445398e475cb3c747e54 : SUCCESS in 2h 21m 08s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a7b1c2acbad94868b14647318037b80d : SUCCESS in 2h 02m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b040caf56be4e21ae97d24c47944d57 : SUCCESS in 4m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/afd5146bc8cf4605b0f88dfe99889d40 : SUCCESS in 1h 18m 17s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c661be9e4b8b4578bd2831ff18558322 : SUCCESS in 9m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ed44aad860ad4bdb8e02226092880930 : SUCCESS in 1h 44m 11s","accounts_in_message":[],"_revision_number":17},{"id":"9c8cd677f85dc9e36565818c0a10098c79c47a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-07 01:00:43.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"77408f891cfc28442bc3918c041b1d5f7a5ccc1b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-07 04:05:08.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/60eb0b0f2f224cb680e8ac3a0c39c903 : SUCCESS in 2h 55m 32s","accounts_in_message":[],"_revision_number":18},{"id":"d21b25ce6a55acb8fe587237b2a70ac0dafa7de3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 05:19:26.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/efcdaf69464943f384585d397280c7b8 : SUCCESS in 3h 19m 55s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/55015b8bb1684cfabc7c0c6cda7fb88c : SUCCESS in 1h 08m 57s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f9def6bee41f470290ccd60fda3ddff8 : FAILURE in 58m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65ae55753a764bbdb9e870d93df2ef97 : SUCCESS in 6m 30s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/56b40bdc3f13469cae7e68eaff1b6212 : SUCCESS in 3h 11m 22s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/fd2c820d8bb04fa5ba49147c348feb50 : SUCCESS in 4h 09m 27s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/00ca06b22be84dfab4790ed76df50f3d : SUCCESS in 3h 05m 48s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7e0f5b8b28f043119189f1d672fc5d2e : SUCCESS in 3h 30m 36s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/8eef5020f7c54dcd894549606032da01 : SUCCESS in 1h 52m 42s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/7f21dbf9b83e45a08e8f04d334e0926a : SUCCESS in 1h 34m 05s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ec5a2a6e980d4daaabd21d613a5db8cc : FAILURE in 1h 54m 15s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3e5a81f266e349fd9bd1535f2360ba81 : SUCCESS in 2h 22m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9009048a6bf40d391b7f7bf461694ef : SUCCESS in 4m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c4da2803ea7c4b26a7527c528e957fc8 : SUCCESS in 1h 37m 31s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/522d5178ec074a8d9126acd56e84fa36 : SUCCESS in 8m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d6ac00a5fe5a474aaae69fb3ba0cbdc0 : SUCCESS in 1h 54m 40s","accounts_in_message":[],"_revision_number":18},{"id":"75780d23311fbffd04abc30716bb566929858223","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-08 15:15:50.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"61ef4f09b692fddaa7f215c5a326811680b3ae09","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-08 17:03:55.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1ae38fc713484a6ab8ec2ce8cd71fe48 : FAILURE in 1h 43m 32s","accounts_in_message":[],"_revision_number":18},{"id":"7aed7a9a4781d9f37d8b3547fa13de6f8b257e75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 18:53:33.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3f043a4d56a04d5d8f6a610c7a6e0d93 : SUCCESS in 3h 27m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/11b63a47b7a8405cae2257a0b696308c : SUCCESS in 1h 10m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6f482edb153f41a18105329717c7ecea : FAILURE in 1h 08m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/be3f133f0fce45c3a43a5cb6be79917f : SUCCESS in 6m 39s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5dc50a91b16346849783e3be11be5db3 : SUCCESS in 1h 59m 30s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7887929ebb2b4949b30dd4511b77d334 : SUCCESS in 3h 04m 58s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/e2d6a189a8554c6ab9fcb8af178f7303 : SUCCESS in 2h 57m 09s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/6149c9aaab62401b8a4903fa29b31a24 : SUCCESS in 3h 00m 41s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/2120865c34434d1b8afb15cb26ba3383 : FAILURE in 52m 26s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/4a128fac57a6455da68b28db02f67f5b : SUCCESS in 1h 56m 47s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6a513ad63ac04ffa974e0ddcd0867005 : SUCCESS in 1h 52m 10s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/71b9c86f3048409fb01d969e2fd44df2 : SUCCESS in 2h 03m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4355171269424d87d2113fb3fe8eb8 : SUCCESS in 4m 31s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6a4e99ac88bf46528562f69e83c5698d : SUCCESS in 1h 28m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/506ea83e4f5643afb6ae4b4136912138 : SUCCESS in 7m 58s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3bb20af4ce134389bfbfd0c2a3d93426 : SUCCESS in 2h 03m 15s","accounts_in_message":[],"_revision_number":18},{"id":"b3fdbf1c2b2021c38c5a9ce18cd8ef0ab4a5fe69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-10 00:14:11.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"bc6ebd23ea43f66ad76849011c23b097ea8298e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-10 02:43:36.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b872aa2de568439a809e573ac5b9d06b : SUCCESS in 2h 27m 13s","accounts_in_message":[],"_revision_number":19},{"id":"f2c3f678870a8ef69870b76bd2b3d0c0ec5211f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 03:51:16.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/718dc9a3223343a88e69c26032f307a1 : SUCCESS in 3h 13m 47s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2450c959dd794a1eb1eada74f96e3b05 : SUCCESS in 1h 16m 53s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b4e6a547195c49fe8a2c358d0934fb22 : SUCCESS in 1h 32m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d94f07b7b36749068b4fb524726aa76c : SUCCESS in 6m 17s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d5968f39f2024d96a33a86e88f4eed76 : SUCCESS in 3h 14m 39s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/b5ad057c5c8347b885cd3cc5a01c2a50 : SUCCESS in 3h 01m 57s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d223376d00394badab7c5db3f1c85d41 : SUCCESS in 3h 21m 35s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/5f8d0b6a26f743b9bb6241d037db10b9 : SUCCESS in 2h 42m 04s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/9934c9bbee7b4458b4804ae453e3a6b9 : SUCCESS in 1h 28m 54s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/062d93c6cec54bc1b2929bbd5906c7a0 : SUCCESS in 1h 54m 54s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/b6b2231cdaf0494fa11809691bef874c : SUCCESS in 2h 04m 33s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/76017cf32a6a44769e717248642e22c9 : SUCCESS in 1h 37m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e882bdc1f64b4134b12fef28b98b8b0a : SUCCESS in 4m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f737ad1b51b74907bf7c7401d0f8db8d : SUCCESS in 1h 09m 02s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/caaf42d230e0439b9bfb71fa5b42a95b : SUCCESS in 9m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9449d157b61d4b96ae379e3e62f2ed84 : SUCCESS in 1h 42m 18s","accounts_in_message":[],"_revision_number":19},{"id":"325cebdaa7cc625425612b47cbccd854aeb86462","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-10 12:17:52.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"5bdbf61bbb338b1fecdeabf981919b4ed1dbcb21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:45:48.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/76a55401e313494a9e08b3962b2ce52c : FAILURE in 1h 06m 10s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aba33f9a698e413fbdf69622b46ebf6f : SUCCESS in 5m 49s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/875905eaed96443a987638a3a3c5c3e9 : FAILURE in 1h 03m 53s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/269bbd88e6194a1ba58ac21f990be108 : FAILURE in 1h 06m 17s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/cd740b6d0862465cb06997e7b4fb4196 : FAILURE in 59m 23s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b178ef86314f4655bd5c7ba378194034 : FAILURE in 58m 13s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d511856c6bb465fbeb629741ce05bed : SUCCESS in 4m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/233b8397219a4ca6b1937060f6f48301 : SUCCESS in 9m 59s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":19},{"id":"ba250757e38eb4ccecef5d0f591ccd87dbf50a77","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-10 13:53:53.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"25d5832923df78da6f664babce60dd209c6d524b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-10 15:04:12.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cbe660fe55974b878bf8b82d487802c5 : FAILURE in 2h 20m 02s","accounts_in_message":[],"_revision_number":19},{"id":"50b6a91ba5b35c970cc70c528703988186013539","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 16:20:44.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cf95f50f40884b14b9c8823af37a7637 : FAILURE in 1h 18m 58s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/92cdbca6b9514e3c9f64b50e03beb34a : SUCCESS in 6m 38s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3cfcba2fd7fd4dc8abd18363d0008952 : FAILURE in 1h 25m 59s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d45c367c684a4b72a033d6d87bcb2290 : FAILURE in 1h 59m 15s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/122f373eab364e8aa90ed7e35891c8ae : FAILURE in 2h 14m 42s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/345cfde738914920886ea523c403d39d : FAILURE in 1h 10m 20s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab3e507a94b4023bd34c27f0d1b5aef : SUCCESS in 5m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/116c2540099a4527a5ab8bde71351a25 : SUCCESS in 8m 18s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":19},{"id":"369d60dde6681de552e58bb46e244aa4b3d89263","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-11 11:17:43.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"5dd91b8b541087015c864365001613fbb1abc95b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 14:36:08.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9166fbb8a80148aeb1e36028eeb6246a : SUCCESS in 2h 33m 29s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/aadc60dc14b24c8b9145afcca1e5f2ed : SUCCESS in 58m 01s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/180dffb7713e425bb64cbc2427ef9839 : FAILURE in 59m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1cdbb0c3ebb54c5db7fb18fbf7fd3bd9 : SUCCESS in 7m 01s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/91baf056b7994c2e9c1c38579f31d540 : SUCCESS in 3h 03m 59s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7bddd7e2417246f1be6d907c41bb3d2a : SUCCESS in 2h 45m 44s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0de0e51c8ce14e11ae820a8769f870ed : SUCCESS in 2h 23m 20s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e9b60c2ad9f44ca997bb868905be64d0 : SUCCESS in 2h 26m 48s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/ef660826b10e440d8944fc978a846ed2 : SUCCESS in 1h 51m 36s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/3f40bae876924fc687479b18aa9f8795 : SUCCESS in 1h 39m 03s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/33e85a30c7694b73bd4f1a1b1c9dfa57 : SUCCESS in 1h 35m 34s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/a6f6578255ad42818a0913c447e10dce : SUCCESS in 1h 40m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39af91ab840d49f48566e7b25467a2cd : SUCCESS in 4m 41s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9a53273add6f4498a3def007da69eff8 : SUCCESS in 1h 15m 34s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/9b9f6fca52804453acc555eb030047d2 : SUCCESS in 9m 41s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/95ec3d29d6ef4f2cb1705d923a48fe37 : SUCCESS in 1h 35m 55s","accounts_in_message":[],"_revision_number":19},{"id":"089dbfa4f5c79f04a199033e138efaaceae74d0e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-11 15:29:42.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cad58b7e9d4f4e38a84b66193f98ee4f : SUCCESS in 2h 57m 10s","accounts_in_message":[],"_revision_number":19},{"id":"134b54b3b34dc1fdc733f796e4ed6423e3594d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-11 22:12:42.000000000","message":"Uploaded patch set 20: Commit message was updated.","accounts_in_message":[],"_revision_number":20},{"id":"ca3e4465372b37b2ee643916a7690df6b33ab678","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-12 01:32:03.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1c7963dc9dbb4f6cab235a60b700d0b5 : SUCCESS in 2h 45m 30s","accounts_in_message":[],"_revision_number":20},{"id":"4eef00160482bc56d16b4ea19886879963c686e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 01:45:03.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9d3a56b4b06d486e892b041927c24787 : SUCCESS in 2h 38m 41s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c384fcac7a5f46a4b63ee7d7a1a6d16f : SUCCESS in 55m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/afc6a11f76ae4f0e9ed8734a5b5fe8b1 : SUCCESS in 1h 29m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/75f11ecd0eca4bf4b4093020be7e9708 : SUCCESS in 6m 46s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e68a8a78f3ce4b439bbf926af19aa2b7 : SUCCESS in 2h 33m 55s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a44659281afa4eb0b9a20ece9c7deedf : SUCCESS in 3h 22m 39s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/06eeeda195794ae29eb478a96fe745cd : SUCCESS in 2h 43m 01s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/b93c167c2ea1414aa00e5a8f8510c63b : SUCCESS in 2h 30m 05s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/2ad7f6825716450785a64d0f65079ec3 : SUCCESS in 1h 30m 37s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/b7606250b1fa477f88bbe4a934c62230 : SUCCESS in 1h 55m 23s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/988c4fe4f49b4c2aae72a7bb08b9bd5c : SUCCESS in 1h 39m 57s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/f983f7aabd2d4f79895802cdbfd1f587 : SUCCESS in 1h 42m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdae586fe802459a8e56aa7ef5ef994c : SUCCESS in 4m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0737b76a79664a0eb1bdb71631d01b66 : SUCCESS in 1h 09m 46s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8311bd1ba3b5448babc27c90577ff7c8 : SUCCESS in 9m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/963573dfb2fa470c8661d43703436b1b : SUCCESS in 1h 32m 54s","accounts_in_message":[],"_revision_number":20},{"id":"6ccf224a9da889a23cda3a48257341a5750b424d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-05-12 10:16:33.000000000","message":"Patch Set 20: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"5bc7269d13f8b7dbdf6c3b51c4fc4cb508205738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-12 11:36:20.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"57e60c9620577aec8247872c85528da2de11fb5d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-12 11:36:35.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"cbf4b7eafd97bf9a14ec2ebd2477c5479e2e5762","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-12 11:41:26.000000000","message":"Patch Set 21: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"a429bed34d0179ded5f15e3012983271adc3634a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:41:55.000000000","message":"Patch Set 21: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":21},{"id":"28d1a4a9cc1d06e391e451c2aa6f4b0c5e6e8a82","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-12 11:42:15.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"b637fb03e6a89db5e07abd06a2ba4fe2e4df4651","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-12 14:07:41.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/dcbb154dde9e43c0a3755521aa896884 : FAILURE in 2h 22m 01s","accounts_in_message":[],"_revision_number":21},{"id":"bbafec9ef348c670463088cebad39c2f7714ae9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:49:40.000000000","message":"Patch Set 21:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b7479c9a73ad417d866162beb0ae436d : SUCCESS in 3h 31m 41s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/562e4e2d7fe24068956522b67a3ec07f : POST_FAILURE in 1h 14m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9bc3d4740e5240f28a9b9725b23fd325 : SUCCESS in 2h 00m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8f6d92a6ff034278901ba4e08125f59a : SUCCESS in 6m 47s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e32930a7dc0a4ab59b52706f2922e53d : SUCCESS in 3h 17m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/356d0245433f494d8abe6cbf18c2cc39 : SUCCESS in 3h 54m 23s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4372384c75f047878b9aa61663185561 : SUCCESS in 3h 40m 20s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7421b25bb29c4814921616753b9a9a82 : SUCCESS in 3h 46m 40s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/108bb5fc42c44741af4332b3d61855de : SUCCESS in 1h 57m 24s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/e8830ffc926745f9ac76ccdd1a5d06a4 : SUCCESS in 2h 20m 07s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f0baed62ec6a4c80ab66ea58e481883d : SUCCESS in 2h 20m 50s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/2b53ffea1d3c4974b003924b018fc74e : SUCCESS in 2h 32m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60f54cfb06d45d89661127f501f0e7c : SUCCESS in 4m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/112ba513849e4a3c8eaa3cc0961590da : SUCCESS in 1h 31m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/46794d26ccf04e3b8724fd5e5671be42 : SUCCESS in 9m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e4a36f5a83f14db9949ff906bc8cbd4c : SUCCESS in 1h 58m 01s","accounts_in_message":[],"_revision_number":21},{"id":"f6959ff1a4595f4dcbe89e68227fc86b8522de17","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-12 18:14:12.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"1133c6c859fcf3bd71afa65abaa2dd116ff334d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-12 20:14:46.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"1ae2c2fe7624d6851ecb793c7ed221c24c9e2751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-13 00:21:15.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"f2c04f97cc44dcc56b342105727974ed781549b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-13 00:23:26.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"d6090347be3dc34324081a10d1a6e6051523a9c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-13 03:08:37.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/59741927ef5e4c7ba6e6c52671449ae6 : SUCCESS in 2h 43m 23s","accounts_in_message":[],"_revision_number":24},{"id":"d60a60c03caebd03059cb03cb8c652d93658250d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 03:52:11.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fc69af724f1144c8b8f0b9ad7f4098dc : SUCCESS in 3h 19m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0480e66f2aa84788b2a0eaf97814fb39 : SUCCESS in 1h 06m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/16c417e4b46e489bb841fd12ab75ea34 : SUCCESS in 1h 40m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/406a572933ca4d28940dcb36f25dd794 : SUCCESS in 7m 03s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f984782a9e0946479141c86a2f826da3 : SUCCESS in 2h 19m 57s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/dc5d06e04fea41abad8cee2e32c4c5d7 : SUCCESS in 2h 49m 17s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/8ae8201ee5f24076be4479962476d9ad : SUCCESS in 2h 49m 55s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1161e212d56747c780d55fdb198b659d : SUCCESS in 3h 09m 33s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/dc35e8b61666469c9cae294bd87aaf89 : SUCCESS in 1h 26m 42s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/32b5fdece46b4aaeb1f743503679d7dc : SUCCESS in 1h 45m 29s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/769ebec4fae44db49b7d5b73b2bbf772 : SUCCESS in 1h 40m 35s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/854c20ece9da4c2b95c7844b9cbf4be6 : SUCCESS in 1h 52m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b5c42db01943bd93c243ced1318ad8 : SUCCESS in 4m 01s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5f289e7d58784a98b10afb460fba7168 : SUCCESS in 1h 23m 22s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7d3186dbe9cd4e68b64f09f8ba601f65 : SUCCESS in 8m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6a383ecf7b294248995fb975c0739e4d : FAILURE in 59m 41s","accounts_in_message":[],"_revision_number":24},{"id":"5b8d3d72310233922e99a33955a79a5c63cb30d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-13 12:49:10.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"42b62f1cc381a8b16ea4a0813eb56202df92d205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 13:19:56.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c76c0328860140c9a195e0897c907ddd : FAILURE in 15m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/970b572b8bc8482497ae6d8ea96c766e : SUCCESS in 9m 08s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/0121274597bd4050bad9c0c108a20c51 : FAILURE in 14m 33s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/36b26e01509646dd85c6cc18bc4b1426 : FAILURE in 14m 48s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/0dcba7f4666e42df94ffec6528736c41 : FAILURE in 15m 06s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/76bb994c4c07476ba95d4028c2b9e7b1 : FAILURE in 12m 28s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de648ac5a1584b109b9966f07d38e401 : SUCCESS in 5m 08s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/7ecdca90601b4b5c9876ebfde8790a27 : SUCCESS in 10m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":25},{"id":"d0128f904d788400705172e03300319765f08fd8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-13 13:44:49.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8f6d4eba97c048e08e4793afee391d5a : FAILURE in 23m 17s","accounts_in_message":[],"_revision_number":25},{"id":"f158a5f3c5565c3364b527853cb54eeca801d42e","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-13 13:48:27.000000000","message":"Patch Set 25:\n\nrecheck","accounts_in_message":[],"_revision_number":25},{"id":"7b6f36fc641fbd35d57717c0b2e7a17651743181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 17:25:25.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/16a9c49149be4196adb21763ef0c602c : SUCCESS in 3h 24m 14s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cf7b839d53024eaaa8d82661a5443a82 : SUCCESS in 57m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e0c072f8cc42461cb78e894d4381b08a : SUCCESS in 1h 37m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ecda7eff4acf43c28ff275f46519d6d4 : SUCCESS in 6m 30s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/dc2a343aaa2c46cbbd1c5564bc77950f : SUCCESS in 2h 36m 35s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/22cff6827d814a4e8c9b60bbe84290e2 : SUCCESS in 3h 16m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/97ec7040dea34a43a68ed24b48581efc : SUCCESS in 3h 20m 03s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/cb794d447c5e47629351df2551e39d4a : SUCCESS in 2h 38m 44s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/e5e19bc7065146b1bed261d487f2c5c5 : SUCCESS in 1h 38m 33s (non-voting)\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/b5b693fb383f463182767af32f1c1631 : SUCCESS in 1h 42m 46s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/a0e37a48c7c84a4f9125a3ebbbaaeb57 : SUCCESS in 1h 46m 34s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/8e121a91494c44d489e5e6a88f825f62 : SUCCESS in 1h 41m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c0f24ed2d34873b19731666dd46e52 : SUCCESS in 5m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2c6b0fa1ae1b4fdc8c7e453cddfef5f7 : SUCCESS in 1h 14m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/47dc24ef339b479692a297a85d6efc97 : SUCCESS in 9m 06s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8a56f7a5935641fda6c51255edd0b01b : SUCCESS in 1h 41m 46s","accounts_in_message":[],"_revision_number":25},{"id":"dfac174901dc437256a430e57e560d5a09f3aa46","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-13 18:29:17.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/78d9391273784348909dfef40b6df821 : SUCCESS in 3h 19m 17s","accounts_in_message":[],"_revision_number":25},{"id":"3be24f0e62273869fabaa407968173a81904aa98","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-14 06:38:15.000000000","message":"Patch Set 25: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"0c2061b1469107899e9a843578b00f0dbdc3e32d","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-05-14 06:48:32.000000000","message":"Patch Set 25:\n\n* \"enable_heat \u003d False\" is correctly set in undercloud.conf\nhttps://f1ad14c2f18111e05a25-c392e4904ca041b2375209c5c066ffc0.ssl.cf1.rackcdn.com/777108/25/check/tripleo-ci-centos-8-containers-multinode/e0c072f/logs/undercloud/home/zuul/undercloud.conf\n~~~\n# Whether to enable Heat in the Undercloud. (boolean value)\nenable_heat \u003d False\n~~~\n\n* No heat container in \"podman ps\"\nhttps://f1ad14c2f18111e05a25-c392e4904ca041b2375209c5c066ffc0.ssl.cf1.rackcdn.com/777108/25/check/tripleo-ci-centos-8-containers-multinode/e0c072f/logs/undercloud/var/log/extra/podman/podman_allinfo.log\n\nhttps://f1ad14c2f18111e05a25-c392e4904ca041b2375209c5c066ffc0.ssl.cf1.rackcdn.com/777108/25/check/tripleo-ci-centos-8-containers-multinode/e0c072f/logs/undercloud/home/zuul/overcloud-deploy.sh\n--heat-type pod in deploy command","accounts_in_message":[],"_revision_number":25},{"id":"db6f333e1a0ddf1414b3c4f8adb9f3dfbac67699","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-05-14 06:58:03.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"b2bf9e9a069b01de5390fba49b8e68762d8e0570","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-14 12:06:05.000000000","message":"Patch Set 25:\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"296819f804d960962d4dae35acd249c990ea09a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-14 12:06:11.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"9027770040fea1bb1026548086f573a874d6aefb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-14 12:14:34.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"88edc61679e2f33c36a8feea980f874b6e0977db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 12:14:59.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"661414ffc34c7ba9c9995ee402806e837098ee8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-14 12:15:39.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"08891f7ddfdf12312a3605c4b029fd869a7630a8","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-05-14 13:27:41.000000000","message":"Patch Set 27: Code-Review+2","accounts_in_message":[],"_revision_number":27},{"id":"c0ba3b3fc4aa6b75637c1c2da68d4b3ebcdc2e22","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-14 15:28:03.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/76af8f5c1e5644ba9aa3b38f1353e633 : SUCCESS in 2h 46m 00s","accounts_in_message":[],"_revision_number":27},{"id":"82e07ca78f987c4f8e67b0aba5e43bacd66142bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 15:38:24.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9d6443512c09407d8db67648b9e31b7e : SUCCESS in 3h 02m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/04e7c3b096b249f19a29eaa1546538c3 : SUCCESS in 55m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/331fd7ac9dc94b1595af43987c77b1db : SUCCESS in 1h 27m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/059941ad0c354468bb828c9c093f96e3 : SUCCESS in 6m 29s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a3d5a38c0e5046d18fed778d7b25fa9b : SUCCESS in 2h 52m 04s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/86a40cc9e3204716b0421082b2cfd919 : SUCCESS in 2h 42m 01s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d209b83f80f540e995c930b8a09115e9 : SUCCESS in 2h 58m 34s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/88dd50f4bd834565b1fd7a81c9587c00 : SUCCESS in 2h 44m 31s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/47c5e68535c14a3e9f63b5ac1428ef81 : SUCCESS in 1h 43m 52s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/621308c9c635434bb29e2638f52ff057 : SUCCESS in 1h 43m 36s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/e657e60e7fff4c24bc7ac0521ab87e8f : SUCCESS in 1h 51m 21s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/7042c63f46a84ba9affa6c3193b863f0 : SUCCESS in 1h 56m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4343b14f75bc43d28309951123a7605f : SUCCESS in 4m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3325ae8e45134f23a3e54b915985057d : SUCCESS in 1h 09m 04s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8899506e6e6d41c292d6d589dc471fd7 : SUCCESS in 8m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/51c7a950d5f04a7ba904fd73980a1499 : SUCCESS in 1h 58m 27s","accounts_in_message":[],"_revision_number":27},{"id":"63c76a42ef445d8062c68a6f4e5e897893d32845","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-14 15:43:08.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"eb5f54422079f9e3d7f91fee525d46395a008db9","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-05-14 16:18:00.000000000","message":"Patch Set 27: Code-Review+1","accounts_in_message":[],"_revision_number":27},{"id":"f8adc18bbdaf0888b1629e2ceb5a8d28ac1b5aae","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-17 06:39:12.000000000","message":"Patch Set 27: Code-Review+2\n\n(2 comments)\n\nk i added some comments in the commit message but only noticed it was workflowed when I came here to vote so +2","accounts_in_message":[],"_revision_number":27},{"id":"9b16d5aabe5798057af87e3b904ccfe3175956f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:13:10.000000000","message":"Patch Set 27: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":27},{"id":"f55285aa610b56fc97fcae25d48c2d1c5e94ca6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 12:16:13.000000000","message":"Patch Set 27: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":27},{"id":"c165beed7336c76979e27c4398c955651a18417e","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-17 12:31:13.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"5d3cd7eb92c98c405a6b5b1d19f12c628faf18fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:18:56.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/febbe8cdc0f643909b315db0f0223d06 : SUCCESS in 3h 11m 37s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4ba71028a569428185fae9ae1d3b8c71 : SUCCESS in 1h 08m 37s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c929410b89db4507a17b78edae7a44b6 : SUCCESS in 1h 53m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cde42dd7f3fb477a855c43175b814b97 : SUCCESS in 7m 00s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/16dcb0ee4699447ca8ae436f2c2d8629 : SUCCESS in 2h 59m 31s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/6912e5a43562484fb63cfb1eb98a0f70 : SUCCESS in 2h 40m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d529ef4a258b4513a19581494f3bb54d : SUCCESS in 2h 43m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/cdeb54a3943a430daeb0199a3173de91 : SUCCESS in 3h 36m 26s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/41dbbc4e2de543459ea04df88934f258 : FAILURE in 1h 16m 59s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/faec0e61903e4fa98ccfb5e9a85aad1b : FAILURE in 1h 28m 47s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/8ef7632dd90b4b549ef65193cd40f916 : FAILURE in 1h 28m 12s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/d360e9c502e04135a72ef73d19deebdb : FAILURE in 2h 07m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f6cb8476e546c79ca36d5b50d55c25 : SUCCESS in 5m 31s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/727b7e6ee6e541a0ac1889950ddc31ca : SUCCESS in 1h 31m 57s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/8d6e4d650721493a8d42d9b853d10dd8 : SUCCESS in 10m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9a35fb725778472cb8908e82bf434223 : SUCCESS in 1h 58m 19s","accounts_in_message":[],"_revision_number":27},{"id":"2b7960d8898bd3019952111842b766f433b72c70","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-17 16:35:40.000000000","message":"Patch Set 27:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c49db5ff46884c448e73e5ae55fb3fcb : SUCCESS in 2h 40m 52s","accounts_in_message":[],"_revision_number":27},{"id":"4a049906f412ab2ebe5b8976cf7d8f1ce42d6bf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-17 17:17:16.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"6dc6c048027ac070acd85cb89b4c0494e029fae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 20:35:04.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b6190de0d709446889cac07c01385919 : SUCCESS in 3h 06m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/300e3b8102a54fbd80cd3379d6e5c045 : SUCCESS in 1h 07m 06s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/52689c1f01d24a178b82d9629d333c2f : SUCCESS in 1h 47m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4367e263e05b4e0fa1ac344ee71fe18f : SUCCESS in 5m 19s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/bb05441b1f1a494c8f5306a67d5d00ed : SUCCESS in 2h 22m 15s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/bc07d04b4dbd40458e2d8304ea31c396 : SUCCESS in 2h 21m 27s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/972365cfe1b54a7489f0eaca7dfce099 : SUCCESS in 2h 24m 39s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d076633a883545cfa9be5421e28b645b : SUCCESS in 2h 40m 14s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/d82c77660a994d01ba4f393753c017f4 : FAILURE in 1h 10m 02s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/9aef7900a1314913a393d1020f725dec : FAILURE in 1h 01m 39s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/18fc8cdfd86c43b0bd36498cfe0d173f : FAILURE in 1h 02m 08s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/0a447a2d44b845199a7177fde66886ab : FAILURE in 1h 29m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43cbba67b97045a488700e213e268e2f : SUCCESS in 4m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fc1b938b0e134e04b6aa74e4de73be2d : SUCCESS in 1h 25m 44s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3255b91bd08e45569ed1d7cbd201c24b : SUCCESS in 9m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f9e5ebef14b640f4aa46c49d5119e509 : SUCCESS in 1h 46m 03s","accounts_in_message":[],"_revision_number":28},{"id":"577b244df760fe69a350f78f10570b91c28e060d","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-17 20:42:13.000000000","message":"Patch Set 28:\n\nrecheck","accounts_in_message":[],"_revision_number":28},{"id":"4bf31554020dce2cf7ad8689f0932dbaa37bddc6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-17 20:43:03.000000000","message":"Patch Set 28: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":28},{"id":"18a6e8108057877a291474becab62370de0a9ca5","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-17 22:05:10.000000000","message":"Patch Set 28:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":28},{"id":"e29932212b798bd1b55a6d4babd69b13c6aa0fce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-17 23:19:54.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/88553a96a3ce431b93909e12eb38b52b : SUCCESS in 2h 34m 20s","accounts_in_message":[],"_revision_number":28},{"id":"f0270734b617281f49f986826cf33a3174bb54ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-18 00:44:42.000000000","message":"Uploaded patch set 29: Commit message was updated.","accounts_in_message":[],"_revision_number":29},{"id":"2c4ee7834c8a27d344088b73d20fe94a38000bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 04:15:51.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ed1db1de8fdc400ab9e615600cdace29 : SUCCESS in 2h 28m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/425809fd3d2b4c3980b034c14afb3237 : SUCCESS in 1h 21m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/576081e866044167a376a2e940ecfab7 : SUCCESS in 1h 36m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0886d6fae6104409964dcc7f1f5024c9 : SUCCESS in 6m 13s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d07a71b19d5c47a68617688a0d2aa702 : SUCCESS in 2h 22m 17s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/1f88fdb4b7e046dca91fcb2a97de03ac : SUCCESS in 3h 23m 19s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/cb9861c7de334d949239b9c084425fc2 : SUCCESS in 2h 43m 56s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/3a7cd0588e724b22a3e62099eb3ee38d : FAILURE in 1h 50m 27s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/43fb5bba438d46eb86602abc12536592 : FAILURE in 1h 25m 37s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/821460847ba5427fb60bd46c00c6c03e : SUCCESS in 1h 29m 29s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/f4259997df0c445d9daa5ccc1ff34417 : SUCCESS in 1h 52m 53s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19d1570ed4c45a7b8c8cca516b3c4f4 : SUCCESS in 3m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/efdfa49425f74103a7633e3a271a6204 : SUCCESS in 1h 09m 27s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/a5c4274e4a7341ed87edfede6d3cf46e : SUCCESS in 8m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a793b30a3d3a437588871b318f4153cf : SUCCESS in 1h 34m 45s","accounts_in_message":[],"_revision_number":29},{"id":"8c760ad01c02e8da3861e00463dcf978d55c815b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-18 04:45:32.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b87e3de25dd74f03ba58dc7b0418135f : SUCCESS in 2h 58m 01s","accounts_in_message":[],"_revision_number":29},{"id":"f8a5d434643b81a17706e190b93aaea3604e4201","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-18 09:42:30.000000000","message":"Patch Set 29:\n\nrecheck","accounts_in_message":[],"_revision_number":29},{"id":"619ac64a85da161fbf908da2d7dd776bc100e7d0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-18 12:23:07.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1a27e407da2d4a44a8651d601db6a718 : SUCCESS in 2h 39m 01s","accounts_in_message":[],"_revision_number":29},{"id":"1ca426f9f766cb8c72cafee470525c6be925be04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-18 13:11:24.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"5a6637fa80f097ee25ddc60de2dc5e8cc97c1e3a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-18 16:52:49.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2574ae0e8d414b35aa791eb61de5e53b : FAILURE in 2h 15m 51s","accounts_in_message":[],"_revision_number":30},{"id":"019b50735dfeabaf57c82fa82a8f0775b680dcb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 17:42:01.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1a4df3b9d6eb4b02a4c905c7287b0e3a : SUCCESS in 4h 20m 15s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ccf2e2da0d2d46179971474a9dbffeca : SUCCESS in 1h 02m 22s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a95af438bcce4b48865b9272912ad3f5 : SUCCESS in 3h 14m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/345846d1a698402387df5a60fd874939 : SUCCESS in 7m 06s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d4783ced9bac458fba0e6a95bf8af6c4 : SUCCESS in 2h 49m 42s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/17182ec53f6e40cbb5420177fdecd151 : SUCCESS in 2h 47m 59s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6e996f2ccc884cbc9edfe961ca21b20d : SUCCESS in 2h 31m 37s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f5a86479a12d4d3cb7c9ed2190ce8ab7 : SUCCESS in 3h 11m 57s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/3be581fb11a847c393db3363d99a62a3 : SUCCESS in 1h 38m 21s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8a1038b4c5af401fa92dd5411a2eab98 : SUCCESS in 1h 44m 15s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/85c83130ce714809962b24d5bdc1cccd : SUCCESS in 1h 35m 28s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/18212c7bccd64457905d9e770c4333f3 : SUCCESS in 2h 09m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856283c13f0c4a63a60bdfa4428c1e70 : SUCCESS in 5m 30s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4fd9adc4c17d4d38a5b40c6d34aaae79 : SUCCESS in 1h 33m 11s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/5293a6b276f44971b6f8c73d26057f27 : SUCCESS in 10m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/07455270f4134488853ffcb88a3ea6a8 : SUCCESS in 1h 54m 13s","accounts_in_message":[],"_revision_number":30},{"id":"746c3db5234d4c6cf750c4e49701c6d99dc246a3","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-18 20:02:05.000000000","message":"Patch Set 30:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":30},{"id":"20f1ab9ad6131fe62789f6fb154af322e33fe897","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-18 21:00:12.000000000","message":"Patch Set 30: Code-Review+2","accounts_in_message":[],"_revision_number":30},{"id":"643162d9ba9c072a62a6e69326fbc847431c5082","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-18 23:37:31.000000000","message":"Patch Set 30:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/40d43aa2bda646e19a15202e64323815 : FAILURE in 2h 03m 58s","accounts_in_message":[],"_revision_number":30},{"id":"39dbd82e278fd987d3e5331a12f9840916d2142a","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-19 07:26:54.000000000","message":"Patch Set 30: Code-Review-1\n\n(1 comment)\n\ndepends-on are now merged so this is good to go but see inline comment about queens please?\nslightly concerned about the 3 hrs 14 mins of https://zuul.opendev.org/t/openstack/build/a95af438bcce4b48865b9272912ad3f5  but it doesn\u0027t appear to be consistent the previous runs were closer to the 1.5 hours like the job currently takes looking at the history https://zuul.opendev.org/t/openstack/builds?job_name\u003dtripleo-ci-centos-8-containers-multinode\u0026project\u003dopenstack/tripleo-quickstart","accounts_in_message":[],"_revision_number":30},{"id":"f3c1d9de93edeca3deeb78fff0ca2a1a07b287e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-19 12:20:05.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"a8e649337b06232b28667adc7db352a2bd65e711","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-19 12:20:34.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"7df4dd47595e646ad32ac26991aeea9a195a82b7","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-19 15:05:17.000000000","message":"Patch Set 31: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"df1196de2871df231fa05f588165179aaa34fab8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-19 15:14:44.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/919052fc28564265b38080e33f572628 : FAILURE in 2h 32m 42s","accounts_in_message":[],"_revision_number":31},{"id":"0e2a9b037205c591b798ffbb916c8862ca9b8fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 15:40:41.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7d553e668b4644f8b9f87e0f3dc3636a : SUCCESS in 2h 55m 48s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f44dcd32bda940058e56bdf97f142a0a : SUCCESS in 58m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/32b033d1a69a400981b3b014cf8295ca : SUCCESS in 1h 29m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b8ff1c9ba9934a69a4b7c1391a9333fc : SUCCESS in 6m 48s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f177caf4ef1240fb9c77e6725c818f93 : SUCCESS in 2h 49m 45s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/e7338f32e08e4f61a98f881f1c0c680d : SUCCESS in 3h 05m 46s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d10fda38505c463ba496a8d3c973352f : SUCCESS in 2h 50m 14s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7bac6eb52bdc447a922587dfaacf1d54 : SUCCESS in 2h 44m 02s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/0a1e59205bc946eda7734e20441bc74b : SUCCESS in 1h 42m 56s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/1574d3042c22430393410ae53d8085f6 : SUCCESS in 1h 48m 52s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/88925df3414843c09dd9823cdaf826c4 : SUCCESS in 1h 45m 23s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/defbc0df83584395ad8d44934cef580d : SUCCESS in 1h 41m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9def6aea93d8463692f3c2e439fe4c1b : SUCCESS in 4m 06s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/44046877ac6342af9261becc55b7c1fc : SUCCESS in 1h 14m 23s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/3aed76c5994b4adfb376ee9b6a3afec7 : SUCCESS in 9m 04s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/850e86e7c07540ff8a731c3399055765 : SUCCESS in 1h 41m 17s","accounts_in_message":[],"_revision_number":31},{"id":"2a4975161f51ff6653062f9dd86464e12a6b3292","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-05-19 17:25:33.000000000","message":"Patch Set 31: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":31},{"id":"fb36c7b32987b86e028f6c0b0edb84cc4bf40064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:26:03.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"b5d31ac99ce897da9bf055992528fc75372cb506","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-05-19 18:40:03.000000000","message":"Patch Set 31: Code-Review+2","accounts_in_message":[],"_revision_number":31},{"id":"38d0efea0605f9e110208fa5afead4bf0d3fbb1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 21:30:58.000000000","message":"Patch Set 31: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6a41fc131a6040ed8b7ae8ac6cb955ed : SUCCESS in 3h 59m 43s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/548d17fb97c54802a8debdb0dca2e7c2 : SUCCESS in 1h 11m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c3f0293291cc4efc9446043409a85cbc : SUCCESS in 1h 55m 53s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c94193bd2da642f580b1f299ff426955 : SUCCESS in 3h 38m 47s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/3300161c8fe74a868033aa0bdd92db96 : SUCCESS in 3h 03m 51s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/f4520addcbec477a8cefa981bdaa77d1 : SUCCESS in 3h 13m 48s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/21f070695d284658adec08310d14d27d : SUCCESS in 3h 56m 36s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/1e36735e6a4041b18bd4fc73024770e5 : SUCCESS in 1h 39m 37s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/89c522f50e2b4899bb99566a0d263f39 : SUCCESS in 1h 43m 14s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/ac6ff65c4ffb45e6a57b1c69edef1fd3 : SUCCESS in 1h 38m 36s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/7f5fefc3e3f442db95e5f85a78a64d90 : SUCCESS in 2h 30m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7a6a3d0d3d045dab20b39253278cea1 : SUCCESS in 5m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/eae776fe5e4543e4860d2989a1c7e026 : SUCCESS in 1h 38m 29s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4b0f41d9e6e54c6d96c25a0bde4afd5f : SUCCESS in 9m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/195b066203bd40ce993b3a6308a57a07 : SUCCESS in 2h 10m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/da8d2b48c57443febb578bd441bbbc1e : SUCCESS in 7m 56s","accounts_in_message":[],"_revision_number":31},{"id":"dcc0841f87083f1c6eac4aaec53b2002400e18d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 21:30:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":31},{"id":"f2c7c180892777fdcca216ebfb2a1f755be2dc09","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 21:33:42.000000000","message":"Patch Set 31:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944d0d1607fa4d49914241688030766f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"2289341399b6efad8725b2de28f1c416a07f3649","revisions":{"2ae05c7e7dedc79463a6d31f5334a8446ea6421e":{"kind":"REWORK","_number":1,"created":"2021-02-23 13:46:28.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/1"}}},"commit":{"parents":[{"commit":"f75ebea5240cdcf690be503913095c92c8c4d757","subject":"Merge \"update default libvirt nodepool to centos8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f75ebea5240cdcf690be503913095c92c8c4d757"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:46:12.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I8fb6ca088b1052488ff4f9ada4d3ab29c0be4735\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2ae05c7e7dedc79463a6d31f5334a8446ea6421e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2ae05c7e7dedc79463a6d31f5334a8446ea6421e"}]},"branch":"refs/heads/master"},"8a26a4952b9051aa84ffee8e7f77684850dc7dcd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-24 13:20:08.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/2"}}},"commit":{"parents":[{"commit":"f130a51703b6f4f7dfc040c8217292de8d625593","subject":"DNM: test podman3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f130a51703b6f4f7dfc040c8217292de8d625593"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-24 13:19:54.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I8fb6ca088b1052488ff4f9ada4d3ab29c0be4735\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8a26a4952b9051aa84ffee8e7f77684850dc7dcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8a26a4952b9051aa84ffee8e7f77684850dc7dcd"}]},"branch":"refs/heads/master"},"d49722936e569e37ef0506c276bee90fd55aec32":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-24 18:33:49.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/3"}}},"commit":{"parents":[{"commit":"80c9442b1afec8ee21dce96503b0868ecd6c8d83","subject":"Ensure the gating repo is release aware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/80c9442b1afec8ee21dce96503b0868ecd6c8d83"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-24 18:32:16.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I8fb6ca088b1052488ff4f9ada4d3ab29c0be4735\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d49722936e569e37ef0506c276bee90fd55aec32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d49722936e569e37ef0506c276bee90fd55aec32"}]},"branch":"refs/heads/master"},"9dd4fd96b0db37da7d6ad740633bebd844d922de":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-25 02:17:03.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/4"}}},"commit":{"parents":[{"commit":"eb9dc674c0158a71af49678fa4042c2c0fee6cee","subject":"Merge \"Fix network environment ordering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb9dc674c0158a71af49678fa4042c2c0fee6cee"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-25 02:16:58.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I8fb6ca088b1052488ff4f9ada4d3ab29c0be4735\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9dd4fd96b0db37da7d6ad740633bebd844d922de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9dd4fd96b0db37da7d6ad740633bebd844d922de"}]},"branch":"refs/heads/master"},"e9d18ae51c2ad8b66932c26185ec6194171109f1":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-02-26 00:30:59.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/5"}}},"commit":{"parents":[{"commit":"eb9dc674c0158a71af49678fa4042c2c0fee6cee","subject":"Merge \"Fix network environment ordering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb9dc674c0158a71af49678fa4042c2c0fee6cee"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-26 00:30:44.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: Ia6636bf0223f7e51b6b50950a5c8c1413b4a54b0\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I8fb6ca088b1052488ff4f9ada4d3ab29c0be4735\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e9d18ae51c2ad8b66932c26185ec6194171109f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e9d18ae51c2ad8b66932c26185ec6194171109f1"}]},"branch":"refs/heads/master"},"9bc2e339daa3d2035f5f23fb6900c6025018af27":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-02-26 02:17:09.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/6"}}},"commit":{"parents":[{"commit":"eb9dc674c0158a71af49678fa4042c2c0fee6cee","subject":"Merge \"Fix network environment ordering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb9dc674c0158a71af49678fa4042c2c0fee6cee"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-26 02:16:59.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: If7f40f27aba9edbe6e6604afd244626a238922fc\nDepends-On: Ia6636bf0223f7e51b6b50950a5c8c1413b4a54b0\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I8fb6ca088b1052488ff4f9ada4d3ab29c0be4735\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9bc2e339daa3d2035f5f23fb6900c6025018af27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9bc2e339daa3d2035f5f23fb6900c6025018af27"}]},"branch":"refs/heads/master"},"dc1559d2979bf274b5cc8325c5a434d10a5d7684":{"kind":"REWORK","_number":7,"created":"2021-03-01 12:47:05.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/7"}}},"commit":{"parents":[{"commit":"eb9dc674c0158a71af49678fa4042c2c0fee6cee","subject":"Merge \"Fix network environment ordering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb9dc674c0158a71af49678fa4042c2c0fee6cee"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-03-01 12:46:59.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: If7f40f27aba9edbe6e6604afd244626a238922fc\nDepends-On: Ia6636bf0223f7e51b6b50950a5c8c1413b4a54b0\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I8fb6ca088b1052488ff4f9ada4d3ab29c0be4735\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dc1559d2979bf274b5cc8325c5a434d10a5d7684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dc1559d2979bf274b5cc8325c5a434d10a5d7684"}]},"branch":"refs/heads/master"},"c66baf61067d7ea680b102e3c0430e524734b7bd":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-03-01 16:27:40.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/8"}}},"commit":{"parents":[{"commit":"eb9dc674c0158a71af49678fa4042c2c0fee6cee","subject":"Merge \"Fix network environment ordering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb9dc674c0158a71af49678fa4042c2c0fee6cee"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-03-01 16:26:50.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Idc326cb7baa54a7d2d01a2be002dfffa6a59ef6a\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c66baf61067d7ea680b102e3c0430e524734b7bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c66baf61067d7ea680b102e3c0430e524734b7bd"}]},"branch":"refs/heads/master"},"c3844d3e77a6aabf66cbeb28e61e864bd395e84c":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2021-03-02 12:48:06.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/9"}}},"commit":{"parents":[{"commit":"eb9dc674c0158a71af49678fa4042c2c0fee6cee","subject":"Merge \"Fix network environment ordering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb9dc674c0158a71af49678fa4042c2c0fee6cee"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-03-02 12:47:55.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I2c9cf50363579d4e18aaeaf783770ca7a4266622\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Idc326cb7baa54a7d2d01a2be002dfffa6a59ef6a\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c3844d3e77a6aabf66cbeb28e61e864bd395e84c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/c3844d3e77a6aabf66cbeb28e61e864bd395e84c"}]},"branch":"refs/heads/master"},"9146828ccffaab1decff074a394064cf30c89522":{"kind":"REWORK","_number":10,"created":"2021-03-04 01:58:06.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/10"}}},"commit":{"parents":[{"commit":"eb9dc674c0158a71af49678fa4042c2c0fee6cee","subject":"Merge \"Fix network environment ordering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/eb9dc674c0158a71af49678fa4042c2c0fee6cee"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-03-04 01:58:01.000000000","tz":-300},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I2c9cf50363579d4e18aaeaf783770ca7a4266622\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Idc326cb7baa54a7d2d01a2be002dfffa6a59ef6a\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9146828ccffaab1decff074a394064cf30c89522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9146828ccffaab1decff074a394064cf30c89522"}]},"branch":"refs/heads/master"},"cfddeccf88dfe098caae36deba92f11b3d159b5b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":11,"created":"2021-04-14 15:19:42.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/11"}}},"commit":{"parents":[{"commit":"5a955af4be604951bd13c3592f0794f9eac0823a","subject":"Merge \"Move Centos repo to Pacific\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/5a955af4be604951bd13c3592f0794f9eac0823a"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-04-14 14:44:00.000000000","tz":-240},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I384d15f819f8f5f5654ea8d98a67cdaaec1cff2e\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Idc326cb7baa54a7d2d01a2be002dfffa6a59ef6a\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cfddeccf88dfe098caae36deba92f11b3d159b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cfddeccf88dfe098caae36deba92f11b3d159b5b"}]},"branch":"refs/heads/master"},"513367bc5c790507ca5bbb9da4bcc8fdf3c67aea":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":12,"created":"2021-04-20 15:51:03.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/12"}}},"commit":{"parents":[{"commit":"fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35","subject":"Merge \"Move all the octavia tests to fs062\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-04-20 15:50:48.000000000","tz":-240},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Idc326cb7baa54a7d2d01a2be002dfffa6a59ef6a\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/513367bc5c790507ca5bbb9da4bcc8fdf3c67aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/513367bc5c790507ca5bbb9da4bcc8fdf3c67aea"}]},"branch":"refs/heads/master"},"1049c7a7ec1578067660caafb5202ff6ce758d71":{"kind":"REWORK","_number":13,"created":"2021-04-21 15:32:45.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/13"}}},"commit":{"parents":[{"commit":"fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35","subject":"Merge \"Move all the octavia tests to fs062\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-04-21 14:23:51.000000000","tz":-240},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Idc326cb7baa54a7d2d01a2be002dfffa6a59ef6a\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1049c7a7ec1578067660caafb5202ff6ce758d71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1049c7a7ec1578067660caafb5202ff6ce758d71"}]},"branch":"refs/heads/master"},"6f0e9770d8f2a1518e92590d697337734ef40b54":{"kind":"REWORK","_number":14,"created":"2021-04-21 16:07:08.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/14"}}},"commit":{"parents":[{"commit":"fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35","subject":"Merge \"Move all the octavia tests to fs062\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-04-21 16:07:03.000000000","tz":-240},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Idc326cb7baa54a7d2d01a2be002dfffa6a59ef6a\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6f0e9770d8f2a1518e92590d697337734ef40b54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6f0e9770d8f2a1518e92590d697337734ef40b54"}]},"branch":"refs/heads/master"},"b2951673fbf5aff7266b6d4e9995a5d1b8d2feb8":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2021-04-23 20:48:30.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/15"}}},"commit":{"parents":[{"commit":"fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35","subject":"Merge \"Move all the octavia tests to fs062\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fdccdccb796ff30cf2d3f9a1d4956c0f9639ff35"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-04-23 20:48:14.000000000","tz":-240},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: I7a94e0f72ed42b70468a361536bfbf88d158dc14\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b2951673fbf5aff7266b6d4e9995a5d1b8d2feb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b2951673fbf5aff7266b6d4e9995a5d1b8d2feb8"}]},"branch":"refs/heads/master"},"b853b5859288345422dc88d2290401144aa1964a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":16,"created":"2021-04-29 22:29:21.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/16"}}},"commit":{"parents":[{"commit":"7ceaa0c0a5d0e528c71fc130573fef10677ca71f","subject":"Addl wallaby release file for centosci","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7ceaa0c0a5d0e528c71fc130573fef10677ca71f"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-04-29 22:28:41.000000000","tz":-240},"subject":"DNM: Support using ephemeral Heat","message":"DNM: Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b853b5859288345422dc88d2290401144aa1964a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b853b5859288345422dc88d2290401144aa1964a"}]},"branch":"refs/heads/master"},"0b37024f90baad3bf1235f40ab898447f9a58927":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2021-05-03 19:07:08.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/17"}}},"commit":{"parents":[{"commit":"7ceaa0c0a5d0e528c71fc130573fef10677ca71f","subject":"Addl wallaby release file for centosci","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7ceaa0c0a5d0e528c71fc130573fef10677ca71f"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-03 19:06:48.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0b37024f90baad3bf1235f40ab898447f9a58927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/0b37024f90baad3bf1235f40ab898447f9a58927"}]},"branch":"refs/heads/master"},"fb6cb8cf9ab05a25c868972b7c08231d2fb0eea0":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-05-07 01:00:43.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/18"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-07 01:00:25.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fb6cb8cf9ab05a25c868972b7c08231d2fb0eea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/fb6cb8cf9ab05a25c868972b7c08231d2fb0eea0"}]},"branch":"refs/heads/master"},"4689c5af3184d9e7268607a995badebe2f3a0999":{"kind":"REWORK","_number":19,"created":"2021-05-10 00:14:11.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/19"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-10 00:14:06.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4689c5af3184d9e7268607a995badebe2f3a0999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4689c5af3184d9e7268607a995badebe2f3a0999"}]},"branch":"refs/heads/master"},"82e8e1e1e07a1a34a4c49606ce137760a32e20db":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2021-05-11 22:12:42.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/20"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-11 22:12:26.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/82e8e1e1e07a1a34a4c49606ce137760a32e20db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/82e8e1e1e07a1a34a4c49606ce137760a32e20db"}]},"branch":"refs/heads/master"},"bf85384962e2c577c4ec88c229ccab751d015d3a":{"kind":"REWORK","_number":21,"created":"2021-05-12 11:36:20.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/21"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-12 11:36:13.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdate featureset010 for the tripleo-ci-centos-8-containers-multinode\njob to enable using ephemeral heat using the newly added\nephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bf85384962e2c577c4ec88c229ccab751d015d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/bf85384962e2c577c4ec88c229ccab751d015d3a"}]},"branch":"refs/heads/master"},"78987cf0ebf7a0969db4beb57313c823ec6921f3":{"kind":"REWORK","_number":22,"created":"2021-05-12 20:14:46.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/22"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-12 20:13:31.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the following job to enable using ephemeral heat using the newly\nadded ephemeral_heat/ephemeral_heat_args variable.\n\n- featureset010: tripleo-ci-centos-8-containers-multinode\n- featureset037: tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78987cf0ebf7a0969db4beb57313c823ec6921f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78987cf0ebf7a0969db4beb57313c823ec6921f3"}]},"branch":"refs/heads/master"},"1086bd799c3f09e34ddd2e8ed04797f613eddfdd":{"kind":"REWORK","_number":23,"created":"2021-05-13 00:21:15.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/23"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-13 00:20:24.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nSets the\ntripleo-ci-centos-8-scenario000-multinode-oooq-container-updates\n(featureset037) to not use ephemeral heat for now.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1086bd799c3f09e34ddd2e8ed04797f613eddfdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/1086bd799c3f09e34ddd2e8ed04797f613eddfdd"}]},"branch":"refs/heads/master"},"149a07bdcb7d42b220af78517d07fd150e81db15":{"kind":"REWORK","_number":24,"created":"2021-05-13 00:23:26.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/24"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-13 00:23:19.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nSets the\ntripleo-ci-centos-8-scenario000-multinode-oooq-container-updates\n(featureset037) to not use ephemeral heat for now.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/149a07bdcb7d42b220af78517d07fd150e81db15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/149a07bdcb7d42b220af78517d07fd150e81db15"}]},"branch":"refs/heads/master"},"80f0a36e41f19ae869700e88a23a97ea7f512ce1":{"kind":"REWORK","_number":25,"created":"2021-05-13 12:49:10.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/25"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-13 12:47:44.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nSets the\ntripleo-ci-centos-8-scenario000-multinode-oooq-container-updates\n(featureset037) and tripleo-ci-centos-8-undercloud-containers\n(featureset027) to not use ephemeral heat for now so that we can move\nforward with changing the default in tripleoclient\n(c9c0a5adac08ea2e8465447b4c4a221d8fe4548e)\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/80f0a36e41f19ae869700e88a23a97ea7f512ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/80f0a36e41f19ae869700e88a23a97ea7f512ce1"}]},"branch":"refs/heads/master"},"8b685b04b0decd767b0f51a4ea264aeded5aecde":{"kind":"REWORK","_number":26,"created":"2021-05-14 12:06:11.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/26"}}},"commit":{"parents":[{"commit":"851a548a53a8bb784fdf64b29751b507a388b02c","subject":"Remove tripleo-ansible dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/851a548a53a8bb784fdf64b29751b507a388b02c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-14 12:03:54.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nSets the\ntripleo-ci-centos-8-scenario000-multinode-oooq-container-updates\n(featureset037) and tripleo-ci-centos-8-undercloud-containers\n(featureset027) to not use ephemeral heat for now so that we can move\nforward with changing the default in tripleoclient\n(c9c0a5adac08ea2e8465447b4c4a221d8fe4548e)\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8b685b04b0decd767b0f51a4ea264aeded5aecde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8b685b04b0decd767b0f51a4ea264aeded5aecde"}]},"branch":"refs/heads/master"},"e80d9efcfb0fbff5d614a2096e9191cbd6b72860":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2021-05-14 12:15:39.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/27"}}},"commit":{"parents":[{"commit":"151abc3cb84292d2caecd6967e2a5e0341b99a13","subject":"Merge \"Remove no_log: result.rc from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/151abc3cb84292d2caecd6967e2a5e0341b99a13"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-14 12:15:14.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nSets the\ntripleo-ci-centos-8-scenario000-multinode-oooq-container-updates\n(featureset037) and tripleo-ci-centos-8-undercloud-containers\n(featureset027) to not use ephemeral heat for now so that we can move\nforward with changing the default in tripleoclient\n(c9c0a5adac08ea2e8465447b4c4a221d8fe4548e)\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e80d9efcfb0fbff5d614a2096e9191cbd6b72860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e80d9efcfb0fbff5d614a2096e9191cbd6b72860"}]},"branch":"refs/heads/master"},"78157effb966f2b62c8cbf1cd702652f6e270d4d":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":28,"created":"2021-05-17 17:17:16.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/28"}}},"commit":{"parents":[{"commit":"78551d6c2ead03092b8851554acc3a20f4c8b041","subject":"Merge \"Don\u0027t include net-iso absolute in fs001\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78551d6c2ead03092b8851554acc3a20f4c8b041"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-17 17:16:40.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78157effb966f2b62c8cbf1cd702652f6e270d4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78157effb966f2b62c8cbf1cd702652f6e270d4d"}]},"branch":"refs/heads/master"},"008fbc69cf50998baddca900f950219629c836d8":{"kind":"NO_CODE_CHANGE","_number":29,"created":"2021-05-18 00:44:42.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/29"}}},"commit":{"parents":[{"commit":"78551d6c2ead03092b8851554acc3a20f4c8b041","subject":"Merge \"Don\u0027t include net-iso absolute in fs001\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/78551d6c2ead03092b8851554acc3a20f4c8b041"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-18 00:44:32.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/008fbc69cf50998baddca900f950219629c836d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/008fbc69cf50998baddca900f950219629c836d8"}]},"branch":"refs/heads/master"},"f692fc0579a25e8440ee77c628a0726db2c6d6ec":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2021-05-18 13:11:24.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/30"}}},"commit":{"parents":[{"commit":"9dcefb2973d71406b548def51161855b87a4b90c","subject":"Merge \"Fix featureset002 for network_provision/baremetal_provision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9dcefb2973d71406b548def51161855b87a4b90c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-18 13:11:20.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nDepends-On: I29cf6dd8a32242f27a40946688f9dede8968c7f9\nDepends-On: Ib63813b63c37fa2cee57a211535d43d605131529\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f692fc0579a25e8440ee77c628a0726db2c6d6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f692fc0579a25e8440ee77c628a0726db2c6d6ec"}]},"branch":"refs/heads/master"},"2289341399b6efad8725b2de28f1c416a07f3649":{"kind":"REWORK","_number":31,"created":"2021-05-19 12:20:05.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/08/777108/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/08/777108/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/08/777108/31"}}},"commit":{"parents":[{"commit":"9dcefb2973d71406b548def51161855b87a4b90c","subject":"Merge \"Fix featureset002 for network_provision/baremetal_provision\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9dcefb2973d71406b548def51161855b87a4b90c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-02-23 13:40:33.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-19 12:17:15.000000000","tz":-240},"subject":"Support using ephemeral Heat","message":"Support using ephemeral Heat\n\nUpdates the containers-multinode job to enable using ephemeral heat\nusing the newly added ephemeral_heat/ephemeral_heat_args variable.\n\nImplements: blueprint ephemeral-heat-overcloud\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Ide514b7f2ea94a1a82bd414840a9adb0a79ade98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2289341399b6efad8725b2de28f1c416a07f3649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/2289341399b6efad8725b2de28f1c416a07f3649"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
