)]}'
{"id":"openstack%2Ftripleo-quickstart~784323","triplet_id":"openstack%2Ftripleo-quickstart~master~I9116b37128919e3c6c3114a7ce87c0982ff033ae","project":"openstack/tripleo-quickstart","branch":"master","topic":"fix_octavia_kvm","hashtags":[],"change_id":"I9116b37128919e3c6c3114a7ce87c0982ff033ae","subject":"Include nested_virt in default opt tags","status":"MERGED","created":"2021-04-01 09:34:56.000000000","updated":"2021-04-08 17:16:24.000000000","submitted":"2021-04-08 17:14:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784323-fix_octavia_kvm","meta_rev_id":"0b58918f981d7c310f71aabd50003ac9d8afc2c5","_number":784323,"virtual_id_number":784323,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 17:14:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-06 11:40:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-04-06 09:12:27.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-04-08 09:36:43.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-08 17:16:24.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2021-04-08 09:36:43.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 10:40:16.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-04-01 12:49:40.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-01 13:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 09:12:27.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-04-08 04:36:32.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-04-08 04:36:32.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"}],"messages":[{"id":"cf3f5c03230ef64874d3703e52ee0482af75034c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-01 09:34:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a334f409a35e4c68de07704e2820a0b3a06460d","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-01 10:40:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b89f63d189d96784e07b85316a3f028c029f874e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-01 12:49:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0906b9104c084e06a35df647fa6be998 : SUCCESS in 3h 11m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0d1d5970cf3f4e734f237446aa602e714d3848a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:42:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/03fb9f8bade041d9afead8d33cc74978 : SUCCESS in 3h 30m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/16ac5db8f8184b6a841e40255fc4f127 : SUCCESS in 1h 17m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/23ab2d7c110949b79971685ccdd1c849 : SUCCESS in 1h 59m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8416be253c424941a697b2d6c2ecd715 : SUCCESS in 7m 56s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/4a0beb517352406993c1989aa7203612 : SUCCESS in 3h 25m 36s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/aa961bbb59734fc89a00fb72370d02b1 : SUCCESS in 3h 27m 02s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d6a0bbe600a1492cbfb02207f66293ad : SUCCESS in 3h 30m 28s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/3d669a5f21164451ae76034f1e50a10f : SUCCESS in 2h 05m 00s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/cfa0f84fcaa94d04bd3307d2d50f3af7 : SUCCESS in 2h 05m 09s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/667236c764c14011872eca81174d76c4 : SUCCESS in 2h 14m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04fd0f4d7054119bed927132079d088 : SUCCESS in 5m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/44550c85339a4c7eaf8802d41e3bc581 : FAILURE in 20m 54s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/227a6c47f3f24c4ca5f38a562e0fe8d8 : SUCCESS in 10m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f0f253322dab4924a527905b7abed007 : SUCCESS in 2h 02m 45s","accounts_in_message":[],"_revision_number":1},{"id":"756f68cf09a3dd8ab04d9804342be50208c2a794","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-06 09:00:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"999689aff4e5cb080ea808d8d8322d86bd36dbfb","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-04-06 09:12:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e1ebbaf26c32a1f5b362f8d5013dbd3dbd8f2d7a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-06 11:40:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bfbab971f6ea463b901b16e8ff3a404a : SUCCESS in 2h 37m 43s","accounts_in_message":[],"_revision_number":2},{"id":"0bfd4be959de87a094558c5ab57c3fc2aa975f07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 12:21:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/57b29ceb3ace491c8cfba63bd0c0e7a4 : SUCCESS in 3h 03m 30s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c99b82b57918497db31a5dd57a778f95 : SUCCESS in 1h 17m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8b9c692ba9404cf4a9bbbb8b437eadf5 : SUCCESS in 1h 53m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7f0d90b2dc994c1aa8c9908cd4e4a427 : SUCCESS in 8m 15s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/f693ca1e340747d9bd0aad80a60033e3 : SUCCESS in 2h 33m 05s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/476f136c92d9479697274d5c86f71ae5 : SUCCESS in 2h 55m 05s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/21a3d4189d354769b1459d451887aafc : SUCCESS in 2h 54m 56s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/87761e7fff244dd8be25b5d09e559102 : SUCCESS in 1h 35m 40s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6b4081960bd647a999f951298e6e74a1 : SUCCESS in 1h 55m 59s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/2fa7d17412de4abea70b67aaba1ebe7c : SUCCESS in 1h 55m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0effbbef19d4171b8e364793b66f9ac : SUCCESS in 4m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/01b97bf7748b453f9a80b68b261a174d : SUCCESS in 1h 34m 55s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/f1c6685a092149a0a9d81aa931963ef4 : SUCCESS in 22m 01s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e570e4ee8c5b45218609e2615e057cbb : SUCCESS in 1h 55m 01s","accounts_in_message":[],"_revision_number":2},{"id":"a56b1ab6c0c820a3055098649ef3cbe4bc191c00","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-04-08 09:36:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ef2e14812f59ec1679e4ad035ff6b7949619de2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:37:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18ba518176731a9045a7266e41d0c84b1beb6dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:14:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7fb0dcf2503d4718a9f1138f1e6a70a9 : SUCCESS in 3h 08m 09s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/333b61b00cdf4592aca6e75412b375cb : SUCCESS in 58m 51s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8ecd9c38f55d43e38de99f75a51ffde7 : SUCCESS in 1h 48m 14s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/46c05b8cd04e44c19ec8d2099c07fa60 : SUCCESS in 3h 43m 23s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2013bb0b156b471bb607ea3aa8a2b2fb : SUCCESS in 3h 09m 01s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/63c0b2ba99ce4fff91da118a08a9eec3 : SUCCESS in 3h 07m 29s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/ac03915b65404be69bc696221d7f9a14 : SUCCESS in 1h 56m 48s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/979a5018594e46a6938762a1f964aa27 : SUCCESS in 2h 00m 58s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/9267bf5b909a4c62bdf966dda04aa77f : SUCCESS in 2h 05m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d915e619e9524cef807d5920e390afb1 : SUCCESS in 8m 22s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/121b4ba363db4e87b690812302f59149 : SUCCESS in 1h 10m 48s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/ac50af4389984e49a36185efe2fb1976 : SUCCESS in 15m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/895832133a8243d1a4e3a66acd6d775a : SUCCESS in 1h 34m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2dc53678ac7f47ebab841485e073e554 : SUCCESS in 8m 23s","accounts_in_message":[],"_revision_number":2},{"id":"18b4fc6d96f92fb48aee21d7377eb684dc140130","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:14:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b570e9500e063b74fee70dd29f3a98449268195f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:16:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5114bd297847436ca5adc70c532b7472 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7cc521a93c9950504b86585e02f64fb72e781f69","revisions":{"38dcac2d8c6cc14cc483ec75f6b29c8280bffbcd":{"kind":"REWORK","_number":1,"created":"2021-04-01 09:34:56.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/23/784323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/784323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/1"}}},"commit":{"parents":[{"commit":"9c734b154a3516bba55209af0cdcb5daeb795d47","subject":"Merge \"Fix quotes issue in stein release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9c734b154a3516bba55209af0cdcb5daeb795d47"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-01 09:32:53.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-01 09:32:53.000000000","tz":330},"subject":"Include nested_virt in default opt tags","message":"Include nested_virt in default opt tags\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/782020\nenables nested virt for standalone octavia jobs. In order to run it\nvia quickstart, it adds the nested_virt tag to default opt tags.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/782020\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I9116b37128919e3c6c3114a7ce87c0982ff033ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/38dcac2d8c6cc14cc483ec75f6b29c8280bffbcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/38dcac2d8c6cc14cc483ec75f6b29c8280bffbcd"}]},"branch":"refs/heads/master"},"7cc521a93c9950504b86585e02f64fb72e781f69":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-06 09:00:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/23/784323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/23/784323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/23/784323/2"}}},"commit":{"parents":[{"commit":"8a15c2d0265692d6263b754086383559f4076888","subject":"Bump container-tools:3.0 for integration line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/8a15c2d0265692d6263b754086383559f4076888"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-01 09:32:53.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-04-06 09:00:46.000000000","tz":0},"subject":"Include nested_virt in default opt tags","message":"Include nested_virt in default opt tags\n\nhttps://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/782020\nenables nested virt for standalone octavia jobs. In order to run it\nvia quickstart, it adds the nested_virt tag to default opt tags.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/782020\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I9116b37128919e3c6c3114a7ce87c0982ff033ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7cc521a93c9950504b86585e02f64fb72e781f69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/7cc521a93c9950504b86585e02f64fb72e781f69"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
