)]}'
{"id":"openstack%2Ftripleo-quickstart~784768","triplet_id":"openstack%2Ftripleo-quickstart~master~Ifdaa7d59f9a4a1dc54c32532e79fb8a10c20f637","project":"openstack/tripleo-quickstart","branch":"master","topic":"bump_container_tools","hashtags":[],"change_id":"Ifdaa7d59f9a4a1dc54c32532e79fb8a10c20f637","subject":"Bump to container-tools:3.0 for upstream CI","status":"ABANDONED","created":"2021-04-05 13:09:06.000000000","updated":"2021-04-29 12:04:14.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"66a7f3d6a98ffdfa9b6402d36a774877719b1062","_number":784768,"virtual_id_number":784768,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-08 11:34:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":1,"date":"2021-04-07 11:49:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2021-04-08 02:29:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":2,"date":"2021-04-07 12:10:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-07 11:30:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":-1,"date":"2021-04-08 13:17:09.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":-1,"date":"2021-04-08 02:32:23.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"date":"2021-04-08 11:58:43.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-05 13:10:08.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2021-04-05 13:10:08.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2021-04-05 13:10:33.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-04-05 15:59:57.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-05 16:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 08:26:39.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2021-04-06 08:29:42.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-04-07 11:30:27.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2021-04-08 13:17:09.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-04-19 20:27:41.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REMOVED"}],"messages":[{"id":"fd858ab4fa634f4d6cdca77b1378fe6b62d44211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-05 13:09:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf1d47d3b2a764fa20703ebc0998c5f64700e249","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-05 13:10:33.000000000","message":"Patch Set 1: Code-Review-1\n\nlet\u0027s break this into two patches...\n1. for integration\n2. one for upstream","accounts_in_message":[],"_revision_number":1},{"id":"1f1d31fe85202a477229cd557b97d334188ed981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-05 13:18:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ce498ed59744c21c53dd7bdb760635b547dc2a7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-05 15:59:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f90774266a744692816230cf97eebb5c : SUCCESS in 2h 40m 22s","accounts_in_message":[],"_revision_number":2},{"id":"ff78c26cff1483d62ccbfe9cfabaed1317ff2ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 16:20:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4a7ee3e6dfbe4a91aa88b069fe39e7ff : SUCCESS in 2h 49m 33s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4bd359ce3c904244863e55225b387495 : SUCCESS in 1h 05m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5b1fa6d8b9604fc38f17c448f0629583 : SUCCESS in 1h 46m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/38de1e4ba97e449d947c4dafc71d8ab9 : SUCCESS in 6m 24s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/a6f265a49bef41c687f1e44013b86204 : SUCCESS in 2h 40m 45s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/846a868aa27a4a148c013371759f49d8 : SUCCESS in 2h 53m 08s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/102e2914a2454f4eb81b41527d796552 : SUCCESS in 2h 48m 04s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/5272ab13211c4731bab8b99cb843ef81 : SUCCESS in 1h 34m 16s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/be844b9e1e674195aa653d3145284311 : SUCCESS in 1h 48m 56s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3d853aa14f4b494d9c08654245182567 : SUCCESS in 1h 54m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39320885d07d4a17b4e3eca8b3bb7e65 : SUCCESS in 4m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a792a1b12e9f4a61916a92e7c04cb85c : SUCCESS in 1h 21m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/e4b588c07a5942d68c1e12287c65314a : SUCCESS in 8m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3e608d16d3864655ab4648b5b21be891 : FAILURE in 1h 19m 38s","accounts_in_message":[],"_revision_number":2},{"id":"0745d8a527fb0e42f7e1e4f74cd58d47acc1bcd9","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-04-06 07:20:42.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f4f04bcde017aeff402e754ba8bb75cfee3179d6","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-06 08:26:39.000000000","message":"Patch Set 2:\n\nthe update job failure seems real\n2021-04-05 15:29:41 | 2021-04-05 15:29:41.706267 | fa163e80-a0fb-4f6d-0f81-00000000002c |      FATAL | Ensure DNF modules have the right stream | centos-8-stream-inap-mtl01-0023852791 | item\u003d{\u0027module\u0027: \u0027container-tools\u0027, \u0027stream\u0027: \u00272.0\u0027} | error\u003d{\"ansible_loop_var\": \"item\", \"changed\": false, \"failures\": [], \"item\": {\"module\": \"container-tools\", \"stream\": \"2.0\"}, \"msg\": \"Depsolve Error occured: \\n Problem: cannot install both podman-1.6.4-23.module_el8.4.0+643+ef36beeb.x86_64 and podman-3.0.1-6.module_el8.5.0+736+58cc1a5a.x86_64\\n  - package podman-catatonit-3.0.1-6.module_el8.5.0+736+58cc1a5a.x86_64 requires podman \u003d 3.0.1-6.module_el8.5.0+736+58cc1a5a, but none of the providers can be installed\\n  - conflicting requests\\n  - problem with installed package podman-catatonit-3.0.1-6.module_el8.5.0+736+58cc1a5a.x86_64\\n  - package podman-3.0.1-6.module_el8.5.0+736+58cc1a5a.x86_64 is filtered out by modular filtering\", \"rc\": 1, \"results\": []}","accounts_in_message":[],"_revision_number":2},{"id":"8aa2ad9165ff54132b256e9c4027ff1b2529e799","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-06 08:26:53.000000000","message":"Patch Set 2:\n\n( https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_3e6/784768/2/check/tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates/3e608d1/logs/undercloud/home/zuul/overcloud_update_run-Controller.log )","accounts_in_message":[],"_revision_number":2},{"id":"52443e4adbd469ff48ab8bf640718908249303d0","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-06 08:29:42.000000000","message":"Patch Set 2: Code-Review-1\n\nDnfStreams in tht needs to be updated as well","accounts_in_message":[],"_revision_number":2},{"id":"ceb0fc7fe633ee5a00d717a0ed4c947f6b08729b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-04-06 08:31:08.000000000","message":"Patch Set 2:\n\nhttps://review.opendev.org/c/openstack/tripleo-heat-templates/+/784890","accounts_in_message":[],"_revision_number":2},{"id":"70fa577be64a5d3f18189cef84ac31e66976ba9d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-06 08:31:12.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e DnfStreams in tht needs to be updated as well\n\nhttps://review.opendev.org/c/openstack/tripleo-heat-templates/+/784800","accounts_in_message":[],"_revision_number":2},{"id":"81364ec19d786c014f8ab7def2c0fcfe681f11ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-06 09:36:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"07190d9ab4e3c4c52970ed894c6b3ce23852824f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-06 12:02:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/22a26b92f63f4557b191c3c1bf8ce78e : FAILURE in 2h 23m 59s","accounts_in_message":[],"_revision_number":3},{"id":"0af8e160492ab47c1c321ee03a5a41d4cebe2264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:21:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8530d42a148a4658a22cb7c040a8f6bb : SUCCESS in 3h 28m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4037615959564671ab0ab4972ce0f8fe : SUCCESS in 1h 16m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d2af7f05da2045709c80407b38521f4e : SUCCESS in 1h 51m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f1cf1dbf5c8945f6b2efab9e31c64338 : SUCCESS in 6m 55s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/16a7f56e4de64402b28922954663dedb : SUCCESS in 3h 11m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c2c6d7f7b99943fd902a2c0c69d5fa1c : SUCCESS in 3h 10m 08s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1b77596be45c4f6aa03b2eca796ac838 : SUCCESS in 3h 18m 32s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/a49f107c1c034c03909269e292267f98 : SUCCESS in 1h 58m 43s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/1575abe6352243d5a94a8a8e9b7a7ac6 : SUCCESS in 1h 59m 48s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/3968d56ed3ad476284667b9125f752c3 : SUCCESS in 2h 04m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0e2e2d391b04b5b9cc73f18cfae9b2d : SUCCESS in 4m 52s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/7db4d6a5e4cc491593803a641c0bb30e : SUCCESS in 1h 39m 16s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/fb9dfe6cafa6465bb48c8fb8691cc46e : SUCCESS in 11m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e2e4b74fadbb405e8e50a4b31feb192f : SUCCESS in 1h 56m 32s","accounts_in_message":[],"_revision_number":3},{"id":"826deaa0998e6cdbf9c00f94ec0212f2a325e256","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-04-07 08:29:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9b90ed2fbe71e366cdff3552724cb1e899035aea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-07 08:31:20.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"5a18136f5789ca47a287a4fe4796f724902103eb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-07 08:32:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1c6f83f0e4b8f0b96231e9d5ceab8c12ca4c4ddd","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2021-04-07 11:30:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c22a093e05c4e067a119d0b36338fef98c557a77","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-07 11:49:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a87ade48d82e4a3181ae81e15d2cc358 : SUCCESS in 3h 15m 53s","accounts_in_message":[],"_revision_number":4},{"id":"3dd1292c4eaed216529ccf9cb3c5abcfcdc8c358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:01:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ec332faffb7b4c0cb8e49cf2a5debe91 : SUCCESS in 3h 07m 59s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/611895c8f4a7451f826fd2c85f949b41 : SUCCESS in 1h 08m 45s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2250f87ae23441fbb337125722da702a : SUCCESS in 1h 41m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77a3df3427a842b4b66b50ab5b064779 : SUCCESS in 7m 43s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/23344a6cdace44e5846269370a604c8f : SUCCESS in 3h 18m 56s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/4ba4bbeeab554c2c889857831c1c8af6 : SUCCESS in 3h 19m 06s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1da9bcc3fa78444da83324d6cdae382c : SUCCESS in 3h 17m 21s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/cbc3db0e49f94395bd219d5109c70f0e : SUCCESS in 1h 50m 50s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/6a4ac44fd7ec43f09c5525c0a7049d20 : SUCCESS in 1h 53m 09s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/d323058555a141eb91b545d06852324e : SUCCESS in 1h 57m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31d8ae522064e85b109783dc8306d75 : SUCCESS in 5m 45s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2cfc3753583f406fad25fe1ab346178d : SUCCESS in 1h 20m 59s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/df40af40b4b247008fe1899656e67684 : SUCCESS in 9m 31s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/39020dcfc1994469b78367f304b28090 : FAILURE in 1h 19m 32s","accounts_in_message":[],"_revision_number":4},{"id":"a55d6c9143c2df7a3d1207d8567f810f0286c5af","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-04-07 12:10:07.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"1631ed26082a209e96ecf9386d5078710defa3bf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-07 15:09:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1316a9e93d5f4ee3816c82a2afbacc12 : SUCCESS in 2h 46m 36s","accounts_in_message":[],"_revision_number":4},{"id":"9a9fb3874407ac9dbce8cb8fb19b7de5000afff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 15:31:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/893a93c0de8245a0b338a0ce7d9d00a4 : SUCCESS in 2h 52m 49s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/04552b39e66141d7ab4201ad33927983 : SUCCESS in 1h 40m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c919a22db2f643d482c0339b7138a6f7 : SUCCESS in 1h 37m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ae5971b9058e4a1994bade3e32dde0d0 : SUCCESS in 6m 00s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/50f273a0265d4c09acc52b95e6ea5aae : SUCCESS in 3h 12m 58s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a32b319c38a9490dadaecf01531c712b : SUCCESS in 2h 57m 37s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1f2cb2be0727485dbc55aeefab63211e : SUCCESS in 2h 42m 48s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/2ec3f419c5f9482e8b1a74d580f6db79 : SUCCESS in 1h 55m 40s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/11ec3e159940433aaff18df5f34ef290 : SUCCESS in 1h 53m 00s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/8eacce1e395f432dba93261ab2eb8f60 : SUCCESS in 1h 48m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ed3c0eea7b4fa2ac4afff6c99cebb3 : SUCCESS in 5m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/282faa956dbf4cdba8b84854de2cbbb7 : SUCCESS in 1h 20m 05s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/2f41a966df194cd295c84b1b385994b1 : SUCCESS in 10m 01s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b0be74e2ddc74489882a0c072dc01012 : SUCCESS in 1h 35m 50s","accounts_in_message":[],"_revision_number":4},{"id":"f1079203e14372eb14f178246456cdf9f4803bc5","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2021-04-07 15:55:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"647905212df4548a36fe24265c476ba824dff295","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 02:29:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7af83c7ffbcaee9afd4b076e3657d43aeea0a5c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 02:29:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"42673260160265ec47dec05bc8d9bda11b42ed14","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 02:31:58.000000000","message":"Abandoned\n\nfak.. promoter fail","accounts_in_message":[],"_revision_number":4},{"id":"8ebe0e3eea0ce2330f9d7f6d1e6d34c4ec02a8e5","tag":"autogenerated:gerrit:restore","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 02:32:16.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"629f2aaea56e561f08a23732590f5665e7a45798","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 02:32:23.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"eea46d42144dfad53c0c53a2d380d6c3dc02132d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 02:39:49.000000000","message":"Abandoned\n\nfak.. promoter fail","accounts_in_message":[],"_revision_number":4},{"id":"b540f49ff456632fa688ae9038173dab8447bff4","tag":"autogenerated:gerrit:restore","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-04-08 02:39:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"18d7dbfc978a2d940573297150da3cb376e657f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-08 05:05:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8d95342ddbbb4f82951e04b114da350e : SUCCESS in 2h 23m 34s","accounts_in_message":[],"_revision_number":4},{"id":"7a05473d28ea5112d10a4780397fe8efeb14a2f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 05:39:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/91345a400fe84b769771f259175ee143 : SUCCESS in 2h 47m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2af921d20625436f81ff91c61dad80b7 : SUCCESS in 1h 11m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f561246e06dd426eb4f083cf27fc2647 : SUCCESS in 1h 39m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59725560e2cf4fe1885012a49b09b2b6 : SUCCESS in 7m 58s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d6421436c4454a94953ec4cd15db8b2e : SUCCESS in 2h 46m 55s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6585f84553354a798ca7df5eca77e417 : SUCCESS in 2h 34m 04s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d1532725aad040459d05a541c098ec36 : SUCCESS in 2h 48m 10s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/0fddb74103564361a0193f314c55de72 : SUCCESS in 1h 46m 51s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/d66858df5cd446a9921a035a5a755a2d : FAILURE in 1h 29m 01s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/59d469c15cdc47d596667491d7132aaa : SUCCESS in 1h 51m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f88d8a4a4ba4b1ba8518cc3529d799b : SUCCESS in 5m 41s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e135a65fff844d569e23aac7896b8a87 : SUCCESS in 1h 16m 19s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/85788b52ab8d4f6b85b9a2096490234b : SUCCESS in 9m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/dc72659374d34c0ebfbc215e9f1c75b7 : SUCCESS in 1h 38m 07s","accounts_in_message":[],"_revision_number":4},{"id":"ed09e90f458ef478b2f1c59703d02faec95a263a","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-04-08 07:39:05.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d91f53be7ef9472e97cd72eb9d0c3380c1e5dd62","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-08 11:05:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/090e5723ab95478697a3518472a28ebf : SUCCESS in 2h 50m 10s","accounts_in_message":[],"_revision_number":4},{"id":"2cebaa98dcfb53b6253d8614a1280fcc6b25e728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:34:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8cf42daaa6a444e59514e5c5ef3b7845 : SUCCESS in 3h 05m 55s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cb83a43f904a49ebabc838c1fa313621 : SUCCESS in 1h 10m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/151bddf5349446dbb732ed4cad4f3921 : SUCCESS in 1h 46m 57s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/327723cc93724b1ea7b7cc2e0bd969fb : SUCCESS in 6m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/f6d0438240224a05817ebc0db1c1ebb2 : SUCCESS in 3h 16m 49s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/9601aca7d1bc4c2a9ef06d94865a1ad7 : SUCCESS in 3h 01m 59s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/207e76320d384b77ba05d2f18dc8c78f : SUCCESS in 3h 32m 38s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/52b6510a07b542109540cf98ff82b951 : SUCCESS in 1h 59m 42s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/767f9cc8a68b4b67a62ed0d3873e04f6 : SUCCESS in 1h 49m 53s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/676a0b91ea58479384384ce4fe3eadd6 : SUCCESS in 2h 06m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19beead5cf54ffc9ec79fe858ec5a59 : SUCCESS in 4m 30s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4e86ce73c08d4abeb8968533740ab7bf : SUCCESS in 1h 20m 20s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/6fd90b38721c4f51b00546d5f57e79de : SUCCESS in 9m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e81bad549b064d35835d1ed2719f6d3f : SUCCESS in 1h 44m 06s","accounts_in_message":[],"_revision_number":4},{"id":"5477cb70095fb29f4e86fb67e7d049f58d092762","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-04-08 11:57:50.000000000","message":"Patch Set 4: Workflow+1\n\njsut reapplying +W","accounts_in_message":[],"_revision_number":4},{"id":"9fd736fd1a3308884a80ed82c5e3e4515fc30859","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-04-08 11:58:03.000000000","message":"Removed Workflow+1 by Francesco Pantano \u003cfpantano@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"ef969b49ef3dc3f2e5c883735cb983468b62dfba","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-04-08 11:58:43.000000000","message":"Patch Set 4:\n\nIt\u0027s better asking if for tripleo-ci team is safe merging this change now","accounts_in_message":[],"_revision_number":4},{"id":"916331c3289664b09d335293e98ca9fd42131bbf","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-04-08 13:17:09.000000000","message":"Patch Set 4: Workflow-1\n\nsplitting these changes into separate commits","accounts_in_message":[],"_revision_number":4},{"id":"82717c3adc0b6ee900addaff5bf0bcb87c55e61b","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2021-04-19 20:27:41.000000000","message":"Removed reviewer John Fulton with the following votes:\n\n* Code-Review+2 by John Fulton \u003cfulton@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"66a7f3d6a98ffdfa9b6402d36a774877719b1062","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-04-29 12:04:14.000000000","message":"Abandoned\n\nAll seperate patches are merged, abandoning it.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4a2ad62f4351fc6370c1373c734931b140354313","revisions":{"56c9766b6dbbbc51be90766899d48823a87355cd":{"kind":"REWORK","_number":1,"created":"2021-04-05 13:09:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/68/784768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/68/784768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/1"}}},"commit":{"parents":[{"commit":"9c734b154a3516bba55209af0cdcb5daeb795d47","subject":"Merge \"Fix quotes issue in stein release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9c734b154a3516bba55209af0cdcb5daeb795d47"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-05 13:08:31.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-05 13:08:31.000000000","tz":330},"subject":"[WIP] Bump to container-tools:3.0","message":"[WIP] Bump to container-tools:3.0\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ifdaa7d59f9a4a1dc54c32532e79fb8a10c20f637\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/56c9766b6dbbbc51be90766899d48823a87355cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/56c9766b6dbbbc51be90766899d48823a87355cd"}]},"branch":"refs/heads/master"},"f5cf3f38b6488910ac92db4e30a995eb66d9f427":{"kind":"REWORK","_number":2,"created":"2021-04-05 13:18:35.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/68/784768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/68/784768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/2"}}},"commit":{"parents":[{"commit":"9c734b154a3516bba55209af0cdcb5daeb795d47","subject":"Merge \"Fix quotes issue in stein release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9c734b154a3516bba55209af0cdcb5daeb795d47"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-05 13:17:02.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-05 13:17:02.000000000","tz":330},"subject":"Bump to container-tools:3.0 for upstream CI","message":"Bump to container-tools:3.0 for upstream CI\n\nChange-Id: Ifdaa7d59f9a4a1dc54c32532e79fb8a10c20f637\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f5cf3f38b6488910ac92db4e30a995eb66d9f427"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/f5cf3f38b6488910ac92db4e30a995eb66d9f427"}]},"branch":"refs/heads/master"},"caae4378554a41dfb49e87577818c5f02e125d5e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-04-06 09:36:24.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/68/784768/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/68/784768/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/3"}}},"commit":{"parents":[{"commit":"9c734b154a3516bba55209af0cdcb5daeb795d47","subject":"Merge \"Fix quotes issue in stein release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9c734b154a3516bba55209af0cdcb5daeb795d47"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-05 13:17:02.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-04-06 09:36:16.000000000","tz":0},"subject":"Bump to container-tools:3.0 for upstream CI","message":"Bump to container-tools:3.0 for upstream CI\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-heat-templates/+/784800\n\nChange-Id: Ifdaa7d59f9a4a1dc54c32532e79fb8a10c20f637\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/caae4378554a41dfb49e87577818c5f02e125d5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/caae4378554a41dfb49e87577818c5f02e125d5e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"4a2ad62f4351fc6370c1373c734931b140354313":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-04-07 08:31:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/68/784768/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/68/784768/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/68/784768/4"}}},"commit":{"parents":[{"commit":"9c734b154a3516bba55209af0cdcb5daeb795d47","subject":"Merge \"Fix quotes issue in stein release file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/9c734b154a3516bba55209af0cdcb5daeb795d47"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-04-05 13:17:02.000000000","tz":330},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2021-04-07 08:31:15.000000000","tz":0},"subject":"Bump to container-tools:3.0 for upstream CI","message":"Bump to container-tools:3.0 for upstream CI\n\nChange-Id: Ifdaa7d59f9a4a1dc54c32532e79fb8a10c20f637\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4a2ad62f4351fc6370c1373c734931b140354313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/4a2ad62f4351fc6370c1373c734931b140354313"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
