)]}'
{"id":"openstack%2Ftripleo-quickstart~802100","triplet_id":"openstack%2Ftripleo-quickstart~master~I8756800c26649e1dffb1aa367c9ad07422927692","project":"openstack/tripleo-quickstart","branch":"master","hashtags":[],"change_id":"I8756800c26649e1dffb1aa367c9ad07422927692","subject":"Switch default registry to quay.io","status":"MERGED","created":"2021-07-23 19:07:57.000000000","updated":"2021-07-27 00:39:44.000000000","submitted":"2021-07-27 00:38:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802100","meta_rev_id":"236e27b75bd61e3ee1b8f44faf8c197dca2bd6d1","_number":802100,"virtual_id_number":802100,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-27 00:38:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-07-26 12:26:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":2,"date":"2021-07-23 19:16:54.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2021-07-23 19:14:06.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2021-07-26 09:43:31.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-26 09:42:20.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2021-07-26 11:44:13.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-23 19:14:06.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-07-23 19:16:54.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-07-23 21:40:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 22:03:23.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2021-07-23 22:03:23.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-07-23 22:03:23.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-07-23 23:21:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-26 09:42:20.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"}],"messages":[{"id":"114252a836c9e1e58eeabd08d3984cca93bce207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-23 19:07:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18677dc3e84557e0756729d4e06f2e4bfdce86fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-23 19:11:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20836f09f2326a82f7d5c7bfb5448720ba5d1450","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-23 19:14:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"864b6a2b5cb03d197cd8868e7598538a66bb8a33","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-07-23 19:16:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6eae492a9c7ec6eac6452376a368c9417e7e9094","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-23 21:40:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2cf50fbc106041a4952f7e6392f0b28e : SUCCESS in 2h 26m 05s","accounts_in_message":[],"_revision_number":2},{"id":"3a3da827dde0e28f1349a332825274e3fc3b5ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 23:21:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/53dbc582e15a4773914c3cd4c759f876 : SUCCESS in 3h 44m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f4f757adb6244e05aaa41fb61a2b02a7 : SUCCESS in 43m 37s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5ed69e6a66234675a92e4d9a39e19afa : SUCCESS in 1h 31m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bfe030f237984732918ca5b718bba12a : SUCCESS in 5m 37s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/28e5d99e69d94802b97a7e7dbe2556cd : SUCCESS in 3h 00m 13s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/44182ed6acef4e6e91e6c3a364238f9b : SUCCESS in 3h 50m 01s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/89feeb029c9e482ea98ea78953d91527 : SUCCESS in 3h 58m 08s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/85cebf2fe8e24375b7cc9e0c04f23bba : SUCCESS in 3h 08m 59s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/f6effdf8df074bae86506edc28619760 : SUCCESS in 1h 45m 13s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/b7a8ac3eb9d145c3a190255c5019dca2 : SUCCESS in 2h 17m 34s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/73fd03f2695147ecabeb611f840277f2 : SUCCESS in 2h 10m 58s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/dcbad84b684e4c7998535bd008f421da : SUCCESS in 2h 05m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/480d200a65104f25aaaf1daf456496bb : SUCCESS in 4m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3a1ba5151b0249d1887c59c4b7cc51c3 : SUCCESS in 1h 13m 05s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4c93146390d34b46b4ba98cb70afe4e1 : SUCCESS in 9m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/171bf66587c844dbb78409d708b55f80 : SUCCESS in 1h 42m 34s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/87392d7a57004ab58fe06b0a181bd63d : FAILURE in 31m 43s","accounts_in_message":[],"_revision_number":2},{"id":"790a5de4f585cfe76da4469445bbfabf3b89c9e0","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-07-26 09:42:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3a454c1fa6b71b9733d43da729eeb2412c5fe9a4","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-07-26 09:42:40.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"91c3ec33fb3e803d252e94a95be6af9bbeb56647","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-26 09:43:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"13addcc8618af5252295a83d755b6f4c45e2efd9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-07-26 09:50:25.000000000","message":"Patch Set 2:\n\nNeeded for https://a574f9c0fd4ca92b7603-2045be852d43868eb95da6cc3429b40d.ssl.cf2.rackcdn.com/801738/3/check/tripleo-ci-centos-8-standalone/44d1279/logs/undercloud/home/zuul/standalone_deploy.log and https://review.opendev.org/c/openstack/ansible-role-python_venv_build/+/801738/","accounts_in_message":[],"_revision_number":2},{"id":"5e4c890b832a2879b0831dd6b1ab21ddc62ec562","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-07-26 11:44:13.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d9a54eddadf8fe5a2cb50090b45d82c38f7c6b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-26 12:26:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6535e736d6314879a6ff6907e8b7fce6 : FAILURE in 2h 43m 00s","accounts_in_message":[],"_revision_number":2},{"id":"6823b1fef1715d6cf78ae784596f5d65a0f51cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 13:20:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bf8bf0cd5dd14d76bc8d0fad0c39aca0 : SUCCESS in 2h 49m 31s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4a0b28af3b644f6ea0470faa7375eed2 : SUCCESS in 46m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/8b799815b08442a08631014ab84c9004 : SUCCESS in 1h 38m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2921c7a9cbad410da3dccbdd1cd028e1 : SUCCESS in 6m 00s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e086fc8e12254e4e9b7ba0591acb22a4 : SUCCESS in 2h 58m 12s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/9a86c7fc9f9f448994bca77d2b20cb52 : SUCCESS in 3h 28m 49s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/6ec64351083d4848bf09b91ae0d3042c : SUCCESS in 3h 25m 44s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/54b102203c72412c97f5fb690a87b90f : SUCCESS in 3h 24m 27s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/c7372dc33c60426389f80cf4643cd2ea : SUCCESS in 1h 44m 30s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/8ccde5895e444e7dad30473411d7efa9 : SUCCESS in 2h 01m 30s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/1a66a3d95acd4150b8f40495e359d3f0 : SUCCESS in 2h 01m 03s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/ef8f964625b5465696e91ab965e7937a : SUCCESS in 2h 07m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c3f1bf3f65e49a1a9bdcd0d3cd7bd86 : SUCCESS in 4m 51s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4b0cda74185d4f818ee2bef500af6445 : SUCCESS in 1h 20m 05s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/c246721e57ac40a2a9e9b64a59a3f79e : SUCCESS in 12m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/087993d5e1434d8bacdc64fef1559f3c : SUCCESS in 1h 30m 51s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/867d79be1e92433da4b2fb504173bdb5 : SUCCESS in 1h 13m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d300a63f0440c573788231519e017b3f50dd0abf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 13:20:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1a2feaa5189c0293af7c0ec4b3b4674c2c1c635f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 00:38:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4cbe001ea2ea444084447b81728dccef : SUCCESS in 3h 13m 16s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a06a43c66c424035a4cefbbd3a81d4b0 : SUCCESS in 53m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1b77b9c935ce4196b83a7b25fc14025e : SUCCESS in 1h 48m 39s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e032a4335f7a4c85b1d240298abcb131 : SUCCESS in 3h 21m 50s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ea89bd913a19400aa26e72bf16319976 : SUCCESS in 3h 57m 35s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ec22f7b7c7fc42928c85ec7ddf662629 : SUCCESS in 3h 41m 03s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/3d78029dc30c43a6afda2f576086f63e : SUCCESS in 3h 29m 42s\n- tripleo-ci-centos-8-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/42879510d8ca48788b00fa29e28c4103 : SUCCESS in 1h 57m 41s\n- tripleo-ci-centos-8-containers-multinode-victoria https://zuul.opendev.org/t/openstack/build/62b745c28bdb4b0d92d1921ab9e46107 : SUCCESS in 2h 17m 02s\n- tripleo-ci-centos-8-containers-multinode-ussuri https://zuul.opendev.org/t/openstack/build/894e71fa954f498eb65d9f827ea26797 : SUCCESS in 2h 07m 24s\n- tripleo-ci-centos-8-containers-multinode-train https://zuul.opendev.org/t/openstack/build/2f50620b1a8b40b08ef9bfea81e48a8f : SUCCESS in 2h 20m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c10f0a02b44719885fdfcd1cac9acb : SUCCESS in 4m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/78114bef9fd84674beac1e5408b4faaf : SUCCESS in 1h 28m 15s\n- openstack-tox-molecule https://zuul.opendev.org/t/openstack/build/4acc60660511497dbb4b23073ccc7a10 : SUCCESS in 9m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/997f1f7c65f74e08a029e8f7c81b3b83 : SUCCESS in 1h 39m 54s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b2eec60c871644e78f96462325f862a2 : SUCCESS in 1h 26m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b734f8a65e4b4a83a0620cff65388d03 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"702870e1d70bfd5203c0bead5f78d979e5ea02d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 00:38:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"236e27b75bd61e3ee1b8f44faf8c197dca2bd6d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 00:39:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad2efcb0d2b4f30868d7c412de51c5d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dab5d693607fbe2f9a0ed0bc45993acdb47f7a2b","revisions":{"29f28f7eb53407700519ee86d8239ecb75b3e3f2":{"kind":"REWORK","_number":1,"created":"2021-07-23 19:07:57.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/00/802100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/00/802100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/1"}}},"commit":{"parents":[{"commit":"d5504218ef647ea8410640e4d95d4a621cead70d","subject":"Merge \"Disable delorean repo in component jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d5504218ef647ea8410640e4d95d4a621cead70d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-07-23 19:07:22.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-07-23 19:07:22.000000000","tz":-360},"subject":"Switch default registry to quay.io","message":"Switch default registry to quay.io\n\nIf you don\u0027t build the containers, it\u0027s using docker.io by default even\nthough we\u0027ve switched to quay.io a while back.\n\nChange-Id: I8756800c26649e1dffb1aa367c9ad07422927692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/29f28f7eb53407700519ee86d8239ecb75b3e3f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/29f28f7eb53407700519ee86d8239ecb75b3e3f2"}]},"branch":"refs/heads/master"},"dab5d693607fbe2f9a0ed0bc45993acdb47f7a2b":{"kind":"REWORK","_number":2,"created":"2021-07-23 19:11:54.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/00/802100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/00/802100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/00/802100/2"}}},"commit":{"parents":[{"commit":"d5504218ef647ea8410640e4d95d4a621cead70d","subject":"Merge \"Disable delorean repo in component jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/d5504218ef647ea8410640e4d95d4a621cead70d"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-07-23 19:07:22.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-07-23 19:11:49.000000000","tz":-360},"subject":"Switch default registry to quay.io","message":"Switch default registry to quay.io\n\nIf you don\u0027t build the containers, it\u0027s using docker.io by default even\nthough we\u0027ve switched to quay.io a while back.\n\nChange-Id: I8756800c26649e1dffb1aa367c9ad07422927692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dab5d693607fbe2f9a0ed0bc45993acdb47f7a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/dab5d693607fbe2f9a0ed0bc45993acdb47f7a2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
