)]}'
{"id":"openstack%2Ftripleo-quickstart~843102","triplet_id":"openstack%2Ftripleo-quickstart~master~I91b8169602315cbcbee1931719ca70657796ab2c","project":"openstack/tripleo-quickstart","branch":"master","topic":"c9s","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-05-30 07:42:48.000000000","reason":"Marios Andreou replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-05-30 15:43:06.000000000","reason":"Change was submitted"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-05-30 15:43:06.000000000","reason":"Change was submitted"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-05-24 10:47:48.000000000","reason":"Chandan Kumar replied on the change"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-05-30 15:43:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91b8169602315cbcbee1931719ca70657796ab2c","subject":"Use Nodepool CentOS mirror for 9-stream SIG repos","status":"MERGED","created":"2022-05-24 06:12:41.000000000","updated":"2022-05-30 15:44:48.000000000","submitted":"2022-05-30 15:43:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"843102-c9s","meta_rev_id":"6b3f1d29a2fdc60ef14122a861f7c94aec1f41d2","_number":843102,"virtual_id_number":843102,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-30 15:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-05-24 08:25:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2022-05-30 07:42:48.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-05-24 10:47:48.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2022-05-30 07:42:48.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"CC":[{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-24 06:12:58.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-24 08:25:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 08:30:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-24 10:32:49.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-05-24 10:32:49.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-05-24 10:32:49.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-05-24 10:32:49.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"2df692077be99c4ba0c180e4b718ab074ccbec48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-24 06:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2f60c630401d3db6cbc5bfbbdfa1e842871e357","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-24 06:12:58.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1027/","accounts_in_message":[],"_revision_number":1},{"id":"7fb50ee8ea2a410fb851666ced1a5e2e0739fb5a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-05-24 06:40:58.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1027/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"d1a30c858b657a0999a34e357a9d2ec64de80dc5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-24 08:25:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/70ecc6c9a64d40ec845a278ac4cad5e9 : FAILURE in 1h 18m 20s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/7d508e3758d646dc97093a10861bd8ed : SUCCESS in 2h 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"cfb75ed2b22185a6c1370330b061dbb2b196f5f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 08:30:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4f97d82c58604770999a0de4dc1a0c93 : SUCCESS in 2h 11m 36s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/aacee942320e4758bcd7d2c621553300 : SUCCESS in 39m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/bef2548b4a7e4341bc331bf23c950cfb : SUCCESS in 1h 24m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e7de01d6b474488bad30c605ae69d279 : SUCCESS in 4m 18s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ea3cfd0302fd47578478ef6ee32863da : SUCCESS in 2h 06m 54s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/3a66fbe2c8fd475683b0191ceb41eb68 : SUCCESS in 1h 19m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cab299b3d1da4f1fbc9549af209aecdd : SUCCESS in 4m 33s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4eacbed9ad28449f96f33b7945eec722 : SUCCESS in 1h 14m 10s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/062f0dfe1b084d08ae0c6d19a1393a95 : SUCCESS in 1h 08m 19s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/8d02ef6b6e62416cb9bf0be90162baf3 : SUCCESS in 1h 03m 42s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/90c6b03f13994b209ab0cb8ec9deb447 : SUCCESS in 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"22f9888c3972f4e954bcd1991063a6c047d398d2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-24 10:47:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d377a6377592c100aa1eb9c25985c9e3c433a6b","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-05-30 07:42:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8275a4aedc31c2a1c96a1318b4fd3a0c9ba3ed64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:43:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b7cdc89c2fefc2f8ec7be0bcdb8255c76a43d59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 15:43:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7adb749f9cff443ba056c4895831c940 : SUCCESS in 2h 49m 13s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c5c6e3a958c542d2a2ba3a9ddcf23dfc : SUCCESS in 56m 12s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/656db01125c1479789481937250c6ddd : SUCCESS in 1h 38m 44s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/36c35e7c0d3143f2b15aa67c727ac638 : SUCCESS in 3h 12m 03s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/3509daae63a6472aaf02195e31458c6c : SUCCESS in 1h 47m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35391d5a4b84f7f82c7e011a6c277b7 : SUCCESS in 4m 38s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/5cc61d7597a84b31affb39b64c701d58 : SUCCESS in 1h 26m 47s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/d8c21265c7dc4242bf5d38723a7f695d : SUCCESS in 1h 14m 22s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4b950cd462484c9cb723eb864a616b70 : SUCCESS in 1h 30m 01s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/14e9e389db1d4e4f92cd15199ab0d252 : SUCCESS in 13m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/52f4329873804c6a8bcd4e6480cc9e1f : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"d8172c39d8949098612a96344ae499ffc92eb10f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 15:43:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6b3f1d29a2fdc60ef14122a861f7c94aec1f41d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 15:44:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5ea15cc96a42878a536af9e6db4e52 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cadea41d28e361baca26f7445835d9417c752a32","revisions":{"cadea41d28e361baca26f7445835d9417c752a32":{"kind":"REWORK","_number":1,"created":"2022-05-24 06:12:41.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/02/843102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/02/843102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/843102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/843102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/843102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/02/843102/1"}}},"commit":{"parents":[{"commit":"e2708e09d8c43b81c9f0c473cd5b9a08086487da","subject":"Merge \"Changing container registry for master c9 to quay\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/e2708e09d8c43b81c9f0c473cd5b9a08086487da"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-05-24 06:07:29.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-05-24 06:10:23.000000000","tz":330},"subject":"Use Nodepool CentOS mirror for 9-stream SIG repos","message":"Use Nodepool CentOS mirror for 9-stream SIG repos\n\ndelorean-deps repo contains SIG repos, let\u0027s replace\nmirror.stream.centos.org to nodepool centos proxy\nto avoid external mirror specific issues.\n8-stream release files already have this fixed.\n\nChange-Id: I91b8169602315cbcbee1931719ca70657796ab2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cadea41d28e361baca26f7445835d9417c752a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cadea41d28e361baca26f7445835d9417c752a32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
