)]}'
{"id":"openstack%2Ftripleo-quickstart~850398","triplet_id":"openstack%2Ftripleo-quickstart~master~I9b70723e17db851104811a65a0933bd86ee500b2","project":"openstack/tripleo-quickstart","branch":"master","topic":"collections-override","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2023-12-04 05:07:28.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9b70723e17db851104811a65a0933bd86ee500b2","subject":"[DNM] Test collections update in master","status":"ABANDONED","created":"2022-07-19 15:50:59.000000000","updated":"2023-12-04 05:07:28.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f922fd49dd72af0b575a36d391c9add038264ffd","_number":850398,"virtual_id_number":850398,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-09 11:28:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 15:51:17.000000000","updated_by":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 18:10:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 19:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 06:38:01.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"}],"messages":[{"id":"88cdeefbb9ecab6a64e6010fe950f4c8b8636ea4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-19 15:50:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7727a0bed310411649ad5e9746244e1b4fd70271","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-19 15:51:17.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1199/","accounts_in_message":[],"_revision_number":1},{"id":"9a3ba77535e225a4fbe9a1f9d3d3ab54e4917a93","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-19 15:55:00.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1199/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bf14705de24fb396d7e4387da66736acf64d62b9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-19 18:10:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c96e15d95f3e4ab0949455baeedd4117 : SUCCESS in 2h 17m 53s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/2683d10df4e444eb9eae987041f12ae1 : FAILURE in 1h 09m 16s","accounts_in_message":[],"_revision_number":1},{"id":"307a2615be0f132310c3843476902e1dec9eca97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 19:19:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d91c0b3ecb314b909d85072a26653cab\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/fc35c234811b4b0e83cde85c676916de : SUCCESS in 2h 17m 48s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/eaea8b7bf51b426d9047ba4e82fb9081 : FAILURE in 50m 30s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7660d5914b2644c1b00d769bac1bd66b : FAILURE in 55m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55c2e15189694087956207be37860fc9 : SUCCESS in 6m 13s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/be3bd02d7ca147e398ea4a8baa9c5eb7 : SUCCESS in 3h 15m 01s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/da2bd37bad144ed298150969a7a73fcf : FAILURE in 1h 57m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0662303931a34fffb402d9f7aad89d98 : SUCCESS in 6m 10s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/e2b2682edc184d78a68de948dd1212b7 : FAILURE in 52m 32s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/a5fdc41af5814decb86f6d261468f655 : FAILURE in 1h 01m 02s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3c71241c169b4790bb1d4bbd1d4acb44 : FAILURE in 54m 35s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/82b0fdab5fb849d3b7f2615e7f9ce1e7 : FAILURE in 46m 16s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/2dd00826bc924c0d84de07ba98a50f03 : SUCCESS in 18m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ae61f9a4be806027dd347db66617fc37c458f834","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2022-07-20 06:38:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b815567dc3494378b8be15683b22a7c772481d28","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-20 06:38:18.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1206/","accounts_in_message":[],"_revision_number":1},{"id":"41bde22b5faa39fb2f086f1a276afdfbb958475d","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-20 06:41:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"746e197b038f1452264d636e3b24dbb81196c76e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-20 06:42:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1206/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7efeeac447049e98146b96ffedc9cfecbf37b8c3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-20 06:42:14.000000000","message":"Patch Set 1:\n\nBuild Started https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1207/","accounts_in_message":[],"_revision_number":1},{"id":"74e68a65e481f152b7160db45bc9ee3e8e306ff0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":18846,"name":"RedHat RDO CI","email":"rdo-ci-admins@redhat.com","username":"rdo-ci","tags":["SERVICE_USER"]},"date":"2022-07-20 06:45:42.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052 https://jenkins-cloudsig-ci.apps.ocp.ci.centos.org/job/tripleo-quickstart-gate-master-tripleo-ci-delorean-full-featureset052/1207/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"dbcef19dfd2fa20744f4ab4a203dc45bab05afc7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-20 08:40:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/42910ca3ecbb49bb87a3387850716cbd : SUCCESS in 2h 01m 00s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001-wallaby https://review.rdoproject.org/zuul/build/8e22bd9d20454f18bc5e73abc9c30b4c : FAILURE in 1h 11m 45s","accounts_in_message":[],"_revision_number":1},{"id":"a50119349f7be009a756b7328a67feb020dcd9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 10:00:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5d294b171f1442f93ed0b1d89bbb035\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/34ff5962d71b48f4986857fc3ae1bf91 : SUCCESS in 2h 55m 55s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8074bc20abfb4640b4673798e7af1af4 : SUCCESS in 50m 02s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/fcabb392b4244f0f89f6576155ab7079 : SUCCESS in 1h 42m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5ae65220505d4c7db2f5adf2a9c2a635 : SUCCESS in 5m 51s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/882d7de4ce4740438e58a8e19c332913 : SUCCESS in 3h 11m 36s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/f1ce5abb000f404597154fc8cc27ae10 : FAILURE in 2h 01m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d90a5919284ecaaa45d5ada06b48c1 : SUCCESS in 5m 03s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/5b4a226cbb2e4b7fa48468e35ff065e8 : SUCCESS in 1h 25m 06s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/a9c70a1fa80841378408c53b860ebcb9 : FAILURE in 1h 46m 30s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c7fb4727312c44be94cc6fae568ef706 : SUCCESS in 1h 38m 09s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/fee5042e4950404db4f7de853dc9e1aa : FAILURE in 47m 40s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/73a92c28f77c48d8869236ae2c74a397 : SUCCESS in 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fe45d3b95ef2e23d95b914480205615400738b83","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-25 05:03:28.000000000","message":"Abandoned\n\nJust for tests, master now switched to ovs2.17 and ovn22.06 with https://review.rdoproject.org/r/c/rdoinfo/+/43728","accounts_in_message":[],"_revision_number":1},{"id":"9daca56291a5c1406f1b44caad5e55de3fa1c7d2","tag":"autogenerated:gerrit:restore","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-02 05:52:49.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"20813884e1a1859b57cbd2149fa217da28fe996a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2023-03-02 05:53:04.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/tripleo-quickstart-extras for 850308,1","accounts_in_message":[],"_revision_number":1},{"id":"7dfba960ef7c36985d62c939dc9724e4958afe56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-03-02 05:53:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f9816b155b2ae6e2d175336a6f0c76303ad8d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 05:53:09.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/tripleo-quickstart-extras for 850308,1","accounts_in_message":[],"_revision_number":1},{"id":"1a68d3a204bf6348f4e6c598b77c571d088b9e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 09:34:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1f54ef920424851b948528f1d61a0ec\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b98b01bba92845b68fcf2847244402ce : SUCCESS in 3h 24m 28s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/53119d26aa08467a9aaf81ff642ec952 : SUCCESS in 52m 06s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2604222c8fb64a6f8b7f8c9a86f966e3 : SUCCESS in 1h 49m 27s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56ef28ff2c5b4cfea632965683537b01 : SUCCESS in 3m 59s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/a75a6406f9054ce8958e2d81ed95f5fa : SUCCESS in 2h 28m 34s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/890eace9cd52468abf9130d535b69822 : SUCCESS in 3h 03m 19s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/d4bdb2970af94c2989c6262ac00ea2e4 : SUCCESS in 1h 34m 30s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/4bbae6dd27994a05b54a37ac6f478665 : SUCCESS in 1h 33m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b6417dc2774fc08a26dcc9f3cd40a6 : SUCCESS in 3m 46s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/837841bcdb55481b8ba24da9771f6d44 : SUCCESS in 1h 27m 03s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/c6f62f86df4c46cd9104a40f561cbe96 : FAILURE in 2h 00m 18s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/89597c2fbfb745edb1b4ae3635dd8dc9 : SUCCESS in 1h 44m 27s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/539b692cfbf74a018ae62e2d2f040e6e : SUCCESS in 59m 17s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/db175d417dfe4a2cb1caa263d122a14a : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a97bad61de3cef1a21c79bab2e96e8b319df3aa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2023-03-09 07:39:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"41444d800947f21f863dee43e9bb2b5d2adbff0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 11:28:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43279c5ccead4f72a5bc7dc3b2b6c2df\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/cd7794f147a141f2ae809e500e6dafed : SUCCESS in 3h 34m 39s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0b419c56f7ab4425b0629b2b06ef093c : SUCCESS in 56m 16s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/09da1447ae7a463daf8acdcf54a6135c : SUCCESS in 1h 45m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6adf4127ba9b4e448a566a8e498e3472 : SUCCESS in 3m 36s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/0982fff86a7b46caad80aa25a7f5c30c : SUCCESS in 3h 07m 43s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/408f53f3331c4728ba69d59f671a1258 : SUCCESS in 3h 04m 27s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/c14aa2801deb4df6a9c83bcfe44fcfdf : SUCCESS in 1h 50m 03s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/87be746ca10047b091bbee46a01ca8a5 : SUCCESS in 1h 48m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fbb036602d475fba02009c4d2ed0ce : SUCCESS in 3m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ce51a8a4d3e643348a717844228db7df : SUCCESS in 1h 40m 05s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/17f2e761bc8f479786d35ab0e0386e8e : SUCCESS in 2h 15m 25s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cb08499b7ddc4338bcb480815c128c81 : SUCCESS in 1h 51m 01s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/140c1caeadf1442a8d0c4a93da7c7e91 : SUCCESS in 1h 17m 10s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/c68be15c5fac4259a9a09deed75a951d : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f922fd49dd72af0b575a36d391c9add038264ffd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2023-12-04 05:07:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"919b527118171d758a1cc1b856dafdde35d5d130","revisions":{"6589630d3cee29869989b344e2572813a049b079":{"kind":"REWORK","_number":1,"created":"2022-07-19 15:50:59.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/98/850398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/98/850398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/1"}}},"commit":{"parents":[{"commit":"46b5df17c37a294f4c6ad91bf9f780c7115c6395","subject":"Merge \"Add new featureset66 and update tripleo-inventory for mixed OS jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/46b5df17c37a294f4c6ad91bf9f780c7115c6395"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-19 15:43:20.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-19 15:50:25.000000000","tz":330},"subject":"[DNM] Test ovs2.17/ovn22.06 in master","message":"[DNM] Test ovs2.17/ovn22.06 in master\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/850308\nChange-Id: I9b70723e17db851104811a65a0933bd86ee500b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6589630d3cee29869989b344e2572813a049b079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/6589630d3cee29869989b344e2572813a049b079"}]},"branch":"refs/heads/master"},"cb3585209dab81b0ce825979041c15f977d98cba":{"kind":"REWORK","_number":2,"created":"2023-03-02 05:53:07.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/98/850398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/98/850398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/2"}}},"commit":{"parents":[{"commit":"3ea6e712d20e9e928585f4f8b7f5ce048e181752","subject":"Merge \"Add NovaHWMachineType pin for mixed os job on rhel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3ea6e712d20e9e928585f4f8b7f5ce048e181752"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-19 15:43:20.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2023-03-02 05:53:13.000000000","tz":330},"subject":"[DNM] Test ovs3.1 in master","message":"[DNM] Test ovs3.1 in master\n\nChange-Id: I9b70723e17db851104811a65a0933bd86ee500b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cb3585209dab81b0ce825979041c15f977d98cba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/cb3585209dab81b0ce825979041c15f977d98cba"}]},"branch":"refs/heads/master"},"919b527118171d758a1cc1b856dafdde35d5d130":{"kind":"REWORK","_number":3,"created":"2023-03-09 07:39:07.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/98/850398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/98/850398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/98/850398/3"}}},"commit":{"parents":[{"commit":"3ea6e712d20e9e928585f4f8b7f5ce048e181752","subject":"Merge \"Add NovaHWMachineType pin for mixed os job on rhel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/3ea6e712d20e9e928585f4f8b7f5ce048e181752"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-19 15:43:20.000000000","tz":330},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2023-03-09 07:38:29.000000000","tz":60},"subject":"[DNM] Test collections update in master","message":"[DNM] Test collections update in master\n\nhttps://review.rdoproject.org/r/c/rdoinfo/+/47569\n\nChange-Id: I9b70723e17db851104811a65a0933bd86ee500b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/919b527118171d758a1cc1b856dafdde35d5d130"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/919b527118171d758a1cc1b856dafdde35d5d130"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
