)]}'
{"id":"openstack%2Ftripleo-quickstart~880224","triplet_id":"openstack%2Ftripleo-quickstart~master~I6860592ee05b73f855ee4371e5d6af2a1f6d8ed8","project":"openstack/tripleo-quickstart","branch":"master","topic":"retire-patrole","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-04-25 22:03:58.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2024-02-26 01:50:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6860592ee05b73f855ee4371e5d6af2a1f6d8ed8","subject":"Retire patrole projects","status":"ABANDONED","created":"2023-04-12 19:00:10.000000000","updated":"2024-02-26 01:50:19.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"58958fb9ada9765e97884a7ab5c6b91d0108cbab","_number":880224,"virtual_id_number":880224,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-18 18:02:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2023-04-13 08:46:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-12 22:02:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 08:46:24.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-04-25 22:03:58.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"}],"messages":[{"id":"3dba1e6a1841f87a3eb4c327d4d3bc21773fc75e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-04-12 19:00:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"315c3866d82131ade5b35565c7ed00220c4b69e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 22:02:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/589c65fe35834ea7ab1318e9f5ba7464\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/21407e26ed514af786c39d52bb199806 : FAILURE in 22m 31s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c305e70899b34f39977f2b784d791cd1 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/5568bf39bf8047ce8dcb6f51e9dc68e1 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cf7fc8c863f24731a987974058aa727c : SUCCESS in 3m 24s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/f777930f5b534269a3f9101cfa8577cb : SUCCESS in 2h 30m 54s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/560ef469fa78488b9213cacc4657134e : SUCCESS in 2h 12m 37s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/eaaac235066646c8a3cb957ca18a1815 : SUCCESS in 1h 28m 37s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/411311bea55340008406ae083edad778 : SUCCESS in 1h 23m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2bc35891f9645ca9996761fad6fb7a4 : SUCCESS in 3m 42s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ecfcbf23e8184a5e8471026e42d15b16 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/ec8d36ab680d4b02954a4d59e021e879 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b532bf46c710401bbb7371100e4fc34d : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/413cc939f6db4320a0b1d83573beab8e : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/59450c9bb04d4e2caa44deba33a35795 : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"589257bc9f33713c3262b4305db66bd160df1afc","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-13 08:46:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"66a64a26661ec513d4c199453e605c40fffcb0c9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-18 15:06:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cf96f62d0fed5640c55cbec68c1071a8cbc5ec0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 18:02:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80cda693d32943dd8e64998a54aac6ac\n\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/29bd7a9242a94bee9aec1d49ba8c3cf8 : SUCCESS in 2h 22m 32s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/dfbbcee75ecf4b91877d430674269ffb : SUCCESS in 39m 03s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3d33b4c5b9ee49a387ef71771a10df90 : SUCCESS in 1h 32m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa209ca31807420ba182db4c1152a0bb : SUCCESS in 3m 33s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/bc5cca87e5194222a2d504bc281f5106 : SUCCESS in 2h 36m 57s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1393b48aa9cf458586e71bcca449f223 : SUCCESS in 2h 12m 44s\n- tripleo-ci-centos-9-containers-multinode-zed https://zuul.opendev.org/t/openstack/build/06e22b9485ca44608f093393117945cb : SUCCESS in 1h 37m 59s\n- tripleo-ci-centos-9-containers-multinode-wallaby https://zuul.opendev.org/t/openstack/build/c619c8adceea46e6916b59677b175dc8 : SUCCESS in 1h 30m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df00fdbfe71c4980ac5f2c4165a77ac9 : SUCCESS in 3m 35s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a1656117278a4ac5a1e324d79a58f5aa : SUCCESS in 1h 18m 07s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/8c1e788b9ccd4a519b27e437ad473237 : FAILURE in 1h 27m 10s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b42389aedf548f4aaa0a19b2104bf1f : SUCCESS in 1h 23m 04s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/3d6788a86e8c42118895efd30306a35d : SUCCESS in 59m 18s\n- tripleo-tox-molecule https://zuul.opendev.org/t/openstack/build/092f41c73dfd45bf90526acac0884443 : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"58958fb9ada9765e97884a7ab5c6b91d0108cbab","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-26 01:50:19.000000000","message":"Abandoned\n\nTrieplO project is retiring now, for details, please see https://review.opendev.org/c/openstack/governance/+/905145  or reach out to OpenStack TC.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"25b87f7db94388da35df2694514dc1b521fdcafd","revisions":{"25b87f7db94388da35df2694514dc1b521fdcafd":{"kind":"REWORK","_number":1,"created":"2023-04-12 19:00:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/24/880224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-quickstart","ref":"refs/changes/24/880224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/880224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/880224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/880224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-quickstart refs/changes/24/880224/1"}}},"commit":{"parents":[{"commit":"b97252d4d9bd52882037305e9ac876f2345ae025","subject":"Drop python3-sqlalchemy exclude from wallaby release file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/b97252d4d9bd52882037305e9ac876f2345ae025"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-04-12 18:59:06.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-04-12 18:59:06.000000000","tz":-300},"subject":"Retire patrole projects","message":"Retire patrole projects\n\npatrole project is not maintained and in QA\nPTG, we decided to retire it.\n\n- https://etherpad.opendev.org/p/qa-bobcat-ptg\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/880014\nChange-Id: I6860592ee05b73f855ee4371e5d6af2a1f6d8ed8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/25b87f7db94388da35df2694514dc1b521fdcafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-quickstart/commit/25b87f7db94388da35df2694514dc1b521fdcafd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
