)]}'
{"id":"openstack%2Ftripleo-ui~559021","triplet_id":"openstack%2Ftripleo-ui~master~I748f5151949327ace7f7d824cc5c29a00df0437a","project":"openstack/tripleo-ui","branch":"master","topic":"bp/config-download-ui","hashtags":[],"change_id":"I748f5151949327ace7f7d824cc5c29a00df0437a","subject":"Add deployment status tracking infrastructure","status":"MERGED","created":"2018-04-05 08:28:18.000000000","updated":"2018-05-16 15:26:56.000000000","submitted":"2018-05-16 15:26:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"559021-1526484416694-4330fb6e","meta_rev_id":"65d7b4c560209885ae38adb1113bf30cd6c2bd87","_number":559021,"virtual_id_number":559021,"owner":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2018-05-16 05:37:37.000000000","_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},{"value":0,"date":"2018-05-14 16:42:26.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"value":2,"date":"2018-05-16 15:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"recommended":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"all":[{"value":1,"date":"2018-05-16 06:42:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":2,"date":"2018-05-16 12:23:27.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":2,"date":"2018-05-15 15:08:08.000000000","_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"all":[{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":1,"date":"2018-05-16 13:04:40.000000000","_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-09 14:11:55.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2018-05-14 16:42:26.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2018-05-16 06:42:58.000000000","updated_by":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"reviewer":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"state":"REVIEWER"},{"updated":"2018-05-16 12:23:27.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2018-05-16 13:04:40.000000000","updated_by":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"reviewer":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"state":"REVIEWER"},{"updated":"2018-05-16 15:26:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6f77e1aa801ee99ca12d652c36f2b83c6d5eaee","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-05 08:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"af47e7bf7703df7af33bd0d34bc76d7da192a310","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-05 08:37:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/1/check/nodejs-npm-run-lint/f28a2f7/ : SUCCESS in 3m 55s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/1/check/nodejs-npm-run-test/0781322/ : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"fbbd7f0f237f3da54f9f93e2cae7419aa2fae6cd","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-05 09:40:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"75b07912fbeb75b906b559d71397f26575581542","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-05 09:55:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/2/check/nodejs-npm-run-lint/3ac36fc/ : SUCCESS in 12m 53s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/2/check/nodejs-npm-run-test/77bd70e/ : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0219a784adf23891e9495ff97dcafa2d791da1a1","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-05 12:34:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6413b9d8be3941dedd867af4f036a7ad65e33a73","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-05 12:36:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ca11f6ed9e11051ffe01599a9ae2a523582d9080","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-05 12:47:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/4/check/nodejs-npm-run-lint/d6ff1c2/ : SUCCESS in 4m 32s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/4/check/nodejs-npm-run-test/f1cf319/ : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":4},{"id":"c3ad8d36e1972627bcdcbd28cded89ac8ad26fd3","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-04-05 13:37:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nMissing release note","accounts_in_message":[],"_revision_number":4},{"id":"a4e7eb5a6caf090743c13bbbe60cfbf81d3b35b7","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-06 09:14:12.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b5b72733c09e0f830585a4e41aa31ed504ed4c0f","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-06 09:19:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ce74d55a2f24287fc90dfe2829c302dddad5e4df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-06 09:27:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/5/check/nodejs-npm-run-lint/200b9c5/ : SUCCESS in 5m 00s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/5/check/nodejs-npm-run-test/f54bd36/ : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":5},{"id":"25134494851020da02f12065eb4f5faa867903c9","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-12 11:33:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f31ffd78ea13d1d88394f06c2f8f123fef009811","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-12 12:02:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/6/check/nodejs-npm-run-lint/7f90419/ : SUCCESS in 4m 24s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/6/check/nodejs-npm-run-test/de2738c/ : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":6},{"id":"68410790b90f83145ddda7b927a4636b3746306f","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-04-16 12:48:32.000000000","message":"Patch Set 6:\n\nIs this ready for review?  You said that you\u0027d turn it into WIP but I don\u0027t see any indication of that.","accounts_in_message":[],"_revision_number":6},{"id":"18ba3e42aed94ce9219f6f6408a9208ecb25e2e6","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-26 13:23:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3ba3a476a0971625b893a8eb0c66ea205b275524","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-26 13:40:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/7/check/nodejs-npm-run-lint/419f14f/ : SUCCESS in 5m 03s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/7/check/nodejs-npm-run-test/91d728b/ : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":7},{"id":"7ff152f7502f9eadfd80666d2cd90bbe7170a711","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-26 14:09:34.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"afe91a7fb41ab0bd23c987357a840700baf5f64c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-26 14:18:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/8/check/nodejs-npm-run-lint/e444c61/ : SUCCESS in 4m 32s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/8/check/nodejs-npm-run-test/c2c1bfc/ : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":8},{"id":"9de4d7c3f13ade588ef1ea626b31a46d9c4da2fd","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2018-04-26 22:02:56.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"17ac22ffffd1d2aaaf62d4a1f867c54d74ac7c19","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-27 08:19:52.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2980ef0138c6f68cc7358d32d1cd843dede23659","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-04-27 08:21:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"02084dd662a23dd64499ace780e937cb644a9d42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 09:40:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/9/check/nodejs-npm-run-lint/65bce1f/ : SUCCESS in 4m 17s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/9/check/nodejs-npm-run-test/7bdd455/ : SUCCESS in 6m 05s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/9/check/tripleo-ci-centos-7-undercloud-oooq/3d137ba/ : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":9},{"id":"e8f5f9e2f79797c19637ec9bab9693116f1c8930","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2018-05-02 13:15:47.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"5255ae6b49ee4c503aff4fefa9e930e2bafc7549","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-02 13:38:16.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"950dbcea1edda2a70ceed9fb1518047516631ddf","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-02 13:39:07.000000000","message":"Patch Set 9:\n\nWe could also use some tests here.","accounts_in_message":[],"_revision_number":9},{"id":"53307d3c73636b6c836fdf42d4d5e386d1db29ce","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-02 13:50:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2a581e0e2fbbaa6a593f0f79705036e97e79e563","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 15:10:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/10/check/nodejs-npm-run-lint/3972fb2/ : SUCCESS in 4m 28s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/10/check/nodejs-npm-run-test/f9b8ad5/ : SUCCESS in 6m 57s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/10/check/tripleo-ci-centos-7-undercloud-oooq/151367a/ : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":10},{"id":"c45e626bb10a6865301d819ec7324750433c301f","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-02 15:54:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"283151c057035c6fc7b62db9b7b4e6abb6a2e99c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-02 17:27:24.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/11/check/nodejs-npm-run-lint/6eaf69e/ : SUCCESS in 5m 16s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/11/check/nodejs-npm-run-test/96577af/ : SUCCESS in 6m 24s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/11/check/tripleo-ci-centos-7-undercloud-oooq/8a22ac3/ : SUCCESS in 1h 22m 56s","accounts_in_message":[],"_revision_number":11},{"id":"41beac8add3e4c641cb10ba0ecb01940a7757fc0","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-02 17:39:39.000000000","message":"Patch Set 11: Code-Review-1\n\nWhen the deployment_status.yaml file can\u0027t be found in Swift, we cannot ignore the error, and dispatch the success function.  This masks the error, and wrongly sets the deployment as UNDEPLOYED.  The code works correctly when there is actual data coming from the server: if the payload is undefined, the Map() is empty, and the deployment status is set to the default of the DeploymentStatus record.  When the deployment_status.yaml file can\u0027t be found, we should set the status to UNKNOWN.","accounts_in_message":[],"_revision_number":11},{"id":"6970b34a345f5cc0d6de5eb2dd3581db9891fdf0","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-03 08:21:50.000000000","message":"Patch Set 11:\n\n\u003e When the deployment_status.yaml file can\u0027t be found in Swift, we\n \u003e cannot ignore the error, and dispatch the success function.  This\n \u003e masks the error, and wrongly sets the deployment as UNDEPLOYED. \n \u003e The code works correctly when there is actual data coming from the\n \u003e server: if the payload is undefined, the Map() is empty, and the\n \u003e deployment status is set to the default of the DeploymentStatus\n \u003e record.  When the deployment_status.yaml file can\u0027t be found, we\n \u003e should set the status to UNKNOWN.\n\nI agree, problem is that deployment_status.yaml is currently created when deploy_plan workflow is started. Until then, that object in swift does not exist. That\u0027s why non existence of the object is currently treated as UNDEPLOYED. (The error is not ignored, it is explicitly treated as a valid option which determines non deployed deployment)\n\nTo change this, we would have to update create_plan workflow to set deployment status and update delete_plan workflow to allow deleting plan only if deployment status is UNDEPLOYED and as part of it delete whole \u003cdeploymentPlanName\u003e-messages swift container which stores deployment status as well as zaqar messages.\n\nIf you agree, how about doing these changes as part of subsequent patches which also introduce undeploy_plan workflow?","accounts_in_message":[],"_revision_number":11},{"id":"67e419c829c50037d8f06692cc3739160fdd5047","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2018-05-03 18:57:29.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"1bbf062ae2bfd4a714825074c6d8515d3c161679","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2018-05-03 20:21:28.000000000","message":"Patch Set 11:\n\nI agree, let\u0027s follow-up with subsequent patches","accounts_in_message":[],"_revision_number":11},{"id":"0ce122a3a3b0bb6881c4dc7b2f34cd72fc079662","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-04 13:11:30.000000000","message":"Patch Set 11:\n\n\u003e a valid option which determines non deployed deployment\n\nI have demonstrated this to be false for certain deployments.\n\nAll I\u0027m asking is that we make the UI more defensive when getting strange data from the server. This patch introduces a breaking change, and I\u0027d like to provide a smoother transition.  If you\u0027re happy to break existing deployments, and to remove the possibility of removing said deployments by using the UI, then fine, let\u0027s leave it alone.\n\nWe\u0027re also still missing tests, and you haven\u0027t addressed the *UI naming scheme comment (it breaks away from existing patterns).","accounts_in_message":[],"_revision_number":11},{"id":"98d8f9f92d48ddbfd0ce0ee270af8180512725a9","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-09 12:11:11.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f396567e8995d70097935f8b82cdd4f56d122c22","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-09 12:52:39.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"d7bedb78504b6cbd0a368972fde60b9e28b47636","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-09 13:47:44.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/12/check/nodejs-npm-run-lint/17d71d7/ : SUCCESS in 3m 46s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/12/check/nodejs-npm-run-test/c058804/ : SUCCESS in 7m 06s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/12/check/tripleo-ci-centos-7-undercloud-oooq/eded57e/ : SUCCESS in 1h 30m 37s","accounts_in_message":[],"_revision_number":12},{"id":"bdec358310b8def1621e28e9b88a9569f049fa4c","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2018-05-09 14:11:55.000000000","message":"Patch Set 12: Code-Review+1","accounts_in_message":[],"_revision_number":12},{"id":"b833bd130db1292be9430455ca444132f9284340","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-11 13:56:47.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"b04baab5c09f9559fa701e4d1f9ea67b7043b8d3","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-11 14:05:46.000000000","message":"Patch Set 13:\n\n(1 comment)\n\nI have added a set of tests. I thing it would be good to keep deployment status app state split into 2 reducers (deploymentStatusByPlan and deploymentStatusUI) as it nicely separates the user interface state and deployment status state. This makes reducers simpler. Both are tracked per deployment plan. We could even go one step further and separate configDownload messages into a separate reducer which would make CONFIG_DOWNLOAD_MESSAGE part of reducer even simpler as we could just set deployment status on GET_DEPLOYMENT_STATUS_SUCCESS instead of updating it due to keeping the configDownload messages around.","accounts_in_message":[],"_revision_number":13},{"id":"fb5ca7f0d282d9cb47a25e6016386381dc50af9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-11 15:19:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/13/check/nodejs-npm-run-lint/db8a225/ : SUCCESS in 4m 42s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/13/check/nodejs-npm-run-test/4dfdfc3/ : SUCCESS in 7m 24s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/13/check/tripleo-ci-centos-7-undercloud-oooq/ea285fc/ : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":13},{"id":"c614a722faacd4fcc3e2399458c33e25817d58e7","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2018-05-11 20:59:11.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"8c21fc8c3cf329e8893bded7968d41e5eead807f","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-14 13:33:12.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"1457066f534a6f6f736020b2f42d31f6b68795c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-14 16:38:34.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/14/check/nodejs-npm-run-lint/73c34c8/ : SUCCESS in 4m 48s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/14/check/nodejs-npm-run-test/f54bed2/ : SUCCESS in 6m 25s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/14/check/tripleo-ci-centos-7-undercloud-oooq/28342c1/ : FAILURE in 2h 57m 15s","accounts_in_message":[],"_revision_number":14},{"id":"ac447e64c7e5f0fd2c4154d3513416b1632f949b","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2018-05-14 16:42:26.000000000","message":"Patch Set 14:\n\nI noticed Zuul failed, I think you hit bug(s):\n\n- tripleo-ci-centos-7-undercloud-oooq: https://bugs.launchpad.net/bugs/1686542\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":14},{"id":"151dbe10ab62e9ffa9a50045a32e0d19e4731b08","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-14 16:53:25.000000000","message":"Patch Set 14: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"c99533620aab6ac62359cdeb0f059ca22e233b06","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-15 11:51:55.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"0636aa000af9b4e4afab3826af26b5484331c263","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-15 15:08:08.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"e686b0a197faa31eb108180ff20ffc325f62eb9b","author":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"date":"2018-05-16 05:37:37.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"997aa1184786f40e2d9a817bf96c090514baa374","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2018-05-16 06:42:58.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"c9c9c1461034360b27130d3ef2db59a6d5babff2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 06:55:57.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/14/check/nodejs-npm-run-lint/d821777/ : SUCCESS in 4m 34s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/14/check/nodejs-npm-run-test/ce4765b/ : SUCCESS in 7m 50s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/14/check/tripleo-ci-centos-7-undercloud-oooq/386c4d2/ : SUCCESS in 1h 14m 16s","accounts_in_message":[],"_revision_number":14},{"id":"3ee73d6b114fe55bc2994a0a744b656cae78796d","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2018-05-16 12:23:27.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"5bb506eb1beb3db4f1cee70eefa5957a839760d1","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2018-05-16 13:04:40.000000000","message":"Patch Set 14: Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"ae043b37ad3da99155099f739902c21bfbf61200","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 13:04:54.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"3ee28ae6e9dd824df19628cc53754dde213149dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 15:26:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"b7641c329be9fc07e3d9d35acb90e8c6ff6dffcd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-16 15:26:56.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- nodejs-npm-run-lint http://logs.openstack.org/21/559021/14/gate/nodejs-npm-run-lint/030e24c/ : SUCCESS in 4m 01s\n- nodejs-npm-run-test http://logs.openstack.org/21/559021/14/gate/nodejs-npm-run-test/b7861dd/ : SUCCESS in 7m 15s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/21/559021/14/gate/tripleo-ci-centos-7-undercloud-oooq/7ae557a/ : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"a50d4b71dc14f7213d1659fe993b729a4c7607df","revisions":{"becab351ab4e339418f8c2681d0bf20d6e9bfd22":{"kind":"REWORK","_number":1,"created":"2018-04-05 08:28:18.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/1"}}},"commit":{"parents":[{"commit":"efc5befca805be0a9b19ea003897b1659a62cd55","subject":"Don\u0027t pass history with message actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/efc5befca805be0a9b19ea003897b1659a62cd55"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download, it introduces actions, reducers, selectors and immutable records to track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/becab351ab4e339418f8c2681d0bf20d6e9bfd22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/becab351ab4e339418f8c2681d0bf20d6e9bfd22"}]},"branch":"refs/heads/master"},"3e914ecda530aad406d748794ea4e46ced2118b4":{"kind":"REWORK","_number":2,"created":"2018-04-05 09:40:48.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/2"}}},"commit":{"parents":[{"commit":"efc5befca805be0a9b19ea003897b1659a62cd55","subject":"Don\u0027t pass history with message actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/efc5befca805be0a9b19ea003897b1659a62cd55"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 09:40:25.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download, it introduces actions, reducers, selectors and immutable records to track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/3e914ecda530aad406d748794ea4e46ced2118b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/3e914ecda530aad406d748794ea4e46ced2118b4"}]},"branch":"refs/heads/master"},"0c6591fcd8beecde9bb4bdec7d453a48c96b6b86":{"kind":"REWORK","_number":3,"created":"2018-04-05 12:34:39.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/3"}}},"commit":{"parents":[{"commit":"efc5befca805be0a9b19ea003897b1659a62cd55","subject":"Don\u0027t pass history with message actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/efc5befca805be0a9b19ea003897b1659a62cd55"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 12:33:48.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download, it introduces actions, reducers, selectors and immutable records to track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/0c6591fcd8beecde9bb4bdec7d453a48c96b6b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/0c6591fcd8beecde9bb4bdec7d453a48c96b6b86"}]},"branch":"refs/heads/master"},"a54416610a2d3fd881f5b03f3bb7f617edce4896":{"kind":"REWORK","_number":4,"created":"2018-04-05 12:36:37.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/4"}}},"commit":{"parents":[{"commit":"efc5befca805be0a9b19ea003897b1659a62cd55","subject":"Don\u0027t pass history with message actions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/efc5befca805be0a9b19ea003897b1659a62cd55"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 12:36:21.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download, it introduces actions, reducers, selectors and immutable records to track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/a54416610a2d3fd881f5b03f3bb7f617edce4896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/a54416610a2d3fd881f5b03f3bb7f617edce4896"}]},"branch":"refs/heads/master"},"016bd9b7323b297f4ab0788e9d641923cf1089e1":{"kind":"REWORK","_number":5,"created":"2018-04-06 09:19:16.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/5"}}},"commit":{"parents":[{"commit":"1178bba4ed48b35b5019325844842077e07ca2e1","subject":"Merge \"Upgrade webpack to 4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/1178bba4ed48b35b5019325844842077e07ca2e1"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-06 09:14:32.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/016bd9b7323b297f4ab0788e9d641923cf1089e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/016bd9b7323b297f4ab0788e9d641923cf1089e1"}]},"branch":"refs/heads/master"},"956ff9dc7a452a3086042c7abea1fbeee45ed9de":{"kind":"REWORK","_number":6,"created":"2018-04-12 11:33:04.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/6"}}},"commit":{"parents":[{"commit":"1178bba4ed48b35b5019325844842077e07ca2e1","subject":"Merge \"Upgrade webpack to 4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/1178bba4ed48b35b5019325844842077e07ca2e1"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-12 11:32:47.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/956ff9dc7a452a3086042c7abea1fbeee45ed9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/956ff9dc7a452a3086042c7abea1fbeee45ed9de"}]},"branch":"refs/heads/master"},"325dcd6c8770be34f879b4d916755a35f530b4b6":{"kind":"REWORK","_number":7,"created":"2018-04-26 13:23:31.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/7"}}},"commit":{"parents":[{"commit":"aaf4698172c932c83db6fac45f7ffeb79865286c","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/aaf4698172c932c83db6fac45f7ffeb79865286c"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-26 13:22:38.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/325dcd6c8770be34f879b4d916755a35f530b4b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/325dcd6c8770be34f879b4d916755a35f530b4b6"}]},"branch":"refs/heads/master"},"e43b1498183fe34fde6a5914d3062c06d13eb95a":{"kind":"REWORK","_number":8,"created":"2018-04-26 14:09:34.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/8"}}},"commit":{"parents":[{"commit":"aaf4698172c932c83db6fac45f7ffeb79865286c","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/aaf4698172c932c83db6fac45f7ffeb79865286c"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-26 13:32:05.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/e43b1498183fe34fde6a5914d3062c06d13eb95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/e43b1498183fe34fde6a5914d3062c06d13eb95a"}]},"branch":"refs/heads/master"},"d899b82b08c92c7758bd10e2ace4af3ab8a710c5":{"kind":"REWORK","_number":9,"created":"2018-04-27 08:19:52.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/9"}}},"commit":{"parents":[{"commit":"b9b16b294f6c1286347236aa1978096119589226","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/b9b16b294f6c1286347236aa1978096119589226"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-27 08:13:21.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/d899b82b08c92c7758bd10e2ace4af3ab8a710c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/d899b82b08c92c7758bd10e2ace4af3ab8a710c5"}]},"branch":"refs/heads/master"},"3566d3af25284d2ca2a995d67cc7bd39962527c1":{"kind":"REWORK","_number":10,"created":"2018-05-02 13:50:54.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/10"}}},"commit":{"parents":[{"commit":"24640a2ed47fe75b73205908b03b70587d59853f","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/24640a2ed47fe75b73205908b03b70587d59853f"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-05-02 11:18:31.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/3566d3af25284d2ca2a995d67cc7bd39962527c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/3566d3af25284d2ca2a995d67cc7bd39962527c1"}]},"branch":"refs/heads/master"},"1509583eae36a06ce6ecccaa3015d694feeb3765":{"kind":"REWORK","_number":11,"created":"2018-05-02 15:54:15.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/11"}}},"commit":{"parents":[{"commit":"24640a2ed47fe75b73205908b03b70587d59853f","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/24640a2ed47fe75b73205908b03b70587d59853f"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-05-02 15:49:46.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/1509583eae36a06ce6ecccaa3015d694feeb3765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/1509583eae36a06ce6ecccaa3015d694feeb3765"}]},"branch":"refs/heads/master"},"d477004681885d13a2bb85f767e05b6794f49bbc":{"kind":"REWORK","_number":12,"created":"2018-05-09 12:11:11.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/12"}}},"commit":{"parents":[{"commit":"699ef97f55f196b634bfb30bf111da86aa5da661","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/699ef97f55f196b634bfb30bf111da86aa5da661"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-05-09 12:10:54.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/d477004681885d13a2bb85f767e05b6794f49bbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/d477004681885d13a2bb85f767e05b6794f49bbc"}]},"branch":"refs/heads/master"},"e5cda08547edbaebfd846ff57d17e261cea731f1":{"kind":"REWORK","_number":13,"created":"2018-05-11 13:56:47.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/13"}}},"commit":{"parents":[{"commit":"699ef97f55f196b634bfb30bf111da86aa5da661","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/699ef97f55f196b634bfb30bf111da86aa5da661"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-05-11 13:56:09.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/e5cda08547edbaebfd846ff57d17e261cea731f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/e5cda08547edbaebfd846ff57d17e261cea731f1"}]},"branch":"refs/heads/master"},"a50d4b71dc14f7213d1659fe993b729a4c7607df":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2018-05-14 13:33:12.000000000","uploader":{"_account_id":7509,"name":"Jiri Tomasek","email":"jtomasek@redhat.com","username":"jtomasek"},"ref":"refs/changes/21/559021/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/21/559021/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/21/559021/14"}}},"commit":{"parents":[{"commit":"699ef97f55f196b634bfb30bf111da86aa5da661","subject":"Add sanitizeMessage function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/699ef97f55f196b634bfb30bf111da86aa5da661"}]}],"author":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-04-05 07:30:00.000000000","tz":120},"committer":{"name":"Jiri Tomasek","email":"jtomasek@redhat.com","date":"2018-05-14 13:20:47.000000000","tz":120},"subject":"Add deployment status tracking infrastructure","message":"Add deployment status tracking infrastructure\n\nThis change is a first step of adding support for config-download,\nit introduces actions, reducers, selectors and immutable records\nto track deployment status\n\nImplements: blueprint config-download-ui\nChange-Id: I748f5151949327ace7f7d824cc5c29a00df0437a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/a50d4b71dc14f7213d1659fe993b729a4c7607df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/a50d4b71dc14f7213d1659fe993b729a4c7607df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
