)]}'
{"id":"openstack%2Ftripleo-ui~661113","triplet_id":"openstack%2Ftripleo-ui~master~Id342707ec1f32cdb2914fdcec4f1531e551ebd98","project":"openstack/tripleo-ui","branch":"master","topic":"retire_tripleo_ui","hashtags":[],"change_id":"Id342707ec1f32cdb2914fdcec4f1531e551ebd98","subject":"Retiring TripleO-UI","status":"MERGED","created":"2019-05-23 20:32:06.000000000","updated":"2019-06-03 15:51:05.000000000","submitted":"2019-06-03 15:51:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661113-1559577065771-af150b49","meta_rev_id":"2743e9f4fa74a21334dd1a2ba4d50cdb7a040a74","_number":661113,"virtual_id_number":661113,"owner":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"value":2,"date":"2019-06-03 15:51:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-06-03 15:27:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-06-03 15:50:10.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"value":1,"date":"2019-06-03 15:50:49.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 21:24:57.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-24 15:22:21.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-05-30 16:56:40.000000000","updated_by":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"reviewer":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"state":"REVIEWER"},{"updated":"2019-06-03 15:27:09.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-06-03 15:50:10.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-03 15:50:49.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-03 15:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee0c6e31d3060151e50380f4123f1be36f978cd3","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-05-23 20:32:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"80b2444562c7a649d1cd78cbbceb09e33b69fefc","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-23 21:24:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1b8fc683c77c23e09fed1bfe32c572a7080f42bf","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 04:57:26.000000000","message":"Patch Set 1:\n\nthis needs noop-jobs template in project-config/zuul.d/projects.yaml to work","accounts_in_message":[],"_revision_number":1},{"id":"527527329d3e4af56b937556f1b5bc4f75026030","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-24 04:57:58.000000000","message":"Patch Set 1: Code-Review-1\n\nMissing README - read https://docs.openstack.org/infra/manual/drivers.html#step-3-remove-project-content","accounts_in_message":[],"_revision_number":1},{"id":"f7964f0b81983bc55f675343bd02d31d052a214f","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-05-24 15:19:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b4f33493caca74665af4b4152c0b3e08d6913378","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2019-05-24 15:22:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9010f237e89ffdd191f473290fecf491a9619a72","author":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"date":"2019-05-24 15:23:04.000000000","message":"Patch Set 2:\n\nIt was already turned off, but I have gone one step further and removed all jobs. https://review.opendev.org/661332","accounts_in_message":[],"_revision_number":2},{"id":"d5539ae4f1ca3abf00ac0b43e28d92de25e8ebb3","author":{"_account_id":20970,"name":"Honza Pokorny","email":"honza@redhat.com","username":"hpokorny"},"date":"2019-05-30 16:56:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nWe should also retire tempest-tripleo-ui.","accounts_in_message":[],"_revision_number":2},{"id":"3080b9115faac501626c5ca539686cc402e7d66f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-05-31 06:33:09.000000000","message":"Patch Set 2:\n\nDon\u0027t you need tempest-tripleo-ui for stable branches?","accounts_in_message":[],"_revision_number":2},{"id":"c2d203aed70bd26fbe603b2fd6f12ea17fb5a6cd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-31 21:44:09.000000000","message":"Patch Set 2:\n\nSo i don\u0027t think we properly noop\u0027d the repo","accounts_in_message":[],"_revision_number":2},{"id":"4837c7feb60854beed07f2d1116c0c49864a21c9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-31 21:44:42.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"e5ad24cfaeda9f0257e69d16b2a6a64e5a162bd0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-06-03 15:27:09.000000000","message":"Patch Set 3: Code-Review+1\n\nrecheck - dependency merged","accounts_in_message":[],"_revision_number":3},{"id":"dbcc6b9911c13848ac702ce4a0ff647ca8f1b733","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:27:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop noop : SUCCESS in 0s","accounts_in_message":[],"_revision_number":3},{"id":"29b330f824cab097695a56678a67e4e7a1024ca6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-03 15:50:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f7d919484db2135f00defed6d95d23fb893ee175","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-03 15:50:49.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bf52c7e52cf8d0e36d2a483c81a64a19442c0821","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:51:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8273a0881997f2ca8331a2e355a84ba7142fa0cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:51:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop noop : SUCCESS in 0s","accounts_in_message":[],"_revision_number":3},{"id":"1a83edfc73e4353cd74950548f56d4f2ddbba40b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:51:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b6367383e4f8f1f51fecfdd90d3a4dc0d2cef073","revisions":{"5e9c566b6c958eae5038cfbbfeedea393b10d5c2":{"kind":"REWORK","_number":1,"created":"2019-05-23 20:32:06.000000000","uploader":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"ref":"refs/changes/13/661113/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/13/661113/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/1"}}},"commit":{"parents":[{"commit":"d1baef537f0746efceacfddd2fc671e3efa478d0","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/d1baef537f0746efceacfddd2fc671e3efa478d0"}]}],"author":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-23 20:31:07.000000000","tz":-360},"committer":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-23 20:31:07.000000000","tz":-360},"subject":"Retiring TripleO-UI","message":"Retiring TripleO-UI\n\nChange-Id: Id342707ec1f32cdb2914fdcec4f1531e551ebd98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/5e9c566b6c958eae5038cfbbfeedea393b10d5c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/5e9c566b6c958eae5038cfbbfeedea393b10d5c2"}]},"branch":"refs/heads/master"},"78d658f3c671f41aef43ba732bc6b3cb74cfb352":{"kind":"REWORK","_number":2,"created":"2019-05-24 15:19:00.000000000","uploader":{"_account_id":10112,"name":"Jason E. Rist","email":"jrist@redhat.com","username":"jrist"},"ref":"refs/changes/13/661113/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/13/661113/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/2"}}},"commit":{"parents":[{"commit":"d1baef537f0746efceacfddd2fc671e3efa478d0","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/d1baef537f0746efceacfddd2fc671e3efa478d0"}]}],"author":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-23 20:31:07.000000000","tz":-360},"committer":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-24 15:18:50.000000000","tz":-360},"subject":"Retiring TripleO-UI","message":"Retiring TripleO-UI\n\nChange-Id: Id342707ec1f32cdb2914fdcec4f1531e551ebd98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/78d658f3c671f41aef43ba732bc6b3cb74cfb352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/78d658f3c671f41aef43ba732bc6b3cb74cfb352"}]},"branch":"refs/heads/master"},"b6367383e4f8f1f51fecfdd90d3a4dc0d2cef073":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-31 21:44:42.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/13/661113/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-ui","ref":"refs/changes/13/661113/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-ui refs/changes/13/661113/3"}}},"commit":{"parents":[{"commit":"d1baef537f0746efceacfddd2fc671e3efa478d0","subject":"Merge \"Replace git.openstack.org URLs with opendev.org URLs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/d1baef537f0746efceacfddd2fc671e3efa478d0"}]}],"author":{"name":"Jason E. Rist","email":"jrist@redhat.com","date":"2019-05-23 20:31:07.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-05-31 21:44:40.000000000","tz":0},"subject":"Retiring TripleO-UI","message":"Retiring TripleO-UI\n\nDepends-On: https://review.opendev.org/#/c/661332\nChange-Id: Id342707ec1f32cdb2914fdcec4f1531e551ebd98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/b6367383e4f8f1f51fecfdd90d3a4dc0d2cef073"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-ui/commit/b6367383e4f8f1f51fecfdd90d3a4dc0d2cef073"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
