)]}'
{"id":"openstack%2Ftripleo-validations~486140","triplet_id":"openstack%2Ftripleo-validations~master~I2bba5f66e653d9ec4e26f74887261c8541054800","project":"openstack/tripleo-validations","branch":"master","hashtags":[],"change_id":"I2bba5f66e653d9ec4e26f74887261c8541054800","subject":"Don\u0027t modify case of role names in inventory","status":"MERGED","created":"2017-07-21 13:03:57.000000000","updated":"2017-07-26 13:56:23.000000000","submitted":"2017-07-26 13:56:23.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2377e1fd1072227a144abed1d4e3a9f838e2586","_number":486140,"virtual_id_number":486140,"owner":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-07-26 13:56:23.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"all":[{"value":2,"date":"2017-07-21 16:52:33.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":2,"date":"2017-07-24 08:28:25.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-07-26 12:24:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":2,"date":"2017-07-24 14:20:07.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2017-07-26 13:49:31.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-07-21 16:52:33.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2017-07-24 08:28:25.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2017-07-24 14:20:07.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2017-07-26 12:24:55.000000000","updated_by":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"reviewer":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"state":"REVIEWER"},{"updated":"2017-07-26 13:49:31.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-07-26 13:56:23.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"1128b8ed455a8deb47a9c6fdd904856b58b16362","author":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"date":"2017-07-21 13:03:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9ff9507a05bccb344cb015c51893f513be3e51f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-21 13:12:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-validations-docs-ubuntu-xenial http://docs-draft.openstack.org/40/486140/1/check/gate-tripleo-validations-docs-ubuntu-xenial/24fc50b//doc/build/html/ : SUCCESS in 2m 20s\n- gate-tripleo-validations-pep8-ubuntu-xenial http://logs.openstack.org/40/486140/1/check/gate-tripleo-validations-pep8-ubuntu-xenial/70e940a/ : SUCCESS in 2m 50s\n- gate-tripleo-validations-python27-ubuntu-xenial http://logs.openstack.org/40/486140/1/check/gate-tripleo-validations-python27-ubuntu-xenial/e690e42/ : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"08670b04649314d40722cb25e532fb894163187d","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2017-07-21 13:21:49.000000000","message":"Patch Set 1: Code-Review-1\n\nLooks good, but could you change the `hosts` values in the playbooks accordingly so they match what\u0027s in the inventory?","accounts_in_message":[],"_revision_number":1},{"id":"a4712e608fd9494534ef82283d73c84a07b97cce","author":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"date":"2017-07-21 15:46:59.000000000","message":"Patch Set 1:\n\n\u003e Looks good, but could you change the `hosts` values in the playbooks accordingly so they match what\u0027s in the inventory?\n\nAh yes, I didn\u0027t realize they were hard-coded - it\u0027d be good to figure out ways to avoid that in future so we can support validations of custom roles.","accounts_in_message":[],"_revision_number":1},{"id":"b8e3a1d64867829aecf905ac0082776c4a599fdb","author":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"date":"2017-07-21 15:50:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6bda0048244fad5ca5e02327964f82e1466a293d","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2017-07-21 16:07:51.000000000","message":"Patch Set 2: Code-Review-1\n\nThanks! I found just one more controller entry: \n\n  validations/controller-token.yaml\n\nCould you add that one as well?","accounts_in_message":[],"_revision_number":2},{"id":"08b6f442ea286be62188b1b9ffb8787792810866","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-21 16:23:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-validations-docs-ubuntu-xenial http://docs-draft.openstack.org/40/486140/2/check/gate-tripleo-validations-docs-ubuntu-xenial/7232923//doc/build/html/ : SUCCESS in 3m 19s\n- gate-tripleo-validations-pep8-ubuntu-xenial http://logs.openstack.org/40/486140/2/check/gate-tripleo-validations-pep8-ubuntu-xenial/015ee68/ : SUCCESS in 3m 51s\n- gate-tripleo-validations-python27-ubuntu-xenial http://logs.openstack.org/40/486140/2/check/gate-tripleo-validations-python27-ubuntu-xenial/da8dae9/ : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":2},{"id":"dbdb174e56414243afb546e457abe0cf319e8e5b","author":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"date":"2017-07-21 16:51:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"63e2bf41b29eb41df863356258bd69abd7ceb2ed","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2017-07-21 16:52:33.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"067b6838ae2eafa2de4caaac901ac9122dabd42a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-21 17:03:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-validations-docs-ubuntu-xenial http://docs-draft.openstack.org/40/486140/3/check/gate-tripleo-validations-docs-ubuntu-xenial/9db9b72//doc/build/html/ : SUCCESS in 3m 19s\n- gate-tripleo-validations-pep8-ubuntu-xenial http://logs.openstack.org/40/486140/3/check/gate-tripleo-validations-pep8-ubuntu-xenial/fda8f2e/ : SUCCESS in 4m 00s\n- gate-tripleo-validations-python27-ubuntu-xenial http://logs.openstack.org/40/486140/3/check/gate-tripleo-validations-python27-ubuntu-xenial/3cab666/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":3},{"id":"51b2b86ac7aa36ccf6c11c977e1e81d15ffad934","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2017-07-24 08:28:25.000000000","message":"Patch Set 3: Code-Review+2\n\nlgtm lets see if there are more comments otherwise we can merge later/tomorrow","accounts_in_message":[],"_revision_number":3},{"id":"dca2d87d1c90b7b2e6ff747ae79e86338ab4adf3","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2017-07-24 14:20:07.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5299e57e4c5cb2ef42ef9630966765d8d0dee3aa","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2017-07-26 12:24:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"16954c20d764fcacb57b8b1fad88c95e6a71a904","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-26 13:49:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"98dd3ab63353fe2de78918173a7dec70ea296dea","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-26 13:49:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"241fa612e3baa246d928c559738fe55fd990c699","date":"2017-07-26 13:56:23.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"a2377e1fd1072227a144abed1d4e3a9f838e2586","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-26 13:56:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tripleo-validations-docs-ubuntu-xenial http://docs-draft.openstack.org/40/486140/3/gate/gate-tripleo-validations-docs-ubuntu-xenial/76d9db0//doc/build/html/ : SUCCESS in 5m 01s\n- gate-tripleo-validations-pep8-ubuntu-xenial http://logs.openstack.org/40/486140/3/gate/gate-tripleo-validations-pep8-ubuntu-xenial/6a5a5cd/ : SUCCESS in 5m 38s\n- gate-tripleo-validations-python27-ubuntu-xenial http://logs.openstack.org/40/486140/3/gate/gate-tripleo-validations-python27-ubuntu-xenial/2dc1869/ : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25a20b672c66ed235364cccf454007a2529a75c8","revisions":{"01b3590b41812d7c45d4a9f304e8bed17143a70e":{"kind":"REWORK","_number":1,"created":"2017-07-21 13:03:57.000000000","uploader":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"ref":"refs/changes/40/486140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/40/486140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/1"}}},"commit":{"parents":[{"commit":"e6ebdd37bde1289380cdde7e1334a440a1144256","subject":"Add jsonschema validation for os-net-config data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e6ebdd37bde1289380cdde7e1334a440a1144256"}]}],"author":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2017-07-21 13:02:47.000000000","tz":60},"committer":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2017-07-21 13:02:47.000000000","tz":60},"subject":"Don\u0027t modify case of role names in inventory","message":"Don\u0027t modify case of role names in inventory\n\nWhen doing --limit \u003crole\u003e for ansible-playbook, it\u0027s quite unintuitive\nas you need to know to change the case, I would prefer if we made\nthese groups match exactly with what is in the roles_data.yaml and\noutput by the heat RoleData outputs.\n\nChange-Id: I2bba5f66e653d9ec4e26f74887261c8541054800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/01b3590b41812d7c45d4a9f304e8bed17143a70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/01b3590b41812d7c45d4a9f304e8bed17143a70e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e6ebdd37bde1289380cdde7e1334a440a1144256","is_merged_in_target_branch":true,"change_id":"Ic16ee0bc353c46f8fe512454176a07ee95347346","change_number":351345,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fe3bdbfe6b929245a7507c65f1b909b71e9cd4a3":{"kind":"REWORK","_number":2,"created":"2017-07-21 15:50:07.000000000","uploader":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"ref":"refs/changes/40/486140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/40/486140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/2"}}},"commit":{"parents":[{"commit":"e6ebdd37bde1289380cdde7e1334a440a1144256","subject":"Add jsonschema validation for os-net-config data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e6ebdd37bde1289380cdde7e1334a440a1144256"}]}],"author":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2017-07-21 13:02:47.000000000","tz":60},"committer":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2017-07-21 15:50:00.000000000","tz":60},"subject":"Don\u0027t modify case of role names in inventory","message":"Don\u0027t modify case of role names in inventory\n\nWhen doing --limit \u003crole\u003e for ansible-playbook, it\u0027s quite unintuitive\nas you need to know to change the case, I would prefer if we made\nthese groups match exactly with what is in the roles_data.yaml and\noutput by the heat RoleData outputs.\n\nChange-Id: I2bba5f66e653d9ec4e26f74887261c8541054800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fe3bdbfe6b929245a7507c65f1b909b71e9cd4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fe3bdbfe6b929245a7507c65f1b909b71e9cd4a3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e6ebdd37bde1289380cdde7e1334a440a1144256","is_merged_in_target_branch":true,"change_id":"Ic16ee0bc353c46f8fe512454176a07ee95347346","change_number":351345,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"},"25a20b672c66ed235364cccf454007a2529a75c8":{"kind":"REWORK","_number":3,"created":"2017-07-21 16:51:31.000000000","uploader":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"ref":"refs/changes/40/486140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/40/486140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/40/486140/3"}}},"commit":{"parents":[{"commit":"e6ebdd37bde1289380cdde7e1334a440a1144256","subject":"Add jsonschema validation for os-net-config data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e6ebdd37bde1289380cdde7e1334a440a1144256"}]}],"author":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2017-07-21 13:02:47.000000000","tz":60},"committer":{"name":"Steven Hardy","email":"shardy@redhat.com","date":"2017-07-21 16:51:25.000000000","tz":60},"subject":"Don\u0027t modify case of role names in inventory","message":"Don\u0027t modify case of role names in inventory\n\nWhen doing --limit \u003crole\u003e for ansible-playbook, it\u0027s quite unintuitive\nas you need to know to change the case, I would prefer if we made\nthese groups match exactly with what is in the roles_data.yaml and\noutput by the heat RoleData outputs.\n\nChange-Id: I2bba5f66e653d9ec4e26f74887261c8541054800\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/25a20b672c66ed235364cccf454007a2529a75c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/25a20b672c66ed235364cccf454007a2529a75c8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e6ebdd37bde1289380cdde7e1334a440a1144256","is_merged_in_target_branch":true,"change_id":"Ic16ee0bc353c46f8fe512454176a07ee95347346","change_number":351345,"patch_set_number":13,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
