)]}'
{"id":"openstack%2Ftripleo-validations~499352","triplet_id":"openstack%2Ftripleo-validations~master~I3102753b83ad80dc442f662aebc8b2484b6f824d","project":"openstack/tripleo-validations","branch":"master","topic":"tripleo/reno-versionless","hashtags":[],"change_id":"I3102753b83ad80dc442f662aebc8b2484b6f824d","subject":"Use Python to compute release notes version","status":"MERGED","created":"2017-08-30 21:32:51.000000000","updated":"2017-08-30 21:53:43.000000000","submitted":"2017-08-30 21:53:43.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8b494afb397fb588b95b90edadb1b019b9da5f67","_number":499352,"virtual_id_number":499352,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2017-08-30 21:53:43.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2017-08-30 21:39:38.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-08-30 21:48:32.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-08-30 21:48:32.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-30 21:39:38.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2017-08-30 21:48:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2017-08-30 21:53:43.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"df221364a06beb58f41da764f85800ed4a17e177","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-08-30 21:32:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd00d1389be9e5663f890c19fde3f8d5549c881c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-08-30 21:33:21.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"a36f708294eeeb578ac96210fe5e63c1a0bb712f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-08-30 21:39:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"031809532ce8c99fea4ed9efb4c3d8293c45985d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 21:43:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-validations-docs-ubuntu-xenial http://docs-draft.openstack.org/52/499352/2/check/gate-tripleo-validations-docs-ubuntu-xenial/f476e54//doc/build/html/ : SUCCESS in 4m 36s\n- gate-tripleo-validations-pep8-ubuntu-xenial http://logs.openstack.org/52/499352/2/check/gate-tripleo-validations-pep8-ubuntu-xenial/db0fb80/ : SUCCESS in 4m 49s\n- gate-tripleo-validations-python27-ubuntu-xenial http://logs.openstack.org/52/499352/2/check/gate-tripleo-validations-python27-ubuntu-xenial/c96f5de/ : SUCCESS in 4m 39s\n- gate-tripleo-validations-releasenotes http://docs-draft.openstack.org/52/499352/2/check/gate-tripleo-validations-releasenotes/9d8cb8b//releasenotes/build/html/ : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"6f72595409f14c65cf33193f9696bbcc1d919f92","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-08-30 21:48:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"120f0e6aab8830e46ef8e8a6705559bdf0fd3c9c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 21:48:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"f6f569530e2f034b773af38836585ce768c3dd1c","date":"2017-08-30 21:53:43.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"56c9055b006a21655104a54ebb8a3975b0616b10","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-30 21:53:43.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tripleo-validations-docs-ubuntu-xenial http://docs-draft.openstack.org/52/499352/2/gate/gate-tripleo-validations-docs-ubuntu-xenial/3aab328//doc/build/html/ : SUCCESS in 3m 40s\n- gate-tripleo-validations-pep8-ubuntu-xenial http://logs.openstack.org/52/499352/2/gate/gate-tripleo-validations-pep8-ubuntu-xenial/6ec92ad/ : SUCCESS in 3m 45s\n- gate-tripleo-validations-python27-ubuntu-xenial http://logs.openstack.org/52/499352/2/gate/gate-tripleo-validations-python27-ubuntu-xenial/7296795/ : SUCCESS in 4m 06s\n- gate-tripleo-validations-releasenotes http://docs-draft.openstack.org/52/499352/2/gate/gate-tripleo-validations-releasenotes/0996f66//releasenotes/build/html/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ec736ed1c57a16945e0c62b1187224ebd89c084","revisions":{"3686fb6c6100c31faabc27a34dcc865cc7b9e070":{"kind":"REWORK","_number":1,"created":"2017-08-30 21:32:51.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/52/499352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/52/499352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/1"}}},"commit":{"parents":[{"commit":"ddc025d4720e9d4fe276b6a82a91c08e4d3e5e46","subject":"Make sure to keep case when for host definition in host inventory.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ddc025d4720e9d4fe276b6a82a91c08e4d3e5e46"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-08-30 21:32:46.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-08-30 21:32:46.000000000","tz":-420},"subject":"Use Python to compute release notes version","message":"Use Python to compute release notes version\n\nLeave the version fields blank, since the release notes document\napplies to all versions.\nThat will avoid manual changes in the future like we did until now.\n\nChange-Id: I3102753b83ad80dc442f662aebc8b2484b6f824d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3686fb6c6100c31faabc27a34dcc865cc7b9e070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3686fb6c6100c31faabc27a34dcc865cc7b9e070"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ddc025d4720e9d4fe276b6a82a91c08e4d3e5e46","is_merged_in_target_branch":true,"change_id":"I7f46e4b7bdac5110d111537a044473c362d059f6","change_number":495975,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0ec736ed1c57a16945e0c62b1187224ebd89c084":{"kind":"REWORK","_number":2,"created":"2017-08-30 21:33:21.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/52/499352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/52/499352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/52/499352/2"}}},"commit":{"parents":[{"commit":"ddc025d4720e9d4fe276b6a82a91c08e4d3e5e46","subject":"Make sure to keep case when for host definition in host inventory.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ddc025d4720e9d4fe276b6a82a91c08e4d3e5e46"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-08-30 21:32:46.000000000","tz":-420},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2017-08-30 21:33:16.000000000","tz":0},"subject":"Use Python to compute release notes version","message":"Use Python to compute release notes version\n\nLeave the version fields blank, since the release notes document\napplies to all versions.\nThat will avoid manual changes in the future like we did until now.\n\nChange-Id: I3102753b83ad80dc442f662aebc8b2484b6f824d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0ec736ed1c57a16945e0c62b1187224ebd89c084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/0ec736ed1c57a16945e0c62b1187224ebd89c084"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"ddc025d4720e9d4fe276b6a82a91c08e4d3e5e46","is_merged_in_target_branch":true,"change_id":"I7f46e4b7bdac5110d111537a044473c362d059f6","change_number":495975,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
