)]}'
{"id":"openstack%2Ftripleo-validations~564739","triplet_id":"openstack%2Ftripleo-validations~stable%2Fnewton~I895fc83d148523c283664080b7b1bf75d53d9edb","project":"openstack/tripleo-validations","branch":"stable/newton","topic":"bug/1765121-stable/queens-stable/pike-stable/ocata-stable/newton","hashtags":[],"change_id":"I895fc83d148523c283664080b7b1bf75d53d9edb","subject":"Validate that there should not be XFS volumes with ftype\u003d0","status":"MERGED","created":"2018-04-27 11:52:50.000000000","updated":"2018-05-21 10:48:41.000000000","submitted":"2018-05-21 10:48:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"564739-1526899721735-211805f1","meta_rev_id":"c60a6e6d7972ab44e88ff194b54cce947ba5b04c","_number":564739,"virtual_id_number":564739,"owner":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":2,"date":"2018-05-21 10:48:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"all":[{"value":2,"date":"2018-05-21 08:08:47.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"all":[{"value":1,"date":"2018-05-21 08:08:47.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-21 10:48:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67c6e40d5df6554084532e75dc4a3d5bc5473e4b","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2018-04-27 11:52:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ocata.","accounts_in_message":[],"_revision_number":1},{"id":"4a503bdaed63c9fcfd756ed3c114112510ef7119","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2018-04-27 11:52:57.000000000","message":"Patch Set 1: Code-Review-2","accounts_in_message":[],"_revision_number":1},{"id":"71deb91e2d2f2d429799580fad01aaa48cd5657e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-27 13:49:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/39/564739/1/check/openstack-tox-pep8/218afd8/ : SUCCESS in 3m 38s\n- openstack-tox-py27 http://logs.openstack.org/39/564739/1/check/openstack-tox-py27/e3025b7/ : SUCCESS in 4m 09s\n- openstack-tox-py35 http://logs.openstack.org/39/564739/1/check/openstack-tox-py35/43358dd/ : SUCCESS in 3m 24s\n- build-openstack-sphinx-docs http://logs.openstack.org/39/564739/1/check/build-openstack-sphinx-docs/ab6c51b/html/ : SUCCESS in 4m 02s\n- build-openstack-releasenotes http://logs.openstack.org/39/564739/1/check/build-openstack-releasenotes/d3b3301/html/ : SUCCESS in 3m 24s\n- tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/39/564739/1/check/tripleo-ci-centos-7-nonha-multinode-oooq/f7b6287/ : SUCCESS in 1h 54m 46s","accounts_in_message":[],"_revision_number":1},{"id":"03561144c183b0820110e07eedf182fbf79e3cd9","author":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"date":"2018-05-21 08:08:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"049779bc3c2be58f7e79428813333564e90f39cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-21 08:09:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2832ac152990186e97230619ff7fa22d6917742","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-21 10:48:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a825a4c40923113b5603866609b1c4f0e83adc7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-21 10:48:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/39/564739/1/gate/openstack-tox-pep8/12cc447/ : SUCCESS in 3m 30s\n- openstack-tox-py27 http://logs.openstack.org/39/564739/1/gate/openstack-tox-py27/6c99a91/ : SUCCESS in 3m 57s\n- openstack-tox-py35 http://logs.openstack.org/39/564739/1/gate/openstack-tox-py35/e5d129d/ : SUCCESS in 4m 02s\n- build-openstack-sphinx-docs http://logs.openstack.org/39/564739/1/gate/build-openstack-sphinx-docs/6f0c0ca/html/ : SUCCESS in 3m 44s\n- build-openstack-releasenotes http://logs.openstack.org/39/564739/1/gate/build-openstack-releasenotes/18c5ced/html/ : SUCCESS in 3m 25s\n- tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/39/564739/1/gate/tripleo-ci-centos-7-nonha-multinode-oooq/82a0219/ : SUCCESS in 2h 10m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3130bc412917b7591b532ca792f6df909df21615","revisions":{"3130bc412917b7591b532ca792f6df909df21615":{"kind":"REWORK","_number":1,"created":"2018-04-27 11:52:50.000000000","uploader":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"ref":"refs/changes/39/564739/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/39/564739/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/39/564739/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/39/564739/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/39/564739/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/39/564739/1"}}},"commit":{"parents":[{"commit":"952d63318e97112a2bf7a4ca663f9b55778aba19","subject":"Merge \"Zuul: Remove project name\" into stable/newton","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/952d63318e97112a2bf7a4ca663f9b55778aba19"}]}],"author":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2018-04-18 13:41:58.000000000","tz":120},"committer":{"name":"Carlos Camacho","email":"ccamacho@redhat.com","date":"2018-04-27 11:52:50.000000000","tz":0},"subject":"Validate that there should not be XFS volumes with ftype\u003d0","message":"Validate that there should not be XFS volumes with ftype\u003d0\n\nIf  we have XFS partitions formatted with ftype\u003d0,\nthe upgrades will fail because ftype\u003d0 is incompatible\nwith the docker overlayfs driver.\n\nRelated-Bug: 1765121\nChange-Id: I895fc83d148523c283664080b7b1bf75d53d9edb\n(cherry picked from commit a5e63ee72531c85ee527ce721022fffce3cfcf75)\n(cherry picked from commit d54ae2d1e21245f7033107e1bae327621f54f82f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3130bc412917b7591b532ca792f6df909df21615"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3130bc412917b7591b532ca792f6df909df21615"}]},"parents_data":[{"branch_name":"refs/heads/stable/newton","commit_id":"952d63318e97112a2bf7a4ca663f9b55778aba19","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/newton"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
