)]}'
{"id":"openstack%2Ftripleo-validations~596120","triplet_id":"openstack%2Ftripleo-validations~stable%2Fpike~I2b59edfb4a90973e49e3c0e930d182baf6c275e6","project":"openstack/tripleo-validations","branch":"stable/pike","topic":"python3-first","hashtags":[],"change_id":"I2b59edfb4a90973e49e3c0e930d182baf6c275e6","subject":"import zuul job settings from project-config","status":"MERGED","created":"2018-08-24 06:14:56.000000000","updated":"2018-08-29 21:46:59.000000000","submitted":"2018-08-29 21:46:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"596120-1535579219486-be4e5e24","meta_rev_id":"c67c678d26ab8ae418bfd31a113979e95e496734","_number":596120,"virtual_id_number":596120,"owner":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2018-08-29 21:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"all":[{"value":2,"date":"2018-08-28 08:36:46.000000000","_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":1,"date":"2018-08-29 19:25:26.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-28 08:36:46.000000000","updated_by":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"reviewer":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"state":"REVIEWER"},{"updated":"2018-08-29 19:25:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2018-08-29 21:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"594a1be5b1b20637b6d08b1a33d370dba40ad4cd","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2018-08-24 06:14:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55a99b890522613bcd5ae9177a75314e555a77c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-24 09:09:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/596120/1/check/openstack-tox-pep8/0c673e5/ : SUCCESS in 5m 15s\n- openstack-tox-py27 http://logs.openstack.org/20/596120/1/check/openstack-tox-py27/e9d2cbc/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/20/596120/1/check/openstack-tox-py35/a0b31d1/ : SUCCESS in 13m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/596120/1/check/build-openstack-sphinx-docs/4075d95/html/ : SUCCESS in 4m 38s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/596120/1/check/tripleo-ci-centos-7-containers-multinode/d6ff58a/ : SUCCESS in 2h 29m 19s","accounts_in_message":[],"_revision_number":1},{"id":"2b2a5e4eddf0174e13c140d64d2d5766aa69b892","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2018-08-28 08:36:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90abd3b9ddfa38c792547a279e9ebfefe57523f1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-08-29 19:25:26.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2fa91b291d010b2d034ef8a604b4257fa5556205","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-29 19:25:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06e334b75a67fdb99a4b7541d894e4ff70b64b63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-29 21:46:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c7d1fa42d36bc42d6be6b1ced7bfabc772a19db7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-29 21:46:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/20/596120/1/gate/openstack-tox-pep8/3f4c267/ : SUCCESS in 5m 55s\n- openstack-tox-py27 http://logs.openstack.org/20/596120/1/gate/openstack-tox-py27/d857955/ : SUCCESS in 5m 27s\n- openstack-tox-py35 http://logs.openstack.org/20/596120/1/gate/openstack-tox-py35/e78795b/ : SUCCESS in 4m 35s\n- build-openstack-sphinx-docs http://logs.openstack.org/20/596120/1/gate/build-openstack-sphinx-docs/17cec67/html/ : SUCCESS in 4m 24s\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/20/596120/1/gate/tripleo-ci-centos-7-containers-multinode/254f219/ : SUCCESS in 2h 17m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"050a26fd252a50bba8211fb8b78f5db19bcce48e","revisions":{"050a26fd252a50bba8211fb8b78f5db19bcce48e":{"kind":"REWORK","_number":1,"created":"2018-08-24 06:14:56.000000000","uploader":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"ref":"refs/changes/20/596120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/20/596120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/596120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/596120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/20/596120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/20/596120/1"}}},"commit":{"parents":[{"commit":"578618e4c7598d3282c14c49cc341d14f4cb174f","subject":"Validate that there should not be XFS volumes with ftype\u003d0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/578618e4c7598d3282c14c49cc341d14f4cb174f"}]}],"author":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-08-24 05:05:08.000000000","tz":0},"committer":{"name":"qingszhao","email":"zhao.daqing@99cloud.net","date":"2018-08-24 05:05:08.000000000","tz":0},"subject":"import zuul job settings from project-config","message":"import zuul job settings from project-config\n\nThis is a mechanically generated patch to complete step 1 of moving\nthe zuul job settings out of project-config and into each project\nrepository.\n\nBecause there will be a separate patch on each branch, the branch\nspecifiers for branch-specific jobs have been removed.\n\nBecause this patch is generated by a script, there may be some\ncosmetic changes to the layout of the YAML file(s) as the contents are\nnormalized.\n\nSee the python3-first goal document for details:\nhttps://governance.openstack.org/tc/goals/stein/python3-first.html\n\nChange-Id: I2b59edfb4a90973e49e3c0e930d182baf6c275e6\nStory: #2002586\nTask: #24341\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/050a26fd252a50bba8211fb8b78f5db19bcce48e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/050a26fd252a50bba8211fb8b78f5db19bcce48e"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"578618e4c7598d3282c14c49cc341d14f4cb174f","is_merged_in_target_branch":true,"change_id":"I895fc83d148523c283664080b7b1bf75d53d9edb","change_number":564698,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
