)]}'
{"id":"openstack%2Ftripleo-validations~620227","triplet_id":"openstack%2Ftripleo-validations~stable%2Frocky~I8b19c905bda1a5590309d9fc1aa32ef5a22a25a7","project":"openstack/tripleo-validations","branch":"stable/rocky","topic":"fix-uc-container-check-stable/rocky","hashtags":[],"change_id":"I8b19c905bda1a5590309d9fc1aa32ef5a22a25a7","subject":"Fix containerized services task","status":"MERGED","created":"2018-11-27 07:14:26.000000000","updated":"2018-11-28 03:13:39.000000000","submitted":"2018-11-28 03:13:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"620227-1543374819427-125d40d0","meta_rev_id":"31e584a7aab2071c4e5f05fbc565ef027c5b25b2","_number":620227,"virtual_id_number":620227,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":2,"date":"2018-11-28 03:13:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"all":[{"value":2,"date":"2018-11-27 09:54:59.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"all":[{"value":1,"date":"2018-11-27 09:54:59.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2018-11-27 07:14:39.000000000","_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-27 07:14:39.000000000","updated_by":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"reviewer":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"state":"REVIEWER"},{"updated":"2018-11-27 09:54:59.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2018-11-28 03:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6eaeb902cbcbdf81bf351e0dccec5785f1962799","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2018-11-27 07:14:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5e728ac137f37aa931e50c93484f0151dcf08338","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2018-11-27 09:54:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"712023701fe7cd30342057fbc363f27a475e22fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 10:02:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/620227/1/check/tripleo-ci-centos-7-containers-multinode/32ac0d2/ : SUCCESS in 2h 34m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/27/620227/1/check/tripleo-ci-centos-7-undercloud-containers/3ef4f06/ : SUCCESS in 1h 30m 47s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/27/620227/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/610e25c/ : SUCCESS in 2h 30m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/27/620227/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/91be98c/ : SUCCESS in 2h 07m 28s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/27/620227/1/check/tripleo-ci-centos-7-standalone/4009b86/ : SUCCESS in 1h 45m 42s\n- openstack-tox-pep8 http://logs.openstack.org/27/620227/1/check/openstack-tox-pep8/c096876/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/27/620227/1/check/openstack-tox-py27/eb9ef3a/ : SUCCESS in 4m 53s\n- openstack-tox-py35 http://logs.openstack.org/27/620227/1/check/openstack-tox-py35/946cfa9/ : SUCCESS in 5m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/27/620227/1/check/build-openstack-sphinx-docs/4253b9a/html/ : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/620227/1/check/openstack-tox-lower-constraints/4a3ec27/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b3674c52f4a8444dfe00b282945f66f018c70be6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 10:02:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad70977114b21a629df485c8efeac374533d3b9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 03:13:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/620227/1/gate/tripleo-ci-centos-7-containers-multinode/aeefb4e/ : SUCCESS in 2h 19m 45s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/27/620227/1/gate/tripleo-ci-centos-7-standalone/6b7a058/ : SUCCESS in 1h 38m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/27/620227/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/08fe06b/ : SUCCESS in 2h 27m 16s\n- openstack-tox-pep8 http://logs.openstack.org/27/620227/1/gate/openstack-tox-pep8/5b910b3/ : SUCCESS in 7m 05s\n- openstack-tox-py27 http://logs.openstack.org/27/620227/1/gate/openstack-tox-py27/60c03c1/ : SUCCESS in 9m 16s\n- openstack-tox-py35 http://logs.openstack.org/27/620227/1/gate/openstack-tox-py35/0d9c340/ : SUCCESS in 6m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/27/620227/1/gate/build-openstack-sphinx-docs/d93df03/html/ : SUCCESS in 4m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/620227/1/gate/openstack-tox-lower-constraints/c86192e/ : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"366e1ce33c831bb1b5a46e9fc7e374e4d4bb1794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-28 03:13:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7c57888e66b60e7411285afd2a5dce1cea76f91","revisions":{"f7c57888e66b60e7411285afd2a5dce1cea76f91":{"kind":"REWORK","_number":1,"created":"2018-11-27 07:14:26.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/27/620227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/27/620227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/27/620227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/27/620227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/27/620227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/27/620227/1"}}},"commit":{"parents":[{"commit":"d396dfd5092c0b91af7e37bb01fe1c7bb982b027","subject":"Merge \"Fix the introspection interface look up\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d396dfd5092c0b91af7e37bb01fe1c7bb982b027"}]}],"author":{"name":"Florian Fuchs","email":"flfuchs@redhat.com","date":"2018-11-21 08:49:56.000000000","tz":60},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2018-11-27 07:14:26.000000000","tz":0},"subject":"Fix containerized services task","message":"Fix containerized services task\n\nOn a containerized undercloud the docker_facts module needs to be used\nwith become: yes.\n\nChange-Id: I8b19c905bda1a5590309d9fc1aa32ef5a22a25a7\nStory: #2004395\n(cherry picked from commit 48434586360b3785630317faa94d2024ddd8b311)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f7c57888e66b60e7411285afd2a5dce1cea76f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f7c57888e66b60e7411285afd2a5dce1cea76f91"}]},"parents_data":[{"branch_name":"refs/heads/stable/rocky","commit_id":"d396dfd5092c0b91af7e37bb01fe1c7bb982b027","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
