)]}'
{"id":"openstack%2Ftripleo-validations~625862","triplet_id":"openstack%2Ftripleo-validations~master~Id5bbbcc5bcf1fe247169c22504f4840f3403de09","project":"openstack/tripleo-validations","branch":"master","topic":"undercloud_conf","hashtags":[],"change_id":"Id5bbbcc5bcf1fe247169c22504f4840f3403de09","subject":"Retire undercloud_conf custom plugin","status":"MERGED","created":"2018-12-18 09:50:27.000000000","updated":"2018-12-21 21:39:29.000000000","submitted":"2018-12-21 21:39:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"625862-1545428369824-ba8ca464","meta_rev_id":"1d9e45383b549d2eda1dc1324861c0b0689ff1da","_number":625862,"virtual_id_number":625862,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":2,"date":"2018-12-21 21:39:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"all":[{"value":2,"date":"2018-12-18 11:16:37.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":2,"date":"2018-12-19 15:05:49.000000000","_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2018-12-19 15:05:49.000000000","_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-12-18 11:16:37.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2018-12-19 15:05:49.000000000","updated_by":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"reviewer":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"state":"REVIEWER"},{"updated":"2018-12-21 21:39:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81afb2f930d4f317ee2a6e35f38243359f32f117","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2018-12-18 09:50:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e43e565e8489d2fcf0ab3b8f2cd0ab41decf6be7","author":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"date":"2018-12-18 11:16:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b05cca4ed89a2f71baf5ce5a5feb5ee5de69990","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-18 14:23:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/625862/1/check/tripleo-ci-centos-7-containers-multinode/76e82e5/ : SUCCESS in 2h 18m 17s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/625862/1/check/tripleo-ci-centos-7-undercloud-containers/6b69a02/ : SUCCESS in 1h 43m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/625862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/8dc3e3d/ : SUCCESS in 2h 49m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/625862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/42cbf28/ : SUCCESS in 2h 01m 08s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/625862/1/check/tripleo-ci-centos-7-standalone/c1651bb/ : SUCCESS in 1h 41m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/625862/1/check/tripleo-ci-fedora-28-standalone/1aa5837/ : FAILURE in 33m 33s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/62/625862/1/check/tripleo-ci-centos-7-standalone-upgrade/89281a9/ : SUCCESS in 1h 56m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/62/625862/1/check/openstack-tox-pep8/067da65/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/62/625862/1/check/openstack-tox-py27/f54a752/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/62/625862/1/check/openstack-tox-py35/1cb69fa/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/62/625862/1/check/openstack-tox-py36/a17f844/ : SUCCESS in 4m 35s\n- openstack-tox-docs http://logs.openstack.org/62/625862/1/check/openstack-tox-docs/6cb0b26/html/ : SUCCESS in 4m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/625862/1/check/openstack-tox-lower-constraints/acb5b0b/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d14e8e1a1e249cbc78f8b02e95561f12f2581d5a","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2018-12-19 15:05:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"614fbd4ef1968d1f098f60e88dcbc45e36d4619d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 17:08:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbc09be915142e335f8a75b2bbaa6f0dba957dc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 21:39:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/625862/1/gate/tripleo-ci-centos-7-containers-multinode/98b00fd/ : SUCCESS in 1h 46m 03s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/625862/1/gate/tripleo-ci-centos-7-standalone/e3d500c/ : SUCCESS in 2h 53m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/625862/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/52e1f40/ : SUCCESS in 1h 54m 46s\n- openstack-tox-pep8 http://logs.openstack.org/62/625862/1/gate/openstack-tox-pep8/0359953/ : SUCCESS in 6m 16s\n- openstack-tox-py27 http://logs.openstack.org/62/625862/1/gate/openstack-tox-py27/c1301c6/ : SUCCESS in 4m 18s\n- openstack-tox-py35 http://logs.openstack.org/62/625862/1/gate/openstack-tox-py35/6f02c11/ : SUCCESS in 5m 00s\n- openstack-tox-py36 http://logs.openstack.org/62/625862/1/gate/openstack-tox-py36/0b0b07b/ : SUCCESS in 4m 14s\n- openstack-tox-docs http://logs.openstack.org/62/625862/1/gate/openstack-tox-docs/1ce9e5b/html/ : SUCCESS in 4m 13s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/625862/1/gate/openstack-tox-lower-constraints/5e465e6/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"5dc26a08d0200bcefc9be08d2e32fe1b5690ba58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-12-21 21:39:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d1c7ce44b79680162db43f9ed009afd3e0b2761b","revisions":{"d1c7ce44b79680162db43f9ed009afd3e0b2761b":{"kind":"REWORK","_number":1,"created":"2018-12-18 09:50:27.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/62/625862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/62/625862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/625862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/625862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/625862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/62/625862/1"}}},"commit":{"parents":[{"commit":"e69d6045a4d3ea130f193be365d00356be3e2d2e","subject":"Fix dhcp-provisioning validation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e69d6045a4d3ea130f193be365d00356be3e2d2e"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2018-12-18 09:46:19.000000000","tz":60},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2018-12-18 09:46:19.000000000","tz":60},"subject":"Retire undercloud_conf custom plugin","message":"Retire undercloud_conf custom plugin\n\nThe \u0027undercloud_conf\u0027 custom plugin is buggy and has been replaced by the \u0027ini\u0027\nplugin on the each validations which was using the \u0027undercloud.conf\u0027.\n\nChange-Id: Id5bbbcc5bcf1fe247169c22504f4840f3403de09\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d1c7ce44b79680162db43f9ed009afd3e0b2761b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d1c7ce44b79680162db43f9ed009afd3e0b2761b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e69d6045a4d3ea130f193be365d00356be3e2d2e","is_merged_in_target_branch":true,"change_id":"Ieb38d974fbe2291a2cd3bb1a9d4df9cd971bd098","change_number":625859,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
