)]}'
{"id":"openstack%2Ftripleo-validations~651500","triplet_id":"openstack%2Ftripleo-validations~master~I0dae1b4dc3c945c66e03aa7b361b329f9a25cc99","project":"openstack/tripleo-validations","branch":"master","topic":"bp/validation-framework","hashtags":[],"change_id":"I0dae1b4dc3c945c66e03aa7b361b329f9a25cc99","subject":"Add nova-status Ansible role","status":"MERGED","created":"2019-04-10 11:16:59.000000000","updated":"2019-06-13 09:29:40.000000000","submitted":"2019-06-12 17:47:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651500-1560361655396-91b8215b","meta_rev_id":"2bd7a360b75e8f0ad1fa200bf6d6f3538287350b","_number":651500,"virtual_id_number":651500,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-05-22 07:29:13.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":2,"date":"2019-06-12 17:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2019-06-12 14:54:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 09:53:43.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":1,"date":"2019-06-12 14:54:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-11 15:22:55.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2019-05-20 12:58:43.000000000","updated_by":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"reviewer":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"state":"REVIEWER"},{"updated":"2019-05-20 12:58:51.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-05-22 07:29:13.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-06-07 09:53:43.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-12 14:54:23.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-12 17:47:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bc54e35c8158b0032b6700831881771d23b1231","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-04-10 11:16:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c567ed70983629badedef9d52cfb92f5e5545ca","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-04-10 11:17:25.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"759cc4f56a4db03ddc464d405072808df1be70e0","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-04-10 11:42:32.000000000","message":"Patch Set 1:\n\nW-1 because needs some real tests","accounts_in_message":[],"_revision_number":1},{"id":"e8a73be96c71076ec54ad28574b617e1d8df0d50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-10 14:02:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/651500/1/check/tripleo-ci-centos-7-containers-multinode/3b0ecd7/ : SUCCESS in 2h 16m 27s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/651500/1/check/tripleo-ci-centos-7-undercloud-containers/ea9a011/ : SUCCESS in 1h 31m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/00/651500/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/938482a/ : SUCCESS in 2h 17m 34s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/00/651500/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8decd47/ : SUCCESS in 2h 44m 37s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/00/651500/1/check/tripleo-ci-centos-7-standalone/4cbfd73/ : SUCCESS in 1h 40m 39s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/00/651500/1/check/tripleo-ci-fedora-28-standalone/0bb26d5/ : SUCCESS in 1h 45m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/00/651500/1/check/tripleo-ci-centos-7-standalone-upgrade/745017c/ : FAILURE in 1h 10m 56s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/00/651500/1/check/openstack-tox-pep8/b855fff/ : SUCCESS in 6m 45s\n- openstack-tox-py27 http://logs.openstack.org/00/651500/1/check/openstack-tox-py27/bb0e1a0/ : SUCCESS in 4m 43s\n- openstack-tox-py35 http://logs.openstack.org/00/651500/1/check/openstack-tox-py35/f026bb8/ : SUCCESS in 4m 56s\n- openstack-tox-py36 http://logs.openstack.org/00/651500/1/check/openstack-tox-py36/6ec6ac5/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/00/651500/1/check/openstack-tox-docs/dce0a02/html/ : SUCCESS in 5m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/651500/1/check/openstack-tox-lower-constraints/1dc5a80/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a3bf48c3d5edc59615c9325da03bafba3ce0e3ba","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-04-10 16:48:42.000000000","message":"Patch Set 1:\n\nFWIW I was also looking to do this during deployment via Ic694dbcf454df21de1ca0a40e1459644a3aee17e","accounts_in_message":[],"_revision_number":1},{"id":"37a79b0b039386f94c1f0b889fddb45ed6bce988","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2019-04-11 15:22:55.000000000","message":"Patch Set 1: Code-Review+1\n\nOne question as I am not familiar with the validation framework. What happens when someone has split off the nova services from the Controller role and we can not run there the requested container?","accounts_in_message":[],"_revision_number":1},{"id":"318c62c69baac9163d8b6567832ab1e622c275c6","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-16 09:54:26.000000000","message":"Patch Set 1:\n\n\u003e FWIW I was also looking to do this during deployment via\n \u003e Ic694dbcf454df21de1ca0a40e1459644a3aee17e\n\nHave a look at the in-flight validations (doc still under review: https://review.opendev.org/653350","accounts_in_message":[],"_revision_number":1},{"id":"a26d6cdb047e8ef2aacc8241b6895700241079ca","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-20 11:44:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1ed6cf113735653b04ab6085c892ea09ccca51b8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-20 12:01:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c09c5a94830a838fd7092eaf2a1781054b0306ca","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-20 12:08:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a5ec2bc8e2848218aa4147c28dd02a46bdc10ba3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 15:11:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/651500/4/check/tripleo-ci-centos-7-containers-multinode/ecec35a/ : SUCCESS in 2h 14m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/651500/4/check/tripleo-ci-centos-7-undercloud-containers/bec9f01/ : SUCCESS in 1h 28m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/00/651500/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3d41f4e/ : SUCCESS in 2h 03m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/00/651500/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0faf8fb/ : SUCCESS in 2h 33m 12s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/00/651500/4/check/tripleo-ci-centos-7-standalone/9677806/ : SUCCESS in 1h 49m 25s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/00/651500/4/check/tripleo-ci-fedora-28-standalone/8752e72/ : SUCCESS in 1h 57m 58s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/00/651500/4/check/tripleo-ci-centos-7-standalone-upgrade/caac760/ : FAILURE in 1h 36m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/00/651500/4/check/tripleo-ci-centos-7-standalone-upgrade-stein/c60d65b/ : SUCCESS in 2h 02m 30s\n- openstack-tox-pep8 http://logs.openstack.org/00/651500/4/check/openstack-tox-pep8/4ce160b/ : SUCCESS in 6m 40s\n- openstack-tox-py27 http://logs.openstack.org/00/651500/4/check/openstack-tox-py27/a86d251/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/00/651500/4/check/openstack-tox-py35/2f17664/ : SUCCESS in 5m 34s\n- openstack-tox-py36 http://logs.openstack.org/00/651500/4/check/openstack-tox-py36/6e64b9d/ : SUCCESS in 5m 24s\n- openstack-tox-docs http://logs.openstack.org/00/651500/4/check/openstack-tox-docs/7d70d0f/html/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/651500/4/check/openstack-tox-lower-constraints/ac647ec/ : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":4},{"id":"c70cd3319ef5d8a8e1b6a41155ca64032b5532e0","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-21 08:42:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c829e8428c8628f043583cc1889d342b49a9fbaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:37:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/651500/5/check/tripleo-ci-centos-7-containers-multinode/21a23a4/ : SUCCESS in 2h 06m 22s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/00/651500/5/check/tripleo-ci-centos-7-undercloud-containers/1b340fc/ : SUCCESS in 1h 29m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/00/651500/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/2d24589/ : SUCCESS in 2h 11m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/00/651500/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/42e9000/ : FAILURE in 1h 57m 16s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/00/651500/5/check/tripleo-ci-centos-7-standalone/a983e45/ : SUCCESS in 1h 48m 10s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/00/651500/5/check/tripleo-ci-fedora-28-standalone/0849f71/ : SUCCESS in 1h 53m 06s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/00/651500/5/check/tripleo-ci-centos-7-standalone-upgrade/3f07bfc/ : SUCCESS in 2h 02m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/00/651500/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/5065d6a/ : SUCCESS in 2h 10m 42s\n- openstack-tox-pep8 http://logs.openstack.org/00/651500/5/check/openstack-tox-pep8/b5e02d7/ : SUCCESS in 5m 38s\n- openstack-tox-py27 http://logs.openstack.org/00/651500/5/check/openstack-tox-py27/f8f36d5/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/00/651500/5/check/openstack-tox-py35/30e9b8d/ : SUCCESS in 5m 59s\n- openstack-tox-py36 http://logs.openstack.org/00/651500/5/check/openstack-tox-py36/8869450/ : SUCCESS in 5m 07s\n- openstack-tox-docs http://logs.openstack.org/00/651500/5/check/openstack-tox-docs/6aa2782/html/ : SUCCESS in 4m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/651500/5/check/openstack-tox-lower-constraints/07ac75d/ : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":5},{"id":"06fe716806a3d42021f9f1047020f76d470f4790","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-05-22 07:29:13.000000000","message":"Patch Set 5:\n\nDo you have anything posted that\u0027s using this role btw?","accounts_in_message":[],"_revision_number":5},{"id":"39ebb6ed923b18b7f9b9f3afb0ad8c508651db78","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 09:53:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"a3131425864d3ea22bf196d5c8e96375918e334c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-12 14:54:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c7996418a56839623560136ce93e280ac484f97e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 14:54:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d6c328f8504507feb4b5b9467e220c115a81f88b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 17:47:35.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/00/651500/5/gate/tripleo-ci-centos-7-containers-multinode/12499aa/ : SUCCESS in 2h 20m 00s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/00/651500/5/gate/tripleo-ci-centos-7-standalone/af66277/ : SUCCESS in 2h 12m 17s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/00/651500/5/gate/tripleo-ci-fedora-28-standalone/04ce543/ : SUCCESS in 1h 52m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/00/651500/5/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/90f1c0f/ : SUCCESS in 2h 20m 13s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/00/651500/5/gate/tripleo-ci-centos-7-standalone-upgrade-stein/033a586/ : SUCCESS in 2h 07m 51s\n- openstack-tox-pep8 http://logs.openstack.org/00/651500/5/gate/openstack-tox-pep8/ecc350d/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/00/651500/5/gate/openstack-tox-py27/59bb831/ : SUCCESS in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/00/651500/5/gate/openstack-tox-py35/30f0f23/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/00/651500/5/gate/openstack-tox-py36/4e46920/ : SUCCESS in 5m 29s\n- openstack-tox-docs http://logs.openstack.org/00/651500/5/gate/openstack-tox-docs/87bd63d/html/ : SUCCESS in 4m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/00/651500/5/gate/openstack-tox-lower-constraints/5ca9723/ : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":5},{"id":"a7b6df5ec777895469ec941a14cebb021fe12c6f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 17:47:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"b549b692218b447c2941a2cc57402b0ec13dabde","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-13 09:29:40.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit d0cece2233aa3f01715cc3f1b7d8b0b123470396","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e25845d9676d96b2e2484e46782cbf4e8a2149d4","revisions":{"b32f2d6d05bfa76d3722b891c1d968faa17b5812":{"kind":"REWORK","_number":1,"created":"2019-04-10 11:16:59.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/651500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/651500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/1"}}},"commit":{"parents":[{"commit":"b4c2270cc6bec2aaa3018e55173017c6428237a5","subject":"Merge \"Add openstack-endpoints role\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b4c2270cc6bec2aaa3018e55173017c6428237a5"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-04-10 11:10:37.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-04-10 11:16:56.000000000","tz":120},"subject":"Add nova-status Ansible role","message":"Add nova-status Ansible role\n\nThis patch adds a new role which performs a pre-upgrade check on the\nnova-api service.\n\nDepends-on: I7541bfd244acaf1d00d18b3eef031dbadfe4dd4a\nChange-Id: I0dae1b4dc3c945c66e03aa7b361b329f9a25cc99\nImplements: blueprint validation-framework\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b32f2d6d05bfa76d3722b891c1d968faa17b5812"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/b32f2d6d05bfa76d3722b891c1d968faa17b5812"}]},"branch":"refs/heads/master"},"762669b2ac53d3215d4cc4d4f6e29d6241d69356":{"kind":"REWORK","_number":2,"created":"2019-05-20 11:44:20.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/651500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/651500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/2"}}},"commit":{"parents":[{"commit":"cec7b84484f3b93d9f8b6848d9233bde741cfcf5","subject":"Merge \"Fix Jinja2\u0027s default filter usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cec7b84484f3b93d9f8b6848d9233bde741cfcf5"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-04-10 11:10:37.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-20 11:43:27.000000000","tz":120},"subject":"Add nova-status Ansible role","message":"Add nova-status Ansible role\n\nThis patch adds a new role which performs a pre-upgrade check on the\nnova-api service.\n\nDepends-on: I7541bfd244acaf1d00d18b3eef031dbadfe4dd4a\nChange-Id: I0dae1b4dc3c945c66e03aa7b361b329f9a25cc99\nImplements: blueprint validation-framework\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/762669b2ac53d3215d4cc4d4f6e29d6241d69356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/762669b2ac53d3215d4cc4d4f6e29d6241d69356"}]},"branch":"refs/heads/master"},"d4303f735dfa79efa12f9860ec4bf39d5cd0b85c":{"kind":"REWORK","_number":3,"created":"2019-05-20 12:01:15.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/651500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/651500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/3"}}},"commit":{"parents":[{"commit":"cec7b84484f3b93d9f8b6848d9233bde741cfcf5","subject":"Merge \"Fix Jinja2\u0027s default filter usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cec7b84484f3b93d9f8b6848d9233bde741cfcf5"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-04-10 11:10:37.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-20 12:00:59.000000000","tz":120},"subject":"Add nova-status Ansible role","message":"Add nova-status Ansible role\n\nThis patch adds a new role which performs a pre-upgrade check on the\nnova-api service.\n\nDepends-on: I7541bfd244acaf1d00d18b3eef031dbadfe4dd4a\nChange-Id: I0dae1b4dc3c945c66e03aa7b361b329f9a25cc99\nImplements: blueprint validation-framework\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d4303f735dfa79efa12f9860ec4bf39d5cd0b85c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d4303f735dfa79efa12f9860ec4bf39d5cd0b85c"}]},"branch":"refs/heads/master"},"a077bce6c85d94b9d9bfae1047debc0b827f5c9d":{"kind":"REWORK","_number":4,"created":"2019-05-20 12:08:42.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/651500/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/651500/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/4"}}},"commit":{"parents":[{"commit":"cec7b84484f3b93d9f8b6848d9233bde741cfcf5","subject":"Merge \"Fix Jinja2\u0027s default filter usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cec7b84484f3b93d9f8b6848d9233bde741cfcf5"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-04-10 11:10:37.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-20 12:08:32.000000000","tz":120},"subject":"Add nova-status Ansible role","message":"Add nova-status Ansible role\n\nThis patch adds a new role which performs a pre-upgrade check on the\nnova-api service.\n\nDepends-on: I7541bfd244acaf1d00d18b3eef031dbadfe4dd4a\nChange-Id: I0dae1b4dc3c945c66e03aa7b361b329f9a25cc99\nImplements: blueprint validation-framework\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a077bce6c85d94b9d9bfae1047debc0b827f5c9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a077bce6c85d94b9d9bfae1047debc0b827f5c9d"}]},"branch":"refs/heads/master"},"e25845d9676d96b2e2484e46782cbf4e8a2149d4":{"kind":"REWORK","_number":5,"created":"2019-05-21 08:42:58.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/00/651500/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/00/651500/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/00/651500/5"}}},"commit":{"parents":[{"commit":"cec7b84484f3b93d9f8b6848d9233bde741cfcf5","subject":"Merge \"Fix Jinja2\u0027s default filter usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cec7b84484f3b93d9f8b6848d9233bde741cfcf5"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-04-10 11:10:37.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-21 08:42:44.000000000","tz":120},"subject":"Add nova-status Ansible role","message":"Add nova-status Ansible role\n\nThis patch adds a new role which performs a pre-upgrade check on the\nnova-api service.\n\nChange-Id: I0dae1b4dc3c945c66e03aa7b361b329f9a25cc99\nImplements: blueprint validation-framework\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e25845d9676d96b2e2484e46782cbf4e8a2149d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e25845d9676d96b2e2484e46782cbf4e8a2149d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
