)]}'
{"id":"openstack%2Ftripleo-validations~656784","triplet_id":"openstack%2Ftripleo-validations~master~Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf","project":"openstack/tripleo-validations","branch":"master","topic":"validate/containers","hashtags":[],"change_id":"Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf","subject":"New validation: ensure services status","status":"MERGED","created":"2019-05-02 14:41:56.000000000","updated":"2019-08-01 11:01:17.000000000","submitted":"2019-07-29 21:09:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656784-1564434578752-bfa72627","meta_rev_id":"27d74b260f5a06b9c2a2b9000d8632e71654cb95","_number":656784,"virtual_id_number":656784,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":2,"date":"2019-07-29 21:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":2,"date":"2019-07-29 18:05:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":2,"date":"2019-07-29 16:58:38.000000000","_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":2,"date":"2019-07-26 18:23:58.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-07-29 18:05:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-05-13 09:35:24.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"date":"2019-05-13 09:35:27.000000000","_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-13 09:35:38.000000000","_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 11:59:01.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-05-10 13:16:43.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-05-13 09:35:24.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2019-05-13 09:35:27.000000000","updated_by":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"reviewer":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"state":"REVIEWER"},{"updated":"2019-05-13 09:35:38.000000000","updated_by":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"reviewer":{"_account_id":11166,"name":"Lukas Bezdicka","email":"social@v3.sk","username":"xbezdick"},"state":"REVIEWER"},{"updated":"2019-07-26 18:23:58.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-07-29 16:58:38.000000000","updated_by":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"reviewer":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"state":"REVIEWER"},{"updated":"2019-07-29 18:05:59.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-29 21:09:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"199d9d3bea18e81c920730b5ef7dafcf2c87a4fd","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-02 14:41:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94575e466314569eaf598b57af3b7a0c15537bcc","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-02 15:01:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d17a5a75a6c3957c5db30cec6d51ff9a4a3dce58","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-02 15:12:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c8178c74eef532cb4dfce11f0585e08ac6bfa46f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-02 15:13:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e2c877d49efddbc97be649b0c471bd44a17330ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 18:28:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/656784/2/check/tripleo-ci-centos-7-containers-multinode/6fb6334/ : SUCCESS in 2h 07m 10s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/656784/2/check/tripleo-ci-centos-7-undercloud-containers/9eb6f68/ : SUCCESS in 1h 31m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/656784/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/466300a/ : SUCCESS in 2h 37m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/656784/2/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e44f4d6/ : SUCCESS in 3h 06m 12s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/656784/2/check/tripleo-ci-centos-7-standalone/b9fab07/ : SUCCESS in 2h 13m 13s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/656784/2/check/tripleo-ci-fedora-28-standalone/6bb6f39/ : SUCCESS in 1h 51m 15s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/656784/2/check/tripleo-ci-centos-7-standalone-upgrade/66e0513/ : FAILURE in 1h 39m 42s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/84/656784/2/check/openstack-tox-pep8/a1c3f20/ : SUCCESS in 4m 57s\n- openstack-tox-py27 http://logs.openstack.org/84/656784/2/check/openstack-tox-py27/cdf67f1/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/84/656784/2/check/openstack-tox-py35/550684a/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/84/656784/2/check/openstack-tox-py36/319f55f/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/84/656784/2/check/openstack-tox-docs/38722de/html/ : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/656784/2/check/openstack-tox-lower-constraints/967c4e3/ : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e97503333f7820dc0ae9fbb0a544ac9602c1f00a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-10 12:53:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dde4a66687e1a04de59283dc96c91b6433c540df","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-10 12:55:11.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"8153bc03fecb778ccd1e10e81da1cd777c3f623b","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-10 13:16:43.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"2e84ae2284327f58d30dcb1ba059ad5789b01230","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-13 07:23:10.000000000","message":"Patch Set 2: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac1d8fee07847363d6ed0cda55c7defa0079a827","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-13 07:56:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"efe834971c4ce179a2e9a758c0a648931bc4319c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-13 09:34:35.000000000","message":"Patch Set 2:\n\nCédric Jeanneret wrote:\n\n\u003e (2 comments)\nI am just a little worried about that validations in, I hope ;-), a CI-mode soon. I\u0027ve tested it\non RHOS-15/RHEL8 and 5 or 6 tripleo.*healthcheck.service were failing and a NetworkManager (wait-online one)\nservice too.\n\nMy point is, if that validation fails often, we could potentially have issue with the CI itself blocking patches\ndue to some unrelated or minor failed services. So we do need to armored that one, don\u0027t we?\n\nWe maybe need to get some feedbacks from the Upgrades guys on this one?","accounts_in_message":[],"_revision_number":2},{"id":"ca8b091a13afe67154e7aab0673aad7208eafa87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-13 10:29:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/656784/3/check/tripleo-ci-centos-7-containers-multinode/f3d747b/ : SUCCESS in 1h 53m 43s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/84/656784/3/check/tripleo-ci-centos-7-undercloud-containers/00d8dc3/ : SUCCESS in 1h 25m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/656784/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f7918c6/ : SUCCESS in 2h 15m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/84/656784/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f50b6af/ : SUCCESS in 2h 27m 22s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/656784/3/check/tripleo-ci-centos-7-standalone/d60d114/ : SUCCESS in 1h 34m 50s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/84/656784/3/check/tripleo-ci-fedora-28-standalone/1f9a419/ : SUCCESS in 1h 52m 19s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/84/656784/3/check/tripleo-ci-centos-7-standalone-upgrade/257e787/ : FAILURE in 1h 38m 20s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/84/656784/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/f89c3ce/ : SUCCESS in 2h 03m 56s\n- openstack-tox-pep8 http://logs.openstack.org/84/656784/3/check/openstack-tox-pep8/e1f0bb0/ : SUCCESS in 6m 35s\n- openstack-tox-py27 http://logs.openstack.org/84/656784/3/check/openstack-tox-py27/e8b626a/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/84/656784/3/check/openstack-tox-py35/1f0af97/ : SUCCESS in 5m 49s\n- openstack-tox-py36 http://logs.openstack.org/84/656784/3/check/openstack-tox-py36/486e4bc/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/84/656784/3/check/openstack-tox-docs/6f5c019/html/ : SUCCESS in 7m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/656784/3/check/openstack-tox-lower-constraints/f5e634d/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"811b6f12ebd68b82382c9478b85cbcc38058baa8","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-07-26 18:23:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"87d21b455c342e69827db38c43eb9e239fa84551","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-07-29 16:58:38.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nGreat validation Cédric! If we plan to backport this validation to rocky or queens we need to adapt it to check the docker container healthchecks for the containerized services, as pre-stein we didn\u0027t have systemd services mapped to containers.","accounts_in_message":[],"_revision_number":3},{"id":"c6ec2fe39cf5d5730a73d60b53bd49410869631d","author":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"date":"2019-07-29 17:47:35.000000000","message":"Patch Set 3:\n\nrecheck - as results are pretty old already","accounts_in_message":[],"_revision_number":3},{"id":"c0b376b68b685f64a4c63c3c1285604ca44576eb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-29 18:05:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"29b8ad50c935a13b4709223e6b30317c1387a1cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 18:06:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28915ec175eb8d2d0ebc3d87b0dace4a27bded28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:09:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/84/656784/3/gate/tripleo-ci-centos-7-containers-multinode/7c46a0c/ : SUCCESS in 2h 46m 13s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/84/656784/3/gate/tripleo-ci-centos-7-standalone/7a43d6a/ : SUCCESS in 1h 57m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/84/656784/3/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/892c719/ : SUCCESS in 2h 22m 00s\n- openstack-tox-pep8 http://logs.openstack.org/84/656784/3/gate/openstack-tox-pep8/38cbb98/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/84/656784/3/gate/openstack-tox-py27/4bbcd65/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/84/656784/3/gate/openstack-tox-py36/e3f8d4b/ : SUCCESS in 4m 04s\n- openstack-tox-py37 http://logs.openstack.org/84/656784/3/gate/openstack-tox-py37/f8515bf/ : SUCCESS in 4m 51s\n- openstack-tox-molecule http://logs.openstack.org/84/656784/3/gate/openstack-tox-molecule/f297d1c/tox/reports.html : SUCCESS in 26m 11s\n- openstack-tox-docs http://logs.openstack.org/84/656784/3/gate/openstack-tox-docs/500c400/html/ : SUCCESS in 5m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/84/656784/3/gate/openstack-tox-lower-constraints/21939d2/ : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"34408983a945720856594eb97bc1c9013be6cd66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 21:09:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"02ee48db32eb36e2b04f7290db2ac79016983e8d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-08-01 11:01:17.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 87c293917e081f106b202eeb986a0316139752e0","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1265e2f6675b95bf7487fe041066d8de3164e39e","revisions":{"96216ec19fceab6985a2958fb39abf29597b37c8":{"kind":"REWORK","_number":1,"created":"2019-05-02 14:41:56.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/84/656784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/84/656784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/1"}}},"commit":{"parents":[{"commit":"2be2f9dc5bd75b9141bfecea835c0f376194d42c","subject":"Merge \"Retire the undercloud-selinux-mode validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2be2f9dc5bd75b9141bfecea835c0f376194d42c"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-01 19:19:28.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-02 14:41:42.000000000","tz":-360},"subject":"New validation: ensure services status","message":"New validation: ensure services status\n\nThis new validation ensures we don\u0027t have any failed systemd unit.\n\nFailed units might lead to unexpected errors when we want to update\nor upgrade the system, so we have to ensure we don\u0027t have anything\nin a failed state.\n\nSince this check is pretty generic and useful, it can also be called\nas a post-deploy, post-upgrade task. This ensures we have a corret\nstate AFTER such actions as well.\n\nChange-Id: Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/96216ec19fceab6985a2958fb39abf29597b37c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/96216ec19fceab6985a2958fb39abf29597b37c8"}]},"branch":"refs/heads/master"},"3b4afa60d7b1253d24905656ff377336f978d9b7":{"kind":"REWORK","_number":2,"created":"2019-05-02 15:13:24.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/84/656784/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/84/656784/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/2"}}},"commit":{"parents":[{"commit":"2be2f9dc5bd75b9141bfecea835c0f376194d42c","subject":"Merge \"Retire the undercloud-selinux-mode validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2be2f9dc5bd75b9141bfecea835c0f376194d42c"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-01 19:19:28.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-02 15:12:33.000000000","tz":-360},"subject":"New validation: ensure services status","message":"New validation: ensure services status\n\nThis new validation ensures we don\u0027t have any failed systemd unit.\n\nFailed units might lead to unexpected errors when we want to update\nor upgrade the system, so we have to ensure we don\u0027t have anything\nin a failed state.\n\nSince this check is pretty generic and useful, it can also be called\nas a post-deploy, post-upgrade task. This ensures we have a corret\nstate AFTER such actions as well.\n\nChange-Id: Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3b4afa60d7b1253d24905656ff377336f978d9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3b4afa60d7b1253d24905656ff377336f978d9b7"}]},"branch":"refs/heads/master"},"1265e2f6675b95bf7487fe041066d8de3164e39e":{"kind":"REWORK","_number":3,"created":"2019-05-13 07:56:47.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/84/656784/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/84/656784/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/84/656784/3"}}},"commit":{"parents":[{"commit":"2be2f9dc5bd75b9141bfecea835c0f376194d42c","subject":"Merge \"Retire the undercloud-selinux-mode validation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2be2f9dc5bd75b9141bfecea835c0f376194d42c"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-01 19:19:28.000000000","tz":120},"committer":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2019-05-13 07:38:46.000000000","tz":120},"subject":"New validation: ensure services status","message":"New validation: ensure services status\n\nThis new validation ensures we don\u0027t have any failed systemd unit.\n\nFailed units might lead to unexpected errors when we want to update\nor upgrade the system, so we have to ensure we don\u0027t have anything\nin a failed state.\n\nSince this check is pretty generic and useful, it can also be called\nas a post-deploy, post-upgrade task. This ensures we have a corret\nstate AFTER such actions as well.\n\nChange-Id: Id5bb2cce6b4ba41b6c1d860303d9e655ef23f1cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1265e2f6675b95bf7487fe041066d8de3164e39e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1265e2f6675b95bf7487fe041066d8de3164e39e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
