)]}'
{"id":"openstack%2Ftripleo-validations~657173","triplet_id":"openstack%2Ftripleo-validations~master~I9a2a60e645443d4c052321c8edc1ec46943b39e9","project":"openstack/tripleo-validations","branch":"master","topic":"bp/inflight-validations","hashtags":[],"change_id":"I9a2a60e645443d4c052321c8edc1ec46943b39e9","subject":"Update ceph-health for containers and failure options","status":"MERGED","created":"2019-05-04 23:36:45.000000000","updated":"2019-06-12 12:51:55.000000000","submitted":"2019-06-12 12:48:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"657173-1560343722783-216c3fa0","meta_rev_id":"b95f49fcecc62b07fd7193ffc631e69abc300589","_number":657173,"virtual_id_number":657173,"owner":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"date":"2019-06-11 13:59:31.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":2,"date":"2019-06-12 12:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":2,"date":"2019-06-07 13:21:25.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-07 14:10:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 14:03:43.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":1,"date":"2019-06-10 16:36:27.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-04 23:45:51.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 13:21:25.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-06-07 14:03:43.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-06-07 14:10:01.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-06-11 13:59:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-12 12:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"424bf21d110a14c7c17863b142d8155bc3ee37fb","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-04 23:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a6f3809dc3678bb83a32fd6e6e9af7dae417b9d","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-04 23:39:10.000000000","message":"Patch Set 1: Workflow-1\n\nCedric,\nAny early feedback on this WIP?","accounts_in_message":[],"_revision_number":1},{"id":"825b53980afe401ce89d94a887be1cd32eefd1ad","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-04 23:45:51.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"11ea7026b4e849db84d84b8f91a93c45a358e0b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 01:58:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/1/check/tripleo-ci-centos-7-containers-multinode/a16eaf8/ : SUCCESS in 2h 03m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/657173/1/check/tripleo-ci-centos-7-undercloud-containers/410679a/ : SUCCESS in 1h 32m 18s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/989419d/ : SUCCESS in 2h 10m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/657173/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/362b435/ : SUCCESS in 2h 17m 20s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/1/check/tripleo-ci-centos-7-standalone/b1a28e0/ : SUCCESS in 1h 47m 03s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/1/check/tripleo-ci-fedora-28-standalone/945723f/ : SUCCESS in 1h 51m 11s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/657173/1/check/tripleo-ci-centos-7-standalone-upgrade/f85dcce/ : FAILURE in 1h 33m 15s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/1/check/openstack-tox-pep8/d1ce77a/ : FAILURE in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/1/check/openstack-tox-py27/c8a1bc5/ : SUCCESS in 7m 29s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/1/check/openstack-tox-py35/8dd483e/ : SUCCESS in 5m 41s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/1/check/openstack-tox-py36/d5437b2/ : SUCCESS in 7m 19s\n- openstack-tox-docs http://logs.openstack.org/73/657173/1/check/openstack-tox-docs/1c10349/html/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/1/check/openstack-tox-lower-constraints/7c08ae8/ : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0e9d6c6d9ade7c03bc42c91fc5a7571cc76463a0","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-05 13:17:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7caddc55d836d275b1c0d2ff2f0eb07d1c069e10","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-05 18:11:28.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7c125355fae0878527b846df8459d78b51d2e368","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-05 18:12:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d6c34bc744979f3181210b368e1ecb379e9f5e3a","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-05 18:15:52.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7769eaea1eae427bf677c3f6bd485fed3330b0c6","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-05 18:18:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dde039c3c9e5f6fb1f5499e30af4e4bbda829aef","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-05 18:19:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"98d659e8e3edf2d0d0b58b6b39695c98df82d1ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-05 20:49:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/4/check/tripleo-ci-centos-7-containers-multinode/49ff03f/ : SUCCESS in 1h 52m 51s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/657173/4/check/tripleo-ci-centos-7-undercloud-containers/cb6d76d/ : SUCCESS in 1h 26m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/e5ae37a/ : SUCCESS in 1h 56m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/657173/4/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4659394/ : SUCCESS in 2h 28m 18s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/4/check/tripleo-ci-centos-7-standalone/a65b2b1/ : SUCCESS in 1h 51m 31s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/4/check/tripleo-ci-fedora-28-standalone/c9723a7/ : SUCCESS in 1h 58m 08s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/657173/4/check/tripleo-ci-centos-7-standalone-upgrade/0c4e857/ : FAILURE in 1h 27m 02s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/4/check/openstack-tox-pep8/e210427/ : SUCCESS in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/4/check/openstack-tox-py27/23df634/ : SUCCESS in 4m 47s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/4/check/openstack-tox-py35/0d82612/ : SUCCESS in 6m 13s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/4/check/openstack-tox-py36/9ca7f88/ : SUCCESS in 4m 32s\n- openstack-tox-docs http://logs.openstack.org/73/657173/4/check/openstack-tox-docs/51d3f67/html/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/4/check/openstack-tox-lower-constraints/f12df8e/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":4},{"id":"02d96e8dc2a158291ff06196b16baa37a19fe6e0","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-08 12:08:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4e3d4fcb362d6b5ffb0e98a7645e7cc4bba3020a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-08 12:31:19.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nThis validation should be run against the overcloud controller(s). The inventory provides us the container_cli var used for the overcloud, not for the undercloud which is provided by the undercloud.conf file on the undercloud.\n\nThat\u0027s why it\u0027s failing. The include_roles should be coming with a host definition.","accounts_in_message":[],"_revision_number":4},{"id":"e4b5dadb7f4bb389df5c3820fb920732aba77da6","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-14 21:14:56.000000000","message":"Patch Set 4:\n\n(1 comment)\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e This validation should be run against the overcloud controller(s).\n\nTechnically, it should really be run against wherever the Mon service is running, but yes I agree it shouldn\u0027t be run on the undercloud.\n\nWhen you said earlier \"The ceph-health playbook is already targeting the Controllers group. So Ansible will execute those tasks on the Overcloud Controllers.\" I think I misunderstood.\n\n \u003e The inventory provides us the container_cli var used for the\n \u003e overcloud, not for the undercloud which is provided by the\n \u003e undercloud.conf file on the undercloud.\n \u003e \n \u003e That\u0027s why it\u0027s failing. The include_roles should be coming with a\n \u003e host definition.\n\nOK, I\u0027ll update the patch that depends-on this with a when or a delegate_to. AFAICT hosts entry would be for a play, but I want to use it with just a task and not introduce a whole play if I can help it. Also, I just need to run it on one host.","accounts_in_message":[],"_revision_number":4},{"id":"f5a4789164b832d79ff92296fe14f57ea768f47c","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-20 17:02:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"696e22fc498709a9c8223a25bd3861e05e2d8451","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-03 20:51:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"61b6e7beeff3797b949f0059411e627474e38752","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-03 20:55:37.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ccbec755a65e9ebbacacc578f9f424dc1c5c0f05","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-03 20:56:12.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"48e687f0d171a6743a574c86c550679f6c878246","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 01:11:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/7/check/tripleo-ci-centos-7-containers-multinode/f441f03/ : SUCCESS in 1h 57m 38s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/657173/7/check/tripleo-ci-centos-7-undercloud-containers/dbf1c87/ : SUCCESS in 1h 38m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d2f5d19/ : SUCCESS in 2h 15m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/657173/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8aaa0de/ : SUCCESS in 2h 27m 04s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/7/check/tripleo-ci-centos-7-standalone/172196d/ : SUCCESS in 1h 46m 37s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/7/check/tripleo-ci-fedora-28-standalone/4846366/ : SUCCESS in 1h 46m 10s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/657173/7/check/tripleo-ci-centos-7-standalone-upgrade/9e2dcc5/ : SUCCESS in 2h 05m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/657173/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/b911ac1/ : SUCCESS in 2h 08m 41s\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/7/check/openstack-tox-pep8/119c7ca/ : FAILURE in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/7/check/openstack-tox-py27/df83fe9/ : SUCCESS in 5m 28s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/7/check/openstack-tox-py35/8e17afd/ : SUCCESS in 5m 16s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/7/check/openstack-tox-py36/b7cc271/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/73/657173/7/check/openstack-tox-docs/0094fea/html/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/7/check/openstack-tox-lower-constraints/d475534/ : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":7},{"id":"7b15ad3446a741e88f7586ec8ac6df1b4bb70b2b","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-04 01:43:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7e17f17a2868abe80d483d24445a480deaa50b43","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-04 01:45:53.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"26b756955f13c2eed4a13cccadd1be3cd22ea44d","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-04 03:45:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c6b3b6ade2cff0930f14a8bf912c1d85b70f0dfa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 06:36:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/9/check/tripleo-ci-centos-7-containers-multinode/992c681/ : SUCCESS in 2h 00m 59s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/657173/9/check/tripleo-ci-centos-7-undercloud-containers/9b1f01f/ : SUCCESS in 1h 32m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b58da92/ : SUCCESS in 2h 18m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/657173/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/8d33a09/ : SUCCESS in 2h 24m 48s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/9/check/tripleo-ci-centos-7-standalone/a90277d/ : SUCCESS in 1h 40m 50s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/9/check/tripleo-ci-fedora-28-standalone/5484ac4/ : FAILURE in 1h 47m 58s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/657173/9/check/tripleo-ci-centos-7-standalone-upgrade/47b2a9b/ : FAILURE in 1h 27m 50s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/657173/9/check/tripleo-ci-centos-7-standalone-upgrade-stein/a3dd039/ : SUCCESS in 1h 53m 31s\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/9/check/openstack-tox-pep8/2189c75/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/9/check/openstack-tox-py27/09b8808/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/9/check/openstack-tox-py35/2b59643/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/9/check/openstack-tox-py36/958c17d/ : SUCCESS in 4m 58s\n- openstack-tox-docs http://logs.openstack.org/73/657173/9/check/openstack-tox-docs/f922323/html/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/9/check/openstack-tox-lower-constraints/54de508/ : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":9},{"id":"5aff4be8312c3528fa8ada668df2fd0fae084ea5","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-04 13:01:38.000000000","message":"Patch Set 9:\n\nf28 failure was environmental\n\nhttp://logs.openstack.org/73/657173/9/check/tripleo-ci-fedora-28-standalone/5484ac4/job-output.txt.gz#_2019-06-04_03_56_55_342961","accounts_in_message":[],"_revision_number":9},{"id":"cf723164c5c3f1f0a9a099b09ca990a1563f7816","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-05 14:10:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"163f9e717b79c0f4d7c387afa7eb14c5be83caa2","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-05 14:17:24.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7420cab52b51f84961f1852f1d86b95f7b499fe8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 20:26:07.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/10/check/tripleo-ci-centos-7-containers-multinode/bcf45ff/ : SUCCESS in 2h 01m 41s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/657173/10/check/tripleo-ci-centos-7-undercloud-containers/6172b32/ : SUCCESS in 1h 29m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c8c6997/ : SUCCESS in 2h 12m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/657173/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/455dd24/ : SUCCESS in 2h 21m 57s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/10/check/tripleo-ci-centos-7-standalone/77786df/ : SUCCESS in 1h 53m 12s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/10/check/tripleo-ci-fedora-28-standalone/76b8d35/ : SUCCESS in 1h 52m 33s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/657173/10/check/tripleo-ci-centos-7-standalone-upgrade/65bc35d/ : SUCCESS in 2h 08m 06s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/657173/10/check/tripleo-ci-centos-7-standalone-upgrade-stein/e2010a9/ : SUCCESS in 2h 08m 15s\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/10/check/openstack-tox-pep8/2d63158/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/10/check/openstack-tox-py27/83a9f9f/ : SUCCESS in 4m 56s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/10/check/openstack-tox-py35/a3c4e53/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/10/check/openstack-tox-py36/ab149c7/ : SUCCESS in 4m 44s\n- openstack-tox-docs http://logs.openstack.org/73/657173/10/check/openstack-tox-docs/4ddc160/html/ : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/10/check/openstack-tox-lower-constraints/013c42c/ : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":10},{"id":"01af1bceb53d6f1389a586ed5de1887d95d1299c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 10:50:45.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b8d96f353a41a7dccff65dc2cc28bd033a6fe7dc","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 10:56:19.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"894a14b768992880ec63d3cd7b928628fabc7da9","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-07 13:12:06.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"fc760342b13a3cd86282279ce134c7a453faa4e3","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-07 13:12:23.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"c34d4fcb63cf1b59d8b18d59d2e404ce277399e8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 13:21:25.000000000","message":"Patch Set 11: Code-Review+2\n\nThanks John.","accounts_in_message":[],"_revision_number":11},{"id":"341aa127501993ad62471428f8e7262f8f334204","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-06-07 14:03:43.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"e458a3363ece9bdc03763168a6659673d2c93f39","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:05:28.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"23fae06d5dee7ed8dea9342f5f851f41c11d14cd","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-07 14:06:36.000000000","message":"Removed Workflow+1 by Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":11},{"id":"f841edd9297db9b1677b75e6e1c5eafd592114fd","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-07 14:07:22.000000000","message":"Patch Set 11:\n\nI want to see this green in CI and then I\u0027ll +w it. Thanks for the reviews","accounts_in_message":[],"_revision_number":11},{"id":"91a091e05846aea1bc7245231746005cf8fb51a4","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-06-07 14:10:01.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"41d9850e4fb07a8ae3a86a52aebcdb9e92168526","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:22:24.000000000","message":"Patch Set 11:\n\nJohn Fulton wrote:\n\n\u003e I want to see this green in CI and then I\u0027ll +w it. Thanks for the\n\u003e reviews\n\nFair enough, but FYI, unfortunately the CI doesn\u0027t test the validations yet!\nWe are working on it though.","accounts_in_message":[],"_revision_number":11},{"id":"39d65c07e34dc48a62d2d48f8e3f5a282b49ecec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 02:34:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-containers-multinode/78bfadc/ : SUCCESS in 2h 03m 36s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-undercloud-containers/136a14c/ : SUCCESS in 1h 33m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/25392cb/ : SUCCESS in 2h 13m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/516323f/ : SUCCESS in 2h 26m 03s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-standalone/fa7df2a/ : SUCCESS in 1h 51m 53s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/11/check/tripleo-ci-fedora-28-standalone/bf1235b/ : SUCCESS in 1h 46m 03s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-standalone-upgrade/daf8716/ : SUCCESS in 2h 14m 07s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-standalone-upgrade-stein/3bc701b/ : SUCCESS in 2h 02m 31s\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/11/check/openstack-tox-pep8/3644fce/ : SUCCESS in 7m 46s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/11/check/openstack-tox-py27/3bb984b/ : SUCCESS in 6m 01s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/11/check/openstack-tox-py35/b6a2373/ : SUCCESS in 5m 39s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/11/check/openstack-tox-py36/6235fcb/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/73/657173/11/check/openstack-tox-docs/f461ad3/html/ : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/11/check/openstack-tox-lower-constraints/16e1eed/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":11},{"id":"8ade18cad7c501fe5ac18823c9fe5516265df909","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-06-10 16:36:27.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"ff2f23c61d95d5e714610342829ad5b87d1f450b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 16:36:47.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"ba20177ba04ac0477549a2664a13848e62e2af3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 05:34:55.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-containers-multinode/b35771b/ : SUCCESS in 2h 08m 56s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-standalone/fd1a8f4/ : SUCCESS in 2h 03m 56s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/11/gate/tripleo-ci-fedora-28-standalone/05b16e5/ : SUCCESS in 1h 57m 37s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/59fbd51/ : FAILURE in 18m 50s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-standalone-upgrade-stein/a2b8aee/ : FAILURE in 16m 13s\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/11/gate/openstack-tox-pep8/76ab22a/ : SUCCESS in 6m 18s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/11/gate/openstack-tox-py27/e90cbe6/ : SUCCESS in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/11/gate/openstack-tox-py35/41cc903/ : SUCCESS in 5m 03s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/11/gate/openstack-tox-py36/1a6743d/ : SUCCESS in 5m 34s\n- openstack-tox-docs http://logs.openstack.org/73/657173/11/gate/openstack-tox-docs/d09dd03/html/ : SUCCESS in 6m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/11/gate/openstack-tox-lower-constraints/089c390/ : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":11},{"id":"e1534cc47397d3389ff52e498e9fca7648784021","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-11 13:59:31.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"bf70280417138d0e79723e603adbdd454c3a7f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 01:28:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-containers-multinode/9ae3f1d/ : SUCCESS in 2h 17m 24s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-undercloud-containers/2e39194/ : SUCCESS in 1h 35m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/4bb00d1/ : SUCCESS in 2h 28m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5b1ccb3/ : SUCCESS in 2h 30m 19s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-standalone/1e02e83/ : SUCCESS in 1h 56m 46s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/11/check/tripleo-ci-fedora-28-standalone/a6ceff2/ : SUCCESS in 1h 37m 17s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-standalone-upgrade/d26a6f3/ : FAILURE in 1h 36m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/657173/11/check/tripleo-ci-centos-7-standalone-upgrade-stein/9bf83f7/ : SUCCESS in 2h 15m 01s\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/11/check/openstack-tox-pep8/f7a5ec3/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/11/check/openstack-tox-py27/5826902/ : SUCCESS in 5m 00s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/11/check/openstack-tox-py35/2221d0d/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/11/check/openstack-tox-py36/282a07a/ : SUCCESS in 6m 40s\n- openstack-tox-docs http://logs.openstack.org/73/657173/11/check/openstack-tox-docs/d0d13b4/html/ : SUCCESS in 5m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/11/check/openstack-tox-lower-constraints/1bf6776/ : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":11},{"id":"4365cef29ad86c7e86026b12aef4f7e4b87dd0ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 01:29:19.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4b0fadf522b57fb976368d62033fb1c0026821e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:48:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"6b509625e2b1e91cf517b2e42d43db70d70e9ca3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 12:48:42.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-containers-multinode/4d21668/ : SUCCESS in 2h 18m 10s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-standalone/0eb69fb/ : SUCCESS in 2h 19m 07s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/73/657173/11/gate/tripleo-ci-fedora-28-standalone/b005bea/ : SUCCESS in 2h 11m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c739171/ : SUCCESS in 2h 46m 08s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/73/657173/11/gate/tripleo-ci-centos-7-standalone-upgrade-stein/113e788/ : SUCCESS in 1h 48m 09s\n- openstack-tox-pep8 http://logs.openstack.org/73/657173/11/gate/openstack-tox-pep8/613fca1/ : SUCCESS in 5m 59s\n- openstack-tox-py27 http://logs.openstack.org/73/657173/11/gate/openstack-tox-py27/c7bac85/ : SUCCESS in 8m 26s\n- openstack-tox-py35 http://logs.openstack.org/73/657173/11/gate/openstack-tox-py35/69dec2b/ : SUCCESS in 7m 44s\n- openstack-tox-py36 http://logs.openstack.org/73/657173/11/gate/openstack-tox-py36/fe445e7/ : SUCCESS in 6m 21s\n- openstack-tox-docs http://logs.openstack.org/73/657173/11/gate/openstack-tox-docs/7a8fd27/html/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/73/657173/11/gate/openstack-tox-lower-constraints/e2dc414/ : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":11},{"id":"75363791d53657fac1474c27033bbf466af99cfd","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-12 12:51:55.000000000","message":"Patch Set 11: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 853d7316f506435dd4a24a7ff3c2d93a1d9598ff","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"52eeb5b5fc88dd4a99d0dfcfd988552668300e11","revisions":{"d6ab18b6ed30c55f8113fe20238b764a32a0aa01":{"kind":"REWORK","_number":1,"created":"2019-05-04 23:36:45.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/1"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"subject":"WIP/DNM: Update ceph-health to work with containers","message":"WIP/DNM: Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nceph-ansible deployed ceph will not benefit from hiera data so\nremove task to get it from the enabled_services list and instead\nassume a ceph_mon is deployed because the inflight-validations\nwill call it only for the appropriate node. We will also worry\nabout identifying the correct node by using delegate_to\n\nTODO: fix delegate_to so it uses the correct node\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d6ab18b6ed30c55f8113fe20238b764a32a0aa01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/d6ab18b6ed30c55f8113fe20238b764a32a0aa01"}]},"branch":"refs/heads/master"},"88ba8c3da783875ba384e818bbe238e9d08d9562":{"kind":"REWORK","_number":2,"created":"2019-05-05 18:12:35.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/2"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-05 18:11:07.000000000","tz":-360},"subject":"WIP/DNM: Update ceph-health to work with containers","message":"WIP/DNM: Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/88ba8c3da783875ba384e818bbe238e9d08d9562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/88ba8c3da783875ba384e818bbe238e9d08d9562"}]},"branch":"refs/heads/master"},"fa50d430017a83f9b5ed7a39c877f979bfd6a4cd":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-05-05 18:15:52.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/3"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-05 18:14:11.000000000","tz":-360},"subject":"Update ceph-health to work with containers","message":"Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa50d430017a83f9b5ed7a39c877f979bfd6a4cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fa50d430017a83f9b5ed7a39c877f979bfd6a4cd"}]},"branch":"refs/heads/master"},"a5f581f1fc94cde3bb8067a788620090141ee89e":{"kind":"REWORK","_number":4,"created":"2019-05-05 18:18:57.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/4"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-05 18:17:29.000000000","tz":-360},"subject":"Update ceph-health to work with containers","message":"Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a5f581f1fc94cde3bb8067a788620090141ee89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a5f581f1fc94cde3bb8067a788620090141ee89e"}]},"branch":"refs/heads/master"},"ac3c2521119ee1c312c33c09b8097bd1df280eba":{"kind":"REWORK","_number":5,"created":"2019-06-03 20:51:31.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/5"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-03 20:50:24.000000000","tz":0},"subject":"Update ceph-health to work with containers","message":"Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nAdd new option, which defaults to false, to fail if ceph\nhealth is in error.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ac3c2521119ee1c312c33c09b8097bd1df280eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ac3c2521119ee1c312c33c09b8097bd1df280eba"}]},"branch":"refs/heads/master"},"ec0a6b77673876d6a9ca52768c51841acdf7aba7":{"kind":"REWORK","_number":6,"created":"2019-06-03 20:55:37.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/6"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-03 20:55:27.000000000","tz":0},"subject":"Update ceph-health to work with containers","message":"Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nAdd new option, which defaults to false, to fail if ceph\nhealth is in error.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ec0a6b77673876d6a9ca52768c51841acdf7aba7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ec0a6b77673876d6a9ca52768c51841acdf7aba7"}]},"branch":"refs/heads/master"},"3d433c4421a1db7a7092d6f6c9190a1bd7980022":{"kind":"REWORK","_number":7,"created":"2019-06-03 20:56:12.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/7"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-03 20:56:06.000000000","tz":0},"subject":"Update ceph-health to work with containers","message":"Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nAdd new option, which defaults to false, to fail if ceph\nhealth is in error.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3d433c4421a1db7a7092d6f6c9190a1bd7980022"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/3d433c4421a1db7a7092d6f6c9190a1bd7980022"}]},"branch":"refs/heads/master"},"ce6141673ae3fa743ffd4f4d59dc0603a7597b98":{"kind":"REWORK","_number":8,"created":"2019-06-04 01:43:28.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/8"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-04 01:43:16.000000000","tz":-240},"subject":"Update ceph-health to work with containers","message":"Update ceph-health to work with containers\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nAdd new option, which defaults to false, to fail if ceph\nhealth is in error.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ce6141673ae3fa743ffd4f4d59dc0603a7597b98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ce6141673ae3fa743ffd4f4d59dc0603a7597b98"}]},"branch":"refs/heads/master"},"78fb6465a8933c00d1383df5f16c2c4debeaa5ae":{"kind":"REWORK","_number":9,"created":"2019-06-04 03:45:36.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/9"}}},"commit":{"parents":[{"commit":"6d61d8d4ddd7db799fe5e43abe0823bd495236d2","subject":"Merge \"New validation: ensure image-serve service is working as expected\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/6d61d8d4ddd7db799fe5e43abe0823bd495236d2"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-04 03:39:36.000000000","tz":0},"subject":"Update ceph-health for containers and failure options","message":"Update ceph-health for containers and failure options\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nAdd new option, fail_on_ceph_health_err (default: false), to\nfail when ceph health is in error.\n\nAdd new option, osd_percentage_min (default: 0), to fail\nwhen a certain perecentage of OSDs are not in. When\nosd_percentage_min is 0, this check is not performed.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/78fb6465a8933c00d1383df5f16c2c4debeaa5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/78fb6465a8933c00d1383df5f16c2c4debeaa5ae"}]},"branch":"refs/heads/master"},"47bca50da323734cb718bcde7500b8208900afad":{"kind":"REWORK","_number":10,"created":"2019-06-05 14:10:28.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/10"}}},"commit":{"parents":[{"commit":"bf206aa97a5cce5b04de505c294d4d634f05d135","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/bf206aa97a5cce5b04de505c294d4d634f05d135"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-05 14:10:19.000000000","tz":-240},"subject":"Update ceph-health for containers and failure options","message":"Update ceph-health for containers and failure options\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nAdd new option, fail_on_ceph_health_err (default: false), to\nfail when ceph health is in error.\n\nAdd new option, osd_percentage_min (default: 0), to fail\nwhen a certain perecentage of OSDs are not in. When\nosd_percentage_min is 0, this check is not performed.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/47bca50da323734cb718bcde7500b8208900afad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/47bca50da323734cb718bcde7500b8208900afad"}]},"branch":"refs/heads/master"},"52eeb5b5fc88dd4a99d0dfcfd988552668300e11":{"kind":"REWORK","_number":11,"created":"2019-06-07 13:12:06.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/73/657173/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/73/657173/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/73/657173/11"}}},"commit":{"parents":[{"commit":"e819f040e43d29c44f73b0342ada1193c8fcad46","subject":"Merge \"Add run-validations.sh script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e819f040e43d29c44f73b0342ada1193c8fcad46"}]}],"author":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-04 23:35:08.000000000","tz":-360},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-06-07 13:11:58.000000000","tz":-240},"subject":"Update ceph-health for containers and failure options","message":"Update ceph-health for containers and failure options\n\nCall to \u0027ceph health\u0027 should be wrapped in a container exec.\nDisplay warning for any \u0027ceph health\u0027 status that is not OK.\n\nAdd new option, fail_on_ceph_health_err (default: false), to\nfail when ceph health is in error.\n\nAdd new option, osd_percentage_min (default: 0), to fail\nwhen a certain perecentage of OSDs are not in. When\nosd_percentage_min is 0, this check is not performed.\n\nChange-Id: I9a2a60e645443d4c052321c8edc1ec46943b39e9\nImplements: blueprint inflight-validations\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/52eeb5b5fc88dd4a99d0dfcfd988552668300e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/52eeb5b5fc88dd4a99d0dfcfd988552668300e11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
