)]}'
{"id":"openstack%2Ftripleo-validations~659885","triplet_id":"openstack%2Ftripleo-validations~master~I31feafcdfd8467a00aa96bcafa567ca06b1bd86a","project":"openstack/tripleo-validations","branch":"master","topic":"tls-everywhere","hashtags":[],"change_id":"I31feafcdfd8467a00aa96bcafa567ca06b1bd86a","subject":"Add new role to validate tls-everywhere","status":"MERGED","created":"2019-05-17 21:59:34.000000000","updated":"2019-07-23 17:52:09.000000000","submitted":"2019-07-23 17:52:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":40,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659885-1563904329571-31ef7618","meta_rev_id":"d2d6ba8f5e0e8fe3c665cd41396a6615ce072710","_number":659885,"virtual_id_number":659885,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2019-07-23 17:52:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":2,"date":"2019-07-22 13:02:02.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"value":1,"date":"2019-07-23 15:20:08.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-17 22:00:51.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-05-20 07:08:34.000000000","updated_by":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2019-05-20 07:09:25.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2019-06-04 12:56:25.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2019-06-04 13:28:01.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2019-06-07 09:21:52.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2019-07-15 14:12:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-15 14:12:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-15 15:27:52.000000000","updated_by":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"reviewer":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"state":"REVIEWER"},{"updated":"2019-07-16 15:12:29.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2019-07-18 01:56:43.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-07-23 15:20:08.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-07-23 17:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e81f864ba86f576b7142e793e161456f8e443556","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-05-17 21:59:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14bbf7384010a55c1a99f70e5cd74fcfde727da2","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-05-17 22:00:23.000000000","message":"Patch Set 1: Workflow-1\n\nStill WIP pending the questions I had in the comments.","accounts_in_message":[],"_revision_number":1},{"id":"805385214cd6ec2e505a2ab3a98cf592361c1975","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-18 00:44:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/1/check/tripleo-ci-centos-7-containers-multinode/9138172/ : FAILURE in 1h 48m 06s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/1/check/tripleo-ci-centos-7-undercloud-containers/1dbbe8a/ : SUCCESS in 1h 38m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d60a307/ : SUCCESS in 2h 15m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a2da988/ : SUCCESS in 2h 23m 54s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/1/check/tripleo-ci-centos-7-standalone/eade9ea/ : FAILURE in 1h 43m 38s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/1/check/tripleo-ci-fedora-28-standalone/688107d/ : SUCCESS in 1h 43m 02s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/1/check/tripleo-ci-centos-7-standalone-upgrade/643f256/ : SUCCESS in 2h 05m 39s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/f44a21f/ : SUCCESS in 2h 06m 04s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/1/check/openstack-tox-pep8/e03b441/ : FAILURE in 6m 46s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/1/check/openstack-tox-py27/5d4e0ea/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/1/check/openstack-tox-py35/b667ab9/ : SUCCESS in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/1/check/openstack-tox-py36/5a1646f/ : SUCCESS in 6m 39s\n- openstack-tox-docs http://logs.openstack.org/85/659885/1/check/openstack-tox-docs/c2f9168/html/ : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/1/check/openstack-tox-lower-constraints/fe4dbc5/ : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"dab80521f167550161b30984191f1a69b0b666e9","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-20 08:06:44.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nFor sure, this is a serious one! ;-)\n\nThanks for your great work here, Ade \u0026 Juan!\n\nHere is a first batch of comments, others will follow.","accounts_in_message":[],"_revision_number":1},{"id":"c646017103c0cc50f30543384aa3bbb437060b25","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-20 09:02:55.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8000bac322efef31135453c28b32dffc58c5a254","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-05-20 09:13:49.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4394bdeade89c29241e65811931ce44a26f17919","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-05-20 22:10:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"959b9185bd177795f029db7adb1391161dfcfe9c","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-05-20 22:12:20.000000000","message":"Patch Set 2: Workflow-1\n\nThanks for all the comments.  I think I got all of them.\nLet me know if I missed anything, or anything further needs to be changed.\n\nThis was tested on OSP13.  Going to test on OSP 15 tomorrow to make sure it works in a containerized undercloud.","accounts_in_message":[],"_revision_number":2},{"id":"90256497617b91895d3c31cdd7f54d0230132ab3","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-05-20 22:13:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"39d17db3c282a3da097bec561ae1c72225dfb2fe","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-05-20 22:14:43.000000000","message":"Patch Set 3: Workflow-1\n\nThanks for all the comments.  I think I got all of them.\nLet me know if I missed anything, or anything further needs to be changed.\n\nThis was tested on OSP13.  Going to test on OSP 15 tomorrow to make sure it works in a containerized undercloud.","accounts_in_message":[],"_revision_number":3},{"id":"25044d034ff4f2d96b6c482c75cc8e4b343e0c43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 01:19:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/3/check/tripleo-ci-centos-7-containers-multinode/5272c8e/ : SUCCESS in 2h 01m 58s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/3/check/tripleo-ci-centos-7-undercloud-containers/d82a3d1/ : SUCCESS in 1h 43m 44s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/eb72273/ : SUCCESS in 2h 23m 01s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/3/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/7341a1d/ : SUCCESS in 2h 31m 00s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/3/check/tripleo-ci-centos-7-standalone/d39966e/ : SUCCESS in 2h 01m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/3/check/tripleo-ci-fedora-28-standalone/496ef0f/ : SUCCESS in 1h 45m 30s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/3/check/tripleo-ci-centos-7-standalone-upgrade/f942468/ : FAILURE in 1h 57m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/3/check/tripleo-ci-centos-7-standalone-upgrade-stein/c150364/ : SUCCESS in 2h 13m 44s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/3/check/openstack-tox-pep8/1634774/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/3/check/openstack-tox-py27/b8d027e/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/3/check/openstack-tox-py35/482831f/ : SUCCESS in 5m 33s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/3/check/openstack-tox-py36/672b9f6/ : SUCCESS in 5m 19s\n- openstack-tox-docs http://logs.openstack.org/85/659885/3/check/openstack-tox-docs/a661597/html/ : SUCCESS in 5m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/3/check/openstack-tox-lower-constraints/7e7b3eb/ : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":3},{"id":"7919660217712a7fac58fde72b5fb026423494e0","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-05-21 20:47:25.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nWe should probably look closer at the rest of the tasks that made need priv. escalation","accounts_in_message":[],"_revision_number":3},{"id":"dcb46529c0e38ed42ae8ad37898f050010c66d5d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-22 10:21:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8e53b9b028687b3ae614b98974ab64e46f910ed8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-22 10:23:20.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"363f1b162b8072cb497a7c39411ac83a9e95f7bb","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-05-26 16:17:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f381a55b98075ac2dccbbf6a36a622eb7307e9f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-27 11:01:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"29dcc8c32720abb6fe3b6925f791807a952e01a9","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-27 11:15:47.000000000","message":"Patch Set 3:\n\nHarry Rybacki wrote:\n\n\u003e (1 comment)\n\nIn the patchset #4, I\u0027ve added a new custom lib called \u0027reportentry.py\u0027 which include the ability to add \u0027Recommendations\u0027.\n\nIt gives this in conjunction with our callback plugin:\n\n```\n(undercloud) [stack@undercloud ~]$ ./run-validations.sh tls-everywhere-prep\nTask \u0027tls-everywhere : Report Warning DNS setup in undercloud.conf\u0027 succeeded, but had some warnings:\nHost: undercloud\nWarnings:\n* [SKIPPED] \u0027DNS is not set up correctly in undercloud.conf\u0027\n- RECOMMENDATION: Please set the \u0027undercloud_nameservers\u0027 parameter to point to IdM/FreeIPA in undercloud.conf\n- RECOMMENDATION: Please set the super magic variable to get a super magic IdM\n- RECOMMENDATION: Please don\u0027t follow my recommendations here, I am a joke\n\nTask \u0027tls-everywhere : Report Error DNS setup in undercloud.conf\u0027 failed:\nHost: undercloud\nMessage: \n* [ERROR] \u0027DNS is not set up correctly in undercloud.conf\u0027\n- RECOMMENDATION: Please set the \u0027undercloud_nameservers\u0027 parameter to point to IdM/FreeIPA in undercloud.conf\n- RECOMMENDATION: Please set the super magic variable to get a super magic IdM\n- RECOMMENDATION: Please don\u0027t follow my recommendations here, I am a joke\n\n\n+-----------------------------------------------+\n| Failure! The validation failed for all hosts: |\n+-----------------------------------------------+\n* undercloud\n```","accounts_in_message":[],"_revision_number":3},{"id":"53efea0039ade68e7ba5723a5b6bec1041f4fe4a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-27 11:24:47.000000000","message":"Patch Set 4: Workflow-1\n\nSorry for the formatting above!\n\nUnfortunately, I pressed the \u0027save\u0027 button too quickly.\n\nThis custom lib will implement the same reporting like you were and will work \nwith our custom callback.\n\nFeel free to change/improve the logic of this new lib.\n\nNote: to get the same formatting, you will need [1].\n\n[1] - Change-Id: I643552d2fd91b1c986584aa72bcbcb4e9ab7a7bf","accounts_in_message":[],"_revision_number":4},{"id":"421032d6c27d4e6e2f5878547b89e77ea33137ee","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-27 12:12:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f64517d4af4fbb41323d7bf3bc926007d11e586a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-27 12:12:26.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"9c9759ff5699c611281a731e497e64a7154b8902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 14:50:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/5/check/tripleo-ci-centos-7-containers-multinode/8162083/ : SUCCESS in 2h 26m 57s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/5/check/tripleo-ci-centos-7-undercloud-containers/5c3e23f/ : SUCCESS in 1h 39m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/14890af/ : SUCCESS in 2h 14m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/5/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9657281/ : SUCCESS in 2h 26m 36s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/5/check/tripleo-ci-centos-7-standalone/7c09568/ : SUCCESS in 1h 34m 03s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/5/check/tripleo-ci-fedora-28-standalone/e81dde5/ : SUCCESS in 1h 42m 51s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/5/check/tripleo-ci-centos-7-standalone-upgrade/3d909b6/ : SUCCESS in 2h 07m 24s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/5/check/tripleo-ci-centos-7-standalone-upgrade-stein/33a5ebd/ : SUCCESS in 2h 15m 23s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/5/check/openstack-tox-pep8/534a305/ : FAILURE in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/5/check/openstack-tox-py27/717e966/ : SUCCESS in 4m 50s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/5/check/openstack-tox-py35/41d006a/ : SUCCESS in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/5/check/openstack-tox-py36/66bcbf6/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/85/659885/5/check/openstack-tox-docs/f1212f4/html/ : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/5/check/openstack-tox-lower-constraints/3ba96b0/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":5},{"id":"d5ad51f1a822cbb2057d9622acaf240829a2069f","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-05-27 14:59:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cf7dbb9b33f058bc56de8ce1f66c910b10405ac6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 17:58:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-containers-multinode/e1d4db2/ : SUCCESS in 2h 06m 53s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-undercloud-containers/0f057c0/ : SUCCESS in 1h 53m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/22ec28d/ : SUCCESS in 2h 32m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0fe27aa/ : SUCCESS in 2h 10m 55s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-standalone/2bf831c/ : SUCCESS in 1h 53m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/6/check/tripleo-ci-fedora-28-standalone/82d60d6/ : FAILURE in 1h 46m 05s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-standalone-upgrade/76a9941/ : SUCCESS in 2h 24m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/10b40fc/ : SUCCESS in 2h 05m 22s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/6/check/openstack-tox-pep8/2f043bc/ : SUCCESS in 7m 35s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/6/check/openstack-tox-py27/4a19e5c/ : SUCCESS in 6m 54s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/6/check/openstack-tox-py35/4182f95/ : SUCCESS in 8m 11s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/6/check/openstack-tox-py36/c944c05/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/85/659885/6/check/openstack-tox-docs/867c630/html/ : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/6/check/openstack-tox-lower-constraints/a9ecb19/ : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":6},{"id":"e71a85f8612527861f716182448546433abc9509","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-03 14:43:58.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"08875270b88632952701045d09e10693821fa14a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 19:31:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-containers-multinode/8963afe/ : SUCCESS in 2h 10m 00s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-undercloud-containers/92c9f50/ : SUCCESS in 1h 57m 16s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f3b2c63/ : SUCCESS in 2h 25m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/906cfce/ : SUCCESS in 2h 19m 15s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-standalone/386c839/ : SUCCESS in 1h 51m 07s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/6/check/tripleo-ci-fedora-28-standalone/66a370e/ : SUCCESS in 1h 49m 59s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-standalone-upgrade/4cc2bcc/ : SUCCESS in 2h 05m 00s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/6/check/tripleo-ci-centos-7-standalone-upgrade-stein/f710f92/ : SUCCESS in 2h 03m 20s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/6/check/openstack-tox-pep8/2169290/ : SUCCESS in 8m 19s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/6/check/openstack-tox-py27/96ac54c/ : SUCCESS in 4m 48s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/6/check/openstack-tox-py35/6cc75bd/ : SUCCESS in 8m 01s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/6/check/openstack-tox-py36/c885252/ : SUCCESS in 7m 43s\n- openstack-tox-docs http://logs.openstack.org/85/659885/6/check/openstack-tox-docs/074eca4/html/ : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/6/check/openstack-tox-lower-constraints/0234720/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":6},{"id":"e5147918ce04d0be4aaf0cca145e308c4d052dd4","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-04 12:59:27.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"10f8fa1296b14cdd8040fc30518e14f5eeeb1d99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-04 20:55:17.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/7/check/tripleo-ci-centos-7-containers-multinode/e822257/ : SUCCESS in 2h 00m 54s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/7/check/tripleo-ci-centos-7-undercloud-containers/6669b28/ : SUCCESS in 1h 32m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/d90d876/ : SUCCESS in 2h 13m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/7/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/980476d/ : SUCCESS in 2h 32m 53s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/7/check/tripleo-ci-centos-7-standalone/c716987/ : SUCCESS in 1h 48m 57s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/7/check/tripleo-ci-fedora-28-standalone/94ae681/ : FAILURE in 1h 59m 34s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/7/check/tripleo-ci-centos-7-standalone-upgrade/cb687e8/ : SUCCESS in 2h 10m 21s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/7/check/tripleo-ci-centos-7-standalone-upgrade-stein/4efe6d1/ : FAILURE in 2h 28m 15s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/7/check/openstack-tox-pep8/600bcb0/ : SUCCESS in 7m 59s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/7/check/openstack-tox-py27/6b7e191/ : SUCCESS in 7m 38s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/7/check/openstack-tox-py35/8a07ae9/ : SUCCESS in 5m 37s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/7/check/openstack-tox-py36/56a3c92/ : SUCCESS in 6m 02s\n- openstack-tox-docs http://logs.openstack.org/85/659885/7/check/openstack-tox-docs/c5a1425/html/ : SUCCESS in 6m 10s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/7/check/openstack-tox-lower-constraints/7d91bf8/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":7},{"id":"3a4c6a027d7a92b0209303c958f8a0918ae3ae84","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-06-05 09:58:48.000000000","message":"Patch Set 7:\n\nExample output format:\n\nTask \u0027tls-everywhere : Test Enable novajoin setup in undercloud.conf check\u0027 succeeded, but had some warnings:\nHost: undercloud\nWarnings: [SKIPPED] \u0027enable_novajoin is not enabled undercloud.conf\u0027\n - RECOMMENDATION: Please set \u0027enable_novajoin\u0027 to True in undercloud.conf\n\n\n\nTask \u0027tls-everywhere : Enable novajoin setup in undercloud.conf check\u0027 failed:\nHost: undercloud\nMessage: [ERROR] \u0027enable_novajoin is not enabled undercloud.conf\u0027\n - RECOMMENDATION: Please set \u0027enable_novajoin\u0027 to True in undercloud.conf\n\n\nFailure! The validation failed for all hosts:\n* undercloud","accounts_in_message":[],"_revision_number":7},{"id":"5dbb26a1d7a36b64487593617c57b97065a69f8d","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-06-05 10:26:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"145d80ac4f4b41d3f3758a1a6b9f88956b4f50fc","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-05 10:42:40.000000000","message":"Patch Set 8: Code-Review+1\n\nGood port, Greg.","accounts_in_message":[],"_revision_number":8},{"id":"4e9eb2007d7784e564fd3d3e1e14e3c2f13fd647","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-05 13:15:21.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4bb68baa1377f9d922cf672024150e46674f1336","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-05 13:43:30.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3af420aae6bb920adde9de8c478660562b5da284","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 19:02:03.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/9/check/tripleo-ci-centos-7-containers-multinode/124bafb/ : SUCCESS in 1h 54m 49s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/9/check/tripleo-ci-centos-7-undercloud-containers/d5f6619/ : SUCCESS in 1h 27m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/57f679b/ : SUCCESS in 2h 06m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/9/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d47989f/ : SUCCESS in 2h 24m 43s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/9/check/tripleo-ci-centos-7-standalone/71dbd38/ : SUCCESS in 1h 59m 26s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/9/check/tripleo-ci-fedora-28-standalone/44444e6/ : SUCCESS in 1h 44m 17s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/9/check/tripleo-ci-centos-7-standalone-upgrade/df9c743/ : SUCCESS in 2h 00m 58s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/9/check/tripleo-ci-centos-7-standalone-upgrade-stein/6b732e5/ : SUCCESS in 2h 06m 35s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/9/check/openstack-tox-pep8/354a8e5/ : FAILURE in 6m 09s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/9/check/openstack-tox-py27/1b70230/ : SUCCESS in 5m 55s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/9/check/openstack-tox-py35/5383097/ : SUCCESS in 4m 54s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/9/check/openstack-tox-py36/1e5527e/ : SUCCESS in 5m 57s\n- openstack-tox-docs http://logs.openstack.org/85/659885/9/check/openstack-tox-docs/e3874f5/html/ : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/9/check/openstack-tox-lower-constraints/50d742f/ : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":9},{"id":"ea6860c9d5ec3163125d58dad75baebb73b90140","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 05:59:01.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"780af84dcfe5ad0eb2b5ea3c65da5e82d5e8fe89","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 06:34:48.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8d9d1d12b03c2c5a19b073fcabd7c675e656a98b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-06-06 07:26:21.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"00caec305488a3c0d41bc8ed01b9206dca63ee02","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-06-06 07:26:51.000000000","message":"Patch Set 9:\n\nThanks for the feedback Gael!","accounts_in_message":[],"_revision_number":9},{"id":"c5b6570eaafd485747402740af306573c2c297ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 10:08:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/10/check/tripleo-ci-centos-7-containers-multinode/300f301/ : SUCCESS in 2h 06m 28s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/10/check/tripleo-ci-centos-7-undercloud-containers/1a8e90b/ : SUCCESS in 1h 25m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/493289c/ : SUCCESS in 2h 10m 23s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/10/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f14861d/ : SUCCESS in 2h 23m 08s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/10/check/tripleo-ci-centos-7-standalone/f64fab3/ : SUCCESS in 1h 50m 41s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/10/check/tripleo-ci-fedora-28-standalone/aee81ed/ : SUCCESS in 2h 07m 00s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/10/check/tripleo-ci-centos-7-standalone-upgrade/ea4ed73/ : SUCCESS in 2h 03m 26s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/10/check/tripleo-ci-centos-7-standalone-upgrade-stein/eb7f8a3/ : SUCCESS in 2h 07m 02s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/10/check/openstack-tox-pep8/defdccf/ : SUCCESS in 5m 23s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/10/check/openstack-tox-py27/01502cc/ : SUCCESS in 4m 45s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/10/check/openstack-tox-py35/5a1df41/ : SUCCESS in 7m 20s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/10/check/openstack-tox-py36/c4a183f/ : SUCCESS in 4m 25s\n- openstack-tox-docs http://logs.openstack.org/85/659885/10/check/openstack-tox-docs/f381715/html/ : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/10/check/openstack-tox-lower-constraints/1ce2b4b/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":10},{"id":"8df9380ed50561de469d823b50a61cfea17f2d49","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-06 14:45:21.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"09c7c83025d36746c7fd56ef0c7e05ed7626c4e4","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-06 14:47:33.000000000","message":"Patch Set 11: Code-Review-1\n\nPlease disregard some of the inline comments and \"containerized|non-containerized\" bits. They are solely to aid in debugging these tasks.\n\nI\u0027ll be going back and making some of the flows more consistent.\n\n@Ade there is a specific question in pre-deployment.yml for you.","accounts_in_message":[],"_revision_number":11},{"id":"b5d2dd2328d86921e10a882715d5ea4cfacdf72f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 22:17:14.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/11/check/tripleo-ci-centos-7-containers-multinode/7add24b/ : SUCCESS in 1h 55m 42s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/11/check/tripleo-ci-centos-7-undercloud-containers/258992b/ : SUCCESS in 1h 30m 59s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/11562aa/ : SUCCESS in 2h 31m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/11/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e2b085b/ : SUCCESS in 2h 29m 44s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/11/check/tripleo-ci-centos-7-standalone/e1613ed/ : SUCCESS in 1h 55m 18s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/11/check/tripleo-ci-fedora-28-standalone/b158c4e/ : SUCCESS in 2h 05m 54s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/11/check/tripleo-ci-centos-7-standalone-upgrade/a3b8023/ : SUCCESS in 2h 08m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/11/check/tripleo-ci-centos-7-standalone-upgrade-stein/772ba01/ : SUCCESS in 2h 12m 40s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/11/check/openstack-tox-pep8/6b7d4ac/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/11/check/openstack-tox-py27/e4b5053/ : SUCCESS in 6m 11s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/11/check/openstack-tox-py35/d246dd2/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/11/check/openstack-tox-py36/1c6d2a4/ : SUCCESS in 5m 44s\n- openstack-tox-docs http://logs.openstack.org/85/659885/11/check/openstack-tox-docs/c58bd84/html/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/11/check/openstack-tox-lower-constraints/c82c65b/ : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":11},{"id":"96497771aeff54afff99325e2f8145280b408d16","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2019-06-07 09:21:52.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"22b0753ffb724049173f6dc202ab7bab31335fd6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 09:33:39.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"ed58f3c0ae56ada2d3e27ae1b677e0d17fbb565b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:00:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/12/check/tripleo-ci-centos-7-containers-multinode/15b6617/ : SUCCESS in 2h 13m 02s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/12/check/tripleo-ci-centos-7-undercloud-containers/36b0a4f/ : SUCCESS in 1h 38m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c6b88d6/ : SUCCESS in 2h 13m 03s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/12/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c69508b/ : SUCCESS in 2h 19m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/12/check/tripleo-ci-centos-7-standalone/2c3dc8f/ : SUCCESS in 2h 03m 18s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/12/check/tripleo-ci-fedora-28-standalone/c3e1b96/ : SUCCESS in 2h 01m 11s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/12/check/tripleo-ci-centos-7-standalone-upgrade/8b12293/ : SUCCESS in 2h 10m 59s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/12/check/tripleo-ci-centos-7-standalone-upgrade-stein/429d5ac/ : SUCCESS in 2h 06m 41s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/12/check/openstack-tox-pep8/8cda891/ : SUCCESS in 5m 35s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/12/check/openstack-tox-py27/ff62140/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/12/check/openstack-tox-py35/fcc46aa/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/12/check/openstack-tox-py36/d0012fa/ : SUCCESS in 5m 35s\n- openstack-tox-docs http://logs.openstack.org/85/659885/12/check/openstack-tox-docs/2e143f8/html/ : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/12/check/openstack-tox-lower-constraints/03a92ec/ : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":12},{"id":"6d0f9c3f745465e491462dcb85322dabbba69196","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-10 12:03:11.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"98d2750c91626916043002854110479905d92c29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 14:42:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/13/check/tripleo-ci-centos-7-containers-multinode/56ecbe1/ : SUCCESS in 2h 32m 20s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/13/check/tripleo-ci-centos-7-undercloud-containers/5fa0446/ : SUCCESS in 1h 44m 11s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ba268e5/ : SUCCESS in 2h 17m 22s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/13/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d7dc8ca/ : FAILURE in 1h 55m 31s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/13/check/tripleo-ci-centos-7-standalone/e0055e9/ : SUCCESS in 2h 02m 10s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/13/check/tripleo-ci-fedora-28-standalone/0bc968c/ : SUCCESS in 1h 33m 43s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/13/check/tripleo-ci-centos-7-standalone-upgrade/fbc5bf1/ : SUCCESS in 1h 58m 31s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/13/check/tripleo-ci-centos-7-standalone-upgrade-stein/6244897/ : SUCCESS in 2h 08m 42s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/13/check/openstack-tox-pep8/5bcf5ff/ : FAILURE in 8m 49s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/13/check/openstack-tox-py27/267a1fd/ : SUCCESS in 7m 32s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/13/check/openstack-tox-py35/0e73533/ : SUCCESS in 5m 15s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/13/check/openstack-tox-py36/c1fd64b/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/85/659885/13/check/openstack-tox-docs/57c9933/html/ : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/13/check/openstack-tox-lower-constraints/b775346/ : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":13},{"id":"b06db7e9677b2b6db6b1353b7905af401b35cf79","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-11 07:55:12.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7e1fe85e28c988ba84ebb3f884f7fcb347169247","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 13:31:30.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/14/check/tripleo-ci-centos-7-containers-multinode/e5e41ac/ : SUCCESS in 2h 15m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/14/check/tripleo-ci-centos-7-undercloud-containers/2dbc5a7/ : SUCCESS in 1h 36m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/106c323/ : SUCCESS in 2h 28m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/14/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/90c2c23/ : SUCCESS in 2h 37m 34s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/14/check/tripleo-ci-centos-7-standalone/9a8dc0d/ : SUCCESS in 1h 59m 30s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/14/check/tripleo-ci-fedora-28-standalone/2e7dbde/ : SUCCESS in 1h 45m 25s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/14/check/tripleo-ci-centos-7-standalone-upgrade/760fd95/ : SUCCESS in 2h 05m 27s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/14/check/tripleo-ci-centos-7-standalone-upgrade-stein/f506e64/ : SUCCESS in 2h 25m 57s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/14/check/openstack-tox-pep8/7b8959b/ : SUCCESS in 5m 48s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/14/check/openstack-tox-py27/2861610/ : SUCCESS in 5m 01s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/14/check/openstack-tox-py35/50e05da/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/14/check/openstack-tox-py36/3840ba1/ : SUCCESS in 4m 57s\n- openstack-tox-docs http://logs.openstack.org/85/659885/14/check/openstack-tox-docs/27d97b1/html/ : SUCCESS in 6m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/14/check/openstack-tox-lower-constraints/c9f7e32/ : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":14},{"id":"16839d83205eb64a8c3dd6ed7ae2f097136c4875","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-14 09:17:00.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"3652d5ace1e016664c317a30b2ebf83b51ab0cd2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-14 09:55:02.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8e34b160ea8215b9a5729fd3beb9fabceb3b7cdb","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-06-14 10:50:03.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4826b2ec0e80c9122b6ecf5f30e1de74b1d4b655","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-14 11:35:36.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/15/check/tripleo-ci-centos-7-containers-multinode/ed07204/ : SUCCESS in 1h 57m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/15/check/tripleo-ci-centos-7-undercloud-containers/70a8f92/ : SUCCESS in 1h 30m 49s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/15/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/b8bea21/ : SUCCESS in 2h 07m 00s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/15/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/f479a54/ : SUCCESS in 1h 58m 49s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/15/check/tripleo-ci-centos-7-standalone/678673a/ : SUCCESS in 1h 53m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/15/check/tripleo-ci-fedora-28-standalone/0c14219/ : SUCCESS in 1h 48m 29s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/15/check/tripleo-ci-centos-7-standalone-upgrade/36a3564/ : SUCCESS in 2h 05m 47s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/15/check/tripleo-ci-centos-7-standalone-upgrade-stein/87eca35/ : SUCCESS in 2h 04m 27s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/15/check/openstack-tox-pep8/109d297/ : SUCCESS in 11m 01s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/15/check/openstack-tox-py27/bd9895c/ : SUCCESS in 5m 07s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/15/check/openstack-tox-py35/6cfacfb/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/15/check/openstack-tox-py36/ba4edb2/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/85/659885/15/check/openstack-tox-docs/08011e6/html/ : SUCCESS in 5m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/15/check/openstack-tox-lower-constraints/5a7a413/ : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":15},{"id":"a3d64a327a11e4d769d6757d04070c3a01037987","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2019-06-17 18:29:51.000000000","message":"Patch Set 15: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":15},{"id":"5e340d1f6ece9936df3e946c90e9e0dc7756d145","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-07-08 19:57:12.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"5145d812d533092c3ee1aebf196bacab670d591c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 02:23:41.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/16/check/tripleo-ci-centos-7-containers-multinode/f0d3d9f/ : SUCCESS in 2h 22m 14s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/16/check/tripleo-ci-centos-7-undercloud-containers/2d9947e/ : FAILURE in 1h 05m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1a1465a/ : SUCCESS in 2h 34m 12s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/16/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c27a3e2/ : FAILURE in 59m 00s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/16/check/tripleo-ci-centos-7-standalone/525e4e4/ : SUCCESS in 1h 59m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/16/check/tripleo-ci-fedora-28-standalone/fdb380e/ : FAILURE in 29m 32s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/16/check/tripleo-ci-centos-7-standalone-upgrade/46280b5/ : SUCCESS in 2h 24m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/16/check/tripleo-ci-centos-7-standalone-upgrade-stein/802ca91/ : SUCCESS in 2h 17m 27s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/16/check/openstack-tox-pep8/a0ed30c/ : SUCCESS in 4m 45s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/16/check/openstack-tox-py27/f605260/ : SUCCESS in 4m 01s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/16/check/openstack-tox-py35/7472fd5/ : SUCCESS in 4m 46s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/16/check/openstack-tox-py36/b31b513/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/85/659885/16/check/openstack-tox-docs/9006cc7/html/ : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/16/check/openstack-tox-lower-constraints/3f583d3/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":16},{"id":"ccc0d931febf896cb1f33b748e45eb87d092fa46","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-07-09 18:51:33.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"5ec714cc1c0319891c80cb2cdc2eb7bbf84f99ac","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-07-09 19:03:20.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c50ede79ab543d84df58e3d0fa79c7a181ac9a09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 02:52:56.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/17/check/tripleo-ci-centos-7-containers-multinode/a4f1ad3/ : SUCCESS in 2h 00m 25s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/17/check/tripleo-ci-centos-7-undercloud-containers/6e0f7f4/ : SUCCESS in 1h 49m 30s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/285da01/ : SUCCESS in 2h 32m 20s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/17/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/cfe385c/ : SUCCESS in 2h 39m 20s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/17/check/tripleo-ci-centos-7-standalone/de89996/ : SUCCESS in 2h 01m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/17/check/tripleo-ci-fedora-28-standalone/a14f0c7/ : SUCCESS in 1h 45m 41s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/17/check/tripleo-ci-centos-7-standalone-upgrade/b478341/ : SUCCESS in 2h 36m 12s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/17/check/tripleo-ci-centos-7-standalone-upgrade-stein/dcead1d/ : SUCCESS in 2h 28m 42s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/17/check/openstack-tox-pep8/1cb441a/ : SUCCESS in 6m 08s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/17/check/openstack-tox-py27/510dff5/ : SUCCESS in 4m 34s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/17/check/openstack-tox-py35/1fd9bd8/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/17/check/openstack-tox-py36/a4518f2/ : SUCCESS in 4m 59s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/17/check/openstack-tox-molecule/dd3e522/tox/reports.html : SUCCESS in 7m 57s\n- openstack-tox-docs http://logs.openstack.org/85/659885/17/check/openstack-tox-docs/081f5e5/html/ : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/17/check/openstack-tox-lower-constraints/d69ebd0/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":17},{"id":"30ae7091581695e1bd3d69c7e71f2e6206e59406","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-07-10 16:57:09.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"ea5defdadb1348abf1791d428bb729576911d8b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 22:07:58.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/18/check/tripleo-ci-centos-7-containers-multinode/b47ccc9/ : SUCCESS in 2h 12m 35s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/18/check/tripleo-ci-centos-7-undercloud-containers/1181906/ : SUCCESS in 1h 37m 05s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/18/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f1889ae/ : SUCCESS in 2h 26m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/18/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/3da7333/ : FAILURE in 2h 31m 00s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/18/check/tripleo-ci-centos-7-standalone/6141671/ : SUCCESS in 1h 58m 16s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/18/check/tripleo-ci-fedora-28-standalone/b65922b/ : FAILURE in 1h 50m 29s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/18/check/tripleo-ci-centos-7-standalone-upgrade/f1a45a0/ : FAILURE in 2h 10m 15s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/85/659885/18/check/tripleo-ci-centos-7-standalone-upgrade-stein/5ed9711/ : SUCCESS in 2h 18m 23s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/18/check/openstack-tox-pep8/e58897b/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/18/check/openstack-tox-py27/84c31bb/ : SUCCESS in 4m 16s\n- openstack-tox-py35 http://logs.openstack.org/85/659885/18/check/openstack-tox-py35/3cca258/ : SUCCESS in 6m 41s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/18/check/openstack-tox-py36/efe0019/ : SUCCESS in 5m 03s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/18/check/openstack-tox-molecule/18c4d24/tox/reports.html : SUCCESS in 8m 48s\n- openstack-tox-docs http://logs.openstack.org/85/659885/18/check/openstack-tox-docs/f24138e/html/ : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/18/check/openstack-tox-lower-constraints/4a5f2c3/ : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":18},{"id":"70ba574717c8557d706e5b6914651d30cabe038d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-15 15:18:56.000000000","message":"Patch Set 18:\n\nAde, Harry,\n\nIs this patch ready for review ?\n\nWhatever your answers, I am gonna test and review it tomorrow morning.","accounts_in_message":[],"_revision_number":18},{"id":"2ab77020d54b8e656b04687f83ee11cc8d45a180","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-07-15 15:27:52.000000000","message":"Patch Set 18:\n\nGael, yep! Reviews are very much welcome. At this point I\u0027m just writing up some complimentary documentation to accompany the work.","accounts_in_message":[],"_revision_number":18},{"id":"3e98eae2a558a521eba06ae23cdf9c6723565429","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-07-16 15:07:52.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5216a35f1dd044b0bade156e81c34114a986b6f6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-16 15:12:29.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"c5e5382bacbaedcae2622d2f3c3ea8ecf4732fed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 23:06:23.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-containers-multinode/eed243d/ : SUCCESS in 2h 43m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-undercloud-containers/245c5d8/ : SUCCESS in 2h 00m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1717740/ : SUCCESS in 3h 09m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/d3e8087/ : FAILURE in 2h 49m 36s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-standalone/7dbf546/ : SUCCESS in 1h 50m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/19/check/tripleo-ci-fedora-28-standalone/b1bfd9e/ : SUCCESS in 1h 44m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-standalone-upgrade/9a6feec/ : FAILURE in 2h 01m 05s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/19/check/openstack-tox-pep8/d5ffbd5/ : SUCCESS in 5m 13s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/19/check/openstack-tox-py27/8f1b6b2/ : SUCCESS in 4m 49s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/19/check/openstack-tox-py36/dcbbfbc/ : SUCCESS in 4m 08s\n- openstack-tox-py37 http://logs.openstack.org/85/659885/19/check/openstack-tox-py37/9717595/ : SUCCESS in 4m 04s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/19/check/openstack-tox-molecule/6f172e2/tox/reports.html : SUCCESS in 10m 40s\n- openstack-tox-docs http://logs.openstack.org/85/659885/19/check/openstack-tox-docs/65f292f/html/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/19/check/openstack-tox-lower-constraints/d807eb3/ : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":19},{"id":"5063af92e23a3d6bf9f28e1ea91b095c114c45c6","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-17 05:54:30.000000000","message":"Patch Set 19: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"87c887b0eb08cd57e4381363e19b6c7d5c9f0a3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 05:54:44.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"43a3384602aa00fe0b44143f2f3879028119aa4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 08:57:33.000000000","message":"Patch Set 19: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/19/gate/tripleo-ci-centos-7-containers-multinode/a34cdcb/ : FAILURE in 1h 46m 35s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/19/gate/tripleo-ci-centos-7-standalone/3cf1c82/ : SUCCESS in 2h 09m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/19/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bce6bc0/ : SUCCESS in 2h 51m 10s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/19/gate/openstack-tox-pep8/3cf4778/ : SUCCESS in 5m 14s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/19/gate/openstack-tox-py27/eed6dbb/ : SUCCESS in 4m 47s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/19/gate/openstack-tox-py36/d9e0474/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/85/659885/19/gate/openstack-tox-py37/e39b5e1/ : SUCCESS in 4m 14s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/19/gate/openstack-tox-molecule/10961a3/tox/reports.html : FAILURE in 10m 59s\n- openstack-tox-docs http://logs.openstack.org/85/659885/19/gate/openstack-tox-docs/ab2b8fa/html/ : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/19/gate/openstack-tox-lower-constraints/b5e38f5/ : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":19},{"id":"081571921827f68c526946ad52b73c06907c6be7","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-07-18 01:56:43.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"4e233634194b2e9c58040bbeec6912a9b91c6e6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 04:58:09.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-containers-multinode/ccd8dc0/ : FAILURE in 1h 02m 28s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-undercloud-containers/4b4fc00/ : SUCCESS in 1h 50m 54s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/12c0c4b/ : FAILURE in 1h 00m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/e730ef2/ : FAILURE in 2h 55m 56s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-standalone/561b35d/ : SUCCESS in 1h 58m 01s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/19/check/tripleo-ci-fedora-28-standalone/2107c29/ : SUCCESS in 2h 05m 48s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-standalone-upgrade/11d6f39/ : FAILURE in 2h 08m 18s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/19/check/openstack-tox-pep8/a9370f6/ : SUCCESS in 5m 22s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/19/check/openstack-tox-py27/6544d12/ : SUCCESS in 6m 23s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/19/check/openstack-tox-py36/f313494/ : SUCCESS in 4m 56s\n- openstack-tox-py37 http://logs.openstack.org/85/659885/19/check/openstack-tox-py37/b8c5884/ : SUCCESS in 6m 05s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/19/check/openstack-tox-molecule/2305ae0/tox/reports.html : FAILURE in 11m 59s\n- openstack-tox-docs http://logs.openstack.org/85/659885/19/check/openstack-tox-docs/0915810/html/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/19/check/openstack-tox-lower-constraints/2754a2f/ : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":19},{"id":"eba09c66bcba02cb13614692254c19e3968878e2","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-07-19 10:14:54.000000000","message":"Patch Set 19:\n\nrecheck","accounts_in_message":[],"_revision_number":19},{"id":"f0180fd0818506aad328e0c8c14129d624981f9b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:27:49.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-containers-multinode/2b212ac/ : SUCCESS in 2h 08m 57s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-undercloud-containers/41f175a/ : SUCCESS in 2h 02m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/5cccc34/ : SUCCESS in 2h 09m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c3220e5/ : FAILURE in 3h 09m 58s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-standalone/80450e6/ : SUCCESS in 1h 43m 02s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/19/check/tripleo-ci-fedora-28-standalone/1ee0db0/ : FAILURE in 33m 23s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/19/check/tripleo-ci-centos-7-standalone-upgrade/8cec6e6/ : FAILURE in 1h 48m 04s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/19/check/openstack-tox-pep8/086b600/ : SUCCESS in 5m 32s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/19/check/openstack-tox-py27/9cb13cc/ : SUCCESS in 3m 59s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/19/check/openstack-tox-py36/c2b486e/ : SUCCESS in 5m 03s\n- openstack-tox-py37 http://logs.openstack.org/85/659885/19/check/openstack-tox-py37/18f99e1/ : SUCCESS in 4m 09s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/19/check/openstack-tox-molecule/cec681c/tox/reports.html : FAILURE in 11m 56s\n- openstack-tox-docs http://logs.openstack.org/85/659885/19/check/openstack-tox-docs/0780279/html/ : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/19/check/openstack-tox-lower-constraints/49f2492/ : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":19},{"id":"f68a5a1d6d6b88a00b77ba8cb060baf0ba65dd96","author":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"date":"2019-07-22 13:02:02.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"e61f4df5d43c9516eb9a85e52debfda8c7888b8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 15:59:33.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/20/check/tripleo-ci-centos-7-containers-multinode/0686f64/ : SUCCESS in 2h 29m 07s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/85/659885/20/check/tripleo-ci-centos-7-undercloud-containers/f14dbee/ : SUCCESS in 1h 26m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/20/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c733455/ : SUCCESS in 2h 19m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/85/659885/20/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/5f5d7bc/ : SUCCESS in 2h 36m 42s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/20/check/tripleo-ci-centos-7-standalone/eaa664a/ : SUCCESS in 1h 45m 30s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/85/659885/20/check/tripleo-ci-fedora-28-standalone/0bd51d7/ : SUCCESS in 1h 47m 51s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/85/659885/20/check/tripleo-ci-centos-7-standalone-upgrade/adba6e1/ : FAILURE in 2h 15m 58s (non-voting)\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/20/check/openstack-tox-pep8/a267d28/ : SUCCESS in 5m 44s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/20/check/openstack-tox-py27/a57523d/ : SUCCESS in 5m 05s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/20/check/openstack-tox-py36/a28729b/ : SUCCESS in 5m 19s\n- openstack-tox-py37 http://logs.openstack.org/85/659885/20/check/openstack-tox-py37/f5d927d/ : SUCCESS in 6m 04s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/20/check/openstack-tox-molecule/f8091c0/tox/reports.html : SUCCESS in 11m 55s\n- openstack-tox-docs http://logs.openstack.org/85/659885/20/check/openstack-tox-docs/6e0fde2/html/ : SUCCESS in 3m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/20/check/openstack-tox-lower-constraints/ce9e380/ : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":20},{"id":"62996ca4e47fa80457c340f9bb18dea6c4cfe009","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-07-23 15:20:08.000000000","message":"Patch Set 20: Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"1720ae70a267f8bbc6abca390e11c875d2d1992b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 15:20:26.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"5ff274271d063c126b06d4f91598437b87ae3dcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:52:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":20},{"id":"049362098933c3715ff3aa88d5c619bc5045f766","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 17:52:09.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/85/659885/20/gate/tripleo-ci-centos-7-containers-multinode/57e2743/ : SUCCESS in 2h 25m 52s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/85/659885/20/gate/tripleo-ci-centos-7-standalone/8c3168d/ : SUCCESS in 1h 51m 55s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/85/659885/20/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0ba5a1f/ : SUCCESS in 2h 23m 00s\n- openstack-tox-pep8 http://logs.openstack.org/85/659885/20/gate/openstack-tox-pep8/c739850/ : SUCCESS in 4m 34s\n- openstack-tox-py27 http://logs.openstack.org/85/659885/20/gate/openstack-tox-py27/66d331d/ : SUCCESS in 8m 42s\n- openstack-tox-py36 http://logs.openstack.org/85/659885/20/gate/openstack-tox-py36/0945dcf/ : SUCCESS in 6m 48s\n- openstack-tox-py37 http://logs.openstack.org/85/659885/20/gate/openstack-tox-py37/29068c0/ : SUCCESS in 5m 22s\n- openstack-tox-molecule http://logs.openstack.org/85/659885/20/gate/openstack-tox-molecule/493cfbe/tox/reports.html : SUCCESS in 21m 56s\n- openstack-tox-docs http://logs.openstack.org/85/659885/20/gate/openstack-tox-docs/b384343/html/ : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/85/659885/20/gate/openstack-tox-lower-constraints/aa8fbcc/ : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"fe467b21e0a9f100ca7544944fd2efb13d089269","revisions":{"e664df0ec1dcd01842bc613d7733be094cf63da2":{"kind":"REWORK","_number":1,"created":"2019-05-17 21:59:34.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/85/659885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/1"}}},"commit":{"parents":[{"commit":"cec7b84484f3b93d9f8b6848d9233bde741cfcf5","subject":"Merge \"Fix Jinja2\u0027s default filter usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cec7b84484f3b93d9f8b6848d9233bde741cfcf5"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 21:59:33.000000000","tz":-240},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.  I still have a number of\nquestions including:\n-- do we need special tags ?  how do we make sure these tests\n   only run when deploying tls-everywhere?\n-- how is output handled?\n\nRight now, this gives you an idea how things are shaping up.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e664df0ec1dcd01842bc613d7733be094cf63da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e664df0ec1dcd01842bc613d7733be094cf63da2"}]},"branch":"refs/heads/master"},"2250785b3abfdd27c8fadf0c54286e233f8a3af9":{"kind":"REWORK","_number":2,"created":"2019-05-20 22:10:27.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/85/659885/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/2"}}},"commit":{"parents":[{"commit":"cec7b84484f3b93d9f8b6848d9233bde741cfcf5","subject":"Merge \"Fix Jinja2\u0027s default filter usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cec7b84484f3b93d9f8b6848d9233bde741cfcf5"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-20 22:09:58.000000000","tz":-240},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2250785b3abfdd27c8fadf0c54286e233f8a3af9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2250785b3abfdd27c8fadf0c54286e233f8a3af9"}]},"branch":"refs/heads/master"},"ddade8948a9a89bddd1a5e40e96a4e395f08a166":{"kind":"REWORK","_number":3,"created":"2019-05-20 22:13:57.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/85/659885/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/3"}}},"commit":{"parents":[{"commit":"cec7b84484f3b93d9f8b6848d9233bde741cfcf5","subject":"Merge \"Fix Jinja2\u0027s default filter usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/cec7b84484f3b93d9f8b6848d9233bde741cfcf5"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-20 22:13:12.000000000","tz":-240},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ddade8948a9a89bddd1a5e40e96a4e395f08a166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/ddade8948a9a89bddd1a5e40e96a4e395f08a166"}]},"branch":"refs/heads/master"},"1c21268f78a063c0443709017cd3a1cc4876e9a6":{"kind":"REWORK","_number":4,"created":"2019-05-27 11:01:17.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/85/659885/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/4"}}},"commit":{"parents":[{"commit":"eb9ca985ef5b6653fc5fa61eb3a451e30316de6c","subject":"Merge \"Makes rogue_dhcp.py script Python3 aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/eb9ca985ef5b6653fc5fa61eb3a451e30316de6c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-27 11:01:02.000000000","tz":120},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1c21268f78a063c0443709017cd3a1cc4876e9a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/1c21268f78a063c0443709017cd3a1cc4876e9a6"}]},"branch":"refs/heads/master"},"7be55bc93b26082ee6c2cb1b4a0e98c0791432d7":{"kind":"REWORK","_number":5,"created":"2019-05-27 12:12:13.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/85/659885/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/5"}}},"commit":{"parents":[{"commit":"eb9ca985ef5b6653fc5fa61eb3a451e30316de6c","subject":"Merge \"Makes rogue_dhcp.py script Python3 aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/eb9ca985ef5b6653fc5fa61eb3a451e30316de6c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-27 12:11:23.000000000","tz":120},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7be55bc93b26082ee6c2cb1b4a0e98c0791432d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/7be55bc93b26082ee6c2cb1b4a0e98c0791432d7"}]},"branch":"refs/heads/master"},"991374745f2449f896042a979ec5da8232a4c5f4":{"kind":"REWORK","_number":6,"created":"2019-05-27 14:59:29.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/85/659885/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/6"}}},"commit":{"parents":[{"commit":"eb9ca985ef5b6653fc5fa61eb3a451e30316de6c","subject":"Merge \"Makes rogue_dhcp.py script Python3 aware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/eb9ca985ef5b6653fc5fa61eb3a451e30316de6c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-27 14:59:04.000000000","tz":120},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/991374745f2449f896042a979ec5da8232a4c5f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/991374745f2449f896042a979ec5da8232a4c5f4"}]},"branch":"refs/heads/master"},"c34dd391d34f6ba3e1f290305e1772e0cad28317":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-06-04 12:59:27.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/7"}}},"commit":{"parents":[{"commit":"8b78b96ab4899b9adc01dcedf1f8d228884e32cf","subject":"Fix validations-specific Ansible output callback","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8b78b96ab4899b9adc01dcedf1f8d228884e32cf"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-06-04 12:59:05.000000000","tz":-240},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c34dd391d34f6ba3e1f290305e1772e0cad28317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c34dd391d34f6ba3e1f290305e1772e0cad28317"}]},"branch":"refs/heads/master"},"8cc2eaf2f2c4b20c1d7b251ebb72e04b35ce2fe0":{"kind":"REWORK","_number":8,"created":"2019-06-05 10:26:48.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/85/659885/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/8"}}},"commit":{"parents":[{"commit":"c18eadb78ab4085c199f5f80ea55f0aedb633001","subject":"Fix validations-specific Ansible output callback","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c18eadb78ab4085c199f5f80ea55f0aedb633001"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-06-05 10:24:33.000000000","tz":0},"subject":"Add new role for tls-everywhere","message":"Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cc2eaf2f2c4b20c1d7b251ebb72e04b35ce2fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cc2eaf2f2c4b20c1d7b251ebb72e04b35ce2fe0"}]},"branch":"refs/heads/master"},"2cc820135395c117749d4dd5e7fa81cafe05f81a":{"kind":"REWORK","_number":9,"created":"2019-06-05 13:15:21.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/9"}}},"commit":{"parents":[{"commit":"c18eadb78ab4085c199f5f80ea55f0aedb633001","subject":"Fix validations-specific Ansible output callback","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c18eadb78ab4085c199f5f80ea55f0aedb633001"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-06-05 13:15:07.000000000","tz":-240},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2cc820135395c117749d4dd5e7fa81cafe05f81a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2cc820135395c117749d4dd5e7fa81cafe05f81a"}]},"branch":"refs/heads/master"},"8e53da07921e6a2cb5b8c60268340c5d7d77dc11":{"kind":"REWORK","_number":10,"created":"2019-06-06 07:26:21.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/85/659885/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/10"}}},"commit":{"parents":[{"commit":"c18eadb78ab4085c199f5f80ea55f0aedb633001","subject":"Fix validations-specific Ansible output callback","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c18eadb78ab4085c199f5f80ea55f0aedb633001"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2019-06-06 07:25:51.000000000","tz":120},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8e53da07921e6a2cb5b8c60268340c5d7d77dc11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8e53da07921e6a2cb5b8c60268340c5d7d77dc11"}]},"branch":"refs/heads/master"},"c68f4be2a9460c9b384ee3a716fa54ea1b1f9825":{"kind":"REWORK","_number":11,"created":"2019-06-06 14:45:21.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/11"}}},"commit":{"parents":[{"commit":"c18eadb78ab4085c199f5f80ea55f0aedb633001","subject":"Fix validations-specific Ansible output callback","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c18eadb78ab4085c199f5f80ea55f0aedb633001"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-06-06 14:44:52.000000000","tz":-240},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c68f4be2a9460c9b384ee3a716fa54ea1b1f9825"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/c68f4be2a9460c9b384ee3a716fa54ea1b1f9825"}]},"branch":"refs/heads/master"},"2ccc67ac177d0e7f32c34dff6f1b44c96070f0e4":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-06-07 09:33:39.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/85/659885/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/12"}}},"commit":{"parents":[{"commit":"8cad90854ff9eefe7fb19b6678737f55a620021c","subject":"Merge \"Fix validations-specific Ansible output callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cad90854ff9eefe7fb19b6678737f55a620021c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 09:33:39.000000000","tz":0},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2ccc67ac177d0e7f32c34dff6f1b44c96070f0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2ccc67ac177d0e7f32c34dff6f1b44c96070f0e4"}]},"branch":"refs/heads/master"},"9b96bfbf684d673eb7e439813c3a53c74d01abac":{"kind":"REWORK","_number":13,"created":"2019-06-10 12:03:11.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/13"}}},"commit":{"parents":[{"commit":"8cad90854ff9eefe7fb19b6678737f55a620021c","subject":"Merge \"Fix validations-specific Ansible output callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cad90854ff9eefe7fb19b6678737f55a620021c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-06-10 12:02:43.000000000","tz":-240},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9b96bfbf684d673eb7e439813c3a53c74d01abac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9b96bfbf684d673eb7e439813c3a53c74d01abac"}]},"branch":"refs/heads/master"},"a5d38799737766b2ae1aae558d6931c255f6618d":{"kind":"REWORK","_number":14,"created":"2019-06-11 07:55:12.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/14"}}},"commit":{"parents":[{"commit":"8cad90854ff9eefe7fb19b6678737f55a620021c","subject":"Merge \"Fix validations-specific Ansible output callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cad90854ff9eefe7fb19b6678737f55a620021c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-06-11 07:55:01.000000000","tz":-240},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a5d38799737766b2ae1aae558d6931c255f6618d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a5d38799737766b2ae1aae558d6931c255f6618d"}]},"branch":"refs/heads/master"},"e146c14d10d58e08c86ec397a881e3adbe7531f2":{"kind":"REWORK","_number":15,"created":"2019-06-14 09:17:00.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/15"}}},"commit":{"parents":[{"commit":"8cad90854ff9eefe7fb19b6678737f55a620021c","subject":"Merge \"Fix validations-specific Ansible output callback\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8cad90854ff9eefe7fb19b6678737f55a620021c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-06-14 09:16:46.000000000","tz":-240},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e146c14d10d58e08c86ec397a881e3adbe7531f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e146c14d10d58e08c86ec397a881e3adbe7531f2"}]},"branch":"refs/heads/master"},"aa3c4c4210bee961e89248e8b4878fb65cad0870":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2019-07-08 19:57:12.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/16"}}},"commit":{"parents":[{"commit":"42b2a2068cb4c2c0369a47b5c7fc3d376e97c7ef","subject":"Merge \"Deletes validations directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/42b2a2068cb4c2c0369a47b5c7fc3d376e97c7ef"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-07-08 19:57:07.000000000","tz":-240},"subject":"WIP: Add new role for tls-everywhere","message":"WIP: Add new role for tls-everywhere\n\nThis provides a role and playbook for various validations\nwhen deploying tls-everywhere.\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/aa3c4c4210bee961e89248e8b4878fb65cad0870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/aa3c4c4210bee961e89248e8b4878fb65cad0870"}]},"branch":"refs/heads/master"},"11b7c78b90b207696f203646441be531c9aef900":{"kind":"REWORK","_number":17,"created":"2019-07-09 19:03:20.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/17"}}},"commit":{"parents":[{"commit":"42b2a2068cb4c2c0369a47b5c7fc3d376e97c7ef","subject":"Merge \"Deletes validations directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/42b2a2068cb4c2c0369a47b5c7fc3d376e97c7ef"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-07-09 19:01:25.000000000","tz":-240},"subject":"Add new role to validate tls-everywhere","message":"Add new role to validate tls-everywhere\n\nThis provides a role and playbooks for various validations\nwhen deploying tls-everywhere.\n\nEssentially this is broken into three phases (playbooks):\n- Verify the the undercloud node is prepared for novajoin prior\n  to installing the undercloud\n- Verify novajoin is setup correctly and that the overcloud is\n  ready to deploy and enroll nodes with the IdM\n- Verify overcloud nodes are enrolled in the IdM with certs\n  tracked an in a MONITOR state by certmonger\n\nCo-Authored-By: Harry Rybacki hrybacki@redhat.com\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/11b7c78b90b207696f203646441be531c9aef900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/11b7c78b90b207696f203646441be531c9aef900"}]},"branch":"refs/heads/master"},"167eb23b654e608d33fbfc088123ce285cc7d964":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2019-07-10 16:57:09.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/18"}}},"commit":{"parents":[{"commit":"2f4921b62dc2fc71822b3375bc112f202d76478c","subject":"Merge \"Add Molecule tests for the role xfs-check-ftype.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2f4921b62dc2fc71822b3375bc112f202d76478c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-07-10 16:57:09.000000000","tz":0},"subject":"Add new role to validate tls-everywhere","message":"Add new role to validate tls-everywhere\n\nThis provides a role and playbooks for various validations\nwhen deploying tls-everywhere.\n\nEssentially this is broken into three phases (playbooks):\n- Verify the the undercloud node is prepared for novajoin prior\n  to installing the undercloud\n- Verify novajoin is setup correctly and that the overcloud is\n  ready to deploy and enroll nodes with the IdM\n- Verify overcloud nodes are enrolled in the IdM with certs\n  tracked an in a MONITOR state by certmonger\n\nCo-Authored-By: Harry Rybacki hrybacki@redhat.com\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/167eb23b654e608d33fbfc088123ce285cc7d964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/167eb23b654e608d33fbfc088123ce285cc7d964"}]},"branch":"refs/heads/master"},"656bac791747cbf9cf60225d0e1ec49c330f4228":{"kind":"REWORK","_number":19,"created":"2019-07-16 15:07:52.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/19"}}},"commit":{"parents":[{"commit":"2f4921b62dc2fc71822b3375bc112f202d76478c","subject":"Merge \"Add Molecule tests for the role xfs-check-ftype.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/2f4921b62dc2fc71822b3375bc112f202d76478c"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-07-16 15:07:47.000000000","tz":-240},"subject":"Add new role to validate tls-everywhere","message":"Add new role to validate tls-everywhere\n\nThis provides a role and playbooks for various validations\nwhen deploying tls-everywhere.\n\nEssentially this is broken into three phases (playbooks):\n- Verify the the undercloud node is prepared for novajoin prior\n  to installing the undercloud\n- Verify novajoin is setup correctly and that the overcloud is\n  ready to deploy and enroll nodes with the IdM\n- Verify overcloud nodes are enrolled in the IdM with certs\n  tracked an in a MONITOR state by certmonger\n\nCo-Authored-By: Harry Rybacki hrybacki@redhat.com\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/656bac791747cbf9cf60225d0e1ec49c330f4228"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/656bac791747cbf9cf60225d0e1ec49c330f4228"}]},"branch":"refs/heads/master"},"fe467b21e0a9f100ca7544944fd2efb13d089269":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2019-07-22 13:02:02.000000000","uploader":{"_account_id":11589,"name":"Harry Rybacki","email":"hrybacki@redhat.com","username":"hrybacki"},"ref":"refs/changes/85/659885/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/85/659885/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/85/659885/20"}}},"commit":{"parents":[{"commit":"e0cbceaf523a7cc24fbd4656ab196778187e7df9","subject":"Merge \"Support multi-arch in deploy image validations\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/e0cbceaf523a7cc24fbd4656ab196778187e7df9"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2019-05-17 20:51:16.000000000","tz":-240},"committer":{"name":"Harry Rybacki","email":"hrybacki@redhat.com","date":"2019-07-22 13:02:02.000000000","tz":0},"subject":"Add new role to validate tls-everywhere","message":"Add new role to validate tls-everywhere\n\nThis provides a role and playbooks for various validations\nwhen deploying tls-everywhere.\n\nEssentially this is broken into three phases (playbooks):\n- Verify the the undercloud node is prepared for novajoin prior\n  to installing the undercloud\n- Verify novajoin is setup correctly and that the overcloud is\n  ready to deploy and enroll nodes with the IdM\n- Verify overcloud nodes are enrolled in the IdM with certs\n  tracked an in a MONITOR state by certmonger\n\nCo-Authored-By: Harry Rybacki hrybacki@redhat.com\n\nChange-Id: I31feafcdfd8467a00aa96bcafa567ca06b1bd86a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fe467b21e0a9f100ca7544944fd2efb13d089269"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/fe467b21e0a9f100ca7544944fd2efb13d089269"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
