)]}'
{"id":"openstack%2Ftripleo-validations~663619","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I91d326d05b919e45fad19732bf9bdd5e6a032636","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I91d326d05b919e45fad19732bf9bdd5e6a032636","subject":"Sync \u0027validations/undercloud-ram.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-06 12:04:01.000000000","updated":"2019-06-08 01:38:27.000000000","submitted":"2019-06-08 01:38:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663619-1559957907815-2d8656ec","meta_rev_id":"33bb8bde7c2e3f5b9075d93df09ea941528a29da","_number":663619,"virtual_id_number":663619,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-08 01:38:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 09:40:58.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-06 14:39:16.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-06-07 13:38:33.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 14:39:16.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 09:40:58.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 13:38:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-08 01:38:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d39b9aa8d60e350e02d5a75ddc6aeec1a486184c","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 12:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b0e4794f052eb893f844f19f7a7681adaf25a04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 15:46:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/663619/1/check/tripleo-ci-centos-7-containers-multinode/aeb1010/ : SUCCESS in 1h 54m 34s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/19/663619/1/check/tripleo-ci-centos-7-undercloud-containers/c826bdf/ : SUCCESS in 1h 31m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/663619/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6a3337f/ : SUCCESS in 2h 00m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/19/663619/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0908355/ : FAILURE in 2h 34m 26s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/663619/1/check/tripleo-ci-centos-7-standalone/773bd5e/ : SUCCESS in 1h 46m 30s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/663619/1/check/tripleo-ci-fedora-28-standalone/0e721fe/ : SUCCESS in 1h 52m 07s\n- openstack-tox-pep8 http://logs.openstack.org/19/663619/1/check/openstack-tox-pep8/5674afb/ : SUCCESS in 6m 34s\n- openstack-tox-py27 http://logs.openstack.org/19/663619/1/check/openstack-tox-py27/001ac40/ : SUCCESS in 5m 57s\n- openstack-tox-py35 http://logs.openstack.org/19/663619/1/check/openstack-tox-py35/c54f9bf/ : SUCCESS in 5m 20s\n- openstack-tox-py36 http://logs.openstack.org/19/663619/1/check/openstack-tox-py36/044770c/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/19/663619/1/check/openstack-tox-docs/bb6de6c/html/ : SUCCESS in 4m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/663619/1/check/openstack-tox-lower-constraints/1b05cfe/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a2916b9167aa0099c05c6bf3fd533607c11571dd","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-06 19:44:31.000000000","message":"Patch Set 1:\n\na general question for all those \"sync\" patches: can\u0027t we tweak the package in order to make symlinks instead?","accounts_in_message":[],"_revision_number":1},{"id":"e5dc66344ed8115d8050c86397b83d32eb7f1a96","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 09:40:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e9b86abab06cffabc7cbf70e1c7db7f3dc50d239","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 13:38:33.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71d783cb3a3dfa543b1977c716b229d35ee7cf32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:38:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec05019833c0fdbbe48701345cb2a2ff1e8eec11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 21:06:27.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c870b1b13f8dfd91bfeef78fe2f8b1730f58460e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 01:38:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/19/663619/1/gate/tripleo-ci-centos-7-containers-multinode/2512446/ : SUCCESS in 2h 00m 15s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/19/663619/1/gate/tripleo-ci-centos-7-standalone/28acfe7/ : SUCCESS in 1h 49m 44s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/19/663619/1/gate/tripleo-ci-fedora-28-standalone/43a616e/ : SUCCESS in 2h 00m 14s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/19/663619/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/31f99ff/ : SUCCESS in 2h 09m 16s\n- openstack-tox-pep8 http://logs.openstack.org/19/663619/1/gate/openstack-tox-pep8/07268dd/ : SUCCESS in 6m 50s\n- openstack-tox-py27 http://logs.openstack.org/19/663619/1/gate/openstack-tox-py27/038ce44/ : SUCCESS in 5m 47s\n- openstack-tox-py35 http://logs.openstack.org/19/663619/1/gate/openstack-tox-py35/5710edc/ : SUCCESS in 6m 09s\n- openstack-tox-py36 http://logs.openstack.org/19/663619/1/gate/openstack-tox-py36/be6c304/ : SUCCESS in 6m 13s\n- openstack-tox-docs http://logs.openstack.org/19/663619/1/gate/openstack-tox-docs/ea36245/html/ : SUCCESS in 5m 56s\n- openstack-tox-lower-constraints http://logs.openstack.org/19/663619/1/gate/openstack-tox-lower-constraints/30487ce/ : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8847749da95626a792106b2e0f4bd0193c970f3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 01:38:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9aeeb97847f14501afb8034d300f7004297f0caf","revisions":{"9aeeb97847f14501afb8034d300f7004297f0caf":{"kind":"REWORK","_number":1,"created":"2019-06-06 12:04:01.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/19/663619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/19/663619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/19/663619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/19/663619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/19/663619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/19/663619/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 12:02:22.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 12:02:22.000000000","tz":120},"subject":"Sync \u0027validations/undercloud-ram.yaml\u0027 with the new role","message":"Sync \u0027validations/undercloud-ram.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the undercloud-ram validation.\n\nChange-Id: I91d326d05b919e45fad19732bf9bdd5e6a032636\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9aeeb97847f14501afb8034d300f7004297f0caf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9aeeb97847f14501afb8034d300f7004297f0caf"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
