)]}'
{"id":"openstack%2Ftripleo-validations~663623","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Ie4f7e371ba9dc51a1be1473c18ea596e19fa233f","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"Ie4f7e371ba9dc51a1be1473c18ea596e19fa233f","subject":"Sync \u0027validations/check-ftype.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-06 12:10:50.000000000","updated":"2019-06-08 01:45:57.000000000","submitted":"2019-06-08 01:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663623-1559958357985-b5d80b2d","meta_rev_id":"dae7867dd7fbe2f0e49bf12e5a1c04f2aedea94d","_number":663623,"virtual_id_number":663623,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-08 01:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 09:42:48.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-06 14:39:09.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-06-07 14:08:11.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 14:39:09.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 09:42:48.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 14:08:11.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-08 01:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3f701aac48e20db9b4041449f9fd8dac4927819","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 12:10:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47cd511ffd249cae8653563d0ad85cf31b7c3744","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 16:11:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/23/663623/1/check/tripleo-ci-centos-7-containers-multinode/b0aad68/ : SUCCESS in 2h 05m 01s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/23/663623/1/check/tripleo-ci-centos-7-undercloud-containers/819d77b/ : SUCCESS in 1h 32m 41s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/23/663623/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/c14ee10/ : SUCCESS in 2h 06m 52s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/23/663623/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/80e1048/ : SUCCESS in 2h 30m 52s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/23/663623/1/check/tripleo-ci-centos-7-standalone/3cf4da4/ : SUCCESS in 1h 45m 29s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/23/663623/1/check/tripleo-ci-fedora-28-standalone/6f3d608/ : SUCCESS in 1h 54m 59s\n- openstack-tox-pep8 http://logs.openstack.org/23/663623/1/check/openstack-tox-pep8/29f2fde/ : SUCCESS in 5m 19s\n- openstack-tox-py27 http://logs.openstack.org/23/663623/1/check/openstack-tox-py27/7f43418/ : SUCCESS in 4m 51s\n- openstack-tox-py35 http://logs.openstack.org/23/663623/1/check/openstack-tox-py35/31b41c9/ : SUCCESS in 9m 38s\n- openstack-tox-py36 http://logs.openstack.org/23/663623/1/check/openstack-tox-py36/faf8437/ : SUCCESS in 6m 32s\n- openstack-tox-docs http://logs.openstack.org/23/663623/1/check/openstack-tox-docs/9c40801/html/ : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/663623/1/check/openstack-tox-lower-constraints/a8328fd/ : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a6ac5f7b0d6213b7298eded84203d5fb33793a8b","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 09:42:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"33b73e859357ab145ece8c337ec36c5fd5645d47","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 14:08:11.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0556530de070898bf4a66532325d64dc85d57ad4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:08:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a20e0455ff401d6779a9f7ec5e09f3bd284685d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 21:06:40.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc26edbf55f0b98fabe1f93aef81f8378ae4eeff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 01:45:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6790d9749f7b3622284c28c219783e40bc9fd6e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 01:45:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/23/663623/1/gate/tripleo-ci-centos-7-containers-multinode/9c600af/ : SUCCESS in 2h 09m 53s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/23/663623/1/gate/tripleo-ci-centos-7-standalone/09434cc/ : SUCCESS in 1h 48m 21s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/23/663623/1/gate/tripleo-ci-fedora-28-standalone/e40981d/ : SUCCESS in 1h 46m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/23/663623/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/ad5e07c/ : SUCCESS in 2h 16m 15s\n- openstack-tox-pep8 http://logs.openstack.org/23/663623/1/gate/openstack-tox-pep8/cadcb8d/ : SUCCESS in 5m 39s\n- openstack-tox-py27 http://logs.openstack.org/23/663623/1/gate/openstack-tox-py27/2d5b7a4/ : SUCCESS in 5m 46s\n- openstack-tox-py35 http://logs.openstack.org/23/663623/1/gate/openstack-tox-py35/57a9986/ : SUCCESS in 5m 53s\n- openstack-tox-py36 http://logs.openstack.org/23/663623/1/gate/openstack-tox-py36/4b8fc72/ : SUCCESS in 7m 01s\n- openstack-tox-docs http://logs.openstack.org/23/663623/1/gate/openstack-tox-docs/2a7a22c/html/ : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/663623/1/gate/openstack-tox-lower-constraints/0521c30/ : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6ee958a681aaaa269210ac455169c1b0ba0c01d","revisions":{"a6ee958a681aaaa269210ac455169c1b0ba0c01d":{"kind":"REWORK","_number":1,"created":"2019-06-06 12:10:50.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/23/663623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/23/663623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/23/663623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/23/663623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/23/663623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/23/663623/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 12:09:56.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 12:09:56.000000000","tz":120},"subject":"Sync \u0027validations/check-ftype.yaml\u0027 with the new role","message":"Sync \u0027validations/check-ftype.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the check-ftype validation.\n\nChange-Id: Ie4f7e371ba9dc51a1be1473c18ea596e19fa233f\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a6ee958a681aaaa269210ac455169c1b0ba0c01d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/a6ee958a681aaaa269210ac455169c1b0ba0c01d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
