)]}'
{"id":"openstack%2Ftripleo-validations~663662","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I886d26776a9f3e9d9a49f753b8dd434fc8b5b9fe","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I886d26776a9f3e9d9a49f753b8dd434fc8b5b9fe","subject":"Sync \u0027validations/repos.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-06 13:34:21.000000000","updated":"2019-06-07 23:28:54.000000000","submitted":"2019-06-07 23:28:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663662-1559950134596-edb3e276","meta_rev_id":"4a96f7738d820a5bfc6a0d337f689bbcac85f4ec","_number":663662,"virtual_id_number":663662,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-07 23:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 09:39:34.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-06 14:38:52.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-06-07 13:38:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 14:38:52.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 09:39:34.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 13:38:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-07 23:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3754b0249ae3ab10a5153fd22f04b575ec4dc3f8","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 13:34:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48cfb676fa77f00a2e9b89e94188af0e70fa4f1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 21:21:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/663662/1/check/tripleo-ci-centos-7-containers-multinode/90b7141/ : SUCCESS in 2h 12m 46s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/663662/1/check/tripleo-ci-centos-7-undercloud-containers/2a3ed84/ : SUCCESS in 1h 26m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/663662/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/790f50b/ : SUCCESS in 2h 09m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/663662/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/0e7555d/ : SUCCESS in 2h 19m 51s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/663662/1/check/tripleo-ci-centos-7-standalone/b6eda1a/ : SUCCESS in 1h 50m 30s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/663662/1/check/tripleo-ci-fedora-28-standalone/a1b88cb/ : SUCCESS in 2h 03m 22s\n- openstack-tox-pep8 http://logs.openstack.org/62/663662/1/check/openstack-tox-pep8/ae54d52/ : SUCCESS in 7m 19s\n- openstack-tox-py27 http://logs.openstack.org/62/663662/1/check/openstack-tox-py27/a52016f/ : SUCCESS in 5m 09s\n- openstack-tox-py35 http://logs.openstack.org/62/663662/1/check/openstack-tox-py35/8ddf6b9/ : SUCCESS in 7m 02s\n- openstack-tox-py36 http://logs.openstack.org/62/663662/1/check/openstack-tox-py36/974dc40/ : SUCCESS in 5m 12s\n- openstack-tox-docs http://logs.openstack.org/62/663662/1/check/openstack-tox-docs/918c60b/html/ : SUCCESS in 5m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/663662/1/check/openstack-tox-lower-constraints/ca0ed85/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6915916feeb762914058fb4b417c3e3308e5b833","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 09:39:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ac1005e79e4b81f531d6cf06569b80e9870379c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 13:38:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff78349c2e51bc8ba89a055696ba3f489515785b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:38:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54bce95a58454fa4a0ed8928818c5b1d5654131a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 21:06:18.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de4822771bc8f41f62c245dc8be5df1dff3078c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:28:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"799b2b4969918308dc6708bf65a7587048659aa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 23:28:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/663662/1/gate/tripleo-ci-centos-7-containers-multinode/2ab5efe/ : SUCCESS in 1h 55m 04s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/663662/1/gate/tripleo-ci-centos-7-standalone/52fe935/ : SUCCESS in 1h 59m 37s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/663662/1/gate/tripleo-ci-fedora-28-standalone/803db93/ : SUCCESS in 1h 56m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/663662/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/774188d/ : SUCCESS in 2h 14m 55s\n- openstack-tox-pep8 http://logs.openstack.org/62/663662/1/gate/openstack-tox-pep8/057ae49/ : SUCCESS in 8m 18s\n- openstack-tox-py27 http://logs.openstack.org/62/663662/1/gate/openstack-tox-py27/7f49e79/ : SUCCESS in 7m 15s\n- openstack-tox-py35 http://logs.openstack.org/62/663662/1/gate/openstack-tox-py35/4e8e296/ : SUCCESS in 6m 08s\n- openstack-tox-py36 http://logs.openstack.org/62/663662/1/gate/openstack-tox-py36/c05a4cd/ : SUCCESS in 5m 50s\n- openstack-tox-docs http://logs.openstack.org/62/663662/1/gate/openstack-tox-docs/aa3a222/html/ : SUCCESS in 8m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/663662/1/gate/openstack-tox-lower-constraints/33fd641/ : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"00b4857314bac624face5c782248e2114d9ac747","revisions":{"00b4857314bac624face5c782248e2114d9ac747":{"kind":"REWORK","_number":1,"created":"2019-06-06 13:34:21.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/62/663662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/62/663662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663662/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 13:33:32.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 13:33:32.000000000","tz":120},"subject":"Sync \u0027validations/repos.yaml\u0027 with the new role","message":"Sync \u0027validations/repos.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the repos validation.\n\nChange-Id: I886d26776a9f3e9d9a49f753b8dd434fc8b5b9fe\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/00b4857314bac624face5c782248e2114d9ac747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/00b4857314bac624face5c782248e2114d9ac747"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
