)]}'
{"id":"openstack%2Ftripleo-validations~663680","triplet_id":"openstack%2Ftripleo-validations~master~Id6beb76dc756db1bdbddbbbba86ff001fc79aad6","project":"openstack/tripleo-validations","branch":"master","hashtags":[],"change_id":"Id6beb76dc756db1bdbddbbbba86ff001fc79aad6","subject":"Fix typo in the command to get the file_descriptors total_limit","status":"MERGED","created":"2019-06-06 14:06:53.000000000","updated":"2019-06-07 11:02:29.000000000","submitted":"2019-06-07 11:01:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663680-1559905315939-94b57270","meta_rev_id":"7f8cce6b65ca2ac12b1c3c049d155c87d813974d","_number":663680,"virtual_id_number":663680,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-07 11:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":2,"date":"2019-06-06 14:47:50.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-06 14:13:54.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"date":"2019-06-06 14:07:25.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-07 08:34:51.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-06 14:07:25.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-06 14:47:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-07 08:34:51.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 11:01:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16e21e5dd703dbaacbd5cd6c9a87e9dc995b93b5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-06 14:06:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b4d5fc8e5a42d39c8472e58cfa8d8c5463e9c2a","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-06 14:13:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bf642ea6bb5e5f82f5c32e3a85fc93a8c0ac328b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-06 14:47:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cfa8597e13630f02fd102e5843085e7b301e4ef8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-06 22:01:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/80/663680/1/check/tripleo-ci-centos-7-containers-multinode/91d5d5d/ : SUCCESS in 2h 00m 08s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/80/663680/1/check/tripleo-ci-centos-7-undercloud-containers/7759422/ : SUCCESS in 1h 31m 38s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/80/663680/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/6478c8e/ : SUCCESS in 2h 07m 29s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/80/663680/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9354eca/ : SUCCESS in 2h 40m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/80/663680/1/check/tripleo-ci-centos-7-standalone/f5acf8a/ : SUCCESS in 1h 57m 42s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/80/663680/1/check/tripleo-ci-fedora-28-standalone/1f4d78c/ : SUCCESS in 1h 46m 43s\n- tripleo-ci-centos-7-standalone-upgrade http://logs.openstack.org/80/663680/1/check/tripleo-ci-centos-7-standalone-upgrade/6942388/ : SUCCESS in 2h 00m 25s (non-voting)\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/80/663680/1/check/tripleo-ci-centos-7-standalone-upgrade-stein/c4c0792/ : SUCCESS in 2h 01m 16s\n- openstack-tox-pep8 http://logs.openstack.org/80/663680/1/check/openstack-tox-pep8/a4db2bf/ : SUCCESS in 5m 18s\n- openstack-tox-py27 http://logs.openstack.org/80/663680/1/check/openstack-tox-py27/59e5247/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/80/663680/1/check/openstack-tox-py35/5661162/ : SUCCESS in 5m 10s\n- openstack-tox-py36 http://logs.openstack.org/80/663680/1/check/openstack-tox-py36/e738cf8/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/80/663680/1/check/openstack-tox-docs/503ddbb/html/ : SUCCESS in 6m 38s\n- openstack-tox-lower-constraints http://logs.openstack.org/80/663680/1/check/openstack-tox-lower-constraints/21d1ff2/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7ef145b3a1463ae14d08122490258022f8bf91a1","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 08:34:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4bdbf3f46187cd19d8f870b00462fd3ef924acc6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 08:35:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a3c16f27f8a55fcc97054cc01456f89cca4e8c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:01:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/80/663680/1/gate/tripleo-ci-centos-7-containers-multinode/fa7a9e2/ : SUCCESS in 2h 02m 28s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/80/663680/1/gate/tripleo-ci-centos-7-standalone/65552b6/ : SUCCESS in 2h 20m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/80/663680/1/gate/tripleo-ci-fedora-28-standalone/e37556c/ : SUCCESS in 1h 55m 48s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/80/663680/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/97a87c6/ : SUCCESS in 2h 08m 35s\n- tripleo-ci-centos-7-standalone-upgrade-stein http://logs.openstack.org/80/663680/1/gate/tripleo-ci-centos-7-standalone-upgrade-stein/17f325f/ : SUCCESS in 2h 15m 48s\n- openstack-tox-pep8 http://logs.openstack.org/80/663680/1/gate/openstack-tox-pep8/a0cd41d/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/80/663680/1/gate/openstack-tox-py27/797a536/ : SUCCESS in 5m 30s\n- openstack-tox-py35 http://logs.openstack.org/80/663680/1/gate/openstack-tox-py35/1a8c06c/ : SUCCESS in 5m 25s\n- openstack-tox-py36 http://logs.openstack.org/80/663680/1/gate/openstack-tox-py36/a98f8a9/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/80/663680/1/gate/openstack-tox-docs/60c3687/html/ : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/80/663680/1/gate/openstack-tox-lower-constraints/99e7cca/ : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"39683e5100cf08b7f2d6b1627330086953a6861d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:01:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7f8cce6b65ca2ac12b1c3c049d155c87d813974d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:02:29.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 4ab67012910821c27e064134b3debc1ba04dd869","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"39c530ce0e35edfb98fb46bc722ea75feb57b643","revisions":{"39c530ce0e35edfb98fb46bc722ea75feb57b643":{"kind":"REWORK","_number":1,"created":"2019-06-06 14:06:53.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/80/663680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/80/663680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/80/663680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/80/663680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/80/663680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/80/663680/1"}}},"commit":{"parents":[{"commit":"f5e3d1374becd75ff74f62886cfca721e2850456","subject":"Merge \"Sync Sphinx requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f5e3d1374becd75ff74f62886cfca721e2850456"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 14:01:22.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 14:01:22.000000000","tz":120},"subject":"Fix typo in the command to get the file_descriptors total_limit","message":"Fix typo in the command to get the file_descriptors total_limit\n\nThis patch fixes a templating string error due to a typo.\n\n- default(\u0027podman\u0027) instead of default\u0027(podman\u0027)\n\nChange-Id: Id6beb76dc756db1bdbddbbbba86ff001fc79aad6\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/39c530ce0e35edfb98fb46bc722ea75feb57b643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/39c530ce0e35edfb98fb46bc722ea75feb57b643"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
