)]}'
{"id":"openstack%2Ftripleo-validations~663862","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I643552d2fd91b1c986584aa72bcbcb4e9ab7a7bf","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"tls-everywhere-stable/stein","hashtags":[],"change_id":"I643552d2fd91b1c986584aa72bcbcb4e9ab7a7bf","subject":"Fix validations-specific Ansible output callback","status":"MERGED","created":"2019-06-07 08:51:22.000000000","updated":"2019-06-09 19:14:34.000000000","submitted":"2019-06-09 19:14:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663862-1560107674403-7c5f8a7f","meta_rev_id":"27ec0ece94e4064f092656688a908a1512274161","_number":663862,"virtual_id_number":663862,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-09 08:37:16.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-09 19:14:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 12:27:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":2,"date":"2019-06-09 16:18:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 08:59:11.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-06-09 16:18:07.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 08:59:11.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 12:27:12.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-06-09 16:18:07.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-09 19:14:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0186138c9175fac129806b951fe1b499193821b6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 08:51:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1e96c4b6c60a36e510447c9f916fe9858d05f5e5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 08:59:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d59887a9762ed08bb7931dde678857a939e7a8fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 11:20:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-containers-multinode/37bc724/ : SUCCESS in 1h 59m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-undercloud-containers/2d6d50a/ : SUCCESS in 1h 35m 53s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/15aee04/ : SUCCESS in 2h 11m 57s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/23612a5/ : SUCCESS in 2h 21m 35s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-standalone/58a1aa6/ : FAILURE in 1h 46m 00s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/663862/1/check/tripleo-ci-fedora-28-standalone/00217bd/ : SUCCESS in 1h 35m 41s\n- openstack-tox-pep8 http://logs.openstack.org/62/663862/1/check/openstack-tox-pep8/b3aa1cd/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/62/663862/1/check/openstack-tox-py27/d3bec0e/ : SUCCESS in 4m 58s\n- openstack-tox-py35 http://logs.openstack.org/62/663862/1/check/openstack-tox-py35/ad872a9/ : SUCCESS in 5m 02s\n- openstack-tox-py36 http://logs.openstack.org/62/663862/1/check/openstack-tox-py36/2c5292a/ : SUCCESS in 5m 01s\n- openstack-tox-docs http://logs.openstack.org/62/663862/1/check/openstack-tox-docs/b1a7342/html/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/663862/1/check/openstack-tox-lower-constraints/64c4a32/ : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"9893d53f8bae30681ca0371a4982ec907bcbc872","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 12:50:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2094df403459dfe86f22db993d492207e8c3a1ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 02:55:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-containers-multinode/ce26a0a/ : SUCCESS in 2h 03m 12s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-undercloud-containers/21dba22/ : SUCCESS in 1h 33m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/93090c2/ : SUCCESS in 2h 10m 42s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/6c46285/ : SUCCESS in 2h 18m 44s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-standalone/9547bf0/ : SUCCESS in 2h 02m 54s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/663862/1/check/tripleo-ci-fedora-28-standalone/6c8a5be/ : FAILURE in 2h 23m 16s\n- openstack-tox-pep8 http://logs.openstack.org/62/663862/1/check/openstack-tox-pep8/dcb9f9f/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/62/663862/1/check/openstack-tox-py27/07f4105/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/62/663862/1/check/openstack-tox-py35/388e9db/ : SUCCESS in 5m 53s\n- openstack-tox-py36 http://logs.openstack.org/62/663862/1/check/openstack-tox-py36/700e1cb/ : SUCCESS in 5m 49s\n- openstack-tox-docs http://logs.openstack.org/62/663862/1/check/openstack-tox-docs/50da1cd/html/ : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/663862/1/check/openstack-tox-lower-constraints/9f6c057/ : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"d5c052da6d8f72bbf5ca67847f2f06d5aaaa83d4","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:37:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db9e9fbf0f0728cd6d7696ae57435a57ac26ef51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 11:00:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-containers-multinode/c9e00b6/ : SUCCESS in 1h 57m 04s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-undercloud-containers/ebd3101/ : SUCCESS in 1h 38m 06s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/a38124b/ : SUCCESS in 2h 01m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/761105a/ : SUCCESS in 2h 17m 59s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/663862/1/check/tripleo-ci-centos-7-standalone/3f29664/ : SUCCESS in 1h 44m 51s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/663862/1/check/tripleo-ci-fedora-28-standalone/69a28d7/ : SUCCESS in 1h 45m 01s\n- openstack-tox-pep8 http://logs.openstack.org/62/663862/1/check/openstack-tox-pep8/108cb8e/ : SUCCESS in 6m 01s\n- openstack-tox-py27 http://logs.openstack.org/62/663862/1/check/openstack-tox-py27/ca9105f/ : SUCCESS in 5m 15s\n- openstack-tox-py35 http://logs.openstack.org/62/663862/1/check/openstack-tox-py35/5ed66f5/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/62/663862/1/check/openstack-tox-py36/0b14044/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/62/663862/1/check/openstack-tox-docs/6b87b74/html/ : SUCCESS in 4m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/663862/1/check/openstack-tox-lower-constraints/5bbd06f/ : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"4a7c68ed733fd1bbd5aaa1ffe74e46e848267b53","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-06-09 12:27:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3accfb90b1045e4f647c2e76949595f2214957a5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:18:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae398d28a5ee9730409c033371decf31e6dc2c40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:18:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"88d2c210e5ef8a12ccef9ea6ab14f71707cd0b51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 19:14:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cfd5ac1a916dacbbf4d29ad5695ad40bc13e55cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 19:14:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/62/663862/1/gate/tripleo-ci-centos-7-containers-multinode/0845b17/ : SUCCESS in 2h 27m 51s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/62/663862/1/gate/tripleo-ci-centos-7-standalone/be31608/ : SUCCESS in 1h 50m 23s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/62/663862/1/gate/tripleo-ci-fedora-28-standalone/1ec1547/ : SUCCESS in 1h 55m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/62/663862/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0a78c67/ : SUCCESS in 2h 27m 49s\n- openstack-tox-pep8 http://logs.openstack.org/62/663862/1/gate/openstack-tox-pep8/ca9b5e3/ : SUCCESS in 5m 28s\n- openstack-tox-py27 http://logs.openstack.org/62/663862/1/gate/openstack-tox-py27/0a71e91/ : SUCCESS in 6m 31s\n- openstack-tox-py35 http://logs.openstack.org/62/663862/1/gate/openstack-tox-py35/dcbed3f/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/62/663862/1/gate/openstack-tox-py36/008f062/ : SUCCESS in 7m 19s\n- openstack-tox-docs http://logs.openstack.org/62/663862/1/gate/openstack-tox-docs/ef1cd7f/html/ : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/62/663862/1/gate/openstack-tox-lower-constraints/278361e/ : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ee7475e34a01cec8bc0c25c7939bffb0241dc3c","revisions":{"8ee7475e34a01cec8bc0c25c7939bffb0241dc3c":{"kind":"REWORK","_number":1,"created":"2019-06-07 08:51:22.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/62/663862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/62/663862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/62/663862/1"}}},"commit":{"parents":[{"commit":"9d3d67647a8e93c495bcdcbdbaaa77fa64db069b","subject":"New validation: ensure image-serve service is working as expected","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9d3d67647a8e93c495bcdcbdbaaa77fa64db069b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-05-24 13:22:18.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 08:51:22.000000000","tz":0},"subject":"Fix validations-specific Ansible output callback","message":"Fix validations-specific Ansible output callback\n\nSince newer Ansible release, the tasks results data structure have\nsomewhat changed.\n\nThis patch fixes the way we output debug, warning and failure messages\nwhen running the validations. It also brings some colors for this\ndifferent kind of message.\n\nChange-Id: I643552d2fd91b1c986584aa72bcbcb4e9ab7a7bf\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit c18eadb78ab4085c199f5f80ea55f0aedb633001)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8ee7475e34a01cec8bc0c25c7939bffb0241dc3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/8ee7475e34a01cec8bc0c25c7939bffb0241dc3c"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
