)]}'
{"id":"openstack%2Ftripleo-validations~663870","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I21952cbc8788fba07fdb30cf6c0f130de1d1f966","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I21952cbc8788fba07fdb30cf6c0f130de1d1f966","subject":"Sync \u0027validations/rabbitmq-limits.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 09:40:07.000000000","updated":"2019-06-11 03:24:13.000000000","submitted":"2019-06-11 03:24:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663870-1560223453864-d9eb55b5","meta_rev_id":"c612200500a5002044ec2b1b0c5f8951ef408631","_number":663870,"virtual_id_number":663870,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-11 03:24:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"recommended":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-09 23:14:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":2,"date":"2019-06-07 10:49:56.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 09:57:30.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-06-10 13:15:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 09:57:30.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 10:49:56.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 23:14:15.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-06-10 13:15:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-11 03:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e3f6c9fba03a71b37789401642c1fbc45cccf78","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 09:40:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4942bb2d847ee44c1588ffafed6ea8f1f73e124c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 10:49:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"427bed771f3d2c3e3c83edc36250c86d08b52b04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 13:26:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-containers-multinode/751d6d4/ : SUCCESS in 1h 51m 55s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-undercloud-containers/d1c9a2f/ : SUCCESS in 1h 31m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/592817a/ : SUCCESS in 2h 10m 46s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/9e1b6e8/ : SUCCESS in 2h 41m 19s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-standalone/9aabe8d/ : FAILURE in 1h 44m 32s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/663870/1/check/tripleo-ci-fedora-28-standalone/d1bdfc9/ : SUCCESS in 1h 56m 56s\n- openstack-tox-pep8 http://logs.openstack.org/70/663870/1/check/openstack-tox-pep8/b08de35/ : SUCCESS in 8m 53s\n- openstack-tox-py27 http://logs.openstack.org/70/663870/1/check/openstack-tox-py27/49c77be/ : SUCCESS in 4m 46s\n- openstack-tox-py35 http://logs.openstack.org/70/663870/1/check/openstack-tox-py35/f5ad3f1/ : SUCCESS in 5m 17s\n- openstack-tox-py36 http://logs.openstack.org/70/663870/1/check/openstack-tox-py36/2dbf8f7/ : SUCCESS in 5m 00s\n- openstack-tox-docs http://logs.openstack.org/70/663870/1/check/openstack-tox-docs/c8e8bdb/html/ : SUCCESS in 4m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/663870/1/check/openstack-tox-lower-constraints/097660d/ : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"fcbffd4d72a996b3a731d638c53c09a863566cc1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:18:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"40c60aa6d800975c70f49090f8a27a8e071b16e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 18:38:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-containers-multinode/2342f25/ : SUCCESS in 2h 02m 51s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-undercloud-containers/fa67efd/ : SUCCESS in 1h 22m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f88a3a0/ : SUCCESS in 2h 07m 13s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/2b12888/ : SUCCESS in 2h 12m 15s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/663870/1/check/tripleo-ci-centos-7-standalone/eacb9ca/ : SUCCESS in 1h 37m 24s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/663870/1/check/tripleo-ci-fedora-28-standalone/cc5fd2d/ : SUCCESS in 1h 43m 12s\n- openstack-tox-pep8 http://logs.openstack.org/70/663870/1/check/openstack-tox-pep8/9c663cb/ : SUCCESS in 6m 05s\n- openstack-tox-py27 http://logs.openstack.org/70/663870/1/check/openstack-tox-py27/4dba8c1/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/70/663870/1/check/openstack-tox-py35/1da9b63/ : SUCCESS in 5m 57s\n- openstack-tox-py36 http://logs.openstack.org/70/663870/1/check/openstack-tox-py36/90b18b1/ : SUCCESS in 5m 02s\n- openstack-tox-docs http://logs.openstack.org/70/663870/1/check/openstack-tox-docs/8b5717c/html/ : SUCCESS in 4m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/663870/1/check/openstack-tox-lower-constraints/782ca0d/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a3b9f7f471c5b9807698d69b6deb86ae54e0c58f","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-09 23:14:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"595644225c6a9f8c16e7adf871ba2c98ff2a223f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:15:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"448920c9f1c98b185c6bad4e1e43b026e5937652","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:15:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"399e0b420c0b62d79f15a0e20131a0126d20bc7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 03:24:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/70/663870/1/gate/tripleo-ci-centos-7-containers-multinode/5ce880d/ : SUCCESS in 2h 16m 27s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/70/663870/1/gate/tripleo-ci-centos-7-standalone/ae8875f/ : SUCCESS in 1h 55m 05s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/70/663870/1/gate/tripleo-ci-fedora-28-standalone/5d0a289/ : SUCCESS in 1h 44m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/70/663870/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/1a61b41/ : SUCCESS in 2h 30m 50s\n- openstack-tox-pep8 http://logs.openstack.org/70/663870/1/gate/openstack-tox-pep8/b607c37/ : SUCCESS in 5m 34s\n- openstack-tox-py27 http://logs.openstack.org/70/663870/1/gate/openstack-tox-py27/b93264e/ : SUCCESS in 5m 29s\n- openstack-tox-py35 http://logs.openstack.org/70/663870/1/gate/openstack-tox-py35/6e7d005/ : SUCCESS in 5m 45s\n- openstack-tox-py36 http://logs.openstack.org/70/663870/1/gate/openstack-tox-py36/1f70c41/ : SUCCESS in 6m 06s\n- openstack-tox-docs http://logs.openstack.org/70/663870/1/gate/openstack-tox-docs/a9e004d/html/ : SUCCESS in 6m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/70/663870/1/gate/openstack-tox-lower-constraints/04ba4bf/ : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":1},{"id":"f938b3577eb5d79aa5f17fa63213158a2d981722","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 03:24:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"396e54a1c4c95f2772ea452602d684cb79b54480","revisions":{"396e54a1c4c95f2772ea452602d684cb79b54480":{"kind":"REWORK","_number":1,"created":"2019-06-07 09:40:07.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/70/663870/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/70/663870/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/70/663870/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/70/663870/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/70/663870/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/70/663870/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 09:38:36.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 09:38:36.000000000","tz":120},"subject":"Sync \u0027validations/rabbitmq-limits.yaml\u0027 with the new role","message":"Sync \u0027validations/rabbitmq-limits.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the rabbitmq-limits validation.\n\nChange-Id: I21952cbc8788fba07fdb30cf6c0f130de1d1f966\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/396e54a1c4c95f2772ea452602d684cb79b54480"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/396e54a1c4c95f2772ea452602d684cb79b54480"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
