)]}'
{"id":"openstack%2Ftripleo-validations~663881","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~If2cc183ee55f8385475c706eb2ceb6178cf2e00f","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"If2cc183ee55f8385475c706eb2ceb6178cf2e00f","subject":"Sync \u0027validations/nova-event-callback.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 10:14:20.000000000","updated":"2019-06-10 17:06:57.000000000","submitted":"2019-06-10 17:06:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663881-1560186417536-f0d02f45","meta_rev_id":"53b9ad5229f9a682638364bf47cdefc33f78c34f","_number":663881,"virtual_id_number":663881,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-10 09:55:52.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 17:06:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-07 14:17:33.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:12:16.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:18:22.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:12:16.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:12:16.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 12:12:16.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:18:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 17:06:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d88a3d40f8e4f098378f4e4f254ee54dd084e260","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 10:14:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11435a9a1edb97da9013ccc14f49fededbd57133","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 14:09:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-containers-multinode/307d00c/ : SUCCESS in 1h 51m 31s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-undercloud-containers/3511d1e/ : SUCCESS in 1h 35m 40s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3965158/ : SUCCESS in 2h 09m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/c3ad577/ : SUCCESS in 2h 21m 27s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-standalone/bd9b830/ : SUCCESS in 1h 50m 25s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/81/663881/1/check/tripleo-ci-fedora-28-standalone/fbcd58c/ : SUCCESS in 1h 38m 34s\n- openstack-tox-pep8 http://logs.openstack.org/81/663881/1/check/openstack-tox-pep8/1c47207/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/81/663881/1/check/openstack-tox-py27/ffea703/ : SUCCESS in 5m 06s\n- openstack-tox-py35 http://logs.openstack.org/81/663881/1/check/openstack-tox-py35/af69b06/ : SUCCESS in 6m 14s\n- openstack-tox-py36 http://logs.openstack.org/81/663881/1/check/openstack-tox-py36/f6f381a/ : SUCCESS in 5m 18s\n- openstack-tox-docs http://logs.openstack.org/81/663881/1/check/openstack-tox-docs/2902c9e/html/ : SUCCESS in 5m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/663881/1/check/openstack-tox-lower-constraints/3a93a1f/ : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e29f64e7f05ccefb5ede039db8a91bb6f54bd8b7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 14:17:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5eef2a716d6775e8211aed1edc856f7d45efea78","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:18:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96e9a4085ffea315e36e8dba499ba389320cc998","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:18:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"451a9403ad0d0b5d36c9b2152de12d2af23c2f54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 19:55:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/81/663881/1/gate/tripleo-ci-centos-7-containers-multinode/30e19f3/ : SUCCESS in 2h 20m 29s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/81/663881/1/gate/tripleo-ci-centos-7-standalone/f055952/ : SUCCESS in 1h 53m 41s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/81/663881/1/gate/tripleo-ci-fedora-28-standalone/4d310ee/ : TIMED_OUT in 3h 04m 36s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/81/663881/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/338bf65/ : SUCCESS in 2h 10m 29s\n- openstack-tox-pep8 http://logs.openstack.org/81/663881/1/gate/openstack-tox-pep8/27222cb/ : SUCCESS in 7m 24s\n- openstack-tox-py27 http://logs.openstack.org/81/663881/1/gate/openstack-tox-py27/c3b9e50/ : SUCCESS in 6m 12s\n- openstack-tox-py35 http://logs.openstack.org/81/663881/1/gate/openstack-tox-py35/cfe97fc/ : SUCCESS in 5m 18s\n- openstack-tox-py36 http://logs.openstack.org/81/663881/1/gate/openstack-tox-py36/8ebfc1b/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/81/663881/1/gate/openstack-tox-docs/8441a44/html/ : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/663881/1/gate/openstack-tox-lower-constraints/13954f8/ : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a661a3fd1108043625663bbae1cfefa149afa48d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-10 09:55:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9146038976b612d72e94b27e6d7154bd0d04792","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:22:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-containers-multinode/40dab08/ : SUCCESS in 2h 01m 33s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-undercloud-containers/f27a6d6/ : SUCCESS in 1h 36m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/cdf5ef3/ : SUCCESS in 2h 10m 50s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/740d2d7/ : SUCCESS in 2h 21m 11s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/81/663881/1/check/tripleo-ci-centos-7-standalone/93e71ef/ : SUCCESS in 1h 45m 31s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/81/663881/1/check/tripleo-ci-fedora-28-standalone/0c7a335/ : SUCCESS in 2h 04m 45s\n- openstack-tox-pep8 http://logs.openstack.org/81/663881/1/check/openstack-tox-pep8/0149fa8/ : SUCCESS in 6m 47s\n- openstack-tox-py27 http://logs.openstack.org/81/663881/1/check/openstack-tox-py27/5117ce4/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/81/663881/1/check/openstack-tox-py35/da001a2/ : SUCCESS in 5m 12s\n- openstack-tox-py36 http://logs.openstack.org/81/663881/1/check/openstack-tox-py36/a618ba5/ : SUCCESS in 4m 59s\n- openstack-tox-docs http://logs.openstack.org/81/663881/1/check/openstack-tox-docs/0201b85/html/ : SUCCESS in 5m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/663881/1/check/openstack-tox-lower-constraints/950c3e6/ : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":1},{"id":"324571e8ae81614e19a58900d46c1796a7cc4fdc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 12:22:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76a3ec67b6758755d667decb630980ec1966c90d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:06:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/81/663881/1/gate/tripleo-ci-centos-7-containers-multinode/1b157c7/ : SUCCESS in 1h 55m 48s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/81/663881/1/gate/tripleo-ci-centos-7-standalone/8fcca9c/ : SUCCESS in 2h 14m 48s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/81/663881/1/gate/tripleo-ci-fedora-28-standalone/e4cdd32/ : SUCCESS in 1h 53m 31s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/81/663881/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/fd24862/ : SUCCESS in 2h 09m 01s\n- openstack-tox-pep8 http://logs.openstack.org/81/663881/1/gate/openstack-tox-pep8/7b23c3b/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/81/663881/1/gate/openstack-tox-py27/366f431/ : SUCCESS in 7m 06s\n- openstack-tox-py35 http://logs.openstack.org/81/663881/1/gate/openstack-tox-py35/4e98450/ : SUCCESS in 5m 59s\n- openstack-tox-py36 http://logs.openstack.org/81/663881/1/gate/openstack-tox-py36/91b163f/ : SUCCESS in 5m 47s\n- openstack-tox-docs http://logs.openstack.org/81/663881/1/gate/openstack-tox-docs/745a57f/html/ : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/81/663881/1/gate/openstack-tox-lower-constraints/dff0895/ : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c17aab7451692197cf8ea0e74c097736e8e51838","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:06:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f0edec74d9ecf461f05b5da806722e919f9fc00","revisions":{"4f0edec74d9ecf461f05b5da806722e919f9fc00":{"kind":"REWORK","_number":1,"created":"2019-06-07 10:14:20.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/81/663881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/81/663881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/81/663881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/81/663881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/81/663881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/81/663881/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 10:10:55.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 10:10:55.000000000","tz":120},"subject":"Sync \u0027validations/nova-event-callback.yaml\u0027 with the new role","message":"Sync \u0027validations/nova-event-callback.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the nova-event-callback validation.\n\nChange-Id: If2cc183ee55f8385475c706eb2ceb6178cf2e00f\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4f0edec74d9ecf461f05b5da806722e919f9fc00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4f0edec74d9ecf461f05b5da806722e919f9fc00"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
