)]}'
{"id":"openstack%2Ftripleo-validations~663890","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~I9142c1eea46f85aa36c49be41131785dc5343b95","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"I9142c1eea46f85aa36c49be41131785dc5343b95","subject":"Sync \u0027validations/mysql-open-files-limit.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 10:42:47.000000000","updated":"2019-06-09 22:43:38.000000000","submitted":"2019-06-09 22:43:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663890-1560120218622-f0d02f45","meta_rev_id":"eb18526908d36bd0bf9b3a48c505719c31675269","_number":663890,"virtual_id_number":663890,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-09 22:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":2,"date":"2019-06-09 16:18:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-07 10:51:05.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:12:08.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":1,"date":"2019-06-09 16:18:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 10:51:05.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 12:12:08.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-09 16:18:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-09 22:43:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"101355ad6da10f108d68532099195cffa2d994b5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 10:42:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f76d159eee270fb10a6939239501b575172ce5b4","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2019-06-07 10:51:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"318213f4532e9daac73c6fe98a75211bb321de74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:09:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/90/663890/1/check/tripleo-ci-centos-7-containers-multinode/2b4115b/ : SUCCESS in 1h 58m 17s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/90/663890/1/check/tripleo-ci-centos-7-undercloud-containers/f999785/ : SUCCESS in 1h 35m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/90/663890/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0526986/ : SUCCESS in 2h 25m 58s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/90/663890/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/1bac999/ : SUCCESS in 2h 24m 04s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/90/663890/1/check/tripleo-ci-centos-7-standalone/988a505/ : SUCCESS in 2h 01m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/90/663890/1/check/tripleo-ci-fedora-28-standalone/64531b7/ : SUCCESS in 2h 09m 19s\n- openstack-tox-pep8 http://logs.openstack.org/90/663890/1/check/openstack-tox-pep8/5d5214e/ : SUCCESS in 5m 49s\n- openstack-tox-py27 http://logs.openstack.org/90/663890/1/check/openstack-tox-py27/8541b46/ : SUCCESS in 5m 12s\n- openstack-tox-py35 http://logs.openstack.org/90/663890/1/check/openstack-tox-py35/3b7954a/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/90/663890/1/check/openstack-tox-py36/58f435f/ : SUCCESS in 6m 01s\n- openstack-tox-docs http://logs.openstack.org/90/663890/1/check/openstack-tox-docs/e1611a5/html/ : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/663890/1/check/openstack-tox-lower-constraints/49a9207/ : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e74c17b7a370508abf8580f69e0dcc81f8aa862d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:18:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5756285580887c135563eded5bc6a68f53aec6c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:18:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31aeafaf49eef9841b3776966039306beb48b2c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:43:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6a83af76a96196b1743913ef15e429317036eb85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:43:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/90/663890/1/gate/tripleo-ci-centos-7-containers-multinode/a8673e4/ : SUCCESS in 1h 57m 56s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/90/663890/1/gate/tripleo-ci-centos-7-standalone/658ce0f/ : SUCCESS in 1h 53m 15s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/90/663890/1/gate/tripleo-ci-fedora-28-standalone/fbf7751/ : SUCCESS in 1h 44m 25s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/90/663890/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/9225245/ : SUCCESS in 1h 53m 56s\n- openstack-tox-pep8 http://logs.openstack.org/90/663890/1/gate/openstack-tox-pep8/04ec4a0/ : SUCCESS in 9m 44s\n- openstack-tox-py27 http://logs.openstack.org/90/663890/1/gate/openstack-tox-py27/dc0c9b6/ : SUCCESS in 4m 57s\n- openstack-tox-py35 http://logs.openstack.org/90/663890/1/gate/openstack-tox-py35/b951248/ : SUCCESS in 5m 44s\n- openstack-tox-py36 http://logs.openstack.org/90/663890/1/gate/openstack-tox-py36/f4eb37a/ : SUCCESS in 5m 23s\n- openstack-tox-docs http://logs.openstack.org/90/663890/1/gate/openstack-tox-docs/db85b1b/html/ : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/90/663890/1/gate/openstack-tox-lower-constraints/42b922b/ : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"619e4d7e069182e454f6f2f5bb589a8a85c33759","revisions":{"619e4d7e069182e454f6f2f5bb589a8a85c33759":{"kind":"REWORK","_number":1,"created":"2019-06-07 10:42:47.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/90/663890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/90/663890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/90/663890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/90/663890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/90/663890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/90/663890/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 10:41:44.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 10:41:44.000000000","tz":120},"subject":"Sync \u0027validations/mysql-open-files-limit.yaml\u0027 with the new role","message":"Sync \u0027validations/mysql-open-files-limit.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the mysql-open-files-limit validation.\n\nChange-Id: I9142c1eea46f85aa36c49be41131785dc5343b95\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/619e4d7e069182e454f6f2f5bb589a8a85c33759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/619e4d7e069182e454f6f2f5bb589a8a85c33759"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
