)]}'
{"id":"openstack%2Ftripleo-validations~663895","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Id6beb76dc756db1bdbddbbbba86ff001fc79aad6","project":"openstack/tripleo-validations","branch":"stable/stein","hashtags":[],"change_id":"Id6beb76dc756db1bdbddbbbba86ff001fc79aad6","subject":"Fix typo in the command to get the file_descriptors total_limit","status":"MERGED","created":"2019-06-07 11:02:29.000000000","updated":"2019-06-11 10:04:33.000000000","submitted":"2019-06-11 10:04:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663895-1560247473542-b5d80b2d","meta_rev_id":"eb95e46d03228b4657c6dc51702d162a48162250","_number":663895,"virtual_id_number":663895,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-09 08:39:39.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-11 10:04:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-07 13:51:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-09 23:13:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 11:02:45.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-10 13:14:46.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"date":"2019-06-07 11:02:43.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 11:02:43.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-07 11:02:45.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-09 23:13:05.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-06-10 13:14:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-11 10:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6557beadad395bf9d00f1d7ecfe14072752c59a","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:02:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d288110f18786a864ebe56716009c940906fc4cd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-07 13:51:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eb5df80b28bdfbcbd8c020d01caff99bdb1e207c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 17:42:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-containers-multinode/0483e32/ : SUCCESS in 1h 51m 03s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-undercloud-containers/b16c4f9/ : SUCCESS in 1h 30m 39s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/247957e/ : SUCCESS in 2h 00m 35s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/a4e90b5/ : SUCCESS in 2h 39m 30s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-standalone/6192508/ : FAILURE in 1h 52m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-fedora-28-standalone/523920d/ : SUCCESS in 1h 51m 48s\n- openstack-tox-pep8 http://logs.openstack.org/95/663895/1/check/openstack-tox-pep8/4a577c2/ : SUCCESS in 6m 27s\n- openstack-tox-py27 http://logs.openstack.org/95/663895/1/check/openstack-tox-py27/5000ed9/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/95/663895/1/check/openstack-tox-py35/1a48a98/ : SUCCESS in 6m 40s\n- openstack-tox-py36 http://logs.openstack.org/95/663895/1/check/openstack-tox-py36/eb21728/ : SUCCESS in 7m 02s\n- openstack-tox-docs http://logs.openstack.org/95/663895/1/check/openstack-tox-docs/32692f7/html/ : SUCCESS in 4m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/663895/1/check/openstack-tox-lower-constraints/2b6dac0/ : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"40bd05c18e752021f98724d0bac6eadc8f34eac6","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:39:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7444d0b89810b042b825eca31ebfb511d7e0790e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 11:28:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-containers-multinode/39c7745/ : SUCCESS in 1h 49m 37s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-undercloud-containers/bb19363/ : SUCCESS in 1h 19m 08s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/0e4ee49/ : SUCCESS in 2h 15m 45s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/b355ce6/ : SUCCESS in 2h 12m 13s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-standalone/7fb5664/ : FAILURE in 1h 53m 47s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-fedora-28-standalone/9a07707/ : SUCCESS in 1h 43m 35s\n- openstack-tox-pep8 http://logs.openstack.org/95/663895/1/check/openstack-tox-pep8/98e3b21/ : SUCCESS in 9m 12s\n- openstack-tox-py27 http://logs.openstack.org/95/663895/1/check/openstack-tox-py27/2d86720/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/95/663895/1/check/openstack-tox-py35/976e131/ : SUCCESS in 5m 35s\n- openstack-tox-py36 http://logs.openstack.org/95/663895/1/check/openstack-tox-py36/95cc418/ : SUCCESS in 5m 26s\n- openstack-tox-docs http://logs.openstack.org/95/663895/1/check/openstack-tox-docs/785deae/html/ : SUCCESS in 4m 53s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/663895/1/check/openstack-tox-lower-constraints/5fbf824/ : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9059187a353f287473f96bbd621982f8ceba1ebd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:18:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3847c16aaff1cde91e40af90f5e28bb0937146a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 18:52:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-containers-multinode/364a397/ : SUCCESS in 1h 54m 32s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-undercloud-containers/2688b2c/ : SUCCESS in 1h 50m 07s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/3614c2b/ : SUCCESS in 2h 10m 28s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/49d88af/ : SUCCESS in 2h 18m 05s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-standalone/b2231f9/ : SUCCESS in 1h 59m 24s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-fedora-28-standalone/9022e52/ : SUCCESS in 1h 54m 05s\n- openstack-tox-pep8 http://logs.openstack.org/95/663895/1/check/openstack-tox-pep8/18c6577/ : SUCCESS in 5m 29s\n- openstack-tox-py27 http://logs.openstack.org/95/663895/1/check/openstack-tox-py27/12365f6/ : SUCCESS in 5m 08s\n- openstack-tox-py35 http://logs.openstack.org/95/663895/1/check/openstack-tox-py35/7571be0/ : SUCCESS in 6m 33s\n- openstack-tox-py36 http://logs.openstack.org/95/663895/1/check/openstack-tox-py36/c654233/ : SUCCESS in 5m 06s\n- openstack-tox-docs http://logs.openstack.org/95/663895/1/check/openstack-tox-docs/8f0bb3c/html/ : SUCCESS in 4m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/663895/1/check/openstack-tox-lower-constraints/d7b16fb/ : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"d24f5e5cb545a423216c39a24d9af93c271bfe9b","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-09 23:13:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"139f377a46d6142512e043c2f90dc91585c89436","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:14:46.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb4adea5bcc15a2bc22b9812136da28eebf503a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:15:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72c427c2b1b12130330e8cd8c12f926ec1ededf9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 22:14:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/663895/1/gate/tripleo-ci-centos-7-containers-multinode/01913f6/ : FAILURE in 2h 09m 24s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/663895/1/gate/tripleo-ci-centos-7-standalone/2a9ee22/ : SUCCESS in 1h 53m 27s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/663895/1/gate/tripleo-ci-fedora-28-standalone/7f12765/ : SUCCESS in 1h 54m 10s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/663895/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/f5fff74/ : SUCCESS in 2h 28m 12s\n- openstack-tox-pep8 http://logs.openstack.org/95/663895/1/gate/openstack-tox-pep8/f4d7e95/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/95/663895/1/gate/openstack-tox-py27/955fc3c/ : SUCCESS in 5m 50s\n- openstack-tox-py35 http://logs.openstack.org/95/663895/1/gate/openstack-tox-py35/952c867/ : SUCCESS in 5m 23s\n- openstack-tox-py36 http://logs.openstack.org/95/663895/1/gate/openstack-tox-py36/9d591ee/ : SUCCESS in 4m 56s\n- openstack-tox-docs http://logs.openstack.org/95/663895/1/gate/openstack-tox-docs/7a4b7bc/html/ : SUCCESS in 5m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/663895/1/gate/openstack-tox-lower-constraints/4fb9fee/ : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"37b60a07b5368e3b7750c12b70ff97d588b9fa79","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 22:18:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8b49d023c49b47af7d61edf6e0e6e52837a5f0e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 03:46:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-containers-multinode/73703ed/ : SUCCESS in 2h 08m 28s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-undercloud-containers/0551034/ : SUCCESS in 1h 47m 19s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/20fd0ee/ : SUCCESS in 2h 20m 09s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/138b333/ : SUCCESS in 2h 41m 33s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-centos-7-standalone/3b49c2e/ : SUCCESS in 1h 51m 39s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/663895/1/check/tripleo-ci-fedora-28-standalone/03d0b75/ : SUCCESS in 1h 55m 55s\n- openstack-tox-pep8 http://logs.openstack.org/95/663895/1/check/openstack-tox-pep8/ff09dfe/ : SUCCESS in 7m 12s\n- openstack-tox-py27 http://logs.openstack.org/95/663895/1/check/openstack-tox-py27/30e9062/ : SUCCESS in 7m 14s\n- openstack-tox-py35 http://logs.openstack.org/95/663895/1/check/openstack-tox-py35/a267179/ : SUCCESS in 6m 32s\n- openstack-tox-py36 http://logs.openstack.org/95/663895/1/check/openstack-tox-py36/c0a4ba1/ : SUCCESS in 7m 09s\n- openstack-tox-docs http://logs.openstack.org/95/663895/1/check/openstack-tox-docs/a3eb6e6/html/ : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/663895/1/check/openstack-tox-lower-constraints/25fc44d/ : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":1},{"id":"61a670c65d9b01987edb0ec6415efad5bf02fdc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 05:56:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcb7368898719fbfd4d97688c4fa82ea0bd77210","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:04:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/95/663895/1/gate/tripleo-ci-centos-7-containers-multinode/e7610f5/ : SUCCESS in 2h 17m 42s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/95/663895/1/gate/tripleo-ci-centos-7-standalone/3991611/ : SUCCESS in 1h 44m 45s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/95/663895/1/gate/tripleo-ci-fedora-28-standalone/867dd26/ : SUCCESS in 2h 01m 27s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/95/663895/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/736d8d8/ : SUCCESS in 2h 23m 17s\n- openstack-tox-pep8 http://logs.openstack.org/95/663895/1/gate/openstack-tox-pep8/bb51741/ : SUCCESS in 6m 26s\n- openstack-tox-py27 http://logs.openstack.org/95/663895/1/gate/openstack-tox-py27/7fb6152/ : SUCCESS in 5m 42s\n- openstack-tox-py35 http://logs.openstack.org/95/663895/1/gate/openstack-tox-py35/237871d/ : SUCCESS in 5m 32s\n- openstack-tox-py36 http://logs.openstack.org/95/663895/1/gate/openstack-tox-py36/6596034/ : SUCCESS in 5m 53s\n- openstack-tox-docs http://logs.openstack.org/95/663895/1/gate/openstack-tox-docs/0c4d3a0/html/ : SUCCESS in 6m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/95/663895/1/gate/openstack-tox-lower-constraints/1c7a81b/ : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"eb95e46d03228b4657c6dc51702d162a48162250","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 10:04:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ab67012910821c27e064134b3debc1ba04dd869","revisions":{"4ab67012910821c27e064134b3debc1ba04dd869":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:02:29.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/95/663895/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/95/663895/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/95/663895/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/95/663895/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/95/663895/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/95/663895/1"}}},"commit":{"parents":[{"commit":"9d3d67647a8e93c495bcdcbdbaaa77fa64db069b","subject":"New validation: ensure image-serve service is working as expected","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/9d3d67647a8e93c495bcdcbdbaaa77fa64db069b"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-06 14:01:22.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:02:29.000000000","tz":0},"subject":"Fix typo in the command to get the file_descriptors total_limit","message":"Fix typo in the command to get the file_descriptors total_limit\n\nThis patch fixes a templating string error due to a typo.\n\n- default(\u0027podman\u0027) instead of default\u0027(podman\u0027)\n\nChange-Id: Id6beb76dc756db1bdbddbbbba86ff001fc79aad6\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 39c530ce0e35edfb98fb46bc722ea75feb57b643)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4ab67012910821c27e064134b3debc1ba04dd869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4ab67012910821c27e064134b3debc1ba04dd869"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
