)]}'
{"id":"openstack%2Ftripleo-validations~663896","triplet_id":"openstack%2Ftripleo-validations~stable%2Fstein~Ib452bbf1142ea59eb34ed043f463da48a21081df","project":"openstack/tripleo-validations","branch":"stable/stein","topic":"stable/stein/backports_validations_dir","hashtags":[],"change_id":"Ib452bbf1142ea59eb34ed043f463da48a21081df","subject":"Sync \u0027validations/no-op-firewall-nova-driver.yaml\u0027 with the new role","status":"MERGED","created":"2019-06-07 11:09:08.000000000","updated":"2019-06-09 22:43:39.000000000","submitted":"2019-06-09 22:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663896-1560120219286-edb3e276","meta_rev_id":"975f6808574fc464d06141a47eacca38d94ee535","_number":663896,"virtual_id_number":663896,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-09 22:43:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2019-06-09 16:18:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"date":"2019-06-07 12:12:06.000000000","_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"date":"2019-06-09 08:28:59.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2019-06-09 16:18:44.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-06-07 12:12:06.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-07 12:12:06.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2019-06-07 12:12:06.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2019-06-09 16:18:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-09 22:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c98e20b5a565c2844b5c64475f2f1a68d7d5a8ca","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 11:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd27e3bd073384c716b7e773750557477bf971d5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-07 14:26:30.000000000","message":"Patch Set 1: Workflow-1\n\nTo not hurt the CI","accounts_in_message":[],"_revision_number":1},{"id":"863be3c798e56f8103335cb94c7cc218b1590920","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 18:14:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/96/663896/1/check/tripleo-ci-centos-7-containers-multinode/85f5bfb/ : SUCCESS in 2h 00m 21s\n- tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/96/663896/1/check/tripleo-ci-centos-7-undercloud-containers/e5eaf0f/ : SUCCESS in 1h 30m 02s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/96/663896/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/96546b9/ : SUCCESS in 2h 05m 43s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades http://logs.openstack.org/96/663896/1/check/tripleo-ci-centos-7-scenario000-multinode-oooq-container-upgrades/4cc1a0a/ : SUCCESS in 2h 31m 55s (non-voting)\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/96/663896/1/check/tripleo-ci-centos-7-standalone/c448cb4/ : SUCCESS in 1h 59m 42s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/96/663896/1/check/tripleo-ci-fedora-28-standalone/ecbcb97/ : SUCCESS in 1h 59m 34s\n- openstack-tox-pep8 http://logs.openstack.org/96/663896/1/check/openstack-tox-pep8/ca7d60b/ : SUCCESS in 5m 31s\n- openstack-tox-py27 http://logs.openstack.org/96/663896/1/check/openstack-tox-py27/316d93d/ : SUCCESS in 4m 49s\n- openstack-tox-py35 http://logs.openstack.org/96/663896/1/check/openstack-tox-py35/5973358/ : SUCCESS in 5m 11s\n- openstack-tox-py36 http://logs.openstack.org/96/663896/1/check/openstack-tox-py36/eea7a07/ : SUCCESS in 5m 20s\n- openstack-tox-docs http://logs.openstack.org/96/663896/1/check/openstack-tox-docs/7f3396e/html/ : SUCCESS in 4m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/663896/1/check/openstack-tox-lower-constraints/c1c5ee4/ : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"1a1f4490b19766e06521e3e86b0516f82640dc02","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2019-06-09 08:28:59.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"47ae74105853e7e4311111fefc149c514331e129","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-09 16:18:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9275caf0f6aaf9a42eb72e7cb8c897786d9c8ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 16:19:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"688d361e635e470283af309083913ffac00f1399","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:43:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c2d59cc6d1418a6fc3f916c2104ed7f14c0d92d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-09 22:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/96/663896/1/gate/tripleo-ci-centos-7-containers-multinode/8febb12/ : SUCCESS in 1h 53m 32s\n- tripleo-ci-centos-7-standalone http://logs.openstack.org/96/663896/1/gate/tripleo-ci-centos-7-standalone/f93424b/ : SUCCESS in 1h 49m 36s\n- tripleo-ci-fedora-28-standalone http://logs.openstack.org/96/663896/1/gate/tripleo-ci-fedora-28-standalone/879a87d/ : SUCCESS in 1h 47m 04s\n- tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates http://logs.openstack.org/96/663896/1/gate/tripleo-ci-centos-7-scenario000-multinode-oooq-container-updates/bac2da9/ : SUCCESS in 2h 19m 00s\n- openstack-tox-pep8 http://logs.openstack.org/96/663896/1/gate/openstack-tox-pep8/5417ad4/ : SUCCESS in 5m 12s\n- openstack-tox-py27 http://logs.openstack.org/96/663896/1/gate/openstack-tox-py27/44ec46e/ : SUCCESS in 5m 02s\n- openstack-tox-py35 http://logs.openstack.org/96/663896/1/gate/openstack-tox-py35/3896095/ : SUCCESS in 5m 14s\n- openstack-tox-py36 http://logs.openstack.org/96/663896/1/gate/openstack-tox-py36/c28e595/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/96/663896/1/gate/openstack-tox-docs/b3d5c8f/html/ : SUCCESS in 4m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/663896/1/gate/openstack-tox-lower-constraints/6936e62/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f702793833a00199f98de8f93f2a4570528f9df1","revisions":{"f702793833a00199f98de8f93f2a4570528f9df1":{"kind":"REWORK","_number":1,"created":"2019-06-07 11:09:08.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/96/663896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-validations","ref":"refs/changes/96/663896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/663896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/663896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-validations refs/changes/96/663896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-validations refs/changes/96/663896/1"}}},"commit":{"parents":[{"commit":"4af560c8c2dc830afbbe20dc3b2d7606eab7af75","subject":"Add option to fail if ceph-ansible is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/4af560c8c2dc830afbbe20dc3b2d7606eab7af75"}]}],"author":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:06:37.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2019-06-07 11:06:37.000000000","tz":120},"subject":"Sync \u0027validations/no-op-firewall-nova-driver.yaml\u0027 with the new role","message":"Sync \u0027validations/no-op-firewall-nova-driver.yaml\u0027 with the new role\n\nTo be able to maintain stable branches, the validations in the \u0027validations\u0027\ndirectory have to be synchronized with the new role created with the\nvalidation framework.\n\nThis patch synchronizes the no-op-firewall-nova-driver validation.\n\nChange-Id: Ib452bbf1142ea59eb34ed043f463da48a21081df\nSigned-off-by: Gael Chamoulaud \u003cgchamoul@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f702793833a00199f98de8f93f2a4570528f9df1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-validations/commit/f702793833a00199f98de8f93f2a4570528f9df1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
